)]}'
{"id":"openstack%2Fneutron~707898","triplet_id":"openstack%2Fneutron~master~I85a02d8b59e8a758826c110e00df84672fa93676","project":"openstack/neutron","branch":"master","topic":"bug/1828205","hashtags":[],"change_id":"I85a02d8b59e8a758826c110e00df84672fa93676","subject":"Workaround in NetworkSegmentRange OVO until \"project_id\" migration","status":"MERGED","created":"2020-02-14 18:50:44.000000000","updated":"2020-02-19 21:09:52.000000000","submitted":"2020-02-19 21:07:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707898-1582146426702-7c03db43","meta_rev_id":"7f63719b03fa881c40f8a6900da8915716a18e63","_number":707898,"virtual_id_number":707898,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-19 21:07:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-17 14:09:31.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-19 15:55:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-02-19 13:11:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-02-19 18:33:51.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-02-19 18:33:51.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-14 21:36:28.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-17 14:09:31.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-02-19 13:11:15.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-02-19 15:55:17.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-02-19 18:33:51.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-02-19 21:07:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"15bee5303c16b49dbdc5b23e2b4898f53051023d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-02-14 18:50:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ea88af7eb0e169717d402e9fc931112d44b3ec7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-14 19:52:51.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/98/707898/1/SRIOV-neutron-macvtap : SUCCESS\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/98/707898/1/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"23ee7986920c59db1f7700a9cf8cf3e92727e9e2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-14 20:36:06.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/707898/1 : FAILURE in 1h 44m 45s","accounts_in_message":[],"_revision_number":1},{"id":"eda4f7d628fc152d1ff4e78f9080a9420401d2a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 21:24:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/19fbc363e4314da69b1279282e389254 : SUCCESS in 50m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d9fd58cae30147ee99ef926ceaebad9d : SUCCESS in 38m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/883b8b6be1d54c1381971d6defbfefe3 : FAILURE in 1h 44m 32s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e8c5df25ff8b43f08ca5436b8ecb3eb2 : FAILURE in 1h 21m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/df4c34fe303841aea325ef2b18a04719 : SUCCESS in 1h 11m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f538e80b8f7c4a78b8bfb31e7720d1fe : SUCCESS in 1h 07m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6641f8eec92404d84133d0c03486dbd : FAILURE in 36m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0b79d3a7a0d4c31874066deb19d2d29 : FAILURE in 31m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21a1ec3e8fdc4adf96f7ce2b5d81b9a9 : SUCCESS in 14m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8109d1f5bc664e68857e9b052f092dba : FAILURE in 27m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d0c9bb5f98c49fd8f0a3ef7e10d8cb4 : FAILURE in 30m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f724cd69dc84fd18aef74ea37dfdcb9 : FAILURE in 26m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be63fa9275e7434bbc5761ff5bda24ce : SUCCESS in 10m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/71e01e4f0ada4f5ba222edd29eea1143 : SUCCESS in 40m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/aba242fa56834893a2fb8ec4f4660b0d : SUCCESS in 1h 03m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/01f016fb4a5148ed80e669ef011f4968 : SUCCESS in 1h 56m 27s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ef6a61f6649a4a6e86eb2d6a1d247f22 : SUCCESS in 1h 17m 23s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/56ff8c06f79e465e8cbda569fa39866d : SUCCESS in 1h 19m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2421add6eef54620b6d4c335e043a99f : SUCCESS in 1h 02m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/687f5f245b7e42b7a4d2d7ede0d9103a : SUCCESS in 1h 10m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/16335612b6aa4bb4b3bc9f5ab0c54268 : SUCCESS in 1h 08m 14s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ba12a72e2cd74143bc86926b69f2676b : SUCCESS in 1h 41m 18s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b8fb57c6f5f34d0ca0d50763d840de39 : SUCCESS in 1h 49m 05s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2490d783ac5a431e97991e5b3dd47c56 : SUCCESS in 1h 20m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e757acaaf6264e0cb8875283b96dad2d : SUCCESS in 53m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/945634458e6f49a2b4b5179969393f0c : SUCCESS in 2h 23m 55s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d43c485b90fb41a2907675179ef1e3f7 : SUCCESS in 1h 50m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bd95b6c438c54133b0da34ddc51b1c2e : SUCCESS in 2h 00m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/426d008807f14cee8a7e87579e96bcfc : SUCCESS in 1h 09m 21s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8a2b45c052b44264a1a092ed13bd4725 : SUCCESS in 1h 14m 50s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/efec36f041754b7b966b2b179dd36d6e : SUCCESS in 41m 12s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a14f19b10c514739aea3640a9324eef2 : SUCCESS in 1h 25m 47s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/dcdbcdb23fc74e83a5887aaeb7555584 : SUCCESS in 1h 11m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2801535876d748c69f2e3788db1af06d : FAILURE in 2h 27m 25s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/ba580ce3678d471e80aed75773eada1a : FAILURE in 2h 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8168fd877efaf70b6ec5d24160951e4782ff828","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-14 21:36:28.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/98/707898/1/SRIOV-neutron-macvtap : SUCCESS\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/98/707898/1/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"5b020e33a9a78b5c1222bccdc2f48519d511abf7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-02-17 12:10:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c96e43c175aec3217b3eed377b01943a8e001498","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-17 14:09:31.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/707898/2 : FAILURE in 1h 58m 54s","accounts_in_message":[],"_revision_number":2},{"id":"2006b5de111dbd4ae48ffb3638da4289ec758d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 16:07:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/70b6f4003fa04943bce99e0e71ab80b4 : SUCCESS in 51m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b469742ae90a4ec89b1c43aa51d8d341 : SUCCESS in 37m 53s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/1c671032a34a4bbcabb731a51d20c7b9 : SUCCESS in 1h 37m 44s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a6d16fea18d04636b2c0315d998ecc4e : SUCCESS in 1h 26m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7786469211b14d3e9e4a3c99df3e6655 : SUCCESS in 1h 19m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e50b8c2f5cf14e30b7f9832b1b7c1712 : SUCCESS in 1h 18m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50496cf0e92a4762b5c2668153d23df5 : SUCCESS in 38m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/027a1c8ee12b4e2aab5dedcc73a38429 : SUCCESS in 31m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0854fe84b374d54977e2e9b70ab4b8a : SUCCESS in 16m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ba06897c4424b5086f7faf2b7e9d21c : SUCCESS in 29m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71a57a9d557c47d083b66a020108bf37 : SUCCESS in 32m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98b0bc843d6d4549b25d8e12f690ec93 : SUCCESS in 26m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7853a9bfb3c4431ea456eddb19bb0c5a : SUCCESS in 10m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/158507a3df5c40b693d8934d9945d2ed : SUCCESS in 46m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1a88359c87ec4bcca76870338ce0fbd3 : SUCCESS in 1h 12m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/977d3fbfac4949f483e63cf228b5d39c : SUCCESS in 1h 59m 41s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ec4e8928277e4ba6b594d12722294a7a : SUCCESS in 1h 26m 32s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/af85ecc57ef640ea9668cf3ea5be8bee : SUCCESS in 1h 14m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/67152c424c634259907648d00ff51bca : SUCCESS in 1h 10m 49s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/50fbd0b9985a4a68b707022771b651df : SUCCESS in 1h 13m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/84cd568c677345c9a7818bc344c7123b : SUCCESS in 1h 26m 06s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f64a81c36a2441289700d91e23068e11 : SUCCESS in 2h 07m 56s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9ed232fbbdc04a69acbf00e38d7cd728 : SUCCESS in 2h 04m 32s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/14b2697855324b7496ea2b0e790d7472 : SUCCESS in 1h 20m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/84ba28d0fb0841ac9f0194ac0281bc57 : SUCCESS in 56m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a18a85a466224dda8ed03160a4b08399 : SUCCESS in 2h 31m 11s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bacb1d72588b4ff291eb943309d90166 : SUCCESS in 1h 42m 11s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f4de9b103d4d4af385cdfce078e0facc : FAILURE in 2h 31m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b21dc2edc4264214b48fd0d9a9497562 : SUCCESS in 1h 07m 39s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/64a26a6daa084e2e96032f5166e93378 : SUCCESS in 1h 09m 57s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/11e1e7f4fea04178a8326c81223d7665 : SUCCESS in 49m 11s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0ef9c716453247ca99dcc7f265790a82 : SUCCESS in 1h 13m 53s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/16e64eec5a0b4973a244950c48559a53 : SUCCESS in 1h 22m 31s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/01bf82ee266e43798a2bfe5f5eb1f552 : FAILURE in 2h 15m 16s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/662560402fd0465fa09475a5c62e6f14 : FAILURE in 2h 02m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2bb4cdc275d4179ee5e7350912b2148abc1b35ce","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-19 13:11:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8f2f1626850c516828989cb93024a9cd7cd2bdce","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-02-19 15:55:17.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c3ee2ce83ad26aae41f75769fd9374630972883b","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-02-19 18:33:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6e2713ce32ffd7f12d55b3f5b9dd025954fbccbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 18:34:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3118129731360ed60779638a9c79e40f356524fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 21:07:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e96bc7511bc5965701aeadd9be62e30e89a99649","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 21:07:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/674e8ec191104dc0bcd348c5c1237065 : SUCCESS in 1h 05m 26s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bb2835fb49dd47cca85cb93bab5309a4 : SUCCESS in 1h 04m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e23d9ef119d347f6ae07b2798b1d5bf0 : SUCCESS in 1h 18m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/27bc94bf03a14e9e9e881cab407211b9 : SUCCESS in 1h 09m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43875c05a7a64ae0bdba7c1639f52503 : SUCCESS in 28m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3872ad3e2e574202be000fe68a741c3f : SUCCESS in 15m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08e100c4c025490581c605a17590e241 : SUCCESS in 23m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21f2c9c4896746c69f1ed09e4cc79c68 : SUCCESS in 29m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb4cc8c83cb6483587b47a5f8f2cd61c : SUCCESS in 11m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d4a09016e427453fb1cd180de33a3570 : SUCCESS in 52m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/07050f71f013438bb555981f4a15217b : SUCCESS in 1h 04m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dbf726fb8104412c97bce9892e561e2f : SUCCESS in 1h 33m 41s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1a7238e907374ca6bd681459e4bf70af : SUCCESS in 1h 11m 11s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/290d2e8224b54d218efa1354f61f94d3 : SUCCESS in 1h 11m 24s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8b496ffceca041f8a6ab09e9cbc8bb1e : SUCCESS in 1h 14m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/efbbbb0a08554c6f96ba42bdb50af156 : SUCCESS in 1h 22m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ea14f5036a31477982a912e7d3d63912 : SUCCESS in 1h 11m 12s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/08716268eeba49539fd301e81f88f098 : SUCCESS in 2h 25m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69170cbd2560423ca3f3b864e0ba6413 : SUCCESS in 1h 21m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fc9b16ebd83d4e29a2c46032ff675c07 : SUCCESS in 1h 46m 07s","accounts_in_message":[],"_revision_number":2},{"id":"e01ccd736bf113c55b774584316b3b62dfb40655","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 21:09:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fb5a4819b1f4255b6283c31a0ac9e6a : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3053caaa02adaadfffa53fb1070aaa60f49419e8","revisions":{"ba565b1dfe908c1cf8a9a3add1becd0a86639f12":{"kind":"REWORK","_number":1,"created":"2020-02-14 18:50:44.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/98/707898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/98/707898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/707898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/707898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/707898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/98/707898/1"}}},"commit":{"parents":[{"commit":"7a0e5185c6cf7b5f8bcfe50576e86798947a7ba7","subject":"Merge \"Add \"ncat\" rootwrap filter for debug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a0e5185c6cf7b5f8bcfe50576e86798947a7ba7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-02-14 18:40:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-02-14 18:50:36.000000000","tz":0},"subject":"Workaround in NetworkSegmentRange OVO until \"project_id\" migration","message":"Workaround in NetworkSegmentRange OVO until \"project_id\" migration\n\nThe \"network_segment_range\" service plugin API does not have\n\"tenant_id\" as parameter. However, because the migration from\n\"tenant_id\" to \"project_id\" did not finish, every OVO with\n\"project_id\" will also have \"tenant_id\" as an extra field. This\nextra field will be populated with the same value as the\n\"project_id\" field.\n\nIf the Neutron API WSGI is used and \"tenant_id\" is not present in\nthe API (defined in neutron-lib), both \"tenant_id\" and \"project_id\"\nparameters will be removed from the response.\n\nThis patch introduces a workaround for this OVO until the migration\nis over.\n\nMore information about this patch can be found in the bug c#4.\n\nChange-Id: I85a02d8b59e8a758826c110e00df84672fa93676\nCloses-Bug: #1828205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba565b1dfe908c1cf8a9a3add1becd0a86639f12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba565b1dfe908c1cf8a9a3add1becd0a86639f12"}]},"branch":"refs/heads/master"},"3053caaa02adaadfffa53fb1070aaa60f49419e8":{"kind":"REWORK","_number":2,"created":"2020-02-17 12:10:06.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/98/707898/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/98/707898/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/707898/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/707898/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/707898/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/98/707898/2"}}},"commit":{"parents":[{"commit":"7a0e5185c6cf7b5f8bcfe50576e86798947a7ba7","subject":"Merge \"Add \"ncat\" rootwrap filter for debug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a0e5185c6cf7b5f8bcfe50576e86798947a7ba7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-02-14 18:40:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-02-17 12:09:50.000000000","tz":0},"subject":"Workaround in NetworkSegmentRange OVO until \"project_id\" migration","message":"Workaround in NetworkSegmentRange OVO until \"project_id\" migration\n\nThe \"network_segment_range\" service plugin API does not have\n\"tenant_id\" as parameter. However, because the migration from\n\"tenant_id\" to \"project_id\" did not finish, every OVO with\n\"project_id\" will also have \"tenant_id\" as an extra field. This\nextra field will be populated with the same value as the\n\"project_id\" field.\n\nIf the Neutron API WSGI is used and \"tenant_id\" is not present in\nthe API (defined in neutron-lib), both \"tenant_id\" and \"project_id\"\nparameters will be removed from the response.\n\nThis patch introduces a workaround for this OVO until the migration\nis over.\n\nMore information about this patch can be found in the bug c#4.\n\nChange-Id: I85a02d8b59e8a758826c110e00df84672fa93676\nCloses-Bug: #1828205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3053caaa02adaadfffa53fb1070aaa60f49419e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3053caaa02adaadfffa53fb1070aaa60f49419e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
