)]}'
{"id":"openstack%2Fneutron~708442","triplet_id":"openstack%2Fneutron~master~Ib8076ee711a0e5712d30193aca3335477856e61d","project":"openstack/neutron","branch":"master","topic":"bp/neutron-ovn-merge","hashtags":[],"change_id":"Ib8076ee711a0e5712d30193aca3335477856e61d","subject":"Fix compute-only ovn sample config","status":"ABANDONED","created":"2020-02-18 16:21:43.000000000","updated":"2020-02-19 11:52:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d6ddaa8f40e111d1239fb9dd2ccb5bb0c747d2b4","_number":708442,"virtual_id_number":708442,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-19 02:10:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-02-18 18:22:06.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":-1,"date":"2020-02-18 16:32:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-18 16:32:33.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-02-18 18:22:06.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-02-19 02:10:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d7dcd8ec46cfec9c96ccfac154638cb73d762c5","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-02-18 16:21:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5450a3556a2ccdad1cb9da68ef9d48bfb60e259","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-18 16:32:33.000000000","message":"Patch Set 1: Code-Review-1\n\nThanks Bence for the patch!\n\nI\u0027ve recently played with this sample file and pushed a fix (which include ur fix) at https://review.opendev.org/#/c/706824/","accounts_in_message":[],"_revision_number":1},{"id":"43416730bf6d4c03c6068c440e563360009ac7b9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-18 18:22:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/708442/1 : SUCCESS in 1h 50m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1ae36bfc75101a4b18652cdad1c16d41e0683964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 02:10:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/26041fad9aea465c8d59434c82744b4b : SUCCESS in 48m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1f8a6944f6fa4d64912d2826361a0543 : SUCCESS in 45m 06s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/023a2a1f5c3f47579547555993c91648 : FAILURE in 2h 03m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4077f289bf5447faae4344fbc5ba7f1a : SUCCESS in 1h 20m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/25e02348634842b5a1e64b8994c584f5 : SUCCESS in 1h 12m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b8b554ac517349788371f59dd62d19b6 : SUCCESS in 1h 07m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ced191be9b84acabb6230d1e4d35aa6 : SUCCESS in 36m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f76096e0944e424ca1218e6d5456805b : SUCCESS in 25m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17608720d9a240e1bf98f40ac9aa5afd : SUCCESS in 13m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f5a70092225442c9e014befe291a99f : SUCCESS in 27m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a4f87f44745455f8d0d5b81a9194f6a : SUCCESS in 29m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf5927d7573746bca8561dd19949def4 : SUCCESS in 27m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7b33841e91a46ccb69e0b09ce6eecf6 : SUCCESS in 14m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/983eef588a0741a2bbc104ba332e3156 : SUCCESS in 44m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/177965cb52294973b74a7894e8b969d8 : SUCCESS in 1h 00m 06s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d1d95d7afd6d420c86e1b8d614ea9bbf : SUCCESS in 1h 56m 11s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/9745636889dd4c02a47ee61a25f2f307 : SUCCESS in 1h 14m 33s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9ffb36f03c81409180cc093c118d7f9d : SUCCESS in 1h 11m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e8aa0cd8238841519b1bbefba3580f55 : FAILURE in 45m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/22a22e1bd83a49af9a7d2dfdc5574c23 : SUCCESS in 1h 20m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4a116c76afa044ca980ec794f933a7ee : SUCCESS in 1h 17m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ee78261729db42fd810cc39f8c9b5cc9 : SUCCESS in 1h 49m 25s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2022aa419a0c462ba6c28793b0e94d41 : FAILURE in 1h 27m 24s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d2af5edea7ee4d4895bb394471010fd9 : SUCCESS in 1h 10m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fafdbb79e8db471da247fba5e049961e : SUCCESS in 59m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5d33d72c43e14be2b9121b6e633b163e : SUCCESS in 2h 24m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ab4aa7a2364044f1a2581792b608d518 : SUCCESS in 1h 42m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f33f630a3d0441d7949489527d48176e : SUCCESS in 1h 52m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e6e19d3669544379873cb9d726f281d : SUCCESS in 1h 00m 56s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/89d924ec949c40ff877ef1d8da482e04 : SUCCESS in 1h 21m 19s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b566b4af9a4643a6b54563d09e505015 : SUCCESS in 40m 45s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2829516ff18248d084571c2face28b69 : SUCCESS in 1h 00m 54s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/208c1c5c360c45e2948c5786252d4410 : SUCCESS in 1h 09m 10s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/03469c6332a741ce8804eb19aad53e0f : FAILURE in 2h 53m 30s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/58fb6f0fb0bb4897b01f7abc85b819e3 : FAILURE in 2h 01m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fb493a89ce5319338d13f9bad086172dae435a9","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-02-19 11:52:29.000000000","message":"Abandoned\n\nAbandoning this in favor of Lucas\u0027 change. Thanks!","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e6ea2df1a46ee62f3ae64b689fd246daa32617ca","revisions":{"e6ea2df1a46ee62f3ae64b689fd246daa32617ca":{"kind":"REWORK","_number":1,"created":"2020-02-18 16:21:43.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/42/708442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/42/708442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/42/708442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/42/708442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/42/708442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/42/708442/1"}}},"commit":{"parents":[{"commit":"7b567d64492ee8237686627f4b33764cdc99dec0","subject":"Limit ovn sanity checks to q-svc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b567d64492ee8237686627f4b33764cdc99dec0"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2020-02-18 14:18:36.000000000","tz":60},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2020-02-18 15:08:48.000000000","tz":60},"subject":"Fix compute-only ovn sample config","message":"Fix compute-only ovn sample config\n\nWithout setting Q_AGENT\u003dovn devstack/lib/ovn_agent is not called\ntherefore ovn-* ovs* and q-* devstack services are not started.\n\nChange-Id: Ib8076ee711a0e5712d30193aca3335477856e61d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6ea2df1a46ee62f3ae64b689fd246daa32617ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6ea2df1a46ee62f3ae64b689fd246daa32617ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
