)]}'
{"id":"openstack%2Fneutron~709409","triplet_id":"openstack%2Fneutron~stable%2Fqueens~Ieded0654d0ad16235ec923b822dcd842bd7735e5","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1732067","hashtags":[],"change_id":"Ieded0654d0ad16235ec923b822dcd842bd7735e5","subject":"Add VLAN type conntrack direct flow","status":"MERGED","created":"2020-02-24 05:16:50.000000000","updated":"2020-03-09 23:42:21.000000000","submitted":"2020-03-09 23:39:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709409-1583797193266-dc82b96e","meta_rev_id":"9b59cf2ed38092def8f839eb2cfded36d9b5c84d","_number":709409,"virtual_id_number":709409,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"date":"2020-02-28 13:03:34.000000000","_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-09 23:39:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-04 09:35:59.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-03-03 16:38:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-03-09 22:08:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-03-09 22:08:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-03 16:38:45.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-03-04 09:35:59.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-03-09 22:08:07.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-03-09 23:39:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9dea82b28b9c970e6d543f448c01574e68118a55","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-02-24 05:16:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b548b86f5f9b3e30d76143b76d7c6a36147247b","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-02-24 05:22:47.000000000","message":"Topic changed from bug/1732067 to bug/1831534","accounts_in_message":[],"_revision_number":1},{"id":"50f2da2499a2e8d38d855690b51a007875f567f3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-24 05:24:51.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/709409/1 : FAILURE in 7m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ef5dc3f84631e868f5694cbd7c6fe58c016ab287","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 07:16:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/4fffe58e06794afa93175553f54d4bee : SUCCESS in 1h 00m 10s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/dec66e10ae3e4e67b2f6fbb73802f64e : SUCCESS in 42m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/ee21da342695413386a5d4880a5e8611 : FAILURE in 1h 18m 38s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/3f955413e5f94dcd9659bab7e797acb6 : SUCCESS in 56m 07s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/b412f5c67f90432091ec9cdcb2284d1e : SUCCESS in 56m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1713abd935f4ed6b80a34e14c25d9dd : SUCCESS in 9m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3269308aa50e4fc397e850f02d9bca55 : SUCCESS in 29m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2709c27ad83b4227b980637b7d8d050c : SUCCESS in 24m 33s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/305bcba8f12547af842c5f704f18b703 : SUCCESS in 6m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/80d1ca2f7a6e4ff08eb48af6877c50e3 : SUCCESS in 1h 17m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d3be4726821e42e0ac8a051ba25f0187 : SUCCESS in 55m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5697a1c600e64f3bb820779ab76d84af : SUCCESS in 1h 12m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ce0b7243e4a4ce784307675fe9ba3c5 : SUCCESS in 12m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7d9b4c7a3b4d4f0f8e69f8627544cc0f : FAILURE in 47m 19s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8c8c5a3da3cd440cb4ace44a806a5454 : FAILURE in 28m 51s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/9c3eb7850cbf472ea1192c58edbf327c : SUCCESS in 1h 37m 40s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b5dc1deb2a524865a8fe51837f949918 : SUCCESS in 1h 21m 11s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b04fa369d8af499c86d3094dfabfdb96 : SUCCESS in 1h 11m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/700b2c1978c742b58c8118bca2f74a08 : SUCCESS in 1h 12m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cc846a1e45264c5b8d87a022f51b6351 : SUCCESS in 1h 09m 51s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3ea7bce3a60245f29a67773f4fd518b5 : SUCCESS in 1h 20m 21s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3f7e5e8834c743ad87af991e13f152eb : SUCCESS in 1h 33m 06s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/e89cb9697268476c8195e73b4f094631 : SUCCESS in 1h 09m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1705b7802e7450d9db0db79c2bde51b : SUCCESS in 35m 05s","accounts_in_message":[],"_revision_number":1},{"id":"61a1ccef29c717d3acc07df77ebdf3e69030192e","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-02-24 07:44:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a3df67ae4d8c51b5012b9234b4a961cf55257bd9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-24 07:52:11.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/709409/1 : FAILURE in 7m 03s","accounts_in_message":[],"_revision_number":1},{"id":"03eec761a4b2895141d66cfd6648616ba7341a95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 09:28:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/9b49425f823d4042b2048579b7ec93b5 : SUCCESS in 41m 43s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/f07d03a1ad564b35adc1a6d8f1fb29b1 : SUCCESS in 34m 25s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/c5c84860e96947c6af0850504a875ff5 : SUCCESS in 1h 09m 18s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/2e89bdf7ad9240cf9cf1fbb67e200132 : SUCCESS in 46m 47s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/c9d68c77b2a44da69e4087274dc2dfee : SUCCESS in 50m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72242f8daa004245866279b34bd63ab7 : SUCCESS in 12m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0badd13336ef4122b450b4ed69110eb7 : SUCCESS in 36m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f4739f3799cf4e098e70d7e93418ee9f : SUCCESS in 25m 09s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/698f0f9afea74c2aa2800c1563d2facd : SUCCESS in 7m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e721673c0fb34118959384bb6cb5d110 : SUCCESS in 1h 27m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/13985f6a84e7437fb18eed59f3147a81 : SUCCESS in 52m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/881d62f1ac43440c852aaeef8beafe61 : SUCCESS in 1h 16m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a0725f071d443a3952fd5b44bf607ef : SUCCESS in 15m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/770476e4a1ca4d93806f027a6c852935 : SUCCESS in 37m 22s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3a5187e407254fadb19a5e5f212103fc : FAILURE in 23m 02s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/53d2599724be461bbbb453d1056d9d84 : SUCCESS in 1h 38m 12s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ec8e7b1ceeb14e86be26b23609eac109 : SUCCESS in 1h 26m 19s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b1ebcf13907d44948075d73abbc7a44f : SUCCESS in 1h 24m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5778e4e12d344b59f8c84cd8b57b684 : SUCCESS in 1h 08m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/414b76480e4948b098204e1e30c6e622 : SUCCESS in 1h 00m 30s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9930a268fc7f48b9bef7a74543c58966 : SUCCESS in 1h 32m 32s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/09e9ca330ec74e9a8152d7170b39007a : SUCCESS in 1h 33m 39s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/5b033ca3d0ac46bbb46b4754899eeb93 : SUCCESS in 1h 27m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f101cef2a9b4b8481e0e11fad67296c : SUCCESS in 44m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7c8a0e6d0ea8164df32e947f8e20b5a110f0be42","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-02-25 15:01:22.000000000","message":"Patch Set 1: Code-Review-1\n\nBackport itself looks good to go (it changes generated flows a bit, but to fix actual issue with VLAN), but missing backports for branches between (original patch is only in train)","accounts_in_message":[],"_revision_number":1},{"id":"84b120d0c3c79d90ca4003a8bf7a15d89058fc3c","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-02-27 05:15:58.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"eb22720a17b61f3ad757d071a737db4b0e6fb6b9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-27 05:24:18.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/709409/2 : FAILURE in 7m 49s","accounts_in_message":[],"_revision_number":2},{"id":"1816ecfaf1212f54ed39bb5d901811f20aa76f3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 12:10:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/5171b053ed36452f88d47d0387889067 : SUCCESS in 45m 53s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/61bc0c4ee75741818eee5da467740c5b : SUCCESS in 34m 42s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/5f48c989276b4dafa407c5ac7882b468 : RETRY_LIMIT in 1h 57m 45s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/d6ba931bb85d4c839cd634adfa198f1c : SUCCESS in 54m 43s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/fe6fd9be9ea9444594575392c5ce3717 : SUCCESS in 47m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/799aa062dcf14391ba6f5ef0524dc4a7 : SUCCESS in 11m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/686d5cae6a8946e0acf705ce0e0b4af8 : SUCCESS in 24m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6f1e43dac72f4b9f872aad4c5319d231 : SUCCESS in 25m 07s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/67f5cffa7d774fe598534326d5d3b27d : SUCCESS in 9m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/aced5848d763408381eb797f8164f879 : SUCCESS in 1h 24m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a4ea0c7989214df09c111746f2562c6b : SUCCESS in 57m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00221e94e1634d33beb19b9b97fd46b5 : SUCCESS in 1h 39m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5277d4064634f509b4af4c941688759 : SUCCESS in 15m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/13be7caf51244fcb9b71f920511b2ee3 : SUCCESS in 43m 33s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3293662a485c461ebc1bdd081c8b052c : SUCCESS in 50m 39s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/cba110f4e1ae4900b567180f6deef15f : RETRY_LIMIT in 1h 39m 05s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/875fe926634a4168a9d4fac442242acd : SUCCESS in 1h 20m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8407685fe0664832b01bd036e1d58620 : SUCCESS in 1h 32m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1e3111f63154be39aed8a85f9ca8d9c : SUCCESS in 1h 08m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7edc0105471e4caeac82cc8e92153580 : SUCCESS in 1h 06m 46s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/235b85b1cc214e53ba751274b463a8ad : FAILURE in 1h 41m 38s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/756f1a32337a42248d71aff38cd7f699 : RETRY_LIMIT in 1h 38m 43s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/b5a5d9930f96450eafbb9aa52f16f51e : SUCCESS in 1h 28m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6ba143609504ad599bb2cd0bdaf96bb : SUCCESS in 31m 34s","accounts_in_message":[],"_revision_number":2},{"id":"737725e69d9ca6afa8c2042f1fd2e8df9e4b1872","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-02-28 01:13:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6d7c985fca34b3a283ada83654cfb5af555ccf51","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-28 01:23:32.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/709409/2 : FAILURE in 9m 09s","accounts_in_message":[],"_revision_number":2},{"id":"2e68e2eda989a462b09c4c5309f7b90451f6219e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 04:42:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/6c0ff50f5b494f00a66f8cc3ea08a1b2 : SUCCESS in 51m 55s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/83b8aeb1ee914a6f9bd75cca081a6b01 : TIMED_OUT in 1h 03m 43s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/9584e04e26f242b1a4724fb1a077ab47 : FAILURE in 1h 20m 45s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/eda94257697e4ba680b94b7c2e348ca0 : SUCCESS in 53m 14s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/e97db3000f0147cb93b3494465220e60 : SUCCESS in 1h 02m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b5a5a46104f4a81bcf8e8c9b45d68dd : SUCCESS in 11m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/83bc9e9550dc4fd2bd901aa4e31d7ced : FAILURE in 25m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/83143a1cf36241e089ab14cea5118012 : SUCCESS in 29m 12s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8c6f79aecdac4f13873d61cc73b6e0ed : SUCCESS in 7m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e73edc80a27c4d1c876afb00b5b31246 : SUCCESS in 1h 24m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/369e6bd472ff43ea949d1e616cd67f90 : SUCCESS in 51m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8704ab7cc96427a82bc2d6f039e2ab3 : SUCCESS in 1h 23m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00e2639a7b9a49e8b42217d931cb9740 : SUCCESS in 14m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/67d3ab1d026c4eff9310de40dba9d8c1 : SUCCESS in 35m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/735e912f27ed44c794e7330c3a257ca5 : SUCCESS in 43m 38s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/fe492c22e1af4c3ca2de1bf63b1c5bb5 : SUCCESS in 1h 48m 14s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3d7607f3f877464b9c6a201d06738db7 : SUCCESS in 1h 24m 16s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/82bf40afe66243d9b6ea95e4984fc6fb : SUCCESS in 1h 19m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4429f89f1a8c41bdaf582457f2564853 : SUCCESS in 1h 16m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bd3d3e25d795430484adf4dd409c2a2d : SUCCESS in 1h 11m 13s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c843ffedd31947e8977ac56f0277bf8b : SUCCESS in 1h 40m 11s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8f6ccb986bc64153a353f0dbe2eeb5bf : SUCCESS in 1h 29m 45s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/fe9459a374654c6d8c67fa9914555c3e : SUCCESS in 1h 30m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0647a6892f0481d8dfa27aeb5cbfe9a : FAILURE in 33m 52s","accounts_in_message":[],"_revision_number":2},{"id":"4ab3519a61d0c3a3535fd48d5fb39ceac593e5cd","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-02-28 13:03:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7bb748f7bb123a98f0375c944e8b5d47ecbb909a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-28 13:55:23.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/709409/2 : FAILURE in 8m 05s","accounts_in_message":[],"_revision_number":2},{"id":"86fbee4cfb3b0e2248fd8fb1b966ad56b55bd856","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 16:08:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/0234ce1efa3246069920266f595ace90 : SUCCESS in 59m 16s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/bbbab518cd644d58bb956cc9c3e7ca43 : SUCCESS in 36m 59s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/4efa6dd5819f49c6955537711fdb8b69 : FAILURE in 1h 26m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/a118ff0a74624ce09df16d36536fa27a : SUCCESS in 1h 03m 10s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/6ba328328f614079ac5ec5d83c342dcc : SUCCESS in 1h 01m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1649caf1766a4ad9966a3d6c02d1b142 : SUCCESS in 12m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7bc349f5dc54360ae126c2ef22dc523 : FAILURE in 28m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/633a41f449c8443e875d5b72ca199bd6 : SUCCESS in 21m 30s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a09ee2adc6764372905d05b4514ee594 : SUCCESS in 7m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ea774138f8664eeb9d764d6d5fdd00b0 : FAILURE in 1h 21m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6a38acfffadb4e16a227dcd9d231cc62 : SUCCESS in 52m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97a8f92ae03b4347ae97146372248900 : SUCCESS in 1h 20m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e56439679e37480dabbaa34e56741880 : SUCCESS in 15m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7578ec2ec28e442fa97316691e1691af : SUCCESS in 43m 19s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4309377f39ec49daa132acbd82b37a23 : FAILURE in 47m 43s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/a7e94bdabbea4125b49d4506d03b1272 : SUCCESS in 1h 45m 21s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/4a5f9e90549e4572af7b2342442478d5 : SUCCESS in 1h 24m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/814278f778504a469865b50e573c62ca : SUCCESS in 1h 25m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/74c6f885df4640d2bdb37f97f4e4ff71 : SUCCESS in 58m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c4bd7162c4e142a3bfd156998c7f34e5 : SUCCESS in 1h 02m 06s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5fc5802904544a1289cf11039afa34f2 : SUCCESS in 1h 37m 01s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/04e6458d376b4cdeb9d9801c41bf12c9 : SUCCESS in 1h 40m 15s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/3faab18423174da2a6cf8234fa8e79f0 : SUCCESS in 1h 15m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1d30648e944496dad532d56ea13a2e4 : FAILURE in 36m 51s","accounts_in_message":[],"_revision_number":2},{"id":"29eda685de5b4a5f722bb3485c919ef3cead2e8f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-01 08:25:08.000000000","message":"Patch Set 2: -Code-Review\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b167a70ba19acec6d4a31a38f81796b0faae1357","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-01 08:34:44.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/709409/2 : FAILURE in 9m 05s","accounts_in_message":[],"_revision_number":2},{"id":"56ddf20f8779b73122e3ef222e3dca025c64c95d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-01 10:09:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/7214eb7de58b4ad8a11ec055b0a52330 : SUCCESS in 53m 27s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/2f6e3fbdc6f144979a538f3125977c2f : SUCCESS in 34m 17s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/4d66b0da69fc488c98ccc419866f8865 : FAILURE in 1h 29m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/a61ee6a4a37a4c9e9b2d973bf52ab78e : SUCCESS in 50m 51s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/ff34cdf4fb304666b096d45cfeb7100e : FAILURE in 55m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71316e5cef88453ea0d9e87dc6eca3be : SUCCESS in 12m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a8a0d98eaed4a2fb06f427141729286 : FAILURE in 31m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9167d61e28a64ef2b5360d0c069d38ac : SUCCESS in 28m 26s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1a589f86e9eb4d98baf7ccab68882df3 : SUCCESS in 7m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/edf74b0798c54c76a0f75b31c7add9fe : SUCCESS in 1h 23m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7ef37de119d24dd796ea18dee2095c93 : SUCCESS in 57m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7185c67315c04f1cace2dc212970ea9f : SUCCESS in 1h 15m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8db21ed269074c5cb84fd3caabe17f48 : SUCCESS in 15m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2bb60e015f9049a598b23bf4b163404f : SUCCESS in 41m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/82ffbffcdb7d487a893f3053cb872a09 : FAILURE in 49m 24s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/97ca502992f44cf4bdebe22f759014d1 : SUCCESS in 1h 35m 40s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5f515cc84ca24f21977202c5e777c6d3 : SUCCESS in 1h 30m 50s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8072df61b71f4ca28b1b8af6a8d15a12 : SUCCESS in 1h 15m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc5b412366a84c819f2e29d08cd8ae58 : SUCCESS in 1h 04m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/afcafd404c2b48d3925d64b76f6267af : SUCCESS in 1h 05m 03s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d7d6edcaeeee48b0a52d4c1a38bcb6f7 : SUCCESS in 1h 31m 08s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/77b61984d6f743c28fd56b414440ab04 : SUCCESS in 1h 35m 44s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/d9bf905665404633945d19f126f78632 : SUCCESS in 1h 20m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d78fe6c5dfd40da9a81ced0e1ab0311 : FAILURE in 32m 27s","accounts_in_message":[],"_revision_number":2},{"id":"e9adb8732773537605639aaf121f66c1da7f1f5c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-03 16:38:45.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9717db1084dd7d608bdd58fb53dc5404f55eb96c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-03 16:47:14.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/709409/2 : FAILURE in 8m 00s","accounts_in_message":[],"_revision_number":2},{"id":"52639ecc9a01ecafcd90ceda850a190974ca2e56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 00:55:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/08194e3b5bea4198830bb780de2bd4d7 : SUCCESS in 50m 47s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/01af57c7980a4f19b0d88e8e9ca62b91 : SUCCESS in 41m 29s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/12ea7e1720ec40adbc4fefb19c2da9e7 : FAILURE in 1h 42m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/e095927479aa4ba6a9f7b60386204dbf : SUCCESS in 58m 37s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/6c9f3047085c44aeb2b735a07935eafa : SUCCESS in 52m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa9e677fc4d044d5b28a0c70dda6fb3c : SUCCESS in 11m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a5df5b80115493182892af1739d5bf0 : SUCCESS in 22m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/66517a15cff5456f8e71baf8168b5f5d : SUCCESS in 22m 52s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/41915c89c9a843478109979e50bd0c19 : SUCCESS in 7m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/eb97322ce06b44f0a139ef83c0e73028 : SUCCESS in 1h 21m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/80eff88e69d94be1995b387b8124865b : SUCCESS in 53m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c58611b336614ddbb223c75a3dd9029b : SUCCESS in 1h 24m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46cc592766fe48c78e59a6f9115c7329 : SUCCESS in 12m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/140fd38dcda440c9a134aecd778b63f8 : FAILURE in 45m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c0408c046d27493bbe59638d7c6c200c : FAILURE in 43m 22s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/49d48bb986154ad9ab34ee68a8232436 : SUCCESS in 1h 43m 15s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c9af1c44a88c4022a86590f5a4cfdf04 : SUCCESS in 1h 22m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a1679dec708e4f5a9a52e72a15490a2f : SUCCESS in 1h 19m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/453dc828514b46e690dc14893ac28735 : SUCCESS in 1h 00m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c3d065a6583f4bb7b65618cab3078366 : SUCCESS in 1h 02m 20s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a58959950d1647ed8c3aab22876cc362 : SUCCESS in 1h 37m 46s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4cc20c83e9d346a08298e7680dc095fb : FAILURE in 1h 35m 05s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/afceb91d997445c098444121b8108de4 : SUCCESS in 1h 17m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f4fc87e1138434b913e48e6cd81650b : SUCCESS in 38m 53s","accounts_in_message":[],"_revision_number":2},{"id":"912e65f3a2687b327e37c15e2f7c763faec27c41","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-04 09:22:41.000000000","message":"Patch Set 2:\n\nrecheck functional","accounts_in_message":[],"_revision_number":2},{"id":"5f0493a8f8a29af7ccb85ff9730ec62b85a5842d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-04 09:35:59.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/709409/2 : FAILURE in 8m 05s","accounts_in_message":[],"_revision_number":2},{"id":"21a60c1dd4b36438cc38dab33841e772b2199dc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 12:23:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/0b3a7918934d4b64b100082094f06ecd : SUCCESS in 50m 00s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/3e1eedf9a9f24dfd81bff74d995457ff : SUCCESS in 40m 04s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/568d5e5439c44fcb9f5d8e8da2058fa9 : SUCCESS in 1h 26m 10s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/848ba785e75d4d20ab3ce11e31316851 : SUCCESS in 52m 16s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/4c7f610d08f64ebda5de289bffddb209 : SUCCESS in 53m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29bfcb39f791485dbb1c7962fc798fcb : SUCCESS in 12m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe2e15c8ec944b759f8c7e3fcd56dfb2 : SUCCESS in 26m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c8e577d5a9dc4150b2468d90dadae753 : SUCCESS in 30m 59s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/bbacf15c7c6349b194d66113d48c3336 : SUCCESS in 7m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cfe2c4f3d00a4dc6a55a4ca4fe5d233f : SUCCESS in 1h 24m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bbb55e371d1940bf89890dcb0c7cf942 : SUCCESS in 58m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef81f3dd082c47fa92378e8fb2d1d74b : SUCCESS in 1h 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3544e12dfd1f4fb2b61de498332ec97d : SUCCESS in 14m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/36340ee0f3dd4a91a9fab208fab94552 : SUCCESS in 1h 13m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4d431e907a454a99a63831e0dcd8f930 : SUCCESS in 47m 36s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/95c6906d0632482395fc8020f3a46d91 : SUCCESS in 2h 00m 28s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1c8ba289ae8a42f6ae0c7599a1d1cfc2 : SUCCESS in 1h 22m 04s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f9398496c9244803b9dc3509e7bf8cf6 : SUCCESS in 1h 26m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4c987e25587340929256bd19f8e9dcc9 : SUCCESS in 1h 18m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b8a6c10e99d2441eb750ddfd2d607dc2 : SUCCESS in 1h 04m 11s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dbff0c9abea74c65b508ee07540e5162 : SUCCESS in 1h 44m 53s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e266f5b8beaa4358b303fb23191682a0 : SUCCESS in 1h 34m 06s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/c46ae76c22734ef9b1035f94327e4c46 : SUCCESS in 1h 27m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e97b0831c134d729353a9e2f3137371 : SUCCESS in 39m 05s","accounts_in_message":[],"_revision_number":2},{"id":"10206ba0466b3ab3c2ef873ccbdd2fbdde3a99e8","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-03-09 22:08:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d1dcabafb930ba73be6f3616d5b5920cb8151d40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 22:09:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c48bac128aa2c45573a860cd3e05f166235a8d3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 23:39:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/a884846cf17b48948690c308396332a7 : SUCCESS in 45m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb522ff7ae1240c0b5faf641d4ead5a3 : SUCCESS in 10m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/78e8be1729234367a27a591e26ee2a50 : SUCCESS in 21m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6c2786e9bc294e79a8f897c3a236df43 : SUCCESS in 23m 14s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/921750d9805d44a2bdc552684a13c82b : SUCCESS in 8m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0d1c61aa70da46c8a7208d9c1c79a133 : SUCCESS in 1h 13m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/02d6a2f9d5974422b27c0ac385984c0d : SUCCESS in 1h 13m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78fcd2920b764ac0ae5a244422bb32ba : SUCCESS in 1h 21m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9a00ceec1624d799c432b818c271025 : SUCCESS in 15m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/323a252550f3411093b94d203573eea5 : SUCCESS in 42m 41s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/472821ad19f04efd9c16d40f3cb53fce : SUCCESS in 1h 12m 23s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d40d6d269e7f4f6a9bf6928d5c478a1c : SUCCESS in 1h 09m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9e478678d7b4ae19f861687b4b600db : SUCCESS in 1h 00m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eb1672ddb6d345b2b82cc781af8c4712 : SUCCESS in 1h 04m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/014f95cd89464c5cb966bbf433b5e8e1 : SUCCESS in 27m 14s","accounts_in_message":[],"_revision_number":2},{"id":"c12638e682785cae079b4478be29f73954ac37f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 23:39:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d8e5099af3a3a6da0d29fad1fdae498b80505627","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 23:42:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00aab6d839b44222867897d17668a311 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d5e168b281e704d5954cd06bcf941b183eb7c8d7","revisions":{"efb396feffda2ce0e9d0bc306b5d22aeaa417c1d":{"kind":"REWORK","_number":1,"created":"2020-02-24 05:16:50.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/09/709409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/709409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/709409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/709409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/709409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/709409/1"}}},"commit":{"parents":[{"commit":"dad3dc704e1a9bb67000e7f5042bbabac0c98991","subject":"Merge \"Check SG members instead of ports to skip flow update\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dad3dc704e1a9bb67000e7f5042bbabac0c98991"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-15 14:48:53.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2020-02-21 18:06:08.000000000","tz":480},"subject":"Add VLAN type conntrack direct flow","message":"Add VLAN type conntrack direct flow\n\nFor vlan type network, we add a segment match flow\nto the openflow security group ingress table. Then\nthe packets will be recorded in conntrack table, and\nthe reply packets can be processed properly.\n\nConflicts:\n\tdoc/source/contributor/internals/openvswitch_firewall.rst\n\nChange-Id: Ieded0654d0ad16235ec923b822dcd842bd7735e5\nCloses-Bug: #1831534\n(cherry picked from commit aa58542e823d23d233524cd5639c7ec4bb757769)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/efb396feffda2ce0e9d0bc306b5d22aeaa417c1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/efb396feffda2ce0e9d0bc306b5d22aeaa417c1d"}]},"branch":"refs/heads/stable/queens"},"d5e168b281e704d5954cd06bcf941b183eb7c8d7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-27 05:15:58.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/09/709409/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/709409/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/709409/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/709409/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/709409/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/709409/2"}}},"commit":{"parents":[{"commit":"eaf7535840c188361788ae11b43c1be8d533a0a0","subject":"Merge \"Improve \"OVSFirewallDriver.process_trusted_ports\"\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eaf7535840c188361788ae11b43c1be8d533a0a0"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-15 14:48:53.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2020-02-24 23:28:12.000000000","tz":480},"subject":"Add VLAN type conntrack direct flow","message":"Add VLAN type conntrack direct flow\n\nFor vlan type network, we add a segment match flow\nto the openflow security group ingress table. Then\nthe packets will be recorded in conntrack table, and\nthe reply packets can be processed properly.\n\nConflicts:\n\tdoc/source/contributor/internals/openvswitch_firewall.rst\n\nChange-Id: Ieded0654d0ad16235ec923b822dcd842bd7735e5\nCloses-Bug: #1831534\n(cherry picked from commit aa58542e823d23d233524cd5639c7ec4bb757769)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d5e168b281e704d5954cd06bcf941b183eb7c8d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d5e168b281e704d5954cd06bcf941b183eb7c8d7"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
