)]}'
{"id":"openstack%2Fneutron~709446","triplet_id":"openstack%2Fneutron~stable%2Fqueens~Iccc4fe4a785ee042fd76a663d0e76a27facd1809","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1853613-stable/queens","hashtags":[],"change_id":"Iccc4fe4a785ee042fd76a663d0e76a27facd1809","subject":"ovs agent: signal to plugin if tunnel refresh needed","status":"MERGED","created":"2020-02-24 10:36:43.000000000","updated":"2020-02-26 01:35:04.000000000","submitted":"2020-02-26 01:35:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709446-1582680904923-31ef7618","meta_rev_id":"483e68b83dfa63dfd0d466eb8bdfee09e87448a6","_number":709446,"virtual_id_number":709446,"owner":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-02-24 17:54:45.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-26 01:35:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"value":0,"date":"2020-02-24 14:40:57.000000000","_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"date":"2020-02-24 15:10:33.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-25 07:42:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-02-25 23:55:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-02-25 23:55:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-24 10:36:43.000000000","_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-24 10:36:43.000000000","updated_by":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"reviewer":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"state":"REVIEWER"},{"updated":"2020-02-24 15:10:33.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-02-24 17:54:45.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-25 07:42:50.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-02-25 23:55:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-02-26 01:35:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a400d9f0ca40d7df473ef465b9d0b7b52ef21879","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-02-24 10:36:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8631bf45f6f02551b69000600adceb95471020e6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-24 10:45:15.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/709446/1 : FAILURE in 7m 45s","accounts_in_message":[],"_revision_number":1},{"id":"49cd50e1146eb289585bc9861eff0c3d23c57e91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 12:32:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/24f0cb28fc5c42199383895eb3bf07b6 : SUCCESS in 56m 00s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/0f5da94e50b5448e8e93fbf91dd20e3b : TIMED_OUT in 1h 04m 25s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/112ef9f6529e41ee90ba6987ba09b744 : FAILURE in 1h 16m 58s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/7e00f367c9504985a9978cf7936a3510 : SUCCESS in 57m 36s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/fba2d11f84d74da3a3e8011fe795747a : SUCCESS in 1h 02m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58936e5f21544480bc445866dc6ec513 : SUCCESS in 13m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46f3aedf375749e993aa2a0dc8b7aee3 : SUCCESS in 30m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/55222b44457b49a88c5d8867295c77d1 : SUCCESS in 30m 14s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/06c455ec15da463494c952c5c6ccddd2 : SUCCESS in 9m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c26a8f1c6254435cb59309d0df3492fd : SUCCESS in 1h 18m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0e974ecf3ed84e92b8c4bfa412fca36f : SUCCESS in 57m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e64675e2c3694028a6008234b78ef547 : SUCCESS in 1h 17m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3edf09bfe6fe48eab0dc5022990dddac : SUCCESS in 50m 28s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/77ae2c13d48e4de688d40fcf15eedd10 : FAILURE in 30m 05s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/6dfb30351bf6487c8a115f331c92c182 : SUCCESS in 1h 48m 25s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a7d0d1df014c416b8f014f2d93ec48ed : SUCCESS in 1h 41m 02s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d6df3b92dc8f473e86856fea2040525b : SUCCESS in 1h 22m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cce6d4b5d5e24dc1997bfa44e87c790e : SUCCESS in 1h 02m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/72374e22d98b4186accdabaf9e0853fa : SUCCESS in 1h 08m 13s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4299d98652d24841b33bc3ea65f52a22 : SUCCESS in 1h 38m 55s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a0a6a183d72e432fa860a58678ef02f8 : SUCCESS in 1h 46m 24s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/4e3ca712eb204959b8f96ce0c7530566 : SUCCESS in 1h 17m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bde9502b90dd418386c4167bc1927b89 : SUCCESS in 34m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2ff9c67607ddd366856f7ed3fdaf7e4a1c32b61e","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-02-24 12:53:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"11d8a297d121b5d240839933d0ae88b29b5e77f3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-24 13:02:20.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/709446/1 : FAILURE in 8m 15s","accounts_in_message":[],"_revision_number":1},{"id":"3fa675a0ee4dc2c484abf2acd0316a41b4b6e17a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-24 13:12:57.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/46/709446/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/46/709446/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"29fde91806725d4a721fc1e388539f9eea03dc9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 14:38:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/6339bc0929e0416893c74c70127c0f21 : SUCCESS in 46m 02s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/e60722b9c84645e0afd62383af16cfd0 : SUCCESS in 35m 52s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/e04348730838431a96b75a96f7629e1e : POST_FAILURE in 1h 15m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/665f88b9ce104f8081c31603e2683eb9 : SUCCESS in 49m 25s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/c80232e927fa46478eca814d3e2fe5a8 : SUCCESS in 48m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/834a0beae86d417a9707c5111bd5575e : SUCCESS in 12m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a5006a42b9b949d58fa3a8aef51d727d : SUCCESS in 23m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8699413442de4ef2a338946d31cd41bf : SUCCESS in 24m 29s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f59ac919408f471d884619afebf5f79b : SUCCESS in 7m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/65c73234e8c94c89b50783a5bcdbd003 : FAILURE in 1h 17m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/69a039d9d8b54dedbce0b740bdd13238 : SUCCESS in 54m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed02986e93144374a01c9c392e5de013 : SUCCESS in 1h 15m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/95d9107de79344b186948f591cba1a8c : SUCCESS in 37m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d27a859fcf1c4e179388094e18de4491 : FAILURE in 32m 39s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/d0e87452288c43bcabc01a8bcd7733ca : SUCCESS in 1h 34m 49s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3a3e701eeb134113a13208cd81bf80c0 : SUCCESS in 1h 13m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/82b7ff9cc3f948788f70ad2a781eca2f : SUCCESS in 1h 28m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f069b71da696468991e8605e7e0fd058 : SUCCESS in 1h 05m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0a657f1b4a9a461e90ea5835ff35b85f : SUCCESS in 1h 13m 45s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7a137e3f5e864c2084bdfaf40ac787ca : SUCCESS in 1h 34m 53s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/590cd1d0d088469aa5947a3ca4a2cebf : SUCCESS in 1h 32m 08s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/f408df9a75404d08915e3d11a3b69182 : SUCCESS in 1h 41m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fd0ea95e7274f18adefefd11a97fea0 : SUCCESS in 37m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9cfed06fbd8d5bb2a463ceb88d4ca0170fca7d7e","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-02-24 14:40:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e1c2354110cc696b6d254d01f3dd5cfa8daac88f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-24 15:10:33.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/709446/1 : FAILURE in 7m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c16b7aaef350dd9fcb995c09e04a1bf953ce4146","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-24 17:54:45.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/46/709446/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/46/709446/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d9a292895665e9d4bfcf3e78a8175203bf6cc0f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 20:23:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/b25b285f7ecf48b18229208a4bfaef7f : SUCCESS in 48m 07s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/8ee69e3921a34a9aa7e1630bc44bca7c : SUCCESS in 40m 32s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/e351f3deca0d4e32919c5c7a5dde1d60 : FAILURE in 1h 24m 24s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/dd83af2ade95439d8f176f46b166128a : SUCCESS in 52m 37s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/0721c6d68b4b4047849e3d924d166681 : SUCCESS in 54m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96cab9627fe44cd0a732147cb4893dea : SUCCESS in 12m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/751a2383b1f74809877c629334cdc014 : SUCCESS in 33m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ef890ac6f5e440a088b60ca076ffc4af : SUCCESS in 35m 50s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/03f7185669664304839a14964b137c58 : SUCCESS in 7m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/37af8fe2f56b4f5ca6c1700a17ae0719 : SUCCESS in 1h 27m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/82c3c53560ff4e0ba66cc73027be20cf : SUCCESS in 1h 02m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/483d8e198ebe4855b50a389acdf7273b : SUCCESS in 1h 21m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ce64419b21684dc7a52fb06c07f4256b : SUCCESS in 41m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3db6f402eb234bad80a222fadde4d517 : FAILURE in 25m 06s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/12f5fced75d04c52b8e66137663c49d1 : SUCCESS in 1h 51m 59s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/4efddfc83dcc447d9c1c3a8f9fa631bf : SUCCESS in 1h 25m 43s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/15815974a0884b86964187221c322a54 : SUCCESS in 1h 30m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a28de9c412aa426694fc6c9e2a0f143e : SUCCESS in 1h 10m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f8177166fd6045058c3e142901c95c8c : SUCCESS in 1h 10m 33s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/32c2add8003b48a8967b90271a3f957d : SUCCESS in 1h 36m 07s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4f6bca351e2f45679dc7ef684b7786b8 : SUCCESS in 1h 29m 44s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/2681afbe60ca48b58867e173f5032ef1 : SUCCESS in 1h 25m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/988615b1a7f34b4c8fdd7315adfab5b9 : SUCCESS in 34m 35s","accounts_in_message":[],"_revision_number":1},{"id":"dae7860064e24da7a21fcf60fdc0779b30557d8d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-02-25 07:42:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"72d72f40cb411b3d52459a90a7cc8d096758f2e9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-25 23:55:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c4ad5dcd594993d681886e44aaacdd325495b4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 23:55:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2515c8ec2f72a34c8c7c701c2a2d2eb90691d70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 01:35:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/ec2dc7e47ada4a73bb5f8947f720f39e : SUCCESS in 43m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1734f356b3a3434bbff06ac8b845489c : SUCCESS in 11m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/14bbc79f22084097a7dd5d1fd72f8ede : SUCCESS in 22m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d5dfbfcc0a424c6faf5eda7e9551081e : SUCCESS in 25m 05s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2c22f073096044078eed9e0c4dbdaa13 : SUCCESS in 6m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e5b874170dd54d92bbbc79bb8fb68577 : SUCCESS in 1h 14m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/94e2c04cab34443191820b7980b90d10 : SUCCESS in 49m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/618c2ce30a09427fbe50c3bc0c9f024b : SUCCESS in 1h 08m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6513bad86f1743a5aeb8fb2434df1711 : SUCCESS in 36m 52s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e6ba1b6960934d85a27b1d235ce676f7 : SUCCESS in 1h 28m 20s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0d714b1b0fe5471aaf9420abae0b3724 : SUCCESS in 1h 12m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dce03a3ed4b4461a882554e1a76e69c7 : SUCCESS in 1h 00m 49s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ca5420353b904ae987053448952ffa6d : SUCCESS in 1h 04m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e874efc5afd4eeeac29d1c3df6db6b2 : SUCCESS in 28m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a0c34261d80323a18213d1915e221de9c3a46077","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 01:35:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"576e76f209ce69d99606ce126cfd4a96d99f8e88","revisions":{"576e76f209ce69d99606ce126cfd4a96d99f8e88":{"kind":"REWORK","_number":1,"created":"2020-02-24 10:36:43.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/46/709446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/709446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/709446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/709446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/709446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/709446/1"}}},"commit":{"parents":[{"commit":"dad3dc704e1a9bb67000e7f5042bbabac0c98991","subject":"Merge \"Check SG members instead of ports to skip flow update\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dad3dc704e1a9bb67000e7f5042bbabac0c98991"}]}],"author":{"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","date":"2020-01-24 16:19:30.000000000","tz":0},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2020-02-24 10:36:20.000000000","tz":180},"subject":"ovs agent: signal to plugin if tunnel refresh needed","message":"ovs agent: signal to plugin if tunnel refresh needed\n\nPatch https://review.opendev.org/#/c/697655/ cannot be backported\nbecause it includes an RPC version change. This patch is for the\nstable branches.\n\nCurrently the ovs agent calls update_device_list with the\nagent_restarted flag set only on the first loop iteration. Then the\nserver knows to send the l2pop flooding entries for the network to\nthe agent. But when a compute node with many instances on many\nnetworks reboots, it takes time to readd all the active devices and\nsome may be readded after the first loop iteration. Then the server\ncan fail to send the flooding entries which means there will be no\nflood_to_tuns flow and broadcasts like dhcp will fail.\n\nThis patch fixes that by also setting the agent_restarted flag if\nthe agent has not received the flooding entries for a network.\n\nChange-Id: Iccc4fe4a785ee042fd76a663d0e76a27facd1809\nCloses-Bug: #1853613\n(cherry picked from commit bc0ab0fcd721d3fb01fd83291269a586f50efa0e)\n(cherry picked from commit aee87e72b1d456da66d719ccf19054ac1f285a7b)\n(cherry picked from commit d1e2b840b57ba9425d5ea710a0509b677eefcc08)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/576e76f209ce69d99606ce126cfd4a96d99f8e88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/576e76f209ce69d99606ce126cfd4a96d99f8e88"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
