)]}'
{"id":"openstack%2Fneutron~710458","triplet_id":"openstack%2Fneutron~master~Id6c103b7bcebe441c27c6049d349d84ba7fd15a6","project":"openstack/neutron","branch":"master","topic":"metadata-agent-tests-assertEqual","hashtags":[],"change_id":"Id6c103b7bcebe441c27c6049d349d84ba7fd15a6","subject":"Remove extra header fields in proxied metadata requests","status":"MERGED","created":"2020-02-28 14:48:38.000000000","updated":"2020-03-03 20:11:49.000000000","submitted":"2020-03-03 19:47:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710458-1583264841805-1d1e633c","meta_rev_id":"87abe277b4a1128206cf6f64727301676221839f","_number":710458,"virtual_id_number":710458,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"date":"2020-03-02 23:13:51.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-03 19:47:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-03 01:10:38.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2020-03-02 20:29:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-03-02 17:11:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2020-03-02 19:46:56.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2020-03-02 20:29:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-02 17:11:31.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-02 19:46:56.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-03-02 20:29:20.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-03-03 01:10:38.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-03-03 19:47:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fae6c42ac43532a0ec7171e0dbbe682fbbf1a57d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-28 14:48:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"517beb7a9ef2d413bee3f36363e068a893b3ffeb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-28 16:40:40.000000000","message":"Patch Set 1: Code-Review+1\n\nI would also patch agent.py to expect only one header at a time, but otherwise lgtm. ;-)","accounts_in_message":[],"_revision_number":1},{"id":"3eb49427b5147c40dd28f894992f9f6d72d2e367","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-28 17:43:33.000000000","message":"Patch Set 1:\n\nRadoslaw - I see what you mean - _get_instance_and_tenant_id() assumes only one will be present, and prefers the network over the router ID.  But I think after this change it should only get one.  I could have it return nothing if it somehow sees both, other opinions?","accounts_in_message":[],"_revision_number":1},{"id":"de0bc7479de74da5cea7621d0004fb5ebcb371f7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-28 18:02:45.000000000","message":"Patch Set 1:\n\nFine by me, Brian.\n\nBTW, should not it have a reno as well?","accounts_in_message":[],"_revision_number":1},{"id":"22dc4159b75cf24c162ea640158c22736b6fef96","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-28 19:35:42.000000000","message":"Patch Set 1:\n\nSure, I can add a release note, and since I\u0027m doing that might as well change the agent to drop the request if both headers are seen.","accounts_in_message":[],"_revision_number":1},{"id":"43090bf1037ade44921ea427c9a64c641b1005a3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-28 19:36:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"735ab18d98cbe9f17943ee3d6712f56d6c744661","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-28 19:47:58.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nThanks, some minor comment inline.","accounts_in_message":[],"_revision_number":2},{"id":"3a05d46e183e9b4ee5423214c15b99e2c5af4823","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-28 20:12:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa7a9cb2a060dbb8c388f92dc3c29ed0ae498735","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-28 20:13:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"58e4eb4b87468cacd9ccd1d5349e4170f7f84a59","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-28 20:25:31.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks, Brian. Looking great.","accounts_in_message":[],"_revision_number":3},{"id":"9b11c95bd86eeff8707798ca25cb5e62a90de42f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-28 22:09:25.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/710458/3 : FAILURE in 1h 55m 49s","accounts_in_message":[],"_revision_number":3},{"id":"d70a7aa01daa8f5bed6e6764c88746086809b6c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 01:42:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/12bbde252831468086c84f23a098319a : SUCCESS in 1h 01m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e2fc775ae1cb45e99fb5bcb601057491 : SUCCESS in 39m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/21b3902017814fd99a8912722f882a3a : FAILURE in 1h 49m 42s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/11da02ae6f9f4a16bf0495d69b0beccf : SUCCESS in 1h 09m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b9aeb029946f4cb08c8497124a213c0f : SUCCESS in 1h 25m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d8af9d603f8240edbaaa43c68999fdb0 : SUCCESS in 1h 38m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a55f1585de3439ea2bb41131c11ec0f : SUCCESS in 35m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2458470b7bad4cf6a25e394b3e4c6598 : SUCCESS in 29m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebed540536bb49eda41c053a1da7022b : SUCCESS in 15m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ae60140dfca45c499f153c49b1ae1f0 : SUCCESS in 23m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e8ce1349f424b3499005813b6c85529 : SUCCESS in 26m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c692175b21704224a2514f73f17e4a60 : SUCCESS in 34m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4200851b9b945bdb413586c1d4c0397 : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff446e8af8f24ce9ab4bfd8dfb006b47 : SUCCESS in 14m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/65151458ed6d4c69a3412dd3d58e7fea : SUCCESS in 44m 04s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7045b573951548fe9d05c1bda9ce6510 : SUCCESS in 1h 02m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5e3d5cd6f3ce4c6cac893de7ee40a5ac : SUCCESS in 1h 40m 53s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e85c2b2bd6ea4e6497050c931dd03cac : SUCCESS in 1h 20m 54s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/60b935b3d6da4e0381e849301b92b785 : SUCCESS in 1h 26m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/493e9a0dd32b49e9b403e58c4ed68b1d : SUCCESS in 1h 13m 40s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/75bce9b44ad6425f9f741866370a464d : SUCCESS in 1h 11m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/36ac1eae2cb6445a8797e5a99d200046 : SUCCESS in 1h 20m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/94a8ed1351144a02ba5a097675d147f8 : SUCCESS in 1h 50m 19s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2ab872091f4249bbbac007052b6daa76 : SUCCESS in 1h 54m 07s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2ee517a8afda4364b5e5e97e82495ec5 : SUCCESS in 1h 08m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c496b277433b4a02b3de44587f021df3 : SUCCESS in 53m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b003201edfa6480d8f59c9680745bc69 : SUCCESS in 2h 14m 42s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5aa391f877fe40898f3d57a473c8ca1a : SUCCESS in 1h 38m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/74ad5febe78e4bd0a39014a0108f4bfb : SUCCESS in 1h 38m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa191713cde8490abb3f5b8f814b2641 : SUCCESS in 1h 02m 09s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3060cad6840c43539b7b7764c9b6cda1 : SUCCESS in 1h 11m 24s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aeb5159e4e884367989d7ab3483b1f95 : SUCCESS in 47m 01s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6f3bfe7d26174cd68055ef56895f1f73 : SUCCESS in 1h 09m 37s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7af1ae26eaff4ddc8b661185a7e2ec6f : SUCCESS in 1h 22m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bdcf7d8158ab44a0bcf2898e3c8e37d4 : FAILURE in 2h 31m 35s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c19078d60cc945b88e0bc538e1fb7df0 : FAILURE in 1h 50m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c839a2a7f844fd69d77896b0d7c49efa74a54839","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-01 09:58:37.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cf7d6e3907a2eedfb96e82ddc68daa8e84939af4","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-02 14:07:25.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nI am opposed to silently dropping messages that could be created by a subsequent bug, making it much harder to debug a problem, especially if the debugging is in a running cloud.","accounts_in_message":[],"_revision_number":3},{"id":"cbc797d63f7c1ef81819a2cb8de50c9c60e2028b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-02 16:27:04.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"168f0cf13677033e2bbcbc3e36cea7d47bcf995b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-02 16:27:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5133bafa22c31678ca12cfbee7cd7aa8e18d973b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-02 17:11:31.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"cee0f68dcd5e60ef4d8c61925bf15c48c6519036","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-02 18:46:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/710458/4 : SUCCESS in 2h 18m 18s","accounts_in_message":[],"_revision_number":4},{"id":"a4c4e0b43c891d2eae761df18d46404e674e6911","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-02 19:46:56.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"25e7b79c33ec4ae655ca5766bc7a6940c90d97a6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-02 20:29:20.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"869b77037148255a8a670a2b529f5c690f058773","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 22:32:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5a56db39b411418cb5372b6db694dc42 : SUCCESS in 55m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e029cbbebdcb45e7b357e27c7f075746 : SUCCESS in 43m 21s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/c05ec5ceef694313af833867d643cca3 : FAILURE in 1h 58m 17s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4f7bdfd6dbfb46ceb4b6e47dd3be2d5c : SUCCESS in 1h 10m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f63c6cb20f204ee39f618a458d0089a6 : SUCCESS in 1h 00m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ec8188fe921143169a21e04c5e99b55f : SUCCESS in 1h 07m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e52cbce157f46bbae4c6a568f053994 : SUCCESS in 30m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d352d313d4de4f4e99afee0c2ce4bdbe : SUCCESS in 25m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc0880aba16444ba802e373e0258147c : SUCCESS in 14m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7aee7d377a5d4fca978f50b4151f1cf6 : SUCCESS in 26m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3964321e379c4f59b5e0e4df45624755 : SUCCESS in 29m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f1eff638b9a46dfa5e3888cd19972d9 : SUCCESS in 26m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5cdbf6453544369bf7c1ee21303c402 : SUCCESS in 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60888dfa4b174288a8432d066f5443c5 : SUCCESS in 17m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d8d6f277733546a5ba537b1038806431 : SUCCESS in 47m 29s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8005b3f6c7194434a3e368dda1ca4402 : SUCCESS in 1h 08m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d1c058f04514405da44a6a2c9e7dc8bd : SUCCESS in 1h 46m 23s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/838d7651b02a4220bffbdcc8c8d1e64a : SUCCESS in 1h 20m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a8dc06129b834f03bc6a38d62be042bb : SUCCESS in 1h 19m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/74eb98a154c648ceba3ca8b49c67538a : SUCCESS in 1h 16m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/94ced0e410aa4f818dc509f1480db06d : SUCCESS in 1h 22m 24s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0c6daf7fc8da49378880b0b58e4695bc : SUCCESS in 2h 06m 39s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9b3a33bd61c849d1ad555acb72512036 : SUCCESS in 1h 39m 37s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5675e960c7714f839226462d9d8c78cd : SUCCESS in 1h 21m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d38b71b592844f57b3e71a6256b509cd : SUCCESS in 57m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f9ca606e85f643b2900a0b11bdac9c63 : SUCCESS in 2h 26m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cc50254eaf0543e0ad178a8a77527db8 : FAILURE in 2h 01m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/272fd3a9ca744dcaa35c3a87a8dc0e36 : SUCCESS in 2h 14m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c48a703cf67b411bbd0b7e33f39da961 : SUCCESS in 1h 11m 34s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6c1c69980392430f8c47e2b83d7871c0 : SUCCESS in 1h 12m 56s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e9febfdc9a5c4439b42c44c2d854d806 : SUCCESS in 55m 02s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d02d3f47cecb4c17980416c12cdeb3d5 : SUCCESS in 1h 12m 43s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/bd08abbb79ae4cb985a052d57c071222 : SUCCESS in 1h 16m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a288935d0efb4d5184285dfad04251da : FAILURE in 2h 39m 21s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/1b59bcfc8ad1428cbad97d469e9255cd : FAILURE in 2h 23m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"713e273fefbb08462fad8a61f53f01bd472b953e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-02 23:13:51.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"169998f46a1bd43dde21051afc28a30ffdf6aca6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-03 01:10:38.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/710458/4 : SUCCESS in 1h 56m 17s","accounts_in_message":[],"_revision_number":4},{"id":"6515e2fdabc4f5465f98dd9a476dd2772c2b1ff7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 02:42:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d1a96ef690ad48a8accfea8a3cd687e9 : SUCCESS in 48m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dabd0bdc588a4aadba601cd77cee43ed : SUCCESS in 35m 50s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/e98f3ef89fa24dde84bbd238c356a0e2 : FAILURE in 2h 27m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a0268e1da1d2455db8a34c111ad3f5d4 : SUCCESS in 1h 27m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dbb9ce104f2049c48a5586200c577ece : SUCCESS in 1h 21m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0b99d960ad214eb38ee9a1a15f579772 : SUCCESS in 1h 25m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b661bbcb348d4087b1e6857ff68944e0 : SUCCESS in 36m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/623fe956f0eb466a9d4351592a42c22f : SUCCESS in 30m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1251cb5dabc7422695fa8395140425b3 : SUCCESS in 16m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f24049bfc56d47348f66418f0e96f683 : SUCCESS in 37m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2909e6a02657440b800de6ee016522d0 : SUCCESS in 28m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3be71ab836d14a61b3e337afd222115e : SUCCESS in 24m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c955df70677448ea95852a13c2eab34 : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d646720563294de19419cf31f14e0cf5 : SUCCESS in 17m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ae78c974a59a46409afe3f8930f1017a : SUCCESS in 43m 00s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e738da026ca8498eb6a502d08db1d498 : SUCCESS in 1h 06m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7dad2dadce244479854d6a4309f7ea8b : SUCCESS in 2h 07m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c02d826415734a68a5d15ab510f0b6c5 : SUCCESS in 1h 09m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f19f2532e60a4a73938898a8a7b8ceab : SUCCESS in 1h 09m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/830e66b2a7244ab8b0a6af79185c6815 : SUCCESS in 1h 16m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/85142bb0612a4f239ce48e4447ea421e : SUCCESS in 1h 17m 22s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/343a0c8da3d2455fa8cfdf0325f6b64b : SUCCESS in 1h 38m 22s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ab1271682d7d40488f932043634f9d0b : FAILURE in 1h 43m 52s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/61508260f9c24fe0957343f90ff3fbda : SUCCESS in 1h 07m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/05956bcbdecc4a7b8918ea40597f4d55 : SUCCESS in 1h 02m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4faa686463b14d46b2da75d5c29e0ff8 : SUCCESS in 2h 13m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/150405e8d3cd4370861ff35dabda9b5d : SUCCESS in 1h 44m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8ec87a2ede1c44f296dc4cbd1808bbdb : SUCCESS in 1h 54m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8038f6b90734d6e9397c1e6fa1716fb : SUCCESS in 1h 07m 56s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/aee8e9ca37e2492a99a9a1321ee4250c : SUCCESS in 1h 24m 42s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f0f2fc0e407047b29ec037a1b966c108 : SUCCESS in 48m 16s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/340cd96dbaaa40d48abb24cf48ecb1a4 : SUCCESS in 1h 09m 15s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/23e06888363e45f8b5b327cc3b835090 : SUCCESS in 1h 24m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/442e1cd5b4f845e68ab62708c3142edd : FAILURE in 2h 21m 15s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/dd0efbccad8842d1b589049a8bc251d6 : FAILURE in 2h 40m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f7d6c91d341199510acfdf87757a46e803c03b9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 02:43:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ffbff0b6a05410478091bc2585515c4971225fb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 15:34:36.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0a84bd86a91f24705f8befcc5469a05f12818575","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 19:47:21.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7bdba495d0aa444e8ef5f27986fb2151 : SUCCESS in 1h 04m 26s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8b732fad6c73465ba1b1fd3a08073da2 : SUCCESS in 1h 34m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/95687f620fdf4191a8e16871a434b6a8 : SUCCESS in 1h 12m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9bca2fafb10447dea36ef893afe88604 : SUCCESS in 1h 10m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84bfc617ca5b475db5379ae810d350e8 : SUCCESS in 31m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/754bf518fa0e4855b73ebe53b5cf6d50 : SUCCESS in 14m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecd00071028f49139e318ff4aa54226e : SUCCESS in 31m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35268de33dc44a018c76f97809e446df : SUCCESS in 31m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/812f9d82043a4b03af6be17886909187 : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43a2685c6d564520879da304b612c3bc : SUCCESS in 18m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5f811406cdf44745956a89baa151df60 : SUCCESS in 1h 01m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/17aa604ea7214b499307d829f867dcbb : SUCCESS in 1h 08m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2b72ab558f6c4195b811c839fc161f7b : SUCCESS in 1h 20m 11s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5dde1fbe68724d45b026f3fadfb3a785 : SUCCESS in 1h 18m 34s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bc95a1270c8d4cbd8747d33e15cbcf2f : SUCCESS in 1h 22m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e8a2419ccc514b039e5c3da197f45823 : SUCCESS in 1h 13m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bb24b221230c4737b80314aee20b3242 : SUCCESS in 1h 13m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ea977fb9d1d349d2b8ff6d9c4bd6e85e : SUCCESS in 2h 22m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d5cd4515f184406b23f0bab2a13ea94 : SUCCESS in 1h 17m 03s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e2833b7456b34f04a3a64766bd883b01 : SUCCESS in 2h 02m 31s","accounts_in_message":[],"_revision_number":4},{"id":"9e08f953e62678c25f1170f1bd15778f440a9f25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 19:47:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c1ebe050154a4afc2a17810048e96635376fc753","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-03 20:11:10.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit bcc4f98a3d7e8f0e23cbf953cc828506a8a8002d","accounts_in_message":[],"_revision_number":4},{"id":"f754df03baa3b99cf8365f9c80eeae9f5b55388a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-03 20:11:30.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 4dc0a61cd5aaa8a0448b19c08b9e8f04d2d11eb4","accounts_in_message":[],"_revision_number":4},{"id":"48b19dd5e653ff8bddbccf02b5d0704a7de06f5a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-03 20:11:38.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 1c13ad05dd937b65330c389301c2f095b7bedcd3","accounts_in_message":[],"_revision_number":4},{"id":"f61dba8d1d47488cb085fdf98a20bfbe11d2bcb8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-03 20:11:49.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit b8de1e6ef01a45f75ad14d0d96cbcf40a8767f5d","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5af046fd4e6387cdbf8bf65ea4c2039a4019b64b","revisions":{"dbcd01a75d2ef78746da17147362363344ba2db2":{"kind":"REWORK","_number":1,"created":"2020-02-28 14:48:38.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/58/710458/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/58/710458/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/710458/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/710458/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/710458/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/58/710458/1"}}},"commit":{"parents":[{"commit":"6b9765c991da8731fe39f7e7eed1ed6e2bca231a","subject":"Merge \"[OVN] Add OVN functional tests - part 1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b9765c991da8731fe39f7e7eed1ed6e2bca231a"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-02-27 22:33:28.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-02-27 22:33:28.000000000","tz":-300},"subject":"Remove extra header fields in proxied metadata requests","message":"Remove extra header fields in proxied metadata requests\n\nIf a user specifies a header in their request for metadata,\nit could override what the proxy would have inserted on their\nbehalf. Make sure to remove any headers we don\u0027t want, and\noverride something that might be present in the request.\n\nChange-Id: Id6c103b7bcebe441c27c6049d349d84ba7fd15a6\nCloses-bug: #1865036\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbcd01a75d2ef78746da17147362363344ba2db2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbcd01a75d2ef78746da17147362363344ba2db2"}]},"branch":"refs/heads/master"},"326052344fd9fd57120d6d9da25cec956e04a50a":{"kind":"REWORK","_number":2,"created":"2020-02-28 19:36:05.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/58/710458/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/58/710458/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/710458/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/710458/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/710458/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/58/710458/2"}}},"commit":{"parents":[{"commit":"6b9765c991da8731fe39f7e7eed1ed6e2bca231a","subject":"Merge \"[OVN] Add OVN functional tests - part 1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b9765c991da8731fe39f7e7eed1ed6e2bca231a"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-02-27 22:33:28.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-02-28 19:34:46.000000000","tz":-300},"subject":"Remove extra header fields in proxied metadata requests","message":"Remove extra header fields in proxied metadata requests\n\nIf a user specifies a header in their request for metadata,\nit could override what the proxy would have inserted on their\nbehalf. Make sure to remove any headers we don\u0027t want, and\noverride something that might be present in the request.\nIf the agent somehow gets a request with both headers it will\nsilently drop it.\n\nChange-Id: Id6c103b7bcebe441c27c6049d349d84ba7fd15a6\nCloses-bug: #1865036\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/326052344fd9fd57120d6d9da25cec956e04a50a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/326052344fd9fd57120d6d9da25cec956e04a50a"}]},"branch":"refs/heads/master"},"8fbf90665a16e35bf4f7082b72772e11404fa18f":{"kind":"REWORK","_number":3,"created":"2020-02-28 20:13:05.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/58/710458/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/58/710458/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/710458/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/710458/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/710458/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/58/710458/3"}}},"commit":{"parents":[{"commit":"6b9765c991da8731fe39f7e7eed1ed6e2bca231a","subject":"Merge \"[OVN] Add OVN functional tests - part 1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b9765c991da8731fe39f7e7eed1ed6e2bca231a"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-02-27 22:33:28.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-02-28 20:11:57.000000000","tz":-300},"subject":"Remove extra header fields in proxied metadata requests","message":"Remove extra header fields in proxied metadata requests\n\nIf a user specifies a header in their request for metadata,\nit could override what the proxy would have inserted on their\nbehalf. Make sure to remove any headers we don\u0027t want, and\noverride something that might be present in the request.\nIf the agent somehow gets a request with both headers it will\nsilently drop it.\n\nChange-Id: Id6c103b7bcebe441c27c6049d349d84ba7fd15a6\nCloses-bug: #1865036\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8fbf90665a16e35bf4f7082b72772e11404fa18f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8fbf90665a16e35bf4f7082b72772e11404fa18f"}]},"branch":"refs/heads/master"},"5af046fd4e6387cdbf8bf65ea4c2039a4019b64b":{"kind":"REWORK","_number":4,"created":"2020-03-02 16:27:10.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/58/710458/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/58/710458/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/710458/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/710458/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/710458/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/58/710458/4"}}},"commit":{"parents":[{"commit":"6b9765c991da8731fe39f7e7eed1ed6e2bca231a","subject":"Merge \"[OVN] Add OVN functional tests - part 1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b9765c991da8731fe39f7e7eed1ed6e2bca231a"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-02-27 22:33:28.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-02 16:20:25.000000000","tz":-300},"subject":"Remove extra header fields in proxied metadata requests","message":"Remove extra header fields in proxied metadata requests\n\nIf a user specifies a header in their request for metadata,\nit could override what the proxy would have inserted on their\nbehalf. Make sure to remove any headers we don\u0027t want, and\noverride something that might be present in the request.\nIf the agent somehow gets a request with both headers it will\nsilently drop it.\n\nChange-Id: Id6c103b7bcebe441c27c6049d349d84ba7fd15a6\nCloses-bug: #1865036\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5af046fd4e6387cdbf8bf65ea4c2039a4019b64b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5af046fd4e6387cdbf8bf65ea4c2039a4019b64b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
