)]}'
{"id":"openstack%2Fneutron~710554","triplet_id":"openstack%2Fneutron~master~If08909dfbb841ad941e6aaea0990e9e98354e3d9","project":"openstack/neutron","branch":"master","topic":"bug/1863987","hashtags":[],"change_id":"If08909dfbb841ad941e6aaea0990e9e98354e3d9","subject":"[OVN] Remove DNS dependency on original port dict","status":"ABANDONED","created":"2020-02-29 10:49:11.000000000","updated":"2023-09-22 08:03:28.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"92244565c998e91815c77b63b769810d21149f64","_number":710554,"virtual_id_number":710554,"owner":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":29874,"name":"Taoyunxiang","email":"taoyunxiang_yewu@cmss.chinamobile.com","username":"Yunxiang"},{"value":1,"date":"2020-03-19 03:31:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"date":"2020-03-19 01:49:36.000000000","_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-19 04:19:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"date":"2020-03-19 03:33:45.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29874,"name":"Taoyunxiang","email":"taoyunxiang_yewu@cmss.chinamobile.com","username":"Yunxiang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":29874,"name":"Taoyunxiang","email":"taoyunxiang_yewu@cmss.chinamobile.com","username":"Yunxiang"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":29874,"name":"Taoyunxiang","email":"taoyunxiang_yewu@cmss.chinamobile.com","username":"Yunxiang"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"date":"2020-05-25 09:22:22.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"date":"2020-05-25 09:22:56.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"_account_id":29874,"name":"Taoyunxiang","email":"taoyunxiang_yewu@cmss.chinamobile.com","username":"Yunxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-29 12:32:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-03-17 11:58:09.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-03-17 11:58:19.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-03-17 11:59:10.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-03-17 11:59:42.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-03-17 12:00:18.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-03-18 08:21:59.000000000","updated_by":{"_account_id":29874,"name":"Taoyunxiang","email":"taoyunxiang_yewu@cmss.chinamobile.com","username":"Yunxiang"},"reviewer":{"_account_id":29874,"name":"Taoyunxiang","email":"taoyunxiang_yewu@cmss.chinamobile.com","username":"Yunxiang"},"state":"REVIEWER"},{"updated":"2020-03-19 03:31:51.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-19 03:33:45.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-03-19 04:19:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-25 09:22:22.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-05-25 09:22:56.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2021-07-29 22:23:12.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REMOVED"}],"messages":[{"id":"519ee5b0c7f633145f67a0da7a0c3d98b41e1ff7","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-02-29 10:49:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31cd54dd827e432746909e7e33a09f05064345a0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-02-29 12:32:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"523c0204169f53b13b5a92204125d43368bf401f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-29 12:34:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/710554/1 : SUCCESS in 1h 44m 57s","accounts_in_message":[],"_revision_number":1},{"id":"18aedf4c7ffb7a655a4ac5acf9040ce688473f2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 14:01:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3216bcccadca44d3ae0601cca7b46a35 : SUCCESS in 53m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/042e3c72e0564e129bda0467228c2be7 : SUCCESS in 33m 51s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/0000ce500eab4b4599cdafc727571d4b : FAILURE in 2h 12m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/91cd5dc323f941f1aaa8a609ac0702d3 : SUCCESS in 1h 25m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/39be60a664824fd49aecb2754e25b34a : SUCCESS in 56m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1d57259dd64d43fcb6a5cc997b37e536 : SUCCESS in 1h 03m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc07efad4f6444fe91335f94be8df594 : SUCCESS in 37m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c88a34fc9ea4de183c90268774ee1b9 : SUCCESS in 24m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3815282ea12479bacbd822069a0e561 : FAILURE in 12m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8485d362424e4f8cb0b0666b0a3a9e9d : SUCCESS in 35m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f30a634183f848d68f360c5e52c53523 : SUCCESS in 27m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af98f8c7c6cc45eb85737eff9e779e5e : SUCCESS in 29m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f25feb85e394634ac732aed462a3281 : SUCCESS in 10m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7d6b14bc83614ad094a538034b3ac129 : FAILURE in 1h 05m 22s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/00ede4bade834e3b874ec6a76e9afc14 : SUCCESS in 1h 08m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/016c5c9fca33411eaa5638acd815dbf9 : SUCCESS in 1h 50m 38s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/57986d98148f44a69c6452a80ef1abb4 : SUCCESS in 1h 12m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/934cbf7bcc5e496284900d054b012b56 : SUCCESS in 1h 15m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/273ac53e3218453891567d3807d8e690 : FAILURE in 1h 18m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/98c9c30288df43bf8882ab7dde2a5735 : SUCCESS in 1h 08m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b606201d17fe4e87849f270375f6a0e9 : SUCCESS in 1h 03m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1b35af9de5e44c0885497842eeb79f48 : SUCCESS in 1h 43m 25s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/605d5dd6698743529c179b5ce2faff6c : SUCCESS in 2h 22m 17s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/041bd53d657145c988ca50dbe6ff74f3 : SUCCESS in 1h 39m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6b6275d4de44418290553e6d57278de7 : SUCCESS in 1h 07m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/79d86b632df046338eefa1aedeacbae5 : SUCCESS in 2h 51m 24s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bb878798365d4e49be5578bcf181b34f : SUCCESS in 1h 30m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/69effe56e1814907a88175c8a8dc6062 : SUCCESS in 2h 11m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8500173d58742f3a1418235b13dc6b8 : SUCCESS in 1h 02m 22s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e821a9db4ef64bb7af27646c07226e14 : SUCCESS in 1h 21m 07s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b214e881d48e454d90c017db21970297 : SUCCESS in 40m 02s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cf4e4bc74b354334a52ddf2f5193f6be : SUCCESS in 1h 16m 17s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/de933e79aeb043e6a26662fd571b5e43 : SUCCESS in 1h 24m 08s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3e8057fe26324b19ad97192f6caeccf6 : FAILURE in 3h 06m 07s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/57ab9ddfb0514ce5bd5e047a52d8ba97 : FAILURE in 1h 50m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9547016e563d6829cf735ffda70571f47e6271f7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-12 00:49:38.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/54/710554/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"0e59fa65e195239bf6b9a107bca9cb581f8b0fcd","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-03-17 08:16:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e8b2eb77142794c42d424e2083249f5102c4b5ec","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-03-17 08:18:10.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"94f717ed5715dc16e5bc015f2ad9d6f377713dd2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-17 09:30:09.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/54/710554/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"752dc9048c7f8f231febe3f286a3507b10d1a7cc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-17 09:58:35.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/710554/2 : FAILURE in 1h 42m 00s","accounts_in_message":[],"_revision_number":2},{"id":"c6ad819b9364fdae51def01c8cf05c71d1b71b8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 10:50:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dd8b635adbf747c9af3803ab36f5b24e : SUCCESS in 47m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e3d3fb3f3e134b4e911e39c6a9a81e86 : SUCCESS in 34m 16s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/9d96bb276b24472599662150ee3c313f : FAILURE in 2h 11m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/47de684d11664db587887a24a758efa8 : SUCCESS in 1h 28m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d906150097e54f2d9eadd7f496f0c4ca : SUCCESS in 1h 20m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/163bc37a87804102a3665c1793ca3613 : SUCCESS in 1h 01m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a1178a9b7b54a8aa9f630cacb464b40 : SUCCESS in 43m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a9558b715a4426cb870f28815db4cec : SUCCESS in 23m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36b0a78c305d4a819b316cf4dc029c51 : FAILURE in 12m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a72b1a97910404c8ff5030f47e217be : SUCCESS in 25m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46b7d7dfc4714702ad91d270f70800f4 : SUCCESS in 25m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23b99b5a79174e9a84b09dbb614b681a : SUCCESS in 27m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/760b00d0adc54581af6341749e98bc28 : SUCCESS in 11m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4ba23f7de1c343178ab0acab69cd90ec : SUCCESS in 51m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/65ccd0ccb62a4fe1be1019ed990c436d : SUCCESS in 1h 05m 16s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/427d3b3eee154cd78a5df6d297a736d6 : SUCCESS in 1h 46m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/52203942855646bda6f9796f712ba2a3 : SUCCESS in 1h 39m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae9a14cf63314b5bb085a9ea38039c95 : SUCCESS in 1h 10m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/00fea135c32f4ce6a840591e5f58ea9a : SUCCESS in 1h 12m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dda936a0a8e443ccab56685d3f794f54 : SUCCESS in 1h 23m 22s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ef3694621fb44ba0812e8aafb7ee217c : SUCCESS in 1h 42m 32s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9a19f572059f4f96b44dab2e9787a5fa : FAILURE in 1h 47m 52s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d8ec1e339f0c42bfbbba82e3a37b42e0 : SUCCESS in 1h 17m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e83c3d78b64e41509cf9c206814b3765 : SUCCESS in 55m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b6ae72b6b9904b72a00cbba790dcbcd5 : SUCCESS in 2h 31m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8dd72cde9011441995ea36076bf6c7bd : SUCCESS in 2h 11m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1684990a35d4408a8dce39e8dbf5ee3f : SUCCESS in 1h 54m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6950eb1a07a54cb3b55996786414a189 : SUCCESS in 1h 11m 27s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/69ebf828f75347f4918c6154ac78f79f : SUCCESS in 1h 26m 24s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/30af4b65f3604f1fabebb8288c66e9d4 : SUCCESS in 53m 16s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d270a18e0153410693b6784a23530dff : SUCCESS in 1h 10m 03s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e320d33d7ea451b882933d4d7da5f76 : SUCCESS in 1h 19m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1e9c6a8962af4a588d84585f643346cf : SUCCESS in 2h 06m 10s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f8d7848c9bfc4a38a7e2f697a53d7034 : SUCCESS in 2h 27m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d39be9d5af757427e0b2e376cb8c5bb2ba801b0f","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-03-17 11:57:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8faf0bb5028b166b67e3d39ba344aefd767cc7b1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-17 13:13:05.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/54/710554/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"86e97527a8654e08b1905e8525861cc075c8bf1c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-17 13:52:08.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/710554/3 : FAILURE in 1h 54m 23s","accounts_in_message":[],"_revision_number":3},{"id":"65ee35d4d671b1e2d0c7ec6bdc71e0a3e0aa5772","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 14:28:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a556468e4b5a4ff3a0c51a72a5456589 : SUCCESS in 1h 04m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/22c4e7954be6430ea84093643100bd98 : SUCCESS in 37m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/2f27b1a7224942b2ace5bd655f1aa1e6 : FAILURE in 1h 55m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/86627da576ca4d318a1081fdb6645d87 : SUCCESS in 1h 23m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d28df36ee789493cace79cb176446b1c : SUCCESS in 1h 08m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8cbfe0c23d0d49fb8285d82c0f1fa5c1 : SUCCESS in 1h 16m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5534e6acefaf4e26ae9e0b6f8b7470ad : SUCCESS in 37m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db9be7208d8645aca890076630df7785 : SUCCESS in 30m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0566142c86b406b91fb4f09324337b8 : SUCCESS in 15m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32714a9a013541baa1e33430dcc58708 : SUCCESS in 27m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3e2827d407c44a1b4f5d7d893306bb9 : SUCCESS in 27m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bc81bdfd6d744609dd4f030dd699644 : SUCCESS in 28m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a94a850461e4745abe082cd7f8d6ee2 : SUCCESS in 9m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a9f3f66d7e0b45799d95c7382d725e30 : SUCCESS in 52m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/81878100c4e74bb6b8c07aecdef39ba8 : SUCCESS in 1h 09m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/895aef89ccd94d0494826a6bac742375 : SUCCESS in 2h 15m 42s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2837eb9bc8eb47ce8687b2378114d820 : SUCCESS in 1h 15m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c487370e423643b89d181337a405ee59 : SUCCESS in 1h 13m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5eef78134910495b946cd793eaf06f28 : SUCCESS in 1h 06m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e694e4f4cadd4a23bc07e17e2fa16068 : SUCCESS in 1h 15m 53s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e599b626d02b474b9024949a256e7cb3 : SUCCESS in 1h 46m 25s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f0d837137b11482d93925b1f1cc4aac7 : SUCCESS in 1h 39m 57s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e2aa30cc7240428cb773cd769f7bcded : SUCCESS in 1h 14m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bbe8521939324bc686d27d360dfdce72 : SUCCESS in 53m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a11a7c2025e94aed8c5db84e12e15c91 : SUCCESS in 2h 25m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/088bba5900154d1a8001673cad1ad428 : SUCCESS in 1h 54m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2e5c7aa9a1e145e5a21fc80f17a61601 : SUCCESS in 1h 56m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc1eda13b37f4daa9e39c265d87082b1 : SUCCESS in 1h 11m 22s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1e38d3ded7914a06bff446ad9298349d : SUCCESS in 1h 20m 08s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d4970284b8a9433c8927fa2b6a7b3919 : SUCCESS in 48m 33s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/34ad206b197a4e808cc9bd4b9dd80ff3 : SUCCESS in 1h 06m 59s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/2ec23592af3d43a483757c8709b8e5ef : SUCCESS in 1h 15m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/31f334e71a8c4ae1b8e37d5da14e4531 : SUCCESS in 1h 58m 34s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/bc8696d8d76c4ac8aae099c961752129 : SUCCESS in 2h 15m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5144f6ce583aca8fd433b1d5c38b192627435fba","author":{"_account_id":29874,"name":"Taoyunxiang","email":"taoyunxiang_yewu@cmss.chinamobile.com","username":"Yunxiang"},"date":"2020-03-18 08:21:59.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"5d5362c92785944aec3df517fc2687a2037115ce","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-03-19 01:47:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"99f6dc431b5f2159d5f7a2d3ed210f7bf7c3daa2","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-03-19 01:49:36.000000000","message":"Patch Set 4:\n\nresolve the conflicts.","accounts_in_message":[],"_revision_number":4},{"id":"c6797411587f1e1c57e021ad9824a47010f258a4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-19 03:31:51.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/54/710554/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"0082b0b1a61703e6f8b65a1816fcf3d69a7ea03d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-19 03:33:45.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/710554/4 : SUCCESS in 1h 45m 15s","accounts_in_message":[],"_revision_number":4},{"id":"57009da130a0b436fa0222e469ed7464d09fd1e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 04:19:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b11c4c9f41fa4d91ae75ad7b40d8a385 : SUCCESS in 54m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9ba116e84b8c4957b4959c985d30f536 : SUCCESS in 41m 50s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/d0c0a9078a4e47d8b9d7771531a60e6d : FAILURE in 1h 59m 41s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ae664f97898b48cea431649473dff212 : SUCCESS in 1h 16m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b988d8a6800c44859780c25585e00ec3 : SUCCESS in 1h 31m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/988263cb2b604efdb0d8caf93a9e01f7 : SUCCESS in 1h 08m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76b8ba4dd27f45e291e7f7fb670dfbd0 : SUCCESS in 34m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd3e4fb08bc74c81a80e1c75ecad3c6d : SUCCESS in 25m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ab6ff7294ee497a98c35fd8e314185f : SUCCESS in 13m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1ac014ba945481e80795d67dc468f57 : SUCCESS in 25m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/140dae86232540a5aed2bf785821e863 : SUCCESS in 29m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/189aca29277e4070ae637c9011d06bed : SUCCESS in 24m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/314a71af25b24e2eadd4bb2025b57117 : SUCCESS in 10m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4b148fbd7e1d4eac9418e837e5d3b18e : SUCCESS in 47m 26s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/885616cdcc124b3db879070e846c82b9 : SUCCESS in 57m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ee3b4311635945feb35cb954a7af14bf : SUCCESS in 1h 48m 02s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c7dfb630399245d8a5cf06e8a92e750e : SUCCESS in 1h 14m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8bc448f428c94df39993343f9b34b062 : SUCCESS in 1h 09m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/347aeb78ff9d4c0f893479fd02ab7439 : SUCCESS in 1h 10m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b0bd64f25abc4f3bb7a523687809d7b5 : SUCCESS in 1h 14m 17s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8be14e79ed94423390a68c3fc9e47165 : SUCCESS in 1h 35m 16s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fb8712284b5a4dd6b0d385e131a0482b : SUCCESS in 2h 02m 52s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/62f7c4a7fead48adb3407bbc5a11ba2b : SUCCESS in 1h 48m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b2d764291bed4684a7d647c7b6272e29 : FAILURE in 1h 27m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ea52c8f0d70949e08fb4c1bae5afba50 : SUCCESS in 2h 10m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c164bef5ca114ed385b056b18789087d : SUCCESS in 2h 10m 33s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c68397a32b4c4cb9baf39cc960413403 : FAILURE in 2h 24m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b8f25996dfd461a87c3718d301e5229 : SUCCESS in 1h 14m 40s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ea743453bb324d27bfd2285ffde7a7f2 : SUCCESS in 1h 07m 26s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4fd296ef1f0a4e2aa5a35a0125d3b7dd : SUCCESS in 1h 13m 00s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/33da96e2b96f4fcd8785059dffc038ea : SUCCESS in 1h 00m 19s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/cc7692e1d85d45879f4aec69ec495f5c : SUCCESS in 1h 41m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ea86bbc73dcf4723a8f752fb4befc6b9 : FAILURE in 2h 13m 45s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e1bbfa3f92b94e50b762de935bc8f3e1 : SUCCESS in 2h 00m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cef51f2b29906f82fec12266c3ff140677219876","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-29 22:23:12.000000000","message":"Removed reviewer Brian Haley.","accounts_in_message":[],"_revision_number":4},{"id":"92244565c998e91815c77b63b769810d21149f64","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-22 08:03:28.000000000","message":"Abandoned\n\nPlease, feel free to restore the patch, address the comments and rebase the patch.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"da9b0ef6f8600b3ad336d751b91e36c10bc10a4c","revisions":{"999a11d7196d8832264e117873766dcafeed5e09":{"kind":"REWORK","_number":1,"created":"2020-02-29 10:49:11.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/54/710554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/54/710554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/710554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/710554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/710554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/54/710554/1"}}},"commit":{"parents":[{"commit":"4d5817003cc2766839f7629234d7a72697821ae9","subject":"Blacklist test_multiple_ports_portrange_remote for OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d5817003cc2766839f7629234d7a72697821ae9"}]}],"author":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-01-30 06:29:41.000000000","tz":-300},"committer":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-02-29 10:49:03.000000000","tz":-300},"subject":"[OVN] Remove DNS dependency on original port dict","message":"[OVN] Remove DNS dependency on original port dict\n\nThe \u0027port_object\u0027 needs to be removed, so the current DNS dependency\non it needs to be resolved first. The \u0027dns_domain\u0027 is stored in the\n\u0027external_ids\u0027 of the logical switch. The \u0027dns_name\u0027 and \u0027dns_assignment\u0027\nare stored in the \u0027external_ids\u0027 of the logical switch port.\n\nChange-Id: If08909dfbb841ad941e6aaea0990e9e98354e3d9\nCloses-Bug: #1863987\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/999a11d7196d8832264e117873766dcafeed5e09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/999a11d7196d8832264e117873766dcafeed5e09"}]},"branch":"refs/heads/master"},"1f1ceeb9c38be9cb9ceb8d4065af3e12680315a5":{"kind":"REWORK","_number":2,"created":"2020-03-17 08:16:04.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/54/710554/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/54/710554/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/710554/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/710554/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/710554/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/54/710554/2"}}},"commit":{"parents":[{"commit":"0bc66b5fe88d14e68a09f2a4b847de56ae3c0b91","subject":"Merge \"[OvS] Handle re_added multi ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0bc66b5fe88d14e68a09f2a4b847de56ae3c0b91"}]}],"author":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-03-13 07:28:44.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-03-13 07:28:44.000000000","tz":-240},"subject":"[OVN] Remove DNS dependency on original port dict","message":"[OVN] Remove DNS dependency on original port dict\n\nThe \u0027port_object\u0027 needs to be removed, so the current DNS dependency\non it needs to be resolved first. The \u0027dns_domain\u0027 is stored in the\n\u0027external_ids\u0027 of the logical switch. The \u0027dns_name\u0027 and \u0027dns_assignment\u0027\nare stored in the \u0027external_ids\u0027 of the logical switch port.\n\nChange-Id: If08909dfbb841ad941e6aaea0990e9e98354e3d9\nPartial-Bug: #1863987\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f1ceeb9c38be9cb9ceb8d4065af3e12680315a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f1ceeb9c38be9cb9ceb8d4065af3e12680315a5"}]},"branch":"refs/heads/master"},"6a3ebbfb80387b4afc307b739b7dc73dd02f7de8":{"kind":"REWORK","_number":3,"created":"2020-03-17 11:57:05.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/54/710554/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/54/710554/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/710554/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/710554/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/710554/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/54/710554/3"}}},"commit":{"parents":[{"commit":"0bc66b5fe88d14e68a09f2a4b847de56ae3c0b91","subject":"Merge \"[OvS] Handle re_added multi ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0bc66b5fe88d14e68a09f2a4b847de56ae3c0b91"}]}],"author":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-03-13 07:28:44.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-03-17 03:36:18.000000000","tz":-240},"subject":"[OVN] Remove DNS dependency on original port dict","message":"[OVN] Remove DNS dependency on original port dict\n\nThe \u0027port_object\u0027 needs to be removed, so the current DNS dependency\non it needs to be resolved first. The \u0027dns_domain\u0027 is stored in the\n\u0027external_ids\u0027 of the logical switch. The \u0027dns_name\u0027 and \u0027dns_assignment\u0027\nare stored in the \u0027external_ids\u0027 of the logical switch port.\n\nChange-Id: If08909dfbb841ad941e6aaea0990e9e98354e3d9\nPartial-Bug: #1863987\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a3ebbfb80387b4afc307b739b7dc73dd02f7de8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a3ebbfb80387b4afc307b739b7dc73dd02f7de8"}]},"branch":"refs/heads/master"},"da9b0ef6f8600b3ad336d751b91e36c10bc10a4c":{"kind":"REWORK","_number":4,"created":"2020-03-19 01:47:59.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/54/710554/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/54/710554/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/710554/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/710554/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/710554/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/54/710554/4"}}},"commit":{"parents":[{"commit":"2d9b2ffcee737ce2035bdaba3844ce977d82c3ae","subject":"Merge \"Clean dvr fip gateway entry when fip agent gw port is deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d9b2ffcee737ce2035bdaba3844ce977d82c3ae"}]}],"author":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-03-18 16:23:14.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-03-18 16:49:23.000000000","tz":-240},"subject":"[OVN] Remove DNS dependency on original port dict","message":"[OVN] Remove DNS dependency on original port dict\n\nThe \u0027port_object\u0027 needs to be removed, so the current DNS dependency\non it needs to be resolved first. The \u0027dns_domain\u0027 is stored in the\n\u0027external_ids\u0027 of the logical switch. The \u0027dns_name\u0027 and \u0027dns_assignment\u0027\nare stored in the \u0027external_ids\u0027 of the logical switch port.\n\nChange-Id: If08909dfbb841ad941e6aaea0990e9e98354e3d9\nPartial-Bug: #1863987\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da9b0ef6f8600b3ad336d751b91e36c10bc10a4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da9b0ef6f8600b3ad336d751b91e36c10bc10a4c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
