)]}'
{"id":"openstack%2Fneutron~712508","triplet_id":"openstack%2Fneutron~stable%2Ftrain~I953062d9ee8ee5ee9a9f07aff4a8222ac63ed525","project":"openstack/neutron","branch":"stable/train","topic":"bug/1863423","hashtags":[],"change_id":"I953062d9ee8ee5ee9a9f07aff4a8222ac63ed525","subject":"Fix queries to retrieve allocations with network_segment_range","status":"MERGED","created":"2020-03-11 17:07:30.000000000","updated":"2020-03-26 00:51:25.000000000","submitted":"2020-03-26 00:49:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712508-1585183759407-fe09d30a","meta_rev_id":"1d5b7d659829c39e6e2f4a99b7e66f2dd2338fac","_number":712508,"virtual_id_number":712508,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-26 00:49:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-16 16:43:42.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-03-17 16:18:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-03-25 21:42:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-03-25 21:42:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-16 16:43:42.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-03-17 16:18:48.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-03-25 21:42:22.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-03-26 00:49:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f12a47812a02bde37d85b4737bd7a3c9398d0ce","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-03-11 17:07:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cb06149f96b8d72a2062f588e1ee558685d634b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-11 17:55:29.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712508/1 : FAILURE in 47m 24s","accounts_in_message":[],"_revision_number":1},{"id":"cbafe194880c7ff747360b20b9706484d88c6760","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 23:39:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/670ed42e52554f37802f6351c1c82bc2 : SUCCESS in 1h 06m 25s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/67d83a356db04b83833f8b991dfbfe6b : SUCCESS in 46m 26s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/fff3918d70c9423997fef08ae4bbe7ce : FAILURE in 2h 16m 52s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/90509a9010644baf8af0f2744d5de1f4 : SUCCESS in 1h 24m 25s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/3be4fb8cabcd4ee69c2ab7e74a0edcce : SUCCESS in 1h 24m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/4c383fb961e54dbb865ebbe604041923 : SUCCESS in 1h 15m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f27351c7aff143c58129001d4e087175 : SUCCESS in 30m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa8e9838454e42988a648da88b1e9575 : SUCCESS in 20m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/516c5876e00d4efe8844b13529c42267 : FAILURE in 7m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8f13e6630c56405fab5cfa9733ac8287 : FAILURE in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3f044e2508b473b85cacc828820adf8 : SUCCESS in 27m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/93a19140731f477688a3560bf17d6132 : SUCCESS in 21m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3173b672dbfc4e27a671732cf62a1f19 : SUCCESS in 8m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4c28aa0fb92a461391867c27a115c476 : SUCCESS in 1h 03m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8bda749815854c419d6c0a9e7a8a3256 : SUCCESS in 1h 16m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/97e2c0bfee0d410c93fd50ab9e416bb0 : SUCCESS in 42m 54s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/f5311ac7d3ef41e091ce6930f8d7c2d3 : FAILURE in 16m 28s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3c5e12fab3d443a991ad5ccadd2a5b6c : SUCCESS in 1h 14m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a6f8d5b5710440c59fba67d83b4b7a0c : SUCCESS in 1h 47m 21s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/38f50b1f8dcd48149d522120cde5a780 : SUCCESS in 1h 13m 43s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d0c117b863f04575b4d095bea67b1d55 : SUCCESS in 1h 16m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b52368cb40a445ead8cde13eb9cc909 : SUCCESS in 1h 16m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c527859800884f1499ab594bf66da858 : SUCCESS in 1h 21m 24s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ca2942ffdeb14ede9b1e5e88a0958f78 : SUCCESS in 1h 33m 56s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/61f0d00fda3a4944a908067ae97439ca : SUCCESS in 2h 00m 18s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9bf02419b5cd4cc99136818883420662 : SUCCESS in 1h 13m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c71e88266d1541de8be59c26728d2653 : FAILURE in 46m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/eabff90a6bbf43179ee99836a24c76ba : FAILURE in 31m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a90c869d15724099a40dff9ba3a1ca77 : SUCCESS in 2h 23m 39s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/048c6d3dd90946769867d0fe28ca7c27 : SUCCESS in 2h 43m 37s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/14c122a5362b4ae58ff294155c17abd1 : FAILURE in 28m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/757e4bfc87a24337a191e4046d206cc9 : SUCCESS in 1h 10m 32s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7d29c5a0212649cba929e7cdebf6ab85 : FAILURE in 31m 37s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a7a231db2c9c45039e234e4b0a6fab74 : SUCCESS in 41m 13s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d99671865ce143c39f53f2e7d78724b9 : FAILURE in 1h 12m 05s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/dcd1f96e8c3646cc9608b4e61c0272a3 : SUCCESS in 1h 12m 58s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ebe189bcc88744ecbd7dafe938950fda : FAILURE in 28m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e81a187f61664ba92bbc0e10e1cf08634668d657","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-03-12 11:10:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f3e91e75bb76114261583868dff5535f75f9cf6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-12 11:34:32.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712508/2 : FAILURE in 23m 27s","accounts_in_message":[],"_revision_number":2},{"id":"8278e33a016d195c8be9e842889de2050c7a7f7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 14:05:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/4ed8419c654045f98aa0444f6d0d2361 : SUCCESS in 51m 34s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/058c32a1783d4f5390364d82ccf6d4e6 : SUCCESS in 39m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/7a9db637df594ec695fc69a72f865aad : FAILURE in 2h 01m 23s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/b9a1feac728b4cc799015a5db761df9a : SUCCESS in 1h 27m 24s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/4f0298b9b19b4a1a84c424130cd6b653 : SUCCESS in 1h 21m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/9e02b6b0c5de43c69da0bcfb9e0d5bf5 : SUCCESS in 1h 08m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b50d32f6371c4e7e9b260256181c5ed2 : SUCCESS in 35m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d46681f2ae2402ba2a7f2d80385e375 : SUCCESS in 26m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cb31522f77a4672bb0a671c6e749940 : SUCCESS in 17m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bab98dd45e3e4aed927e0ce50d1bb55f : FAILURE in 39m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b6084ab783e44c79c05dfc8a87030fd : SUCCESS in 38m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8021520c88bb49e1a01825a9ae184050 : SUCCESS in 39m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d457af965817462786c192a965382789 : SUCCESS in 10m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f84e3a0a1ea5495c9b9dc3a41c92a619 : SUCCESS in 1h 11m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6a3213fc079a469ca215416578e7ea5f : SUCCESS in 1h 26m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9e387f37f06d4db985b35994775d4be2 : SUCCESS in 45m 41s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/7c316acd9e28467c9301f4b4892e76e0 : SUCCESS in 42m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1b7e5c52cd734b59a13acc02bd5318fb : SUCCESS in 1h 03m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e47305902c05463eb33199a592900318 : FAILURE in 2h 17m 47s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/38dcfc8ee91645c29d416ebacccb770b : SUCCESS in 1h 24m 00s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e93d4ec15cc84dcfa052b6a5d8209062 : SUCCESS in 1h 33m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bd8371c2efde493383636edd61447692 : SUCCESS in 1h 11m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cca1f653ec8443dd9a328dc758704fdf : SUCCESS in 1h 15m 48s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/80d81a7514ef4e40bd4de7cc46aef91c : SUCCESS in 1h 35m 44s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b885420912dd42de8373e01163e5d608 : SUCCESS in 1h 43m 20s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/79e0b7e8040f44f199ebac9172631bfd : SUCCESS in 1h 16m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a06e7cdc073047cab155506b9c31a633 : SUCCESS in 1h 02m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c4cb41d373f04f85a9c62042922cbe1c : SUCCESS in 58m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d71935fb8c81428ebf5d59a3f16434b5 : SUCCESS in 2h 17m 57s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/47000ff9f148434aa6fb4610ce5d43d2 : SUCCESS in 2h 42m 22s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/3fdc899354094b2382dddc80919a7c86 : FAILURE in 44m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36680dfdb7c744ca8734e216e07b1eb5 : SUCCESS in 1h 00m 52s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/13616429e4c2415687f4376ac237f7ab : SUCCESS in 1h 19m 24s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d1303ca84aa54d2492690e16d8b96221 : SUCCESS in 36m 21s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/62da96a1821a4a71b21cf1875a6a3a9d : SUCCESS in 1h 03m 13s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/6ab4fa502c6e435bb21b91330fdd67cd : SUCCESS in 1h 15m 34s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ef4a1b9db3924197a9777bb6f6d9c5cf : FAILURE in 1h 10m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d4c767b1c02d83750dff27e89538d84c119e43c4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-03-12 15:01:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d5201f060af7eb4e98203003de26ad92d493c627","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-12 15:58:08.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712508/3 : FAILURE in 54m 22s","accounts_in_message":[],"_revision_number":3},{"id":"6613709a8a0e2a1f9c5bcfdcdd68d4a02922de33","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-03-12 17:09:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"be02e3ad3795ee8f466494cddd9fae645e14e4b8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-12 17:57:20.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712508/4 : FAILURE in 45m 37s","accounts_in_message":[],"_revision_number":4},{"id":"4853a77370f44bd4bb713da80ecf996de4aaa14b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 20:51:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/8008cbf6289e4fd7bee7fd49997851bc : SUCCESS in 1h 12m 00s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/0843a556e24447f1b9b4324900e4d01d : SUCCESS in 49m 43s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/c7e0400ff8df433f9af68c311379960c : SUCCESS in 2h 12m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/b8154d6341e942609d53844f01aba577 : SUCCESS in 1h 18m 06s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/cfbe15b5a4574e6bba6757797f184991 : SUCCESS in 1h 12m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/e2c44377c7ee4f16b7f37875b4353143 : SUCCESS in 1h 13m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d2e5c7ad912483aaef124332bfbb0de : SUCCESS in 33m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5973af37a0ce44bbbfd717d50c094bc1 : SUCCESS in 25m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f57b3f9439ff4b6b91802b538c4dea22 : SUCCESS in 12m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/79d4d4b3bd5044f0ba2d09ad82b7b491 : FAILURE in 25m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2ea30ab466a40a08adbb572b1b71e53 : SUCCESS in 25m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed40f6bfbb4f4aec8227ca2e69f6c2ab : SUCCESS in 22m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ec7f40ac2ae4b1eb633e52dfc5c1351 : SUCCESS in 12m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/42f55886b2e74cc28cd36807cf332401 : SUCCESS in 1h 08m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e9a5f23b7df74401a0bcfe9b0ea8b140 : SUCCESS in 1h 22m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7bfd70b1f21e43be96db2a409ac6c5eb : SUCCESS in 50m 02s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/209e311a82df47f4b7b1ccc162aef02c : SUCCESS in 42m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9aad662b2bb340eba9b625fb5f3d1763 : SUCCESS in 1h 01m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9dfcd9b1934c4e3883ba11bbf4f0c4d1 : SUCCESS in 2h 05m 50s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/53462c4538f9454f98ea67c48fb12dd1 : SUCCESS in 1h 15m 50s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/bcdf2fd020a64354bc75a6db8bb285dc : SUCCESS in 1h 11m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a832b1f131ed46bd824f1205a27b5a0a : SUCCESS in 1h 13m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6dfc664e4ae441e394bafd8f1e8aab7a : SUCCESS in 1h 24m 34s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ab7c8474af9f40bb84c0496f374befb6 : SUCCESS in 1h 49m 15s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fa1a0aaa05534053ae4543c984bc066e : SUCCESS in 1h 57m 34s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4903cb961d6845b994383299878f116a : SUCCESS in 1h 19m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/679c779e672849c1896c03cd17451b7f : SUCCESS in 1h 03m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/51285001d3304222b56780b588110368 : SUCCESS in 52m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/783aa0c74b4c4127b1e68f5773523d0b : SUCCESS in 2h 19m 38s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/aa722b84de2449ac98169870ba4626b6 : SUCCESS in 2h 53m 54s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/cd452ee902274c4981cf5af860ea19e7 : FAILURE in 52m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6070a6d657040c289ba5bfd4c5bc79a : SUCCESS in 1h 04m 52s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/47c10bb3cfed47299c956f2a8d9f5e76 : SUCCESS in 1h 22m 24s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/50970213023e434887938aef21f5651f : SUCCESS in 39m 51s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ebc4d745187a489e86a1314bc6a5088f : SUCCESS in 1h 09m 25s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/adcc1e5d970c46df9d29dcf05b095f46 : SUCCESS in 1h 23m 45s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8cb9f0c91d57487fa3363b667b118902 : FAILURE in 1h 22m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2b81fbe1405598c111551ca3864112fda51a16e3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-03-13 12:01:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"352d8436519bc463bc71de44f7b51c72eb98b5cd","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-13 12:47:51.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712508/5 : FAILURE in 46m 15s","accounts_in_message":[],"_revision_number":5},{"id":"3a3ffca7765074cf9cda57f678befa2d07d754ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 14:28:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/7a4cd780ba0242d78f6fdb42f41de0c2 : SUCCESS in 1h 07m 20s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/f24b8060af384e499842a9bffb57b7b1 : SUCCESS in 38m 08s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/01bc80aa38f14cf1b98d48a8873d7c4c : FAILURE in 1h 50m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/bf3998b54d2e46588df99190aebf10d4 : SUCCESS in 1h 54m 59s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/3d32af9992c747fe998540bd86ff52a1 : SUCCESS in 1h 57m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/e952dea214644545b2eb2962ed833bd4 : SUCCESS in 1h 14m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb62650cc3b54ac78c3d30feb0dd385b : SUCCESS in 26m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db9d0d93cd94441d875c9603b19633f3 : SUCCESS in 29m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2acd09ba3f324b55a843b742c9153108 : SUCCESS in 16m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fb160ea59f1544a388a1ab028ce523ef : SUCCESS in 31m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ff806b072bc4dc699d2186ddb0f787b : SUCCESS in 30m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/edd4dcd3ef0840eba182f765ab1c7067 : SUCCESS in 26m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca2c935a7fe04cccbaa8d9e8f67c5943 : SUCCESS in 7m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3a819d2043ef47609bf4836006f9d468 : SUCCESS in 1h 20m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/64cc1bb36a884930b256da6e8da4f9bc : SUCCESS in 1h 35m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/49ffd3f5ac2e416daf975d5ddf69588f : SUCCESS in 1h 02m 34s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/e74f0a798261480e8532a5f2f92d2b2c : SUCCESS in 40m 29s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e0dce5e4b12d40078477423ca3111402 : SUCCESS in 1h 20m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8b8e9b40c26f404ba6d71cd89506b868 : SUCCESS in 1h 55m 20s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bc73501b4794439facdd4cbc811dd58c : SUCCESS in 1h 08m 43s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/610ed0baabe84790ae8fbe05211714a4 : SUCCESS in 1h 16m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a9f179345bcc403db69cd0ef11974346 : SUCCESS in 1h 16m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e2b97b6b10e646ea88463926988f32a2 : SUCCESS in 1h 12m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9af1493f31ac4349825d46fcce2cab0a : SUCCESS in 1h 41m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c3e42fc206bc445fbdcb2538c1ba3d0f : SUCCESS in 1h 55m 45s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/649177a807d64cdab3c2b160d4e4ccc8 : SUCCESS in 1h 03m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a8e520cc91114584a6205497ead73374 : SUCCESS in 58m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/40ac46a987f248da8a613d60559422f5 : SUCCESS in 53m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8645db649ce44ac983d62e64299c1986 : SUCCESS in 2h 19m 02s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a31186b2a78047c6b924d52d5be1df85 : SUCCESS in 2h 13m 06s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/0001d7a55b9f4b6b9e590be9f00dacf9 : FAILURE in 43m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e318573148d4625ac5950e281e43128 : SUCCESS in 1h 29m 44s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/46e204d64dc848beac6d0c82dc11b93b : SUCCESS in 1h 29m 54s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f7badefd7a3f4107852551925d2e455e : SUCCESS in 1h 01m 14s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/74884f876b1c44df889eb36524010ba2 : SUCCESS in 1h 09m 12s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b04db4700b3e49238dedbac71af3c441 : SUCCESS in 1h 12m 35s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3c0a96f1c32e479da37e7629e7716314 : FAILURE in 1h 49m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"834322312efae9d029af6656bdfbdb71d1a785ad","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-16 14:06:41.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)\n\nLarge backport, but this actually makes the feature work. Code changes look OK to backport too.\n\n2 fixes needed: Conflicts or similar section to show where changes were required, and one file looks like it is coming from a separate change\n\nSide note: a bit late, but should this have a release note? (in master) \"Fixing this feature so it actually works\" seems worthy","accounts_in_message":[],"_revision_number":5},{"id":"f83a3e6f4059f5fa9a29fa3c5466d4c88fde97e5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-03-16 16:18:45.000000000","message":"Patch Set 5:\n\n(3 comments)\n\n\u003e (3 comments)\n \u003e \n \u003e Large backport, but this actually makes the feature work. Code\n \u003e changes look OK to backport too.\n \u003e \n \u003e 2 fixes needed: Conflicts or similar section to show where changes\n \u003e were required, and one file looks like it is coming from a separate\n \u003e change\n \u003e \n \u003e Side note: a bit late, but should this have a release note? (in\n \u003e master) \"Fixing this feature so it actually works\" seems worthy\n\nI\u0027ll push a release note in master and then I\u0027ll backport it to T.","accounts_in_message":[],"_revision_number":5},{"id":"3bcb51374cf371338284149b1674c9588a7d4777","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-03-16 16:19:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2f28f129192f457e6261cdab393e2d6f88604497","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-16 16:43:42.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712508/6 : FAILURE in 23m 53s","accounts_in_message":[],"_revision_number":6},{"id":"58de519256ad278f6dba5a34db694339c18417cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 19:34:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/dd3a65c2feff4ae8bdacb49e91b1cb44 : SUCCESS in 1h 05m 03s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/77f582cfab7946faa1fe43d13f83dbd7 : SUCCESS in 37m 18s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/493c57aefc7e40a28fba430dcd839e1c : FAILURE in 2h 04m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/420beb2a6e114999be8b9e0c9d451d47 : SUCCESS in 1h 34m 16s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/0c43e533c56143b89be1693c5588f00c : SUCCESS in 1h 14m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/dfe0a68daf2441c9bfca7c2a9871cd52 : SUCCESS in 1h 24m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a44eaf90295543eb8b647a74cb12bb79 : SUCCESS in 34m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03f3bfe68a6b49c384bf0b6eb507d83d : SUCCESS in 23m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e61ede04ace6425f90bbfa0c5f42a86a : SUCCESS in 13m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6f8d0db46fdc4f6291cf5b41459fbd2f : SUCCESS in 23m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9eab19f6946404091169e81b48c250c : SUCCESS in 20m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/27a6b72000344596a97573d97aabef3c : SUCCESS in 22m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd60e1e0ecab4fdda06b4d84c6f54082 : SUCCESS in 11m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/65e1721185f3402c8721a370505caedd : SUCCESS in 1h 03m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/822cd2e1200742c6a4236f273fe45679 : SUCCESS in 1h 23m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f1d09b5af1c5498abef0d96781f64c97 : SUCCESS in 44m 50s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/7091beeba63d406e9381afbfe9efca2c : SUCCESS in 44m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d84de596773b4255be211b57e8133652 : SUCCESS in 1h 09m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/327dcc6f12ed4292bbf2ec6b68cc2e79 : SUCCESS in 2h 01m 48s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e280af8c0135474b8a284e43e2e691e0 : SUCCESS in 1h 26m 27s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/208f0600c265430b9555ad4dc0cb3f16 : SUCCESS in 1h 18m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d664c7c5d46d441e95644f4a8b7ee8e8 : SUCCESS in 1h 43m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ba8205ee31de4e57999f9294744dada3 : SUCCESS in 1h 14m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e26fc473d9c4499db1078a771333627d : SUCCESS in 1h 52m 10s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2d4b760aa4ce4e9aac65acd60949ef08 : SUCCESS in 2h 18m 39s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9d9e5b19d3014f93bea68b800dfaacd3 : SUCCESS in 1h 30m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3f33467f9938469e8dd83c5e97f2adbf : SUCCESS in 1h 04m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/69e682db9d02408aa18f87ecf26edb9b : SUCCESS in 1h 09m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4f8953234e4f49a88da35f03e1b19e1b : SUCCESS in 2h 17m 11s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/368f0cbac4d24f90b825ee5b249dcb78 : SUCCESS in 2h 21m 30s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/16904bfc8f034c6783a4f262c3d0863b : FAILURE in 47m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42b187a30d204544a6ef09724c4a58af : SUCCESS in 1h 03m 28s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b3c78a46c68447fc96d46da44c601de1 : SUCCESS in 1h 22m 02s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4cae303f49da4668bf43970e61d1f690 : SUCCESS in 44m 16s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c003f808c6124b86a0d5a3c103785d57 : SUCCESS in 1h 06m 26s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/af63e6255e924da4aa3f8e8058c6112e : SUCCESS in 1h 19m 31s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1aaec47f5a98480c8c89cc65719b40d8 : FAILURE in 1h 31m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"80b96cc4a66bb136e23bed3ac41352aeaf83ec1e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-17 16:18:48.000000000","message":"Patch Set 6: Code-Review+2\n\nThanks, LGTM now!","accounts_in_message":[],"_revision_number":6},{"id":"0d636d097f3d1dbcde9243a39aa7e5c222612785","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-25 21:42:22.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"02376d93cf56d40ec54b4edd26d083675e58dd65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 21:42:52.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"517f51ae008d6977131265784b81d73f5b7a8bd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 00:49:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"b9a13d6db3b1966af7b9141330394e8b62622c58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 00:49:19.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/520e530f55af4337bf70ccf9fa6822ca : SUCCESS in 53m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38f9849046b24c5a9cc3aaf4f2384b97 : SUCCESS in 20m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a07e3a5a3b94fe59f7378c312843509 : SUCCESS in 21m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9727c1c7c71a4d048cf271342bd17018 : SUCCESS in 33m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61fa622f1bde441c9e2e8d2c2c77c77c : SUCCESS in 33m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51279bb94e88476aaa4adc45b706c842 : SUCCESS in 22m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dc7a03db0a04e819e1b0290dfd15ada : SUCCESS in 9m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b3baae7ba5114917a3057a81265f04dd : SUCCESS in 1h 03m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/afcb8e63c74b4fefbe9f96f9a4f94a4a : SUCCESS in 1h 13m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ca37e7b76c2e4a2388f4f484f57da194 : SUCCESS in 41m 33s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/65f6ee03de8b405b854048e3461bd717 : SUCCESS in 42m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/356592fc769a42b886a8613a6ae778e7 : SUCCESS in 56m 09s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8cdf75539fa844d18b83bb3cd4c01854 : SUCCESS in 1h 35m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a832510bcf9a4f6f8036657e85cd94e7 : SUCCESS in 1h 17m 35s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/02dd940a3af0496694935ed6a7908085 : SUCCESS in 1h 11m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7dcdac5c2334a22b28b89ced336d4d7 : SUCCESS in 1h 04m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/945b20cec2c041c1961eef952c8f2d6b : SUCCESS in 1h 17m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4b73db8ef566402ba44c633198de72a1 : SUCCESS in 59m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cca235e3f625444ab3ed091193f30131 : SUCCESS in 2h 15m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a619cae0ebee47ef9a1ac69d43d30545 : SUCCESS in 1h 06m 27s","accounts_in_message":[],"_revision_number":6},{"id":"897906748e0cf944482c314c429e85407ada5326","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 00:51:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b27070800eac4bd386a9ec1814602bca : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"bbe401aaf9bfdd77e1d43d547b2cdb436b1440c8","revisions":{"3ae08224037370f08f8137343c05b8e684c056be":{"kind":"REWORK","_number":1,"created":"2020-03-11 17:07:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/08/712508/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/712508/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/712508/1"}}},"commit":{"parents":[{"commit":"6a1c6b6e58cee49ea47b6b1e61f82d24a2aa50e6","subject":"Merge \"Remove extra header fields in proxied metadata requests\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a1c6b6e58cee49ea47b6b1e61f82d24a2aa50e6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-02-26 10:39:19.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-03-11 16:37:34.000000000","tz":0},"subject":"Fix queries to retrieve allocations with network_segment_range","message":"Fix queries to retrieve allocations with network_segment_range\n\nFixed the queries to retrieve the segment ID allocations when service\nplugin network_segment_range is enabled. With the previous\nimplementation, a project user was able to allocate a segment ID\nbelonging to other project segment range.\n\nThe solution implemented was discussed in [1]:\n- A project user will retrieve segments from the project ranges.\n- When depleted, the segment IDs will be retrieved from the shared\n  range, never using another project segment ID.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-February/012736.html\n\nChange-Id: I953062d9ee8ee5ee9a9f07aff4a8222ac63ed525\nCloses-Bug: #1863423\n(cherry picked from commit 046672247de56bad950e8267a57bd26205f354a0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ae08224037370f08f8137343c05b8e684c056be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ae08224037370f08f8137343c05b8e684c056be"}]},"branch":"refs/heads/stable/train"},"12367032858deb9926a11aaffdd66aed6fafa5d0":{"kind":"REWORK","_number":2,"created":"2020-03-12 11:10:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/08/712508/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/712508/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/712508/2"}}},"commit":{"parents":[{"commit":"6a1c6b6e58cee49ea47b6b1e61f82d24a2aa50e6","subject":"Merge \"Remove extra header fields in proxied metadata requests\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a1c6b6e58cee49ea47b6b1e61f82d24a2aa50e6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-02-26 10:39:19.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-03-12 11:10:14.000000000","tz":0},"subject":"Fix queries to retrieve allocations with network_segment_range","message":"Fix queries to retrieve allocations with network_segment_range\n\nFixed the queries to retrieve the segment ID allocations when service\nplugin network_segment_range is enabled. With the previous\nimplementation, a project user was able to allocate a segment ID\nbelonging to other project segment range.\n\nThe solution implemented was discussed in [1]:\n- A project user will retrieve segments from the project ranges.\n- When depleted, the segment IDs will be retrieved from the shared\n  range, never using another project segment ID.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-February/012736.html\n\nChange-Id: I953062d9ee8ee5ee9a9f07aff4a8222ac63ed525\nCloses-Bug: #1863423\n(cherry picked from commit 046672247de56bad950e8267a57bd26205f354a0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/12367032858deb9926a11aaffdd66aed6fafa5d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/12367032858deb9926a11aaffdd66aed6fafa5d0"}]},"branch":"refs/heads/stable/train"},"a5950490c9843aae4590bc1e24aa5106bef05688":{"kind":"REWORK","_number":3,"created":"2020-03-12 15:01:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/08/712508/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/712508/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/712508/3"}}},"commit":{"parents":[{"commit":"6a1c6b6e58cee49ea47b6b1e61f82d24a2aa50e6","subject":"Merge \"Remove extra header fields in proxied metadata requests\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a1c6b6e58cee49ea47b6b1e61f82d24a2aa50e6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-02-26 10:39:19.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-03-12 14:59:20.000000000","tz":0},"subject":"Fix queries to retrieve allocations with network_segment_range","message":"Fix queries to retrieve allocations with network_segment_range\n\nFixed the queries to retrieve the segment ID allocations when service\nplugin network_segment_range is enabled. With the previous\nimplementation, a project user was able to allocate a segment ID\nbelonging to other project segment range.\n\nThe solution implemented was discussed in [1]:\n- A project user will retrieve segments from the project ranges.\n- When depleted, the segment IDs will be retrieved from the shared\n  range, never using another project segment ID.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-February/012736.html\n\nChange-Id: I953062d9ee8ee5ee9a9f07aff4a8222ac63ed525\nCloses-Bug: #1863423\n(cherry picked from commit 046672247de56bad950e8267a57bd26205f354a0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5950490c9843aae4590bc1e24aa5106bef05688"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5950490c9843aae4590bc1e24aa5106bef05688"}]},"branch":"refs/heads/stable/train"},"ec26d4e2cf13cc99e06f126b55c0d49134053e29":{"kind":"REWORK","_number":4,"created":"2020-03-12 17:09:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/08/712508/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/712508/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/712508/4"}}},"commit":{"parents":[{"commit":"6a1c6b6e58cee49ea47b6b1e61f82d24a2aa50e6","subject":"Merge \"Remove extra header fields in proxied metadata requests\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a1c6b6e58cee49ea47b6b1e61f82d24a2aa50e6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-02-26 10:39:19.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-03-12 17:09:41.000000000","tz":0},"subject":"Fix queries to retrieve allocations with network_segment_range","message":"Fix queries to retrieve allocations with network_segment_range\n\nFixed the queries to retrieve the segment ID allocations when service\nplugin network_segment_range is enabled. With the previous\nimplementation, a project user was able to allocate a segment ID\nbelonging to other project segment range.\n\nThe solution implemented was discussed in [1]:\n- A project user will retrieve segments from the project ranges.\n- When depleted, the segment IDs will be retrieved from the shared\n  range, never using another project segment ID.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-February/012736.html\n\nChange-Id: I953062d9ee8ee5ee9a9f07aff4a8222ac63ed525\nCloses-Bug: #1863423\n(cherry picked from commit 046672247de56bad950e8267a57bd26205f354a0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec26d4e2cf13cc99e06f126b55c0d49134053e29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec26d4e2cf13cc99e06f126b55c0d49134053e29"}]},"branch":"refs/heads/stable/train"},"72936ffcbae7f4d508f62b5fbad6ff1a1e06fe86":{"kind":"REWORK","_number":5,"created":"2020-03-13 12:01:04.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/08/712508/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/712508/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/712508/5"}}},"commit":{"parents":[{"commit":"6a1c6b6e58cee49ea47b6b1e61f82d24a2aa50e6","subject":"Merge \"Remove extra header fields in proxied metadata requests\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a1c6b6e58cee49ea47b6b1e61f82d24a2aa50e6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-02-26 10:39:19.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-03-13 11:58:57.000000000","tz":0},"subject":"Fix queries to retrieve allocations with network_segment_range","message":"Fix queries to retrieve allocations with network_segment_range\n\nFixed the queries to retrieve the segment ID allocations when service\nplugin network_segment_range is enabled. With the previous\nimplementation, a project user was able to allocate a segment ID\nbelonging to other project segment range.\n\nThe solution implemented was discussed in [1]:\n- A project user will retrieve segments from the project ranges.\n- When depleted, the segment IDs will be retrieved from the shared\n  range, never using another project segment ID.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-February/012736.html\n\nChange-Id: I953062d9ee8ee5ee9a9f07aff4a8222ac63ed525\nCloses-Bug: #1863423\n(cherry picked from commit 046672247de56bad950e8267a57bd26205f354a0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72936ffcbae7f4d508f62b5fbad6ff1a1e06fe86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72936ffcbae7f4d508f62b5fbad6ff1a1e06fe86"}]},"branch":"refs/heads/stable/train"},"bbe401aaf9bfdd77e1d43d547b2cdb436b1440c8":{"kind":"REWORK","_number":6,"created":"2020-03-16 16:19:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/08/712508/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/712508/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/712508/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/712508/6"}}},"commit":{"parents":[{"commit":"651eb12bec5e5e986bed79f3a5c006c617e79bda","subject":"Improve VLAN allocations synchronization","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/651eb12bec5e5e986bed79f3a5c006c617e79bda"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-02-26 10:39:19.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-03-16 16:16:30.000000000","tz":0},"subject":"Fix queries to retrieve allocations with network_segment_range","message":"Fix queries to retrieve allocations with network_segment_range\n\nFixed the queries to retrieve the segment ID allocations when service\nplugin network_segment_range is enabled. With the previous\nimplementation, a project user was able to allocate a segment ID\nbelonging to other project segment range.\n\nThe solution implemented was discussed in [1]:\n- A project user will retrieve segments from the project ranges.\n- When depleted, the segment IDs will be retrieved from the shared\n  range, never using another project segment ID.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-February/012736.html\n\nConflicts:\n    neutron/objects/network_segment_range.py\n    neutron/objects/plugins/ml2/base.py\n    neutron/objects/plugins/ml2/vlanallocation.py\n    neutron/objects/plugins/ml2/vxlanallocation.py\n\nChange-Id: I953062d9ee8ee5ee9a9f07aff4a8222ac63ed525\nCloses-Bug: #1863423\n(cherry picked from commit 046672247de56bad950e8267a57bd26205f354a0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bbe401aaf9bfdd77e1d43d547b2cdb436b1440c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bbe401aaf9bfdd77e1d43d547b2cdb436b1440c8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
