)]}'
{"id":"openstack%2Fneutron~712640","triplet_id":"openstack%2Fneutron~master~I9cc05e9e89b806263e9cba31ed9e16b1d2b455f6","project":"openstack/neutron","branch":"master","topic":"bug/1732067","hashtags":[],"change_id":"I9cc05e9e89b806263e9cba31ed9e16b1d2b455f6","subject":"Enable egress direct flow for iptables_hybrid","status":"ABANDONED","created":"2020-03-12 10:22:23.000000000","updated":"2020-10-19 02:04:23.000000000","total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4939a4a04d183b66aa8018a89a15bd1e0c1a3138","_number":712640,"virtual_id_number":712640,"owner":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-08 00:01:12.000000000","_account_id":21907,"name":"Arjun Baindur","email":"xagent@gmail.com","username":"arjunbaindur"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":1,"date":"2020-04-08 03:17:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"date":"2020-10-09 01:13:56.000000000","_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},{"date":"2020-09-25 15:40:40.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"date":"2020-10-02 17:35:29.000000000","_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-03 00:35:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"date":"2020-09-27 01:35:49.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"date":"2020-06-09 03:14:12.000000000","_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21907,"name":"Arjun Baindur","email":"xagent@gmail.com","username":"arjunbaindur"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-10-15 15:05:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":21907,"name":"Arjun Baindur","email":"xagent@gmail.com","username":"arjunbaindur"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":21907,"name":"Arjun Baindur","email":"xagent@gmail.com","username":"arjunbaindur"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":21907,"name":"Arjun Baindur","email":"xagent@gmail.com","username":"arjunbaindur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-12 10:22:40.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2020-04-02 08:26:21.000000000","updated_by":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"reviewer":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"state":"REVIEWER"},{"updated":"2020-04-03 16:46:42.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-08 03:17:55.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-09 03:14:12.000000000","updated_by":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"reviewer":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"state":"REVIEWER"},{"updated":"2020-09-25 15:40:40.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-09-27 01:35:49.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-10-02 17:35:29.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2020-10-03 00:35:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-08 00:01:12.000000000","updated_by":{"_account_id":21907,"name":"Arjun Baindur","email":"xagent@gmail.com","username":"arjunbaindur"},"reviewer":{"_account_id":21907,"name":"Arjun Baindur","email":"xagent@gmail.com","username":"arjunbaindur"},"state":"REVIEWER"},{"updated":"2020-10-15 15:05:55.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"}],"messages":[{"id":"ccac4583877a2c6b353982a1612a5c9d25013a8c","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-03-12 10:22:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab6ce9c6fa6e895246ce50c7abcd88588e097daf","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2020-03-12 10:22:40.000000000","message":"Patch Set 1:\n\nCongratulations, you\u0027ve proposed your first change in OpenDev.\n\nYour submission will now be tested automatically by Zuul, our gatekeeper,\nand reviewed by other friendly developers. They will give you feedback and\nmay require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches often take days (and sometimes weeks) to get reviewed, so be\npatient. Don\u0027t hesitate to ask for help, and answer questions about your\nwork promptly if you can. The more you get to know reviewers and get to be\nknown by them, the smoother the review and approval process will become. The\nfastest way to accomplish this is by reviewing other proposed changes\nyourself: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for participating in OpenDev, we look forward to seeing you\naround.\n\nWorkflow Guide: https://docs.openstack.org/infra/manual/developers.html","accounts_in_message":[],"_revision_number":1},{"id":"97f6f889aaeb57ec4be7ea3de323f8a9c76f502e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-12 11:09:40.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712640/1 : FAILURE in 46m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5f030b76663ff5e95fb573313e344449854a74ab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-12 11:10:58.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/40/712640/1/SRIOV-neutron-macvtap : ABORTED\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/40/712640/1/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"efcf56cd25d118ff253c6f6da1dd632a55991515","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-12 12:58:23.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/40/712640/1/SRIOV-neutron-macvtap : SUCCESS\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/40/712640/1/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"16186cbafe0c9ed010948d0c59ac7a7b0f6d72dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 13:13:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2cd92aef16d94e1b85d4cb5766e67f11 : SUCCESS in 1h 01m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e3a7aafeeb4646e2bb145680419291b8 : SUCCESS in 36m 54s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/25064c6253704d45910185a07bd0ea3c : FAILURE in 1h 35m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a56a7b880b5d44e4986c316bf1cd83b0 : SUCCESS in 1h 13m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2fe7a8b39164432c926af8623489b955 : SUCCESS in 1h 03m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4e1dd5ed59794e4587e11f2f761e6ef6 : SUCCESS in 1h 11m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a04bf1c4aeb4c5590eef227343b9c55 : SUCCESS in 36m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3e9ba9ff39f44519893da0ccc32e530 : SUCCESS in 25m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69dea1a91fca4407b4d2f33bf8d4b81b : FAILURE in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c433ffa7c0fc4c0994f6d0ceb133d66e : SUCCESS in 28m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/461a15e5025944e7a43ec020fab87d75 : SUCCESS in 34m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d4ac19dab63445580addcc30fd9c462 : SUCCESS in 29m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2e6db47aa6f4beda5796d70376ac915 : SUCCESS in 10m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/17dbbf8a515e45e9ac0724f1d595427b : FAILURE in 48m 13s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8a4958e87f5742a9b849209a79df71b1 : SUCCESS in 1h 04m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8a4cfe7281bd4726be5f3344c80c82fd : SUCCESS in 2h 12m 44s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/40939c56398a426aa04c9dc3c39d8602 : SUCCESS in 1h 36m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d25cd3775cc943c393f086bc72c33735 : SUCCESS in 1h 11m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fbc318ea1ab74e1381229a320aa22f20 : SUCCESS in 1h 16m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8ba8e9e9ac7146c3a1e894f4c34dac9a : SUCCESS in 1h 08m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9b124a4f3f794b98b0ab62d289c5a253 : SUCCESS in 1h 39m 45s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/763a99663b8c4e238ce88d7f584bab76 : SUCCESS in 1h 41m 43s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/119da85022524c92894d3daca7255741 : SUCCESS in 1h 46m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ae4692e5122e419190127980d35c1f98 : SUCCESS in 1h 08m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/00ad807dd85840738922a988de8d65ef : SUCCESS in 2h 19m 24s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fac3f6e8aab3450c980fac4f815e8221 : SUCCESS in 2h 28m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6f87c3d2ae6c409cb6def5fee0d20bac : SUCCESS in 2h 17m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f126b82d3c474a78ad1845474debbfb1 : SUCCESS in 1h 14m 57s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/dc9cb013d150482782b95f3c9f6a518d : SUCCESS in 1h 31m 53s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6bf98cf45ad7479da66c7ffd8bf18c5c : SUCCESS in 57m 59s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c59ea0f21c441789c2679b93197c577 : SUCCESS in 1h 15m 18s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/779afb32ff3f465faebc20e59df61db3 : SUCCESS in 1h 33m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bf0ab34b113340c59bad7384b03899f5 : SUCCESS in 2h 12m 38s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d18386e5da354c809ae8d35e7de225d9 : SUCCESS in 1h 52m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2bc709b4d8f64d5ff0d8e3ebdb438950d665972","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-15 15:24:53.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/40/712640/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"cced47e414a9c9b1c3966cc6c88df2dcd7d5c270","author":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"date":"2020-03-30 10:39:54.000000000","message":"Patch Set 1: Code-Review-1\n\nThis patch fixes the problem when the subnet is attached to a dvr router. But there is flooding if it is not.\n\nI think the solution in https://review.opendev.org/#/c/715419 is better than the \"accepted egress direct flow\" way.","accounts_in_message":[],"_revision_number":1},{"id":"7c6d6979cda395e6638e9b37a815af6bdd5ad143","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-01 01:11:09.000000000","message":"Patch Set 1:\n\n\u003e This patch fixes the problem when the subnet is attached to a dvr\n \u003e router. But there is flooding if it is not.\n \u003e \n \u003e I think the solution in https://review.opendev.org/#/c/715419 is\n \u003e better than the \"accepted egress direct flow\" way.\n\nMy patch just makes sure Yulong\u0027s fix patch can work for firewall_driver\u003diptables_hybrid, it isn\u0027t to fix MAC learning issue. You know Yulong\u0027s fix patch has been merged into multiple branches, so I think my patch is quick path to fix our current issue. I can confirm it did fix L2 and L3 flood, but MAC indeed can\u0027t be learnt.\n\nAbout Li Yajie\u0027s fix patch, it is to fix MAC learning issue, it is duplicate of Yulong\u0027s fix + my fix, if you guys think Yajie\u0027s solution is better, we need to revert Yulong\u0027s merged patch and apply Yajie\u0027s.\n\nMaybe you can help check why Yajie\u0027s patch has L3 performance drop, we don\u0027t know why so far.","accounts_in_message":[],"_revision_number":1},{"id":"a51593d34fd7deb201e52a784349b68627dc105f","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-04-01 06:08:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"369d22c17ce160d9ae8182ce49a13ab34ae3e123","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-04-01 06:42:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ea17782dd825f73c3a81a2df834e497c3719d1da","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-01 06:51:43.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n\u003e (3 comments)\n\nNo problem, I will update it per your comments. I\u0027m not familiar with pep8, how can I check pep8 style issue on my local machine.","accounts_in_message":[],"_revision_number":1},{"id":"469ec741bbe1c67c978f693604ee149991b8c284","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-04-01 08:11:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6fc8d56bfdeeb2a4503b329f13fabe1a2d391c2e","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-01 08:14:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"83ec8d027ab2ed41de22fe988a61c94c0329f500","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-01 08:16:51.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712640/3 : FAILURE in 1m 27s","accounts_in_message":[],"_revision_number":3},{"id":"94eb0ad56a56eebe273423353f9832b7c8abd1c2","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-01 08:19:45.000000000","message":"Patch Set 2:\n\n\u003e Uploaded patch set 2.\n\nYulong, I don\u0027t know you\u0027re changing when I\u0027m changing it, sorry for conflict, I checked pep8 and uploaded update 3, it should be ok to you.","accounts_in_message":[],"_revision_number":2},{"id":"dbb524a5609ce40ed2b024b21aad20a25abff53b","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-01 08:41:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0dbdb5783fcc8536654dff0bf3308b2884f0aa53","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-01 08:42:47.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712640/3 : FAILURE in 38s","accounts_in_message":[],"_revision_number":3},{"id":"09db51ae170163fd8e5807ed044e7d41f870741f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-01 09:48:39.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/40/712640/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"c703c4bd5ba54fdf4fb12852ace7118e74f32efc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-01 09:58:02.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/40/712640/3/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/40/712640/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"4f7ef90e0d67b3ce694a6fcecf1d45b4a0f679ea","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-04-01 10:20:12.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"81a4c9c2e235ee69f7098f8559cf39d783a7330b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 10:36:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2a4b752129554950929b61531631d92c : SUCCESS in 39m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0bde58334dfd41d981c530e7490a9da2 : SUCCESS in 32m 16s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/4d1f8229941549de90f21489a86a722f : FAILURE in 2h 19m 58s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/99d87b787ce049529797cd04d837b60b : SUCCESS in 1h 10m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bbe737d551ba4d2895ae44f06ae416da : SUCCESS in 1h 08m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bb0bbd418ade4e6eaf6d41494820873d : SUCCESS in 1h 05m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db47a98d660e42b8a12844fc17cc9fe5 : SUCCESS in 30m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9cafb91756bc4c5da1a037fef69454e0 : SUCCESS in 31m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f0a57d8cd1c426c857706784581491a : SUCCESS in 15m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43e76677c6004db7b93a9b19f5089d0d : SUCCESS in 32m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3aeb7aeb31bb4e84b93b0f8fbc2c7687 : SUCCESS in 29m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dfa312d15874aeebe4f4e38cecae650 : SUCCESS in 25m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bf96ccd1c524b68affe739025f72e57 : SUCCESS in 10m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bf8be1cf62114460a0d40ae91c941ef3 : SUCCESS in 51m 33s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c01efc4207634b56b8d109f4ea6e949a : FAILURE in 1h 01m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ee62f61197f84b729f173fa0fed35d13 : FAILURE in 1h 51m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7c14e93311cd43ad8367c7ec3aae3e4d : SUCCESS in 1h 07m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/938ea8f9b46e4fc3811924be6599bc8e : SUCCESS in 1h 06m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/71639639a9574ee090caee4c7052797e : SUCCESS in 1h 13m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4e52e4bf0c3047b4ae3cae6b038f7827 : SUCCESS in 1h 16m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6c0abadcd0cb417fb5755266b3de1d24 : FAILURE in 1h 26m 21s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fd727af4eb9f466e81629b128eb5a585 : SUCCESS in 1h 32m 45s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/067d3aa708a340b29f61c715eb15e708 : SUCCESS in 1h 09m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f94855d5a1e74c3fa3e693758eaa85a1 : SUCCESS in 52m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2e63b0863cbf4fd6964f52493fa7836e : SUCCESS in 2h 08m 41s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/65902dc7fdee44e2818debc43c26a893 : SUCCESS in 2h 01m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/37e44d085bd242be938dcaa95d82d343 : SUCCESS in 1h 38m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d884eb5e016145d3b48ee42b2693ae93 : SUCCESS in 1h 08m 11s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/16f4e88fba674d57a2f970c0c398457c : SUCCESS in 1h 19m 10s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bedcad9f48a54c1e9a5390555d52a353 : FAILURE in 52m 58s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/58e65c68b37045d19199f129d0db1d26 : SUCCESS in 1h 16m 55s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b8048b6d81a04ac390a146525bc45061 : SUCCESS in 1h 16m 36s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4eefa06238cf4d80b346e3b93f1f7bd6 : SUCCESS in 1h 39m 57s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/31c6806939584dfca95b05070c282ebf : SUCCESS in 1h 59m 52s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/50cb9f60b7e24792ae17646e1d86cd65 : SUCCESS in 2h 12m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"454550041bb0e592763a6e9a5b04af263234615d","author":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"date":"2020-04-01 10:55:58.000000000","message":"Patch Set 3: Code-Review-1\n\nWhen the subnet is **not** attached to a dvr router there is flooding. Bridge br-int does not learn the VM macs from the egress packets - as intended in this patch. But packets from patch-tun hit the NORMAL action. These packets are flooded because the VM macs were not learned.","accounts_in_message":[],"_revision_number":3},{"id":"a014143f950b83bbca9952e87870ace3905d7d9d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-01 11:10:44.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/40/712640/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"8a2b21640ef8c146bd7b676a0ef6fd67ca4807d5","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-04-01 12:18:57.000000000","message":"Patch Set 3:\n\nHi guys, I will bring up this issue to the L3 meeting [1] today, please attend if possible.\n[1] http://eavesdrop.openstack.org/#Neutron_L3_Sub-team_Meeting","accounts_in_message":[],"_revision_number":3},{"id":"af33ba5ac6a6dc97b3a80e6dd2aeb832c12fd6c5","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-04-01 13:23:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9c7ea779c8ffaa02255ea567efcfdbfa14519f8c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-01 13:25:05.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712640/4 : FAILURE in 40s","accounts_in_message":[],"_revision_number":4},{"id":"1835ffacf36aff5e93aff692ca5e14db5b642a93","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-01 14:44:07.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/40/712640/4/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/40/712640/4/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"00bf0f2befb945a4a05e09d5217f8af08da1d6cd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-01 15:08:58.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/40/712640/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"38360f67028c9bdecd4c4c5c15f47259034f57e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 16:37:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6e9387ef8d544befb8c1d8ab6d81cbf0 : SUCCESS in 1h 09m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/641a62688d5f444e9234a7c8b3ba4005 : SUCCESS in 48m 23s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/cb273c05e4f049f38e57d810a1c25562 : FAILURE in 1h 47m 49s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/31124a8b989c455eb45fdfac51541d1f : FAILURE in 1h 29m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/36278314081c44a89cfa6f845eda6af8 : SUCCESS in 1h 22m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2ff08c2db96b4365aae436c706568ca8 : SUCCESS in 1h 23m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/546c44fbe6f44cd89cc0b2c24b920743 : SUCCESS in 40m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a9533ab5cb7486eb9a582dd39b04133 : SUCCESS in 39m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66ed72a0939c4c8990be0761d1d46d8c : SUCCESS in 14m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eaf3ed088f5344c6adce5bc6bd2dd36e : SUCCESS in 29m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ceb8a1c4ea3043e28b7524eb71130097 : SUCCESS in 33m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8c1b2cbb3604a3eba3b57c20fa2d465 : SUCCESS in 27m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b82fb26d14a4b80b0f36496c8842f3a : SUCCESS in 11m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5bcd446e0662497eada43b00fa5cb0d9 : FAILURE in 1h 28m 23s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/72cc0a04749648b6aee6deb8d16cb7f8 : SUCCESS in 2h 10m 14s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/cd1df17dceab41f09fab55cf49466461 : FAILURE in 2h 57m 54s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1e6ccc43caff483d8d7cd751b542a26a : SUCCESS in 2h 15m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e0d68cd48444175b9d49184125936c2 : FAILURE in 1h 46m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5ed0a60a28b34669813a1ae6dc35c73b : SUCCESS in 2h 00m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a35e7042e1a04f5f95758c7c92283429 : SUCCESS in 1h 14m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d35ff9ae016d492c8489790f9264ba3a : SUCCESS in 2h 36m 41s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/37e8dde5decf439782734790300dfe50 : SUCCESS in 1h 33m 48s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e963b53de96241daabfb1b4b0eaabe67 : SUCCESS in 1h 17m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/454333df8c18429ab528767c35e0ea74 : SUCCESS in 55m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/68d60b73c92743d5b8dfe240202ffdb1 : SUCCESS in 2h 35m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/494f97c891be47adb93df484c07fac6c : SUCCESS in 1h 47m 33s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ef519d52affc4ae3812c849e3ca7538b : SUCCESS in 1h 47m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb8ab4f10cf64e01be1fa9ddc0d23925 : SUCCESS in 1h 04m 12s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/077f44f651de4612a491fa765d1228a4 : SUCCESS in 1h 20m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b401f3d33fa40ffbdaeaa5b4b2d25dc : SUCCESS in 54m 39s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/825f92a68a4f4b3b835448c37850e4e4 : SUCCESS in 1h 09m 14s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5b957e8567d4305b1571350792c0029 : SUCCESS in 1h 26m 00s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/aa7782c7c26546acaac269fa3039e8f1 : SUCCESS in 1h 48m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ff23ffd0c4c1434c95d76c5dc25c1d1b : SUCCESS in 1h 50m 00s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/5c2afd8206624751b368afeeee17aa26 : SUCCESS in 2h 33m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5cbab2f95fdf7820272995d517a4a69d68b361bd","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-02 03:26:45.000000000","message":"Patch Set 4:\n\nHi, guys, I need to figure out issues here.\n\nYajie is my colleague, his fix patch https://review.opendev.org/#/c/715419/ is actually duplicate of https://review.opendev.org/#/c/666991/ for our user scenario, so if it is applied, for firewall_driver\u003diptables_hybrid, only one fix should work, but there are duplicate flows there.\n\nI think we can base on https://review.opendev.org/#/c/666991/ and add some flows to handle corner cases, which will be better.\n\nYajie\u0027s fix https://review.opendev.org/#/c/715419/ is also drv only if you carefully read his changes, moreover we have found it have big performance issue for L3 VM communication (i.e. vm1 in subnet1 to vm2 in subnet2).\n\nI have verified https://review.opendev.org/#/c/666991/ + https://review.opendev.org/#/c/712640/ in our environment, it fixed all the issues we encountered (L2 flood and L3 flood, BTW, we don\u0027t have the case in which a subnet doesn\u0027t have a router).\n\nSo my idea is to submit a new commit to cover corner case Darragh O\u0027Reilly mentioned, we needn\u0027t duplicate code and duplicate flows.\n\nEven if L2 MAC learning  is fixed, for L3 case, we still need direct egress flows to steer the traffic to another VM in another subnet in another compute node because qrouter set static ARP entries, it never will send ARP request, so that VM (another VM in another subnet in another compute node) MAC never will be learnt unless that VM sends traffic to a VM in same subnet in this compute node, it is very likely there isn\u0027t a VM which has same subnet as that VM.","accounts_in_message":[],"_revision_number":4},{"id":"c2f27dd36b64eeb9d49f4013281b0c6e9721d459","author":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"date":"2020-04-02 08:26:21.000000000","message":"Patch Set 4:\n\nHi Yi Yang, okay good. Please add me as a reviewer to any other patchsets.\n\nAbout your last point:\n\u003e Even if L2 MAC learning  is fixed, for L3 case, we still need direct egress flows to steer the traffic to another VM in another subnet in another compute node...\n\nHmm - if I understand you correctly, that would be a fundamental change. Currently the L3 routing for east/west is done by local qrouter namespace on the sending side. \n\nvm1 --\u003e ovs --\u003e qrouter --\u003e ovs --\u003e tunnel --\u003e ovs --\u003e vm2\nvm2 --\u003e ovs --\u003e qrouter --\u003e ovs --\u003e tunnel --\u003e ovs --\u003e vm1\n \nOn the sending side, ovs processes the same packet twice. Each time it has different src and dst mac addresses. So a direct flow can\u0027t be used to forward packets from VM to tunnel.","accounts_in_message":[],"_revision_number":4},{"id":"5d45547acecc49bb5df19678823c23be19a50bfe","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-02 08:39:53.000000000","message":"Patch Set 4:\n\n\u003e Hi Yi Yang, okay good. Please add me as a reviewer to any other\n \u003e patchsets.\n \u003e \n \u003e About your last point:\n \u003e \u003e Even if L2 MAC learning  is fixed, for L3 case, we still need\n \u003e direct egress flows to steer the traffic to another VM in another\n \u003e subnet in another compute node...\n \u003e \n \u003e Hmm - if I understand you correctly, that would be a fundamental\n \u003e change. Currently the L3 routing for east/west is done by local\n \u003e qrouter namespace on the sending side.\n \u003e \n \u003e vm1 --\u003e ovs --\u003e qrouter --\u003e ovs --\u003e tunnel --\u003e ovs --\u003e vm2\n \u003e vm2 --\u003e ovs --\u003e qrouter --\u003e ovs --\u003e tunnel --\u003e ovs --\u003e vm1\n \u003e \n \u003e On the sending side, ovs processes the same packet twice. Each time\n \u003e it has different src and dst mac addresses. So a direct flow can\u0027t\n \u003e be used to forward packets from VM to tunnel.\n\nDarragh O\u0027Reilly， here direct flow means the flow which is in table 61 for iptables_hybrid (table 59 for Yajie\u0027s patch), it will avoid default normal action and change vlan id to another subnet VLAN id and output to patch-tun, here is an example for this.\n\ncookie\u003d0x3a40b98b2a93627f, duration\u003d43848.669s, table\u003d61, n_packets\u003d63600558, n_bytes\u003d559586562325, priority\u003d10,in_port\u003d1,dl_src\u003dfa:16:3e:c8:63:0b,dl_dst\u003d00:00:00:00:00:00/01:00:00:00:00:00 actions\u003dpush_vlan:0x8100,set_field:4097-\u003evlan_vid,output:37\ncookie\u003d0x3a40b98b2a93627f, duration\u003d43814.270s, table\u003d61, n_packets\u003d353147, n_bytes\u003d28560000, priority\u003d10,in_port\u003d5,dl_src\u003dfa:16:3e:4a:69:d0,dl_dst\u003d00:00:00:00:00:00/01:00:00:00:00:00 actions\u003dpush_vlan:0x8100,set_field:4097-\u003evlan_vid,output:37\ncookie\u003d0x3a40b98b2a93627f, duration\u003d43780.222s, table\u003d61, n_packets\u003d0, n_bytes\u003d0, priority\u003d10,in_port\u003d3,dl_src\u003dfa:16:3e:f6:7b:01,dl_dst\u003d00:00:00:00:00:00/01:00:00:00:00:00 actions\u003dpush_vlan:0x8100,set_field:4097-\u003evlan_vid,output:37\ncookie\u003d0x3a40b98b2a93627f, duration\u003d43778.024s, table\u003d61, n_packets\u003d0, n_bytes\u003d0, priority\u003d10,in_port\u003d4,dl_src\u003dfa:16:3e:02:17:e8,dl_dst\u003d00:00:00:00:00:00/01:00:00:00:00:00 actions\u003dpush_vlan:0x8100,set_field:4097-\u003evlan_vid,output:37\ncookie\u003d0x3a40b98b2a93627f, duration\u003d43474.241s, table\u003d61, n_packets\u003d350393, n_bytes\u003d28340052, priority\u003d10,in_port\u003d6,dl_src\u003dfa:16:3e:5b:0e:f0,dl_dst\u003d00:00:00:00:00:00/01:00:00:00:00:00 actions\u003dpush_vlan:0x8100,set_field:4097-\u003evlan_vid,output:37\ncookie\u003d0x3a40b98b2a93627f, duration\u003d43448.240s, table\u003d61, n_packets\u003d350148, n_bytes\u003d28320372, priority\u003d10,in_port\u003d7,dl_src\u003dfa:16:3e:4d:62:dd,dl_dst\u003d00:00:00:00:00:00/01:00:00:00:00:00 actions\u003dpush_vlan:0x8100,set_field:4097-\u003evlan_vid,output:37\n\n\nFor Yajie\u0027s patch, it is\n\ntable\u003d59, n_packets\u003d3, n_bytes\u003d294, priority\u003d19,dl_src\u003dfa:16:3e:6f:7b:39 actions\u003dmod_vlan_vid:3,output:\"patch-tun\"\n\nSo you can see they are duplicate egress flows, my direct flows mean these flows.","accounts_in_message":[],"_revision_number":4},{"id":"ff92e451ecac20cd8f6371f4b3932aa1b04236c3","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-04-02 09:14:44.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"0d2bdbb4b693c948e0d898c9c4c8e18d7a50aca3","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-04-02 09:24:24.000000000","message":"Patch Set 4:\n\n\u003e Hi, guys, I need to figure out issues here.\n \u003e \n \u003e Yajie is my colleague, his fix patch https://review.opendev.org/#/c/715419/\n \u003e is actually duplicate of https://review.opendev.org/#/c/666991/ for\n \u003e our user scenario, so if it is applied, for firewall_driver\u003diptables_hybrid,\n \u003e only one fix should work, but there are duplicate flows there.\n \u003e \n\nThis one is the right approach, we do not need duplicated fixes.\n\n \u003e I think we can base on https://review.opendev.org/#/c/666991/ and\n \u003e add some flows to handle corner cases, which will be better.\n\nWhat\u0027s the corner cases?\n\n \u003e \n \u003e Yajie\u0027s fix https://review.opendev.org/#/c/715419/ is also drv only\n \u003e if you carefully read his changes, moreover we have found it have\n \u003e big performance issue for L3 VM communication (i.e. vm1 in subnet1\n \u003e to vm2 in subnet2).\n \u003e \n \u003e I have verified https://review.opendev.org/#/c/666991/ +\n \u003e https://review.opendev.org/#/c/712640/ in our environment, it fixed\n \u003e all the issues we encountered (L2 flood and L3 flood, BTW, we don\u0027t\n \u003e have the case in which a subnet doesn\u0027t have a router).\n \u003e \n\nPlease elaborate the issues\u0027 detail, I\u0027m confusing now. What\u0027s L2 flooding? What L3 flooding? How it happen? Why related to DVR?\n\n \u003e So my idea is to submit a new commit to cover corner case Darragh\n \u003e O\u0027Reilly mentioned, we needn\u0027t duplicate code and duplicate flows.\n \u003e \n \u003e Even if L2 MAC learning  is fixed, for L3 case, we still need\n \u003e direct egress flows to steer the traffic to another VM in another\n \u003e subnet in another compute node because qrouter set static ARP\n \u003e entries, it never will send ARP request, so that VM (another VM in\n \u003e another subnet in another compute node) MAC never will be learnt\n \u003e unless that VM sends traffic to a VM in same subnet in this compute\n \u003e node, it is very likely there isn\u0027t a VM which has same subnet as\n \u003e that VM.\n\n\nI just want to say I\u0027m lost now. In this comment we can see flooding issue, performance, arp, firewall_driver, cases and cases. We should handle the problem one by one in small pieces. So it\u0027s better to file new bugs for these different cases.","accounts_in_message":[],"_revision_number":4},{"id":"9a7d51ea24059e4ca7d51278031d3b8496ce9fe3","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-04-02 09:36:18.000000000","message":"Patch Set 4:\n\nPlease allow me to make things clear here. In the L3 meeting, we have a conclusion. This patch will be used for iptables_hybrid driver to enable egress direct flows, aka make \"explicitly_egress_direct \u003d True\" work. It will fix the \"egress\" (in the perspective of VM) flooding issue on br-int when iptables_hybrid securitg group driver applied.\n\nAnother is found by Darragh O\u0027Reilly, if subnet has no router connected, it will case the a new \"ingress\" flooding. This will be addressed in another bug or patch.","accounts_in_message":[],"_revision_number":4},{"id":"822da0f42742e3b5c2ceea58baf05e2c2a143f2f","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-02 09:43:36.000000000","message":"Patch Set 4:\n\n\u003e \u003e Hi, guys, I need to figure out issues here.\n \u003e \u003e\n \u003e \u003e Yajie is my colleague, his fix patch https://review.opendev.org/#/c/715419/\n \u003e \u003e is actually duplicate of https://review.opendev.org/#/c/666991/\n \u003e for\n \u003e \u003e our user scenario, so if it is applied, for firewall_driver\u003diptables_hybrid,\n \u003e \u003e only one fix should work, but there are duplicate flows there.\n \u003e \u003e\n \u003e \n \u003e This one is the right approach, we do not need duplicated fixes.\n \u003e \n \u003e \u003e I think we can base on https://review.opendev.org/#/c/666991/ and\n \u003e \u003e add some flows to handle corner cases, which will be better.\n \u003e \n \u003e What\u0027s the corner cases?\n\nAs Darragh O\u0027Reilly said, if a subnet isn\u0027t added into router, your fix won\u0027t install flows to handle this case. Your fix has a side effect, it blocked MAC learning. If ovs bridge can\u0027t find destination MAC in fdb table, it has to do flood unless there are special flows to steer it to patch-tun.\n\n\n \u003e \n \u003e \u003e\n \u003e \u003e Yajie\u0027s fix https://review.opendev.org/#/c/715419/ is also drv\n \u003e only\n \u003e \u003e if you carefully read his changes, moreover we have found it have\n \u003e \u003e big performance issue for L3 VM communication (i.e. vm1 in\n \u003e subnet1\n \u003e \u003e to vm2 in subnet2).\n \u003e \u003e\n \u003e \u003e I have verified https://review.opendev.org/#/c/666991/ +\n \u003e \u003e https://review.opendev.org/#/c/712640/ in our environment, it\n \u003e fixed\n \u003e \u003e all the issues we encountered (L2 flood and L3 flood, BTW, we\n \u003e don\u0027t\n \u003e \u003e have the case in which a subnet doesn\u0027t have a router).\n \u003e \u003e\n \u003e \n \u003e Please elaborate the issues\u0027 detail, I\u0027m confusing now. What\u0027s L2\n \u003e flooding? What L3 flooding? How it happen? Why related to DVR?\n \u003e \n\nDVR blocked MAC learning, this is root cause, in your environment, you can check it by \"ovs-appctl fdb/show br-int\", in non-DVR mode or in the case a subnet isn\u0027t added into router, it should show all learnt MACs from local VMs and remote VMs.\n\nL2 flood means ovs bridge will flood the traffic to all the local VMs in same subnet, your egress direct flow just fixes this issue.\n\nL3 flood means different subnet VM traffic, it also will flood to local VMs in same subnet as destination VM is in unless you installed egress direct flow to handle it.\n\n \u003e \u003e So my idea is to submit a new commit to cover corner case Darragh\n \u003e \u003e O\u0027Reilly mentioned, we needn\u0027t duplicate code and duplicate\n \u003e flows.\n \u003e \u003e\n \u003e \u003e Even if L2 MAC learning  is fixed, for L3 case, we still need\n \u003e \u003e direct egress flows to steer the traffic to another VM in another\n \u003e \u003e subnet in another compute node because qrouter set static ARP\n \u003e \u003e entries, it never will send ARP request, so that VM (another VM\n \u003e in\n \u003e \u003e another subnet in another compute node) MAC never will be learnt\n \u003e \u003e unless that VM sends traffic to a VM in same subnet in this\n \u003e compute\n \u003e \u003e node, it is very likely there isn\u0027t a VM which has same subnet as\n \u003e \u003e that VM.\n \u003e \n \u003e \n \u003e I just want to say I\u0027m lost now. In this comment we can see\n \u003e flooding issue, performance, arp, firewall_driver, cases and cases.\n \u003e We should handle the problem one by one in small pieces. So it\u0027s\n \u003e better to file new bugs for these different cases.\n\nThe root cause is DVR blocked MAC learning, no MAC in ovs bridge fdb table, so ovs bridge will flood it to many of ports, this is the flood,the flood the same subnet VM results in is L2 flood, the flood different subnet VM results in is L3 flood.","accounts_in_message":[],"_revision_number":4},{"id":"e1834907d95ba27f4bfcb1e0da5f27101c54eb12","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-02 09:46:10.000000000","message":"Patch Set 4:\n\n\u003e Please allow me to make things clear here. In the L3 meeting, we\n \u003e have a conclusion. This patch will be used for iptables_hybrid\n \u003e driver to enable egress direct flows, aka make \"explicitly_egress_direct\n \u003e \u003d True\" work. It will fix the \"egress\" (in the perspective of VM)\n \u003e flooding issue on br-int when iptables_hybrid securitg group driver\n \u003e applied.\n \u003e \n \u003e Another is found by Darragh O\u0027Reilly, if subnet has no router\n \u003e connected, it will case the a new \"ingress\" flooding. This will be\n \u003e addressed in another bug or patch.\n\nYes, this is the conclusion we all agreed.","accounts_in_message":[],"_revision_number":4},{"id":"2e7707cca1e2c79d891355c19f77241760ea5c65","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-02 09:58:41.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712640/4 : FAILURE in 43m 24s","accounts_in_message":[],"_revision_number":4},{"id":"ae37b04ff2596c4200a59f6cdbd5d6501a3ac4e4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-02 10:07:22.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/40/712640/4/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/40/712640/4/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"00145495df35e2bbc1f58e77e3322b721c408086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 11:46:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9533c5e2677a47c790907c53cf9bfbfa : SUCCESS in 57m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dd06c9eb084c46ec920ead987f07d9e8 : SUCCESS in 34m 19s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/f2248e66ad494593b09c9088079c8324 : FAILURE in 2h 06m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/23fd76f700cf4ff18c91372e82deab02 : SUCCESS in 1h 10m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/81e7b93dc69340009afe50207cc24a60 : SUCCESS in 1h 20m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c8f3c33173914e4ebd898276818b622b : SUCCESS in 1h 17m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b5ab96da83f43bdb89f35540e0a3944 : SUCCESS in 34m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c8b79fcd648439e9aab3628a8396144 : SUCCESS in 25m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38f588e2af3e4c31aaa1b2d0ca756536 : SUCCESS in 13m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3935f8a1fe29415fb9219fac0de2e87f : SUCCESS in 24m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a99d148ca37247d7870aaa733af81bea : SUCCESS in 34m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/737e946e06ca447d977acdcad02cea7f : SUCCESS in 30m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9979e3952a0e45b9b9520ce61d42af1d : SUCCESS in 9m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3f96fa8b23214741b9dd44d960a8529f : SUCCESS in 1h 34m 49s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/00e903bba35540f48d55b2905bbeca7e : SUCCESS in 1h 16m 46s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ddc50016a3924dffb0042e1e0bf5abd9 : SUCCESS in 1h 52m 03s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f0e2e56df5464a499d1d0df3e9b9dfa7 : SUCCESS in 1h 15m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ce9c0b41e05470898f6f5f098b1f340 : SUCCESS in 1h 06m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5df3f786916a4f228ecf7e703e789278 : SUCCESS in 1h 16m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dd3b83e6d23a41ceb447a4b4d5acebee : SUCCESS in 1h 14m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7c0778ac2ea341a1a343ce65f3246a45 : SUCCESS in 1h 45m 02s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eb128fabd57340c8ba23d8d0cfd1ab23 : SUCCESS in 1h 31m 42s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d275819b85f74d4fa3169e5476f9fabc : SUCCESS in 1h 15m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ec07167816b946bd90d3821ec983924d : SUCCESS in 53m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ed2d92dc5fdd480e93610226c5174b5e : SUCCESS in 2h 23m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4b77256c9742479d84243604ab22e695 : SUCCESS in 2h 03m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/657e0d20eca04177ae31c0ce2c5ff73b : SUCCESS in 1h 59m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd1ad2e4dd0d4c78a92a81e574048e9b : SUCCESS in 1h 09m 52s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6802e6381e7e46a0993324222f01130d : SUCCESS in 1h 16m 39s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b1280b113a3b46f880d58b94702ced8d : SUCCESS in 1h 06m 08s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/354e6b86ab9d4c8781f1841f20343241 : SUCCESS in 1h 07m 58s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5ed01ade6f8413298d32d0844b971eb : SUCCESS in 1h 25m 09s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/11d3abc5b0294514ac0ed9eb4ea50525 : SUCCESS in 1h 39m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/54687fa36cb149208a8c506d1ccda110 : SUCCESS in 2h 10m 22s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/90f4bc67913e4b37b40fed80cdbf4323 : SUCCESS in 2h 17m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2e293fb10bef88777aff66e1e53246014706aa05","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-03 14:32:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7e8779c0c9c580d3a5dc42641baab7d7e33346e4","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-03 14:34:26.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"7cfea3c29348dbf328b8bc8e868c027a0f88e7a0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-03 14:35:48.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/40/712640/5/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"7cafe748c0a9aa3bf2fef5da7fb3af469797a222","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-03 15:16:55.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712640/6 : FAILURE in 41m 53s","accounts_in_message":[],"_revision_number":6},{"id":"bf67c4f07804bced0ac552387a5237237ff85b05","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-03 16:46:42.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/40/712640/6/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/40/712640/6/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"a85b67ead546e31d3661cad0ae1de4ea5bdae392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 19:05:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5ec28290d951426894d1b80036c116c2 : SUCCESS in 1h 06m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3d8d75dba2d94482b1cc18e6fd597624 : TIMED_OUT in 1h 05m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/2c414fc5b8b34a61ab1fd4d90a41d95f : FAILURE in 2h 01m 20s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e0fc37cc70d54857aed1dd63e501ddd2 : SUCCESS in 1h 52m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4891074582c24f98b3b20726dbabbbe3 : SUCCESS in 1h 12m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3b4a0cd42e3a405f873e43fef248c91d : SUCCESS in 53m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20d9a047d2a7496f9b6716f21c743424 : FAILURE in 33m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e82aa0b7dc384d8abe8bc15c8c2a5be5 : FAILURE in 31m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee0bf2eb6b694fcf947d8f42fe610823 : SUCCESS in 13m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39eb3e681a1c436d84073fced02d9578 : FAILURE in 28m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ae088b770584725aae13f1cd5569540 : FAILURE in 30m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/859790e857334f699acb2eef6b3899c5 : FAILURE in 32m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4742047457542e7aca1dfff3d89bc8e : SUCCESS in 14m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ba4844bbe7034490af252a6de10db111 : SUCCESS in 51m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f7ae27d2f64e4771ab111c4ce8d75b61 : SUCCESS in 1h 11m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bdbc9dd5102b492c8fd53a6916fc662b : SUCCESS in 1h 45m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4fe5a2358afc4f368d1c51df1b79530b : SUCCESS in 1h 10m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce76b1de9015482a9b186df120cdd740 : SUCCESS in 1h 04m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6d7f51b6ec774a418ed6ac6e907a09d7 : SUCCESS in 1h 16m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/223423c498c74a668c443fa383b99826 : SUCCESS in 1h 05m 01s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/29ccb3613e494a06a3788f80706d2fdb : SUCCESS in 1h 34m 27s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8d4517d1cc3b474598b32d2b007e0187 : SUCCESS in 2h 07m 15s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c119e4d97db445e29e951d82676fcd4d : SUCCESS in 1h 13m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/51582dd0fa24405eb40f3e8cef0abbc6 : SUCCESS in 54m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6fdcd112977e4a069e1ff1ae90a138f7 : SUCCESS in 2h 18m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/408021e3cb554301a005daa8840f3e3d : SUCCESS in 1h 40m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/32a0e8143e1847b2a17325d837699a8d : SUCCESS in 2h 25m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d73077001284476a14f19fbd3a6bbd2 : SUCCESS in 1h 03m 19s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b5371dcb5a81466d8369a267ad3f4c3b : SUCCESS in 1h 23m 11s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/06765222ef034278959fbc3936bfbcc9 : SUCCESS in 1h 01m 24s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a863feda54b242c2b56307d1e1881b7e : SUCCESS in 1h 18m 23s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5677be782c754337b84445886b899a66 : SUCCESS in 1h 33m 11s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9f9dfb183ca3429a946f5f249c074965 : SUCCESS in 1h 44m 00s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d1ef1ab3cabd4ad699c45b5648577dd5 : SUCCESS in 1h 59m 05s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8ced8673cd554bb8963713f595b44131 : SUCCESS in 2h 59m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5387801efc8b456db18220795b588212 : SUCCESS in 2h 09m 14s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/54e0b4f6aae949b3ae878f78a9c80713 : SUCCESS in 2h 22m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"17a914b278af7a2999b1ac7505d791d2bb8e71f3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-04 15:53:02.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/40/712640/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"0f7c4d3cf2616f00760b4a1137db66c1fe10d1d3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-05 10:01:01.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/40/712640/4/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/40/712640/4/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"9a060ae22da9043e30f7f09e901c24eae1466e4e","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-07 01:29:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3e47782a75c33bc4dcb68c11b5956135ce8ff701","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-07 02:12:44.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712640/7 : FAILURE in 42m 50s","accounts_in_message":[],"_revision_number":7},{"id":"eb0f38c7efc88df10aff8a7e178679a90ec8cd51","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-07 03:01:48.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/40/712640/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"1be4bd1753b26e8eb419683015ddd1571428bf5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 04:15:22.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ac8f59a6340c4923be9a4ab609c330b8 : SUCCESS in 53m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a83ae3649469448ea9702a99a86e4b38 : SUCCESS in 36m 38s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/8258220511e64c159de5f2ef8e47943d : FAILURE in 2h 05m 46s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fc6433cfa4ac4feeae9c93d239250ef3 : SUCCESS in 1h 08m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d82cf2cf5c904793b0033173b91bcec3 : SUCCESS in 50m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d6c75659d233487c8ca26e81bbadbc87 : SUCCESS in 1h 04m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2478f440e45b49238a6363e323a3e893 : FAILURE in 34m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c969ea1c28845d88fceaa8a6dc40300 : FAILURE in 30m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/658f73447b4a439fa56ad6fc99289a9f : SUCCESS in 11m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d7b9ac01b98460ca16257e2c93c3347 : FAILURE in 29m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/632c29bfd6064ebfb1fab2f0a83ebe07 : FAILURE in 31m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e591255121b4ed8982ebf5d3485a6f1 : FAILURE in 30m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7486fcd076c647808393f86084a9ce1c : SUCCESS in 9m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2bf6fb0344a547149812458fd5f0d0cd : SUCCESS in 53m 33s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/11e0eb3892fb4a3f931600fb442fccdc : SUCCESS in 1h 42m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/91af5982a5124d9a80962434471bcaa3 : SUCCESS in 1h 59m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/90fcfed61f8f4b59bc141bcdc965a743 : SUCCESS in 1h 13m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/28da226b90694f00b93427c36719c262 : SUCCESS in 1h 15m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fb3969b28c3f494d8d3f156292bfdaa1 : SUCCESS in 1h 20m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ff69045293d74b0cb923dbc962f78342 : SUCCESS in 1h 15m 06s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/347427b519c443259582d108a2d9d66a : SUCCESS in 1h 35m 22s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1011fda7dd824dc49cf36951a3121a4e : SUCCESS in 1h 51m 01s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8d91ae896f554b1296add993c6f96722 : SUCCESS in 1h 26m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4f39953ee1ba48d68f1a65cd8e83a567 : SUCCESS in 55m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/258eea6506024630aaa78ac09bbb8b81 : SUCCESS in 2h 22m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/30e5c6e1c20746a8b5f08754da226263 : SUCCESS in 1h 42m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2652e422f6a7449cab4d2da9e6f82d6f : SUCCESS in 1h 56m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b278763c090e49af957e82542117e59d : SUCCESS in 1h 01m 49s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fd87020c8d5b45299fb68db43d496b31 : SUCCESS in 1h 17m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/99cebd5c68e84cf594fcebf3733a4554 : SUCCESS in 52m 30s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e42d5165ed2c44d6960440374e9533ee : SUCCESS in 1h 13m 37s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d48854111574401a41d539136ffddcf : SUCCESS in 1h 23m 19s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/664fb41a7a234bdabd6f0e01b4d29cb5 : SUCCESS in 1h 36m 43s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4fcacbcad4b14c71971d19212d6aaa74 : SUCCESS in 1h 54m 31s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/757ad4c4da014e6a97c281a98bb945d3 : SUCCESS in 2h 40m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/27b27f104f584a44b369df6e8b5f75bf : SUCCESS in 1h 57m 13s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/9a4a74c5be8249e48d4839604614fc2f : SUCCESS in 2h 19m 15s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/policy.py\n  Comments left for invalid file neutron/agent/common/ovs_lib.py\n  Comments left for invalid file neutron/objects/base.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_daemon.py\n  Comments left for invalid file neutron/common/eventlet_utils.py\n  Comments left for invalid file neutron/tests/unit/agent/ovsdb/test_impl_idl.py\n  Comments left for invalid file neutron/tests/unit/test_auth.py","accounts_in_message":[],"_revision_number":7},{"id":"8a82517e56eedaa690da12a9943645dcb90727f8","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-07 06:42:39.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"525531a39eefab6d7e87bcf55b84441aa24280f1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-07 07:28:35.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712640/8 : FAILURE in 45m 24s","accounts_in_message":[],"_revision_number":8},{"id":"b30cad3c7151cf1e8041051852d80600a169d80e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-07 07:55:11.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/40/712640/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"7305deb2e1cd5a454685201127058670af1f726e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 09:45:37.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a6694195d1ba459e97249af25a4d744b : SUCCESS in 1h 02m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a722c9948b0b4f73b211b83df2d6033c : SUCCESS in 35m 15s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/caa604adc7d74548afc47fa9979ad7dd : SUCCESS in 1h 44m 38s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4989d66c16b34c349c847515ceca720e : SUCCESS in 1h 06m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7e4d5213c7e7468ca3be2241fc6b294a : SUCCESS in 1h 10m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/731e054af54b41b8bec2934100e94207 : SUCCESS in 1h 08m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2808e1bc018b44de99a1d12d5bc4a6a1 : SUCCESS in 37m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0cf089705684c2397696102b7d5ca56 : SUCCESS in 28m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de9bf71b7ed244ccbd3aa239a26dbcb9 : SUCCESS in 14m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5c0ae60567f467ba6a97f70c71650c1 : SUCCESS in 25m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6912d7e137d14b80ba3e892834909380 : SUCCESS in 29m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/939e672cecac478da61803efe1785ab4 : SUCCESS in 28m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ffca865f167434f9e4a9380d35a6811 : SUCCESS in 9m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c31a5652fce24dc394447ed9bb9b7166 : SUCCESS in 54m 16s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f460870644da4fa8a344caf903c8a715 : SUCCESS in 1h 14m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bc6fee26babd40e2b23feb0e86047522 : SUCCESS in 1h 59m 43s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/68c1e73c7d50490b92be9d580d41fcfb : SUCCESS in 1h 13m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/37cdeca8a2684478b388340e2c0caffe : SUCCESS in 1h 15m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c6077f0ca6d84144b21edd08e9ef55cf : SUCCESS in 1h 22m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/669d61f251aa456bbedaabf8d73a17c8 : SUCCESS in 1h 27m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9de8125a1c2b4c79b5351040b7b705fe : SUCCESS in 1h 57m 57s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/548d7b8b77e84f68bc77ef4e8d8f3e8c : SUCCESS in 1h 51m 21s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/41b604664674402cb5e4fed3f931a4a1 : SUCCESS in 1h 28m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8b0cfafa18a143328144419079880fba : SUCCESS in 1h 18m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/13a8a14ea755484e95894ed82dd475e4 : FAILURE in 2h 51m 38s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8999eaf3e7d24369ad736b6a68b450af : SUCCESS in 2h 08m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0579062ce07348c4b27b466a19b458fd : SUCCESS in 2h 05m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48c315b3946544cc9f2faa49b3ce06cd : SUCCESS in 1h 01m 27s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/57b0e68f734e477cbb16913aa3b6f73e : SUCCESS in 1h 20m 47s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cc3ed3ab8f3c437ea0302dc83f7d7e5a : SUCCESS in 50m 02s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/75ebf1bcd58f456d90d7424538183d22 : SUCCESS in 1h 42m 42s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d7fc9cc7bed041c7afdfbf32204e4b63 : SUCCESS in 1h 17m 46s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/44a57917cb0644b39dfa0816007d0601 : SUCCESS in 1h 41m 00s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/125be6069dee4193b347a7d79942aea9 : SUCCESS in 2h 04m 33s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/03f47240470c40c8bca0815757463772 : SUCCESS in 2h 44m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/937d5b2264824c2db0fabaa39d518044 : SUCCESS in 2h 00m 51s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/77cc42c244f2408ea76513e1c92dc6d0 : SUCCESS in 2h 12m 31s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/common/ovs_lib.py\n  Comments left for invalid file neutron/tests/unit/agent/ovsdb/test_impl_idl.py\n  Comments left for invalid file neutron/tests/unit/test_auth.py\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_daemon.py\n  Comments left for invalid file neutron/objects/base.py\n  Comments left for invalid file neutron/common/eventlet_utils.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file neutron/policy.py","accounts_in_message":[],"_revision_number":8},{"id":"97e90c98966dd2e36826349d718f5a2bc0a4c243","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-04-07 13:14:34.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"10499b75c433aa81270576f35579a038055ad3d7","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-08 01:44:17.000000000","message":"Patch Set 8:\n\n\u003e (1 comment)\nhttps://github.com/openstack/neutron/blob/master/neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py#L2135\n\n    def process_install_ports_egress_flows(self, ports):\n        if not self.conf.AGENT.explicitly_egress_direct:\n            return\n\n        if (isinstance(self.sg_agent.firewall,\n                       agent_firewall.NoopFirewallDriver) or\n                not agent_sg_rpc.is_firewall_enabled()):\n            with self.int_br.deferred(full_ordered\u003dTrue,\n^^^                                      use_bundle\u003dTrue) as int_br:\n                for port in ports:\n                    try:\n                        self.install_accepted_egress_direct_flow(port, int_br)\n                        # give other coroutines a chance to run\n                        eventlet.sleep(0)\n                    except Exception as err:\n                        LOG.debug(\"Failed to install accepted egress flows \"\n                                  \"for port %s, error: %s\",\n                                  port[\u0027port_id\u0027], err)\n\nuse_bundle\u003dtrue means it needs OPENFLOW14, for the case firewall_driver\u003dopenvswitch, it has added OPENFLOW14 to br-int by default, so you didn\u0027t notice that. These flows can\u0027t be installed to br-int if br-int doesn\u0027t have OPENFLOW14 support.","accounts_in_message":[],"_revision_number":8},{"id":"a5bd473cc3da3ac6c9493128f7c491d4cc25bb74","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-08 01:46:02.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"0274dc0e58d35d774d5ecf07bb9b2d47c30fb4fa","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-08 01:47:04.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712640/9 : FAILURE in 28s","accounts_in_message":[],"_revision_number":9},{"id":"fa01bb99bd2f71ff1253dddd4d9d9cfc08a35544","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-08 03:17:55.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/40/712640/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"31f1b689ff5d1d206fe3abfd2638800bbb61bc31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 04:54:32.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b6b386f8bb39484b80781499af84ccbf : SUCCESS in 51m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/82836290db5c4af9b790eae0d5bfe586 : SUCCESS in 36m 04s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/fd295ab177904a6b836112480b69c805 : FAILURE in 1h 57m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9f263b4283d14cabba6aa81ae1bec604 : SUCCESS in 1h 29m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8ed7dd7d9a2a4ec49f52db0f89058867 : SUCCESS in 1h 18m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/648d8fab2e594bdfb6f223ace7809038 : SUCCESS in 1h 19m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3f589d548d94469af3f7a0dd1288378 : SUCCESS in 36m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b273843a6a124ca990bf1cd8ca4dc1bd : SUCCESS in 28m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bebc7de1b3640b7a7a1f2933ec8f0e9 : SUCCESS in 13m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdfccc7a73b647b896dbfdc580c36a20 : SUCCESS in 30m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a31b3b9524c546a7844c326a9978a29f : SUCCESS in 27m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79cd3cc89fdb41aa9b591cf7343330c5 : SUCCESS in 18m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/383fdb4cf4e94f40b741738907543bf8 : SUCCESS in 10m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4dc6780ce25649d6965757a9878f00dc : SUCCESS in 51m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/bf327f2b7bb3484e8771f3c1deb19877 : SUCCESS in 1h 15m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/df25b54464174ea8b6093250eaa57676 : FAILURE in 1h 52m 01s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/fc3558568d17432cb5a842401b0ab2b1 : SUCCESS in 1h 17m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e1a22080cf094bb3a10545dfd3473e2d : SUCCESS in 1h 17m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/262f95f1a9bf4aee85dfcb266823adb7 : SUCCESS in 1h 14m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d0105cb6b1b74d1581dd6c4c50207b82 : SUCCESS in 1h 17m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4343f2e657394ce9a71aaf52a9f338ea : SUCCESS in 2h 09m 17s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/63eaf24226604236b777512837b6d89d : SUCCESS in 1h 35m 32s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0effe6b8b5af4df6843859a54e2bfd8e : SUCCESS in 2h 02m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/728e57dfaad04d0e84eac4a54b8821ed : TIMED_OUT in 1h 34m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/73a024a52bd746f2be5b4e073c20f65b : TIMED_OUT in 3h 04m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d404ea932bf24b36b3817a394cc42c64 : SUCCESS in 2h 58m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f400774ce8f24116bcee262428f46d4a : TIMED_OUT in 3h 05m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68145429973f49d88f73d5a7826b4079 : SUCCESS in 1h 00m 49s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3d4885b6f2e0473599fc3990f57d362c : SUCCESS in 2h 13m 44s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/513942080a9947fbac645768e8df24d3 : FAILURE in 53m 57s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8c9271ff4e0c489b9345d36315e85ddc : SUCCESS in 1h 19m 54s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/644f4ebab91c45b1b85f1d07de6f2da4 : SUCCESS in 2h 14m 53s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d9ad8522c3e745508037b09300d365e7 : FAILURE in 23m 59s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/443fe614d9174653be0fda2843772867 : TIMED_OUT in 2h 14m 06s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/3c63fc34134a4738b20a8f3f515064ca : FAILURE in 29m 54s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/79c92077cc034ca9aeaf292534495624 : SUCCESS in 2h 04m 00s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6bc49f88e2dd4fa686e51d4488927e65 : SUCCESS in 2h 24m 57s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/policy.py\n  Comments left for invalid file neutron/agent/common/ovs_lib.py\n  Comments left for invalid file neutron/objects/base.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_daemon.py\n  Comments left for invalid file neutron/common/eventlet_utils.py\n  Comments left for invalid file neutron/tests/unit/agent/ovsdb/test_impl_idl.py\n  Comments left for invalid file neutron/tests/unit/test_auth.py","accounts_in_message":[],"_revision_number":9},{"id":"1db7ba3ac6f854ddc1e2fba39650627dbb99b274","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-08 12:48:55.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"2c39aab44e921291430ac2bf27075c720b05d6cf","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-08 14:14:47.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712640/9 : FAILURE in 53m 42s","accounts_in_message":[],"_revision_number":9},{"id":"c48cc318db1e8fdd2584104ba59d1cd053cd72cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 21:46:31.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/62107b6292ab4bfeac30729edca342d0 : POST_FAILURE in 55m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/860af7214fb04ac59b55226385e8793f : SUCCESS in 35m 14s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/35da07e82cb54c0bbb438a7ecaddc720 : SUCCESS in 1h 31m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b47eadd51cb64964900996cd89e9f61a : SUCCESS in 1h 10m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/95d27afa6994405382468832ab27df95 : SUCCESS in 1h 40m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/98b18399f73c475ca6b4d142a7d3369c : SUCCESS in 1h 01m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bac4ac6cd2c34fc9b3792cdb246f5093 : SUCCESS in 39m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40ef817a1a1c44e89af829c54e593144 : SUCCESS in 26m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c61a915883d48bd8546a920e57f957d : SUCCESS in 14m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58b619e1e8b448a1904b063d3ed051be : SUCCESS in 28m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d314fdc6bba4413a56e50370332917e : SUCCESS in 32m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1a6bb7c3e314b4ba7394351641356f0 : SUCCESS in 26m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da57478d65554922987db0aa02fd16a0 : SUCCESS in 12m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/55b66730ec004f7e8ff2dd54c274de51 : FAILURE in 1h 22m 13s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/bdda171bcc5e4c64ac87fedaee43f855 : SUCCESS in 1h 16m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c5f17f0c502343a0a3e084605d104500 : SUCCESS in 1h 51m 30s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7bf1ddc021014fec97a3ef6bc8978036 : SUCCESS in 1h 07m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1433bcfa6124449c9780eb73ce527cde : SUCCESS in 1h 13m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5d142160acb14a0f9a9f4d15e7bce011 : SUCCESS in 1h 23m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3bd2c6424dc04ed883f2c6a83c4e2e07 : SUCCESS in 1h 11m 03s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/acca652c76134214b50adaefec2d43dc : SUCCESS in 1h 50m 45s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c6d824362a694c20a3b2be90a55aa261 : SUCCESS in 1h 40m 09s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2290179c9bd44803a6cb4ab0d86e4721 : SUCCESS in 1h 23m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/efd7a86ffa5946649f765ba61f0f7fb8 : SUCCESS in 1h 11m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/20ba6cb20eab436c8167921c1721313d : SUCCESS in 2h 16m 20s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ec9e59c1a362451cabe43cb04a80c1e4 : SUCCESS in 1h 34m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/29678d0f9b724d749c947eff00d39468 : SUCCESS in 1h 43m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4908aa51d6d9498da9d99828d248f449 : SUCCESS in 1h 12m 50s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8966e4e0d0cc4912ac01afe280116949 : SUCCESS in 1h 22m 50s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/91c44a241e0f4f0e92a79b4ff4b6cb43 : SUCCESS in 58m 02s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b5586e47d8d4f2b9c2b5d33ca00fed2 : SUCCESS in 1h 19m 23s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea7853e3406946f887bf4d8d69fbd709 : SUCCESS in 1h 17m 12s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/13bfd1f1bb0945df8744ef3e23a6e8b4 : SUCCESS in 1h 49m 03s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a5614583da0448508e67e00d784ba1e8 : SUCCESS in 2h 08m 18s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f643c4adde0c47b08c9b43ff9aa4e864 : FAILURE in 45m 48s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1bc63b9f9bb84adc96c1a7554eced4cb : SUCCESS in 1h 57m 27s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/432823791f8b44ee841920085a19f95b : FAILURE in 44m 51s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/policy.py\n  Comments left for invalid file neutron/agent/common/ovs_lib.py\n  Comments left for invalid file neutron/objects/base.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_daemon.py\n  Comments left for invalid file neutron/common/eventlet_utils.py\n  Comments left for invalid file neutron/tests/unit/agent/ovsdb/test_impl_idl.py\n  Comments left for invalid file neutron/tests/unit/test_auth.py","accounts_in_message":[],"_revision_number":9},{"id":"6ca5d9bc07dfb8ce0c3542be06c4d31331c9c4dc","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-04-14 23:53:05.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"92187a83da64c0379e8884ea5ee2717989c38549","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-15 00:37:16.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712640/9 : FAILURE in 43m 43s","accounts_in_message":[],"_revision_number":9},{"id":"9ebc8ea33ef9c55ca654e0ef130de0c0abe683f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 03:47:51.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1819ffefcc554cd59394e734dfb71965 : SUCCESS in 51m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/814bfe4c6ecb47cca42cde96cf55e52a : SUCCESS in 33m 01s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/60a424f5625a4bb989c24e33ff86cb7f : FAILURE in 1h 56m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cd8d7fa35cfe4c8eb7deaf48303dcbf2 : SUCCESS in 1h 24m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/041b4a8769884d968bd8c6cdc8ed1817 : SUCCESS in 1h 32m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/029aa3d3cf84487ba19e0bad7fbabf6d : SUCCESS in 1h 30m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cb1dde97de64b8cb2995b71c4785c2a : SUCCESS in 32m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c36d7e719e247be955fc36021f91e3d : SUCCESS in 28m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62eeb81f670a450abbba4cb0c66dd96d : SUCCESS in 14m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c874a91e47a34b8788a1a43371e5f8de : SUCCESS in 28m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ba472a72dfd49c6baac6f286f9362f9 : SUCCESS in 29m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d17ed197626402bb2c5f3782006f504 : SUCCESS in 29m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43c3da553b654faea7ebd0e84414882d : SUCCESS in 9m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d497d6d5c94345db95cee48d8d77665a : SUCCESS in 46m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c4e92f41b08042159b69c2a866463148 : SUCCESS in 1h 07m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/daa7287e71e94541a48a2f071964d2c1 : SUCCESS in 1h 12m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3389484815e042cd80d1bea490f7ca5f : SUCCESS in 1h 46m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/da778c387bc0452e9244c58798fb115b : FAILURE in 47m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0c90369d52c94a7dbe4930f5468796a7 : SUCCESS in 1h 24m 12s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4a2d17d0eaa84f309d38a8fcb33e54f2 : SUCCESS in 1h 15m 37s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3203f61246404b889f7abafe89f3720b : SUCCESS in 1h 15m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a955bc6ea8674bbe9b00712ece06d552 : SUCCESS in 1h 16m 06s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/02b97f0689a042a0a7d9128489c559eb : SUCCESS in 1h 28m 35s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/be4c1a93a5924a419926ff836f202064 : SUCCESS in 1h 42m 00s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f208c943c33347509ecb97a4b791ca97 : SUCCESS in 1h 10m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a3274612d65e403ab17aa70d8727d6f4 : SUCCESS in 52m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9377da7579cf4fa8bc505b4200ca7ce5 : SUCCESS in 2h 15m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1621fda12734b56aa166714ac98d691 : SUCCESS in 1h 11m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9b95dff2013c47c898b126788bce037a : SUCCESS in 1h 55m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/906478b9c6cb454796e450ca6203713f : SUCCESS in 1h 33m 43s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/20da6b7bebb7461d8c8bb99b8ec39a4f : SUCCESS in 1h 31m 14s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3939718e3a7f4e1598c1839c4d532c7b : SUCCESS in 59m 07s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7ec16c040b0943e287e8a17053901bc5 : SUCCESS in 1h 55m 57s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4b44700e786e412a918396ccf6a30758 : SUCCESS in 2h 01m 27s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/42c965bb91834b8c8bb938fad4e21d46 : SUCCESS in 2h 59m 36s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5fddd632c6eb497199baf020d2c8875f : SUCCESS in 1h 51m 04s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/4074600c9fc7421a82e48383ad7569c6 : SUCCESS in 1h 57m 42s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"43c3196ab900d177237d901ac664b27e23cb818c","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-04-15 08:44:04.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"6f30f7d14bf412a85afa0ab7c49f347bb9bb16de","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-15 09:27:26.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712640/9 : FAILURE in 42m 53s","accounts_in_message":[],"_revision_number":9},{"id":"fa836742e80bff15faca479ea25b6d3133af5f09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 13:19:29.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c162a945259e484a8fa3b5e2a3c7858c : SUCCESS in 54m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d71cab73c9a9418f97ce3b949a920be4 : SUCCESS in 34m 14s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/c42a80175d86457b843efc32cdad4c69 : FAILURE in 1h 42m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/27d7b5377c274a54b87c4d6977e4c3fb : SUCCESS in 1h 19m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ccac1784e0914154b362334f82b43eb1 : SUCCESS in 1h 12m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6dffad882d4946478377f0f85efed75f : SUCCESS in 1h 24m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0c18c38ec9c402caa132655afd29a01 : SUCCESS in 34m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33dc5e635c3b46c8a57da4b0db4067c3 : SUCCESS in 32m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/713b356a74c74f4da0cb7f98d9504c7a : SUCCESS in 14m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f829c2684439457b8af99f9d563eb5e4 : SUCCESS in 25m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e770bb1f8d874e2aad415e5637e60562 : SUCCESS in 29m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f6dc833e8ce4743acc22627b8a40a8e : SUCCESS in 28m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32fb951040354a479d9206861661b7b3 : SUCCESS in 10m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/74bff1155acf4ea3912c1e8a388fb019 : SUCCESS in 52m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5ba995008cb546c6b1d05cd351a17491 : SUCCESS in 1h 16m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/eac2d8c87b06411487b00c4eab2fde87 : SUCCESS in 1h 11m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0403c54a4fc44b2b9bae62a4686661b8 : SUCCESS in 2h 06m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4ba2795d32fe4b77a52d6e00f0ac2b6a : SUCCESS in 1h 19m 49s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/57e94d6a15a64ffb92ee085e9e1d14a8 : SUCCESS in 1h 22m 04s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1ec8f9f4a09e4bc3b8f44d887f663751 : SUCCESS in 1h 19m 31s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d58015911b1d4b50b176aa44c6d7f537 : SUCCESS in 1h 09m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/45ab6ad449854d0d970b59a01c91e35e : SUCCESS in 1h 25m 05s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/403a0a500d8b4149b1ad66ecfe5fc7d3 : SUCCESS in 1h 50m 09s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bb376bbb84c048bcb5dd7c25155463b6 : SUCCESS in 1h 48m 02s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/85a36bbfa4be41128ccb787f672769d3 : SUCCESS in 1h 21m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/925440f8610245709066cf64094b12ce : SUCCESS in 59m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e8816a499b99408db82e4cfdd26d2a8d : SUCCESS in 2h 28m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a369d68b89a4449580b95e9eb2fe346d : SUCCESS in 58m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5a3a4ed9a1b9486896636573eaa7379e : FAILURE in 1h 43m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/79bd342515c44c4794a738a46c8c91f1 : FAILURE in 2h 14m 44s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/bf516dd173f84e1eb1f961fba8a5e1e0 : SUCCESS in 1h 25m 53s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7f258c486c564c589b4d0a076d7aec55 : SUCCESS in 1h 01m 00s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3f64cf4470074cdfa90a56922b200cd0 : SUCCESS in 2h 01m 56s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e365d45e56644ef1829f75c97f0401ca : SUCCESS in 2h 04m 57s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1bf8920880224ddb92b204e6d90c21d7 : SUCCESS in 3h 08m 08s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b778f3d784654f9087097b96e2445dd8 : SUCCESS in 2h 01m 13s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/0fe3a7a5a2324c0fa8b4f343b0d874b8 : FAILURE in 2h 11m 20s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"152e0680e57a293cf1ee9856889a7fca4526a43a","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-04-17 02:57:22.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"83f269be0d96c73cccbc9a0f036ed3714712b201","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-17 03:41:47.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712640/9 : FAILURE in 43m 57s","accounts_in_message":[],"_revision_number":9},{"id":"7a0d89a50c317a33b48c700a46d1972f35e9e55c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 05:28:45.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ba14e7d7e25c4a6aa7e4d44dab6406cd : SUCCESS in 50m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8786f05a44974ffe9748bed8b47b3793 : SUCCESS in 30m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/f7743c75423647899690e4747f02ab14 : FAILURE in 2h 08m 24s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e6372c1e89724800ab642ef96229213a : SUCCESS in 1h 12m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/47d1650e75ee4fa7ba970c9a3532bece : SUCCESS in 1h 02m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1dcc3714e9f04f3e8cd48934245afdf3 : SUCCESS in 1h 07m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04dba26802b448d288bb59e34930dea5 : SUCCESS in 38m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90d6c9c7c98242dda95c8acb952e813a : SUCCESS in 32m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cd6161ebb9d49c39076f4cae768675e : SUCCESS in 12m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d89de33be0044e438370962b8e45a75c : SUCCESS in 32m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00a1e86ff8944321a2e0f2829cf3f0f8 : SUCCESS in 31m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/471c322f65494cd7b1662b16234137f7 : SUCCESS in 26m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/179fea26effc4936a3dd92f573fff171 : SUCCESS in 9m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/15be854bb6534616a61a98fc037285c0 : SUCCESS in 53m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/be295b67702941c1a4e6b1d1de11fa7e : SUCCESS in 1h 11m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5443539112ed4f5f8369e4a2f23140d5 : FAILURE in 1h 16m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d060f77582f24a248eafc1c4a172d087 : SUCCESS in 1h 46m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/26d2f013af744af98855528b38502099 : SUCCESS in 1h 28m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/275cb7d03c3d48cdba681214f2cfe3d2 : SUCCESS in 1h 35m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7a743b3cea3d4474b859addf5fa2aba7 : SUCCESS in 1h 43m 47s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5ed8400c3b2e46a5869d7bee2dbd799f : SUCCESS in 1h 08m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0d9975007847424e8f8a72782529ec43 : SUCCESS in 1h 11m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0b2aada2259149fbab0032169fe7f104 : SUCCESS in 1h 41m 46s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/53d94fd819aa4b5383fe2cc8f2983e14 : SUCCESS in 1h 38m 04s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a0babe27f81e4f86a65aaa60c1a5f0b4 : SUCCESS in 1h 17m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9f979b3daa0745c385591966768af3d5 : SUCCESS in 49m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b9a08dc649e5462dbc3cbbdec4507fd3 : SUCCESS in 2h 23m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72f138f001e1451b8b2a1ef5932a03b1 : SUCCESS in 1h 02m 35s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6979379ddb2b47a5ad9ba48a7fcffef2 : SUCCESS in 1h 41m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8670e2814a774e4cb5261ccc6f8ad641 : SUCCESS in 1h 47m 42s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e262dd39bf9a486aad879406e68fab2b : SUCCESS in 1h 21m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/244833fda7ad4dda85c4a6b564486538 : SUCCESS in 51m 25s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f748aa7648264279b4691b19abe9fe21 : SUCCESS in 1h 52m 49s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/05cfa9f3d666443ca29ade8e876d9a37 : SUCCESS in 1h 47m 08s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2b8d5b21ebc8485ea09c40db09f0740e : SUCCESS in 2h 29m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d15c8e459621473e92e4412a73995ecf : SUCCESS in 1h 57m 24s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3d433dfcc0c34e0d9060d826680800c7 : SUCCESS in 2h 08m 09s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7acdba09396e27e1ddc76875a17857333c0fcfe6","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-04-22 01:37:42.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)\n\nSorry for the late reply, the fix looks good to me, thank you.","accounts_in_message":[],"_revision_number":9},{"id":"717db3c7d0c859f222fa8cf478692820b36486f0","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-22 04:08:40.000000000","message":"Patch Set 9:\n\n(1 comment)\n\nIt is impossible because sg initialization is after setup_integration_br().","accounts_in_message":[],"_revision_number":9},{"id":"ea923433dfd5d5989668f2a53632d2c41760925f","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-04-22 13:23:07.000000000","message":"Patch Set 9: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fc7db0a79f026a62fd917d01fe579bbb1ced90ab","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-22 14:03:58.000000000","message":"Patch Set 9:\n\n(1 comment)\n\nThanks, please confirm if you expect the code in my comment.","accounts_in_message":[],"_revision_number":9},{"id":"ab6403774128b0924f5a0b76f7a7dae2ef7a3738","author":{"_account_id":21907,"name":"Arjun Baindur","email":"xagent@gmail.com","username":"arjunbaindur"},"date":"2020-04-28 22:46:10.000000000","message":"Patch Set 9:\n\nHi Yi and Liu, can you see the latest comment I posted on this bug and the original for OVS based firewall?\n\nI have a question about where we can enable this option. Can we enable it on a dvr_snat node that is hosting SNAT routers but one that is not HA? Or can it only be enabled on regular DVR nodes that host qrouter and fip? \n\nFurthermore, what are specifically issues it causes with HA routers and centralized FIPs?  Will it always break an SNAT router hosting centralized Floating IPs, or only when this is an HA router as well...?\n\nThe original commit doc isn\u0027t clear about the effects of turning this on","accounts_in_message":[],"_revision_number":9},{"id":"1330ac7b6e612904b6a7bdc63887c9a059d3a35b","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-29 00:26:05.000000000","message":"Patch Set 9:\n\n\u003e Hi Yi and Liu, can you see the latest comment I posted on this bug\n \u003e and the original for OVS based firewall?\n \u003e \n \u003e I have a question about where we can enable this option. Can we\n \u003e enable it on a dvr_snat node that is hosting SNAT routers but one\n \u003e that is not HA? Or can it only be enabled on regular DVR nodes that\n \u003e host qrouter and fip?\n \u003e \n \u003e Furthermore, what are specifically issues it causes with HA routers\n \u003e and centralized FIPs?  Will it always break an SNAT router hosting\n \u003e centralized Floating IPs, or only when this is an HA router as\n \u003e well...?\n \u003e \n \u003e The original commit doc isn\u0027t clear about the effects of turning\n \u003e this on\n\nI\u0027m not sure your dvr_snat node is compute node or neutron network node, if it is SNAT gateway, I don\u0027t think you need to enable this in that node.\n\nI suggest you try this patch and check what happens.","accounts_in_message":[],"_revision_number":9},{"id":"5fd6d28be697b0385730de2e0bb6e2da4975e3b3","author":{"_account_id":21907,"name":"Arjun Baindur","email":"xagent@gmail.com","username":"arjunbaindur"},"date":"2020-04-29 01:47:13.000000000","message":"Patch Set 9:\n\nI guess I\u0027m confused by the terminology. In DVR, isn\u0027t every node a compute node and a network node? I thought \"network node\" referred only to nodes in a non-DVR topology, where dedicated network nodes hosts centralized l3-agent (and all routers) and DHCP services. \n\nBut in DVR we have the router and FIP namespace located on same nodes as VMs. One of these nodes will also host the snat namespace (based on the l3 agent scheduler), and some will have DHCP services running. we aren\u0027t dedicating one entire node to only host SNAT but no VMs","accounts_in_message":[],"_revision_number":9},{"id":"dee354ec6e895fed10b8f87ebc108cacbd62844e","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-29 02:38:49.000000000","message":"Patch Set 9:\n\n\u003e I guess I\u0027m confused by the terminology. In DVR, isn\u0027t every node a\n \u003e compute node and a network node? I thought \"network node\" referred\n \u003e only to nodes in a non-DVR topology, where dedicated network nodes\n \u003e hosts centralized l3-agent (and all routers) and DHCP services.\n \u003e \n \u003e But in DVR we have the router and FIP namespace located on same\n \u003e nodes as VMs. One of these nodes will also host the snat namespace\n \u003e (based on the l3 agent scheduler), and some will have DHCP services\n \u003e running. we aren\u0027t dedicating one entire node to only host SNAT but\n \u003e no VMs\n\nGot it, for our deployment, we have dedicated network nodes for snat, l3, DHCP, metadata, for these nodes, agent mode is dvr_snat\n\n/etc/neutron/l3_agent.ini:agent_mode \u003d dvr_snat\n\nFor your case, can you show me your /etc/neutron/l3_agent.ini with dvr_snat enabled? I think you only can set agent_mode to dvr or dvr_snat, how do you enable dvr and dvr_snat in an l3 agent config file, they can\u0027t coexist.\n\nFor you case, I think you can turn on it (explicitly_egress_direct \u003d True), please let me know if you have any issue.","accounts_in_message":[],"_revision_number":9},{"id":"9dc386d5fde476515119dcd835368f43793e6372","author":{"_account_id":21907,"name":"Arjun Baindur","email":"xagent@gmail.com","username":"arjunbaindur"},"date":"2020-04-29 02:53:43.000000000","message":"Patch Set 9:\n\nYi, the good news is patch works on both compute nodes hosting SNAT namespace and not. And without it, we see the flooding, regardless of whether snat router is present or not.\n\nwith dvr_snat, the node can also host VMs. We just assign some subset of nodes typically as dvr_snat, the others are just agent_mode \u003d dvr. Same goes for DHCP - we assign some subset (say, at minimum 3) to run DHCP agent. The neutron schedulers will then actually deploy the snat/dhcp namespaces to one or more of them.\n\nMy concern was the comment on the original commit (not yours): https://review.opendev.org/#/c/666991/13/releasenotes/notes/accepted_egress_direct-cc23873e213c6919.yaml\n\n\"A new config option ``explicitly_egress_direct``, with default value False,\n    was added for the aim of distinguishing clouds which are running the\n    network node mixed with compute services, upstream neutron CI should be\n    an example. In such situation, this ``explicitly_egress_direct`` should be\n    set to False, because there are numerous cases from HA routers which can\n    not be covered, particularly when you have centralized floating IPs running\n    in such mixed hosts.\"\n\nThat description is a bit vague and I don\u0027t understand it. We have \"network node\" mixed with compute. Does this mean we should always set this? Or is that *only* the case if we use HA snat routers? Or is it *only* the case with centralized Floating IP AND an HA snat router?\n\nWe definitely see the flooding issue on compute nodes hosting SNAT. And with this fix, the issue definitely goes away. I am worried about these corner cases arising from HA routers, centralized FIPs, etc... as stated in the comment. We can turn the option off on nodes set to agent_mode \u003d dvr_snat, but then we\u0027d have several nodes in each deployment still affected","accounts_in_message":[],"_revision_number":9},{"id":"b22233002c6e52ab339fb5470f85cc8f176b02c5","author":{"_account_id":21907,"name":"Arjun Baindur","email":"xagent@gmail.com","username":"arjunbaindur"},"date":"2020-04-29 02:59:30.000000000","message":"Patch Set 9:\n\nsnat functionality is not broken either, both for VMs located on same node as snat router and remote host. so that is good too.\n\nbut i\u0027ve yet to test more complicated topologies (HA, centralized floating IPs)","accounts_in_message":[],"_revision_number":9},{"id":"d0f13d377300e9ae54182b000146443c9fd1f392","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-06-09 03:14:12.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"796c73a39be5f3049604c13b7704993f5e4e5656","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-09 03:57:44.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712640/9 : FAILURE in 42m 53s","accounts_in_message":[],"_revision_number":9},{"id":"90dc962ee107445071fbdcc3db198f0f6f13c9bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 05:50:18.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/de649b322daa4b1aa658c91924870025 : SUCCESS in 52m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1959544397594c548f654353a48c1a78 : SUCCESS in 31m 10s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/1075132aad2f46328c2de2a4934f9e23 : FAILURE in 1h 44m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/11b3adbdccc34cc8b3645611241bba54 : SUCCESS in 1h 13m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2ad316a8ce9044ce8b693c408d97c9d3 : FAILURE in 1h 05m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7d831e2483e84427bed6d67bfe4741e0 : SUCCESS in 1h 04m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79aaaf75185840d5b4aa40379516416d : SUCCESS in 35m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/439d38cdfc79438aaf5deae4df36e45f : SUCCESS in 35m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6769c1e9e51b4f5db4625511618b79e2 : SUCCESS in 10m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/789e2962ec964c1ab858c7de79ecedfc : SUCCESS in 34m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74fe19ad9e4c4cf7b57affd788c53673 : SUCCESS in 24m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c7338450e6d43d6bb46107675ac34b0 : SUCCESS in 11m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6be7a9c8278649f8b5ef3905233ccb29 : SUCCESS in 1h 03m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b214dfafe61450b8a08537b9e0e16bf : SUCCESS in 56m 37s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d43cd17b98e0457e90202b46a4129df8 : SUCCESS in 1h 48m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/68b316b1dabe41f3a876c8619b1c2946 : FAILURE in 1h 32m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e49eccf160ec4d048cf4100c961d04d5 : SUCCESS in 1h 50m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9cf147f35bf7400881e1146f2f1c6232 : SUCCESS in 1h 44m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1790af7cfb56426e87945da85bfc5573 : SUCCESS in 1h 31m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9b7eac7069bb412aa4809beaa78f7f82 : SUCCESS in 1h 40m 03s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/466e9fc335a94047a2f9b49c62102740 : SUCCESS in 1h 36m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6e05fec0c2024366b25228cf75f928ca : SUCCESS in 1h 45m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1806fd91362f4cfcafc24b23d63e4ebf : SUCCESS in 1h 57m 47s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/01a89b1a0f9846d1be9cb74011fd126d : FAILURE in 1h 07m 14s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/34f85d2b2530460f9392dd12359f95d0 : SUCCESS in 1h 28m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d3d5d279e8714d58b2e3e5cc3014fb7e : SUCCESS in 53m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0e11b9edd174472f8c8d7e41806db9c6 : SUCCESS in 1h 59m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38966dae0f1f46caae03f7fd40960ceb : SUCCESS in 55m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/eb2bcf31d0674cc6aaf1f4523bd81cf8 : SUCCESS in 1h 39m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/60c79ba2902c4a5db724e651f70d9b40 : SUCCESS in 1h 50m 51s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4d8f99d1cb9e44a38bd9868fd16d8dd1 : SUCCESS in 1h 37m 30s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c2ec71bfe9cd4838b8066f635f016c7e : SUCCESS in 1h 44m 04s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a69784b513054d1ea9815b524e2750c4 : SUCCESS in 1h 44m 59s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fc93a8ab4dd74aa3a44c93cbe026402d : SUCCESS in 2h 26m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7055beeb8624443ca89b5f4dcc0feca7 : SUCCESS in 1h 35m 18s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d245e42a09344da5b81109cef70e841d : FAILURE in 2h 35m 25s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9218c24fb175a927c0ba0ea881c59db51df6b887","author":{"_account_id":21907,"name":"Arjun Baindur","email":"xagent@gmail.com","username":"arjunbaindur"},"date":"2020-06-23 07:18:42.000000000","message":"Patch Set 9:\n\nWe are seeing ingress flooding occur due to enabling this fix on other networks. We have a VLAN based network, both attached and unattached to a DVR router. In both cases, on ingress the NORMAL action at table\u003d60 is hit. \n\nThis is because for a provider network, there are no dvr MAC rules with an explicit output: action to a local port. In DVR subnet, these rules are added with the EXTERNAL vlan/segmentation ID. Not with the local vlan. However, there is a table\u003d0, priority\u003d3 rule which converts from the external VLAN to host local VLAN. \n\nNORMAL action at table\u003d60 for ingress, requires the local VM MACs to be learned. But these are never learned, because the fix resubmits egress packets to table 61, which has a direct output: action northbound.\n\nSo now we have flooding in opposite direction. But we still need this fix because in other scenarios, we see the egress flooding.\n\nhttps://bugs.launchpad.net/neutron/+bug/1884708","accounts_in_message":[],"_revision_number":9},{"id":"7b96575ea1cc53d2d28ef0512ae5e6b28750482c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-25 15:40:40.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":9},{"id":"44a03425d524c6ce7c3cd537979a8c9712a0f9f4","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-09-27 00:47:52.000000000","message":"Restored\n\nIt is still valid to us. we don\u0027t know why some test cases failed, anybody can help on this?","accounts_in_message":[],"_revision_number":9},{"id":"83f12614fec89dc5c5bdc7b86f9ce10a5f6bd2fa","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-27 01:35:49.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/712640/9 : FAILURE in 47m 28s","accounts_in_message":[],"_revision_number":9},{"id":"52d7413365ac5a3fc83d45a5e4d6b2412ea7ea5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-27 03:56:09.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/391e8f6718d04c2babdc120507400f3b : SUCCESS in 53m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b3fe3374181d43a9b4c22bbe5c300c63 : SUCCESS in 39m 25s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e844f22111174dfba7d9aef75c65dc72 : FAILURE in 1h 13m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c7cafba773c64545927f63152aa0f8f5 : SUCCESS in 1h 22m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/03f7660ffbaf4666971253ecf8a3e391 : SUCCESS in 1h 05m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0d1904712b2f4c4d80cdd39b7ffe1d0b : FAILURE in 13m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/687caf8d1972428784a0197bb214cfef : SUCCESS in 32m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a23a0aabc11d406692b48c750238fe19 : SUCCESS in 17m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dedcc32589764e809db624a44f286894 : SUCCESS in 10m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff8c1fd9835b4534ba9688775ae1e217 : SUCCESS in 16m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/893438fc771f4fab9c4e712fad0db253 : SUCCESS in 22m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7ca23b8b5684b3b8ace5f1114016762 : SUCCESS in 9m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/29f56161e2614b63af67615a146ff1c1 : SUCCESS in 1h 06m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/09bfe5b397114a9da218f8cce8b88a98 : SUCCESS in 1h 10m 05s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/12476f019fad4abeb37e6cbff1f94b79 : SUCCESS in 1h 20m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/74162af6fa1342a58e757c673387ef0b : SUCCESS in 1h 09m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4faa63d6f1104ee19ceae36aa7f41eb2 : SUCCESS in 1h 10m 42s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/68bce6700945419a9a422171c8a0a832 : SUCCESS in 1h 07m 18s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ddcd79f537f8457a9569947c36bd470b : SUCCESS in 1h 54m 50s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4f63ea9ab31a4f0bac8ebcdb5a43bb9e : SUCCESS in 1h 22m 39s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3de9907bb682455a96c99cfd32dc960e : SUCCESS in 1h 43m 34s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3a4e15f35dcf49318c56b1c8dc9480c4 : SUCCESS in 55m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0dbe73c71c014974a5258400bf11e8c8 : SUCCESS in 1h 11m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f57371a79ab84b03954ea80e9ec29637 : SUCCESS in 2h 10m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95b6f93523114d6a844945e8ff4e8d73 : SUCCESS in 40m 28s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/572ec4706c5e46e1b2f7ac700a85d0c2 : SUCCESS in 1h 10m 33s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/60de763898704e3894fb517410adcf45 : FAILURE in 16m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bf8efe6f540e4739a871390227818448 : FAILURE in 15m 19s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d594b980433540a59a1c8cd366056b3e : SUCCESS in 1h 16m 01s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1727d511a31a448780ecaf3a334bc599 : SUCCESS in 2h 32m 43s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f56d3978a9644d8fa514e517fb9151a9 : SUCCESS in 1h 48m 15s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fe537d6a47d94936b5522393107a003b : SUCCESS in 2h 10m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f10ecab8dfc542339e3ee14918a7fff7 : FAILURE in 22m 46s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/9f3f650473cc46a5bebaee6dd275d942 : TIMED_OUT in 3h 06m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c1504757e7a1f2ba658536dcbdba1f9ecea44a2d","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-10-02 17:35:29.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"06b9fbb96953cb2073da53469b7c2f1763241b85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 00:35:14.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b8339d1422d142bab0e8731b1d3ec506 : SUCCESS in 1h 51m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8dc873541d364735ae76cb2b15cd3f7e : SUCCESS in 36m 38s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f009087ecd104bac8c96999002e56f86 : SUCCESS in 1h 37m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/64c9893ad8e445539cf4ca32c3213b0e : SUCCESS in 1h 23m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bef4c25a26f44c5f96a37df1e381c853 : SUCCESS in 1h 14m 10s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9c9e8b98c4074e77b732c23260ce34dd : SUCCESS in 1h 11m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/beedbc9370774f4d85e0512d275eb29b : SUCCESS in 34m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d11d11c6a26f4b66a9f3bc465e7566b7 : SUCCESS in 26m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79b2c9534de84617a649d1d9049b655b : SUCCESS in 14m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cbbbdf24bf54200b0ae880283a60b12 : SUCCESS in 25m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbab9dc4a8d74f029271db238a841346 : SUCCESS in 26m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04ae56dfc6834f1fbd7e34b000405ad2 : SUCCESS in 9m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b267c392437048c2b48e8de83e7a487d : SUCCESS in 1h 09m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/54447126a6a6410ebe0f6f5fb5268b52 : SUCCESS in 1h 11m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0d40d078641d4dc594c4d43f1fb724c5 : SUCCESS in 1h 38m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/67c8dfdad45042b2a7e4a56d5551af2d : SUCCESS in 1h 03m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dd82982030024de2adbadefbdd3b72a0 : SUCCESS in 1h 10m 51s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/dd844a177f1a4ef7bdb775aa9bde5dec : SUCCESS in 1h 14m 48s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d94d842565244619b3151aa2db0b8b67 : SUCCESS in 1h 08m 06s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1fb27eed5a334d059a42db6726717059 : SUCCESS in 1h 25m 23s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9af8a266611140599265f3447091f4ee : SUCCESS in 1h 36m 11s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/47fac5bfe7b04540941fa9231396d1be : SUCCESS in 1h 07m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4c4116288e0b4cfbbd94d72e2bb8b21c : SUCCESS in 1h 02m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/75349e1053054766b83af13ed1f34336 : SUCCESS in 1h 37m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c8024ef89034a138cc6b7e789e716e6 : SUCCESS in 1h 00m 18s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5fb1100fd79f4c46a03209446343a5aa : SUCCESS in 1h 06m 20s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/93052aa782d94cce8d7a5209d8c12276 : SUCCESS in 1h 17m 07s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e28af09954514dc5955559f2bc7358ce : FAILURE in 54m 33s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7d2566d70b394540965c8ddc44bc6ea1 : SUCCESS in 1h 21m 59s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ba8178a66eb149bd9e7fdaf9e4886730 : SUCCESS in 2h 03m 00s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/728361c0443c4c5998f3f441876c12bf : SUCCESS in 1h 49m 32s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/afa260af1f3c43529b5abdd0f8f8a605 : SUCCESS in 2h 03m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2db5eb1c83b644c18fa112ee29f27dbc : FAILURE in 59m 02s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/7da1040337624180b3b0a5d7952fdd59 : TIMED_OUT in 3h 05m 05s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"16a9c31584df49bcd64b86d3ec3efee423d0b698","author":{"_account_id":21907,"name":"Arjun Baindur","email":"xagent@gmail.com","username":"arjunbaindur"},"date":"2020-10-08 00:01:12.000000000","message":"Patch Set 9:\n\nIs this still valid now that this has finally been merged? https://review.opendev.org/#/c/738551/","accounts_in_message":[],"_revision_number":9},{"id":"62c6191fbf8ea3508967467549cc63e9c9637ee1","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-10-09 01:13:56.000000000","message":"Patch Set 9:\n\n\u003e Is this still valid now that this has finally been merged?\n \u003e https://review.opendev.org/#/c/738551/\n\nFrom source code in https://review.opendev.org/#/c/738551/, it seems this isn\u0027t needed any more, @LIU Yulong, please help double confirm, thanks a lot.","accounts_in_message":[],"_revision_number":9},{"id":"7dfc1a8aa7645a2e903e4a4446b26b50916d97c8","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-10-15 15:05:55.000000000","message":"Patch Set 9:\n\n\u003e \u003e Is this still valid now that this has finally been merged?\n \u003e \u003e https://review.opendev.org/#/c/738551/\n \u003e \n \u003e From source code in https://review.opendev.org/#/c/738551/, it\n \u003e seems this isn\u0027t needed any more, @LIU Yulong, please help double\n \u003e confirm, thanks a lot.\n\nSorry for the late reply, this is not needed anymore, it is covered by 738551.","accounts_in_message":[],"_revision_number":9},{"id":"30a2d191683e7c932114e8074a7a21010908a313","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-10-19 02:04:23.000000000","message":"Abandoned\n\nhttps://review.opendev.org/#/c/738551/ has covered this, so abandon it.","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"01882de1f568fd6037ff459011cca0dbbf9df125","revisions":{"a1a02f15f2ef06e478dad91af4cf859574efa2e1":{"kind":"REWORK","_number":1,"created":"2020-03-12 10:22:23.000000000","uploader":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"ref":"refs/changes/40/712640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/712640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/712640/1"}}},"commit":{"parents":[{"commit":"c90011ee49a00d3b1153afebc171f6f72519dda5","subject":"Merge \"Bump neutron-lib to 2.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c90011ee49a00d3b1153afebc171f6f72519dda5"}]}],"author":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2020-03-12 09:52:49.000000000","tz":480},"committer":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2020-03-12 10:21:32.000000000","tz":480},"subject":"Enable \"accepted egress direct flow\" to work for iptables_hybrid","message":"Enable \"accepted egress direct flow\" to work for iptables_hybrid\n\ncommit efa8dd08957b5b6b1a05f0ed412ff00462a9f216 doesn\u0027t work for\nthe case firewall\u003diptables_hybrid, this patch fixes it and make\nsure it can work in same way for openvswitch and iptables_hybrid\nfirewall.\n\nRelated-Bug: #1732067\nRelated-Bug: #1866445\n\nChange-Id: I9cc05e9e89b806263e9cba31ed9e16b1d2b455f6\nSigned-off-by: Yi Yang \u003cyangyi01@inspur.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1a02f15f2ef06e478dad91af4cf859574efa2e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1a02f15f2ef06e478dad91af4cf859574efa2e1"}]},"branch":"refs/heads/master"},"c4a75b9adc047d4eda8bd753e6a1a451adb0507b":{"kind":"REWORK","_number":2,"created":"2020-04-01 08:11:33.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/40/712640/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/712640/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/712640/2"}}},"commit":{"parents":[{"commit":"8b7d0dcbef75dfdb5dbf8787aa395bf359e1c8eb","subject":"Merge \"Filter subnet by segment ID or None\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b7d0dcbef75dfdb5dbf8787aa395bf359e1c8eb"}]}],"author":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2020-03-12 09:52:49.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2020-04-01 07:58:39.000000000","tz":480},"subject":"Enable egress direct flow for iptables_hybrid","message":"Enable egress direct flow for iptables_hybrid\n\nWhen enable explicitly_egress_direct for iptables_hybrid security group\ndriver, the egress flooding still occurs on br-int. This patch adds the\niptables_hybrid to the driver check list to ensure the direct flows are\ninstalled properly.\n\nRelated-Bug: #1732067\nRelated-Bug: #1866445\n\nCo-Authored-By: LIU Yulong \u003ci@liuyulong.me\u003e\n\nChange-Id: I9cc05e9e89b806263e9cba31ed9e16b1d2b455f6\nSigned-off-by: Yi Yang \u003cyangyi01@inspur.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4a75b9adc047d4eda8bd753e6a1a451adb0507b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4a75b9adc047d4eda8bd753e6a1a451adb0507b"}]},"branch":"refs/heads/master"},"ccd778049f116065b4dd8c9b3390cd369a328683":{"kind":"REWORK","_number":3,"created":"2020-04-01 08:14:52.000000000","uploader":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"ref":"refs/changes/40/712640/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/712640/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/712640/3"}}},"commit":{"parents":[{"commit":"8b7d0dcbef75dfdb5dbf8787aa395bf359e1c8eb","subject":"Merge \"Filter subnet by segment ID or None\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b7d0dcbef75dfdb5dbf8787aa395bf359e1c8eb"}]}],"author":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2020-04-01 07:59:15.000000000","tz":480},"committer":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2020-04-01 07:59:15.000000000","tz":480},"subject":"Enable accepted egress direct flow for iptables_hybrid","message":"Enable accepted egress direct flow for iptables_hybrid\n\nWhen enable explicitly_egress_direct for firewall driver iptables_hybrid, the egress\nflooding still occurs on br-int. This patch adds the iptables_hybrid to the driver\ncheck list to ensure the direct flows are installed properly.\n\nRelated-Bug: #1732067\nRelated-Bug: #1866445\n\nChange-Id: I9cc05e9e89b806263e9cba31ed9e16b1d2b455f6\nSigned-off-by: Yi Yang \u003cyangyi01@inspur.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ccd778049f116065b4dd8c9b3390cd369a328683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ccd778049f116065b4dd8c9b3390cd369a328683"}]},"branch":"refs/heads/master"},"224e7a154feff9ecd6a9b551f865f04b9d1a7fec":{"kind":"REWORK","_number":4,"created":"2020-04-01 13:23:49.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/40/712640/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/712640/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/712640/4"}}},"commit":{"parents":[{"commit":"8b7d0dcbef75dfdb5dbf8787aa395bf359e1c8eb","subject":"Merge \"Filter subnet by segment ID or None\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b7d0dcbef75dfdb5dbf8787aa395bf359e1c8eb"}]}],"author":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2020-03-12 09:52:49.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2020-04-01 13:22:38.000000000","tz":480},"subject":"Enable egress direct flow for iptables_hybrid","message":"Enable egress direct flow for iptables_hybrid\n\nWhen enable explicitly_egress_direct for iptables_hybrid security group\ndriver, the egress flooding still occurs on br-int. This patch adds the\niptables_hybrid to the driver check list to ensure the direct flows are\ninstalled properly.\n\nRelated-Bug: #1732067\nRelated-Bug: #1866445\n\nCo-Authored-By: LIU Yulong \u003ci@liuyulong.me\u003e\n\nChange-Id: I9cc05e9e89b806263e9cba31ed9e16b1d2b455f6\nSigned-off-by: Yi Yang \u003cyangyi01@inspur.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/224e7a154feff9ecd6a9b551f865f04b9d1a7fec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/224e7a154feff9ecd6a9b551f865f04b9d1a7fec"}]},"branch":"refs/heads/master"},"d447e379380232f512937fdc7fa9016a663795c9":{"kind":"REWORK","_number":5,"created":"2020-04-03 14:32:58.000000000","uploader":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"ref":"refs/changes/40/712640/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/712640/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/712640/5"}}},"commit":{"parents":[{"commit":"56bb42fcbc43b619c3c07897c7de88f29158e4b8","subject":"Merge \"Revert \"Switch to use cast method in dhcp_ready_on_ports method\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56bb42fcbc43b619c3c07897c7de88f29158e4b8"}]}],"author":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2020-04-03 14:10:51.000000000","tz":480},"committer":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2020-04-03 14:10:51.000000000","tz":480},"subject":"Enable egress direct flow for iptables_hybrid","message":"Enable egress direct flow for iptables_hybrid\n\nWhen enable explicitly_egress_direct for iptables_hybrid security group\ndriver, the egress flooding still occurs on br-int. This patch adds the\niptables_hybrid to the driver check list to ensure the direct flows are\ninstalled properly.\n\nThese direct flows also need br-int to support OPENFLOW14 because they\nare installed by using bundle, so this patch also adds required protocols\nto br-int. Note: only OPENFLOW10 and OPENFLOW13 are added for br-int in\ncase of firewall_driver\u003diptables_hybrid.\n\nRelated-Bug: #1732067\nRelated-Bug: #1866445\n\nCo-Authored-By: LIU Yulong \u003ci@liuyulong.me\u003e\n\nChange-Id: I9cc05e9e89b806263e9cba31ed9e16b1d2b455f6\nSigned-off-by: Yi Yang \u003cyangyi01@inspur.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d447e379380232f512937fdc7fa9016a663795c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d447e379380232f512937fdc7fa9016a663795c9"}]},"branch":"refs/heads/master"},"6407db2578e6963997b1f87b4ec4901c470e6b87":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-04-03 14:34:26.000000000","uploader":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"ref":"refs/changes/40/712640/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/712640/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/712640/6"}}},"commit":{"parents":[{"commit":"31a6274c1f440e609cdbc277e5aac8d3cb0d0adc","subject":"Merge \"[DVR] Don\u0027t populate unbound ports in router\u0027s ARP cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/31a6274c1f440e609cdbc277e5aac8d3cb0d0adc"}]}],"author":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2020-04-03 14:10:51.000000000","tz":480},"committer":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2020-04-03 14:34:26.000000000","tz":0},"subject":"Enable egress direct flow for iptables_hybrid","message":"Enable egress direct flow for iptables_hybrid\n\nWhen enable explicitly_egress_direct for iptables_hybrid security group\ndriver, the egress flooding still occurs on br-int. This patch adds the\niptables_hybrid to the driver check list to ensure the direct flows are\ninstalled properly.\n\nThese direct flows also need br-int to support OPENFLOW14 because they\nare installed by using bundle, so this patch also adds required protocols\nto br-int. Note: only OPENFLOW10 and OPENFLOW13 are added for br-int in\ncase of firewall_driver\u003diptables_hybrid.\n\nRelated-Bug: #1732067\nRelated-Bug: #1866445\n\nCo-Authored-By: LIU Yulong \u003ci@liuyulong.me\u003e\n\nChange-Id: I9cc05e9e89b806263e9cba31ed9e16b1d2b455f6\nSigned-off-by: Yi Yang \u003cyangyi01@inspur.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6407db2578e6963997b1f87b4ec4901c470e6b87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6407db2578e6963997b1f87b4ec4901c470e6b87"}]},"branch":"refs/heads/master"},"1d8f3616640b2f3b799345e1f61c72e41ae2a8a6":{"kind":"REWORK","_number":7,"created":"2020-04-07 01:29:20.000000000","uploader":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"ref":"refs/changes/40/712640/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/712640/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/712640/7"}}},"commit":{"parents":[{"commit":"fc96f5b2b3a27979d338b558132871ca0c636e5d","subject":"Merge \"\"keepalived_state_change\" needs to use threading to send arping\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc96f5b2b3a27979d338b558132871ca0c636e5d"}]}],"author":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2020-04-07 01:27:36.000000000","tz":480},"committer":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2020-04-07 01:27:36.000000000","tz":480},"subject":"Enable egress direct flow for iptables_hybrid","message":"Enable egress direct flow for iptables_hybrid\n\nWhen enable explicitly_egress_direct for iptables_hybrid security group\ndriver, the egress flooding still occurs on br-int. This patch adds the\niptables_hybrid to the driver check list to ensure the direct flows are\ninstalled properly.\n\nThese direct flows also need br-int to support OPENFLOW14 because they\nare installed by using bundle, so this patch also adds required protocols\nto br-int. Note: only OPENFLOW10 and OPENFLOW13 are added for br-int in\ncase of firewall_driver\u003diptables_hybrid.\n\nRelated-Bug: #1732067\nRelated-Bug: #1866445\n\nCo-Authored-By: LIU Yulong \u003ci@liuyulong.me\u003e\n\nChange-Id: I9cc05e9e89b806263e9cba31ed9e16b1d2b455f6\nSigned-off-by: Yi Yang \u003cyangyi01@inspur.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d8f3616640b2f3b799345e1f61c72e41ae2a8a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d8f3616640b2f3b799345e1f61c72e41ae2a8a6"}]},"branch":"refs/heads/master"},"03557f06a5df99776d4354667c3d56120c357b27":{"kind":"REWORK","_number":8,"created":"2020-04-07 06:42:39.000000000","uploader":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"ref":"refs/changes/40/712640/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/712640/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/712640/8"}}},"commit":{"parents":[{"commit":"fc96f5b2b3a27979d338b558132871ca0c636e5d","subject":"Merge \"\"keepalived_state_change\" needs to use threading to send arping\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc96f5b2b3a27979d338b558132871ca0c636e5d"}]}],"author":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2020-04-07 06:39:45.000000000","tz":480},"committer":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2020-04-07 06:39:45.000000000","tz":480},"subject":"Enable egress direct flow for iptables_hybrid","message":"Enable egress direct flow for iptables_hybrid\n\nWhen enable explicitly_egress_direct for iptables_hybrid security group\ndriver, the egress flooding still occurs on br-int. This patch adds the\niptables_hybrid to the driver check list to ensure the direct flows are\ninstalled properly.\n\nThese direct flows also need br-int to support OPENFLOW14 because they\nare installed by using bundle, so this patch also adds required protocols\nto br-int. Note: only OPENFLOW10 and OPENFLOW13 are added for br-int in\ncase of firewall_driver\u003diptables_hybrid.\n\nRelated-Bug: #1732067\nRelated-Bug: #1866445\n\nCo-Authored-By: LIU Yulong \u003ci@liuyulong.me\u003e\n\nChange-Id: I9cc05e9e89b806263e9cba31ed9e16b1d2b455f6\nSigned-off-by: Yi Yang \u003cyangyi01@inspur.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/03557f06a5df99776d4354667c3d56120c357b27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/03557f06a5df99776d4354667c3d56120c357b27"}]},"branch":"refs/heads/master"},"01882de1f568fd6037ff459011cca0dbbf9df125":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-04-08 01:46:02.000000000","uploader":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"ref":"refs/changes/40/712640/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/712640/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/712640/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/712640/9"}}},"commit":{"parents":[{"commit":"476a0e3214b996e42cc17b7459839e9b08679b71","subject":"Merge \"[fullstack] Add option for custom apipaste.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/476a0e3214b996e42cc17b7459839e9b08679b71"}]}],"author":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2020-04-07 06:39:45.000000000","tz":480},"committer":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2020-04-08 01:46:02.000000000","tz":0},"subject":"Enable egress direct flow for iptables_hybrid","message":"Enable egress direct flow for iptables_hybrid\n\nWhen enable explicitly_egress_direct for iptables_hybrid security group\ndriver, the egress flooding still occurs on br-int. This patch adds the\niptables_hybrid to the driver check list to ensure the direct flows are\ninstalled properly.\n\nThese direct flows also need br-int to support OPENFLOW14 because they\nare installed by using bundle, so this patch also adds required protocols\nto br-int. Note: only OPENFLOW10 and OPENFLOW13 are added for br-int in\ncase of firewall_driver\u003diptables_hybrid.\n\nRelated-Bug: #1732067\nRelated-Bug: #1866445\n\nCo-Authored-By: LIU Yulong \u003ci@liuyulong.me\u003e\n\nChange-Id: I9cc05e9e89b806263e9cba31ed9e16b1d2b455f6\nSigned-off-by: Yi Yang \u003cyangyi01@inspur.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01882de1f568fd6037ff459011cca0dbbf9df125"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01882de1f568fd6037ff459011cca0dbbf9df125"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
