)]}'
{"id":"openstack%2Fneutron~714097","triplet_id":"openstack%2Fneutron~master~I1b95d53f79c2c2b56a119c20017c3bf7fbc8550d","project":"openstack/neutron","branch":"master","topic":"bp/enginefacade-switch","hashtags":[],"change_id":"I1b95d53f79c2c2b56a119c20017c3bf7fbc8550d","subject":"Finish migration to new engine facade in test_db_base_plugin_v2","status":"MERGED","created":"2020-03-20 11:37:13.000000000","updated":"2020-03-25 00:39:25.000000000","submitted":"2020-03-25 00:37:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714097-1585096641205-d1a6a7d1","meta_rev_id":"05764ab9d5d52d8cb3f60e135cffdc6c810c966e","_number":714097,"virtual_id_number":714097,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-03-24 11:44:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-25 00:37:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-24 12:40:34.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-03-24 12:43:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-03-24 19:53:45.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-03-24 19:53:45.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-20 14:55:43.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-03-23 22:50:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-03-24 11:44:13.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-24 12:40:34.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-03-24 12:43:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-03-24 19:53:45.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-03-25 00:37:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b87d773e90c2df2ae052ab8aad2bc708419a19fe","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-20 11:37:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9f898822a529763e810d6589d546a67c6385984","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 12:54:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba2e8855305d49bfa495c2f85ea99fc0 : SUCCESS in 34m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/521fd770780a408bb1d11d52a8977f16 : SUCCESS in 29m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68891d46b5034381b38200ed67462c8b : SUCCESS in 15m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1396698f2dc749b2bc0d2e904fe889bd : SUCCESS in 26m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b6f594008de400c9629df3102d3dabb : SUCCESS in 28m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54b7f5e6b7a248caaa00143fb2b4d2d0 : SUCCESS in 23m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd36ee567a44283a8854c885d303a69 : SUCCESS in 9m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6834f4c23c07418db795200cf841d8fe : SUCCESS in 53m 20s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/358a8f97cb8040c7b47f109223a1ee4b : SUCCESS in 1h 04m 34s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e6533b567db0425c82deb220f4fc43b5 : SUCCESS in 1h 11m 10s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d1445bf2d5644a88b78d52135c8129b9 : SUCCESS in 52m 03s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1e0f023cb293419d8bea9c70f29caff1 : SUCCESS in 1h 02m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c8afecaceb7fb68f492d82841b6e918c5dbc200c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-20 14:24:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/714097/1 : SUCCESS in 2h 02m 30s","accounts_in_message":[],"_revision_number":1},{"id":"477b6731d47d9d497fc641d96f4154cd1890188d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-20 14:55:43.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nOne question on a change that is not directly 1-to-1","accounts_in_message":[],"_revision_number":1},{"id":"1aa9f9273a7792b16921528b96d4874a50400364","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-20 14:57:43.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24873d0036b8c19ac7ec16341635fced34ff9f4b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-20 15:41:04.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/97/714097/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"fd92351d42056db61d22e1da663bd2ef766f47b2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-03-20 15:56:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb45e501f91de49fdad6ab43cad4a04aa481f0a0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-03-20 15:57:05.000000000","message":"Patch Set 1:\n\nSmall -1, rest of the code looks OK.","accounts_in_message":[],"_revision_number":1},{"id":"4aaef1388180173da059ab095b70df0c0bb56686","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-23 22:50:59.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"4b76772cf23997be159d58fbdb77079382a14fdd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-24 10:30:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e9bb177d31de83f8cbd6796348550dfff3705d08","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-24 10:31:23.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"165b4b799b3bf500b1ae5ab04280f9e0cdca5398","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-24 11:44:13.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/97/714097/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"59b05beb8187325f0a5ecb217cf9bfa361a48a9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 12:33:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7ab2596ec964ad9964b97ccba25ded2 : SUCCESS in 44m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1734c4b6c987489e8995d6d44528f036 : SUCCESS in 26m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7303d2e3a91642e28eb3ea44dd5a7705 : SUCCESS in 13m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9902b4d3d984ccc99684db2a00ee28e : SUCCESS in 27m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e68f67cfd7be480a8fd30d3fc5b9305f : SUCCESS in 30m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e27f7ea95f854bcf82921fb65c6789ad : SUCCESS in 28m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec0d5b83092e4b63a6344ff385ab1eb0 : SUCCESS in 9m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7b357e4e45104e4082f61a654da94319 : SUCCESS in 1h 00m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b6004192b40a4c089146790d2647823e : SUCCESS in 1h 16m 57s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fa5739ab57ef4118822790c748d3171a : SUCCESS in 1h 08m 22s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1616dd68fcc647f9b6539a92b37c964a : SUCCESS in 1h 00m 05s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e557c444ff6b4418a3d823d06c99f70a : SUCCESS in 1h 28m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"24343ced2f42ad0da4ed17f701f9639a6e9a9b8c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-24 12:40:34.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/714097/2 : FAILURE in 2h 09m 02s","accounts_in_message":[],"_revision_number":2},{"id":"a7d01f9d949f5895cff5aff37584cf6f7e793130","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-03-24 12:43:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ffe087878fdc83f7ef25c1e21ddfb2811c68a037","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-24 19:53:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b6d12536e75a50bf3254d2df36455cfbc53609a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 19:54:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"01d4a66f7738c3c67051100a4839a976cbdad3bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 00:37:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21d2075cb6294ffe82f2e7283a7feffd : SUCCESS in 27m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e20b00df3e447739e62187e99f12c54 : SUCCESS in 13m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/492346fb489542519302e1945c343861 : SUCCESS in 24m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51007ab857ef4b72980d408ddaeca811 : SUCCESS in 28m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/944fcaa6f9e545c2bc448b898d5170bc : SUCCESS in 10m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/29d02375240e4d41b750944e0b80bd6b : SUCCESS in 1h 07m 37s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/12cf2be550304cf7828183adbdad8f22 : SUCCESS in 1h 27m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a020b6fabe054d12aae03bdb21423911 : SUCCESS in 1h 13m 58s","accounts_in_message":[],"_revision_number":2},{"id":"72c760e16dd2f5ceabe9d4c042a50d2051096d70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 00:37:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b1681d3043a86823f4e7a38eb0aa654fd9a2cd51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 00:39:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a05ef7758d4487eb7f2849342821a9e : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d982a68446ad9b9c432ab2788900993aa21e135a","revisions":{"32fc70a733369e0b81524d832cb3509a4648704f":{"kind":"REWORK","_number":1,"created":"2020-03-20 11:37:13.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/97/714097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/714097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/714097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/714097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/714097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/714097/1"}}},"commit":{"parents":[{"commit":"a4d63e18cfda250faac23bf29a1882c16fb0cb13","subject":"Merge \"Mark TestVirtualPorts tests as unstable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4d63e18cfda250faac23bf29a1882c16fb0cb13"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-20 02:16:20.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-20 02:16:20.000000000","tz":60},"subject":"Finish migration to new engine facade in test_db_base_plugin_v2","message":"Finish migration to new engine facade in test_db_base_plugin_v2\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1b95d53f79c2c2b56a119c20017c3bf7fbc8550d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32fc70a733369e0b81524d832cb3509a4648704f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32fc70a733369e0b81524d832cb3509a4648704f"}]},"branch":"refs/heads/master"},"d982a68446ad9b9c432ab2788900993aa21e135a":{"kind":"REWORK","_number":2,"created":"2020-03-24 10:30:49.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/97/714097/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/714097/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/714097/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/714097/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/714097/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/714097/2"}}},"commit":{"parents":[{"commit":"a4d63e18cfda250faac23bf29a1882c16fb0cb13","subject":"Merge \"Mark TestVirtualPorts tests as unstable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4d63e18cfda250faac23bf29a1882c16fb0cb13"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-20 02:16:20.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-20 07:55:57.000000000","tz":60},"subject":"Finish migration to new engine facade in test_db_base_plugin_v2","message":"Finish migration to new engine facade in test_db_base_plugin_v2\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1b95d53f79c2c2b56a119c20017c3bf7fbc8550d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d982a68446ad9b9c432ab2788900993aa21e135a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d982a68446ad9b9c432ab2788900993aa21e135a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
