)]}'
{"id":"openstack%2Fneutron~714397","triplet_id":"openstack%2Fneutron~master~I008f3ce08346520b415f6db41419a69d472350a9","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"I008f3ce08346520b415f6db41419a69d472350a9","subject":"Enable back mac spoofing and disabled port security tests","status":"MERGED","created":"2020-03-23 08:24:27.000000000","updated":"2020-04-06 00:35:54.000000000","submitted":"2020-04-06 00:33:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714397-1586133190264-40a9e122","meta_rev_id":"d1fe5c48ebc148efc21d5badf430ac6e6dd327f3","_number":714397,"virtual_id_number":714397,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-06 00:33:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-03 21:55:25.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2020-04-03 20:14:32.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2020-04-03 20:14:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-04-03 20:14:32.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2020-04-05 19:58:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-24 10:15:47.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-03-24 11:10:16.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-04-03 21:55:25.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-04-05 19:58:01.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-04-06 00:33:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"15276f5b5815307dfd569047f2799148acad2c7f","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-03-23 08:24:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18257405b87a23b289a46a245e3af07ba3b4830a","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-03-23 08:24:53.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"125196ea61c4c7a421dda053c52205c04a390676","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-23 10:29:13.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/714397/2 : FAILURE in 2h 03m 37s","accounts_in_message":[],"_revision_number":2},{"id":"ca3b51969dff28c04cd17b2e05dd2b3ffd9012ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 11:30:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/34e6fafa1f164dbe874646995c92b7f1 : SUCCESS in 1h 07m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a17631da3e0f47deba66ae5a58b9bfb6 : SUCCESS in 33m 24s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/985bb870023145debbeb0a3bfbf8602b : FAILURE in 2h 05m 24s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b5649cdd3dff4f03be6caae72eacfaa4 : SUCCESS in 1h 27m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cd1fadf0014d4e5cb29f7e110d139f1d : SUCCESS in 1h 06m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/01b85127f76a410cad76625f70230230 : SUCCESS in 1h 27m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c10fb784cf884ea7be804288a82c961b : SUCCESS in 38m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7d8697ed80243079ec842d76dc47414 : SUCCESS in 23m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e215127fe6334b3ca85ffa38d4909a90 : SUCCESS in 11m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf79f17567954e31a7920c45b05bec2e : SUCCESS in 28m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e41228dc75994f75bfb8f77b29c2d89f : SUCCESS in 22m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb27352b9f71427fbdb5c4d5ef333a11 : SUCCESS in 30m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a6d1b6875f94bc08e0092a9b5b3b885 : SUCCESS in 8m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/88a45ed8e7ba44f28f5e9318dfdec4d6 : SUCCESS in 52m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5de17c05d0104e0289a0f73a41c6856d : SUCCESS in 1h 16m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/321892a4b77f4d65bc2ee557cb419d21 : SUCCESS in 2h 57m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d97cd6cff4a1483299ac651c9ed9670c : SUCCESS in 1h 15m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/50272c7cb2a5496583aeff25940da5a2 : FAILURE in 1h 37m 40s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e1c169309a0d44caa2f36cacd0815010 : FAILURE in 1h 06m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a78451f02b764f16ae903223c8d276b7 : TIMED_OUT in 2h 07m 23s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8e50e6c225794748b75b5879fb9c5d00 : SUCCESS in 1h 42m 08s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2edb1c4d198a4e2bb0c25a8fa1ca21f5 : FAILURE in 1h 58m 14s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a22923f8ad594c34aa9c6379eb1f05de : SUCCESS in 2h 04m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/96e593b1fdc049f2ab7247e197ef9f93 : FAILURE in 1h 35m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/efac37f9d6a649cab45c4ca16d84e21b : SUCCESS in 2h 44m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e4680263ab41498ea358c046afd246aa : SUCCESS in 2h 01m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9845708e29fb4b3fb1cad36c8de7f82b : SUCCESS in 2h 58m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7991d8e856642638c78cef12c71ae84 : SUCCESS in 1h 21m 43s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/bf5879f5fd834ded86493fd42897c867 : SUCCESS in 2h 10m 08s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2cef7d0345674cbc8d1b4b6f33a8ad33 : SUCCESS in 50m 35s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/271c96ad6b6747d890d52d0899b758b1 : SUCCESS in 2h 07m 12s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/bfc47448a3d442fdb7700e9e4ce9f978 : SUCCESS in 1h 25m 54s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/94097fa4e0e44c71afe872589324b440 : SUCCESS in 2h 56m 22s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/8043bfe0ce814beb8c1d6302b1d620f1 : SUCCESS in 2h 38m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"73fa813be87cd5cf50ea06f766f6fa37d0157ab9","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-03-23 11:42:37.000000000","message":"Patch Set 2:\n\n\\o/ All ovn-related tests are green!","accounts_in_message":[],"_revision_number":2},{"id":"b8e1297c08f5f74a0cbf186154a0e71a537e4ded","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-03-23 11:42:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3c30129dc30517f12efb1e1ad25e9f9fc74d258e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-23 13:28:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/714397/2 : SUCCESS in 1h 44m 58s","accounts_in_message":[],"_revision_number":2},{"id":"125179f8a18ddd8f0e3d02e9947877c95bf8c59e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:35:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6a9a1793aba749649bf9fa906e6a5536 : SUCCESS in 51m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/45798cd8371c4dd385f7edc2f083fcfb : SUCCESS in 33m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/5724ba6307734336bdb82cdf75e6ad0f : FAILURE in 1h 53m 15s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0e5cc0dfff564c16bc920c418dad3915 : SUCCESS in 1h 16m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ad104fec22774552bfd4cc48c4c79e74 : SUCCESS in 1h 10m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d16137287dcc47bdad16afd6a88c0a88 : SUCCESS in 1h 15m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a157b53cc5f4be490883621204e1c28 : SUCCESS in 32m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/871327cac9e74f9d92a4e77052c5f9c7 : SUCCESS in 35m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b68f4f229e3428da223006478e14c32 : SUCCESS in 14m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f72a43db28c342e2abda2a59bc8b0a58 : SUCCESS in 24m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30702f3932b344178ad672657492400b : SUCCESS in 30m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/868c4cc9f33b4058a887a3c43d65d63a : SUCCESS in 24m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d07841f1cbd41a88730296084a080b2 : SUCCESS in 8m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e11b164571e94c48bde43d9ab843576f : SUCCESS in 54m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/456e11ef82d14b88adad291a2a8d278f : SUCCESS in 1h 25m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/282b0b83919a4c6a830e24a0d0d6986c : SUCCESS in 2h 03m 26s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a344334e3ea04b7f8f0f1e9d1b33ae8a : SUCCESS in 1h 23m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c4c0f8d0eab4e358ea6a212bd33f786 : SUCCESS in 1h 08m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8a74d5c7a58e4f7f904cbf68c6ff1c37 : FAILURE in 1h 14m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c896d629141646e1909ffae26bb62d90 : SUCCESS in 1h 21m 59s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/29ae4e2b647740d399096449a3a016a2 : SUCCESS in 2h 05m 49s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/61d8127b1aa04346b49fa21323951f5c : FAILURE in 1h 18m 10s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/55d956adda194013811edba7afad8edb : SUCCESS in 1h 08m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ea98886d4a864130a5beae50776631e9 : SUCCESS in 1h 11m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c5a020ab3dfa4122a7c1193da2a54f29 : FAILURE in 2h 48m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/18e54fe19ddf4102b3ca28538f53444d : SUCCESS in 1h 55m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/279185fcac7f4717b516bd3e0c7412e6 : SUCCESS in 1h 54m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94b246c458d343a58a2c727b68787035 : SUCCESS in 1h 04m 07s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c452a273bdbf4208ab18933231b7f3dc : SUCCESS in 1h 15m 57s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/46a8d709d63243cfb9157195474f28b8 : SUCCESS in 59m 09s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/988cfdacde234d5a9e6c65e2d34304c1 : FAILURE in 59m 11s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e569b87ab7244ec9916f4d0391548613 : SUCCESS in 1h 14m 39s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/30baef30fb984d75aa79d5ece8a6230f : SUCCESS in 1h 55m 16s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/4998ec23f7d4459e813b856a9e9d53b9 : SUCCESS in 2h 14m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2fda2e680b696cbc9961a01d3ff7fa7630c780e0","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-03-23 17:07:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0bc13c4eaf77765249987d5ec76a31d34cee06a6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-23 18:50:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/714397/2 : SUCCESS in 1h 42m 42s","accounts_in_message":[],"_revision_number":2},{"id":"8451d4f6a70d98cbd569eabbdcddcb398231e486","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 19:46:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9274f40af71349dc9d85c8bcdfdd203d : SUCCESS in 46m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1cf8079086a14e109d7703809b60faae : SUCCESS in 30m 49s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/1d243cc07cd343028879a66ae547ec00 : FAILURE in 2h 04m 11s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/97829e55f0304f46bf888aa18b679e41 : SUCCESS in 1h 27m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/31ad06699d45416e97268df0b96aae6d : SUCCESS in 1h 24m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/795e77e856164700969a6a400d633059 : SUCCESS in 1h 28m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5dec09838bb45a19ba89dc0e468d16a : SUCCESS in 31m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba366c6186a842aca2d3a87d2f0664c6 : SUCCESS in 28m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/154d18ade3d84df49d78a7f9e801a34c : SUCCESS in 13m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7be29518a1f44f28070e8ca0a23781b : SUCCESS in 31m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61c966e6e7b04b5ea0640231eded2577 : SUCCESS in 28m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/036192adb11d440c9ec1ea200e75b66c : SUCCESS in 25m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c2c12870b584303bd58475156ab2f07 : SUCCESS in 8m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ae2620c7dcdc436cbfecdbe76a480797 : SUCCESS in 50m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fc4e84c05da74db1abf552323b44b8fb : SUCCESS in 1h 22m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fed6437a01c14770833aaa3838f46794 : SUCCESS in 1h 53m 26s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/739f0ed055ee46b98811da12f03718f7 : SUCCESS in 1h 12m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a35edb44e534e0889a80389d0b6e36d : SUCCESS in 1h 04m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/aecf7b8f2a854e63a2723e012424c033 : SUCCESS in 1h 22m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/027008bd45144e19b5656818cc3f3393 : SUCCESS in 1h 12m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/60f7c19fdea240c1b277709477bd4e38 : SUCCESS in 2h 04m 13s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f1f0dd117d284da4a6756473a7c5e77a : FAILURE in 1h 30m 23s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c1550d81f54e4282b3aa4ae9d3484037 : SUCCESS in 1h 18m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fbdb9fa969604113bd84e910823a7065 : SUCCESS in 55m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/660af1675d304773b0744b007060a5ca : SUCCESS in 2h 33m 47s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c74cd67ab75241fc82a89dffaabb9f6d : SUCCESS in 1h 45m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6ceca3dd2288480d91a22d16a5e2a850 : SUCCESS in 1h 42m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/406df492e39f48439bf15a732bb88a64 : SUCCESS in 56m 38s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/751ba806eda8425c8e4abe0b125bac5c : SUCCESS in 1h 19m 07s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2585524793c49f1a30e10722b2638bc : SUCCESS in 52m 25s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/71a9fc3b154e428a80969eadff180906 : SUCCESS in 1h 22m 38s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7576294bc0284f3ab93f77f8e9a110db : SUCCESS in 1h 23m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ae6d2fda85314fb59f07e829acbbefac : SUCCESS in 1h 59m 09s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3ff9cdc5498b42799e0c9a48eec174a6 : SUCCESS in 2h 30m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b1fc4c26c7bc5a21b46c886d67243d5cd9e7f8ea","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-03-24 10:15:47.000000000","message":"Patch Set 2: Code-Review+2\n\nGood stuff, thank you!","accounts_in_message":[],"_revision_number":2},{"id":"f8c3f3f7112c2c3bced926c0713fbecfabeed88a","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-03-24 11:10:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks Maciej! Great job","accounts_in_message":[],"_revision_number":2},{"id":"a21b420a011eafd7ea33c8a63034c3744f58e873","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 11:16:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dcccd1095510edb1dd70d2250fe2c60402338b63","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-24 11:37:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"834c718a6a7879de14a63efc5e76e687651a63aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 14:26:22.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"9699fb08be0f9443dc98c983df6e0a8ececf28cc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-25 11:41:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"aeb05af4663bf8f279992a3b230473571a061fea","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-25 11:44:59.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/714397/2 : FAILURE in 2m 36s","accounts_in_message":[],"_revision_number":2},{"id":"5a5986dec7a8251107658fc1a9e74afc10ca15b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 14:12:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6836b03739d74906862e71f346e442cf : SUCCESS in 53m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f583362b61d34e81a68d14578c81165c : SUCCESS in 34m 18s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/7dce9fb223254a06bde04cd076c7c4a6 : FAILURE in 2h 06m 52s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/61e6c4330d454d259f354ba6ca6eb362 : SUCCESS in 1h 10m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b9ecab4164af433dadbe1c187644fcf8 : SUCCESS in 1h 20m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/50a91a8fa2a749fca82613139f258792 : SUCCESS in 55m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/696235acb4ab47e59e5ce851720005a1 : SUCCESS in 35m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a1c5b4df60a418f80a96a39bef34f15 : SUCCESS in 23m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7732e126125d47b693097aabfb2e3d8c : SUCCESS in 14m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/919a46cbca1f451d9171467da846cb30 : SUCCESS in 23m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0942d1ca331d447e8969105bbff8e50a : SUCCESS in 28m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c2e8db78ca24919bf846d4fea1516aa : SUCCESS in 30m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84adb8eba10e498dab0c2444304b49f7 : SUCCESS in 12m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fda79998cd9d4401832f79d5877baa63 : SUCCESS in 49m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c5f2e4101f634244add68758b98a3e3b : SUCCESS in 1h 10m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c6e77ec600e04820ab7673497302f622 : SUCCESS in 1h 45m 16s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/14456c3136154be5944d69ed044ca9b8 : SUCCESS in 1h 14m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e15a66de44e4453398a052538e280508 : SUCCESS in 1h 10m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4eaaf2b56665437b96c07ac2c1f538da : SUCCESS in 1h 03m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a25c35b641934f4d97ceb80620c41f19 : SUCCESS in 1h 08m 20s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a6984edf5dfc4cb39feae4423ffe360b : SUCCESS in 1h 39m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fd0b2059c4b64ad8917bd9af3896d044 : FAILURE in 1h 30m 48s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/850198090b53411da0fc5f2cccffcda2 : SUCCESS in 1h 17m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6203ab391206401da5c74ea995e62007 : SUCCESS in 56m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8546efc99c6c4f32895e391542ee50dd : SUCCESS in 2h 21m 11s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6f93310bb25e432d9199defef70c50a2 : SUCCESS in 2h 08m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d52aac4a74ce4785b2c7d52d617ebf22 : SUCCESS in 2h 06m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c39eb73883fa4b119e30261d40ad19c1 : SUCCESS in 1h 00m 36s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/580c7625d4b84fc3b10a1a13504ab93e : SUCCESS in 1h 32m 07s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/00c76c4c3c314a7abf279ef086980f38 : FAILURE in 45m 48s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a5e8b668422c47a495969fb8773e90cc : SUCCESS in 1h 14m 55s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/af23222f1c5247d0b40afb96953cce1b : SUCCESS in 1h 19m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/72e68789a49c45a1baaf24f3ca227d14 : SUCCESS in 1h 57m 12s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/4ef5b2aef8d44e2ca3bb7c6938bf1782 : SUCCESS in 1h 54m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f53629b4c9c2ec5f0ec0bf6e46101d8355863fad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 16:28:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1f4a0e678a9024123634fbbc3876b040c7b472b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 19:43:35.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"05e39789f1592bd50d4521270ba4af4903bff8a3","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-03 20:14:32.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"01b43da5b264adbca2bb80873f0e8ee8845775db","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-03 21:55:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/714397/3 : SUCCESS in 1h 40m 08s","accounts_in_message":[],"_revision_number":3},{"id":"3ecbd3cfdf3751fb68730182b7a4ec6a88f996a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 23:40:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/95defb92dfe1436285854d215d97e71a : SUCCESS in 53m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7dc4eb37d6cf4b84b7b86b1043b6357f : SUCCESS in 41m 34s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/d933c78df1c84eaf94ed6196f13e2a4e : FAILURE in 1h 56m 40s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/337adece81c14f229e5bd549eda18adf : SUCCESS in 1h 02m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/388396a761904b83be27d2198e3df812 : SUCCESS in 1h 10m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6e26b66cdd31413e992c660714018270 : SUCCESS in 1h 02m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4182297c5ce47449a3f2b9cb05679f1 : SUCCESS in 31m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb1e72f8c832401cbec5d230b2d85bf2 : SUCCESS in 29m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d75738444f0b4e44a546a15f991cf1f3 : SUCCESS in 16m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c68350b7634f4778ba51ffe7df977d4d : SUCCESS in 29m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6e6a9150e7a4da496519657a39ad81f : SUCCESS in 32m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31a5d8f718cc4b0890edc8acb1e228c7 : SUCCESS in 25m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e28af54fd4e4c459c826b23180a873f : SUCCESS in 8m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/614cccaafd654bcf928f65935d53b418 : SUCCESS in 54m 16s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b66e8077ea904721b22ce5ef076b5711 : SUCCESS in 1h 10m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/322dc9eeaf004f3cbfd14489b8f0026a : SUCCESS in 1h 53m 38s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/636980bc9f344ae786696a0ef5e61565 : SUCCESS in 1h 39m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b8d457bfadbc476fbc39b22f3ed5955e : SUCCESS in 1h 03m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7ae8ff2c363a48b9ab96298b60d0c685 : SUCCESS in 1h 15m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8d656245ae084c4bba7c5952cb8985c5 : SUCCESS in 1h 25m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fe8e5f5e8cd04193932e1e85b7c79e1a : SUCCESS in 1h 49m 01s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/16c8849e30b34741a95158216a36a3e0 : SUCCESS in 1h 40m 08s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1f7115a72f7a46ed925fbaa6c6874da8 : SUCCESS in 1h 11m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/359e2de7f0c44b0c9485019e966c3abc : SUCCESS in 56m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/515ad3ff03424d57a97c50cde6630e9d : SUCCESS in 2h 21m 46s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/332fdea9d0c2401fb2a24f32c16f9184 : SUCCESS in 1h 40m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b0d3e243bc3f444f8c2eafb91dc68592 : SUCCESS in 1h 54m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb323475c9e3425c88a8e5db62370138 : SUCCESS in 1h 01m 25s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9f42a5a4ee9540979636d557468daf7a : SUCCESS in 1h 14m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/02caa983640a4f639d43276ba7b6474a : SUCCESS in 49m 41s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3c2b7003ca394563abf4e8cd8a14f7a7 : SUCCESS in 1h 14m 47s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/096c0d1d46324b2180fa80b79563c68b : SUCCESS in 1h 08m 11s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9ee5a8289d234811bf07b8136368f6f4 : SUCCESS in 1h 30m 00s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/653a05ddd2fb4f84b0cf075b0a90c41f : SUCCESS in 1h 58m 03s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/79bd85a9f1534c5cba7388d76772bbad : SUCCESS in 2h 40m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e508616070594ab4af9e6886f51ca3d6 : SUCCESS in 2h 09m 59s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/a6824adf129d4c58b1ece0decb06f901 : SUCCESS in 2h 01m 31s (non-voting)\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/docs/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/shared/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/shared/lib/python3.6/site-packages/astroid/interpreter/_import/spec.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/dns/hash.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/ovs/db/idl.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/unittest2/compatibility.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/sqlalchemy/orm/relationships.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/neutron_lib/context.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/neutron_lib/worker.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/shared/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/shared/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/neutron_lib/context.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/paste/deploy/loadwsgi.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/openstack/config/cloud_region.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_messaging/_drivers/pool.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/ovs/db/idl.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_policy/_checks.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/neutron_lib/context.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/unittest2/compatibility.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/neutron_lib/context.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/sqlalchemy/orm/relationships.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/oslo_context/context.py in the commit\u0027\n  Job neutron-functional-with-uwsgi: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-functional/lib/python3.6/site-packages/neutron_lib/context.py in the commit\u0027\n  Job neutron-functional-with-uwsgi: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-functional/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py in the commit\u0027\n  Job neutron-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-functional/lib/python3.6/site-packages/neutron_lib/context.py in the commit\u0027\n  Job neutron-fullstack: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job neutron-fullstack: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job neutron-fullstack-with-uwsgi: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job neutron-fullstack-with-uwsgi: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job neutron-fullstack-with-uwsgi: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/ovs/reconnect.py in the commit\u0027\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_policy/_checks.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/unittest2/compatibility.py\n  Comments left for invalid file .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/shared/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file neutron/objects/base.py\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_daemon.py\n  Comments left for invalid file .tox/shared/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/eventlet/patcher.py\n  Comments left for invalid file neutron/tests/unit/test_auth.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/openstack/config/cloud_region.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/unittest2/compatibility.py\n  Comments left for invalid file .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/eventlet/patcher.py\n  Comments left for invalid file neutron/common/eventlet_utils.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/ovs/db/idl.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/dns/hash.py\n  Comments left for invalid file .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/ovs/reconnect.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/neutron_lib/worker.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/shared/lib/python3.6/site-packages/astroid/interpreter/_import/spec.py\n  Comments left for invalid file .tox/docs/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_messaging/_drivers/pool.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/sqlalchemy/orm/relationships.py\n  Comments left for invalid file neutron/agent/common/ovs_lib.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/dsvm-functional/lib/python3.6/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/sqlalchemy/orm/relationships.py\n  Comments left for invalid file neutron/policy.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/dsvm-functional/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/ovs/db/idl.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/paste/deploy/loadwsgi.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py\n  Comments left for invalid file neutron/tests/unit/agent/ovsdb/test_impl_idl.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_context/context.py","accounts_in_message":[],"_revision_number":3},{"id":"2cd69f6423931f4d1d11a50570c35448dbc45005","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-05 19:58:01.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"be42e036df558ca1ae1dbd69bef70e8a2e849282","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 19:58:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"778878e40ce97898f5cf6b4d9c7ab4cfae429406","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 00:33:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0ccd2a9511fb487daf1aa9f1ec9dade6 : SUCCESS in 50m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a64336b7076546bf8fa5d1247edbd3e7 : SUCCESS in 1h 16m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9f0fcf3cdfa74210b9018d16d6b270fc : SUCCESS in 1h 16m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6141a2ed1fa944bbae040f3c50059d5c : SUCCESS in 1h 02m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87289db9a4c74002ae81af61cec92995 : SUCCESS in 27m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29b0d59d819b41119ffff3db5d026695 : SUCCESS in 13m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3357899a2f44d2086e811540da645fb : SUCCESS in 29m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc1ea101876d4d06aa85d97d1774ff4c : SUCCESS in 30m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39528262a31c403fae16b7f052fc99b7 : SUCCESS in 9m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a80af8dec819449784177571bc03a3be : SUCCESS in 53m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/075e524a8bed4c0bbcced0a0d57897ad : SUCCESS in 1h 15m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/18bc26a0d57a46728a2be4533af3f529 : SUCCESS in 1h 09m 20s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d8bf5b9cd6b544df97f83d8270071719 : SUCCESS in 1h 13m 49s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9b1c58ee0a394a2a81974457d787868d : SUCCESS in 1h 19m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed05b6e64dca435aa9c9ca73cf393019 : SUCCESS in 1h 06m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/16692c6e46fc4863a1b42e645a30867a : SUCCESS in 1h 01m 39s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9366854eb08745558340ef305ce8c6bc : SUCCESS in 2h 11m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c15307a9e23436c88e5c9ce24ff5add : SUCCESS in 1h 05m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/81663a7faf5641198f9f84584084ba91 : SUCCESS in 2h 04m 59s\n\nWarning:\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file neutron/agent/common/ovs_lib.py\n  Comments left for invalid file neutron/objects/base.py\n  Comments left for invalid file neutron/tests/unit/test_auth.py\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_daemon.py\n  Comments left for invalid file neutron/common/eventlet_utils.py\n  Comments left for invalid file neutron/tests/unit/agent/ovsdb/test_impl_idl.py\n  Comments left for invalid file neutron/policy.py","accounts_in_message":[],"_revision_number":3},{"id":"59c36e6c387a5ad93c0a3e34a2c5265346fcf14e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 00:33:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d1fe5c48ebc148efc21d5badf430ac6e6dd327f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 00:35:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9311c8147e864152a743269b3baeb2d7 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"02a74c3d2ac3febff3b458bbfb6c16f71e0cb3fb","revisions":{"864f4f97c1acb036648d6e625042d2bffc03ac4e":{"kind":"REWORK","_number":1,"created":"2020-03-23 08:24:27.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/97/714397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/714397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/714397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/714397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/714397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/714397/1"}}},"commit":{"parents":[{"commit":"b319d64388cd4e9c1a26247aaf7da0556bcd476e","subject":"Merge \"Run fullstack security group test always serially\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b319d64388cd4e9c1a26247aaf7da0556bcd476e"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-23 08:21:23.000000000","tz":60},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-23 08:26:36.000000000","tz":60},"subject":"Enable back mac spoofing and disabled port security tests","message":"Enable back mac spoofing and disabled port security tests\n\nWe can enable back the mac spoofing test because\n[1] has been merged [2]. Because of [1] we can now\nrollback [3] to make port security tests with\ndisabled security works.\n\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d1805709\n[2] https://patchwork.ozlabs.org/patch/1258152\n[3] https://review.opendev.org/#/c/708852/\n\nChange-Id: I008f3ce08346520b415f6db41419a69d472350a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/864f4f97c1acb036648d6e625042d2bffc03ac4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/864f4f97c1acb036648d6e625042d2bffc03ac4e"}]},"branch":"refs/heads/master"},"5b28e327f9bf65d1fde8f5e6ce57eda6a74dd013":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-23 08:24:53.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/97/714397/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/714397/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/714397/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/714397/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/714397/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/714397/2"}}},"commit":{"parents":[{"commit":"8772d712b583a70e2026a0be9b654ef8719cba0d","subject":"Revert \"[OVN] Set \u0027unknown\u0027 address properly when port sec is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8772d712b583a70e2026a0be9b654ef8719cba0d"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-23 08:21:23.000000000","tz":60},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-23 08:24:53.000000000","tz":0},"subject":"Enable back mac spoofing and disabled port security tests","message":"Enable back mac spoofing and disabled port security tests\n\nWe can enable back the mac spoofing test because\n[1] has been merged [2]. Because of [1] we can now\nrollback [3] to make port security tests with\ndisabled security works.\n\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d1805709\n[2] https://patchwork.ozlabs.org/patch/1258152\n[3] https://review.opendev.org/#/c/708852/\n\nChange-Id: I008f3ce08346520b415f6db41419a69d472350a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b28e327f9bf65d1fde8f5e6ce57eda6a74dd013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b28e327f9bf65d1fde8f5e6ce57eda6a74dd013"}]},"branch":"refs/heads/master"},"02a74c3d2ac3febff3b458bbfb6c16f71e0cb3fb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-03 20:14:32.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/97/714397/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/714397/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/714397/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/714397/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/714397/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/714397/3"}}},"commit":{"parents":[{"commit":"da63adc96bf9f8eafd66d0aeea5d6f8f787420ed","subject":"Merge \"Revert \"[OVN] Set \u0027unknown\u0027 address properly when port sec is disabled\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da63adc96bf9f8eafd66d0aeea5d6f8f787420ed"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-23 08:21:23.000000000","tz":60},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-04-03 20:14:32.000000000","tz":0},"subject":"Enable back mac spoofing and disabled port security tests","message":"Enable back mac spoofing and disabled port security tests\n\nWe can enable back the mac spoofing test because\n[1] has been merged [2]. Because of [1] we can now\nrollback [3] to make port security tests with\ndisabled security works.\n\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d1805709\n[2] https://patchwork.ozlabs.org/patch/1258152\n[3] https://review.opendev.org/#/c/708852/\n\nChange-Id: I008f3ce08346520b415f6db41419a69d472350a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02a74c3d2ac3febff3b458bbfb6c16f71e0cb3fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02a74c3d2ac3febff3b458bbfb6c16f71e0cb3fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
