)]}'
{"id":"openstack%2Fneutron~714716","triplet_id":"openstack%2Fneutron~master~Ia253cc4d85261ce1535f4d27b3da91275d879903","project":"openstack/neutron","branch":"master","topic":"bug/1868137","hashtags":[],"change_id":"Ia253cc4d85261ce1535f4d27b3da91275d879903","subject":"Make OVN driver validate Geneve max_header_size","status":"MERGED","created":"2020-03-24 17:00:20.000000000","updated":"2021-09-03 02:10:11.000000000","submitted":"2021-09-03 02:08:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"714716-bug/1868137","meta_rev_id":"fd52d95f26c939502af8a13c17ed941f158f02f0","_number":714716,"virtual_id_number":714716,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":1,"date":"2021-09-02 09:36:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-03 02:08:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-09-02 16:58:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2021-09-02 19:15:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-09-02 11:41:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-02 11:49:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-09-02 16:58:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-24 18:25:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-06 08:09:28.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-04-07 08:47:59.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2020-04-07 09:23:59.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-07 10:03:09.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-04-07 11:27:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-01 12:59:22.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-09-25 15:42:57.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-09-01 14:05:23.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-01 14:23:41.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2021-09-02 11:49:41.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-09-02 16:58:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-09-02 19:15:16.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"b8bc728a0aa5078f50d56dce92b09b23ff3f5e1b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-24 17:00:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"764911ac6a009f3ffc1f103c8b8ee7c55e5f24b5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-24 18:17:18.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/16/714716/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c68d7009ad4ac8fee1865e0ea1531071e0a58c16","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-24 18:25:52.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0ead045a81ac2473e90cc3111367db9915b3a779","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-24 18:51:27.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"80c93a1873a27291c3d9213b899e5f482a0d3878","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-24 20:22:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/714716/1 : SUCCESS in 1h 49m 40s","accounts_in_message":[],"_revision_number":1},{"id":"db71c5a66e285b0e04a53308ac405a478d58473c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 20:22:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6bb22bc5a2b642dc9bcb6774ea20be43 : SUCCESS in 52m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9efd2c53510c4d4b8354a667b637f380 : SUCCESS in 35m 36s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/f48dbaed1e3f45b2a941a751a4cc2c45 : FAILURE in 1h 59m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d6aebba6171e4dc58a8d4c1bf3945c95 : SUCCESS in 1h 15m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d0e7181e7c62420e8dc69437c26ece2c : SUCCESS in 1h 09m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4e2390dc02de4e7ab308b7cba31284dd : SUCCESS in 58m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db99393672e14e33b684eda9ecebd706 : SUCCESS in 34m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa62bc9b46914b2ca21b5b873ff58739 : SUCCESS in 28m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17d088492da54b8aa51ad07513c8ca3d : SUCCESS in 16m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b01b78f5bf754df893e6d89f8cdd6d93 : SUCCESS in 42m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/06f80be41d924cf2bcb919006599067b : SUCCESS in 27m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ba584a3ba2f417b8271c60720bd542a : SUCCESS in 26m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21732a38b05c417592a67252ce30490f : SUCCESS in 9m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f8e29fffc55a40d982c1e0929b891cca : SUCCESS in 1h 07m 29s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/75ea419f69664dc984d8904e1025277a : SUCCESS in 1h 16m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ddc743849f3c44eebafe992475009f9f : SUCCESS in 2h 27m 33s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/af2620e4f402449ab7aa690242c13838 : SUCCESS in 1h 11m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2154ce732b1642cc92289ed0b33b8268 : SUCCESS in 1h 12m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/72302f0134d342f3b10fbbe9373de48f : SUCCESS in 1h 10m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3ccaf52a647e4141bf01aca588ee518e : SUCCESS in 1h 17m 18s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d15f5e7fda2c4868a38cc22b6e14c4c0 : SUCCESS in 1h 56m 21s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fb12a917d69146f094bbe6507e1180e7 : SUCCESS in 1h 49m 29s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/117d58f459764c53b376fcda874aa366 : SUCCESS in 1h 20m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a6ee395df2574dc2bc6e6c79dfb930fd : SUCCESS in 56m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3d9cfc32fdb14157afe635ca5b8c4dbe : SUCCESS in 2h 28m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7bd503589fad4d9d97d7974172a93de9 : SUCCESS in 2h 01m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/edc673a47e15492ebbc8f92095db18d8 : SUCCESS in 2h 05m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e447e68704d4e3d94ee32ab77aec0dd : SUCCESS in 1h 08m 52s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7013e54c2d2840bcbda04baca014e363 : SUCCESS in 1h 25m 50s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ebf318dc253a4755849ff8432f844f84 : SUCCESS in 51m 39s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/490fdeedcd09461da2f363a6a2cd0d60 : SUCCESS in 1h 22m 51s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5914860339d145f2b4add39e63a28bb8 : FAILURE in 1h 04m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c32bdcd661044218b0797cc0121ce0aa : SUCCESS in 2h 05m 01s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/05503e7ddd5c45feb36fa37cb1fdf137 : SUCCESS in 2h 37m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73f82bcb544ea155da1fd7dac5e830c93af4b88f","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-03-26 13:02:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nThanks for this patch.","accounts_in_message":[],"_revision_number":1},{"id":"b6259c09144c5741f6d68227aa2ac62b92544c70","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-04 15:10:15.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ac6092b242deb8bbe23516ddc2afc37fb4c49ecf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-04 15:45:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4392361ff5c7da91899c5d93214ce42ffb9caf52","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-04 17:24:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/714716/2 : SUCCESS in 1h 39m 05s","accounts_in_message":[],"_revision_number":2},{"id":"f6782c504cf8c6432eca2140b1f8b89fb3e2f923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 20:34:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a41c5dfab8b84d5f979787c156fa27de : SUCCESS in 47m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d04caf49b14e46c28e12e78610291e05 : SUCCESS in 42m 38s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/4c7947b2e826437e825fcedf61b6cdcd : FAILURE in 1h 54m 23s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9fc741eaf6f4486c90db05f83a28163a : SUCCESS in 1h 16m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ecbed1bbc6fc4f4083016777f1391d35 : SUCCESS in 1h 21m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1fc2a37b8d154fd395e996a295b36503 : SUCCESS in 57m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ecc36c5621b403e8ebc55c723c792b3 : FAILURE in 17m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/917b0e1b82ac4348a656d41a9fb47b41 : FAILURE in 21m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/624e7ca2284948cbbf990a0e9ef0c7dd : SUCCESS in 12m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67a2c02276b046ec8e81d9875306d77d : FAILURE in 21m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/794640b37b704ca1b63d8bef4018bb12 : FAILURE in 17m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f520b36c35ab4a6b9cd0c5365ae3a6ec : FAILURE in 19m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3babc1c504f4b6e8df96540bd1b6f29 : SUCCESS in 9m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/05f9931cb61e471085024f4044365746 : FAILURE in 39m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e4269e9b14734433b6c45f5b771e22fc : SUCCESS in 1h 15m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b28fbc216a5e4fa5b20ee4c3be3b7801 : FAILURE in 1h 47m 42s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/251a8745fd5e4918a882d49a8132ef68 : SUCCESS in 1h 18m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d4f7d82e2a34ab7af936b0723c96f09 : SUCCESS in 1h 05m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/21c83e2f550b4051a8b7c075faf93842 : SUCCESS in 1h 09m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d23f94cab8324a8a8a467f19909a6446 : SUCCESS in 1h 15m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c3745d60f9e9429d9d7ef30c6b8055d7 : SUCCESS in 1h 39m 23s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/008ee7755c2b46128e28a6389897d2e3 : SUCCESS in 1h 38m 19s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/72d4981cebd746109484a71f3316cb3b : SUCCESS in 1h 21m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/53292a173bde4f02b0bf4685e009765d : SUCCESS in 51m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/69d6d8ce2a0e4eb8a8b38091150221d8 : SUCCESS in 2h 26m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/02152819c5d9459b9fff96d333c4e205 : SUCCESS in 1h 45m 16s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/82761b33f06d4b33a0981874e1393d80 : SUCCESS in 1h 56m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a709382c3b384c48b76659b0d61a4ef0 : SUCCESS in 55m 29s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6ba912fc3c034287af701c6df723c694 : SUCCESS in 1h 13m 31s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/99cef7e4944f40bab4084c786e2ef839 : FAILURE in 39m 43s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0d1ec2520484484caa6f49ab8dda7d11 : SUCCESS in 1h 33m 18s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/208be1b274964e80bdcadaa4ffe6d0cc : SUCCESS in 1h 21m 29s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d5d61c13b2cf4c0a83e7be457a6246a0 : SUCCESS in 1h 43m 01s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/987eac7f16074184bb8797783e501951 : SUCCESS in 1h 50m 46s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a0def04042a1437e93b939aea31b8642 : SUCCESS in 2h 46m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fa7b08fdc85d4c34952170337ef38c36 : SUCCESS in 1h 53m 45s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6023cec50bf147caaf29fe075a307bbb : SUCCESS in 2h 25m 43s (non-voting)\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/docs/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/shared/lib/python3.6/site-packages/astroid/interpreter/_import/spec.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/shared/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/shared/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/shared/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/neutron_lib/utils/helpers.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/neutron_lib/worker.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/dns/hash.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/sqlalchemy/orm/relationships.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/neutron_lib/context.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/unittest2/compatibility.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/unittest2/compatibility.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/sqlalchemy/orm/relationships.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/neutron_lib/context.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/neutron_lib/context.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/openstack/config/cloud_region.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_policy/_checks.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_messaging/_drivers/pool.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/paste/deploy/loadwsgi.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/neutron_lib/context.py in the commit\u0027\n  Job neutron-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-functional/lib/python3.6/site-packages/neutron_lib/context.py in the commit\u0027\n  Job neutron-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-functional/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py in the commit\u0027\n  Job neutron-functional-with-uwsgi: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-functional/lib/python3.6/site-packages/neutron_lib/context.py in the commit\u0027\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/unittest2/compatibility.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_messaging/_drivers/pool.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/shared/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file neutron/agent/common/ovs_lib.py\n  Comments left for invalid file neutron/objects/base.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/oslo_context/context.py\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_daemon.py\n  Comments left for invalid file .tox/shared/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/dsvm-functional/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/unittest2/compatibility.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/sqlalchemy/orm/relationships.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/openstack/config/cloud_region.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/neutron_lib/context.py\n  Comments left for invalid file neutron/policy.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/dsvm-functional/lib/python3.6/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_policy/_checks.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/eventlet/patcher.py\n  Comments left for invalid file neutron/common/eventlet_utils.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/dns/hash.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/sqlalchemy/orm/relationships.py\n  Comments left for invalid file .tox/docs/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/neutron_lib/worker.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/paste/deploy/loadwsgi.py\n  Comments left for invalid file .tox/shared/lib/python3.6/site-packages/astroid/interpreter/_import/spec.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/neutron_lib/utils/helpers.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py\n  Comments left for invalid file neutron/tests/unit/agent/ovsdb/test_impl_idl.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_context/context.py","accounts_in_message":[],"_revision_number":2},{"id":"d34f86e29b6d31bf8803271eb23fdd7c921d747b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-04 23:58:22.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/16/714716/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"3b115056f61d81a28b517302c1e5229b87138b0c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-05 08:12:28.000000000","message":"Patch Set 2:\n\nhmm, looks like unit tests don\u0027t like the exit (but like having errors logged otherwise)","accounts_in_message":[],"_revision_number":2},{"id":"9882d1dba111bfa31776f62f0470b47213ed1b50","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-06 08:07:20.000000000","message":"Patch Set 2:\n\n\u003e hmm, looks like unit tests don\u0027t like the exit (but like having\n \u003e errors logged otherwise)\n\nI think we need to modify the default config value in the setup of the tests here:\nhttps://github.com/openstack/neutron/blob/da63adc96bf9f8eafd66d0aeea5d6f8f787420ed/neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/test_mech_driver.py#L81","accounts_in_message":[],"_revision_number":2},{"id":"003f3fa5397c209c24f9926587bad6e30e951616","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-06 08:09:25.000000000","message":"Patch Set 2:\n\nLGTM, but leaving -1 to override default max_header_size config value in u/t and functional tests.\n\nThat needs also a release note.","accounts_in_message":[],"_revision_number":2},{"id":"703c590451ed4fd919521a718735221584186bc6","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-06 08:09:28.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"81a756eb0f6ee90cb6f8c562201b8db189bd6ca9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-06 18:27:05.000000000","message":"Patch Set 2:\n\n\u003e LGTM, but leaving -1 to override default max_header_size config\n \u003e value in u/t and functional tests.\n \u003e \n \u003e That needs also a release note.\n\nHmm, what bothers me is that no functional test caught that the value was too low in the first place.","accounts_in_message":[],"_revision_number":2},{"id":"d24070d1f5988f742d91aee4d8d50c6a9c9f59f6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-06 20:15:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f4be34521a318b7f500182ec52e8e7bb09ef0a5e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-06 20:16:21.000000000","message":"Patch Set 3:\n\nargh, something got wrong","accounts_in_message":[],"_revision_number":3},{"id":"e77d96062de7dd80ac36512dc66b7eb8756b91aa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-06 20:19:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e0c18df2c8da99e9de391424ab22886edde60358","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-06 21:32:47.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/16/714716/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"8139c80c4b516c6c0232487094d5bd4d2b84af3a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-06 22:12:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/714716/4 : SUCCESS in 1h 52m 04s","accounts_in_message":[],"_revision_number":4},{"id":"e26d10f3879029757b8b517840537bffbba98e74","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-06 22:28:38.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/16/714716/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"c87b58d1735c388d087bc58a0c51777b200298d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 23:12:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c3c2bb5a81d74efea8d400a6fb3f748b : SUCCESS in 1h 05m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c1c16a9f2f0c4d2384093642600d0687 : SUCCESS in 47m 33s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/1b556f67ee0143a1b8f3adac156127fc : FAILURE in 2h 02m 49s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3feed0f5e8a949a0af2b09028677f037 : SUCCESS in 1h 32m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0d4ce05a49794aaca0d37832697e143e : SUCCESS in 1h 14m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/22aecc10936942a2b65f6224fdc766db : SUCCESS in 1h 02m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82bd4b1f8e70402a87a3d5c41748b567 : FAILURE in 27m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dee3b4927a9f4e0c9a128e8b684fc72e : FAILURE in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bfd4d79659f4631b691a9b58e87ddc8 : SUCCESS in 13m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3065b601918f4c38944a3a6a7fec54eb : FAILURE in 19m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce3b5eed1cdf48f8b1dc0b06b4875c7a : FAILURE in 20m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f3dc51e154141e5acd00726ab7cb503 : FAILURE in 15m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9184d8844f284988a1b2007490978201 : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/182dde6f5b6a46619956720613980d85 : SUCCESS in 13m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/498c680b1bae444da865f624365fc0f8 : FAILURE in 41m 47s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7383c18db7f6482693ea5a49bffe134f : SUCCESS in 1h 11m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0f500de1d3bf4ea49bfb29ad5c783f8a : SUCCESS in 2h 02m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/bd39e63aebb04c8a90b402bc9ef15af0 : SUCCESS in 1h 19m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/86bf71aaa3304a47805044c701ee3ee4 : FAILURE in 1h 16m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2ea73729dfd64a68a31a69e8ac709448 : SUCCESS in 1h 18m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e8209d48e9204b719e77020c715272c0 : SUCCESS in 1h 06m 30s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fd744fcd5da1426f9cbd7fcecfbce968 : SUCCESS in 1h 40m 34s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/597155c4a6294dfdb27ddfd3cce0e0d6 : SUCCESS in 1h 52m 09s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8a0a3643e99c450db11beec1eadf6ee0 : SUCCESS in 1h 13m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/30139b6b899047cba284b11e4cc70218 : SUCCESS in 56m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c9616be86ce849318f2811c73e3e5564 : SUCCESS in 2h 24m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a8bfd74e2ec54898834c1c54399ef661 : SUCCESS in 1h 52m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/313629093bc04574b5a0ad3108bf04fb : POST_FAILURE in 2h 17m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11bb605d7f9d4732a92a037ab83032dc : SUCCESS in 1h 00m 47s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5553ab9eb2e642ee8e65d81e860270e5 : SUCCESS in 1h 17m 54s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ddb2104aed884e8aabe03ff73a184ef2 : FAILURE in 42m 54s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/77bc228106784118a0430facd21fd4b2 : SUCCESS in 1h 07m 17s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b988f12d5a124f0cb9a485ebb52dace5 : SUCCESS in 1h 14m 25s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0fabf129aabc4956aa1195b4fc368669 : SUCCESS in 1h 49m 47s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fec93ba6ae9e49eea023c84c9752996f : SUCCESS in 1h 52m 02s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4fa1ea260ef14f2b9c5e2504c61460f7 : SUCCESS in 2h 46m 57s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9fdbe62680834528aaacf86060846afc : SUCCESS in 1h 55m 33s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/05a3eaffe5ee4d118375595aaddb34ba : SUCCESS in 1h 50m 46s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/common/ovs_lib.py\n  Comments left for invalid file neutron/objects/base.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_daemon.py\n  Comments left for invalid file neutron/common/eventlet_utils.py\n  Comments left for invalid file neutron/tests/unit/agent/ovsdb/test_impl_idl.py\n  Comments left for invalid file neutron/policy.py","accounts_in_message":[],"_revision_number":4},{"id":"f9e821c36b05db2497cccde4f8bc0bcd45f858c1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-07 07:22:26.000000000","message":"Patch Set 4:\n\nHmm, weird, I see slightly different subsets of tests are being run between the different jobs and PSs. Is it something expected or stestr got mad?","accounts_in_message":[],"_revision_number":4},{"id":"217823cf917e61d75341144b5c961643012e91a7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-07 07:33:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bae9bb9e90f698529e3213168a5353a0e4937304","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-07 08:12:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7caa8b8cf3e1022cb13dd941a31106bb73f924b0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-07 08:41:20.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/16/714716/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"a31008121a9417038ca100b34c5c5bb034bb4095","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-04-07 08:47:59.000000000","message":"Patch Set 6: Code-Review+1\n\n\u003e Hmm, what bothers me is that no functional test caught that the\n \u003e value was too low in the first place.\n\nIt\u0027s never a bad idea to write a test that would have caught the bug in the first place. ;-)\n\nBut even without that this is already a nice improvement and I think we can merge this when the test config overrides are all done.","accounts_in_message":[],"_revision_number":6},{"id":"7f3626577a30faef5040279f040257b9d16c6502","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-07 09:23:59.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/16/714716/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"cb8edd4d7f57e028529360de4d3bac47cae0a170","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-07 10:03:09.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/714716/6 : SUCCESS in 1h 50m 22s","accounts_in_message":[],"_revision_number":6},{"id":"0af24754b79cac56826aef1aa120f5d7b1c90bc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 11:27:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6ef095f2ab734c1bb98d0f93f433b9ca : SUCCESS in 54m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6c8fd31ab94e4097bf736bb657861f89 : SUCCESS in 36m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/d189e132c77942afae1afb65ab92ce9c : FAILURE in 2h 06m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/479aea50aaac45fb97121a66884826f6 : SUCCESS in 1h 05m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6bb97fe268414b4db77d57d6b44a5e60 : SUCCESS in 1h 19m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0479047383ab434ab91de160bfe1e123 : SUCCESS in 1h 02m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37cdb5cb75c646268394c9a82949c7f7 : FAILURE in 33m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bf1071032e9467ca5aa95346aea9bf4 : FAILURE in 28m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/193d7bb3ecbc47779d809617ecde9bf8 : SUCCESS in 14m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee044a9bdbc9453dbc923c2a97e0bde8 : FAILURE in 29m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d8292485e336405aad8976ea9fc56064 : FAILURE in 26m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7016dfd158c4be9815f460727bafbf9 : FAILURE in 34m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b8ec48fa2d24831afdf946278cbff5e : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7f73672e271462c8d90174089f4967a : SUCCESS in 15m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d98bb38489954757b4fbb85b4de0782d : FAILURE in 1h 00m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e2baf4b1b946491eb2b72a02b835e623 : SUCCESS in 2h 10m 45s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ff7079c8376145d9afb6a7d4a7222169 : SUCCESS in 2h 26m 33s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3875313005674e8c8340ae8b65bc250f : SUCCESS in 1h 07m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/561200c5210f4147968c5b087c03940f : SUCCESS in 1h 31m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d1bc703452d34e90a15f0cce93ce728d : SUCCESS in 1h 28m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6e063eb36978457cbac1d3b638766d52 : TIMED_OUT in 2h 07m 58s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e81b0eb0e6cb48d79c377494bcf70a59 : TIMED_OUT in 3h 09m 26s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3323acdf423a477e9f92f34adef8ed1b : TIMED_OUT in 3h 08m 03s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4dfc22a833244dd88e170f40a893427c : SUCCESS in 2h 46m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c19989eb5b0c41e8a44ea0e199e55c8b : TIMED_OUT in 1h 37m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5451112749d946608d08f4aee1257ce0 : TIMED_OUT in 3h 06m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/37a7b7846c064ea386592a7233308c04 : TIMED_OUT in 3h 09m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8b93252ad06e4a178025192e6222e631 : TIMED_OUT in 3h 08m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf2754eed2624b61b662ed1bb401cf33 : SUCCESS in 1h 13m 31s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a113520fe6824f609ecb0381df37a96f : SUCCESS in 1h 25m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d806caa5dde4697b7c6bbbb0efb43a2 : FAILURE in 1h 09m 42s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b681187e70fa476587f83926bc788aa9 : SUCCESS in 1h 18m 30s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/90ac51687ac542019501034db93d2f46 : SUCCESS in 1h 21m 53s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/548cae03f0a44f5280f3f1dc2c33c90b : SUCCESS in 1h 57m 56s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/00e6aded41c44c6cae7b4b4343c528cb : SUCCESS in 2h 07m 52s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8916bac7e0cd472e9799d64048ad4088 : SUCCESS in 2h 57m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5699a723abcc4affa35421a95026abad : SUCCESS in 2h 44m 34s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/0c841615a6f74e0ea00a0b15adcb2580 : FAILURE in 2h 17m 12s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/common/ovs_lib.py\n  Comments left for invalid file neutron/tests/unit/agent/ovsdb/test_impl_idl.py\n  Comments left for invalid file neutron/tests/unit/test_auth.py\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_daemon.py\n  Comments left for invalid file neutron/objects/base.py\n  Comments left for invalid file neutron/common/eventlet_utils.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file neutron/policy.py","accounts_in_message":[],"_revision_number":6},{"id":"a2a4901ecff7444e0217238a898daba9efc5d754","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-25 15:42:57.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":6},{"id":"7e47746b37059f2edacb786747648aad47639739","tag":"autogenerated:gerrit:restore","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-01 14:04:13.000000000","message":"Restored","accounts_in_message":[],"_revision_number":6},{"id":"f76a6efa94f129c45e4cb0952c305d23d659eb2e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-01 14:04:30.000000000","message":"Patch Set 6:\n\nelvira asked me to restore this; thus restoring","accounts_in_message":[],"_revision_number":6},{"id":"4addd2f602b42f6fae24281c58f15de2db4171b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 14:04:35.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"281806b97c7d768b81152e6b423d9e01458b07df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 14:04:42.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"ee7ac8c5cc3049a774f44f0c86b67252df2ab9dd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-09-01 14:05:23.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"23fee2de6afc2c01a76f0eb336a890bcad80030e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2021-09-01 14:23:41.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e329ad967c4e6d5d7beb9506c9c31a796531f7fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 15:08:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3db6c7714b904132bf784a367942df86 : SUCCESS in 41m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d60ef89e15294a58a18cc54cc084e7a0 : SUCCESS in 39m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9ac31f75bc41611a68ecdddf1f89b04f5f4aaac7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-01 15:55:22.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/16/714716/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"5edf50f7f21a3a2937abb3eed6e8720a07dd1ba3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-01 16:12:08.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/714716/7 : SUCCESS in 1h 40m 15s","accounts_in_message":[],"_revision_number":7},{"id":"ea6e4f165f36dd5321dfb2681417dcc032c78d20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 16:34:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/64cd2281e99b4d458c03d07d3c557fb9 : SUCCESS in 42m 54s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5c9158c0fe824d7fb0d65a29c743ccf1 : SUCCESS in 1h 52m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b476c8d69a034f41a16a9bb5639ef8db : SUCCESS in 2h 05m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8830bcc153c94a05a1f9e1adb4a36add : SUCCESS in 1h 48m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ee7f09b57879429faa975e4511292d2b : SUCCESS in 1h 41m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d131b0ead1cd4cb283a01d9292ca6c4f : SUCCESS in 33m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0a1acf6addd4237aa34e3247ae18cca : SUCCESS in 30m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa69552cb3d04df3b2674f7c760f1144 : SUCCESS in 37m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82777a9da27c42f1be2d6568ec982032 : SUCCESS in 19m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68e0e7bb963c456cbbfeb26d871db03e : SUCCESS in 28m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51b134ce978146bb9f623dbe525dd7ac : SUCCESS in 30m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0492f880570e4aac9c98853db22213ae : SUCCESS in 29m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bebf21ea4440460d96ac890cc10b8894 : SUCCESS in 13m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/154b0f1eefd441b88d9d1b7396696e06 : SUCCESS in 12m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c1fbf9de8d1b459abcbad0254aced486 : FAILURE in 48m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/65b54806bbc9451bac72040a68c7642d : SUCCESS in 1h 52m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/21e797a12463469798af0a215b02fa46 : SUCCESS in 1h 11m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a1add9e098e84d7296803aba2ddd9f2a : SUCCESS in 1h 00m 54s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/893eee3975614503a17e8d99052404b8 : SUCCESS in 58m 02s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9c67e5c7d0f24ece9444fdb93b1cbdf6 : SUCCESS in 1h 11m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c7d65821b1f34cac98714f110aa945bc : SUCCESS in 1h 14m 26s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/3d48b69e7c6a42e686ff74a8537e7377 : SUCCESS in 1h 25m 26s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f5e18023143482fbc72af4c0b31fab5 : SUCCESS in 1h 03m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5a07897f63a54812a35b25ee7d2131db : SUCCESS in 1h 00m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ac39e98116ae4492b242d384a2e3193e : SUCCESS in 37m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5ae9f23966df4038a9891eae35c22a97 : SUCCESS in 1h 05m 50s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/46e2a3b380a04555ac89f618ae040a90 : SUCCESS in 1h 32m 00s","accounts_in_message":[],"_revision_number":7},{"id":"de758887a734bb9917b4dbae0508bada92d5f905","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2021-09-02 00:03:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"39710bdca67b0d47e64d99ac82ee2e5cb983e3bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 00:46:45.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa8b19af98c64ae08c3e1be7d0809f58 : SUCCESS in 40m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ca226769f8945bf897458f4d9f68513 : SUCCESS in 31m 28s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6aa982f561e6826ad22f61d8ea0998ebf0a316c5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-02 01:03:11.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/16/714716/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"e7a3f69c792caf6c3c0a7db6abb1c85df51339a3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-02 01:32:20.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/714716/8 : SUCCESS in 1h 27m 21s","accounts_in_message":[],"_revision_number":8},{"id":"6a20568dfc7f119bfe85cf620ec1b95c7d6f3266","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 02:39:52.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6e5fd33ca2bf4f618be4cd0e80d942f1 : SUCCESS in 43m 36s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f213e19966f04285aa5efe9e44fa17d9 : SUCCESS in 2h 35m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/11f0456ec44c4987bc3916f262c67c2d : SUCCESS in 1h 30m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/48116d53542e4cd18bfd618e72d61612 : SUCCESS in 1h 25m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e5f380636368461fa5c7d559e8f981c8 : SUCCESS in 1h 36m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/874a542fa6b14459b2a1b1d201692904 : SUCCESS in 28m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7970fd9eb13a44688191b4638af9cbe5 : SUCCESS in 32m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0ea18e3dc414f3aa01bf371e58f0568 : SUCCESS in 36m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1faae9b01f2b4d03b0a8e033bc58aa6e : SUCCESS in 12m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd3ad052c2bd422abd07e265994e47e3 : SUCCESS in 23m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5708575d844741bea2f44a90de7e2bb7 : SUCCESS in 21m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3dde7655cd643619ee9054e2d932f84 : SUCCESS in 23m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4558543d79d4cc3ab3f7b4a1d1c9caf : SUCCESS in 14m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd327cbe33e443a7b2ae0f70bd166a5c : SUCCESS in 10m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7595d1d4555d41ce8e9b967c4547fea6 : FAILURE in 43m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/98b021b48c7a4784b388f986d2335220 : SUCCESS in 1h 37m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/85e92daa6b71447e97e640a1b0c6059a : SUCCESS in 1h 13m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b6bc3ec8e12469297465b382e0471f7 : SUCCESS in 1h 05m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/33f7b49b8b0545039f684ec9e6621cbd : SUCCESS in 1h 05m 22s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/942109b17ea144adb164df67f55dd343 : SUCCESS in 1h 05m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/89c1880fd5b84d5aac33f6a646c79435 : SUCCESS in 1h 23m 17s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/14541896a4314f048339adb888a4613d : SUCCESS in 1h 29m 25s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bbac9eddc9241b49ba09dd03cc13bdd : SUCCESS in 51m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/25e182ad8cb541cd981516b4b4a619c5 : SUCCESS in 1h 09m 07s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d843a46bb0774ad98059b6127c7cf88c : SUCCESS in 26m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1bdbbdb86c6845e2b73adaec13769b2c : SUCCESS in 1h 04m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/84941c10f3b0488f808a86363f7f5453 : SUCCESS in 1h 23m 36s","accounts_in_message":[],"_revision_number":8},{"id":"9c4e791424f63a72db9437260ee45e4de44d47dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2021-09-02 08:41:45.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"cd919ca857ad5fecc4ca8b9158c729ebb169f93d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 09:26:16.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0113fffa472449648b1b970bdb119539 : SUCCESS in 34m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b6b9db077c7f40e4842daf7d5efd0ec7 : SUCCESS in 42m 38s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"146556c38e667db1460728e3718c0bdcca21dc55","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-02 09:36:49.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/16/714716/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"c10ccd37f8796b3f57c9c3085d961e7eab540a16","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-02 10:15:35.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/714716/9 : SUCCESS in 1h 33m 11s","accounts_in_message":[],"_revision_number":9},{"id":"28f5d36725d30974a353c5da4fd6ccff7b716d22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 11:10:31.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9017f77666bc4fd7825b726da50c80db : SUCCESS in 41m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/437adfc9979f4a37a35af82335202f56 : SUCCESS in 2h 22m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c82547770c084ab6b36b1f3ec33dc0d0 : SUCCESS in 1h 50m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5a639745ec524ee6828e26f24b19746f : SUCCESS in 1h 32m 04s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a4c2b1de0727436382cab02c063367e8 : SUCCESS in 2h 07m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2545ea68a6784f54aef7cd1813c6825b : SUCCESS in 46m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c36b97a40e94191abcc29fcab056702 : SUCCESS in 38m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fe65ad9c9884515a413bc2ead8eb5b4 : SUCCESS in 44m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9ed5c0a320e45ea821d21e6ba507391 : SUCCESS in 14m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c707a07c826b4d2a9427e70d9d069ef3 : SUCCESS in 25m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4715fc3ba40c43f2a776b27139017f4a : SUCCESS in 28m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9cd8422ac4d4c12b2e44a86f006b31d : SUCCESS in 26m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68c7f0a0d6a54af0a4af9a08e0b066d8 : SUCCESS in 9m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a2f6c3448034cce8b916877ffec9b9f : SUCCESS in 11m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c82f74bb93c4e2fb325823bce37137f : SUCCESS in 42m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a8ddaae750844575b4da8f3a6dc56d8c : SUCCESS in 1h 52m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/77571ab1f95d4c258a9dd3b7e5c99b81 : SUCCESS in 1h 15m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6daab3cbcc764152a05f6d07aa85b543 : SUCCESS in 56m 57s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/db89603626c14a8e8a2aa6b139b50d5e : SUCCESS in 1h 07m 28s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/db4c333548ff4f12b02abeaff1ddb947 : SUCCESS in 1h 07m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/25cd796abace42d8b326d1e1fa834df3 : SUCCESS in 1h 18m 24s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/c829d588f24843519557d487bf57ecc0 : SUCCESS in 1h 26m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fbf99675639411fb9ef24bf441ab46c : SUCCESS in 55m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1ace82390fae40089afe4d1ddf7dcd32 : SUCCESS in 55m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/14dcc81e2f334d658730994ccdd715ef : SUCCESS in 29m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f463981643574210a0f4111345f0d015 : SUCCESS in 1h 17m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/32e36e1a5b7347e592df3a7192370e15 : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":9},{"id":"a4667598eb1a177518df7d2a7e8db2cd566d135c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-02 11:41:11.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1abd1dd52802d95c410e5ee863dd9cf4d46ca271","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-02 11:42:24.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4cd4ee5b60c1fb9e5a8504111e9394064c01268f","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-09-02 11:49:41.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"7a341ba0eda7f055d77942f058b3e2b8f85dbe7d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-02 16:58:26.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"0ed415fe3caaa02f9ebccf736e193794d9ae57f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 16:59:24.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"01f1b6dd4eba56d06a4f45a24a7aee8ea5722a42","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-02 19:15:16.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"92fb2db41b1882ed449c12d417f92fe6db66b046","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 19:41:27.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c44c3cf1ea714541bd4aab91f5eb7250 : SUCCESS in 50m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3d6d0cc49fda42e3b4bf6da09495297e : FAILURE in 2h 02m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dab476607c1d4c659749fd240af2c006 : SUCCESS in 2h 34m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b52c4ad1fac644b590d791867842a9e6 : SUCCESS in 1h 50m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/63a117d9c7a94b24abc8894a6959b1be : SUCCESS in 1h 27m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3852c26688245cc9f7bde69c241a441 : SUCCESS in 39m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c5cc5f69595419ab717689b12952a21 : SUCCESS in 16m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ecd119105ff4d03a89ca0843d93832c : SUCCESS in 28m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48faedc8cfe24f469588d9eb74c15503 : SUCCESS in 27m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efbf6f3a2ab34805bc184f3e8f8fd5c1 : SUCCESS in 12m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9305c894c2b94f5facd38cd1c47c2d85 : SUCCESS in 13m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f591802c1c5048118d5b483e69a37574 : SUCCESS in 44m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a02c285c6afe4aad88f67a04d27010ac : FAILURE in 1h 35m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cf762017387341d18e1ff6879d2697b9 : SUCCESS in 1h 13m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/35a03f0106474a10b99be1fd183283dc : SUCCESS in 1h 01m 07s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/f27d9323df364a8f98f9e3508c2ca6de : SUCCESS in 1h 29m 16s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b090baa720394d31ae499736e23e94cc : SUCCESS in 59m 42s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7136cb9d4bc8439aa36720db751d67dd : SUCCESS in 53m 51s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/87a9e8922fa044bdb56395c46a4a169e : SUCCESS in 36m 31s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/eccb91d973b14d6990decd1678e25099 : SUCCESS in 1h 16m 05s","accounts_in_message":[],"_revision_number":9},{"id":"d5fced15c5b6451cadb9f38fcd6871eff3800c39","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2021-09-02 20:19:06.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"a7fcf6f77d6982530effca224435b47fa4cad8bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 21:05:42.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f05c073f3e845ae9b8669bcc36c21a3 : SUCCESS in 43m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5898ec0f46d442bd96c7831c9696bf1d : SUCCESS in 44m 12s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"09536ee88ac7571390f8460fb772c7e286783b21","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-02 21:51:45.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/714716/9 : FAILURE in 1h 31m 52s","accounts_in_message":[],"_revision_number":9},{"id":"d38266be1d69e7557752e845cdd824df37691f01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 22:47:11.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/54a1578f3be54824aeac9d1d605650e8 : SUCCESS in 51m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/925969770032451fb23f1c68be5cde02 : SUCCESS in 2h 20m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/45204aed29d344de817fb885220d44ec : SUCCESS in 1h 56m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8e5c0843a20741f696468e0456658bff : SUCCESS in 2h 06m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8a2295ea7ffe4cb99c77453d8784a56f : SUCCESS in 2h 06m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8ca084aee47c4fa58ec9c558f9109b27 : SUCCESS in 39m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dab233db42be4d93891aba1935c79c37 : SUCCESS in 43m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23a00d7fb92e4c76a689b821cbda3113 : SUCCESS in 40m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fdaea9f2baa41e7a156f8318aa5d455 : SUCCESS in 15m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59266c2a11a9401c9e69e29940e41a29 : SUCCESS in 27m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/592267fd3aa345be99fd4fffe9a145cb : SUCCESS in 24m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fa9035b39fa47e8aa5d8e69b7c2fcda : SUCCESS in 28m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/544de5ef964d493b8d19f609dff15dbd : SUCCESS in 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6de7be528615454eba1081453bccaacb : SUCCESS in 13m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bc0ccad568d3455bb3c019e3d47b2da7 : SUCCESS in 46m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/50e5414825964f0ab58be39dd5b451f2 : SUCCESS in 1h 49m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9e20dfceff5f4bffadd255d559229901 : SUCCESS in 1h 16m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d24b53b3c13b469c97113ca6800184bf : SUCCESS in 1h 02m 29s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/da20194d9203410f88569175da8614c5 : SUCCESS in 1h 11m 15s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ded2167fa8cd48f4bf89d8abc3207f2b : SUCCESS in 1h 11m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/22777fc72a9b4255886dc929e2177ae3 : SUCCESS in 1h 21m 11s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/23d660ad40af4e3b8f6e616dc9ca534a : SUCCESS in 1h 33m 46s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f346a7702994c7886fa274610b879ea : SUCCESS in 55m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/09ce6cf7cc8b4658ae9fa8d382624b47 : SUCCESS in 55m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e849e69fd2b740e8afcafadeff123435 : SUCCESS in 33m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f9a07dec36b54b64abf1e2b2bdcc0398 : SUCCESS in 1h 05m 59s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a15ffd23d4784bd09380f01e18674d19 : SUCCESS in 1h 08m 01s","accounts_in_message":[],"_revision_number":9},{"id":"6bb01fe008405577168ba8a1ef14d11e943ff39c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 22:47:53.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"a1e3f588133f0414d86d7817273d4deb8c53ca20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 02:08:17.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1a2fd9e0650240b29f85a8923acf1ac4 : SUCCESS in 44m 49s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7e008fb1f6dc46d1b760d7e75da15954 : SUCCESS in 2h 23m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/43eae45587494895be5153377051eedb : SUCCESS in 2h 01m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5ea3868bd2e6474bb5e1e3e928468cfb : SUCCESS in 2h 22m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a5533f5a5a2643c398659f3f314a6396 : SUCCESS in 1h 28m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39c1ee2f86b1466a820a6750cc8aee8b : SUCCESS in 49m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5250863f561e4829b58c5531896bf805 : SUCCESS in 15m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b68598e8a8414aa597d13909cdac7843 : SUCCESS in 27m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfba1dd7232d4c48afe7539f089c6156 : SUCCESS in 28m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e8d1904b6fa46b38556f571f23172f6 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56fea00e5f8641a59625ba7bf6ebed65 : SUCCESS in 10m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f2a1b218b1414352a22f33948216adc5 : SUCCESS in 55m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d3168eafef2b4e9f90576f312c2f2ace : SUCCESS in 1h 44m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5b07eb53d37747c1803e47129c5c9f5e : SUCCESS in 1h 13m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7149b941eb5b439cb0d56083eb631932 : SUCCESS in 57m 01s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/09d197dc660c4579a0d5dba2b5d17c50 : SUCCESS in 1h 36m 49s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b746cb1bbc24b8595b018203b7f0ab7 : SUCCESS in 50m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3ff1bdf996ef43ce9b89f7cfd7fb642e : SUCCESS in 59m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/174d4685e3b34086a67a92c450b327d8 : SUCCESS in 28m 54s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/465f8718f4534d25852a9fed443b94a5 : SUCCESS in 1h 08m 26s","accounts_in_message":[],"_revision_number":9},{"id":"e7acaca2c1f74eea9a91f556bc9cfd5c88896ea9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 02:08:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"fd52d95f26c939502af8a13c17ed941f158f02f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 02:10:11.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24b3a49c4e9e414797837070113ab1ac : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d1dacac2efc4092ac5f0ebe5f7368e5 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"dc4a57d96661fcb56ff916cafbbbdc683653e9c4","revisions":{"3e1f15e4446c76c768c58182f0d93e0b96d3c0aa":{"kind":"REWORK","_number":1,"created":"2020-03-24 17:00:20.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/16/714716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/714716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/714716/1"}}},"commit":{"parents":[{"commit":"56c7db5e7fafa110a0afc1ad07511fa7f1191137","subject":"Merge \"Change OVS tunnel tests to support unittest.mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56c7db5e7fafa110a0afc1ad07511fa7f1191137"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-24 16:55:23.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-24 17:00:15.000000000","tz":60},"subject":"Make OVN driver validate Geneve max_header_size","message":"Make OVN driver validate Geneve max_header_size\n\nAlso updates the docs to be clearer on OVN-Geneve relation topics.\n\nChange-Id: Ia253cc4d85261ce1535f4d27b3da91275d879903\nCloses-bug: #1868137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e1f15e4446c76c768c58182f0d93e0b96d3c0aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e1f15e4446c76c768c58182f0d93e0b96d3c0aa"}]},"branch":"refs/heads/master"},"9f57a7d99bcf75ce60d0715395f32d1a02d5dfab":{"kind":"REWORK","_number":2,"created":"2020-04-04 15:45:00.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/16/714716/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/714716/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/714716/2"}}},"commit":{"parents":[{"commit":"56c7db5e7fafa110a0afc1ad07511fa7f1191137","subject":"Merge \"Change OVS tunnel tests to support unittest.mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56c7db5e7fafa110a0afc1ad07511fa7f1191137"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-24 16:55:23.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-04 15:44:43.000000000","tz":120},"subject":"Make OVN driver validate Geneve max_header_size","message":"Make OVN driver validate Geneve max_header_size\n\nAlso updates the docs to be clearer on OVN-Geneve relation topics.\n\nChange-Id: Ia253cc4d85261ce1535f4d27b3da91275d879903\nCloses-bug: #1868137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f57a7d99bcf75ce60d0715395f32d1a02d5dfab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f57a7d99bcf75ce60d0715395f32d1a02d5dfab"}]},"branch":"refs/heads/master"},"c5dafdc2fc34a4baedac18065d118e5bda79e4d0":{"kind":"REWORK","_number":3,"created":"2020-04-06 20:15:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/16/714716/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/714716/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/714716/3"}}},"commit":{"parents":[{"commit":"56c7db5e7fafa110a0afc1ad07511fa7f1191137","subject":"Merge \"Change OVS tunnel tests to support unittest.mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56c7db5e7fafa110a0afc1ad07511fa7f1191137"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-24 16:55:23.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-06 20:14:52.000000000","tz":120},"subject":"Make OVN driver validate Geneve max_header_size","message":"Make OVN driver validate Geneve max_header_size\n\nAlso updates the docs to be clearer on OVN-Geneve relation topics.\n\nChange-Id: Ia253cc4d85261ce1535f4d27b3da91275d879903\nCloses-bug: #1868137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5dafdc2fc34a4baedac18065d118e5bda79e4d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5dafdc2fc34a4baedac18065d118e5bda79e4d0"}]},"branch":"refs/heads/master"},"f97abe875f3a209f1451e39bbb7a234144a29a18":{"kind":"REWORK","_number":4,"created":"2020-04-06 20:19:59.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/16/714716/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/714716/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/714716/4"}}},"commit":{"parents":[{"commit":"56c7db5e7fafa110a0afc1ad07511fa7f1191137","subject":"Merge \"Change OVS tunnel tests to support unittest.mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56c7db5e7fafa110a0afc1ad07511fa7f1191137"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-24 16:55:23.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-06 20:18:49.000000000","tz":120},"subject":"Make OVN driver validate Geneve max_header_size","message":"Make OVN driver validate Geneve max_header_size\n\nAlso updates the docs to be clearer on OVN-Geneve relation topics.\n\nChange-Id: Ia253cc4d85261ce1535f4d27b3da91275d879903\nCloses-bug: #1868137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f97abe875f3a209f1451e39bbb7a234144a29a18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f97abe875f3a209f1451e39bbb7a234144a29a18"}]},"branch":"refs/heads/master"},"9079ceda258b9aa22472a8af5e73db47b0125e59":{"kind":"REWORK","_number":5,"created":"2020-04-07 07:33:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/16/714716/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/714716/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/714716/5"}}},"commit":{"parents":[{"commit":"56c7db5e7fafa110a0afc1ad07511fa7f1191137","subject":"Merge \"Change OVS tunnel tests to support unittest.mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56c7db5e7fafa110a0afc1ad07511fa7f1191137"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-24 16:55:23.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-07 07:32:51.000000000","tz":120},"subject":"Make OVN driver validate Geneve max_header_size","message":"Make OVN driver validate Geneve max_header_size\n\nAlso updates the docs to be clearer on OVN-Geneve relation topics.\n\nChange-Id: Ia253cc4d85261ce1535f4d27b3da91275d879903\nCloses-bug: #1868137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9079ceda258b9aa22472a8af5e73db47b0125e59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9079ceda258b9aa22472a8af5e73db47b0125e59"}]},"branch":"refs/heads/master"},"3c71b2d43b63dae6313e747596cf947fc9dd290e":{"kind":"REWORK","_number":6,"created":"2020-04-07 08:12:12.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/16/714716/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/714716/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/714716/6"}}},"commit":{"parents":[{"commit":"56c7db5e7fafa110a0afc1ad07511fa7f1191137","subject":"Merge \"Change OVS tunnel tests to support unittest.mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56c7db5e7fafa110a0afc1ad07511fa7f1191137"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-24 16:55:23.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-07 08:12:03.000000000","tz":120},"subject":"Make OVN driver validate Geneve max_header_size","message":"Make OVN driver validate Geneve max_header_size\n\nAlso updates the docs to be clearer on OVN-Geneve relation topics.\n\nChange-Id: Ia253cc4d85261ce1535f4d27b3da91275d879903\nCloses-bug: #1868137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3c71b2d43b63dae6313e747596cf947fc9dd290e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3c71b2d43b63dae6313e747596cf947fc9dd290e"}]},"branch":"refs/heads/master"},"d9850e498c290fc12d2e5d560955657be335ac1f":{"kind":"REWORK","_number":7,"created":"2021-09-01 14:23:41.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/16/714716/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/714716/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/714716/7"}}},"commit":{"parents":[{"commit":"7e5e6d4929aca442059d6ec1df3ab4405b6b8df9","subject":"Merge \"use payload for OVSDB_RESOURCE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e5e6d4929aca442059d6ec1df3ab4405b6b8df9"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-24 16:55:23.000000000","tz":60},"committer":{"name":"Elvira García Ruiz","email":"egarciar@redhat.com","date":"2021-09-01 14:22:21.000000000","tz":0},"subject":"Make OVN driver validate Geneve max_header_size","message":"Make OVN driver validate Geneve max_header_size\n\nAlso updates the docs to be clearer on OVN-Geneve relation topics.\n\nCo-Authored-By: Elvira García Ruiz \u003cegarciar@redhat.com\u003e\nChange-Id: Ia253cc4d85261ce1535f4d27b3da91275d879903\nCloses-bug: #1868137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9850e498c290fc12d2e5d560955657be335ac1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9850e498c290fc12d2e5d560955657be335ac1f"}]},"branch":"refs/heads/master"},"75d990e50db6db08ccc4948508722bd8e43ff3c5":{"kind":"REWORK","_number":8,"created":"2021-09-02 00:03:47.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/16/714716/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/714716/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/714716/8"}}},"commit":{"parents":[{"commit":"7e5e6d4929aca442059d6ec1df3ab4405b6b8df9","subject":"Merge \"use payload for OVSDB_RESOURCE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e5e6d4929aca442059d6ec1df3ab4405b6b8df9"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-24 16:55:23.000000000","tz":60},"committer":{"name":"Elvira García Ruiz","email":"egarciar@redhat.com","date":"2021-09-02 00:02:06.000000000","tz":0},"subject":"Make OVN driver validate Geneve max_header_size","message":"Make OVN driver validate Geneve max_header_size\n\nAlso updates the docs to be clearer on OVN-Geneve relation topics.\n\nCo-Authored-By: Elvira García Ruiz \u003cegarciar@redhat.com\u003e\nChange-Id: Ia253cc4d85261ce1535f4d27b3da91275d879903\nCloses-bug: #1868137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/75d990e50db6db08ccc4948508722bd8e43ff3c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/75d990e50db6db08ccc4948508722bd8e43ff3c5"}]},"branch":"refs/heads/master"},"dc4a57d96661fcb56ff916cafbbbdc683653e9c4":{"kind":"REWORK","_number":9,"created":"2021-09-02 08:41:45.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/16/714716/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/714716/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/714716/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/714716/9"}}},"commit":{"parents":[{"commit":"7e5e6d4929aca442059d6ec1df3ab4405b6b8df9","subject":"Merge \"use payload for OVSDB_RESOURCE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e5e6d4929aca442059d6ec1df3ab4405b6b8df9"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-24 16:55:23.000000000","tz":60},"committer":{"name":"Elvira García Ruiz","email":"egarciar@redhat.com","date":"2021-09-02 08:41:19.000000000","tz":0},"subject":"Make OVN driver validate Geneve max_header_size","message":"Make OVN driver validate Geneve max_header_size\n\nAlso updates the docs to be clearer on OVN-Geneve relation topics.\n\nCo-Authored-By: Elvira García Ruiz \u003cegarciar@redhat.com\u003e\nChange-Id: Ia253cc4d85261ce1535f4d27b3da91275d879903\nCloses-bug: #1868137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc4a57d96661fcb56ff916cafbbbdc683653e9c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc4a57d96661fcb56ff916cafbbbdc683653e9c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
