)]}'
{"id":"openstack%2Fneutron~715419","triplet_id":"openstack%2Fneutron~master~I257cc4966379ddacbcaa6fbe8552e638cda303f3","project":"openstack/neutron","branch":"master","topic":"bug/1866445","hashtags":[],"change_id":"I257cc4966379ddacbcaa6fbe8552e638cda303f3","subject":"Avoid Layer 2 and Layer 3 traffic flooding in DVR mode.","status":"ABANDONED","created":"2020-03-27 13:07:05.000000000","updated":"2020-09-25 15:43:11.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8451e4d11e1b1346603e346938c925e20dd3bb74","_number":715419,"virtual_id_number":715419,"owner":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2020-04-04 22:34:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},{"date":"2020-09-25 15:43:11.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"date":"2020-04-05 01:55:41.000000000","_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-04 10:30:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"date":"2020-04-05 03:40:20.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 14:54:58.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2020-04-01 01:04:26.000000000","updated_by":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"reviewer":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"state":"REVIEWER"},{"updated":"2020-04-04 10:30:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-04 22:34:16.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-05 03:40:20.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-04-06 08:54:24.000000000","updated_by":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"reviewer":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"state":"REVIEWER"},{"updated":"2020-09-25 15:43:11.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"dd66d8846a828e8e73dca007733bbcf68f730626","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-03-27 13:07:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f399c3c198c6c74237e4210ba3e0b5012fbebec7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-27 13:31:28.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/715419/1 : FAILURE in 22m 49s","accounts_in_message":[],"_revision_number":1},{"id":"926d9de2579933f47459f2b232791199b17761bf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-27 13:46:47.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/19/715419/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"bbb8a107218eb6ce9c2ab03d30736dcedde8fc66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 16:47:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/55c84fa6d7704b0fadde9f6e752b1eec : SUCCESS in 1h 14m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/777a43c2d2ad4ba2880f67862229c4eb : SUCCESS in 40m 03s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/bfa780c5c40e46c4805c93e5cca6dee5 : TIMED_OUT in 3h 04m 12s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/69c3c65dabe44af0b313e37680d7d6f4 : SUCCESS in 1h 10m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/00bf2eada2274b409da92c08bb5ccc91 : SUCCESS in 1h 13m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c08bd0cd9cb44dd2a0066932bcaf1fb3 : SUCCESS in 1h 07m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8990ff16a574bf8ba02b70533ec1d87 : SUCCESS in 43m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16d90e3a78b84a3ea23e7549974ffed6 : SUCCESS in 35m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2814587f36eb450eb601f20f2c7fd9b9 : FAILURE in 9m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fdfc419fe2b4dfb9703d9457a5c1964 : SUCCESS in 35m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d56857b4b9244f329b18bbdd844dd295 : SUCCESS in 33m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/309608847f79458d9520421d55ae6ed2 : SUCCESS in 24m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b82b7072affd4ac7862218629b57653b : SUCCESS in 11m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/481ce1bc42464b1c8eabdd87e510b387 : SUCCESS in 1h 17m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a5c393b504654a86974d7ac7afa23391 : SUCCESS in 1h 17m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ef73c6eb061e49b1898fe31aaa29a617 : SUCCESS in 2h 16m 32s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ac4406322fd341c2af790c31cd70e1fe : SUCCESS in 1h 35m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/80f004c002f14998832bfe502c511139 : SUCCESS in 1h 01m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/670b6057b2a841069ee0c4d14c159658 : FAILURE in 1h 05m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/349c40db142e42f3a337573c634ac63a : SUCCESS in 1h 16m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c445393b61854c98b77a4547221b11cb : SUCCESS in 1h 41m 26s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fa5d213e3bff4bdbb11efc85159ffbc7 : FAILURE in 1h 58m 03s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ac9308a023de44c2b92a893fc9d90375 : SUCCESS in 1h 11m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/784929272c50484e8a753a1883fe049e : SUCCESS in 56m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/899c0d3400f84660862be5216420bdeb : SUCCESS in 2h 28m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2794c513dff44356a4ac777a6a56dd50 : SUCCESS in 1h 51m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/28f2fcf161ca428f93e77fd338b461bc : SUCCESS in 1h 47m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42482cfa04794bdcb619c441eddce89a : SUCCESS in 1h 02m 46s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3bad0cf01c624b6d87ac200e0674bfe9 : FAILURE in 37m 48s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/553be0cb5d8547d596e196bcf1d50a3c : SUCCESS in 50m 28s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a639effc10864030a3ce09c0d4aca737 : SUCCESS in 1h 31m 39s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/04f34747f71849958e4c4ce1e116da3d : SUCCESS in 1h 38m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/dffef0f07f89414a88759ea245209f39 : SUCCESS in 2h 05m 13s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/7df31e917ee5446fae39eb237a823570 : SUCCESS in 2h 53m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a8632b93e4478677e36e4ecf39e123643f000ad","author":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"date":"2020-03-30 10:43:44.000000000","message":"Patch Set 1: Code-Review-1\n\nI like this solution. But the pep8 needs to be be fixed.","accounts_in_message":[],"_revision_number":1},{"id":"b6eebfb2176d9d9965894e31dc621010abe1403a","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-03-30 23:42:07.000000000","message":"Patch Set 1:\n\n\u003e I like this solution. But the pep8 needs to be be fixed.\n\nDarragh O\u0027Reilly, thank for your comments, I did find L3 flood because qrouter used static ARP and couldn\u0027t send ARP request, but this fix resulted in L3 performance drop, about 30% per our test, can you help try it and check if the same performance issue is there?\n\nPer our ovs experience, we can\u0027t directly output the packets to patch port, the behavior is unknown and uncertain, Yajie needs to change code to output it to vxlan port directly. Per Yulong\u0027s patch, it seems that also output the packets to vxlan port but not patch-tun.","accounts_in_message":[],"_revision_number":1},{"id":"5fff9d578f4623c1d1c0acab8ee2908388a8a3a8","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-03-31 14:54:58.000000000","message":"Patch Set 1: Code-Review-1\n\nThe fix of bug #1732067 has a config option ``explicitly_egress_direct`` [1] which can be used for non-OVS firewall as well.\nPlease have a try to add the following config option to your compute nodes to see if the flooding still happen:\n[agent]\n...\nexplicitly_egress_direct \u003d True\n\nFor more details, please see my comment here:\nhttps://bugs.launchpad.net/neutron/+bug/1866445/comments/18","accounts_in_message":[],"_revision_number":1},{"id":"c42bf1af6191f0994e9e24f29c811b006dfdcab5","author":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"date":"2020-03-31 15:18:34.000000000","message":"Patch Set 1:\n\n@Li YaJie: I tested this and I still see flooding in the east/west routing case. Packets from VMs with dl_dst\u003dgateway_mac hit the NORMAL action in table 60 and are flooded because the gateway_mac is not learned. So it seems we need another flow somewhere to direct these packets to the qr- port.","accounts_in_message":[],"_revision_number":1},{"id":"525c95873369776ecb0e46392f2eae1097b25110","author":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"date":"2020-03-31 15:37:18.000000000","message":"Patch Set 1:\n\nI had explicitly_egress_direct\u003dTrue. Your patch does not flood after setting it false.","accounts_in_message":[],"_revision_number":1},{"id":"87fffe54973ebff399acf612f4915eba053b72d1","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2020-04-01 01:04:26.000000000","message":"Patch Set 1:\n\n\u003e The fix of bug #1732067 has a config option ``explicitly_egress_direct``\n \u003e [1] which can be used for non-OVS firewall as well.\n \u003e Please have a try to add the following config option to your\n \u003e compute nodes to see if the flooding still happen:\n \u003e [agent]\n \u003e ...\n \u003e explicitly_egress_direct \u003d True\n \u003e \n \u003e For more details, please see my comment here:\n \u003e https://bugs.launchpad.net/neutron/+bug/1866445/comments/18\n\nYulong, your patch can\u0027t work for firewall_driver\u003diptables_hybrid, I submitted one https://review.opendev.org/#/c/712640/ which can make sure it can work for iptables_hybrid.\n\nI can double confirm your fix patch plus https://review.opendev.org/#/c/712640/ can fix our issue, no L2 and L3 flood.\n\nBut I don\u0027t understand why your patch doesn\u0027t have L3 flood even if you didn\u0027t add flows to handle that case. DVR qrouter used static ARP, it won\u0027t send ARP request, OVS didn\u0027t learn MAC, but it doesn\u0027t have L3 flood, can you help explain why.","accounts_in_message":[],"_revision_number":1},{"id":"a06ea016df7dbff5ec521337666078e45f28000f","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-04-01 03:07:24.000000000","message":"Patch Set 1:\n\n@Darragh O\u0027Reilly: thank for your comments. You are right, the flooding still exists when the packets from VMs with dl_dst\u003dgateway_map. This is my mistake, I forgot to deal with this situation, I will fix it.","accounts_in_message":[],"_revision_number":1},{"id":"9d1e55a01eca59f303acb00400022ea5db13f15e","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-04-01 12:19:11.000000000","message":"Patch Set 1:\n\nHi guys, I will bring up this issue to the L3 meeting [1] today, please attend if possible.\n[1] http://eavesdrop.openstack.org/#Neutron_L3_Sub-team_Meeting","accounts_in_message":[],"_revision_number":1},{"id":"e3232ae9cf4a3427bbd4c4e08a717d1e8ef2fe4b","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-04-02 09:13:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ca78a8cf78c3d814ebfb636bb50a29a132bb103","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-04-04 03:18:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ba3b02e69652e96d4767473a6fca97435e1ec31b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-04 05:04:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/715419/2 : SUCCESS in 1h 45m 39s","accounts_in_message":[],"_revision_number":2},{"id":"e318d80bd80139fab4c5985e29d1e11af6cd6d45","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-04-04 05:21:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f79679edceb83c34eac6fcd94aa11db02eaae1dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 06:44:24.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e990e5fb522a4c9f83aaad312cb7b99c : TIMED_OUT in 2h 05m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/22960b79996847a89a2dde484e3da757 : SUCCESS in 36m 17s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/fab3e1b1a3254a5a8a8970a2f549bfd1 : TIMED_OUT in 3h 05m 05s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2c271cdfb4a949aaab31afc28ed08f4b : SUCCESS in 1h 16m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/732463c209ad433cacb4a9b529c97af3 : SUCCESS in 1h 08m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c72fd9fc05fe4494899a27c40108cfb7 : SUCCESS in 1h 07m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e12d00d7569340628047505363bb7995 : SUCCESS in 37m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68cc51c47b114fc8b7e691deaf9deeb9 : SUCCESS in 24m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56e6a59de3bf4d4fbd40380d8760d248 : FAILURE in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3b2f7ade66e459c955d3e76fb6772e3 : SUCCESS in 25m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b9e9c7d907f4de4bf6cbfdc4e979658 : SUCCESS in 25m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a266d4fc85524823aa20cbffdc658bd0 : SUCCESS in 26m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5afc7f7df88643f2a21332b65144a991 : SUCCESS in 11m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/48118526b3204a61834513dc5c89c8dd : SUCCESS in 47m 04s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e678152840c946bc8176663a4ee3b46d : SUCCESS in 1h 06m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ab99f0bc27e0449b8cc709ca393f45bf : SUCCESS in 1h 50m 32s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1640d791ef144de797d76392ea86d0ea : SUCCESS in 2h 28m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d9cc66b3f5e4cfca592d788d5c862a1 : SUCCESS in 1h 03m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/077752e5d6c34e6bbf04f926964f8282 : FAILURE in 2h 03m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1e3f7429e91d45d596dabc53808e9436 : SUCCESS in 1h 09m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6aa4508ef56e48dd9b9bc7d3f436185d : SUCCESS in 2h 16m 48s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d7b4fd0983b049b48cdb543d3ac4d8f2 : FAILURE in 1h 35m 23s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8ed116413dfd4d63bc135270dc13fe9f : FAILURE in 2h 26m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3be95b16c7a84640a2189aefa3d2d67d : TIMED_OUT in 1h 34m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a8e508ccc2b34b269a4346dd5dab142c : TIMED_OUT in 3h 06m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/818291f6317d46378ce51466c81ed175 : TIMED_OUT in 3h 08m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fdfce91ad36f4d9c96b5003e43f822d9 : TIMED_OUT in 3h 09m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b283cee2f314456288dff1aefdbf9d2f : FAILURE in 2h 26m 16s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/bdef5d8732a241f19ac87494ac190d67 : TIMED_OUT in 2h 36m 18s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ab5328aa91044f41ab5fa9a55534cbfb : FAILURE in 1h 51m 09s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4d74babfeb7481baef7002006a0edf8 : FAILURE in 2h 03m 00s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/14822845f80340b6b92ab646f342f094 : SUCCESS in 1h 05m 51s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e0729c91032f4498ae348be2f81ec0c6 : SUCCESS in 2h 31m 00s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d5df7fa2ed05487b88137eea9f342d4d : SUCCESS in 1h 58m 06s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/3b19a99b77a348f8acebf3b6b5a8cb30 : SUCCESS in 2h 31m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fba81a79fbc64bb19d2cc1472ed2379b : SUCCESS in 1h 46m 18s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/1640abe91e5c40ce9342c3600f39f491 : SUCCESS in 1h 59m 05s (non-voting)\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/shared/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/docs/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_messaging/_drivers/pool.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/paste/deploy/loadwsgi.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_policy/_checks.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/neutron_lib/context.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/ovs/db/idl.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/openstack/config/cloud_region.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/ovs/db/idl.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/sqlalchemy/orm/relationships.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/neutron_lib/worker.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/neutron_lib/context.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/dns/hash.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/unittest2/compatibility.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/unittest2/compatibility.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/neutron_lib/context.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/sqlalchemy/orm/relationships.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/neutron_lib/context.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/shared/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/shared/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job neutron-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-functional/lib/python3.6/site-packages/neutron_lib/context.py in the commit\u0027\n  Job neutron-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-functional/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py in the commit\u0027\n  Job neutron-fullstack: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job neutron-fullstack: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job neutron-functional-with-uwsgi: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-functional/lib/python3.6/site-packages/neutron_lib/context.py in the commit\u0027\n  Job neutron-fullstack-with-uwsgi: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job neutron-fullstack-with-uwsgi: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_policy/_checks.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/unittest2/compatibility.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/shared/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file neutron/objects/base.py\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_daemon.py\n  Comments left for invalid file .tox/shared/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/dns/hash.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/openstack/config/cloud_region.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/unittest2/compatibility.py\n  Comments left for invalid file .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/eventlet/patcher.py\n  Comments left for invalid file neutron/common/eventlet_utils.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/ovs/db/idl.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/neutron_lib/worker.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file neutron/tests/unit/test_auth.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_messaging/_drivers/pool.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/sqlalchemy/orm/relationships.py\n  Comments left for invalid file neutron/agent/common/ovs_lib.py\n  Comments left for invalid file .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/dsvm-functional/lib/python3.6/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/sqlalchemy/orm/relationships.py\n  Comments left for invalid file neutron/policy.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/dsvm-functional/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/ovs/db/idl.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/docs/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/paste/deploy/loadwsgi.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py\n  Comments left for invalid file neutron/tests/unit/agent/ovsdb/test_impl_idl.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_context/context.py","accounts_in_message":[],"_revision_number":2},{"id":"3f857aec94e6bb2d3e9174f5b87dbb6b29eab744","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-04-04 07:19:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"906e7d6a87d69447708e7ee2a8607e4cb707f807","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-04 09:10:42.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/715419/3 : SUCCESS in 1h 50m 26s","accounts_in_message":[],"_revision_number":3},{"id":"ef0b8ccb3dcea9bfc75b7bb9016fe552f0fff75f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 10:30:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/110283a12cff463f810c3ebe371fe0ff : SUCCESS in 1h 50m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c5e7cd2574664aaf9650ffae3b50f45e : TIMED_OUT in 1h 03m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/bf71f5a6f9de4eef8a6eab49b1380edd : TIMED_OUT in 3h 09m 07s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bf2bd9250d7944359c0ffc0def014d5d : SUCCESS in 2h 00m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/99337c23ed7f4c26af91215e1e2c3a59 : SUCCESS in 1h 02m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6bfc7085252c4ec99eabea9c5024dc93 : SUCCESS in 2h 09m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a899effdda584736a63238f357570f86 : SUCCESS in 40m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14d42e6dbeeb4c9dbb1d0689b008df33 : SUCCESS in 26m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36209831603c4f76b948aa007a5e4eba : SUCCESS in 25m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ed582c7bfa3452ab0038fcb2fe8989f : SUCCESS in 45m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a71ea96bb3d743acbf0711c626479088 : SUCCESS in 49m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/181d226510a34b1399f2bd4bc2d68188 : SUCCESS in 24m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07b51c4d64c04b5d82b3ca6b96023061 : SUCCESS in 27m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e155566aa335487db716cd33127742df : SUCCESS in 1h 44m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/75935bc074664b65903dadda301424c1 : FAILURE in 1h 11m 16s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b6ef9c50543346fd85158b948750ed82 : SUCCESS in 2h 28m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1ade10959aee4acf8c1f9e9315347a17 : FAILURE in 57m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/97116a64a9014c85820f732c52525144 : FAILURE in 1h 33m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/60d5e388e2b84592991c88048de4413d : FAILURE in 57m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1ccb838bf4304f4d9d423f797f8cab2e : SUCCESS in 1h 13m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fa2fe0095db445e5bd91d062682a9561 : SUCCESS in 2h 10m 55s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bcb457be06514751b73f2d35b379a60d : FAILURE in 1h 42m 12s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1b420cbfd54544f6910f58a1dbb2a48a : SUCCESS in 1h 25m 32s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fe09e27b333148819df0c0b121e35516 : SUCCESS in 1h 15m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/50a50e9d100c48709dfbc4cabb329a33 : SUCCESS in 2h 33m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/617e5bb7c5ac449381764fe84aca63b8 : SUCCESS in 2h 41m 51s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e046ea1daf6b4d0da0d1a66da6cda56e : SUCCESS in 1h 58m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02b96a9e6ca546ddbe4702949dc5c4a7 : SUCCESS in 59m 53s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/36fbaf6a902046f086893027f472c2df : SUCCESS in 1h 12m 54s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1f9b278e23394b569eb78bac9fb90f86 : SUCCESS in 58m 28s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a824a67f83ab4562b24f2051b766a661 : SUCCESS in 1h 30m 48s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/528fe991ebc14b9d9c081eba81579214 : SUCCESS in 1h 34m 00s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ebf5ce8d2fe14d6293068044bcccb785 : SUCCESS in 1h 46m 05s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0ab76229c7844a749cd82c643fed405d : TIMED_OUT in 2h 14m 34s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0fa566578df845b59455d1ee9c9d9bfd : SUCCESS in 2h 40m 46s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/17a6c448feee485a86f6469484465ba2 : SUCCESS in 2h 00m 23s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/db34be67203a49aaae02d607b4b2dc71 : SUCCESS in 2h 37m 21s (non-voting)\n\nWarning:\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/neutron_lib/context.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/ovs/db/idl.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/dns/hash.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/neutron_lib/worker.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/unittest2/compatibility.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/sqlalchemy/orm/relationships.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/openstack/config/cloud_region.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_policy/_checks.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_messaging/_drivers/pool.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/neutron_lib/context.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/paste/deploy/loadwsgi.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/ovs/db/idl.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/shared/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/shared/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/shared/lib/python3.6/site-packages/astroid/interpreter/_import/spec.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/shared/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/docs/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job neutron-functional-with-uwsgi: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-functional/lib/python3.6/site-packages/neutron_lib/context.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/neutron_lib/context.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job neutron-fullstack: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/ovs/jsonrpc.py in the commit\u0027\n  Job neutron-fullstack: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job neutron-fullstack: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/ovs/reconnect.py in the commit\u0027\n  Job neutron-fullstack: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/neutron_lib/context.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/oslo_context/context.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/unittest2/compatibility.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/sqlalchemy/orm/relationships.py in the commit\u0027\n  Job neutron-fullstack-with-uwsgi: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/ovs/jsonrpc.py in the commit\u0027\n  Job neutron-fullstack-with-uwsgi: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job neutron-fullstack-with-uwsgi: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/ovs/reconnect.py in the commit\u0027\n  Job neutron-fullstack-with-uwsgi: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/ovs/jsonrpc.py in the commit\u0027\n  Job neutron-fullstack-with-uwsgi: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job neutron-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-functional/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py in the commit\u0027\n  Job neutron-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/dsvm-functional/lib/python3.6/site-packages/neutron_lib/context.py in the commit\u0027\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_policy/_checks.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/unittest2/compatibility.py\n  Comments left for invalid file .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/shared/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file neutron/objects/base.py\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_daemon.py\n  Comments left for invalid file .tox/shared/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/eventlet/patcher.py\n  Comments left for invalid file neutron/tests/unit/test_auth.py\n  Comments left for invalid file .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/ovs/jsonrpc.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/openstack/config/cloud_region.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/unittest2/compatibility.py\n  Comments left for invalid file .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/eventlet/patcher.py\n  Comments left for invalid file neutron/common/eventlet_utils.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/ovs/db/idl.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/dns/hash.py\n  Comments left for invalid file .tox/dsvm-fullstack-gate/lib/python3.6/site-packages/ovs/reconnect.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/neutron_lib/worker.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/shared/lib/python3.6/site-packages/astroid/interpreter/_import/spec.py\n  Comments left for invalid file .tox/docs/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_messaging/_drivers/pool.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/sqlalchemy/orm/relationships.py\n  Comments left for invalid file neutron/agent/common/ovs_lib.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/dsvm-functional/lib/python3.6/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/sqlalchemy/orm/relationships.py\n  Comments left for invalid file neutron/policy.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/dsvm-functional/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/ovs/db/idl.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/paste/deploy/loadwsgi.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py\n  Comments left for invalid file neutron/tests/unit/agent/ovsdb/test_impl_idl.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_context/context.py","accounts_in_message":[],"_revision_number":3},{"id":"0442b9088a2873db64c62715b77285ea3f30aec3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-04 22:14:14.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/19/715419/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"2aeaa9be756c7e88d9cea04a7bb3be9bd95e77fa","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-04 22:34:16.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/19/715419/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"a077e20111f16ef01301f7d4fe5a402ced252ef8","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-04-05 01:55:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7c3579ca176002b0d147db1862f3ce7a82c09541","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-05 03:40:20.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/715419/3 : SUCCESS in 1h 43m 58s","accounts_in_message":[],"_revision_number":3},{"id":"aec1aa3734da35048c05e137c1c1df18db224884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 05:13:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ff5bc1c0e1e64f548a8186e60432ca78 : SUCCESS in 47m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fbebc4f601764fecac71b61076eda131 : POST_FAILURE in 48m 29s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/9c7cd50b9a1e4e51b141531bc6676827 : TIMED_OUT in 3h 04m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fe38651a56fd4e91b7c8a51bff10250d : TIMED_OUT in 2h 51m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e8e8e667987d49b79a16c5cd583c8dd8 : SUCCESS in 1h 01m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9fdb2c7d2edd42f78692e7f33f8df686 : SUCCESS in 1h 03m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f14605425e414a9d9e1d39bf09a0f65e : SUCCESS in 28m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18b22cfe45ee43c5ac28689133c9c044 : SUCCESS in 23m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0706f3370bbe4111be77dda723b7c7fa : SUCCESS in 11m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b516919e9ca4895b3801cc32eba1db9 : SUCCESS in 24m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/267b3b0cd49b477fa48b79f6f95a5baa : SUCCESS in 23m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00fe3f277d5f4c55834380fba0dca1e8 : SUCCESS in 29m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/790ab80cc42d4b21be25fe4b169f0f6e : SUCCESS in 9m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0a906af9681a4a70acd2f98d82583ef1 : TIMED_OUT in 2h 18m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b8c62abf7a3c40fa95111f8e8bf39603 : FAILURE in 59m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e67c80ef48ce4430b7e9ce00e0c80776 : TIMED_OUT in 3h 04m 01s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4a5a6d9e36cc447e82d382a4df3958d0 : SUCCESS in 1h 07m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7013f7e7e29c41ad986c018f2ea0acdb : FAILURE in 2h 13m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/16f815e71f8547a58fa5e835b98c1820 : FAILURE in 42m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0db61e891c7e462a9c24e6f9e91d9e79 : TIMED_OUT in 2h 09m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/56ba8cda37014ceeb00a8b0dc553f072 : SUCCESS in 1h 32m 23s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/190748490b3945f08bd014ce26e66510 : FAILURE in 1h 37m 15s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fd5ec647b53649e581bf9d6a48068c79 : SUCCESS in 1h 38m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0f002ac829cf44f494fa5ae6ffb27dbb : TIMED_OUT in 1h 35m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0c3cbefba01842b38a45a513e2af2cc4 : SUCCESS in 2h 35m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7dafa42f6fe04f239ea0e446f65dc8e9 : TIMED_OUT in 3h 06m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0a04a4c8340d4e609b079ad1e5a645ba : TIMED_OUT in 3h 07m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/375f9d164cef4ff685a7dc799e00a9e2 : SUCCESS in 1h 22m 30s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b26cb70bc1ee46f59199dd1f5783cdeb : TIMED_OUT in 2h 35m 05s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e56531cd2e0348a78d3976fa9760ac16 : SUCCESS in 1h 15m 27s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/647d75e2c29948a399c09037dd57879c : TIMED_OUT in 2h 17m 53s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d3e58b06e0942359556ea39ce933096 : SUCCESS in 1h 38m 21s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2406767a4a8d4037bda127e680f83880 : FAILURE in 2h 50m 18s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3f6b34a66d79497d9816ebc19a36a041 : TIMED_OUT in 2h 14m 16s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/aef268c858cb4e0e92e86ad55eba58ac : TIMED_OUT in 3h 12m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/626dccdd76304ccfb7c42b01be04595d : SUCCESS in 2h 11m 15s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3fd300bce48f4c4daf441cc351536809 : TIMED_OUT in 3h 06m 16s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/policy.py\n  Comments left for invalid file neutron/agent/common/ovs_lib.py\n  Comments left for invalid file neutron/objects/base.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_daemon.py\n  Comments left for invalid file neutron/common/eventlet_utils.py\n  Comments left for invalid file neutron/tests/unit/agent/ovsdb/test_impl_idl.py\n  Comments left for invalid file neutron/tests/unit/test_auth.py","accounts_in_message":[],"_revision_number":3},{"id":"c77d4433762802077f4469115b353c675506ccda","author":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"date":"2020-04-06 08:54:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"25da04b837a2e3a9a90de6cd3098f2e31e043069","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-25 15:43:11.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c5300c3ecbe304ed7dc8b3be73478c4c023786dc","revisions":{"0d751e12d9ff6472276f28ffb0ed41e7283f743f":{"kind":"REWORK","_number":1,"created":"2020-03-27 13:07:05.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/19/715419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/19/715419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/715419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/715419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/715419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/19/715419/1"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-03-27 10:55:48.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-03-27 11:01:53.000000000","tz":480},"subject":"Avoid Layer 2 and Layer 3 traffic flooding in DVR mode.","message":"Avoid Layer 2 and Layer 3 traffic flooding in DVR mode.\n\nThis patch add a new flow table AVOID_FLOOD_TABLE(59) which\nprocess the packets with src_mac is gateway_mac. In order to\navoid l2 flooding, move the original flows to local port in\ntable 60 to the table 59 and add the gateway_mac in the match.\nSo the egress and ingress traffic of l2 traffic will always\nuse the NORMAL flow in table 60 and the mac address of remote\nports will be learned.\n\nFor the l3 traffic, this patch add flow in table 0 which\nresubmit the packets from router interface to table 59. There\nare two kinds of flows in table 59. One is to process the\npackets which from the local router to local ports. The other\nis a low priority flow to process the packets which sent from\nlocal router to remote ports and the packets sent to patch_tun_port,\nso the egress traffic will be directed to the patch_tun_port\nand will not to be flood.\n\nChange-Id: I257cc4966379ddacbcaa6fbe8552e638cda303f3\nCloses-bug: #1866445\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d751e12d9ff6472276f28ffb0ed41e7283f743f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d751e12d9ff6472276f28ffb0ed41e7283f743f"}]},"branch":"refs/heads/master"},"72fdf81b29d2aae79ec789bae5a6fa35a0e08a00":{"kind":"REWORK","_number":2,"created":"2020-04-04 03:18:32.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/19/715419/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/19/715419/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/715419/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/715419/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/715419/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/19/715419/2"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-03-27 10:55:48.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-04-04 03:10:10.000000000","tz":480},"subject":"Avoid Layer 2 and Layer 3 traffic flooding in DVR mode.","message":"Avoid Layer 2 and Layer 3 traffic flooding in DVR mode.\n\nThis patch add a new flow table AVOID_FLOOD_TABLE(59) which\nprocess the packets with src_mac is gateway_mac. In order to\navoid l2 flooding, move the original flows to local port in\ntable 60 to the table 59 and add the gateway_mac in the match.\nSo the egress and ingress traffic of l2 traffic will always\nuse the NORMAL flow in table 60 and the mac address of remote\nports will be learned.\n\nFor the l3 traffic, this patch add flow in table 0 which\nresubmit the packets from router interface to table 59. There\nare two kinds of flows in table 59. One is to process the\npackets which from the local router to local ports. The other\nis a low priority flow to process the packets which sent from\nlocal router to remote ports and the packets sent to patch_tun_port,\nso the egress traffic will be directed to the patch_tun_port\nand will not to be flood.\n\nChange-Id: I257cc4966379ddacbcaa6fbe8552e638cda303f3\nCloses-bug: #1866445\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72fdf81b29d2aae79ec789bae5a6fa35a0e08a00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72fdf81b29d2aae79ec789bae5a6fa35a0e08a00"}]},"branch":"refs/heads/master"},"c5300c3ecbe304ed7dc8b3be73478c4c023786dc":{"kind":"REWORK","_number":3,"created":"2020-04-04 07:19:43.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/19/715419/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/19/715419/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/715419/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/715419/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/715419/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/19/715419/3"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-03-27 10:55:48.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-04-04 07:15:56.000000000","tz":480},"subject":"Avoid Layer 2 and Layer 3 traffic flooding in DVR mode.","message":"Avoid Layer 2 and Layer 3 traffic flooding in DVR mode.\n\nThis patch add a new flow table AVOID_FLOOD_TABLE(59) which\nprocess the packets with src_mac is gateway_mac. In order to\navoid l2 flooding, move the original flows to local port in\ntable 60 to the table 59 and add the gateway_mac in the match.\nSo the egress and ingress traffic of l2 traffic will always\nuse the NORMAL flow in table 60 and the mac address of remote\nports will be learned.\n\nFor the l3 traffic, this patch add flow in table 0 which\nresubmit the packets from router interface to table 59. There\nare two kinds of flows in table 59. One is to process the\npackets which from the local router to local ports. The other\nis a low priority flow to process the packets which sent from\nlocal router to remote ports and the packets sent to patch_tun_port,\nso the egress traffic will be directed to the patch_tun_port\nand will not to be flood.\n\nChange-Id: I257cc4966379ddacbcaa6fbe8552e638cda303f3\nCloses-bug: #1866445\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5300c3ecbe304ed7dc8b3be73478c4c023786dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5300c3ecbe304ed7dc8b3be73478c4c023786dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
