)]}'
{"id":"openstack%2Fneutron~715445","triplet_id":"openstack%2Fneutron~master~I3f7de633e5546dc78c3546b9c34ea81d0a0524d3","project":"openstack/neutron","branch":"master","topic":"bug/1869389-ha-chassis-grp-update-no-ext-ids","hashtags":[],"change_id":"I3f7de633e5546dc78c3546b9c34ea81d0a0524d3","subject":"[OVN] HA Chassis Group: Ignore UPDATES when external_ids hasn\u0027t changed","status":"MERGED","created":"2020-03-27 15:27:10.000000000","updated":"2020-03-28 03:43:18.000000000","submitted":"2020-03-28 03:41:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715445-1585366862941-93ed6149","meta_rev_id":"87c91b2b47bad7a7dac7ed101432fd91227d1f1e","_number":715445,"virtual_id_number":715445,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"date":"2020-03-27 17:30:31.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-27 20:52:04.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-28 03:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-27 22:21:13.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-03-27 15:54:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-03-27 19:19:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2020-03-27 15:55:27.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2020-03-27 15:55:27.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-27 15:54:59.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-03-27 15:55:27.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-03-27 17:30:31.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-27 19:19:46.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2020-03-27 22:21:13.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-03-28 03:41:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c53fa613b2a2b231cad4c7b552bd9130b14af441","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-03-27 15:27:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f0962ddef5ddb7d9fd896b0784c3aebb1d710aa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-03-27 15:31:04.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nWaiting for CI","accounts_in_message":[],"_revision_number":1},{"id":"5cd2790a73722fc83ca8ccc14f0c57753cff3969","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-03-27 15:32:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40abf948d59f13bb6f972554fe922bff12d65d3e","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-03-27 15:35:29.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nI\u0027d add some description of the bug that this thing is solving at least in the commit message (or even a LP bug).\n\nie. Any updates on the Chassis table may trigger flapping of all the external ports causing disruption.","accounts_in_message":[],"_revision_number":1},{"id":"842d5c38531bf9b4876521e3599c15cdcdda2fd0","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-03-27 15:53:54.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"890ae1a28080c0a783d41b185b97645310b0a791","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-03-27 15:54:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8b660bcfb1a465923afb5f7606d1433c76eab8f9","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-03-27 15:55:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8e038b88f18166bc4a0165e16e41b5af56205031","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-27 17:03:09.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/45/715445/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"98657c4dd9217f5b688d725210c327caa0428909","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-27 17:06:31.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/715445/2 : FAILURE in 1h 12m 04s","accounts_in_message":[],"_revision_number":2},{"id":"9381af508963ce4c3dbecaf101b6076c6306d976","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-27 17:30:31.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/45/715445/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"46dfb044249e4b71799dc56f67c900444a6a8308","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:52:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/59c0ab2f5f3143cc85981aae319cf738 : SUCCESS in 59m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/98d05faa79a64649af4d37251ef6ffa4 : SUCCESS in 38m 54s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/27121e7f07db4ebf9b2bf3d3b13e3a06 : FAILURE in 1h 54m 45s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/437ea3734efc445cbbae513a39d6c030 : SUCCESS in 1h 05m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7d8c236667e4413889fe4002be13d0c2 : SUCCESS in 1h 09m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/666a5bf8cfc84f8b864fd36b9dd0a2a4 : SUCCESS in 1h 16m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/861905d707d2434dba49c2211208e37c : SUCCESS in 37m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2dbda88a09a4442b4cc5045f6074f17 : SUCCESS in 27m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/425e8ec33cfe4ef68d040a74dfa668ab : SUCCESS in 16m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fd03889802d4d21addee9b4cbbce48a : SUCCESS in 32m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d22bd112ce8a41f5b95cb2d0e235c2b3 : SUCCESS in 30m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a78775693c744436ac4aa28f04d4245f : SUCCESS in 29m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ea2f3785ee240f886c12e1ecba85c8a : SUCCESS in 12m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9115df82a272440999923b4e3b94575b : SUCCESS in 52m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b665eeb217ce4e74b5448a59ea303af9 : SUCCESS in 1h 22m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5e9d84954db54717a401c0633781219c : SUCCESS in 2h 10m 48s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d69397847b0f4e569c8ac7189d43882f : SUCCESS in 1h 21m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d9120b8f5174f96a0f8d43d6dfdf6b1 : FAILURE in 1h 29m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f19520c5e19d463aad1977ff700b3da0 : SUCCESS in 1h 24m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/da8c8051d34c4d3e8d04cbb0c7fcb955 : SUCCESS in 1h 13m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3268e4ce7dfd4ff88845af19e067d5cf : SUCCESS in 1h 42m 40s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0d43c16f3df642ccbaa919d77a747041 : FAILURE in 1h 28m 00s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5e531d2859a8458984fda5fd30c362ac : SUCCESS in 1h 20m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4f1f8b2366ef4824a3b850636e2ca7de : SUCCESS in 58m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3b8ed2ee83854657b03ca720e36adf26 : SUCCESS in 2h 13m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9afa386605c24dca931585788eaf22f2 : SUCCESS in 2h 21m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8b21c3bf746e4d469b5b2b8fc398b4dd : SUCCESS in 2h 16m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1319798e41744e88b3d7ffa5798d962 : SUCCESS in 1h 30m 36s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fdf60d36c93d4bf696774eec634f4fb0 : FAILURE in 39m 58s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f59941eec22b4eb78efd98fcc91c2f96 : SUCCESS in 47m 19s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a0e590600e14457c90e47c867de5905b : SUCCESS in 1h 09m 20s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/554e711e493647deb3af4ed3dca5fa5f : SUCCESS in 1h 29m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bf157a811b264205a33ccb23f9efda38 : SUCCESS in 2h 01m 10s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/51110990af084cc38c9131284a44abbc : SUCCESS in 2h 21m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"03f6bfe598930238914451e9ee8218db49de51f7","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-27 19:19:46.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)\n\njust a nit ;)","accounts_in_message":[],"_revision_number":2},{"id":"969a3c914385b4629185433e2e516e923c65bbdb","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-03-27 20:52:04.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d7217079cb3369fea960b3a8023d1164eee1fb28","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-27 21:34:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"29d9283c8a8327c1c63b9a43581e1e6c7e98e35c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-27 22:21:13.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/715445/2 : FAILURE in 46m 15s","accounts_in_message":[],"_revision_number":2},{"id":"336c42fcc4c00506952ce5c94dd128690413d879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 01:25:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/640542b7dcaf488e93cc8dc2d1ae1561 : SUCCESS in 53m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1d1531f69d4a41f0bf71a0d97d50ee56 : SUCCESS in 48m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/0d0b20966b8744d2948d12dd22112055 : FAILURE in 2h 27m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a48e65b3fb804bf79a0a33da40f504d3 : SUCCESS in 1h 05m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1b1bbef6e1d04fb0adbf4e73f6ad6ff8 : SUCCESS in 1h 48m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/54083971ece74ba8a1de702f3f3a3266 : SUCCESS in 1h 30m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f13ce17d89a4eb68933a852392ee146 : SUCCESS in 37m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a83eb65ef0641ea85029fc4942f9791 : SUCCESS in 24m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/555ae051635547978e3713797413a8d8 : SUCCESS in 14m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acc7e334553e4ef088396be5100ccf53 : SUCCESS in 35m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/588a34960d674fb5b15c9454e30ecc52 : SUCCESS in 28m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29e18b1afc8e48d2bf5acec82662b42d : SUCCESS in 27m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f920c66feca34d17afc70bf02015e5b2 : SUCCESS in 9m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6b4a9c0439c046c8bbdf7efe2d200f30 : SUCCESS in 1h 05m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/751a7da934dd40b7b4f9ea5bcb6835f9 : SUCCESS in 1h 07m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f62217c991474ad3a90bf4f0e5b6fe61 : SUCCESS in 2h 15m 15s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8284410c1af8464189ebb89721c291c7 : SUCCESS in 1h 12m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/03f0805bd55747d5bd7bb6d2129b3942 : SUCCESS in 1h 02m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a0fc76bc56cb4c41871b1ca66d8de565 : SUCCESS in 1h 12m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f2eb8efa9d3f43c4bee5303d182baf61 : SUCCESS in 1h 07m 13s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6edbe25c552a43d4aec35ff235093bd9 : SUCCESS in 1h 55m 42s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/81ec7d4fc8b34c7aa2c360b0608b66f8 : SUCCESS in 2h 15m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/29a64a4c43c4427dbed316afbbb07c9c : SUCCESS in 1h 10m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6adbd2d5f4884bd6b664111ac10a4884 : SUCCESS in 55m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a1d7db3c642d483b9b73d9648b1f2570 : SUCCESS in 2h 39m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d2c14a38910f4f9799e969791a42abc9 : SUCCESS in 1h 51m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3f9cc2b10d204904aca7659d55c2ffd9 : SUCCESS in 2h 14m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bb4a2f6ab0d4e4cabd075d2b488e188 : SUCCESS in 1h 06m 34s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/08255bcc70614825a1fa507d1539f07f : FAILURE in 35m 42s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b062152387544a1f86cbf70b937a59c0 : SUCCESS in 48m 13s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d06a6ae238214160835bb0d904ec2f31 : SUCCESS in 1h 19m 37s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b21c221eab74c6eada90b951aa5e20e : SUCCESS in 1h 09m 45s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7fa865fa910b45f1a9716048d49c2e28 : SUCCESS in 1h 42m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8d55c5f7bc0241d3bf644b9c5015beff : SUCCESS in 1h 59m 51s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d10387fc4bd24a63884903f8a6898c2f : SUCCESS in 2h 48m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"41564e32b26ef1f949dd552671812d3b4c6ecdfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 01:25:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"12af8276765a88c6d4a43d755ef0437ed23072c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 03:41:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4e33827d7f85440c88cad76df2642bfa : SUCCESS in 54m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/13841181c4ae4015b96441020bb2ee6b : SUCCESS in 1h 22m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0491bb2c69dc4ca2839a41d2c88d9a5c : SUCCESS in 1h 13m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/08e3dd1b4a414e6dbf6ff64396eb683f : SUCCESS in 1h 05m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da8effdab0504f5a835105756668f802 : SUCCESS in 25m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8167c3c3fa645bcada6776d9fc682bf : SUCCESS in 13m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cae8d22f66034f7bb99b252be245e09f : SUCCESS in 23m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aeb316487619412d916937470951ae0e : SUCCESS in 27m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92088717992844a6a87fdc8a4426a8e2 : SUCCESS in 8m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e8d9da9f7df8439db546ab79163dbb4a : SUCCESS in 47m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b0cc2659976f45248f08d6c92126e1e6 : SUCCESS in 1h 23m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/68e931099242454f8d3217205557fd50 : SUCCESS in 1h 22m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9a287b3eb4b9400eb9500e4806622bfb : SUCCESS in 1h 26m 25s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/965e20e39f0249829616100e0751c15a : SUCCESS in 1h 07m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3f54b83a16d4c18929543734b5c7219 : SUCCESS in 1h 25m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e8d3ec97985443258ed57937a55e5af4 : SUCCESS in 1h 09m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3cb87053fe2a4a798d0fb01625e7fb8d : SUCCESS in 2h 10m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3427fd6920784292951247935eaba7d2 : SUCCESS in 1h 12m 38s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/80cac9ffa91543319c8d493522f7568c : SUCCESS in 1h 53m 16s","accounts_in_message":[],"_revision_number":2},{"id":"40fb3970b4c17220e248e58037e0d9de4d04c44b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 03:41:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"817ce056e3e986ff31f6cf299c6202d2a575f130","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 03:43:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6d7f9dc3a8a42ddbc09d27521243373 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ea999564a5b80dcf13c0c43f107165f0754210b7","revisions":{"73690fa1da873cad4cce621149c01226089698d9":{"kind":"REWORK","_number":1,"created":"2020-03-27 15:27:10.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/45/715445/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/715445/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/715445/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/715445/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/715445/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/715445/1"}}},"commit":{"parents":[{"commit":"8b2bda7be5dcd1ca1bd67f356cf3e2c8eb00137b","subject":"Merge \"Prioritize port create and update ready messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b2bda7be5dcd1ca1bd67f356cf3e2c8eb00137b"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-03-27 15:23:49.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-03-27 15:23:49.000000000","tz":0},"subject":"[OVN] HA Chassis Group: Ignore UPDATES when external_ids hasn\u0027t changed","message":"[OVN] HA Chassis Group: Ignore UPDATES when external_ids hasn\u0027t changed\n\nThe \"old\" parameter passed to the handle_ha_chassis_group_changes()\nmethod is a delta object and sometimes it does not contain the\n\"external_ids\" column (because it hasn\u0027t changed).\n\nThe absence of that column was misleading that method into believe that\nthe \"old\" object was no longer a gateway chassis and that triggered some\nchanges in the HA group.\n\nThis patch is adding a check to verify that the \"external_ids\" column\nhas changed before acting on it, otherwise just ignore the update and\nreturn.\n\nChange-Id: I3f7de633e5546dc78c3546b9c34ea81d0a0524d3\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73690fa1da873cad4cce621149c01226089698d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73690fa1da873cad4cce621149c01226089698d9"}]},"branch":"refs/heads/master"},"ea999564a5b80dcf13c0c43f107165f0754210b7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-27 15:53:54.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/45/715445/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/715445/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/715445/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/715445/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/715445/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/715445/2"}}},"commit":{"parents":[{"commit":"8b2bda7be5dcd1ca1bd67f356cf3e2c8eb00137b","subject":"Merge \"Prioritize port create and update ready messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b2bda7be5dcd1ca1bd67f356cf3e2c8eb00137b"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-03-27 15:23:49.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-03-27 15:52:13.000000000","tz":0},"subject":"[OVN] HA Chassis Group: Ignore UPDATES when external_ids hasn\u0027t changed","message":"[OVN] HA Chassis Group: Ignore UPDATES when external_ids hasn\u0027t changed\n\nThe \"old\" parameter passed to the handle_ha_chassis_group_changes()\nmethod is a delta object and sometimes it does not contain the\n\"external_ids\" column (because it hasn\u0027t changed).\n\nThe absence of that column was misleading that method into believe that\nthe \"old\" object was no longer a gateway chassis and that triggered some\nchanges in the HA group. Changing the HA group resulted in the SRIOV\n(external in OVN) ports to start flapping between the gateway chassis.\n\nThis patch is adding a check to verify that the \"external_ids\" column\nhas changed before acting on it, otherwise just ignore the update and\nreturn.\n\nCloses-Bug: #1869389\nChange-Id: I3f7de633e5546dc78c3546b9c34ea81d0a0524d3\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea999564a5b80dcf13c0c43f107165f0754210b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea999564a5b80dcf13c0c43f107165f0754210b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
