)]}'
{"id":"openstack%2Fneutron~716049","triplet_id":"openstack%2Fneutron~master~Ifcaf1c21bea0ec3c35278e49cecc90a101a82113","project":"openstack/neutron","branch":"master","topic":"unittest.mock","hashtags":[],"change_id":"Ifcaf1c21bea0ec3c35278e49cecc90a101a82113","subject":"Complete removal of dependency on the \"mock\" package","status":"MERGED","created":"2020-03-30 18:58:10.000000000","updated":"2020-05-10 12:59:14.000000000","submitted":"2020-05-10 12:57:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716049-1589115459144-b16e9fe7","meta_rev_id":"ccbd2fbe83d2f550d9e9b71b95eec363521440f8","_number":716049,"virtual_id_number":716049,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-10 07:58:10.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-01 21:23:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-05-10 08:04:48.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"date":"2020-05-01 21:59:58.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-10 12:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-10 09:19:16.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-05-09 12:33:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2020-05-04 10:19:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-05-09 22:44:36.000000000","_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":2,"date":"2020-05-04 06:18:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-05 12:39:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-05-09 22:44:36.000000000","_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-08 06:41:16.000000000","_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-01 21:23:27.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-04 06:18:35.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-05-04 10:19:50.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-05-05 12:39:59.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-05-08 06:41:16.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-05-09 12:33:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-09 22:44:36.000000000","updated_by":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"reviewer":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"state":"REVIEWER"},{"updated":"2020-05-10 07:58:10.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-05-10 08:04:48.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-10 09:19:16.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-05-10 12:57:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a22561c94726a1d07f9359aa93475df0a13e6302","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-30 18:58:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c1777c66292d39d9dc2e53aff2ea521e67aecab","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-30 19:06:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"37dc34edb37e10d8dcaaa1f30af5239a3f07094d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-30 20:47:07.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/716049/2 : SUCCESS in 1h 39m 50s","accounts_in_message":[],"_revision_number":2},{"id":"93b2ae2da93633b470c074fcc71328e70cab88ee","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-30 21:04:00.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/49/716049/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"0e4cadbb4b6b770eb0334a0de64cc4cc47ad033c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 22:20:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fadade7c48954677a4ef2ce431d9d3f0 : SUCCESS in 51m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/91a8f4a4808d4fc19e251508139536b3 : FAILURE in 34m 04s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/5018ce0d014b460590ab091f6cc6378e : FAILURE in 2h 05m 05s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7d7926a69ea04b70bd8352629667f334 : SUCCESS in 1h 19m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6a582be3f11346e4ad0c32ce7aa51fdc : SUCCESS in 1h 20m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4a585d8bdab44282b7db94819ac0c01a : SUCCESS in 1h 26m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d18e162fdca24261883ca821d54a32d5 : SUCCESS in 45m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f7f5cab495747baa71e36025d4822f9 : SUCCESS in 26m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71d338ee6b364e8dae9e36b88cc0c98b : SUCCESS in 13m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/569b5a1a63fe428ea78f7cd048290c33 : SUCCESS in 25m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd7923509d4f4ec3aa90353d1d82578c : FAILURE in 27m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33b8aa11baa34084b723cc2fea03b271 : FAILURE in 26m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bffa470282d4c5cbb61b20303169ecf : SUCCESS in 10m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9c2043f67b3a4ef5bdc96a82e7626ab8 : SUCCESS in 3m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/795979051b1a4b42a440783ebdc114b0 : SUCCESS in 16m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/08cc8291a4304ad18ae725f687a06ad3 : SUCCESS in 48m 53s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2244dc4c5a16429782a1dd3032d342f3 : SUCCESS in 1h 18m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/aec07150f213451e82d1f85175e6138b : SUCCESS in 1h 54m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/15e0db9eee1540bc91e89ac655b0a89f : SUCCESS in 1h 24m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e81a7dae7a5e466289356d3f2561e354 : FAILURE in 1h 01m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e4aa04cd8097406f987f55f8e180b151 : SUCCESS in 1h 04m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/05bdb0c2653043b0bb24c8603f68c0ac : SUCCESS in 1h 08m 33s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dbdeecec99ef46429d0ce62108ac30a3 : SUCCESS in 1h 47m 37s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b2e749137af34c28a8c39c4313dacf72 : SUCCESS in 1h 55m 51s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6f4eb638c6c84b8ca61b09a5447efe90 : SUCCESS in 1h 20m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3ebb8941c20443d8a12f45cf38effe8d : SUCCESS in 57m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3e9cde09b147480cb6a275e0045949dc : SUCCESS in 2h 16m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/488ad09c067348bdab738f06527695fe : SUCCESS in 2h 30m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f10ff6763b704982962219aaa81c33d1 : SUCCESS in 2h 18m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aef91208ca4a4699a4316b1fa93f237e : SUCCESS in 1h 02m 10s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0c3f070c1c984d5d95752e92c8d83628 : SUCCESS in 1h 14m 25s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c99dcada9d474fd0a7b62a09b848ffa2 : SUCCESS in 58m 42s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd36ce8c0dc24ed1877319db53c88fcc : SUCCESS in 1h 22m 33s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/569588066f9e4b27b3e0a04eefbed78d : SUCCESS in 1h 15m 35s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cea58a6df2044d89a288de2a22f1467e : SUCCESS in 1h 45m 08s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/274ea670b6074fe898af35c3cb6cf295 : SUCCESS in 1h 59m 47s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/258d7ee42e554fd48dbf3999ddd7e992 : SUCCESS in 2h 18m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c8177c61799f51351c01f7608a11f9ae12987717","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-30 22:20:12.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/49/716049/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"d91de4790984ee9e7fa40c7e49c8530e2b147955","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-20 20:41:15.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"39221a00b8371ae50e3116e6392db11b69f0d0d9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-20 22:21:37.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/716049/3 : FAILURE in 1h 39m 42s","accounts_in_message":[],"_revision_number":3},{"id":"ddec57797f7b4b7a8beaa6457acf700b8002f5f9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-20 22:29:27.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/49/716049/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"f906964e0bd39f3d52b63fc0cdd5a69255bbbbc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 02:42:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f10f7270d6064f9b9cb6ed49e578bde4 : SUCCESS in 1h 04m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3109748e9c0a432090f1acddc679113c : SUCCESS in 44m 01s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/0eaddf482d254b289d623389dc180d94 : FAILURE in 2h 07m 03s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b65390ee469940e2811a2c3e44157852 : SUCCESS in 1h 20m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d7b947b136ff48f4b2de12b28e2e9a37 : SUCCESS in 1h 28m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7d3bff08e8554815a98bf4067002c546 : SUCCESS in 1h 01m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ece4a8ed6b38406c9d132c8257914b8d : SUCCESS in 31m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a82fd42a0fe48988ef30b53ab22fb7a : SUCCESS in 24m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f25be56055994caaa124bc1ec1a7e528 : FAILURE in 9m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/756a877bc4634d6c808a759f27544677 : SUCCESS in 28m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ff1e4460a2d4df187d24a9df988bd9f : FAILURE in 28m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/059c58f545b24da5836078f4f308e8ff : FAILURE in 27m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7767b7c1bfbe4c4eafc7e733a71299dc : SUCCESS in 11m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/64d9fdd7a1d0439cbbebf248071b4f8d : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1117c22a752845bb8356c1450d161568 : SUCCESS in 14m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/018f3fac4d1540caa8e9475620c19c23 : SUCCESS in 51m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5b4bf2f1c5cd40f3b3002973a7a082bd : SUCCESS in 1h 35m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a83077ddde48445399eb62fe49d543dc : SUCCESS in 1h 14m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c71453c8b3194492b80bb11c6a01a084 : SUCCESS in 1h 52m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0b163509930041c2b8661adb8b7b45b6 : SUCCESS in 1h 20m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c651831bde184fdeb5dd53491e1cda66 : SUCCESS in 1h 09m 48s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/004025396f89451d90a5357558a5f55e : SUCCESS in 1h 13m 16s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/fcded77b5f7644aeb8db0addf13b35af : SUCCESS in 1h 37m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c91abb811d94477494a14b77a89c4756 : SUCCESS in 1h 05m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2d296b1ceed144ca831e98860697a430 : SUCCESS in 1h 36m 24s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d93afb2f01f046c1979e7e72de1b9347 : SUCCESS in 1h 37m 47s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/867a2fd097914458b6d16e6291262419 : SUCCESS in 1h 17m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f545e27e4f1b43b082a91f6b02cb8a9a : SUCCESS in 54m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fccb709c9bde445aad8b5e786683058f : SUCCESS in 2h 17m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2266aa5be51e42589ce4878ad6f26726 : SUCCESS in 57m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/06a20da0fb28403794d2d40ae3c080ee : SUCCESS in 1h 53m 27s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/80b5cc291fd84eba9f67740431e434dc : SUCCESS in 1h 40m 33s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9f3758fa20524c82b63d260d224c6031 : SUCCESS in 1h 15m 25s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3583fa38e504445d8714cd6415979a66 : SUCCESS in 46m 47s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/32aeb05084434f489957608d5076e8d8 : SUCCESS in 1h 38m 40s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f0c479670572493d9ef28aff37a7bb18 : SUCCESS in 1h 45m 37s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/21f3a30cf3e947a18328bd05d48b29bb : SUCCESS in 2h 34m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/818c91902768470a99828aebb05162de : SUCCESS in 1h 52m 36s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e473a13962d94d878d3868f687e05d0b : SUCCESS in 2h 33m 01s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/tests/unit/objects/test_subnetpool.py\n  Comments left for invalid file neutron/tests/functional/agent/linux/test_tc_lib.py\n  Comments left for invalid file neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/ovsdb/extensions/test_qos.py","accounts_in_message":[],"_revision_number":3},{"id":"cb376c8d4e989c4751eeaf710096d0254cafc603","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-24 21:38:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6ad67fc0b25ea4a37216d89a0fb124ed7700beca","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-24 23:07:01.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/49/716049/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"27ce9c80d398e420d625c3113b0e34c015d944e9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-24 23:25:39.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/716049/4 : FAILURE in 1h 46m 26s","accounts_in_message":[],"_revision_number":4},{"id":"2e1da980abc3c22d6d514f107c836502d49d0139","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-28 21:54:37.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"58b049644e100082cc1f2ac91b0addc8d0ae592a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-28 22:06:22.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"3ac4e99c32dc5ac95c70c658f024d7c29a59aedf","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-28 22:28:26.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/716049/6 : FAILURE in 21m 22s","accounts_in_message":[],"_revision_number":6},{"id":"d04e8bc8402dceee5e756375c1749c9f2f10e654","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-28 23:11:39.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/49/716049/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"2c6935364042248c74ec8e2ab0710caef6399450","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-29 00:37:57.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/49/716049/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"eea5e9c6a171f0e69f4162b83ce24e85e7eca75e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 01:51:36.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5ab44d4cdcc843a79d8ca427b2f29cfb : SUCCESS in 56m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/47907a861ef8438dab96a1cf6a45e55c : SUCCESS in 37m 01s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/020523395210487e844a00c433f5440a : FAILURE in 1h 52m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/50158b5e221848c9a1f5c10a21ec6c5f : SUCCESS in 1h 22m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/da9082c66a98493bbf8206aacb7fe504 : SUCCESS in 1h 23m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1b3e3032c4ae43e494932a87fc6db173 : SUCCESS in 1h 14m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e71dc3c61024fc59d5e439c99f6925e : SUCCESS in 35m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62ed01123bdb443392a172dcc12adcdf : SUCCESS in 30m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f697a88f0c2a401fac18c7b459b14915 : SUCCESS in 14m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00614a92f3a442f896c8e55bdcfb0ded : SUCCESS in 32m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3538eb2f45441249bd7febe785c1379 : SUCCESS in 25m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/563c57ebb9814fbd8450db61915159eb : SUCCESS in 10m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/56c9fa0560364327ba673d6c4492453b : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5026fb555671450dacdf40dda054ba0a : SUCCESS in 28m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0d1d8873edca4476865c3f0f6ef71322 : SUCCESS in 56m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/569301ed37b5463e9a073e04a8cd60e9 : SUCCESS in 1h 15m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/30a5b14f50ed4ef98222d2769da2984f : SUCCESS in 1h 12m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4d07bdf700334d7a973944b3c913e721 : FAILURE in 1h 58m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6a78f64086aa4d6b8a7b1eb65d5507c3 : SUCCESS in 1h 27m 53s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fcd96edef9e046ba94ab405cedc82903 : SUCCESS in 1h 11m 32s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1beaece8eac14b0f9d11a6c7909bcfaf : SUCCESS in 1h 19m 29s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7ff7d8f048d147ca9b2e013dd1cd4ca4 : SUCCESS in 1h 07m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b40726903ad644519f517f7ed3f133da : SUCCESS in 1h 10m 45s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4a5b67318d25430cb0166639c61058c8 : SUCCESS in 1h 42m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f566fad5e90949c9b9b75753ae0d0380 : SUCCESS in 1h 43m 48s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8dce8cf9b3ef4848a9b224d3c865eaa6 : SUCCESS in 1h 11m 01s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5600dcc006bf4da7b24802e2b7f3d409 : SUCCESS in 54m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2a8119f2e1964667ae1d266f7724469c : SUCCESS in 2h 17m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60ee87ba8d9c430f917823b5364a08fe : SUCCESS in 1h 00m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0a3c8a63b50a46a4a2f503a20c9af849 : SUCCESS in 2h 08m 33s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5897116c500944c3a3a7fa09a490b7ff : SUCCESS in 1h 58m 33s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/969ae579d1494002aa217f7181693d74 : SUCCESS in 1h 36m 08s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f2c1173e45e346e2ae8ccf1ccc3601c2 : SUCCESS in 49m 38s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2e620b4cab5947fd99ea6bb7c63841f1 : FAILURE in 1h 16m 48s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/acadc0b3387d40adbb4038e7d7e98414 : SUCCESS in 1h 47m 51s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1ebef09fff93459caf95b6000ab83264 : SUCCESS in 3h 34m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/512d3290a6a34b44b563c7d58125d33e : SUCCESS in 2h 00m 55s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/87e187ca4811486c9e8245c50682ac44 : SUCCESS in 2h 19m 39s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a173ac9563c740d6b1c7ed881b7e8425 : SUCCESS in 28m 44s","accounts_in_message":[],"_revision_number":6},{"id":"a3f7e96c8cd7a05cf3a841f899253f163dc24f0b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-29 08:42:02.000000000","message":"Patch Set 6:\n\nrecheck rally failed but functional/FT passed","accounts_in_message":[],"_revision_number":6},{"id":"b441312658701c7e7bb1bffa60721112ed4656fe","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-29 09:06:32.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/716049/6 : FAILURE in 23m 38s","accounts_in_message":[],"_revision_number":6},{"id":"7a6162a05a7c8e776746985b07bcc6175d3acd5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 11:33:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7052605b86c444eb8c6cf32f67b3ed6b : SUCCESS in 56m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2f4b0317377f4002bada22bc47f2372a : SUCCESS in 56m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/87744e301522459b8a96b320db46da77 : FAILURE in 1h 44m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/304066de3bc04ae1a7bb077bdb39652b : SUCCESS in 1h 28m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a62b816da7304302940b29ea73a13968 : SUCCESS in 1h 42m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0208ae9965f742f89c35927dea6b1807 : SUCCESS in 1h 15m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55cf2d88b81a4cb086e738f129e23a15 : SUCCESS in 31m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0a1a84a08fb4342b253ad113d49d85d : SUCCESS in 25m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16c9c958229445c68a171ce5ba1de973 : SUCCESS in 14m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ffff1ed5c7340239b69827bcf2da2ab : SUCCESS in 27m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff9566aa7b414823bd30b6a9d58f3f07 : SUCCESS in 27m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f5252f5905849cda6ffc709124aff96 : SUCCESS in 9m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a08049e8df0f4512ace1a98dec5e8618 : SUCCESS in 3m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc7ea0a9542840fab3749cc162632f46 : SUCCESS in 21m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/07be3ddb1bc74ecfb007ca80bf525576 : SUCCESS in 55m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5619019b953e4dcb92451d3737bf2b88 : SUCCESS in 1h 11m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d38eea550ce444cb515d0d95c04e4fd : SUCCESS in 1h 09m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ba730ad312c946cf98d1fe2c7b6057a7 : SUCCESS in 1h 46m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/72846481cc1b44a48814af5b65fa4d69 : SUCCESS in 1h 21m 53s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/55a5fa6141ba45d8ae2c933cbbd9dad4 : SUCCESS in 1h 05m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b2634d829a7e417b9ea30d7d1d611d6a : SUCCESS in 1h 22m 30s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3aacc8eb182f44b2a8a83a413bc67850 : SUCCESS in 1h 17m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/739c4b35d52045b69b60d11e1b974e16 : SUCCESS in 1h 09m 06s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0450014b5d6447848532861979871023 : SUCCESS in 1h 37m 59s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1fe068c55c5344a9a8c927b285a1d238 : SUCCESS in 1h 43m 17s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2e0f35c1c67042eeb402870306d9ad42 : SUCCESS in 1h 19m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/974dc30ce6f2431c86ab97c3434f46f1 : SUCCESS in 56m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/542ac548707e43a19f594ceacc847059 : SUCCESS in 2h 16m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/954382bf73904c0aab79eeb4595bbffc : SUCCESS in 1h 02m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0c5a44e79b63486b9a896ae146ae3c02 : SUCCESS in 1h 45m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ee7a3552ad1546328cc042d3dd2d69f6 : SUCCESS in 1h 56m 44s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7421a2e4159049f9922a6457cebd20a1 : SUCCESS in 1h 25m 29s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4cb2e9c3a4944fbfaa1e7f99ea5ff5df : SUCCESS in 53m 59s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e274717e008b4b029b88405c644bfebb : SUCCESS in 1h 44m 17s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bdd87447e60b4e94a012ebc756e2d127 : SUCCESS in 1h 42m 43s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a4f4320c72144e0eb1ed190efe4db200 : SUCCESS in 2h 23m 39s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/dc1cf5295fce4ef2ac02b8aa682794c6 : SUCCESS in 2h 04m 50s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6957782cdb724ce89c25c98af116741d : SUCCESS in 2h 30m 25s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab589c32ef9642f7b7d3346771bcd788 : SUCCESS in 28m 56s","accounts_in_message":[],"_revision_number":6},{"id":"1b9064bacddc772d91fb51763f9ceb1bdd6e79da","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-30 07:17:23.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"d5fe2a9d5066434c1cbfd60b60e2f2eeaa1ae865","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-30 08:13:24.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3b9dd2a880689d18a3736e4195e54042b06bfa15","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-01 16:05:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"567ef37251e3bd1fc60279ad457c8de0ffede150","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-01 16:06:01.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"67a1c69e7b5aee83640472d96912257ba71e8dd3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-01 18:15:34.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/716049/7 : FAILURE in 1h 48m 52s","accounts_in_message":[],"_revision_number":7},{"id":"3ef1616be481fc5bea2f84b4bb8abd4073a12347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 19:44:44.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4c546ff7f623457f8a627a1763b3d6a5 : POST_FAILURE in 49m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/51773fc8dcdd43de956ab646dd43cb53 : POST_FAILURE in 35m 01s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/a2f53e0a32e941969aa1d81d9eec787e : POST_FAILURE in 1h 41m 02s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3b46540a98e64792a1aa1c80c8b27ace : POST_FAILURE in 1h 08m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/eac6c197b3f3414c9b675fbd29b43241 : POST_FAILURE in 1h 19m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/691cc8d80be84feca68f7b43d73ca99c : POST_FAILURE in 1h 05m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fefa78876c244d29c81c37490532de0 : SUCCESS in 48m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29816460c83f41128167ccd90c0513a0 : SUCCESS in 42m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/521f66bce19143f69d517f1f4f826f51 : SUCCESS in 13m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ae1b301f8b942cc9cbac0d79809703d : SUCCESS in 42m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/895fbab574304702aad9fef1114638a2 : SUCCESS in 24m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d39a8938ebbf4e9c80240db56ae560ad : POST_FAILURE in 9m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0945d7e245664b03aa1e3eeabcd3d1ee : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4633d188458046f291b6ed3ef455a602 : SUCCESS in 24m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b533c775a6c14f9c9caf7727d66cee3d : POST_FAILURE in 1h 02m 13s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/dab4b525876f4fa5b32a52c119bbb89b : POST_FAILURE in 1h 06m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/82d87f5b1ee2402fb8a38732df6a924d : POST_FAILURE in 1h 07m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fdbf525405d245d0a0d07bcd8bcf5b2d : POST_FAILURE in 1h 34m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/94190f13012c4ba1b444170dd38a9f59 : SUCCESS in 1h 13m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9b1f4038ee164a21b83afc9a7dac490f : SUCCESS in 1h 11m 47s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7ea7beb41d5349f3b429db0b38480580 : POST_FAILURE in 1h 13m 15s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/720b3e8c6f4b4b4991d05d4c6c4d2985 : POST_FAILURE in 1h 35m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bdcea124ecf44da98fd1fc274951fb85 : POST_FAILURE in 1h 08m 02s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d2b919f869d2421f9969dc9e0d753d81 : POST_FAILURE in 1h 32m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fa07468a544b436992e58d9215229d69 : POST_FAILURE in 1h 34m 56s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8ec2b0e377d64042ad880abcd98def1c : POST_FAILURE in 1h 09m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7fda6e73499e46808d4804ec7a9ac84b : POST_FAILURE in 55m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fc8bc588e5b94de8aed93f80e38c73ae : POST_FAILURE in 2h 08m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe8379820ecb48fe83cb13ce1d946f20 : POST_FAILURE in 54m 03s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6476dbda56f94dcd9cf47b8c4d1cba29 : POST_FAILURE in 2h 01m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7e713c351061457393c04abecaca9d02 : POST_FAILURE in 1h 53m 04s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/06921a9bcae049d9b7ec5335f7e8a655 : POST_FAILURE in 1h 13m 35s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e3d1253f9b314698be1b12a575673a98 : POST_FAILURE in 53m 41s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bccd0b29d37548048c4104a5aaec5c43 : SUCCESS in 1h 43m 58s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c67ae1d6c311446585c706d6e783140c : TIMED_OUT in 2h 12m 45s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ac6f2e5ce972424587660021666c807f : SUCCESS in 2h 43m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1e531536742140c481480db0174eade2 : POST_FAILURE in 2h 00m 43s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d79dc698777a4bc0b08fa2fb6e397b7f : POST_FAILURE in 1h 51m 51s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ac553292e6a45de870e02f30bf877c7 : SUCCESS in 27m 47s","accounts_in_message":[],"_revision_number":7},{"id":"69b4f293ab594cf3c6e5144f67ede38e9f788deb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-05-01 21:23:27.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/May-2020/Arista_Tempest_Testrefs/changes/49/716049/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"5e72600a8d3d8d1adb69df1306b1c9e38747e7ad","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-01 21:59:58.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"9716ee671883cc3c76d45c4167b00606ea67fc8a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-01 23:57:07.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/716049/7 : FAILURE in 1h 56m 31s","accounts_in_message":[],"_revision_number":7},{"id":"eb15fa0afb9268e2a9351d15d7d54a32205ea7a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 00:32:14.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a8a77a8187604c3389911c2e5010e5d3 : SUCCESS in 49m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7316a4e61fae4a12b810b8ff047bd663 : SUCCESS in 34m 18s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/c8e0285fb72646088f29b4415d5de68d : FAILURE in 1h 48m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0fe125434cbf496bb37c74acc6d057ef : SUCCESS in 1h 10m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1f6d531e929842a383157815a9242c71 : SUCCESS in 1h 12m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a9438e69b1c14127972e36ef22643ad2 : SUCCESS in 1h 09m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62f5526805514a499c181fecb713aa10 : SUCCESS in 36m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c8ca28411d64040b93b16562a79b756 : SUCCESS in 26m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22350217eda14ebcbced00e50b6d8cde : SUCCESS in 14m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf43f545ecbf41cb95c48cf13bbd18ba : SUCCESS in 37m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0c1a17f186e4b36933848d8e1fb3a1d : SUCCESS in 37m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61c2d310cc4e45aaa197d5082540189f : POST_FAILURE in 8m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1a9de28e044b4597abead746d8515d7e : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1bb0c0217d8422199a29af86e07ebb2 : SUCCESS in 26m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2d4f3831adf443a18942dcd16933fa4e : SUCCESS in 49m 22s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3596507873a5450792609270fb1b5f2c : SUCCESS in 1h 24m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6e84c59b0ea34be290f813181dccd488 : SUCCESS in 1h 27m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fa3ed33a3f33457c8490dd608d84f7f2 : SUCCESS in 2h 19m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/77cd6a303b6d4386a77c8d2c9f5e1aee : SUCCESS in 1h 07m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5dd8ba404af545718ed834608d53c96d : SUCCESS in 1h 06m 21s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7e98a9d8c38a4f39bd1e19537538af49 : SUCCESS in 1h 15m 36s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/df5bd62906c144d6951baa66905806a8 : SUCCESS in 1h 33m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a042652d9d5a466aaa429aca84006cb5 : SUCCESS in 1h 30m 03s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/98dfe9f8ee9741f29b652d9640afe2d4 : SUCCESS in 1h 35m 49s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/952279c6a6304b6b856f61860c54c5f6 : SUCCESS in 1h 35m 46s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/70ce996c6e0d42b6b25cefc585641c18 : SUCCESS in 1h 10m 32s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/074c0d571be140e2b7c79df505e60875 : SUCCESS in 53m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1617ccba566e4c17954d6112401b00a7 : SUCCESS in 2h 15m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b783c6cdea0d4207a21894274f9b59d9 : SUCCESS in 1h 02m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4f7c2643d2ea46489f6a9f4f0f11826b : SUCCESS in 1h 52m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/194df1133d6345b3a38d9703a5ad3cc2 : SUCCESS in 2h 05m 51s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6098d1ac68ca4467b948ccbd01acb139 : SUCCESS in 1h 17m 04s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/80a838fbbb844884966a3a82a9f3b266 : SUCCESS in 50m 00s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9af455c3396742be88ce33945b1174b9 : SUCCESS in 1h 40m 21s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5f8962079a72442887a09daf850e72b3 : SUCCESS in 1h 47m 46s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/129d07b1806e4444b33fcc6ebe1249e2 : SUCCESS in 2h 27m 10s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b63499f2e405475fb979dfccf0fdd86b : SUCCESS in 1h 43m 25s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/4a966fc0f5824ced98f53f40883a5a86 : SUCCESS in 2h 22m 23s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dda4d3dd42ef473796c2bbf4b537e262 : SUCCESS in 27m 07s","accounts_in_message":[],"_revision_number":7},{"id":"f6d8f04423a3cef4d329addf2cc4233c588cb603","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-02 20:48:30.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"7a3437f902fa649394fa0dd59f3547df6eb65aea","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-02 22:58:00.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/716049/7 : SUCCESS in 2h 09m 01s","accounts_in_message":[],"_revision_number":7},{"id":"bf5b3dd17fa1e78e2701d6c2813f4c9e273511c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 23:52:58.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c64643c0fc3b4abfaa3d9a46e6d6cc93 : SUCCESS in 49m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e6e4b83471f9439284cf1a94f14d8d25 : SUCCESS in 26m 42s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/3a2ff3ec871a4ec6a735e80d6207c45f : FAILURE in 1h 35m 40s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5d2f38cdc7b04778aaa72a20b2efde6e : SUCCESS in 1h 06m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d01cda7e5f374d72ae7172ef0f8ebb09 : SUCCESS in 1h 18m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eafc981f13934ff1ac7940661dd8dee3 : SUCCESS in 54m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/baf52922ccf746bc826739e79d89d6b7 : SUCCESS in 29m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb0da228e90647f5a6f10fde17743855 : SUCCESS in 31m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0caa562bcd84484e8e069b06e88b3db9 : SUCCESS in 13m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f46acff8682f4120b6f01182bb6ecfe9 : SUCCESS in 39m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f347e245f680448d9d13fb0b879b877a : SUCCESS in 28m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/293601c5ceac437b944deab40b0b00c4 : SUCCESS in 10m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/165b0ba0fa9d4d80910b6dea03548ac2 : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/910b888c3ea54068a71b2887fc8a5200 : SUCCESS in 27m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/36fdca9aa3d241b7b5b292d41d0a12b9 : FAILURE in 49m 22s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/80383eed400743febc2956f145fdd92f : SUCCESS in 1h 22m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a20e44ddcd214869aa89f59500e90915 : SUCCESS in 1h 23m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7fb1451699bd4237afb96001c47fd1b0 : SUCCESS in 2h 04m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3046f9b055bf4de891cdcc122ecb15af : SUCCESS in 1h 02m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/14616337dd924e1f92b0ddf4b436dcf2 : SUCCESS in 1h 01m 01s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/da628d59943f4013b5a4930a0bdc8328 : SUCCESS in 1h 16m 55s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4616af5de66a4889aa0ea1f09b0458e6 : SUCCESS in 1h 29m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/271930297e8a4a5899b35e097a568193 : SUCCESS in 1h 10m 14s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7a40f92d807d49a88025cc6779875b67 : SUCCESS in 1h 34m 49s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cc987615511e43159216b048a6a485c2 : SUCCESS in 1h 35m 24s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0e4eef7598654d858c19430e2fa01e84 : SUCCESS in 1h 13m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6ccdce9c622c4203b2a32b8f5e19ff9a : SUCCESS in 49m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dae331530a444bc4a36c18cd9d16ae06 : SUCCESS in 2h 22m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3e9ddc8566946fe874644de5c94ab9b : SUCCESS in 48m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/802f0a0fc6fe44cdaa3c8bb3bc507f59 : SUCCESS in 1h 53m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5299c61d20a949a1b8deab7434455de0 : SUCCESS in 1h 54m 57s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d8d5e7ea8e714d02afb01b92b84ea57c : SUCCESS in 1h 16m 47s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cb2dda310feb4616a61f1a3c4c2ce59a : SUCCESS in 46m 34s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/85d76c0f52874bcdaf262e3d0962c0fd : SUCCESS in 1h 30m 50s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e9b491f6a90349fbaf16d277c232f683 : SUCCESS in 1h 46m 06s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/484fa2c40ee34ba8a9324e60fd8e9d07 : SUCCESS in 2h 18m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fa35aa6cdf784a179bafb47047725b7c : SUCCESS in 1h 51m 57s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/52035057c5e940298149ea9db5c986f5 : SUCCESS in 1h 51m 41s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/931ed1a14805488e863d5a5223fd79da : SUCCESS in 24m 05s","accounts_in_message":[],"_revision_number":7},{"id":"b92a041bc6a2baf64542778ca6b782f349527bd6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-03 10:20:34.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"fb8fe5dc3cfedf5b84ec472febdecdab75beff06","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-03 11:58:55.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/716049/7 : SUCCESS in 1h 37m 52s","accounts_in_message":[],"_revision_number":7},{"id":"baf44daffd2a9f7e58a2adf4b474c6f2a0f6b3da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 12:46:21.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2223711cb2994df2bb482705f3a89a33 : SUCCESS in 46m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c86cf710b8174398bd0dce25403d67b6 : SUCCESS in 34m 03s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/295d992380224aa3b8f7c1259a7301eb : FAILURE in 1h 37m 40s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cdb3b8d29f7545899b5356f646ad2710 : SUCCESS in 1h 25m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5d9b73fd25524263ab148ad401a3ac98 : SUCCESS in 1h 26m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/25d1f1be75734a5fbbf26699fcac9b47 : SUCCESS in 1h 25m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/246bd36e5af14142ae9759d9270e3359 : SUCCESS in 39m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef5234c6ab5d4ec5b31776c80675af4e : SUCCESS in 26m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd770a9f95fc4d17a2fe9046b3e4ef07 : SUCCESS in 11m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5510cd7e7583418e8b5dbab1414cfb24 : SUCCESS in 25m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5eb8dfac73d4432bf201791b005b8d6 : SUCCESS in 32m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d18c8bd25744c19c19be505f791033 : SUCCESS in 7m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c05eb938cd62447cbfe36a27a01d8b8d : SUCCESS in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13abafeb2521436587282e7daf381eba : SUCCESS in 20m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3b48068086ad449db7fa8488b4161a38 : SUCCESS in 54m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b049cc701cf541e9aab6d9929a5107ae : SUCCESS in 1h 47m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/49b570b596b44e8fb05f15df0c4d1c4f : SUCCESS in 1h 10m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/86aaa0e27f1844078e2af45f4beaf1d7 : SUCCESS in 1h 40m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/94e53a2d6e904f5a8179dc68fd326c40 : SUCCESS in 56m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cf8acfd1576d4fb5a4b83138707b0f15 : SUCCESS in 1h 02m 27s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5b86f159cc3a4be3abe487e49cd373aa : SUCCESS in 1h 08m 10s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b8346e565ef141119835d19d81c4f584 : SUCCESS in 1h 08m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/17842d659a9f402b94a30e17dc9e1143 : SUCCESS in 1h 08m 22s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/178cc3357fd64a129a1c9f9157e2a75f : SUCCESS in 1h 30m 08s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dfd4387635f94ec4b87250a011ae03a6 : SUCCESS in 1h 30m 45s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6ba6857f14254e4c8f599c1974f6f6a4 : SUCCESS in 1h 05m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/09a61e8dc8ce42948127fe5e1849e73f : SUCCESS in 51m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1413a2e7ccca48a981f44aee52b629dd : SUCCESS in 2h 12m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23d31958287d48ab99f537bd9f2ea3fc : SUCCESS in 1h 00m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0e32deb35084407eacb5e0b89050134d : SUCCESS in 1h 58m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/da3a49175a2143c9ab93cb5c0f69dd6f : SUCCESS in 1h 43m 59s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cc76174d1b4646ffbdb71e7e8af9ee72 : SUCCESS in 1h 13m 22s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3eb41f9afa3472ca531f2f0c8f61a5f : SUCCESS in 45m 19s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/86e1d7d083f543f6b28f0b7e5f54ae3d : SUCCESS in 1h 29m 35s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/34b09dd757154bdc915c5ff133c319c8 : SUCCESS in 1h 40m 59s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e00c01897b1842e38b1e203f1a43cbe0 : SUCCESS in 2h 22m 31s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ccdd855217b14cf48a6f282cd3615a94 : SUCCESS in 1h 51m 09s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/181ebd58cf704354b0bbc312448c1a01 : SUCCESS in 1h 58m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"88817cf4ceb46bbfd0728cebc42559f2c8e82b85","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-04 06:18:35.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"52f884e28d2d2cfe8cd0fb3ec8f6efb2a29c3f77","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-04 10:19:50.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"8d2ee10830605111c75b80784372d1b2985ec45b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-05 12:39:59.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"657cd32bccecbaece8f0218f5fb001eeefde7376","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-09 12:33:48.000000000","message":"Topic changed from unittest.mock.functional to unittest.mock","accounts_in_message":[],"_revision_number":7},{"id":"ffc7fca2cd364081689ba01227a84bcf5d8fe535","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-09 12:33:52.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"481ec088547935d0578d8b9d46dcc2c1fd752870","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2020-05-09 22:44:36.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"f0e7eeefbdec71c9279b1fdbb22d04fb03d636e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 22:45:10.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f7be5c460b57a284a94db2f5df10c38ae787c4e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 01:52:24.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0a5a566b493c404bb784083d7ab0a02f : TIMED_OUT in 2h 04m 34s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a6deb351bdfa4f8daaeb1be4ab838d59 : SUCCESS in 2h 28m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ab908ce553894b7fb07126b36a50353e : FAILURE in 2h 29m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/02a08d3528e34359b9339f40d36aec89 : FAILURE in 2h 22m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80b0f8c5d3574791893d4b3aa4218b9b : SUCCESS in 54m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/402c85f6950a42978a1037cbe2375fe6 : SUCCESS in 21m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dd7d23f4b414ed8af97ae41c2996a42 : SUCCESS in 52m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d3bcc59396c495c8bc0e0f2553c5abc : TIMED_OUT in 41m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daabea2b3e294642ae6859c593f2557f : SUCCESS in 17m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/18d5abdcec83445aa47977091ed33078 : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ccedf36caec4de1a276dc8b1e25e512 : SUCCESS in 42m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/741985a888ba4ede91b4311a67e559de : FAILURE in 1h 43m 42s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9b216ba83df946de9faeaa26b11ad90a : FAILURE in 2h 09m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0666da9d8611454393a94ef346b451b3 : SUCCESS in 1h 07m 52s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c597a48d6dac468689c6641c6f404ffe : SUCCESS in 2h 38m 37s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/487e3b9cb75f471da5a4b29ad8e7e76b : SUCCESS in 2h 39m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/993bf1afd3c84448bbd7ce2052ad8efd : SUCCESS in 2h 31m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/aad2f95e05e14c08baa741e900767aff : SUCCESS in 1h 06m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7695b4a138d648278eafdd9f346c3b6f : TIMED_OUT in 3h 03m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7c5bfa130d84fd3bc2d7f95ca11e72e : FAILURE in 2h 25m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fe4f0f427fde46b6a34a279ab8bbd31f : TIMED_OUT in 3h 05m 17s","accounts_in_message":[],"_revision_number":7},{"id":"60d0be5c5da537a0563a43d1c3b12d701acc0ef0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-10 07:58:10.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"c19053e793674aafe0489cf1b15908f3cc4c6ad4","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-05-10 08:04:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-05-10/716049/7/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/6f4430d : FAILURE in 6m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a0e71ca5ff0d23b8e90fccba8bfb99ad6aea6b91","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-10 09:19:16.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/716049/7 : FAILURE in 1h 20m 36s","accounts_in_message":[],"_revision_number":7},{"id":"6b2f54cf37817a9ffec366f0a4fde024251eb39d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 10:30:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e6acd8d7993e48d2b06ac2015d2f6d0c : SUCCESS in 51m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aeb5a51679c34c28a3e8d86cb529badf : SUCCESS in 35m 03s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/260d15a04ca14b79b833244048ae6150 : FAILURE in 1h 52m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c83426c89eb44613a6575388edc2eec7 : SUCCESS in 1h 30m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ffba7a9311994efbbe22c3b00a7308ac : SUCCESS in 1h 14m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a16540d61ca048609133f85e5f5033a9 : SUCCESS in 1h 06m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b942122992ea4049afba9eafdda972d2 : SUCCESS in 34m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53f146037ffc4db78e5ac77951576485 : SUCCESS in 26m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09ab07de8a6040dcad6ac55125966110 : SUCCESS in 13m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83d35d6a8e094869bd2435cc5d8084bc : SUCCESS in 26m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fa2121797164ab8ab99160ed0b30dd1 : SUCCESS in 22m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7c409f7710f4433a42910f2d1e12bd2 : SUCCESS in 8m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6c9bfad653494a4c8ae9d68407c6611a : SUCCESS in 2m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95bca9632adb422da57f7974685e8095 : SUCCESS in 24m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d21dc9e44abe473ab405ccac9c3881f8 : SUCCESS in 52m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0bbf37bee9344127bb63f77bc3fd82a7 : SUCCESS in 1h 12m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/47016a809c204964a05e41966c3964a2 : SUCCESS in 1h 13m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/89cbd412f4ef40c188ca74d68cdba10f : SUCCESS in 1h 48m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f1444d8e1a464392a210c17cdb0bb2b9 : SUCCESS in 1h 03m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7dd1314470a54d8d85990a19c2c5be82 : SUCCESS in 1h 08m 22s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7c71da00cf5149698c622498aa5f581e : SUCCESS in 1h 06m 51s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3fee3b7621534619a06d6b9e39f7dac3 : SUCCESS in 1h 13m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/181aae4c53d743b9886b4d29450ff54a : SUCCESS in 1h 04m 32s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1c7ba20fdb714243bba5c6372e237e8e : SUCCESS in 1h 31m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b6ce9bd6fb144f0bab7ae7845dba4056 : SUCCESS in 1h 34m 18s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/af09dca7395443c28f0a442e9b0a990d : SUCCESS in 1h 08m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a9fa2fdf220c42b7a5771f3489729522 : SUCCESS in 54m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2af3ce37267b40adbb58228d4f5b54bf : SUCCESS in 2h 11m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cbbc396080bf4898a328e37677970c20 : SUCCESS in 59m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0724eadb8df34ae28650b346688ddf2e : SUCCESS in 1h 44m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b5a389864b3f49399f2efe0258a779a0 : SUCCESS in 1h 45m 43s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a082c9c3853b40f1b5b7332d82c88d36 : SUCCESS in 1h 26m 55s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3dcc1517785e423b8dbc71663ab50ffc : SUCCESS in 44m 51s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6370173784384ff3aaa0caba15a92b1f : SUCCESS in 1h 33m 39s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1ca51119d603410c88b54eaaf623bb80 : SUCCESS in 1h 57m 44s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ee73962480854ebd9366cd87644e75de : SUCCESS in 2h 26m 55s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/40a806593413402a880ad073e4c26e05 : SUCCESS in 1h 52m 51s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/7e55cb9c2f544cc3bde907cfd487b95c : SUCCESS in 2h 13m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b92ee4fa8f2523710a7143ef553c09337b0cc608","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 10:30:23.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a79215918f1599bfa9703cdd1c3938360bacda67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 12:57:38.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1f64e665d31a45dea4aa8f4e99e7ef92 : SUCCESS in 52m 08s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b07879015ece4dc5807f1bb82485430d : SUCCESS in 1h 28m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/583f0b84bd4a41b1ab1a7b3692e5fd9b : SUCCESS in 1h 02m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/61322555af0646d79dd0a6d0d0352fe6 : SUCCESS in 1h 07m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5076281b0dc8422c9cbb1eef39d90dad : SUCCESS in 24m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9e82368968d4c9a83ba5cc96507fd7b : SUCCESS in 11m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01757b1f1f62417db90ed52e4367dd42 : SUCCESS in 22m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9e1f3e4e23e42958d923a4b8beaeb73 : SUCCESS in 21m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1066e0de1ef4315ab59266c2310a5b9 : SUCCESS in 8m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/53d9d9e08c3343cfb9e672dea52ff44d : SUCCESS in 2m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7dd7a8e0d1f444c98f49124659e92cce : SUCCESS in 26m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3802ee983e5b431fba2487ea58551707 : SUCCESS in 53m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/19248bddd6f44b8fae04470a5a04fb54 : SUCCESS in 1h 13m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1ed6b28eb4b34fe4913797fb50d3af6e : SUCCESS in 1h 16m 23s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a0dbc09346ac43939cefa378e3e085bc : SUCCESS in 1h 11m 11s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a74af136277e46879ca748bb816b7f7d : SUCCESS in 1h 09m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fd5bdde5b0394c9cb7182839527ca58c : SUCCESS in 1h 03m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3496116fe18442b897ab78d5dd2c7290 : SUCCESS in 1h 10m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a38c613ad0694353bab88fc43320431e : SUCCESS in 2h 21m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d8e3fbe31d842eb909e4741a7299586 : SUCCESS in 54m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/da5c5efb4f4843c7aff424335573e047 : SUCCESS in 2h 05m 04s","accounts_in_message":[],"_revision_number":7},{"id":"9caa75202e7929ef13c4b8d9569c4279bf6b2d04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 12:57:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"ccbd2fbe83d2f550d9e9b71b95eec363521440f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 12:59:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac811504cb34621b201428019608352 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e04526650dd04af292d7c06e5f3ec280 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"8126f888945b0f3ac01dca1da0dc46d1f9ec3679","revisions":{"abbe2cb596d281567482ed9234efcca318c4394d":{"kind":"REWORK","_number":1,"created":"2020-03-30 18:58:10.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/49/716049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/716049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/716049/1"}}},"commit":{"parents":[{"commit":"b3dddd8dc323fcec734219ef19bb8611488d7114","subject":"Remove the dependency on the \"mock\" package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3dddd8dc323fcec734219ef19bb8611488d7114"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-30 17:49:50.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-30 18:57:42.000000000","tz":-240},"subject":"Complete removal of dependency on the \"mock\" package","message":"Complete removal of dependency on the \"mock\" package\n\nNow that we are python3 only, we should move to using the built\nin version of mock that supports all of our testing needs and\nremove the dependency on the \"mock\" package.\n\nThis completes removal of all references to \"import mock\",\nchanging to \"from unittest import mock\" in fullstack and\nfunctional tests.\n\nAdded a hacking check to enforce it in future patches.\n\nChange-Id: Ifcaf1c21bea0ec3c35278e49cecc90a101a82113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/abbe2cb596d281567482ed9234efcca318c4394d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/abbe2cb596d281567482ed9234efcca318c4394d"}]},"branch":"refs/heads/master"},"550f7baa22645e4212a76d4ebc5f97a15e9040d2":{"kind":"REWORK","_number":2,"created":"2020-03-30 19:06:40.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/49/716049/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/716049/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/716049/2"}}},"commit":{"parents":[{"commit":"b3dddd8dc323fcec734219ef19bb8611488d7114","subject":"Remove the dependency on the \"mock\" package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3dddd8dc323fcec734219ef19bb8611488d7114"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-30 17:49:50.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-30 19:06:24.000000000","tz":-240},"subject":"Complete removal of dependency on the \"mock\" package","message":"Complete removal of dependency on the \"mock\" package\n\nNow that we are python3 only, we should move to using the built\nin version of mock that supports all of our testing needs and\nremove the dependency on the \"mock\" package.\n\nThis completes removal of all references to \"import mock\",\nchanging to \"from unittest import mock\" in fullstack and\nfunctional tests.\n\nAdded a hacking check to enforce it in future patches.\n\nChange-Id: Ifcaf1c21bea0ec3c35278e49cecc90a101a82113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/550f7baa22645e4212a76d4ebc5f97a15e9040d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/550f7baa22645e4212a76d4ebc5f97a15e9040d2"}]},"branch":"refs/heads/master"},"7ec4a4ac432d46f101c48b291562a26738067449":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-20 20:41:15.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/49/716049/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/716049/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/716049/3"}}},"commit":{"parents":[{"commit":"9f99de1b5d507e6533398d60983066e9f11159d8","subject":"Remove the dependency on the \"mock\" package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f99de1b5d507e6533398d60983066e9f11159d8"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-30 17:49:50.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-04-20 20:41:15.000000000","tz":0},"subject":"Complete removal of dependency on the \"mock\" package","message":"Complete removal of dependency on the \"mock\" package\n\nNow that we are python3 only, we should move to using the built\nin version of mock that supports all of our testing needs and\nremove the dependency on the \"mock\" package.\n\nThis completes removal of all references to \"import mock\",\nchanging to \"from unittest import mock\" in fullstack and\nfunctional tests.\n\nAdded a hacking check to enforce it in future patches.\n\nChange-Id: Ifcaf1c21bea0ec3c35278e49cecc90a101a82113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ec4a4ac432d46f101c48b291562a26738067449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ec4a4ac432d46f101c48b291562a26738067449"}]},"branch":"refs/heads/master"},"6f36a2debb71465ab322ee4027e290080a65c072":{"kind":"REWORK","_number":4,"created":"2020-04-24 21:38:30.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/49/716049/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/716049/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/716049/4"}}},"commit":{"parents":[{"commit":"d59a75e75f43f90678e906162245c41353cb22ad","subject":"Remove the dependency on the \"mock\" package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d59a75e75f43f90678e906162245c41353cb22ad"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-30 17:49:50.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-24 21:37:52.000000000","tz":-240},"subject":"Complete removal of dependency on the \"mock\" package","message":"Complete removal of dependency on the \"mock\" package\n\nNow that we are python3 only, we should move to using the built\nin version of mock that supports all of our testing needs and\nremove the dependency on the \"mock\" package.\n\nThis completes removal of all references to \"import mock\",\nchanging to \"from unittest import mock\" in fullstack and\nfunctional tests.\n\nAdded a hacking check to enforce it in future patches.\n\nChange-Id: Ifcaf1c21bea0ec3c35278e49cecc90a101a82113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6f36a2debb71465ab322ee4027e290080a65c072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6f36a2debb71465ab322ee4027e290080a65c072"}]},"branch":"refs/heads/master"},"9af4ce2ee2b3ee37b04ace7ca6d2eade815d3b55":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-04-28 21:54:37.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/49/716049/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/716049/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/716049/5"}}},"commit":{"parents":[{"commit":"5c9a23b520e0bffbfaedeef5012e0ae15f193b67","subject":"Remove the dependency on the \"mock\" package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c9a23b520e0bffbfaedeef5012e0ae15f193b67"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-30 17:49:50.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-04-28 21:54:37.000000000","tz":0},"subject":"Complete removal of dependency on the \"mock\" package","message":"Complete removal of dependency on the \"mock\" package\n\nNow that we are python3 only, we should move to using the built\nin version of mock that supports all of our testing needs and\nremove the dependency on the \"mock\" package.\n\nThis completes removal of all references to \"import mock\",\nchanging to \"from unittest import mock\" in fullstack and\nfunctional tests.\n\nAdded a hacking check to enforce it in future patches.\n\nChange-Id: Ifcaf1c21bea0ec3c35278e49cecc90a101a82113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9af4ce2ee2b3ee37b04ace7ca6d2eade815d3b55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9af4ce2ee2b3ee37b04ace7ca6d2eade815d3b55"}]},"branch":"refs/heads/master"},"4647b677ea548cc8d6fdcb95258454f2b0fbb05c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-04-28 22:06:22.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/49/716049/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/716049/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/716049/6"}}},"commit":{"parents":[{"commit":"7594bb0627a4fe0a2a63c432cad838454d1fd8e2","subject":"Remove the dependency on the \"mock\" package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7594bb0627a4fe0a2a63c432cad838454d1fd8e2"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-30 17:49:50.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-04-28 22:06:22.000000000","tz":0},"subject":"Complete removal of dependency on the \"mock\" package","message":"Complete removal of dependency on the \"mock\" package\n\nNow that we are python3 only, we should move to using the built\nin version of mock that supports all of our testing needs and\nremove the dependency on the \"mock\" package.\n\nThis completes removal of all references to \"import mock\",\nchanging to \"from unittest import mock\" in fullstack and\nfunctional tests.\n\nAdded a hacking check to enforce it in future patches.\n\nChange-Id: Ifcaf1c21bea0ec3c35278e49cecc90a101a82113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4647b677ea548cc8d6fdcb95258454f2b0fbb05c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4647b677ea548cc8d6fdcb95258454f2b0fbb05c"}]},"branch":"refs/heads/master"},"8126f888945b0f3ac01dca1da0dc46d1f9ec3679":{"kind":"REWORK","_number":7,"created":"2020-05-01 16:06:01.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/49/716049/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/716049/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/716049/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/716049/7"}}},"commit":{"parents":[{"commit":"7594bb0627a4fe0a2a63c432cad838454d1fd8e2","subject":"Remove the dependency on the \"mock\" package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7594bb0627a4fe0a2a63c432cad838454d1fd8e2"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-30 17:49:50.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-01 16:05:34.000000000","tz":-240},"subject":"Complete removal of dependency on the \"mock\" package","message":"Complete removal of dependency on the \"mock\" package\n\nNow that we are python3 only, we should move to using the built\nin version of mock that supports all of our testing needs and\nremove the dependency on the \"mock\" package.\n\nThis completes removal of all references to \"import mock\",\nchanging to \"from unittest import mock\" in fullstack and\nfunctional tests.\n\nAdded a hacking check to enforce it in future patches.\n\nChange-Id: Ifcaf1c21bea0ec3c35278e49cecc90a101a82113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8126f888945b0f3ac01dca1da0dc46d1f9ec3679"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8126f888945b0f3ac01dca1da0dc46d1f9ec3679"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
