)]}'
{"id":"openstack%2Fneutron~718217","triplet_id":"openstack%2Fneutron~stable%2Fstein~Ibd4ef7b0db5b6830383fffcac685b9c709aae350","project":"openstack/neutron","branch":"stable/stein","topic":"bug/1845150-stable/train-stable/stein","hashtags":[],"change_id":"Ibd4ef7b0db5b6830383fffcac685b9c709aae350","subject":"Configure keepalived interfaces according to config file","status":"MERGED","created":"2020-04-07 20:24:32.000000000","updated":"2020-04-10 04:25:45.000000000","submitted":"2020-04-10 04:23:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718217-1586492630225-d5dcdf52","meta_rev_id":"d9ebfaee17c0a5eb3c494f1a514a455d271ff8c1","_number":718217,"virtual_id_number":718217,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-10 04:23:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-08 13:28:16.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-04-08 11:29:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-04-09 23:59:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-04-09 23:59:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-04-07 20:24:32.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 20:24:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-04-08 11:29:25.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-04-08 13:28:16.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-04-09 23:59:19.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-04-10 04:23:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"83d357c61f99f9e722defbfdbe9bff6356240bb6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-07 20:24:32.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"55b7203b355beaee563f321eb567e203e85a3095","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-07 20:25:33.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/718217/1 : FAILURE in 29s","accounts_in_message":[],"_revision_number":1},{"id":"36d6ddb091b7e48fa1482e8307382a19b71224d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 02:11:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/f6bf1a2409ae4c78a30f2283590dc102 : SUCCESS in 1h 09m 08s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/19eb43572bf34c1792922e1ac6ea07bf : SUCCESS in 40m 51s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/611915f22dfe443eb1f1c425c3f20b64 : FAILURE in 1h 50m 22s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/6eba44d1827a4b60841f3858f15d9f18 : SUCCESS in 1h 09m 13s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/98a8f3a38c0246b4903abf1b4c855baa : SUCCESS in 1h 02m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/09d704f52b404f2f93c4ad5e6489f609 : SUCCESS in 58m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/203bd02636634b76b47c321f06689967 : SUCCESS in 25m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb2c29fdef074a47be27fd8ca7c160a4 : SUCCESS in 22m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efba0a86a4c4428db85761c0e1bfdd67 : SUCCESS in 18m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/59631a89680c4d369e340acc3ffbad73 : SUCCESS in 24m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a9f98b81fafc406f9ee48332966d3bab : SUCCESS in 28m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1aaf47685644897a2ce5cc7732fc74a : SUCCESS in 28m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4fdf255017494f8b91613f915d612e40 : SUCCESS in 27m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0f4264a43274bc4a961eccf7d8bcf9c : SUCCESS in 8m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5f5f888453e94c8d85f892b7e1ea70a4 : FAILURE in 29m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9ec2a59d3bd0453597742a21c64a08c8 : SUCCESS in 52m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/72670b668afc476d830767406a3cd584 : SUCCESS in 55m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4ba5385b944493bb18a35a5339ea3e2 : SUCCESS in 1h 35m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e8a8136e93174f6e8c30e1627dc5397c : SUCCESS in 41m 18s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/0c64a3b239e64ef2a8fac825afe117e3 : SUCCESS in 45m 51s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5c701f44c3554c4d8bbde988387ca791 : SUCCESS in 1h 06m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/63bf1d24ef5448fb85eab40beec20a67 : FAILURE in 31m 53s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/face7ebb132c440fa2fc909f92d68d39 : SUCCESS in 1h 26m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2d6056be9f944828a96961203f4ad7a : SUCCESS in 1h 10m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7d413df45f2443cf81a7798f1a619577 : SUCCESS in 1h 06m 28s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d2256ca1f79e4ad1a672e9cde1f5bf10 : SUCCESS in 1h 40m 09s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b7bef3a8adfb4960a49db0b5649dd216 : FAILURE in 38m 12s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a6b6e5c667944b4ea204737ceb8d6d25 : SUCCESS in 1h 22m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1652007a01934b22b74f985f699b8673 : SUCCESS in 49m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2043d0a73517405c8a0f091dc617d0ab : SUCCESS in 2h 31m 26s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ce4ea6fca9344d12993fb44239f63658 : SUCCESS in 2h 09m 27s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/449063ca38684a5a931a972354b68dce : SUCCESS in 1h 19m 01s (non-voting)\n\nWarning:\n  Comments left for invalid file doc/source/ext/support_matrix.py\n  Comments left for invalid file neutron/tests/unit/agent/ovsdb/test_impl_idl.py\n  Comments left for invalid file neutron/tests/unit/test_auth.py\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_daemon.py\n  Comments left for invalid file doc/source/configuration/l3-agent.rst\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py","accounts_in_message":[],"_revision_number":1},{"id":"619907117c50c5bcf6dde62b58bc2016503e09eb","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-08 11:29:25.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck https://review.opendev.org/#/c/718131/ merged","accounts_in_message":[],"_revision_number":1},{"id":"becc887d120021b7d6a9a882c363cd30eab76c15","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-08 13:28:16.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/718217/1 : FAILURE in 10m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6cfdd6c10315e92542b7f0fa42e4c7d10b1fe9a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 20:14:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/6f75c160999243909c88e8caed8009c1 : SUCCESS in 58m 10s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/9fe5dd64b8924020b2517da86228714c : SUCCESS in 44m 15s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/6fbb0d8fe89145a4b276e740123a21b7 : FAILURE in 2h 26m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/e9f868067e7e4f06915cb40dfb8fcaac : SUCCESS in 1h 21m 34s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/464537b5f8fa429eb8627f62f218b932 : SUCCESS in 1h 19m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/8c82878a5c6c445aa74602c260c4cf16 : SUCCESS in 59m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61894627599343afb5f6aa64c45b3202 : SUCCESS in 27m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5225cbca6a7460f955aae61ffb68f29 : SUCCESS in 23m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/346427320fb7466092774db24ba20350 : SUCCESS in 13m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/116af62be900410880cd095ab0165fb2 : SUCCESS in 24m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/35e59690d4c7425a93e370be56021508 : SUCCESS in 24m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbd029bf938a48a59114eea60efc66d9 : SUCCESS in 23m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a28918bc68d4dd08a9012ff9157d740 : SUCCESS in 23m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62256b35bba480ab44f5874da1b9e13 : SUCCESS in 5m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a334ba72df084573ae531cb29b39f100 : SUCCESS in 1h 37m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c8ce96ba47a14b6382a590186b0c1006 : SUCCESS in 54m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/873c8b8523d24e75a237f015a50e95bd : SUCCESS in 56m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef63c4158d804a3aa1b7130f74d74608 : SUCCESS in 1h 20m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c1b37dbc0053486eba0e667d7d0c8395 : SUCCESS in 35m 45s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/3f21c002bb7c466f96d0e34bdbe78a57 : SUCCESS in 39m 03s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e97e0d22e0ca4075aa1edf8244174574 : SUCCESS in 1h 08m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/620da5778e604aaeba289450995ed6e4 : SUCCESS in 1h 30m 01s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b43c15fcf19b42e7a75d6be5ba8dd174 : SUCCESS in 1h 21m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf065840ee4f4ba78b62256cc52e939f : SUCCESS in 1h 01m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/02be8630991141b5bdd9558ff2faf9ca : SUCCESS in 1h 06m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1b5cc74e8124451b9387812cdb3e9357 : SUCCESS in 1h 47m 18s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/45cc7a882eb14c6c83e6d10443bacc6a : SUCCESS in 2h 12m 19s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a2c38e2656e54ae5a17764a7e9c371f3 : SUCCESS in 1h 42m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/20e77eb85f5c4e22b453dff6304c19df : SUCCESS in 51m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/66b8289a7d98421886db12c804782bd4 : SUCCESS in 2h 13m 42s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/6881507ba55c4ddba07a9c4099684a0d : SUCCESS in 1h 58m 32s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1ca9b58142f241bda45429391a1b1147 : SUCCESS in 1h 32m 15s (non-voting)\n\nWarning:\n  Comments left for invalid file doc/source/ext/support_matrix.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_daemon.py\n  Comments left for invalid file doc/source/configuration/l3-agent.rst\n  Comments left for invalid file neutron/tests/unit/agent/ovsdb/test_impl_idl.py\n  Comments left for invalid file neutron/tests/unit/test_auth.py","accounts_in_message":[],"_revision_number":1},{"id":"457434ad754698249b284d7ed6e3cab3e3b151e2","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-04-09 23:59:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f4e1a46fc1957ac6e4580c8ec75acfd2054ff37b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 00:00:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1bd32a414e35d3213f6b98046895c268939b7aca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 04:23:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/b487e7fd6680477da1e689fd2759f9a9 : SUCCESS in 50m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e1ede4d052d43dc93312b5a9e1644ed : SUCCESS in 23m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1d02d0152b14c26a5e891f773b41242 : SUCCESS in 11m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8dcf2de6503d49c9b42951525d484b2a : SUCCESS in 29m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e11cff787bf14e90b1f5a4f61bfd7d1a : SUCCESS in 27m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e75bfb049ae64494b632449427789dd0 : SUCCESS in 23m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf9557e0473e46dda6c0336d1d0cfd53 : SUCCESS in 24m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47ac97c97fd34d5ba1a1f66cdb5d8674 : SUCCESS in 6m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b21ed1d64c774998b501d02c33e62038 : SUCCESS in 1h 22m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ff0da485efd644709b6e8e1902722af0 : SUCCESS in 57m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e6dbc674a6c94cbbac686b986a0842ae : SUCCESS in 1h 04m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d800ad215ee46ba82c2c169ab6d56db : SUCCESS in 1h 23m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/93788a60ebed4a82a44044bcc505030f : SUCCESS in 37m 23s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/b375ec0d01de4d6d93d72ae7ac2eb9cf : SUCCESS in 45m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/250a0addd70f443b97857a69e49f2fe6 : SUCCESS in 1h 02m 17s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/db90702b3be14935a3d994f847f9047b : SUCCESS in 1h 20m 49s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dc70def3ed294c2e94f1038368330134 : SUCCESS in 1h 20m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa10c5e956bb4c1a9086835753892cc1 : SUCCESS in 1h 14m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/56b2b3c336f049d689d69f39acb79e7a : SUCCESS in 1h 10m 49s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/41ef18f0f5df4811bf6777ae01c133ca : SUCCESS in 2h 18m 16s","accounts_in_message":[],"_revision_number":1},{"id":"3ef612f85411c9d96594b1bd4ecbd7bedcd409b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 04:23:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"50e3ce01ce2287016e8cc3fd11e774baf3ec0c19","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 04:25:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c24076b074647f8ae23db07f5bfd7c9 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9dc499da2bb7515ed19204ab4d46e34c7abd1ee3","revisions":{"9dc499da2bb7515ed19204ab4d46e34c7abd1ee3":{"kind":"REWORK","_number":1,"created":"2020-04-07 20:24:32.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/17/718217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/17/718217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/718217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/718217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/718217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/17/718217/1"}}},"commit":{"parents":[{"commit":"2d0adf4a0554e7f8b06a2aaa37d1e772b301de95","subject":"Optimize DVR related port DB query","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d0adf4a0554e7f8b06a2aaa37d1e772b301de95"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-09-23 17:43:44.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-04-07 20:24:32.000000000","tz":0},"subject":"Configure keepalived interfaces according to config file","message":"Configure keepalived interfaces according to config file\n\n\"keepalived\" process stops running if it does not find any of the\ntracked interfaces correctly configured. To prevent this in\n\"KeepalivedManagerTestCase\" test cases, \"eth0\" is set up and an IP\naddress is assigned.\n\nChange-Id: Ibd4ef7b0db5b6830383fffcac685b9c709aae350\nCloses-Bug: #1845150\nRelated-Bug: #1830232\n(cherry picked from commit d8eb42f1eaaa3747691cbb91db49292c66732073)\n(cherry picked from commit eb1d0313d0dddb5e93f4d91ae28390946edab149)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9dc499da2bb7515ed19204ab4d46e34c7abd1ee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9dc499da2bb7515ed19204ab4d46e34c7abd1ee3"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
