)]}'
{"id":"openstack%2Fneutron~718408","triplet_id":"openstack%2Fneutron~stable%2Ftrain~I6ba6309f5b8a33beee8eb2a6b1c39c6353c05814","project":"openstack/neutron","branch":"stable/train","topic":"rally_train","hashtags":[],"change_id":"I6ba6309f5b8a33beee8eb2a6b1c39c6353c05814","subject":"[DNM] Test openstack-rally","status":"ABANDONED","created":"2020-04-08 12:07:16.000000000","updated":"2020-04-09 07:31:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"63fb591e21cd61fb142bc79ebf76f364bb831161","_number":718408,"virtual_id_number":718408,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-08 23:35:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-04-08 15:47:44.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 15:47:44.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-04-08 23:35:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8ef051216dbc84179fb7c6314a8498994dcb06e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-08 12:07:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b276077bddadde7695285902fc47dbef95318793","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-08 13:35:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b7c95deb603954e90c013c832c544fcb1c10a3bb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-08 15:47:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/718408/2 : SUCCESS in 1h 58m 06s","accounts_in_message":[],"_revision_number":2},{"id":"16cbb3e1aa0f981b26ef082bf57ba2100fd147d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 23:35:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/ecaded7aaeb74f41bb4faab87fd4c573 : SUCCESS in 1h 04m 19s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/5f8e8a96489a457cbc4dcb55b2afbc3a : SUCCESS in 42m 28s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/602f202513d14152b9cc4875a3caff33 : FAILURE in 2h 20m 26s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/b42b1018e8bb4e6fbe1ceddc25a09615 : SUCCESS in 1h 13m 21s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/fd3056895bb044d0b527c0cbe546fb74 : SUCCESS in 1h 14m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/0822292243164e7f956b585ee26d6f2a : SUCCESS in 1h 20m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6bb72277ebb4146ada67a4e4cdf0e8a : SUCCESS in 34m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f20a5c15a7e48b5b734722b7f451976 : SUCCESS in 25m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7cfd6db21574ee1b8e7f8f02bc31efb : SUCCESS in 15m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/119f405868914868bb3a03f05d5fa8bc : SUCCESS in 27m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31b9e6543d2043d285c9bd483051c281 : SUCCESS in 26m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba21641e622a48f0add6d5ad7e20b4a4 : SUCCESS in 28m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9fe57261c1646da9424e3be27151c1a : SUCCESS in 11m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/802ccb71e8ce40529e138f43e507d6c2 : SUCCESS in 1h 04m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/52a4bf60a87244e8995e540361b44f6b : SUCCESS in 1h 19m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/441a1f1e2cc2415695e7c0132c275b41 : SUCCESS in 46m 50s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/0396b7cd452747bf94cb15fbcd59038e : SUCCESS in 45m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7979ba2944e94cee8a16ca6f2fa4b471 : SUCCESS in 1h 03m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7eb134c85bee45c394b0354322e199f4 : SUCCESS in 2h 01m 20s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f1d3e3fe9f144baea56fc7d02d4e61f6 : SUCCESS in 1h 23m 02s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/667527dd834b4f378f799bc193b3f0d9 : SUCCESS in 1h 21m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/38ae2aa98a194131b1ad196f3316d14b : FAILURE in 14m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e0726606a60c4975b176965f00d90de8 : SUCCESS in 1h 17m 20s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/04bce131068740d68db0c0b43b9b03b2 : SUCCESS in 2h 12m 52s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/be074415fd8b4d9490db21df852f2657 : SUCCESS in 2h 07m 05s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9a62bbd340ad4a81b2fec7f29890afbd : SUCCESS in 1h 36m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7ad2fef2633b4a7b91f57295fd5fa220 : SUCCESS in 1h 14m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9a1d8ea5f33f439782b98c976416044a : SUCCESS in 1h 11m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ce25fb6466bc47708404db7bebc6f98f : SUCCESS in 2h 33m 57s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/fd03325ea89141aeaa08e4d0e5482a89 : SUCCESS in 2h 32m 05s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/e6e9f0bd3c154ae3849023c791c022b8 : FAILURE in 40m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a8563dae2ac4107bc075ba53107ce4b : SUCCESS in 1h 00m 31s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/33eed0ba092c4f1482848352ddf739ca : SUCCESS in 1h 25m 53s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/45e0202a8e9c4e8aa0ed2ead94db5ad2 : SUCCESS in 49m 45s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/50d7daef566743df9fcd17be42a414bc : SUCCESS in 1h 14m 56s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/73ffb2a884d74603b7259c94e6109a4c : SUCCESS in 1h 33m 36s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/85abd482e82b4d52a11c89463dd1e62c : FAILURE in 1h 36m 20s (non-voting)\n\nWarning:\n  Comments left for invalid file doc/source/ext/support_matrix.py\n  Comments left for invalid file neutron/tests/unit/test_auth.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_daemon.py\n  Comments left for invalid file neutron/common/eventlet_utils.py\n  Comments left for invalid file neutron/tests/unit/agent/ovsdb/test_impl_idl.py\n  Comments left for invalid file doc/source/configuration/l3-agent.rst","accounts_in_message":[],"_revision_number":2},{"id":"9e89d12d367331f2798bc466fa245212f7347654","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-09 07:31:16.000000000","message":"Abandoned\n\nrally passed, all good!","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9c616d17488deff2698665aef06fe7faf3aee8df","revisions":{"f00bc1c96635a59fa452202dfb556aece3ac489c":{"kind":"REWORK","_number":1,"created":"2020-04-08 12:07:16.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/08/718408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/718408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/718408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/718408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/718408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/718408/1"}}},"commit":{"parents":[{"commit":"c4580278bf8f199d0cc3ed5fabdcb3c4fc89d3fe","subject":"Merge \"Revert \"Switch to use cast method in dhcp_ready_on_ports method\"\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4580278bf8f199d0cc3ed5fabdcb3c4fc89d3fe"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-04-08 12:06:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-04-08 12:06:36.000000000","tz":120},"subject":"[DNM] Test rally pinning","message":"[DNM] Test rally pinning\n\nChange-Id: I6ba6309f5b8a33beee8eb2a6b1c39c6353c05814\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f00bc1c96635a59fa452202dfb556aece3ac489c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f00bc1c96635a59fa452202dfb556aece3ac489c"}]},"branch":"refs/heads/stable/train"},"9c616d17488deff2698665aef06fe7faf3aee8df":{"kind":"REWORK","_number":2,"created":"2020-04-08 13:35:03.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/08/718408/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/718408/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/718408/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/718408/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/718408/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/718408/2"}}},"commit":{"parents":[{"commit":"c4580278bf8f199d0cc3ed5fabdcb3c4fc89d3fe","subject":"Merge \"Revert \"Switch to use cast method in dhcp_ready_on_ports method\"\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4580278bf8f199d0cc3ed5fabdcb3c4fc89d3fe"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-04-08 12:06:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-04-08 13:34:48.000000000","tz":120},"subject":"[DNM] Test openstack-rally","message":"[DNM] Test openstack-rally\n\nDepends-On: https://review.opendev.org/718383/\nChange-Id: I6ba6309f5b8a33beee8eb2a6b1c39c6353c05814\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c616d17488deff2698665aef06fe7faf3aee8df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c616d17488deff2698665aef06fe7faf3aee8df"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
