)]}'
{"id":"openstack%2Fneutron~720044","triplet_id":"openstack%2Fneutron~stable%2Ftrain~If60f4ea9793b0872e348a6319f2c24127e93f540","project":"openstack/neutron","branch":"stable/train","topic":"bug/1826565","hashtags":[],"change_id":"If60f4ea9793b0872e348a6319f2c24127e93f540","subject":"Check tc_lib.add_tc_policy_class input parameters","status":"MERGED","created":"2020-04-14 18:04:42.000000000","updated":"2020-04-16 17:20:04.000000000","submitted":"2020-04-16 17:10:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720044-1587057014792-5f04ccdf","meta_rev_id":"d7d5aafdb7fe989c8249add7fd76819c82771398","_number":720044,"virtual_id_number":720044,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-16 17:10:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-15 17:03:25.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-04-15 23:51:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-04-15 14:56:41.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-04-16 03:32:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2020-04-16 03:32:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-15 14:56:41.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-04-15 17:03:25.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-04-15 23:51:36.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-04-16 03:32:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-16 17:10:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"70e540dd5cca3e9e05b009945d23c2bf52ef77ca","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-14 18:04:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b51327beabbbb248c76d7c26139cdf1d8586cad","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-14 19:57:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/720044/1 : SUCCESS in 1h 51m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ecfb880fbe72b72ea6cd339b53b5dcc9f058407d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 02:21:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/f77a7724cb6f45eb9dc7002491c1b42a : SUCCESS in 1h 02m 30s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/5a8968de14a742a289f1ccb30989e44b : SUCCESS in 36m 04s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/83b304ede4104c13ac656cc49fc52aa2 : SUCCESS in 1h 57m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/4089d2cdc9fa49a78e7e2986d4178d92 : SUCCESS in 1h 17m 14s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/24f56c746cc74bdfb03f2e06b1795082 : SUCCESS in 1h 18m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/3a4a168594c34630954d6e583f8048cb : SUCCESS in 1h 14m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e62f5087c9124af99df834e2f60391e1 : SUCCESS in 34m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/572f0e5b1475469cae77167c34526173 : SUCCESS in 20m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fcb85c331de4438a23a9b5a2a9f68f6 : SUCCESS in 14m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e82060a88b564de9a1bf46218e85e780 : SUCCESS in 25m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63c1b1181d22440f9826e4048aa4d497 : SUCCESS in 21m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ef702774a054f8f8e08c0ca96514327 : SUCCESS in 41m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f46dc8c08f14708a20de47a10c8b9e0 : SUCCESS in 12m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/da3721ce755e409b95ead369ccf90956 : FAILURE in 1h 19m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/162f161aa3fa4927be56a74293728239 : SUCCESS in 1h 31m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2410a939d68846b1af5de36ed0cbb79a : SUCCESS in 43m 49s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/95e89d6d80a04f6194de056ea8e336f0 : SUCCESS in 43m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7c44775d6933480d94724575484e19dc : SUCCESS in 1h 02m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1a205df3f50542d0a41b576c30858c25 : SUCCESS in 1h 51m 22s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b766c075a0334b43bd794812157fe85c : SUCCESS in 1h 13m 53s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/176dde72a62a43e1ada7548bc3075844 : SUCCESS in 1h 11m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/00d4348da7e94236a78bfdb20986af88 : SUCCESS in 1h 01m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3de8062f237846bcbd3c20a5c1933eed : SUCCESS in 1h 16m 03s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ef33a7dc402246d5a5e3be1ce2b0b356 : SUCCESS in 1h 47m 56s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e0ad12bfefe0467f85a0d210e23fb7a4 : SUCCESS in 1h 51m 02s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/37ca25a5a3f74e4292a54d5ac041d591 : SUCCESS in 1h 20m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0018707d00a142de8b7228656d5ebbab : SUCCESS in 48m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/30230bfaa1834ec5a51d3c66a2b2be4c : SUCCESS in 1h 04m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/74be65da8fee4872ac3e0d97f266c3e4 : SUCCESS in 1h 48m 10s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/6e62207d72534bcf98538053809ffa88 : TIMED_OUT in 3h 05m 48s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/023d988d3cfe4ebfb7c819ee34c758b9 : FAILURE in 46m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa7127e1263747c8ac616aadc9907a4e : SUCCESS in 1h 11m 52s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c1f51962d91d4e78bc24235fe1fa9956 : SUCCESS in 1h 25m 40s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e60ffe45d024e68b54101309be97dce : SUCCESS in 38m 02s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/add1bfb4513a44daa06ac64e49d2906f : SUCCESS in 1h 17m 31s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d68c048c42a48d983ba3fe5eaa7a15f : SUCCESS in 1h 17m 04s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5549c25a799a43e2a48e721317e30485 : FAILURE in 1h 25m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e08a1aa92fb17edf395654ad3cdf60e7f518052","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-15 12:27:36.000000000","message":"Patch Set 1:\n\nrecheck grenade","accounts_in_message":[],"_revision_number":1},{"id":"9e9565e355f4bfaed31eb2bb7a9af340088a5ceb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-15 14:39:31.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/720044/1 : FAILURE in 2h 08m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ae6060542a56d35a23ab77543a6616c9d4e00afd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-15 14:56:41.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c1d98368d4962ebc7e1c2186974410655cbf8ca8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-15 17:03:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/720044/1 : SUCCESS in 2h 06m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a61ef8fdda3f222959d928cef33f1f7653dd5524","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 17:51:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/2d4b3614015746028b1aace7967adac0 : SUCCESS in 54m 58s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/7ee9b7e523884ab1b930f61f29efde6a : SUCCESS in 35m 38s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/4ee1938edb5a4a7ab1d1e13312a9869c : SUCCESS in 1h 33m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/806fffb33e2645eba238b2f90331ad25 : SUCCESS in 1h 08m 04s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/ca6dbdf3e1ff436980cfac927390f919 : SUCCESS in 1h 07m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/9172282fc02b40cd96992da55fa50d74 : SUCCESS in 1h 08m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d7febc76b0d4a6286b4876b044bcacf : SUCCESS in 33m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/357ec813737940f6bd971c40a7cedae6 : SUCCESS in 23m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5fec4f45a9547b8a736b7e53cabe26b : SUCCESS in 14m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46c6208f11384febafecaf53dc51c866 : SUCCESS in 22m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df6b74a7c7dd43e9ac8329a293754aaf : SUCCESS in 27m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8b0d1613930432a9f8100dcaa7c3ee0 : SUCCESS in 26m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3962ae9f12b4156a811726be2ff83de : SUCCESS in 10m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5f5d114436d44aea975c7c6171bc3da5 : SUCCESS in 1h 04m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4402f49cbaf0453899cf704710f823a6 : SUCCESS in 1h 27m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e0e395f42d6947e1b00740887f640bbd : SUCCESS in 37m 20s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/7fbb2497ead24449999e636f52f7a63b : SUCCESS in 37m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f57bb99bf79d432992ab9dbcf7516f76 : SUCCESS in 59m 06s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e21a7cf6e4f64c84b0a2ecad3ef74123 : SUCCESS in 1h 50m 25s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cdb4b8571d41429fb6251aa7a16f0321 : SUCCESS in 1h 24m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9b7d85713ee94683a0bc50c2822c2a79 : SUCCESS in 1h 26m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/accaeb6493d64cf0bbce49c685573b9b : SUCCESS in 1h 23m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0a4080ea54454e359acf78b5f1dba259 : SUCCESS in 1h 12m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/698c8b86332548ffa985f8b67406e810 : SUCCESS in 2h 02m 02s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1a6fccf63f2f40f1825ca970d9cc3b5f : SUCCESS in 1h 52m 05s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d8ee819f43344031aaac858305c2efeb : SUCCESS in 1h 24m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0af9d4b775f84e6aab63f1b2d05f6ec5 : SUCCESS in 1h 01m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/511186301ea3448a915ddf089942a428 : SUCCESS in 54m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2b717109634a498fb068bf9bfacac750 : SUCCESS in 3h 03m 11s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/3763261f72df4deba7a0281565306492 : SUCCESS in 2h 34m 55s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/91944dc878ff4591a007822735a1618d : FAILURE in 44m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e95f08209eba468ab4329d0a368a2304 : SUCCESS in 1h 08m 28s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/52d768c456174533ac55e7a5e73c9b15 : SUCCESS in 1h 20m 15s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f2274c65293d43f2865e434d6d1e1047 : SUCCESS in 36m 51s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/72b367cc087f45a4b26331829bc8b60f : SUCCESS in 56m 02s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a7a8ac21d9194992a6959f08139c9cf1 : SUCCESS in 1h 21m 29s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/dee2c753e59e41ae8df92307b1552426 : FAILURE in 1h 18m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"835b6c3a534cf43fabe1b9f44bac11653b8b0dd1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-15 23:51:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d458a2a8b1f8995b52d895f2028e77b82bd591a7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-16 03:32:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8e3105d06c45c786c34f17b772ee523a27619da9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 03:33:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac670f4247b65c339f172d189b48d458d8e660f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 17:10:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"60e8fddc1934165142663b775e01c27ab251928f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 17:10:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/26a08d0a57fc4899885abff6254429c5 : SUCCESS in 56m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54b9ce13a4804942828bb2590d161de2 : SUCCESS in 22m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27ccf7b335b14a27bcccfddeeee65429 : SUCCESS in 14m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c18b21000996454381f6a6b12634d31e : SUCCESS in 24m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/872dff5cce18400ab33c8fd68b1085a3 : SUCCESS in 25m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0944c77752f24da1b9f7d52dc430fc6a : SUCCESS in 30m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37264a41b9a74497b07b5b31cd076b10 : SUCCESS in 10m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f365603b186643afaba80518294acd53 : SUCCESS in 1h 03m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/20aab06d30434dd88d09d22ffb806889 : SUCCESS in 1h 20m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7ebb732cbc984aafadf19214598adef6 : SUCCESS in 44m 14s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/67256356200342a19f75eebcc537c593 : SUCCESS in 42m 04s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7abdacdfc7d74b0caca796d7cfcd0a06 : SUCCESS in 1h 09m 33s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/85b1adee4487464eaab8fcc4cfc53cf7 : SUCCESS in 1h 23m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/84cafd0426df4335bfca3809862ad8e7 : SUCCESS in 1h 18m 48s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c00cd6cc061c4d27aca1980f7661e60c : SUCCESS in 1h 18m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a23e0cac4d9744f7b7b3c4ffa0f389f4 : SUCCESS in 1h 26m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5f48b7cabc5c4fbcbb4918886af200a1 : SUCCESS in 1h 11m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c9f01e5d66da41388f002fec89aad9c9 : SUCCESS in 1h 00m 49s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a6b6eecf50c24c678c4547a357410d26 : SUCCESS in 2h 25m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/351f4ae4afc04ab4a0423714dad1924d : SUCCESS in 1h 00m 59s","accounts_in_message":[],"_revision_number":1},{"id":"bbee54eea18ce7f0a3065cc141df17f67844b2a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 17:20:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c243944578a4890a3c8ce759befecbb : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62cbdfbaa2830a526c3390e1d389965133f76cd5","revisions":{"62cbdfbaa2830a526c3390e1d389965133f76cd5":{"kind":"REWORK","_number":1,"created":"2020-04-14 18:04:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/44/720044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/720044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/720044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/720044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/720044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/720044/1"}}},"commit":{"parents":[{"commit":"0471f2c2e22b1c515f0bab0758cc6cd92477f068","subject":"Merge \"Ensure netlink.nla_slot tuple key is a string\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0471f2c2e22b1c515f0bab0758cc6cd92477f068"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-03-03 13:40:32.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-14 18:03:26.000000000","tz":0},"subject":"Check tc_lib.add_tc_policy_class input parameters","message":"Check tc_lib.add_tc_policy_class input parameters\n\nAll input rate parameters (ceil, burst and rate) are mandatory in\nPyroute \"tc\" class [1]. \"tc_lib.add_tc_policy\" should provide them.\n\nmax_kbps: Now \"max_kbps\" is mandatory.\n\nburst_kb: In case this parameter does not exist, the library uses the\nmethod previously defined in Neutron, consisting of asigning the 80%\nof \"max_kbps\".\n\nmin_kbps: In Pyroute2 [2], the value of \"burst\" depends on the \"rate\"\nvalue. If the \"rate\" value is too small, the \"burst\" value will exceed\nthe 2^32 limit defined in the iproute2 structure. In order to avoid\nthe error described in c#4 of the related bug, the method checks if\nthe provided \"rate\" value is greater than the minimum possible value.\nIf not, the minimum value is set and a warning message is logged.\n\n[1]https://github.com/openstack/neutron/blob/fb2453c999f1757cc9085f4533bb2fd56d6b2245/neutron/privileged/agent/linux/tc_lib.py#L107-L108\n[2]https://github.com/svinota/pyroute2/blob/943502a95cbbbac48218d2bde9d92af462682f96/pyroute2/netlink/rtnl/tcmsg/sched_htb.py#L88-L91\n\nChange-Id: If60f4ea9793b0872e348a6319f2c24127e93f540\nRelated-Bug: #1826565\n(cherry picked from commit 1e69279817b4fe978e905987db90dfca9b636b0b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62cbdfbaa2830a526c3390e1d389965133f76cd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62cbdfbaa2830a526c3390e1d389965133f76cd5"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
