)]}'
{"id":"openstack%2Fneutron~720447","triplet_id":"openstack%2Fneutron~stable%2Ftrain~I8fa7176d1d9667c6b5cc95af0e31210d0f1c3662","project":"openstack/neutron","branch":"stable/train","topic":"bug/1873240-stable/train","hashtags":[],"change_id":"I8fa7176d1d9667c6b5cc95af0e31210d0f1c3662","subject":"Add rootwrap filter rule for radvd-kill script","status":"MERGED","created":"2020-04-16 11:51:43.000000000","updated":"2020-04-22 17:09:42.000000000","submitted":"2020-04-22 17:06:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720447-1587575186997-a0c18567","meta_rev_id":"c33bcf7d70c5f02203eb13e1047090bf91e9b1d9","_number":720447,"virtual_id_number":720447,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2020-04-22 11:24:44.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-22 17:06:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-22 13:07:20.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-04-20 10:45:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-04-18 13:20:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-04-20 08:05:11.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-04-21 15:38:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2020-04-21 15:38:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-18 13:20:40.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-04-20 08:05:11.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-04-20 10:45:16.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-04-21 15:38:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-22 13:07:20.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-04-22 17:06:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bddd9df7d7652337f864f9fd8fe46523639c3c3c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-16 11:51:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"29871cbf96ce6bdd9f5efd8e79324e762f948597","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-16 11:51:59.000000000","message":"Patch Set 1: Workflow-1\n\nlets wait to merge it in master first","accounts_in_message":[],"_revision_number":1},{"id":"d5884cabf21d0a979b6026f6c91c0f2444eccd39","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-16 11:52:52.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit dc88432f19fa9e26ead1b05c0168b3a995c1bb10","accounts_in_message":[],"_revision_number":1},{"id":"255dab364e25604532213574537e64d764fc2638","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-16 13:43:35.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/720447/1 : FAILURE in 1h 51m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a8679993952fb125eb985aa463303e54f97ab412","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-16 13:59:23.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7c086421a4971d26c3b980269681146006f87b8a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-16 14:00:00.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"f50b97e96c407700b328835870891c069b7adc60","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-16 14:19:59.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/720447/2 : FAILURE in 20m 03s","accounts_in_message":[],"_revision_number":2},{"id":"cafaed9dd50852006541195ba905ee90d44674ef","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-16 20:14:40.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"61d53931d0dfc9e86c6064840c0b0938b40f9dbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 20:58:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/c630c23d9db24da1b2d0c08aef2c3f27 : SUCCESS in 51m 28s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/9e280e7bfbeb4bd48c9edf259256cf70 : SUCCESS in 44m 52s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/765d0de10eea4c33a4d53d656a4ff980 : FAILURE in 2h 05m 14s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/271534836fbb4378bc06dbd1cb6202a6 : SUCCESS in 1h 09m 30s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/69d5877140f44c9ba9536ab5f8dcc906 : SUCCESS in 1h 52m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/c58f51c4517c403baa516e389f55739b : SUCCESS in 1h 16m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c12bceb8ece423f80e0fdd2d8df09f2 : SUCCESS in 31m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6097e65cc1664e14a92224d3f426ff78 : SUCCESS in 22m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d4396f1a51b403a81c1eb31caeaf379 : SUCCESS in 13m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c1adb6aa8bc47c5af20ac8729688a95 : SUCCESS in 23m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae59249102aa4a8c8c6ebcb1b9602570 : SUCCESS in 23m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/08da7c04a6c0417a81625aeacecdc95d : SUCCESS in 23m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba0a423daa78407c88c959fab08d6356 : SUCCESS in 10m 29s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/23d2833a9a32454a9d90cdbf23996996 : SUCCESS in 52m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d4eb69f39d804703b2cb5b5babdab2eb : SUCCESS in 1h 09m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4cc5fd9031eb44349957fae7ad19d3fc : SUCCESS in 44m 03s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/c352fe62bc3647a084cbad55fa107957 : SUCCESS in 42m 55s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/38c35ccadf1648e0a028c3702466e856 : SUCCESS in 1h 03m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d109b0de7bdc4c779df7121604e0a577 : SUCCESS in 3h 00m 46s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/92a892c917344103bac668f03b56fead : SUCCESS in 1h 17m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9b40b6f6e216424fb198e486c0aaa041 : SUCCESS in 1h 05m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a4db587057a409bb05deb10d8a3ee60 : SUCCESS in 1h 18m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2699a964e3224b148377b3c6830ffe12 : SUCCESS in 1h 22m 36s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/09ba4e1a36d449fd9c230863ff5b4f83 : SUCCESS in 1h 49m 40s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5223f7ae1a01448083af34e12f2ea7cc : SUCCESS in 1h 53m 46s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d33df749db2d4357b4571823ef844c79 : SUCCESS in 1h 11m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9d9ec04a32054597a945484f224deb67 : SUCCESS in 53m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f89947d5bcd84c909a639e31b13d148d : SUCCESS in 1h 02m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1cee5b08841e4cb594540f87738fff82 : SUCCESS in 2h 25m 31s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/081d96120c184215b74dd5a4b52f06bc : SUCCESS in 2h 08m 49s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/516b865dae02465c9930dde7e584f985 : FAILURE in 42m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3e387dfd62a40c1af633dfd4190ccd2 : SUCCESS in 1h 01m 35s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d5cc8765eacf41dbae62c9562587199b : SUCCESS in 1h 19m 59s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b3aa4c325ef4d908308ee6decda6522 : SUCCESS in 38m 54s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3c067967c5e142c38f648281b24501e0 : SUCCESS in 1h 12m 07s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5159b32fbbe344a8b6bb59a2f267978f : SUCCESS in 1h 16m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58d2c2f57bd0158447b85a6909c73b019a92229e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-16 21:03:17.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"75cd1e7834254a0dafbab7090dc9f4a14927a49f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-16 22:49:15.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/720447/3 : FAILURE in 1h 45m 07s","accounts_in_message":[],"_revision_number":3},{"id":"971cdec2f997e6b2489208a43fa9ba24674f3854","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 03:37:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/90140b133df2457989d86f392f720e84 : SUCCESS in 1h 07m 38s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/531d4b9dbbd04aadba970d7256f757ba : SUCCESS in 30m 29s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/ef2e01745e404e79a2801830b7f563d6 : FAILURE in 2h 34m 15s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/113879aa62ef4198925c8bcd24748eed : SUCCESS in 1h 36m 22s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/a514a26aa4574bf0acf18fad6c12f1ca : SUCCESS in 1h 33m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/823c4706608340ab8da951b1a7080dfe : SUCCESS in 1h 24m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54c5547fe04a4b488c6d71eb221b3d0c : SUCCESS in 31m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eead2fc7f041409bacb9e7ef3da4d80b : SUCCESS in 24m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b79adc5b13445a9ae4b93d5ff341dc4 : SUCCESS in 14m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/444993a7cdf14a9baa651faab2c193aa : SUCCESS in 23m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5c1ef7a3edc4213a599372a7db45f83 : SUCCESS in 25m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/338c44f6c5fb4273b6e0cdb1f83a3933 : SUCCESS in 29m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/474d0f7a163f4d8086603d0629915dbf : SUCCESS in 10m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e4dca64715394cbdaebd6e33a1c6bae3 : SUCCESS in 1h 10m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6eb8f57047064267a7d455f719b8da9a : SUCCESS in 1h 13m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9fb863ec48f5483e8ad4f30e9f822acd : SUCCESS in 37m 23s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/8383ea4fbb244f21830ec000562a2fed : SUCCESS in 38m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f30f6a9c3f61436e9bbc6a08e3d7251e : SUCCESS in 1h 00m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8fcc9fea4674493bb8f5f69f471335fa : SUCCESS in 1h 47m 17s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3a391e25adf84d85a6218fa1afd1e0c7 : SUCCESS in 1h 36m 27s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/98618cf417f644fc90480e1388d039d8 : SUCCESS in 1h 18m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f7e35d22cdd4f13a7887518b4132d18 : SUCCESS in 1h 15m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a9ffbfc192d44b28a91f61ce8b895790 : SUCCESS in 1h 08m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2d32211fe89a4bfca8d80104c0a1cdaa : SUCCESS in 1h 38m 40s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1b6cc4c0c65540a59b1006e5da004fc7 : FAILURE in 1h 26m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/137cc434bad441dc9db4f090e5a46e7e : SUCCESS in 1h 30m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b149fb1566984d6fa3d7f4cf4ec2f182 : SUCCESS in 1h 00m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/845a346230564e2fb5934afb820acbeb : SUCCESS in 53m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7cdce703748b4d1da48d332f85d3c625 : SUCCESS in 2h 27m 55s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/cc7453d68d664aff9f7136dac6b6b898 : SUCCESS in 2h 25m 10s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/7b95ed45741342b39b71c22f913c88b2 : FAILURE in 45m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6d3131f76734bb3a494a4cc107d28ab : SUCCESS in 1h 13m 38s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d854bf596aba4e3aaff4fb1c512cbdce : SUCCESS in 1h 28m 19s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fb2c5124ce3a48029af000a6eb9426ce : SUCCESS in 45m 33s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c9bcda9442874ec8a14acbd8897fb693 : SUCCESS in 1h 42m 59s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d0293771925c49eda200a8f1662e33f6 : SUCCESS in 1h 23m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1d85c0a595b6bf49de7ecfcace373acd1388ebfa","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-18 13:20:40.000000000","message":"Patch Set 3: Code-Review+2\n\nOne-liner fix, I only updated master hash (which is merged now)","accounts_in_message":[],"_revision_number":3},{"id":"88f1d2b622cf832ced67f6ba15bebf6723f47956","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-20 08:05:11.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d9a4be8dcd87d844d192d4a2933165ed85a629b1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-20 10:45:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"23f7daf7153371df1bb2c6cf9192e76890f85768","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-21 15:38:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5a2bdb62c394f8b50e5ad4a338e02516c6a55af3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 15:39:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8bdeb4aadfb9590bf23f9ad4ed9c9ef6c3e89f04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 19:05:55.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/aa8efce216714697a918e5eaca3af2e0 : SUCCESS in 54m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a94b77381a5c46a6be32edf56ce28f25 : SUCCESS in 24m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98157208ae4e4af0893881c5da1437d3 : SUCCESS in 14m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/93286b07a3ad474da0e461618335d348 : SUCCESS in 26m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cefc7ee5fc9c4ba384e0c6ed35401cc7 : SUCCESS in 21m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/05b39ad00d2644d69bcc37190a618bf4 : SUCCESS in 25m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caebe5417a114a258ced40b19f01de31 : SUCCESS in 12m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4af75f2dc7a34ca489a4f7ab0199cdb6 : FAILURE in 11m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4794839b46354f0e9a27f4d2a615275a : SUCCESS in 1h 16m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fb7c7b562b4c4f70a9040653ccef3e6e : SUCCESS in 49m 47s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/40baeacc3d324effb8b59ab4986fefd5 : SUCCESS in 49m 39s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/44fe967e5a734e6d9e9ad69fdc7d2cf4 : SUCCESS in 1h 15m 37s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/9cf10790691041ee860afe29f789b7d7 : SUCCESS in 1h 25m 41s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7194efe31c534d538c07a4de500b4b89 : SUCCESS in 1h 16m 38s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1db22dcc73f74c4bac648b2567e7edb4 : SUCCESS in 1h 19m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb281a784c3f497580fa378d1b9018eb : SUCCESS in 1h 24m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2b6bf25fd8e24a67a60fbbf1d1c52e03 : SUCCESS in 1h 08m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/04f0d376ce0c45deb41fbcd81c064a60 : SUCCESS in 1h 03m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/18c7b272dc6a4182a72e81f584ae18a0 : SUCCESS in 2h 17m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b38491297ec54182b09da7f35639ef5c : SUCCESS in 1h 02m 45s","accounts_in_message":[],"_revision_number":3},{"id":"49f89a7fd32f0d356f10bf5cdfb3f6c418b29c11","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-22 11:24:44.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5a0dc6c8f8ef54525ac33b0b6113a153fcce50bf","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-22 13:07:20.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/720447/3 : SUCCESS in 1h 42m 05s","accounts_in_message":[],"_revision_number":3},{"id":"487842afbdf3d8649f774700276c930c3fd36799","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 14:14:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/847440461dda41d39cda1ca08f1f0c25 : SUCCESS in 1h 07m 55s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/c866fbc94b4a4f7abedcdf3a960fb84e : SUCCESS in 47m 45s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/49d336008a5f4891a7abce07b797c0d7 : FAILURE in 2h 37m 32s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/5ec335502e604751880ffab4c067ba7d : SUCCESS in 1h 11m 44s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/c556dcd14508452e94de845bcff5d0a4 : SUCCESS in 1h 26m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/9920444b9dd74fdcb58712054128c93d : SUCCESS in 1h 18m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d2e9e03ec6544c487a9a17ad3f6dcf9 : SUCCESS in 30m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/faa65cd2ff7940759a250352810cff58 : SUCCESS in 23m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8a7c2dcb416448680b4d99ff4e438eb : SUCCESS in 13m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0a7211a0aba446158dd593c98cee34c1 : SUCCESS in 23m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1fee56cff1849c7a71ba3115f915fa5 : SUCCESS in 26m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff52bcf7a35c44948c2fdeabc082cc7d : SUCCESS in 27m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/957441bffd65456d92219f9565090223 : SUCCESS in 8m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0ae89ae7b02949dfad04334834d9ad8f : SUCCESS in 59m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a36006b3bff744c29a7ca3ff4898d793 : SUCCESS in 1h 18m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f6a13973e4bb48268f76295044761c3f : SUCCESS in 40m 42s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/7844f3442c844b93b0491ba9ccb08d59 : SUCCESS in 46m 19s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d920606cbbfa44469b28d1236cfe43ba : SUCCESS in 1h 03m 41s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5701bb7d98314e43ae5e8d203842c430 : SUCCESS in 2h 23m 59s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c4353e6521fb46a19ef5e1412045c781 : SUCCESS in 1h 18m 31s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/66c42a105c48430d914cbb1dbdef38a3 : SUCCESS in 1h 19m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f7fd4cdef7144c3987b090159b231003 : SUCCESS in 1h 09m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6a35bfe89ad14bd7ae5f62cc35c4757e : SUCCESS in 1h 21m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5d37753c3bb84eb7b8abe09fd055027d : FAILURE in 1h 18m 59s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3395b0f5c83b4854b63740e069c89205 : SUCCESS in 2h 02m 17s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a00f5f4a79f746cc9be3a29eed112461 : SUCCESS in 1h 21m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/13bbd40f67fb4c73a3be6e3e68ae8cea : SUCCESS in 1h 00m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/38faadb685214cb0b45bb60ac4fb7dd0 : SUCCESS in 58m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/40acb6157f6d4b9e91a807dfa4b43086 : SUCCESS in 2h 32m 36s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/119135ffdf2d48f799df13967c47bc2a : SUCCESS in 2h 37m 54s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/3dc0a36e1cd344fd93152e16cfce749d : FAILURE in 36m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ab532eea2d34e4ba18cda1e1e96d650 : SUCCESS in 1h 02m 05s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/319dc82608064f8385df64cb01306c5f : SUCCESS in 1h 20m 46s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/04eba2541de34ef8a465e4079aa3054c : SUCCESS in 39m 45s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cd647390445b4504a2bfbcab27c65799 : SUCCESS in 1h 03m 20s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f0718b0496d54881a9ce7212c7497444 : SUCCESS in 1h 15m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5650089d9d6b7c407a6818a58f8116d0e6dc8a9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 14:16:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"92d03fb6b82d68244d487f32006a43138ac2a233","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 17:06:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0e2a5ca6b594e207bf34d47f8ef077c29409c5ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 17:06:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/148a5811039d45ea97d93b1dc630fe79 : SUCCESS in 51m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/493db84c4ae84964bc36e117a72eec89 : SUCCESS in 26m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1b38cda742d4a3bb1dfc0b917fe40de : SUCCESS in 13m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/23c45adc006f4c2981bca028985f0f07 : SUCCESS in 24m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6709c670f52c4aa5adc120ddfd923ad2 : SUCCESS in 29m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9e20de3661a4b6eadd0b40b85dc8acf : SUCCESS in 32m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2aaf13be52e42aea096d97df52d75dd : SUCCESS in 9m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6613e796b6bf4e21bebff289ba903267 : SUCCESS in 1h 03m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/600f8c3f956b4d9fadf16367012a46f8 : SUCCESS in 1h 30m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4c1dc69b93714e8e81ec3a057a024ad8 : SUCCESS in 48m 23s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/a00df20548274883a4d349fee3bbaf2a : SUCCESS in 47m 42s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0ac8daa9ac9b41d78f08ccc0afd9e53f : SUCCESS in 1h 11m 23s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/88f04f5a7d714cc7ac4b014c391a588f : SUCCESS in 1h 17m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/67232949484247b99919b613cad01813 : SUCCESS in 1h 12m 36s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7445e436d23b423d8b4afc2f801351c3 : SUCCESS in 1h 25m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e3c6a7d683a42c2b142172478c23148 : SUCCESS in 1h 11m 24s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a1931835982545faaf5b060ec8b2247d : SUCCESS in 1h 11m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ed13799b58384c61b5d2262503618a89 : SUCCESS in 1h 06m 09s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0c2020317bca41ff87b970a011dbccdb : SUCCESS in 2h 24m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bdcf147e72b748e181765f2ceb529e97 : SUCCESS in 54m 18s","accounts_in_message":[],"_revision_number":3},{"id":"a1c75bfe565b4a96ebb7d9fd82d2be6f822fbd99","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 17:09:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fc5d6c3702c4c3e972992ff30769b46 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"026790adf95973ffeb3a2bfbdbb10cdf5a4916ea","revisions":{"ff0216136f458316b1f0fc7e16924d62381162d2":{"kind":"REWORK","_number":1,"created":"2020-04-16 11:51:43.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/47/720447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/720447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/720447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/720447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/720447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/720447/1"}}},"commit":{"parents":[{"commit":"f964b1c08da7e5ddedd16d69ec287b920d2ce390","subject":"Merge \"Prioritize port create and update ready messages\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f964b1c08da7e5ddedd16d69ec287b920d2ce390"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-04-16 10:25:54.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-04-16 11:51:43.000000000","tz":0},"subject":"Add rootwrap filter rule for radvd-kill script","message":"Add rootwrap filter rule for radvd-kill script\n\nIn patch [1] support for custom kill scripts was added.\nWe also added rootwrap filter rules for such scripts to\nkill dnsmasq, haproxy, dibbler and keepalived processes.\nBut we missed to add rule for radvd-kill so this patch\nadds it (better late than never ;))\n\n[1] https://review.opendev.org/#/c/661760/\n\nCloses-Bug: #1873240\n\nChange-Id: I8fa7176d1d9667c6b5cc95af0e31210d0f1c3662\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ff0216136f458316b1f0fc7e16924d62381162d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ff0216136f458316b1f0fc7e16924d62381162d2"}]},"branch":"refs/heads/stable/train"},"928efe23f4edfb8778ce3fba42a6f903e31be79f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-16 13:59:23.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/47/720447/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/720447/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/720447/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/720447/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/720447/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/720447/2"}}},"commit":{"parents":[{"commit":"f964b1c08da7e5ddedd16d69ec287b920d2ce390","subject":"Merge \"Prioritize port create and update ready messages\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f964b1c08da7e5ddedd16d69ec287b920d2ce390"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-04-16 10:25:54.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-04-16 13:59:19.000000000","tz":0},"subject":"Add rootwrap filter rule for radvd-kill script","message":"Add rootwrap filter rule for radvd-kill script\n\nIn patch [1] support for custom kill scripts was added.\nWe also added rootwrap filter rules for such scripts to\nkill dnsmasq, haproxy, dibbler and keepalived processes.\nBut we missed to add rule for radvd-kill so this patch\nadds it (better late than never ;))\n\n[1] https://review.opendev.org/#/c/661760/\n\nCloses-Bug: #1873240\n\nChange-Id: I8fa7176d1d9667c6b5cc95af0e31210d0f1c3662\n(cherry picked from commit 337abbac4432e2271f8cda03ce287d46f9788072)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/928efe23f4edfb8778ce3fba42a6f903e31be79f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/928efe23f4edfb8778ce3fba42a6f903e31be79f"}]},"branch":"refs/heads/stable/train"},"026790adf95973ffeb3a2bfbdbb10cdf5a4916ea":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-04-16 21:03:17.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/720447/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/720447/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/720447/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/720447/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/720447/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/720447/3"}}},"commit":{"parents":[{"commit":"f964b1c08da7e5ddedd16d69ec287b920d2ce390","subject":"Merge \"Prioritize port create and update ready messages\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f964b1c08da7e5ddedd16d69ec287b920d2ce390"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-04-16 10:25:54.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-04-16 21:03:12.000000000","tz":0},"subject":"Add rootwrap filter rule for radvd-kill script","message":"Add rootwrap filter rule for radvd-kill script\n\nIn patch [1] support for custom kill scripts was added.\nWe also added rootwrap filter rules for such scripts to\nkill dnsmasq, haproxy, dibbler and keepalived processes.\nBut we missed to add rule for radvd-kill so this patch\nadds it (better late than never ;))\n\n[1] https://review.opendev.org/#/c/661760/\n\nCloses-Bug: #1873240\n\nChange-Id: I8fa7176d1d9667c6b5cc95af0e31210d0f1c3662\n(cherry picked from commit 2273499155a8cd83150aa4b4f8ba08ecbbdc9306)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/026790adf95973ffeb3a2bfbdbb10cdf5a4916ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/026790adf95973ffeb3a2bfbdbb10cdf5a4916ea"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
