)]}'
{"id":"openstack%2Fneutron~720692","triplet_id":"openstack%2Fneutron~stable%2Fqueens~Ica953cc2038b24c4b3985447b393763912aa6abd","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1871908-stable/train-stable/rocky-stable/queens","hashtags":[],"change_id":"Ica953cc2038b24c4b3985447b393763912aa6abd","subject":"[Fullstack] Handle properly BrokenPipeError exception","status":"ABANDONED","created":"2020-04-17 11:40:31.000000000","updated":"2020-04-20 14:27:37.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a55826c9a6b7f9408a93ae4de4534e311abc6603","_number":720692,"virtual_id_number":720692,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-17 14:36:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-04-17 11:49:54.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-17 11:49:54.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-04-17 14:36:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"94d0e0ce16536e063f496b89c978c9edb5d2b51c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-17 11:40:31.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"18f8c06507af435362df12165f66d428650a121d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-17 11:49:54.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/720692/1 : FAILURE in 8m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f441313ed3d53d0bd5777e69b91ec648e2b3b534","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 14:36:01.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/2d139a4ad53f420a8397e81742542553 : SUCCESS in 47m 12s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/94155dc788da451b855cf02fea436855 : SUCCESS in 36m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/966b8348fd6c4bde8e4794fd526b2980 : FAILURE in 1h 28m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/6ba5d5351e47482aaf2d5b6788cb3129 : SUCCESS in 54m 39s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/a3448f67dbcd4844bc7ba6b8d2111ba4 : SUCCESS in 46m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f6186ee200f4046bc6e9c50f1a8af04 : FAILURE in 8m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d9179504f6cc4e8b94124bb67198042a : SUCCESS in 26m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0e99d8ec4cff45afaaef19c834452f49 : SUCCESS in 27m 36s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9782d7fd0f6f45c9b163e33a2a500cea : SUCCESS in 6m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a908e6b9154044619c590578b1970a52 : SUCCESS in 1h 14m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/022c6c78aa5b4ac3a36a40261ae7dede : SUCCESS in 54m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b04c3378cd74ef9b9abe745e43f9b19 : SUCCESS in 1h 14m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7e421d427f374138acf759e88f9e919b : FAILURE in 34m 42s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d48f9a7005cd457aaec051eb66bd77d1 : SUCCESS in 43m 38s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/090da61d87734740be5c3807d3c0b412 : SUCCESS in 1h 44m 58s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/fb0ee16e851a45c2814010d55831a7ea : SUCCESS in 1h 24m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f043588db8d34244bbc64678b09a2097 : SUCCESS in 1h 03m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/82a2d9ef50a04a589ea2a4e66c751918 : SUCCESS in 1h 01m 47s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/55a8df22864f48a18e53991adf8ab5f3 : SUCCESS in 1h 31m 29s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6aea309b27b64bd8bcc07fc0325cfb53 : SUCCESS in 1h 36m 49s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/2c423376a01e4d9ea1017ee957c875e7 : SUCCESS in 1h 25m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd0d38d7f1ce44b9bee80b0a980a59eb : SUCCESS in 32m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d9c86196526cc89f826b65659f7b8a09e5ee8c22","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-20 14:27:37.000000000","message":"Abandoned\n\nAs this patch is not compatible with py27 and it isn\u0027t really needed in stable branches, where fullstack security group tests are marked as unstable, I\u0027m going to abandon this and not spent too much time on this one.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"57e700516c35a7472da68958404e6470ad315d7a","revisions":{"57e700516c35a7472da68958404e6470ad315d7a":{"kind":"REWORK","_number":1,"created":"2020-04-17 11:40:31.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/92/720692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/720692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/720692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/720692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/720692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/720692/1"}}},"commit":{"parents":[{"commit":"a96f950e4bbe03495dd80cc541a644dce94528ac","subject":"Merge \"Check dnsmasq process is active when spawned\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a96f950e4bbe03495dd80cc541a644dce94528ac"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-04-09 15:05:53.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-04-17 11:40:31.000000000","tz":0},"subject":"[Fullstack] Handle properly BrokenPipeError exception","message":"[Fullstack] Handle properly BrokenPipeError exception\n\nNetcatTester class should handle BrokenPipeError exception\nand not raise it to fail test immediately if nc process wasn\u0027t\nyet started when it tries first time to read/write something to\nit.\n\nChange-Id: Ica953cc2038b24c4b3985447b393763912aa6abd\nCloses-Bug: #1871908\n(cherry picked from commit c92950b0eea55b8f16ae4ecea9366dba4419545a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57e700516c35a7472da68958404e6470ad315d7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57e700516c35a7472da68958404e6470ad315d7a"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
