)]}'
{"id":"openstack%2Fneutron~720693","triplet_id":"openstack%2Fneutron~stable%2Fstein~Ica953cc2038b24c4b3985447b393763912aa6abd","project":"openstack/neutron","branch":"stable/stein","topic":"bug/1871908-stable/train-stable/stein","hashtags":[],"change_id":"Ica953cc2038b24c4b3985447b393763912aa6abd","subject":"[Fullstack] Handle properly BrokenPipeError exception","status":"ABANDONED","created":"2020-04-17 11:41:20.000000000","updated":"2020-04-20 14:27:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"22a1dc2f248a7c0526cad9727cb5bcd5845afb56","_number":720693,"virtual_id_number":720693,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-17 15:17:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-04-17 11:50:22.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-17 11:50:22.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-04-17 15:17:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd218049b7ccd3b54f7b6a6a663c85b738776558","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-17 11:41:20.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"a1917df49da60699fccca967502c365394f7f0a3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-17 11:50:22.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/720693/1 : FAILURE in 8m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a34377eed51251ecab0aa51f2af3c602918d7d68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 15:17:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/7da7e429f5c84c52aa1205b5372f31be : SUCCESS in 1h 06m 57s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/04fca653c03b4d49ba55149803aa9431 : SUCCESS in 32m 46s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/d2e9844b5de24ca2a3a9212dbf8f0072 : FAILURE in 1h 47m 22s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/d21681f48f464cb591f34937a60c7c53 : SUCCESS in 1h 15m 10s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/bb52a11d06fe49bb972982c4be6b60e8 : SUCCESS in 1h 14m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/ee5c1b13a61943aa80f9bca4a44b91a1 : SUCCESS in 1h 14m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5eecf82310bc484899a652e5685c840c : SUCCESS in 36m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be86cb7da31d432a8b2847091ba8714c : SUCCESS in 27m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05f506c588f34c1aa03bfd4c97564838 : SUCCESS in 13m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19c9a2935ca7413a94b7924ee2c675d2 : SUCCESS in 26m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/684780d987bf422c9a0437bf4b6cef74 : SUCCESS in 28m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6a36dabd3c0426a86c4e79546536687 : SUCCESS in 20m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6e955c052154e77ab77dc483c857597 : SUCCESS in 20m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cfba5b50b5e459dada95d20e6557a1c : SUCCESS in 5m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/615a96d54e5443c19a52e4193e41e5f3 : SUCCESS in 1h 31m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e3bac7ced032422283c49a9e5720a63c : SUCCESS in 53m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2d57a4a5603140e99b1c3df27a8fa137 : SUCCESS in 56m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f34af286c0d46ceba6da6a17c5f79ad : SUCCESS in 1h 20m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d911873855a340e8872b2424a6f91557 : SUCCESS in 42m 42s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/800d1466375e44ada47c3c611b51665c : FAILURE in 38m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9928828755cd4d57952ef37ddb40d2c2 : SUCCESS in 1h 08m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d8a6410a20614591aa658d98198af09f : SUCCESS in 1h 26m 24s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2c04514e70c3472a98118d35f1772a80 : SUCCESS in 1h 31m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4628bf30650c4705bba382a769c22219 : SUCCESS in 1h 11m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/56f882c8ea8343daaf98a739fa1c5584 : SUCCESS in 1h 11m 40s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4b63064b8a86441595b57e1b1a6796fb : SUCCESS in 1h 46m 14s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a3d2fffb60f4417e84fae38053f3fbff : SUCCESS in 1h 54m 43s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/21d6096bf0c74a0b9c9223f3696d3e83 : SUCCESS in 1h 26m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/24dc291b69954bf58c18c63db6bc1171 : SUCCESS in 53m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/41d2a14548b14bea80b83a04a76e7526 : SUCCESS in 2h 18m 55s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/9bf37db625744981995ac05ea5906edc : SUCCESS in 1h 46m 30s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/638303d7e54d4b42b8dcd86ad13e099d : SUCCESS in 1h 12m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3625069277c011788c846f84e5db388bf0f2bc45","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-20 14:27:30.000000000","message":"Abandoned\n\nAs this patch is not compatible with py27 and it isn\u0027t really needed in stable branches, where fullstack security group tests are marked as unstable, I\u0027m going to abandon this and not spent too much time on this one.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd81274691d8c3cd33c174dcf4c345a591da9dda","revisions":{"bd81274691d8c3cd33c174dcf4c345a591da9dda":{"kind":"REWORK","_number":1,"created":"2020-04-17 11:41:20.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/93/720693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/720693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/720693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/720693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/720693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/720693/1"}}},"commit":{"parents":[{"commit":"c13391daa424237c0ee6706986b5ef53b9ca78a5","subject":"Merge \"Prioritize port create and update ready messages\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c13391daa424237c0ee6706986b5ef53b9ca78a5"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-04-09 15:05:53.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-04-17 11:41:20.000000000","tz":0},"subject":"[Fullstack] Handle properly BrokenPipeError exception","message":"[Fullstack] Handle properly BrokenPipeError exception\n\nNetcatTester class should handle BrokenPipeError exception\nand not raise it to fail test immediately if nc process wasn\u0027t\nyet started when it tries first time to read/write something to\nit.\n\nChange-Id: Ica953cc2038b24c4b3985447b393763912aa6abd\nCloses-Bug: #1871908\n(cherry picked from commit c92950b0eea55b8f16ae4ecea9366dba4419545a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd81274691d8c3cd33c174dcf4c345a591da9dda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd81274691d8c3cd33c174dcf4c345a591da9dda"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
