)]}'
{"id":"openstack%2Fneutron~720711","triplet_id":"openstack%2Fneutron~master~I8ed1fc85c2f78710bf6589ba3deca518471339b8","project":"openstack/neutron","branch":"master","topic":"bug/1868686","hashtags":[],"change_id":"I8ed1fc85c2f78710bf6589ba3deca518471339b8","subject":"Set OVS manager command timeout and inactiviy probe","status":"MERGED","created":"2020-04-17 13:33:03.000000000","updated":"2020-07-28 11:30:58.000000000","submitted":"2020-07-28 11:27:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720711-1595935657739-4e4c0980","meta_rev_id":"eba07335fec840d6c6935dff188e72fa93451b90","_number":720711,"virtual_id_number":720711,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-28 11:27:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-27 17:43:02.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-07-16 19:10:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-07-28 09:13:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-07-23 21:23:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-27 14:48:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":1,"date":"2020-07-16 15:56:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-07-28 09:13:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-16 15:56:27.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-17 13:42:33.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2020-04-20 09:16:13.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2020-05-29 14:57:13.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-06-06 11:02:19.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-07 20:02:51.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-07-08 08:36:52.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-07-14 17:40:47.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-07-16 19:10:04.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-23 21:23:13.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-07-27 14:48:26.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2020-07-28 09:13:56.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-07-28 11:27:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"319e50182d70c836871f0fed0e7c586b0db5c106","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-17 13:33:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37d19d788379eb8c079499d2e014e61c73930aed","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-17 14:18:11.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/11/720711/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"3980d8d4728d7b0696690e1a1a1df325937325f7","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-04-17 14:45:12.000000000","message":"Patch Set 1: Code-Review-1\n\nSorry, could you please mention removal of unsed code in commit message?","accounts_in_message":[],"_revision_number":1},{"id":"333bd627e05ace9111294e082e5fdd494360d113","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-04-17 14:52:30.000000000","message":"Patch Set 1:\n\nAlso removed test case was actually verifying the setting of inactivity probe, should we add any similar test?","accounts_in_message":[],"_revision_number":1},{"id":"af32ba01ac272ba345ead0e11824852676761857","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-17 15:29:16.000000000","message":"Patch Set 1: Workflow-1\n\n\u003e Also removed test case was actually verifying the setting of\n \u003e inactivity probe, should we add any similar test?\n\nI forgot to write WIP in the title, I\u0027m still working on this.","accounts_in_message":[],"_revision_number":1},{"id":"20eebff877b56d8d8737ca10e7693a18e1241ca7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-17 16:26:35.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/720711/1 : FAILURE in 2h 52m 59s","accounts_in_message":[],"_revision_number":1},{"id":"c508e68a7678027467f4ff334cb568f2ff7f5c57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 17:09:24.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e57cee0f9cc4474388e07cacb7de5907 : FAILURE in 22m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/05c6e4515fdd4fa78ba743abb5807e9c : FAILURE in 24m 07s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/7d57346417734ca582f941edff1aeab3 : FAILURE in 37m 04s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a43b76435dbb46a9b9dd7318e817bfdc : FAILURE in 33m 46s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/28664f71e5f143fd8eb5a47ead5d9e79 : FAILURE in 31m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/593241ece8fb4b19a59fe2d9dee505c5 : FAILURE in 33m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f096e0b83c148a19bf30679e1054406 : FAILURE in 7m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa2785494b5543e993e858de3f3d7efa : FAILURE in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07ee342170214b94bc9bdfa361204d9c : FAILURE in 10m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51b81c435ac04e918dd4e8dfdfd8642d : FAILURE in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69a237e59685417bbc3c9a7a5b1c6919 : FAILURE in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22b08f7b1faf4480b57a32c0e5ff84d7 : FAILURE in 8m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e63e4a0c6c734ef084d628de15241a98 : FAILURE in 8m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8d760cc0b0374e36be9cddd8b3eb6bf2 : FAILURE in 18m 00s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9fa11bf982a64b80b842b51cf95104b6 : FAILURE in 14m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/99796457f9544fb895db89d70535faf8 : FAILURE in 17m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ee693f7abdae43e7810654cd563f71ef : FAILURE in 27m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2cc8e2553964a09b84a0d4ca0f7a31f : FAILURE in 1h 13m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d7bbb0cf64b14c918958d2587ec598f7 : FAILURE in 1h 10m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a0607c6dd6534639be7d731edc5e43e0 : FAILURE in 29m 49s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/45a2f6a83b3b45dfbebecafbd6622146 : FAILURE in 28m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/63261cbd7ca64561befcbf1b3214e7b3 : FAILURE in 28m 52s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/56802e28157242afb5de54864ce9dfb6 : FAILURE in 30m 27s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b1f2a7b09e324bd8862c59f28ebefd35 : FAILURE in 34m 04s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/801c5278fc6c407e9809fea988d686cc : FAILURE in 24m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0168fcfb75b54cc4b96e0ca49aaab031 : FAILURE in 25m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7b3d33e2b2184e47809e208de931cab3 : FAILURE in 28m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/faae98326de64cd9a3330312649844ca : FAILURE in 22m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2b8ef8b8aca04d0696ceede8d1e63188 : SUCCESS in 1h 43m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ba3d972a86f24cdb9cdbfd1ae93caa25 : SUCCESS in 1h 53m 22s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4135c044e9f94b0fbfd8db67d6a15d96 : FAILURE in 29m 19s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3a07901e7236406bae0aa46a52626ec7 : FAILURE in 16m 55s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/26028664289a461a8098472ed7333ead : FAILURE in 2h 07m 00s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bf8f08dc06ac489580a52391ee22ef42 : SUCCESS in 2h 01m 15s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fb16acd22b0a4e3daffc0cad7a7129a1 : SUCCESS in 3h 21m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c1962a7bcbc44cf9a86d37c36a4f98eb : SUCCESS in 1h 57m 22s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/251b23f3fb254b35aacc883edb30e570 : SUCCESS in 2h 26m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22b9be0f37c7eb1b4fbb8e57bfc94a9f97dc4cb5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-17 17:30:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a763ccd278024ee7e27080a8148ccbda1c30d846","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-17 18:26:53.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/11/720711/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"e8576bccfd0ef871bd9bbd05ee125ae451c5c45b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 20:16:45.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/51089959190b40dbbf2b411e70bc6b5a : FAILURE in 25m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c68383b887394ba5ace1233f56088cd0 : FAILURE in 24m 42s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/faf770060852449ab5a11f24a732bc3a : FAILURE in 35m 12s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d3810eaa7e8c4bf0b86f211181ee8923 : SUCCESS in 1h 15m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8dad158be332447287922ee7d2cd4ecd : FAILURE in 23m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e4117042c6294e2f8c288d185cfbb61a : FAILURE in 20m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c1a44db09914f2bb90c3813eb4e3ece : FAILURE in 31m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8152f008727e4e4bbf38e6b74d185ba6 : FAILURE in 32m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f5f5b4d1480466ca6f9d91d4b42d518 : FAILURE in 8m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dfa505b241a42aeb40a378d821da2db : FAILURE in 24m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5260144aa9e74e48b9ddad437eaf3940 : FAILURE in 28m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d4dce8872fb447f99ac9b4843f0becd : FAILURE in 24m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b6214fdd61640c6b1ef8439ca5befae : SUCCESS in 11m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/31943a9befe94dda800021e4cf4054f5 : FAILURE in 34m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/cd4231f4a6b54e0e99b9b87fee71a323 : FAILURE in 47m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e8aa2899c7284994b6540790758cb1aa : FAILURE in 19m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bfdf7fe7c084417bb09abd39144090c4 : FAILURE in 27m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/52c6bd929ed94d1980f01ad122a12e6f : FAILURE in 1h 07m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ab95af5f2891455f8baf71fc16a386f6 : FAILURE in 1h 20m 23s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/02b4a9a755694774969e29625a9dbfc7 : SUCCESS in 1h 12m 11s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/55efba9474ee4c37872464d550f92695 : FAILURE in 23m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2438b662c149436babe614a3f2b3fd9d : FAILURE in 24m 15s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2bd29d4f26e84a65806b42509571a771 : FAILURE in 30m 52s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c12ab6727faf4eff8e912966592bc89a : FAILURE in 29m 38s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/11e459f8e77048deb7edd4d6d3f4c93d : FAILURE in 21m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/93b1d736949b4ab98f3af9f8937a4f28 : FAILURE in 31m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/81f2c4acd98f4d4584dac81e3222cabc : FAILURE in 26m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c57247156484858ab24ed1dd4c4b692 : FAILURE in 23m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/385132011d8c402eae838e74a9073f95 : SUCCESS in 2h 04m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8d206182c0bb44c39cad9389cb4f2130 : SUCCESS in 1h 48m 58s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/dd0d60d286e94d2bbec6ce9cda18c5f3 : FAILURE in 31m 47s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/85b52bfe970b4842acefcdb25f1ff112 : FAILURE in 35m 32s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ee351aa7fd9f40fb91b99117b1180060 : FAILURE in 48m 59s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ca518dc8bc6e41538abad3e66133f832 : SUCCESS in 2h 10m 25s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/261a893ebf81489dad1c9e397b8132df : FAILURE in 53m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1af4b41e8a044fdab30eda695b77d5f8 : SUCCESS in 1h 53m 26s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/abb8706620064578b06910bac13b96c1 : SUCCESS in 2h 28m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dcb087f32c46c68dfe5d9169fd27a500ff3044e5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-17 20:21:53.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/720711/2 : FAILURE in 2h 50m 32s","accounts_in_message":[],"_revision_number":2},{"id":"d24b0af65b325c4fec670b9fafade206c815c8da","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-18 09:54:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b8074b530434c5cb9120c62c4018106cfb93b629","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-18 11:07:17.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/11/720711/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"cf4c8b0dae8a8e64729bb3d950713b1bce52358e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 12:40:18.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b78ff26b50554cceba1fde7772341cde : FAILURE in 23m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/158b14590b5342ce8a8a2ee1b8c4624f : FAILURE in 46m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/7b598dce8412430db2b9a39696baf1b6 : FAILURE in 28m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/08ce0a249e814ced9f9549556c38d9f4 : SUCCESS in 2h 06m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5ce4d5ad111c4477ad560e95af964eb1 : FAILURE in 26m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4fc75ae997664f97b41acfa82bf336d6 : FAILURE in 27m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fc6184646304964a436a661eaa51f7a : FAILURE in 26m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6dedb76db874bb9806671ed7de03641 : FAILURE in 31m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af969851875e4df2b9f08d07aa0a6c2d : FAILURE in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68042379c23a4f1bbbe1e402bfea5894 : FAILURE in 32m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56d125c6562643c893cc8818294ead00 : FAILURE in 37m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b1f6cf88c31483eacbce48a895fef7d : FAILURE in 35m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46a220b6decf4699968ac0a29db59537 : SUCCESS in 8m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4e8fa6ed11894819be8d980dc36ae6c8 : FAILURE in 29m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6dc15e22462f48d2ab9aa9c83013cb3a : FAILURE in 16m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/60b82c200f7347e480041e0b88d4b7ec : FAILURE in 20m 46s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a7ca3bbd4d434c7d84d61214deb8d9cd : FAILURE in 19m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0922f6482334f4f8721fcba1456155b : FAILURE in 1h 02m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8a44f0fbf6754048b8fd3ca85b171d56 : FAILURE in 1h 21m 47s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/605e3907341b4486bea2a5197eff7556 : SUCCESS in 1h 04m 03s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/0bf488481ac244f4a36c559a4920e175 : FAILURE in 19m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8094aeebde684bca8e9ba706fcf7ab3a : FAILURE in 20m 13s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/22388c6c6ed44f46a9bdb0dce0d6df37 : FAILURE in 26m 32s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ebb31cabbbbd4757a1245611472b2636 : FAILURE in 55m 28s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c3e9fe496af54360a61c41bd0ca96483 : FAILURE in 21m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9b2c64e69a20414a99c058e5a9e1f0ad : FAILURE in 24m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/346f9e7b3baf4869b0d60b70a9d25fa4 : FAILURE in 51m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21d0d4c6c75f44aabb9cf18f1ba13b68 : FAILURE in 22m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8d5f803885ff418286e9676c5472736b : SUCCESS in 2h 09m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/403e073769604ad59b3db544e6e214c4 : SUCCESS in 2h 35m 26s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d4ff3d5f334d49cc9f7a92aa0b73cd91 : FAILURE in 22m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/69f7899d75e2408587612cb9b00edca6 : FAILURE in 1h 16m 41s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/86c4941c23bf4a789591863cd838cda3 : FAILURE in 36m 51s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b2df6932474044d09a815ac2dd69a8f4 : TIMED_OUT in 2h 14m 48s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/884c470951c6419b9191e37fb6d51332 : FAILURE in 39m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/18e523c869d14697a8d6c51501543ecf : SUCCESS in 2h 42m 28s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/dd34a9b9b4a84d38b4e694db748bc63c : SUCCESS in 1h 46m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1ca1470b26c727f1a868394c6b95ae041fe0ac0e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-18 12:45:54.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/720711/3 : FAILURE in 2h 50m 51s","accounts_in_message":[],"_revision_number":3},{"id":"f66031ce6b050d7b4a8c4b2c1bd3be8bf3964e18","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-04-20 09:16:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4b5804809c2953729273efb88136e31bacc7a8be","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-20 09:28:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"94ce3c6ae1e9255df37e818a2da6c50fe071960d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-23 16:28:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0404a0c145bcb91b4780a127b86c1d21437d8564","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 16:33:28.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"b8a65174a41ee9c633cb2ac2b31d783082394b23","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-04-23 18:04:12.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/April-2020/Arista_Tempest_Testrefs/changes/11/720711/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"dc7beea879ca5378dc9e5dd8506e65d4ad7102d1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-23 19:22:54.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/720711/4 : FAILURE in 2h 52m 20s","accounts_in_message":[],"_revision_number":4},{"id":"c22f9dc0cdc21c64c3bb28454a78e30cf11c2e26","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-06 09:59:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f940bf4464c7dcb1d78e370b9a9729396b02d71a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-06 12:55:20.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/720711/4 : FAILURE in 2h 55m 20s","accounts_in_message":[],"_revision_number":4},{"id":"e3c4753b3174db4e9308cb93ecef0503b007fc33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 13:20:23.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ccbebc297ff040ab90c9a775b48392ac : FAILURE in 22m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/322a3e4af22f42a88282a4ab920850d9 : FAILURE in 28m 56s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/78158b95df8342248143e18b0e5952c9 : FAILURE in 38m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/19ef4e3dd4f949ba871a30b84eed0237 : SUCCESS in 1h 33m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3370b4443668447e9c85b0801149695e : FAILURE in 28m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5ec59d351f1b4c17b03b222cdfd360c6 : FAILURE in 30m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8118674a98d4e9da1a177b31bc8bfdc : FAILURE in 38m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/297156b9e549409882a79c61589f09ad : FAILURE in 26m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc303f7b46e046f7bfe6bd96f2d5a25c : FAILURE in 11m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f6317028bf8464788c5dbd94b5d7a81 : FAILURE in 27m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41dd225f7e0b4a2d938cc87f43a7ae53 : FAILURE in 32m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c15109c0069424c9ef6b069d6d433c5 : SUCCESS in 11m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d5de63b1f8a14510be8012eeb5d51cde : FAILURE in 36m 00s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/07bd4703cc704e8793a03af775722732 : FAILURE in 22m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b0b5a922f124bf59418f2f984d1a672 : FAILURE in 24m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/07689522db414b1b91e51b86644d1cbd : FAILURE in 29m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/63f70ade395348debe666582d5faf0ce : FAILURE in 1h 02m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/807c394ee72f4aa0b83aaf4782f542d2 : FAILURE in 1h 01m 16s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/af058221f8d249b0bb5d60c907b7b3af : SUCCESS in 1h 15m 17s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/893082bd46aa4ee19283b638a07e768c : FAILURE in 27m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c35f41dcbddb43469a6742a11b56543e : FAILURE in 22m 29s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/413e3f6752344b6db08ffb30c4ae1f51 : FAILURE in 30m 55s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7d95f9d366f348bcac2d73945024042f : FAILURE in 30m 41s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/59bc33fb09bb4774af78e52226bf55b3 : FAILURE in 21m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4ecba9a549454af9b0e1dad1d208ddee : FAILURE in 25m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f473a9d42bcc4cb0877cdaca0028813a : FAILURE in 27m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0a73c7444a44bb291dfa9766f231fb7 : FAILURE in 23m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/57f439d7be0d41ab980104fcbe7b5461 : SUCCESS in 1h 58m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/158a760043e74afbb67c979cb3e33bb4 : SUCCESS in 2h 34m 39s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/db6dabedf748478c813812899bbd2921 : FAILURE in 24m 03s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c7f39bfdb61240a1bc9fa849aad38bbd : FAILURE in 33m 56s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/80a156b059944b739fa7b503c30fad10 : FAILURE in 1h 33m 13s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e9881f4558da4e3bb0c2598c9469a847 : FAILURE in 1h 58m 50s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b4dfdb078f284fcab3be07c86d95489a : SUCCESS in 2h 31m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5bd5d5c6ec434e7ca00bae2db1f2687f : TIMED_OUT in 3h 03m 38s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/bd32021975e6491c8d8a1be66dcbc8e1 : TIMED_OUT in 3h 04m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cdb32059c41d00ca1ccfa692261ff2806433ff20","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-26 08:55:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a0868ebb7f99f3a30ed5b1bd755421d292008238","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-05-26 09:52:55.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"992238312fbbcf7f45d2aefe28c5a08869e5b684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 11:44:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/513bf3c9b949480a86d118caba9851a0 : FAILURE in 21m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/92568ceb14bb47658f8ad053eedae616 : FAILURE in 23m 34s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/3bbe375d1696432ea1036b49f44e3aa3 : FAILURE in 33m 02s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/211496aa7e4d4419bdc067138c8027f9 : SUCCESS in 1h 18m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a129bf6fb0e34f6c8e0c396a2527f26e : FAILURE in 22m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/05b44be603cd4f3dbdb46fedbad0a16d : FAILURE in 22m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e01e5f53d784f36ba2cfaed1cd909d4 : FAILURE in 6m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f5670a777da42568421068a119e8f61 : FAILURE in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6d24d86a09b4cf4ae1a5abe9886f79d : FAILURE in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77137d714df5499f80f432ab1b6f1417 : FAILURE in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0761e27798f4852b98e85c07e87aeff : FAILURE in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbacda6876ee489faf3e09bd8d5ce652 : FAILURE in 7m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/92096b183b3f49dcbf5b01ceb78eab8a : SUCCESS in 3m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e15d72c9909f40b1810a66a9fc54e620 : FAILURE in 14m 03s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b0568b749da842188f7b8cc0c63eda09 : FAILURE in 13m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b18734297ff0433898725dd7fdc54904 : FAILURE in 14m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f818c08a94e34be1b3b1dea0e2408569 : FAILURE in 20m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0adb1b41f16745c199a9849fb049e4f5 : FAILURE in 1h 03m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/39f996a0522240ebb35698b3082f77fb : FAILURE in 1h 40m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/504cc5c83b43449a9baff14c145c9976 : SUCCESS in 1h 06m 09s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3076145ed3554d4aa8bccaacb0433d11 : FAILURE in 20m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fa65be7b775c40d7bdd3b089509cace2 : FAILURE in 28m 40s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/192dd233eaa14266a1ee5b513d861f11 : FAILURE in 49m 24s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fa9281789bef4083afbd84de926bfdf0 : FAILURE in 25m 51s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2b12cf15c631451386758d0b11a4be46 : FAILURE in 28m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dc420d25f0aa48229c88801f130b7c97 : FAILURE in 49m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2474210f06744e55b213945ca3cec538 : FAILURE in 23m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ecb78efb16a94e81b6c15d7776c5c5bc : FAILURE in 26m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/424f62eb86464bb7ac744459ab1dfee4 : SUCCESS in 1h 54m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e1b80780c97949f89d4cf3942c670d97 : SUCCESS in 2h 38m 30s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2c2ebfd1085e434388b6c5531318f42c : FAILURE in 24m 37s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b0ed424cdbfa469fac0c36bccc26214e : FAILURE in 17m 48s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/831a10fa70b74b588313de4df1184562 : FAILURE in 1h 50m 34s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/db6ea7740dc348b6b69827522b094587 : SUCCESS in 2h 14m 18s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/33cf9694f5784ce4ae70b86b53681f81 : SUCCESS in 2h 47m 13s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/38f3980ac2244aa8b599d5e40287fe5f : FAILURE in 1h 47m 09s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d1256c420cbf43ab88ade382e7ee474d : SUCCESS in 2h 37m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2546afad074d77a463811dc69731fe1719e83278","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-26 12:01:04.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/720711/5 : FAILURE in 3h 05m 05s","accounts_in_message":[],"_revision_number":5},{"id":"c9a380dd334904d2ffe9713a9f19a7909cdd5182","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-27 17:56:30.000000000","message":"Patch Set 5:\n\nok, now we have ovsdbapp 1.3.0, let\u0027s go for it!","accounts_in_message":[],"_revision_number":5},{"id":"6157f5904e263e3d2874b944ff35ddfdaf07bff9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-27 17:56:34.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"a3565cfb11d3cf51e229aba61e587ae16781b9a7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-05-27 18:55:25.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ http://swswe-opsci001p.aristanetworks.com:8000/May-2020/Arista_Tempest_Testrefs/changes/11/720711/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"9ac299ce67d23f2a6f91b05854b7995221424dd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 20:32:41.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0b435b0145484bce99564a51d12e0054 : FAILURE in 25m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1c300dbda9574c41a63332e38e13b2ce : FAILURE in 25m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/cb355f4d1a4f44d1a061ee2f48c7d221 : FAILURE in 33m 00s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3a265dd1b68a4626872646a03810a054 : SUCCESS in 1h 27m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/083dead0c8ac4d8bbeea795a62cb126f : FAILURE in 25m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2ac168dbb8bb4fe1b0626bbe6e7096c3 : FAILURE in 25m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce34f8b6da4c465cadd939de6e98adb8 : FAILURE in 30m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0643163ed9f546b8a75957f04eac0a1c : FAILURE in 31m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f0cf300224402b9e772ab3c557ff96 : SUCCESS in 12m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3b398d7b9ac4e9f9ddb41ec97979a45 : FAILURE in 31m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98a2b613ced74731be6200b27b7705ae : FAILURE in 32m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/284b81160e304256a912664d72348112 : SUCCESS in 10m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/990ada87cc444c649004e1ecda057d3e : SUCCESS in 3m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/75d483752edd4cd2b09d3b83e5815c4b : SUCCESS in 49m 57s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b44e1178a3bb436dab112322bbe763d5 : SUCCESS in 1h 25m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/387b023aef0546ec98e86ff387f49cc1 : SUCCESS in 1h 16m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b604a3146e4e48ddb6c622640439de9c : FAILURE in 24m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/94a71be26f494e74b74f94df4f0fcd51 : FAILURE in 1h 04m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/62e1588306ff492aa85a793d4a63f2f2 : FAILURE in 56m 21s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b46facf8e77a400596a5c65255890cba : SUCCESS in 1h 17m 13s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/251e989bba4b4169957b7313065ee717 : FAILURE in 22m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ddbc4702cedd417497f398cfaff2ea6e : FAILURE in 27m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ae6ea0e6b84f4bdc8e99be6e9f5528a3 : FAILURE in 34m 20s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d59662eb31e84f9a929396d99678bc35 : FAILURE in 26m 06s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/969e8c0dd0fa4d29ac9b595556d79f6b : FAILURE in 22m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7d89f4d50a22421ea13030a7b0cdc62e : FAILURE in 26m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/73e270b2776e412bb0e7c36875dd4da4 : FAILURE in 28m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef6d91a3968940159dda53f5dbef37df : FAILURE in 30m 44s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/097c24d0d56a44d9bf32f793be22fdfb : SUCCESS in 2h 01m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4209cc1ebddf4b5db34887c7086a1961 : SUCCESS in 1h 43m 09s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6f00769cb82a491294717864618be437 : FAILURE in 22m 30s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8eef1c9d4a044009ad622aad01a2b0e7 : SUCCESS in 53m 39s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f4099fdce69c429e8818c0e9e65f897f : FAILURE in 1h 46m 17s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1a5dac61cda14d189323000f08cd0a9e : SUCCESS in 1h 53m 08s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/32fa1f97e2ae4195b89d178b6c6e9407 : SUCCESS in 2h 28m 08s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ec235c7b51b84ce882178259aa6d28f7 : FAILURE in 1h 53m 41s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/8063aad2979945938dfce9ffd78231eb : SUCCESS in 2h 10m 13s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d74b50916ce777e43a9898e4948f018e4c874e85","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-27 21:05:35.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/720711/6 : FAILURE in 3h 08m 28s","accounts_in_message":[],"_revision_number":6},{"id":"f4df50017eabf7e38f1204a0d044746b67f57855","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-28 13:36:10.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a8245a896c374103883973c94e1e9097881663b3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-28 15:20:45.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/720711/7 : SUCCESS in 1h 43m 57s","accounts_in_message":[],"_revision_number":7},{"id":"e540ed1d827cfd69696503aba921005335c7f8f0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-05-28 16:37:30.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/May-2020/Arista_Tempest_Testrefs/changes/11/720711/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"2e05cc2d92c9c074300062efaec1201d13bead3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 17:04:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8f7040c40dcb46bca866851d7484408e : SUCCESS in 57m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0b43f589b7ce4a86b00fcc7e6a2b736c : SUCCESS in 36m 46s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/e0f773e38721457ca2e59cbb37435ee6 : FAILURE in 1h 38m 49s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/319c3650a756437d9a6547379179738d : SUCCESS in 1h 06m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3ad2b807469540dd883c7e400591ca23 : SUCCESS in 41m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b2bcece6c1024ebcb48db4ae1091f897 : SUCCESS in 1h 12m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47b260b3a4bd4a2dae2e0078fdc05b03 : SUCCESS in 33m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cd8fb29412f4b28bfb3037a65d1b20f : SUCCESS in 26m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa44aab1b6e84c8b8682f6b73d7e9f57 : SUCCESS in 15m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de7c90ecb88d42a0aaaaf8d1de7a5803 : SUCCESS in 30m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05076529c7444344a32fb2d3107129f5 : SUCCESS in 32m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/038bbe3287b04343a68c88565c0bf0ff : SUCCESS in 9m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/de7c822c04ae4eddb126d7c636118d72 : SUCCESS in 5m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6fff7819e3914926a4e385f894fae766 : SUCCESS in 1h 17m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/41e955651bd14c76b52fec10113861a0 : SUCCESS in 57m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7a9abb7827d24540a405db22125ccf3b : SUCCESS in 1h 16m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5cb2940fa641455aa125eaf2b8ce6fc1 : SUCCESS in 1h 22m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b38f5072a780483e8973ae1ccf59dae3 : SUCCESS in 1h 44m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3940c9ad74de4aaf870083c5ee56e6f5 : SUCCESS in 1h 04m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3f70f95a9a1e4d83826a901a12079e1e : SUCCESS in 1h 13m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/bfe3877de5d14441a185f3b7b2f820e3 : SUCCESS in 1h 18m 55s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c13d6f614e114bdabb129d91a23b4d35 : SUCCESS in 1h 09m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4511111ca75041cdafe6ca5006f7a23a : SUCCESS in 1h 12m 14s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5d82ea4339d6483baa1d463039b75105 : SUCCESS in 1h 50m 26s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3e1cdca26c114fc9991ec4e661814c21 : FAILURE in 1h 17m 34s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/808c1b55cbbd406da6a7749ddd034216 : SUCCESS in 1h 07m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2e19fede55ca4700af8a3c87ddfe4ddf : SUCCESS in 51m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6c60cf3cce6043a2b3daf884293e5196 : SUCCESS in 2h 00m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/114a140e9d814fe4894b9dfef2d431cd : SUCCESS in 57m 33s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/09b07afc6c51489ea18cde839a0fdcb2 : SUCCESS in 1h 54m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/943ed4d342a6426bab0780ad47b9106d : SUCCESS in 1h 47m 13s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b190389d3e684dfc98cd2f428765f38c : SUCCESS in 1h 18m 06s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4b13119b51fb49b0bacfa5f509d98e23 : FAILURE in 1h 42m 38s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4104df1974d24e76ac166bcb441107db : SUCCESS in 1h 59m 28s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/db7ebc1b1a5f4116bdbaf9d0bf9353ba : SUCCESS in 2h 44m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fc1f92e01fcb4d73866d1f482d18e0d4 : FAILURE in 1h 56m 37s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/03fa6cf5183648c1a5f84952bb2aa979 : SUCCESS in 2h 11m 11s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a52c10836f88dbbc2f7dd9b762aeaf6b1bdcd0be","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-29 14:57:13.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"50f5269fce39788f56ab3b404720a022f9d73cc2","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-06-06 11:02:19.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-06-06/720711/7/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/98b3ce6 : EXCEPTION (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"64b77f1aa316d058bc4a11470e9b630f0b849f1b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-16 19:20:19.000000000","message":"Patch Set 7:\n\nping fellow reviewers","accounts_in_message":[],"_revision_number":7},{"id":"75ef822167bc3cdf3b2d85cdcb42395939ea13af","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-06-17 06:34:05.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c0e8bf557bc92964ec9810fce04f6306d8c5490a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-06 17:37:04.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3f65b1a34e613abac751c89eb2ffa0f89684f2fd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-06 17:40:07.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2f7570e609306b6373a257d59a98a8e23eab7a4b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-07-06 19:25:47.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/720711/8 : SUCCESS in 1h 44m 59s","accounts_in_message":[],"_revision_number":8},{"id":"c3b424c444d775072b625a09d6c65ef8e19bea00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 20:58:12.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9e2bf4bc8ae9476b895fd5469d39b8b4 : SUCCESS in 48m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1899df84c4c243dc946af45acf74330d : SUCCESS in 33m 44s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/beab10d69e774b0b93de2bfbf9ee8f10 : SUCCESS in 1h 07m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a9eafefcfa2a42f5a4905f54362c0ed0 : SUCCESS in 1h 06m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/48d390e8c4c24dc99422a391c8cb2abb : SUCCESS in 1h 21m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/00df298bc0214da1afc259c04b496bcb : SUCCESS in 1h 13m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a25930350d014602be47958c6f9e07b8 : SUCCESS in 38m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da2fb5f2bc2241fcba4ef80fb7797ed6 : SUCCESS in 34m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12dc3bc133174af58ee854f1c5466806 : FAILURE in 10m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c68ffc5ecb3c4513bf4ae141d16a5e38 : SUCCESS in 29m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00d74b205b9b46b1af9724305c1206b8 : SUCCESS in 27m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/891202de9f6c4f8e84311532aebaa0aa : SUCCESS in 12m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/23e3b1eee17745be91794ff82855acbe : SUCCESS in 4m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6ddbbaddedc6409493f11c0194e3de1e : SUCCESS in 1h 03m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7a6cfe3dd1a94f119810c452e7fc72a1 : SUCCESS in 43m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/85d4db881e6b407b96ba61af226e7b22 : SUCCESS in 1h 14m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/35790c4e8be34cf5bec47a35ede9a88f : SUCCESS in 1h 20m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/137d38332662456d989ac56fe15a7d1d : SUCCESS in 2h 02m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/31a7cb1f97ea470c848a03691623d55e : SUCCESS in 1h 14m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/72c2a1301aa64cdc87e300be8996e356 : SUCCESS in 1h 19m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c8fc5d45233d4db6a3d1e0c651adf84e : SUCCESS in 1h 09m 54s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8c7ca194afa94abe807a6f9e27b12331 : SUCCESS in 1h 10m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ac9a276b5c3a443fbcfbab1fc99137af : SUCCESS in 1h 02m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/98d6ad6208744e0c98e663f5ab889aaf : SUCCESS in 2h 05m 50s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2f07d723324a4ee08729bfb3fe8952eb : FAILURE in 1h 34m 02s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/81a9f24b5bb748e99f4656b8ba5661e6 : SUCCESS in 1h 07m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d0fed01d1f8545ec8eccf5c0fb329d25 : SUCCESS in 1h 09m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d5eb519167c043b0ad70b21b5848fd53 : SUCCESS in 2h 01m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/117b154561524c5dac5edd411c23a899 : SUCCESS in 1h 00m 20s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/6075215c622f411ebe75969e302586e5 : SUCCESS in 1h 22m 14s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d35b972b944b429d9ea727fe51353881 : SUCCESS in 2h 05m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/40975176039048888a5c7ac484032403 : SUCCESS in 1h 51m 48s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/eb768c4e8bce4b79ade2645770c9a743 : SUCCESS in 1h 18m 55s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/732ba60520934ac2b34809e84b732f6c : SUCCESS in 1h 50m 23s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d5c961c4b2d64864a027c91f8a28c43f : SUCCESS in 1h 43m 45s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b234175c3cd34f72bba43ae1a537d161 : SUCCESS in 2h 47m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5433a30ca3ab4fe596b08570db1e2461 : SUCCESS in 2h 04m 22s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/31d3f1bfbae5453091802aa074016041 : SUCCESS in 2h 17m 16s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0f4f47aad5fe826ddebe580a205ed8194633a9cc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-07 03:13:55.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/11/720711/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"1f171ca486173fc1d1538e3f94c509dfca09ef29","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-07 18:07:33.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"bc2fef4b9d291b524388d20bb0b549f01a94a756","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-07-07 20:02:51.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/720711/9 : FAILURE in 1h 54m 35s","accounts_in_message":[],"_revision_number":9},{"id":"2d61a7f8b2d324e04e707a9a8b4c98dd62971b6a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-07 22:07:37.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/11/720711/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"d797a894cc82c25f21b3d1dc6d109de9fbd2430c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 01:42:32.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f5b2ddaabdbd4627bd7fafc65d29dabb : SUCCESS in 57m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d723f1505e8940a4a0f3ec8ce9eb2f3b : SUCCESS in 33m 09s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5ef0b0b846424b00ae09715684aad3d0 : SUCCESS in 1h 06m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f7d12e90500f47a69e6a8d7fa9f46615 : SUCCESS in 1h 04m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d837fb89490d46e08d4fb4dec0d1d785 : SUCCESS in 1h 05m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/393f1a30ab4a4e9ba4a5e1d7b57b7366 : SUCCESS in 59m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4af8bb97cd6f44a2986d91a116feb35f : SUCCESS in 31m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3ae7da8a6024151b40206d7dc2ca48a : SUCCESS in 25m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef950a9ce55940f0af827d31df98f5eb : SUCCESS in 14m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/561d350caa274b109e9ea1dad662bb15 : SUCCESS in 28m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ccec9aafe4f43e2875a34ad4a0945a3 : SUCCESS in 26m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840869bce9144d7f8131af7570983667 : SUCCESS in 12m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8ea0b393fba54a2d82a0c9d1dc3f32a2 : SUCCESS in 4m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2c0a56e0bdac4a5d933204f5703582dd : SUCCESS in 56m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eab7f0332aca4fdfa9959b1770710e37 : SUCCESS in 52m 10s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9b626c3bc5a842388c221a65706892bf : SUCCESS in 1h 15m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d2d4fbc4dba444f3b4ae7587fbefca2d : SUCCESS in 1h 12m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2dc6616a28284dc8b958b66f021cb0f4 : SUCCESS in 2h 00m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/61a4297bbadc4c8581b7d2e336a6d105 : SUCCESS in 1h 07m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9cce401e6fa644ae9b622d8996d8f711 : SUCCESS in 1h 18m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a32ccef9819349238bf383a536232f76 : SUCCESS in 1h 03m 45s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5f0f07c55f884b93b2b85fdd2b1f9cce : SUCCESS in 1h 12m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2cae480bf27d4d1d908bf84a0256456a : SUCCESS in 1h 15m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/32ca216eff6c4a76bf56b493670567de : SUCCESS in 1h 36m 09s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/42fc9df440f14ef9aed76716653f5b52 : FAILURE in 1h 34m 42s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0282df4783a741da898d8f527415d65a : SUCCESS in 47m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/990413b3e69a47f997cb2d775f1c8459 : SUCCESS in 46m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/40381623c8d94f7fbfe695082f5cf2b8 : SUCCESS in 1h 53m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9282dfa9467043b2beb3b9842d5168de : SUCCESS in 54m 23s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/bd3ca3e785884b5dacfd3da76d131273 : SUCCESS in 1h 05m 11s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/05a12357b02549fcadbcec97cb6595af : SUCCESS in 2h 07m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e1e1f70005574c12a4672317eb4ca524 : SUCCESS in 1h 53m 09s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b60ccd81f0104c7fbc289f23c5e76eaa : SUCCESS in 1h 17m 57s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ed1a3df17c664c8e9142f692b7b05842 : FAILURE in 2h 18m 52s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4db10ba2985f45f5865280ca18d03c0b : SUCCESS in 2h 01m 34s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/dd0facf66044465c9f59d5d637ac44eb : SUCCESS in 1h 52m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/991aeb335d814f40973ecc929d60b6b1 : SUCCESS in 1h 36m 36s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/7edd9b4b3a554dd0adbea26a8f1c2e4c : TIMED_OUT in 3h 04m 49s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b3f4a8104e75b1446892cdd21cc82838074586e5","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-07-08 08:36:52.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"7560f77474a33d5fd83c26113bcb195ac4e29597","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-16 15:56:27.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"f03b49e53078e54be8457a23753303e1a62151b4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-16 19:10:04.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/11/720711/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"24d0ea82f56a77a9440523ae9d93f5428ef1e4e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 19:43:46.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/933a2b150a8d442f83c35b648d4f9071 : SUCCESS in 38m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ba6a10a263ed4adfa9c2e973a9317a76 : SUCCESS in 31m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f3ba9cac515b43acb040348f5455e3d9 : SUCCESS in 1h 04m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ee5a1c9bab8c4cdc8cd3c86368d52d6e : SUCCESS in 1h 20m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/789ccde3760041af98311090a1a6cb0e : SUCCESS in 1h 04m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/11ab2d12cdde4be988939edd9c64f7c3 : SUCCESS in 1h 07m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11cbbebfc9e14a409a0923f47ff23d76 : SUCCESS in 36m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45702d88a69b46af921cf36c8c13394c : SUCCESS in 26m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ced01012939414ea88811c95887f823 : SUCCESS in 8m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a638658635f5421b8a2e58e7835c860d : SUCCESS in 31m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/875e2ced9f33405982154c2236adcebe : SUCCESS in 29m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49d265eb4d0b4551a253b92f743a0c86 : SUCCESS in 11m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f36593daee8b4a7aa80ae849dd875105 : SUCCESS in 4m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/79805313aee942bf9219b1e83a9240d1 : SUCCESS in 59m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/67a81a852c2a44618786a888ae9f4968 : SUCCESS in 53m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2adb07e3ec4e42f28bc4c8e71ff2f391 : SUCCESS in 1h 12m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7900f91938a744d9a7b96fea3a71d5da : SUCCESS in 1h 05m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/08943fa240b94209bc5859304e40aa98 : SUCCESS in 1h 47m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7f4ff58f6c749f1bcc32f5116867d83 : SUCCESS in 1h 03m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b6259bbe1d154b0eb41fbf4d7f442c76 : SUCCESS in 1h 15m 57s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2591d8068e124b7e8953f91c07dbb9e8 : SUCCESS in 1h 07m 45s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c3dbd9430f74c14a45ceaf87796ee79 : SUCCESS in 1h 06m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0153e6b772b2421e9e2526af96ee21f3 : FAILURE in 1h 10m 39s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3dd8ee734b684a00bbd033f711983b32 : FAILURE in 1h 36m 28s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/09441621fc8945598724ba9fdc4e3b08 : SUCCESS in 1h 03m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1f2a95da9d3847b580599af03d2d5526 : SUCCESS in 1h 13m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f857ed6752b54aa398bdaad91429eaed : SUCCESS in 1h 58m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60503491e80a4fdd8f95c596fbe9fcff : SUCCESS in 1h 01m 04s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/cd6dcfcaf7ba4154922c1f908c1d095a : SUCCESS in 1h 05m 04s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0aa871a2fada444296385e8ec72b0a95 : SUCCESS in 1h 07m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e8e20c0dcbe543f09ce9a9570c09b1a1 : SUCCESS in 42m 45s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/931791a133e340ddb871d68f1f7dea5d : SUCCESS in 1h 19m 19s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/604166abd82b40eba09fbf7fbd1f344b : SUCCESS in 1h 35m 56s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b0131e1b7b234a679876510f79dfcfa8 : SUCCESS in 1h 56m 26s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9995da8fa5b44e869bbfbd1861e25654 : SUCCESS in 2h 28m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/60f3e8bb4bad41a2af0c24584b24c8c0 : SUCCESS in 1h 57m 24s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/889f4359ae114a9ebe875602e774a6c3 : SUCCESS in 1h 57m 21s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a39473b601777c0d9cfae572cb8a496e080fe186","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-07-21 07:25:37.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"a709f436e4965666361bb0b3b00e7f6ecfb563d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 10:13:38.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2d608e2032624abda1e555b01798d513 : SUCCESS in 49m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/57c7790ec8914671bfc2745f07b93f90 : SUCCESS in 42m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/94bee7c389bb47bea77b8ef60cf1a7d4 : SUCCESS in 1h 33m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/882e5d506f904f0d86f7d80196eace66 : SUCCESS in 41m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/389db13df1ed4d209a424d5813b2df7a : SUCCESS in 1h 04m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7ae635c116e54cfbbb73d5109f7bcfb7 : SUCCESS in 1h 04m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/003baaa615b34426918ee3e2b9611fda : SUCCESS in 32m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3dcc2f75068f45e89e65fea9a0ed6a75 : SUCCESS in 35m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/003a623fd16245f98b4290d3900e02ff : SUCCESS in 15m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00361071a27b4e4c8b89fc65b82c6e36 : SUCCESS in 33m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25ea242328d3491cbb5fbe7257d4ff3f : SUCCESS in 32m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b640e032e4c4bd7bf429f34ce14313a : SUCCESS in 18m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7139d9c788094defac4eabe3d1987efa : SUCCESS in 3m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b8aedd74c5914f2787df509745fc3cde : SUCCESS in 1h 03m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a79a0d0bbc734b63806568ddc0f40ad6 : SUCCESS in 1h 02m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a6eebd778b9641f7a8e898a18ad9dbf6 : SUCCESS in 1h 19m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/900837ab2584402bb0a05d0b7bff508b : SUCCESS in 1h 10m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8ef552946e5b423a8295f73f11984eac : SUCCESS in 1h 44m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e384c4f7e4f440a89149804a8032216 : SUCCESS in 1h 09m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/422dc3eca6a64756a658588587280ff5 : SUCCESS in 1h 15m 12s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d5433f21a0da461bb2309fe4c33bd4fb : SUCCESS in 1h 07m 41s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8971f960e5054107aeb354da1f392bca : SUCCESS in 1h 00m 39s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/98bd181d3c1e46c7a898b7949341cd99 : SUCCESS in 1h 44m 45s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/90b2d0014290470d98e656c14a03cbf3 : FAILURE in 1h 19m 53s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9b06e0720f3f4853846a3b5979ec12e6 : SUCCESS in 54m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0187970b3161470b8e1045c86b0aa76d : SUCCESS in 1h 25m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/63c2df69ccbc42459a44495e1e724a8a : SUCCESS in 2h 07m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfd710de80634675beac273c49013677 : SUCCESS in 54m 44s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/ec00c4b8479b4693aec63f096e324c11 : SUCCESS in 1h 32m 55s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8e3cd3da0ef348d78b0fe4b382c1c6bd : SUCCESS in 1h 11m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4a0020d2b9d440848ef58037e9e0842f : SUCCESS in 1h 01m 04s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/67e9f8c127f44e33a431a538c475525e : SUCCESS in 1h 15m 08s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e11390220f3744ef9d415e5a01094644 : SUCCESS in 1h 33m 21s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cd63fed2b70e4c9ab1f6b0e1d9923f35 : SUCCESS in 2h 01m 16s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fc9f29ab38d2494abea200e602651f51 : SUCCESS in 2h 30m 16s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1bb4f763b3a34db28ae358d94d192ed6 : SUCCESS in 1h 13m 59s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/08e0faa161014c66afde095564f6ee87 : SUCCESS in 2h 03m 48s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"40578663025449aa091334809cde208551134b84","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-22 10:21:15.000000000","message":"Patch Set 10:\n\nHi folks, can you take a look at this patch. Thanks!","accounts_in_message":[],"_revision_number":10},{"id":"f20ab708da57f42acb1579f74f615477a5144fef","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-23 21:23:13.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"291c6d5739e299de579757f6bede408c8a05a2e0","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-07-27 14:48:26.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"50d74e8e9d88ae5fb98c1a2b3532726a5fa3e65e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-27 17:43:02.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bf29d57eace9a6aabab0b2fb2c1cc75680055efb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-07-28 09:13:56.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"e92d148aaae1d6e69c51cf9d4d01292b35f09d49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 09:14:16.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"fe66ff9c2ab958790394c4aa8bc2e21327ef230c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 11:27:37.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/89467da2a8ba4c4fa4c9e4cf27221935 : SUCCESS in 56m 54s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a0a5c77315874ae38be8a5aa5ab40d9d : SUCCESS in 47m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c394ae56a77d4beaa358c93986acd0f0 : SUCCESS in 46m 06s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7bb8a1b64b564aeca59ecfd5fa52901b : SUCCESS in 2h 01m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8b88468c8e244b3faf852c4132d324ac : SUCCESS in 43m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7f1ac20a6f84b449f0b7357b4f342b8 : SUCCESS in 26m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e8c35d8e9e84843bebe9a524fa4d99c : SUCCESS in 11m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/334e0ca8214a4e1aa2ccf0cc8fc0a931 : SUCCESS in 27m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8882295914014bb7bfe86dbfe8cd418a : SUCCESS in 25m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7a405bf43d84fd496e85614c82ca113 : SUCCESS in 9m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/15cdbd50476f485f8454d955e1c3fde1 : SUCCESS in 3m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/de8b97d0ff9c42b79eb6b3baa7f50e74 : SUCCESS in 52m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0e6ff6ff2b664915b2fb4fc502f027f5 : SUCCESS in 1h 14m 16s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/00057842e9094e098825698d13a99384 : SUCCESS in 1h 31m 36s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f47223ce7a9e4bc1b16fe0f2fef486eb : SUCCESS in 1h 15m 14s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/127b63c830354dc5b18b47fde5c0046f : SUCCESS in 1h 31m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4abfb4cc8964b429ac7235983606471 : SUCCESS in 1h 10m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dc7e8962b6534bae88885175674c261b : SUCCESS in 1h 14m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bdee8aa35391434e92d760f5b26ea64d : SUCCESS in 2h 02m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5b300b1fa4f4ba89b4913d21cfe0eca : SUCCESS in 1h 00m 38s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5d8cccecb9e84b1cb6d0af1bef61471f : SUCCESS in 1h 15m 04s","accounts_in_message":[],"_revision_number":10},{"id":"3dbe6013ce5c1a20ad81018c7cdfd10238329066","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 11:27:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"1a565ffab650ba9f74b0ec89a3a4337f89d38f73","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 11:30:58.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a260ffbb63a544f8bb7c9f92d96a5586 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"de4c10882fc31f91c58f341d1e25cc769edac5af","revisions":{"e36805d9beee21ea0a7207db43f16ab6c22b4a3e":{"kind":"REWORK","_number":1,"created":"2020-04-17 13:33:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/720711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/720711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/720711/1"}}},"commit":{"parents":[{"commit":"8d85b356293d2a5f702523dfc0d21039abdd84fc","subject":"Add \"rbac-address-scope\" to OVN supported extensions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d85b356293d2a5f702523dfc0d21039abdd84fc"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 13:22:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 13:32:53.000000000","tz":0},"subject":"Set OVS manager command timeout and inactiviy probe","message":"Set OVS manager command timeout and inactiviy probe\n\nWhen creating the OVS manager, define the command timeout\n(CONF.OVS.ovsdb_timeout) and inactivity probe time\n(CONF.OVS.of_inactivity_probe)\n\nNOTE: CONF.OVS.of_inactivity_probe is defined in seconds but the\nparameter should be passed to ovs-vsctl in milliseconds [1].\n\n[1]http://www.openvswitch.org/support/dist-docs/ovs-vswitchd.conf.db.5.txt\n\nChange-Id: I8ed1fc85c2f78710bf6589ba3deca518471339b8\nCloses-Bug: #1868686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e36805d9beee21ea0a7207db43f16ab6c22b4a3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e36805d9beee21ea0a7207db43f16ab6c22b4a3e"}]},"branch":"refs/heads/master"},"b4930b11c0f1cbcf5f8571cd45ac7152dead8f5c":{"kind":"REWORK","_number":2,"created":"2020-04-17 17:30:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/720711/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/720711/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/720711/2"}}},"commit":{"parents":[{"commit":"8d85b356293d2a5f702523dfc0d21039abdd84fc","subject":"Add \"rbac-address-scope\" to OVN supported extensions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d85b356293d2a5f702523dfc0d21039abdd84fc"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 13:22:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 17:29:33.000000000","tz":0},"subject":"Set OVS manager command timeout and inactiviy probe","message":"Set OVS manager command timeout and inactiviy probe\n\nWhen creating the OVS manager, define the command timeout\n(CONF.OVS.ovsdb_timeout) and inactivity probe time\n(CONF.OVS.of_inactivity_probe)\n\nNOTE: CONF.OVS.of_inactivity_probe is defined in seconds but the\nparameter should be passed to ovs-vsctl in milliseconds [1].\n\n[1]http://www.openvswitch.org/support/dist-docs/ovs-vswitchd.conf.db.5.txt\n\nDepends-On: https://review.opendev.org/#/c/720785\n\nChange-Id: I8ed1fc85c2f78710bf6589ba3deca518471339b8\nCloses-Bug: #1868686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b4930b11c0f1cbcf5f8571cd45ac7152dead8f5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b4930b11c0f1cbcf5f8571cd45ac7152dead8f5c"}]},"branch":"refs/heads/master"},"421e62ab915c9d05bc1f7a845d1b13a8d6770f33":{"kind":"REWORK","_number":3,"created":"2020-04-18 09:54:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/720711/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/720711/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/720711/3"}}},"commit":{"parents":[{"commit":"8d85b356293d2a5f702523dfc0d21039abdd84fc","subject":"Add \"rbac-address-scope\" to OVN supported extensions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d85b356293d2a5f702523dfc0d21039abdd84fc"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 13:22:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-18 09:54:13.000000000","tz":0},"subject":"Set OVS manager command timeout and inactiviy probe","message":"Set OVS manager command timeout and inactiviy probe\n\nWhen creating the OVS manager, define the command timeout\n(CONF.OVS.ovsdb_timeout) and inactivity probe time\n(CONF.OVS.of_inactivity_probe)\n\nNOTE: CONF.OVS.of_inactivity_probe is defined in seconds but the\nparameter should be passed to ovs-vsctl in milliseconds [1].\n\n[1]http://www.openvswitch.org/support/dist-docs/ovs-vswitchd.conf.db.5.txt\n\nDepends-On: https://review.opendev.org/#/c/720785\n\nChange-Id: I8ed1fc85c2f78710bf6589ba3deca518471339b8\nCloses-Bug: #1868686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/421e62ab915c9d05bc1f7a845d1b13a8d6770f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/421e62ab915c9d05bc1f7a845d1b13a8d6770f33"}]},"branch":"refs/heads/master"},"e99abba84c7ac33fa9c0bf7d94fdb4199bc430b7":{"kind":"REWORK","_number":4,"created":"2020-04-23 16:28:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/720711/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/720711/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/720711/4"}}},"commit":{"parents":[{"commit":"8d85b356293d2a5f702523dfc0d21039abdd84fc","subject":"Add \"rbac-address-scope\" to OVN supported extensions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d85b356293d2a5f702523dfc0d21039abdd84fc"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 13:22:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-23 16:27:52.000000000","tz":0},"subject":"Set OVS manager command timeout and inactiviy probe","message":"Set OVS manager command timeout and inactiviy probe\n\nWhen creating the OVS manager, define the command timeout\n(CONF.OVS.ovsdb_timeout) and inactivity probe time\n(CONF.OVS.of_inactivity_probe)\n\nNOTE: CONF.OVS.of_inactivity_probe is defined in seconds but the\nparameter should be passed to ovs-vsctl in milliseconds [1].\n\n[1]http://www.openvswitch.org/support/dist-docs/ovs-vswitchd.conf.db.5.txt\n\nDepends-On: https://review.opendev.org/#/c/720785\n\nChange-Id: I8ed1fc85c2f78710bf6589ba3deca518471339b8\nCloses-Bug: #1868686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e99abba84c7ac33fa9c0bf7d94fdb4199bc430b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e99abba84c7ac33fa9c0bf7d94fdb4199bc430b7"}]},"branch":"refs/heads/master"},"5db70c8c4a9323af6cf5854c984db983a9f2e9e8":{"kind":"REWORK","_number":5,"created":"2020-05-26 08:55:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/720711/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/720711/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/720711/5"}}},"commit":{"parents":[{"commit":"4acc6843e849e98cd04a6d01861555c3e120f081","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4acc6843e849e98cd04a6d01861555c3e120f081"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 13:22:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-26 08:55:11.000000000","tz":0},"subject":"Set OVS manager command timeout and inactiviy probe","message":"Set OVS manager command timeout and inactiviy probe\n\nWhen creating the OVS manager, define the command timeout\n(CONF.OVS.ovsdb_timeout) and inactivity probe time\n(CONF.OVS.of_inactivity_probe)\n\nNOTE: CONF.OVS.of_inactivity_probe is defined in seconds but the\nparameter should be passed to ovs-vsctl in milliseconds [1].\n\n[1]http://www.openvswitch.org/support/dist-docs/ovs-vswitchd.conf.db.5.txt\n\nDepends-On: https://review.opendev.org/#/c/720785\n\nChange-Id: I8ed1fc85c2f78710bf6589ba3deca518471339b8\nCloses-Bug: #1868686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5db70c8c4a9323af6cf5854c984db983a9f2e9e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5db70c8c4a9323af6cf5854c984db983a9f2e9e8"}]},"branch":"refs/heads/master"},"4e67a00a1137a67ef1480f92c3940b7f1b5eb911":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-05-27 17:56:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/720711/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/720711/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/720711/6"}}},"commit":{"parents":[{"commit":"3f0fc7f9f267e5b9a5bbf3435e54fce78889cdd7","subject":"Merge \"[OVN] DOC Migration Strategy: Adjust formatting in code-block\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f0fc7f9f267e5b9a5bbf3435e54fce78889cdd7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 13:22:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-27 17:56:34.000000000","tz":0},"subject":"Set OVS manager command timeout and inactiviy probe","message":"Set OVS manager command timeout and inactiviy probe\n\nWhen creating the OVS manager, define the command timeout\n(CONF.OVS.ovsdb_timeout) and inactivity probe time\n(CONF.OVS.of_inactivity_probe)\n\nNOTE: CONF.OVS.of_inactivity_probe is defined in seconds but the\nparameter should be passed to ovs-vsctl in milliseconds [1].\n\n[1]http://www.openvswitch.org/support/dist-docs/ovs-vswitchd.conf.db.5.txt\n\nDepends-On: https://review.opendev.org/#/c/720785\n\nChange-Id: I8ed1fc85c2f78710bf6589ba3deca518471339b8\nCloses-Bug: #1868686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e67a00a1137a67ef1480f92c3940b7f1b5eb911"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e67a00a1137a67ef1480f92c3940b7f1b5eb911"}]},"branch":"refs/heads/master"},"b8cda2536d034987e527d124b349dd8a51a3d95c":{"kind":"REWORK","_number":7,"created":"2020-05-28 13:36:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/720711/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/720711/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/720711/7"}}},"commit":{"parents":[{"commit":"3f0fc7f9f267e5b9a5bbf3435e54fce78889cdd7","subject":"Merge \"[OVN] DOC Migration Strategy: Adjust formatting in code-block\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f0fc7f9f267e5b9a5bbf3435e54fce78889cdd7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 13:22:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-28 13:35:47.000000000","tz":0},"subject":"Set OVS manager command timeout and inactiviy probe","message":"Set OVS manager command timeout and inactiviy probe\n\nWhen creating the OVS manager, define the command timeout\n(CONF.OVS.ovsdb_timeout) and inactivity probe time\n(CONF.OVS.of_inactivity_probe)\n\nNOTE: CONF.OVS.of_inactivity_probe is defined in seconds but the\nparameter should be passed to ovs-vsctl in milliseconds [1].\n\n[1]http://www.openvswitch.org/support/dist-docs/ovs-vswitchd.conf.db.5.txt\n\nDepends-On: https://review.opendev.org/#/c/720785\n\nChange-Id: I8ed1fc85c2f78710bf6589ba3deca518471339b8\nCloses-Bug: #1868686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8cda2536d034987e527d124b349dd8a51a3d95c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8cda2536d034987e527d124b349dd8a51a3d95c"}]},"branch":"refs/heads/master"},"0b1fc6c4181ac1fef79d93e1cfd64a0583cd36a3":{"kind":"REWORK","_number":8,"created":"2020-07-06 17:40:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/720711/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/720711/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/720711/8"}}},"commit":{"parents":[{"commit":"3f0fc7f9f267e5b9a5bbf3435e54fce78889cdd7","subject":"Merge \"[OVN] DOC Migration Strategy: Adjust formatting in code-block\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f0fc7f9f267e5b9a5bbf3435e54fce78889cdd7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 13:22:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-06 17:39:48.000000000","tz":0},"subject":"Set OVS manager command timeout and inactiviy probe","message":"Set OVS manager command timeout and inactiviy probe\n\nWhen creating the OVS manager, define the command timeout\n(CONF.OVS.ovsdb_timeout) and inactivity probe time\n(CONF.OVS.of_inactivity_probe)\n\nNOTE: CONF.OVS.of_inactivity_probe is defined in seconds but the\nparameter should be passed to ovs-vsctl in milliseconds [1].\n\n[1]http://www.openvswitch.org/support/dist-docs/ovs-vswitchd.conf.db.5.txt\n\nDepends-On: https://review.opendev.org/#/c/720785\n\nChange-Id: I8ed1fc85c2f78710bf6589ba3deca518471339b8\nCloses-Bug: #1868686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b1fc6c4181ac1fef79d93e1cfd64a0583cd36a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b1fc6c4181ac1fef79d93e1cfd64a0583cd36a3"}]},"branch":"refs/heads/master"},"a4b2a3756499cf27a2a6cc4fccfff4adf2bf26f9":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-07-07 18:07:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/720711/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/720711/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/720711/9"}}},"commit":{"parents":[{"commit":"75970c4cda175d246b3519171e8cd96da2c65776","subject":"Fix pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/75970c4cda175d246b3519171e8cd96da2c65776"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 13:22:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-07 18:07:33.000000000","tz":0},"subject":"Set OVS manager command timeout and inactiviy probe","message":"Set OVS manager command timeout and inactiviy probe\n\nWhen creating the OVS manager, define the command timeout\n(CONF.OVS.ovsdb_timeout) and inactivity probe time\n(CONF.OVS.of_inactivity_probe)\n\nNOTE: CONF.OVS.of_inactivity_probe is defined in seconds but the\nparameter should be passed to ovs-vsctl in milliseconds [1].\n\n[1]http://www.openvswitch.org/support/dist-docs/ovs-vswitchd.conf.db.5.txt\n\nDepends-On: https://review.opendev.org/#/c/720785\n\nChange-Id: I8ed1fc85c2f78710bf6589ba3deca518471339b8\nCloses-Bug: #1868686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4b2a3756499cf27a2a6cc4fccfff4adf2bf26f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4b2a3756499cf27a2a6cc4fccfff4adf2bf26f9"}]},"branch":"refs/heads/master"},"de4c10882fc31f91c58f341d1e25cc769edac5af":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-07-16 15:56:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/720711/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/720711/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/720711/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/720711/10"}}},"commit":{"parents":[{"commit":"d28654a0c740844ab726d557480ebb3fd3ac66c7","subject":"Merge \"Remove tempest-integrated-networking job from neutron queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d28654a0c740844ab726d557480ebb3fd3ac66c7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 13:22:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-16 15:56:27.000000000","tz":0},"subject":"Set OVS manager command timeout and inactiviy probe","message":"Set OVS manager command timeout and inactiviy probe\n\nWhen creating the OVS manager, define the command timeout\n(CONF.OVS.ovsdb_timeout) and inactivity probe time\n(CONF.OVS.of_inactivity_probe)\n\nNOTE: CONF.OVS.of_inactivity_probe is defined in seconds but the\nparameter should be passed to ovs-vsctl in milliseconds [1].\n\n[1]http://www.openvswitch.org/support/dist-docs/ovs-vswitchd.conf.db.5.txt\n\nDepends-On: https://review.opendev.org/#/c/720785\n\nChange-Id: I8ed1fc85c2f78710bf6589ba3deca518471339b8\nCloses-Bug: #1868686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de4c10882fc31f91c58f341d1e25cc769edac5af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de4c10882fc31f91c58f341d1e25cc769edac5af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
