)]}'
{"id":"openstack%2Fneutron~722215","triplet_id":"openstack%2Fneutron~stable%2Fqueens~Ic16cf83fe1d1576d91047adb2d4f9e07d57185b6","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1874211-stable/train-stable/stein-stable/rocky-stable/queens","hashtags":[],"change_id":"Ic16cf83fe1d1576d91047adb2d4f9e07d57185b6","subject":"[L3 HA] Add \"no_track\" option to VIPs in keepalived config","status":"ABANDONED","created":"2020-04-23 07:15:27.000000000","updated":"2020-04-25 08:08:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"84a0e8fbcefc4bfe23894aa007e5a49c80135f54","_number":722215,"virtual_id_number":722215,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"date":"2020-04-23 15:16:04.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"date":"2020-04-25 08:08:41.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-23 09:40:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-04-23 15:24:14.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2020-04-24 17:48:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-04-23 22:20:24.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 09:40:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 15:16:04.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-04-23 15:24:14.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-04-23 22:20:24.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-24 17:48:41.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"3594ad151ce8fef08c8a4b2c9450df8959b68d61","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-23 07:15:27.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"c6298fee554f3264eaa85fd3c3a0c78df1a595a8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-23 07:24:31.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/722215/1 : FAILURE in 8m 31s","accounts_in_message":[],"_revision_number":1},{"id":"6663706b987d938e8ae44a76dfcc966459032741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 09:40:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/3cd1abaf893246a7a6db08d6a404faba : SUCCESS in 45m 09s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/31940ed450a2402c889ba83856dc46fb : SUCCESS in 40m 22s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/2f88c7be247b429bb60eeaa5b8399dbf : SUCCESS in 1h 39m 20s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/cfd0b7ccdc4c4ee897e8ef309a8acfc8 : SUCCESS in 49m 35s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/720f8da6a6c44848b43c237f9345baf8 : SUCCESS in 49m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f29a11d2e65f4676b1a5cff846020f75 : SUCCESS in 11m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9843bea0700e4d5c9047b2ef317a96b6 : SUCCESS in 22m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6acc4c8f495e4f9bb802ed7dedd8f201 : SUCCESS in 28m 00s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/69c29027887f44cfa9d3cdebaa643928 : SUCCESS in 6m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bc3ed286d47f409ba71ad05703389b55 : SUCCESS in 1h 19m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9a4ff7dc3f954dd99fb45f285dc0a410 : SUCCESS in 53m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17b80edd47bc423aacebc7e59d74d4bb : SUCCESS in 1h 16m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5b733fbcc5e14e278bae6681c66bbde4 : FAILURE in 1h 01m 38s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f8cb33b06f6144bf9af18128cc29fb7c : SUCCESS in 54m 01s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/feccb1204eb749b5917ca322fad561ee : SUCCESS in 1h 55m 44s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f2b33e0a5adb4d539685708ac265dd4d : SUCCESS in 1h 45m 12s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b378392bd5bf45409588041721903968 : SUCCESS in 1h 17m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb18e67b4acc45a0bb439ea76d6aee67 : SUCCESS in 1h 04m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ba3cb0ac09ed45feba220e718283c5d0 : FAILURE in 47m 55s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/77f7c6f377834fd69d64524dfe826d2e : SUCCESS in 1h 57m 17s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/78f0587a366f43f38fbe6d8f87a571cf : SUCCESS in 1h 26m 25s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/07628e7a27144f26af98ea50b2900b40 : SUCCESS in 1h 17m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89a44156c28b4037ac04e86e6b60beed : SUCCESS in 43m 23s","accounts_in_message":[],"_revision_number":1},{"id":"efa3fe75cfd91f0bc55b835c313bc6dba7d8d593","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-23 10:05:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3d332132f9d2cd3386299618c6543221b7821909","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-23 10:13:53.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/722215/1 : FAILURE in 7m 42s","accounts_in_message":[],"_revision_number":1},{"id":"47dba6cf0d47dfcd0c715bd76477da1dbb2275a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 13:19:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/528196202cc44d9e932ceef40f0866c9 : SUCCESS in 52m 35s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/4b0a2aeefde948859754ddd0f7ad7158 : SUCCESS in 36m 59s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/1c9ea0e9c08948758777b9f8ddf08688 : FAILURE in 1h 27m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/5af610b82d7e4d83a2ad07c11df303d7 : SUCCESS in 48m 39s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/4d9d61806f994d8f9fd7023113fc1025 : SUCCESS in 46m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba482cff484a44199a0f671c13619aa5 : SUCCESS in 12m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a92c544117ca422aabd3f606e1226355 : SUCCESS in 30m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d27069976b464b9d8bbade1fc8686d50 : SUCCESS in 30m 37s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1a82156c78054672a9cf71289cda7985 : SUCCESS in 7m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0bc3abc4942446e190b0ee59a772595e : SUCCESS in 1h 18m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e3e4b6ca8ede4cc5a00849c5ad0daefd : SUCCESS in 53m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9f8419621f84626918ddb9a78cc42ee : SUCCESS in 1h 23m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/50394c477c1b41768f573db281469a0f : FAILURE in 53m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/76a4fdd1230143b795652bedc9f0a728 : SUCCESS in 50m 15s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/6fd3c5fece2547c3802fe8f6fce82ab9 : SUCCESS in 1h 44m 44s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/9d309c417a9a412982d21ba67c2831d7 : SUCCESS in 1h 32m 17s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3a2f5d863e2d4b0791f9a4a4b5069d49 : SUCCESS in 1h 38m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/995e4af4d57f4be5a0efaed60c3b7601 : SUCCESS in 1h 14m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d4789f07591b46b8a626e1981c59fdf8 : SUCCESS in 1h 09m 46s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0a469a3f0cde4b97b9aec26201ebc521 : SUCCESS in 1h 27m 05s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5dba9c023538466da6acdd7952dc9393 : SUCCESS in 1h 41m 48s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/7950605413424e548c305fa3c0ba5cf9 : SUCCESS in 1h 15m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ee8a9834775416eacf8667109da0250 : SUCCESS in 28m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f69dd92bac379c41eea3f3f69ff09374f2ffb918","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-23 15:16:04.000000000","message":"Patch Set 1:\n\nrecheck quite a few functional timeout failures","accounts_in_message":[],"_revision_number":1},{"id":"fdb7d63d936e9821125b96406cc009de6a92ed7a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-04-23 15:24:14.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/722215/1 : FAILURE in 7m 31s","accounts_in_message":[],"_revision_number":1},{"id":"42842d5bfe89912a5a8c76a849e61ca0d3acfec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 18:21:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/07757f48c5d14a178109f5d2ce0222ef : SUCCESS in 47m 27s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/33bdf5b47e964d6fb60c37a864cf126f : SUCCESS in 39m 53s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/095cde42b690445a8062b7a2fc480f74 : FAILURE in 1h 25m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/bd01f249d36142f8a7bc5a9fb3640d24 : SUCCESS in 54m 56s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/7672b032e2044dc6b6e401820917967e : SUCCESS in 53m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96b643d1fa614f20860cde05bb47ff5b : SUCCESS in 13m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/91092eee7ed44ad8a9e550292f8c1eb1 : SUCCESS in 25m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f325e0cbbd4d4ee892b80eeed1ada36a : SUCCESS in 23m 07s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/68b52e24825140f5996653bcba57f97a : SUCCESS in 7m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/98ddf2eb04db42bb864e9e3cdfe6755e : SUCCESS in 1h 10m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c54fe60aed264ad48bdce203b0c0d7fd : SUCCESS in 55m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f423cf782b74ebf84fc9ac4984461a5 : SUCCESS in 1h 24m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c416041bb6ae41af985f907df1bc09eb : FAILURE in 39m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a6c02384ce1540ae93f4dd32b30b103e : FAILURE in 45m 02s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/e54d2a91bf9b4969b5c5de6234179361 : SUCCESS in 1h 34m 13s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ada200bd737441b0b795f28c2cc61c7c : SUCCESS in 1h 20m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/109238ff811044ac865ab2cd922eb871 : FAILURE in 1h 14m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e1bf8d5c459f47e3bda5ff5179067d99 : SUCCESS in 1h 02m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6bd5146501cd4501804b97cac4438f69 : SUCCESS in 1h 16m 27s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5738615e3089456289214bfbfa41035c : SUCCESS in 1h 37m 32s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/121cb91cbf134393b8a7a0825a74cf19 : SUCCESS in 1h 36m 25s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/f9ce61dfb08849ac8734608ee655aeb8 : SUCCESS in 1h 24m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7ba64e47d6b44c6aab695bc0ec8f494 : SUCCESS in 37m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ec511f9e04370057c0af9204023283d9f15b9c20","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-24 17:48:41.000000000","message":"Patch Set 1: Code-Review-1\n\nkeepalived is failing [1] with error:\nApr 23 11:47:01 ubuntu-xenial-ovh-bhs1-0016131479 kernel: keepalived[5967]: segfault at 0 ip 000000000041bd32 sp 00007fffaa6dfdc0 error 4 in keepalived[400000+42000]\n\nWe should check if the installed version of keepalived in Queens is not accepting this new parameter. In Queens:\n2020-04-23 16:48:40.021801 | primary | Preparing to unpack .../keepalived_1%3a1.2.24-1ubuntu0.16.04.2_amd64.deb ...\n2020-04-23 16:48:40.028247 | primary | Unpacking keepalived (1:1.2.24-1ubuntu0.16.04.2) ...\n\nIn Rocky:\n2020-04-23 22:42:05.658551 | primary | Preparing to unpack .../keepalived_1%3a1.3.9-1ubuntu0.18.04.2~cloud1_amd64.deb ...\n2020-04-23 22:42:05.849164 | primary | Unpacking keepalived (1:1.3.9-1ubuntu0.18.04.2~cloud1) ...\n\n\n[1]https://e09ab00f519fe294009a-f0730a4947ec1834d620bd3aaae7b9e9.ssl.cf2.rackcdn.com/722215/1/check/neutron-functional/50394c4/logs/syslog.txt","accounts_in_message":[],"_revision_number":1},{"id":"0fe2ca4b754a579dfcd47bd22802aad25d0487cd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-25 08:08:41.000000000","message":"Patch Set 1:\n\n\u003e keepalived is failing [1] with error:\n \u003e Apr 23 11:47:01 ubuntu-xenial-ovh-bhs1-0016131479 kernel:\n \u003e keepalived[5967]: segfault at 0 ip 000000000041bd32 sp\n \u003e 00007fffaa6dfdc0 error 4 in keepalived[400000+42000]\n \u003e \n \u003e We should check if the installed version of keepalived in Queens is\n \u003e not accepting this new parameter. In Queens:\n \u003e 2020-04-23 16:48:40.021801 | primary | Preparing to unpack\n \u003e .../keepalived_1%3a1.2.24-1ubuntu0.16.04.2_amd64.deb ...\n \u003e 2020-04-23 16:48:40.028247 | primary | Unpacking keepalived\n \u003e (1:1.2.24-1ubuntu0.16.04.2) ...\n \u003e \n \u003e In Rocky:\n \u003e 2020-04-23 22:42:05.658551 | primary | Preparing to unpack\n \u003e .../keepalived_1%3a1.3.9-1ubuntu0.18.04.2~cloud1_amd64.deb ...\n \u003e 2020-04-23 22:42:05.849164 | primary | Unpacking keepalived\n \u003e (1:1.3.9-1ubuntu0.18.04.2~cloud1) ...\n \u003e \n \u003e \n \u003e [1]https://e09ab00f519fe294009a-f0730a4947ec1834d620bd3aaae7b9e9.ssl.cf2.rackcdn.com/722215/1/check/neutron-functional/50394c4/logs/syslog.txt\n\nHa, I was checking it with 1.3.9 on Ubuntu Bionic. Queens is still running on Xenial and I never tested it there. But I don\u0027t really think we need this patch on such old release so lets abandon this cherry-pick. Thx Rodolfo for checking this.","accounts_in_message":[],"_revision_number":1},{"id":"bdc54765184bc21f98645cf1beef2190491ef6fa","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-25 08:08:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b59281020a30cabae15afca5aca2e49717a7a4e","revisions":{"5b59281020a30cabae15afca5aca2e49717a7a4e":{"kind":"REWORK","_number":1,"created":"2020-04-23 07:15:27.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/15/722215/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/722215/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/722215/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/722215/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/722215/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/722215/1"}}},"commit":{"parents":[{"commit":"cf9bd62d4d5e674d555f3f5dc88dfed3275e63f3","subject":"Merge \"Refactor the L3 agent batch notifier\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf9bd62d4d5e674d555f3f5dc88dfed3275e63f3"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-04-22 09:56:40.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-04-23 07:15:27.000000000","tz":0},"subject":"[L3 HA] Add \"no_track\" option to VIPs in keepalived config","message":"[L3 HA] Add \"no_track\" option to VIPs in keepalived config\n\nPatch [1] introduced new mechanism which only brings UP interfaces\non master node of HA router. It works fine with keepalived 1.x\nbut it is broken when keepalived 2.x was used (e.g. on Centos 8) as\nin this new version of keepalived by default all interfaces of VIPs\nand routes are tracked, and if one of them is DOWN, keepalived is\ngoing to FAULT state. Because of that router will never be\ntransitioned to MASTER on any node.\n\nThis patch fixes it by adding \"no_track\" option to all VIPs\nand routes in keepalived\u0027s config file.\n\nThis \"no_track\" option isn\u0027t added to ha interface so this one\nis still tracked by keepalived.\n\n[1] https://review.opendev.org/#/c/707406/\n\nCloses-bug: #1874211\n\nChange-Id: Ic16cf83fe1d1576d91047adb2d4f9e07d57185b6\n(cherry picked from commit dc9084a8ec9db08d7ea947e0b73581b748be5819)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b59281020a30cabae15afca5aca2e49717a7a4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b59281020a30cabae15afca5aca2e49717a7a4e"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
