)]}'
{"id":"openstack%2Fneutron~726255","triplet_id":"openstack%2Fneutron~master~I7e4cd88281fb573ca7f91068f1645a474c6be7c3","project":"openstack/neutron","branch":"master","topic":"fix-healthcheck","hashtags":[],"change_id":"I7e4cd88281fb573ca7f91068f1645a474c6be7c3","subject":"WIP: Replace depreacted filter:healthcheck by app:healthcheck","status":"ABANDONED","created":"2020-05-07 22:21:12.000000000","updated":"2020-09-25 15:42:26.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3d00d97f51ef46e58daa549f7b7d16ad83cdd966","_number":726255,"virtual_id_number":726255,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2020-05-08 20:17:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"date":"2020-09-25 15:42:26.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"date":"2020-05-08 16:28:30.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-08 19:01:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-05-08 16:19:35.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-08 16:19:35.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-05-08 16:28:30.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-08 19:01:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-08 20:17:56.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-25 15:42:26.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"ed37218a2ea89eb1af203720e107517a22e49fd4","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-07 22:21:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3fc82e5fa8f287af3b4d7dbb2602d963575309e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-05-07 22:29:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-05-07/726255/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/4bdda43 : FAILURE in 6m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0346a59b5261a4f92e6863a626530dde83f2f7d1","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-07 23:33:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"06190cccbe4816acc853aaafc26a58bfcce7fedf","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-05-07 23:40:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-05-07/726255/2/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/a66f76b : FAILURE in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c3a6b93b71642c28228554e6a8d887f6ca5ff46b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-05-08 00:47:18.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/May-2020/Arista_Tempest_Testrefs/changes/55/726255/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"cd05115a2467ac9c84025cdadb65b540de37a17b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-05-08 00:56:59.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/May-2020/Arista_Tempest_Testrefs/changes/55/726255/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"e9820642edc388cf983cc5e08422d401e875ea2a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-08 01:16:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/726255/2 : SUCCESS in 1h 42m 25s","accounts_in_message":[],"_revision_number":2},{"id":"1ffb2d571f94fb0b5e9af30acf010d2e14195c33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 02:43:20.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4fbee64a1e944acaac45511b5b04e0f0 : SUCCESS in 1h 00m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/97ba6304514742fda1d333da4664be34 : SUCCESS in 39m 58s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/19dca82223484698918c917c7d5610e0 : FAILURE in 2h 10m 04s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a5024a4e3f25407484d901d340995a21 : SUCCESS in 1h 15m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/65d934a44f1c4cf89aeaaf185dad54a7 : SUCCESS in 1h 06m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2324ce1cb6614823b9097e2853961581 : SUCCESS in 1h 04m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b4026ff968d417692e384541a59b448 : SUCCESS in 34m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1348df4a1d6483e94e71d6844c94aa0 : SUCCESS in 27m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d797ebe64a2448a48dd6ad92eca0ef2e : FAILURE in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27e55849500f4f0c8f3055d8a2f0b031 : SUCCESS in 27m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba344b9099204c63b4fa5e69b17914a9 : SUCCESS in 29m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d68005c530b4511907b987769522a0b : SUCCESS in 8m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/10c03cac8e6040d19fd7cb4a94416e8b : FAILURE in 49m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/db3df6adbe164f729764474987fa8104 : SUCCESS in 1h 11m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3f88beedb114130ad9eccf8c0c80084 : SUCCESS in 1h 08m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/eeeebac484264d88bfeb827999abf0d0 : SUCCESS in 1h 52m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb2dbe4239394c519334f4ae66732330 : SUCCESS in 1h 02m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4b5ca863d180414d9040467378e38ab9 : SUCCESS in 1h 50m 40s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/28d98fc1f46b4312b38ddbf92f7e66dd : FAILURE in 1h 33m 49s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4256b35267c04067849feec7fc203c31 : SUCCESS in 1h 12m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e16a9ff0611f458daec97616d655d073 : SUCCESS in 1h 10m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1ac0cbd3d6754b3986657506b4263aec : SUCCESS in 1h 34m 40s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1677435d02c74ece8585e94956dcb59b : SUCCESS in 1h 35m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bd82404cf152404fa8ed5c20a153851a : SUCCESS in 1h 16m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dfb6016fe369477783f3f1926af139f8 : SUCCESS in 52m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7ed734bcb83649a7affd28f61f620a69 : SUCCESS in 2h 35m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45309552b39b4c5aa397970d39bc6fd2 : SUCCESS in 1h 31m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9131288ed1aa4e14bfb4768144c80dde : SUCCESS in 1h 51m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a3aa3f13e784463e84d56758ae87525b : SUCCESS in 2h 18m 36s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f4d1644349bf439aaee6b5ad51b1fb7c : SUCCESS in 1h 23m 47s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6818b88683054e34948ba72ae8054e60 : FAILURE in 57m 01s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0207e2a7a3364efa90aa7c295872907e : FAILURE in 2h 01m 13s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d7be81c951a144cb8dc93f336619c95d : FAILURE in 2h 06m 54s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/cd92e5285d7040188d3d763e1ee3edde : SUCCESS in 2h 37m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/43bdd2d6c0354f56a082664d9a4877c7 : TIMED_OUT in 3h 03m 43s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/41a452f7fe61437e972112161beacf4c : TIMED_OUT in 3h 04m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b0155c633f685171aac3df16c6dbc0c8e8cd2c29","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-08 09:58:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"da7612e7d54ecbd8a03f21be3450df61228c8e18","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-05-08 10:05:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-05-08/726255/3/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/4662501 : FAILURE in 6m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"48fcc91405e32402f775f227c49d84c5591ac48f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-05-08 11:10:48.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/May-2020/Arista_Tempest_Testrefs/changes/55/726255/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"05fb7bb16f829e0cbad000d3570a687f0fd2ebb0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-08 11:44:54.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/726255/3 : FAILURE in 1h 46m 09s","accounts_in_message":[],"_revision_number":3},{"id":"5735ac31e03cc7f97e6758d7fa2a1802d8ead253","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 13:07:27.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f568888cdc6f457dbb0480fca821acb0 : SUCCESS in 1h 07m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/318cc4dc7eb147168afe936be2fd64d1 : SUCCESS in 33m 02s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/18651dd82202411fb4a0620dd39ecf32 : FAILURE in 2h 09m 00s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cb7719b74d92496986b4ed112af1e714 : SUCCESS in 1h 42m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ca8b4a6958d94274889cdcea0c03b0cf : SUCCESS in 59m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a9aa1e29a231442e9fc4e30c201324e1 : SUCCESS in 1h 05m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99e623507a6e4797bf33e3a9100c4c2c : SUCCESS in 55m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/928d24a2f6614715ae5b6b61c6ca5afd : SUCCESS in 27m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aab3dbbc463b4a08872e04585d0e6f1f : FAILURE in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bc13c6a90064b9ca2e80612cace4658 : SUCCESS in 45m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8f4d4b6b01e47bc97b39d92a38845ca : SUCCESS in 44m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c836c94a1c5e4c3fb0c18fe9e314d9d0 : SUCCESS in 18m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1794da295f9a45d7b62ccdd335b8e07b : SUCCESS in 1h 19m 29s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6db3747f0c694e2aa67fdde0f1cbde61 : FAILURE in 56m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ce97932604a044ee8d68a7822ed36c87 : FAILURE in 1h 22m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/04fca249d57d4415ad721a54718ee28f : SUCCESS in 2h 11m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b778db0e697642e3ae6c16b4df3d8cfb : SUCCESS in 1h 05m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/73d8645e9a504864a835aa567e423749 : SUCCESS in 1h 05m 22s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/539c01ee3b954e4499c810c5b011a92a : SUCCESS in 1h 14m 28s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/98063022a46747ccabe2504b65918a2f : SUCCESS in 1h 13m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/99bcaba42ff949ee973f9bcea3b3a898 : SUCCESS in 1h 18m 28s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bf6f629f96b24aa3837df485609a85b4 : SUCCESS in 1h 44m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c92b1c8c481a4549aee994517a2320c4 : SUCCESS in 1h 29m 24s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/531f42e5281a4511a90f4f8bdcd7ea36 : SUCCESS in 1h 24m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d9851c795e4b4ae999cb5fcdc1e496af : SUCCESS in 1h 00m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f80e665fb827448a86ec580f9648e760 : SUCCESS in 2h 18m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09a8b9331dec4cc391977b4c1ddb5f4f : SUCCESS in 1h 06m 03s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/db5e97b9598347639dd22ddad3e5bff8 : FAILURE in 1h 44m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c377f7b838b5410bb5ccf8b3f28adf36 : SUCCESS in 2h 01m 34s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/af22854498cf453c81629758586f59d6 : SUCCESS in 1h 19m 59s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9165ec6ca55f4b95af8ed26bf1e7529a : SUCCESS in 50m 50s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a43a9ad84c2543888c01f3a910178069 : SUCCESS in 1h 45m 02s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7383adba2bc34bb7bed93dace552ae49 : FAILURE in 1h 52m 57s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8b30f360542046159d84852cd07a2949 : SUCCESS in 2h 38m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/dbb867e18a5746299da033e8148b88ce : TIMED_OUT in 3h 03m 36s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/8f4391b1419045c1b9e28f41430b3470 : TIMED_OUT in 3h 05m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a9d53979cb4afe9cc145aabc34c461ed3281d312","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-08 13:46:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3323d7e6b2c6b7db4781b0eab8903666519c33f7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-08 13:52:50.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/726255/4 : FAILURE in 6m 10s","accounts_in_message":[],"_revision_number":4},{"id":"edb9f9421dd7a613c964eb50b5dba2ea79faaae4","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-05-08 13:53:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-05-08/726255/4/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/06404d1 : FAILURE in 6m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"aca9eaa6e508f1e5b56d76f3602cac1d4671fce6","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-08 16:17:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2791305663e6b903d69272ff9649ebd7992f7e29","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-08 16:19:35.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/726255/5 : FAILURE in 1m 30s","accounts_in_message":[],"_revision_number":5},{"id":"02a05493112ad2dce4532a480e4abb662c3df9d4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-05-08 16:23:25.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/May-2020/Arista_Tempest_Testrefs/changes/55/726255/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"9a14b45501ca0960f05ab2c114fa7b9b1c9eb322","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-05-08 16:28:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-05-08/726255/5/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/411ef7e : FAILURE in 6m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a767df90a4081ce10e3e8eeb367c453a3d856de2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 19:01:58.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9d7b4b2e16eb4016b918b805381995e0 : SUCCESS in 1h 04m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8fbdfc3cb6c241458e079e73a2c28e4f : SUCCESS in 38m 35s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/e79563cb062740aa8a82512621ed0922 : FAILURE in 2h 33m 03s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/568904612a164ac38bc9680d55be627f : SUCCESS in 1h 08m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/15b8f5f6bd9c4582b87263bfcf510aec : SUCCESS in 1h 07m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/23556fe72f194111ae46442d9bb7e16a : SUCCESS in 1h 15m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/472acbb77bf44241b9e9bcf7ad8bdc15 : SUCCESS in 32m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c5c25fa53b647a08524f1da9246abcc : SUCCESS in 28m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf2e0761a7a44be2840c676827d365b6 : FAILURE in 8m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc6210f4530c48cf96481282c348a38e : SUCCESS in 24m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bef72b1d2b6464b9818275696c613f1 : SUCCESS in 23m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0f20c33ba2f48f2b7e26b746737725f : SUCCESS in 10m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/df7e45498b3e44f988399f28fce93d90 : SUCCESS in 49m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/daecc97cc8c8428aa2ffe4332beb720d : FAILURE in 56m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9036fac3bad64704a1eccb5930d7d1ae : FAILURE in 1h 27m 21s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6dec1fb5a5e84ce893dcc60700c6e130 : SUCCESS in 1h 54m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/22ab89329e004ad5a41225a1b7061904 : SUCCESS in 1h 10m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/71a7913d5e7941cf86bf9adec25073ef : SUCCESS in 1h 17m 22s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d76f65405aec4238b70a116462d71534 : SUCCESS in 1h 41m 35s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a521c8ab11ea419287a1204be3488d4f : SUCCESS in 1h 22m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7e3ec828a3754890974787cd755837d1 : SUCCESS in 1h 21m 32s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fd84a5dfc7dc45d1a5b9052a848a3604 : SUCCESS in 2h 21m 39s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ef20dcca326e42de9d46e2309f6d954b : FAILURE in 1h 21m 39s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/51a5c26d70994b78b7e54842e93bf9d2 : SUCCESS in 1h 22m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3fd6df5f34d344cd9b82434da551f236 : FAILURE in 1h 04m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/32224580951a40b4832d7589120ed8b9 : SUCCESS in 2h 30m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb771133ac644d16a2e288062e7885d8 : SUCCESS in 1h 10m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c24701025cdf4f51862c266a127fc418 : SUCCESS in 2h 11m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/60610be253a54364a4f06b4c5fb78257 : SUCCESS in 1h 56m 09s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/272d6e48414d46c59790081b82ce429b : SUCCESS in 1h 15m 07s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fb998e7c5bb849d7a81ea79a8bd98d2f : SUCCESS in 55m 12s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0f220b49a7cb4d09a5c6d7e6d61abc78 : SUCCESS in 1h 39m 31s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/56cdc410fe2e4d84aac406cb4c02186b : SUCCESS in 2h 02m 36s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4509378a39cc4b618f27759e2fb85bde : SUCCESS in 2h 39m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/04494e0e4c7048cfb97bb57499ce5b35 : SUCCESS in 2h 03m 39s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/abdb6f8a3971473886ccc5c69f8f099e : SUCCESS in 2h 02m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a527717f698661e6e6a7cefbba8f1e4973c8e0e0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-05-08 20:17:56.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/May-2020/Arista_Tempest_Testrefs/changes/55/726255/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"e919751182c4cae74be9a880f88993b4f411c48f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-25 15:42:26.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"08e537589e03d9c5afb1ef83eb7020c3f83459e8","revisions":{"2f8b1f3a18b5444e1a8e773f71455e1de4364e79":{"kind":"REWORK","_number":1,"created":"2020-05-07 22:21:12.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/55/726255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/726255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/726255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/726255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/726255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/726255/1"}}},"commit":{"parents":[{"commit":"390c4ac55f3ea883882412afdc1b921c4c3614e1","subject":"Merge \"Report L3 extensions enabled in the L3 agent\u0027s config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/390c4ac55f3ea883882412afdc1b921c4c3614e1"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-07 22:04:46.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-07 22:20:39.000000000","tz":120},"subject":"WIP: Replace depreacted filter:healthcheck by app:healthcheck","message":"WIP: Replace depreacted filter:healthcheck by app:healthcheck\n\nAfter applying the same patch on other OpenStack services, it\nappears that healthcheck should be an app, not a filter.\n\nThis patch:\n- Switches to an app, instead of a filter.\n- Add the disable_by_file_path plugin.\n- Add a functional test to make sure all of this is always working.\n\nChange-Id: I7e4cd88281fb573ca7f91068f1645a474c6be7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f8b1f3a18b5444e1a8e773f71455e1de4364e79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f8b1f3a18b5444e1a8e773f71455e1de4364e79"}]},"branch":"refs/heads/master"},"5786a8fb68c31739dc7929e6b9852ff91a1f7839":{"kind":"REWORK","_number":2,"created":"2020-05-07 23:33:38.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/55/726255/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/726255/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/726255/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/726255/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/726255/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/726255/2"}}},"commit":{"parents":[{"commit":"390c4ac55f3ea883882412afdc1b921c4c3614e1","subject":"Merge \"Report L3 extensions enabled in the L3 agent\u0027s config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/390c4ac55f3ea883882412afdc1b921c4c3614e1"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-07 22:04:46.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-07 23:32:54.000000000","tz":120},"subject":"WIP: Replace depreacted filter:healthcheck by app:healthcheck","message":"WIP: Replace depreacted filter:healthcheck by app:healthcheck\n\nAfter applying the same patch on other OpenStack services, it\nappears that healthcheck should be an app, not a filter.\n\nThis patch:\n- Switches to an app, instead of a filter.\n- Add the disable_by_file_path plugin.\n- Add a functional test to make sure all of this is always working.\n\nChange-Id: I7e4cd88281fb573ca7f91068f1645a474c6be7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5786a8fb68c31739dc7929e6b9852ff91a1f7839"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5786a8fb68c31739dc7929e6b9852ff91a1f7839"}]},"branch":"refs/heads/master"},"ea269311dcf6fe44cff8980030922a1580342770":{"kind":"REWORK","_number":3,"created":"2020-05-08 09:58:12.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/55/726255/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/726255/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/726255/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/726255/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/726255/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/726255/3"}}},"commit":{"parents":[{"commit":"390c4ac55f3ea883882412afdc1b921c4c3614e1","subject":"Merge \"Report L3 extensions enabled in the L3 agent\u0027s config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/390c4ac55f3ea883882412afdc1b921c4c3614e1"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-07 22:04:46.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-08 09:57:34.000000000","tz":120},"subject":"WIP: Replace depreacted filter:healthcheck by app:healthcheck","message":"WIP: Replace depreacted filter:healthcheck by app:healthcheck\n\nAfter applying the same patch on other OpenStack services, it\nappears that healthcheck should be an app, not a filter.\n\nThis patch:\n- Switches to an app, instead of a filter.\n- Add the disable_by_file_path plugin.\n- Add a functional test to make sure all of this is always working.\n\nChange-Id: I7e4cd88281fb573ca7f91068f1645a474c6be7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea269311dcf6fe44cff8980030922a1580342770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea269311dcf6fe44cff8980030922a1580342770"}]},"branch":"refs/heads/master"},"978bbaf1e9be307050f7dcd5e61ff20b0087809e":{"kind":"REWORK","_number":4,"created":"2020-05-08 13:46:06.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/55/726255/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/726255/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/726255/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/726255/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/726255/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/726255/4"}}},"commit":{"parents":[{"commit":"390c4ac55f3ea883882412afdc1b921c4c3614e1","subject":"Merge \"Report L3 extensions enabled in the L3 agent\u0027s config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/390c4ac55f3ea883882412afdc1b921c4c3614e1"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-07 22:04:46.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-08 13:45:19.000000000","tz":120},"subject":"WIP: Replace depreacted filter:healthcheck by app:healthcheck","message":"WIP: Replace depreacted filter:healthcheck by app:healthcheck\n\nAfter applying the same patch on other OpenStack services, it\nappears that healthcheck should be an app, not a filter.\n\nThis patch:\n- Switches to an app, instead of a filter.\n- Add the disable_by_file_path plugin.\n- Add a functional test to make sure all of this is always working.\n\nChange-Id: I7e4cd88281fb573ca7f91068f1645a474c6be7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/978bbaf1e9be307050f7dcd5e61ff20b0087809e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/978bbaf1e9be307050f7dcd5e61ff20b0087809e"}]},"branch":"refs/heads/master"},"08e537589e03d9c5afb1ef83eb7020c3f83459e8":{"kind":"REWORK","_number":5,"created":"2020-05-08 16:17:20.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/55/726255/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/726255/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/726255/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/726255/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/726255/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/726255/5"}}},"commit":{"parents":[{"commit":"390c4ac55f3ea883882412afdc1b921c4c3614e1","subject":"Merge \"Report L3 extensions enabled in the L3 agent\u0027s config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/390c4ac55f3ea883882412afdc1b921c4c3614e1"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-07 22:04:46.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-08 16:15:56.000000000","tz":120},"subject":"WIP: Replace depreacted filter:healthcheck by app:healthcheck","message":"WIP: Replace depreacted filter:healthcheck by app:healthcheck\n\nAfter applying the same patch on other OpenStack services, it\nappears that healthcheck should be an app, not a filter.\n\nThis patch:\n- Switches to an app, instead of a filter.\n- Add the disable_by_file_path plugin.\n- Add a functional test to make sure all of this is always working.\n\nChange-Id: I7e4cd88281fb573ca7f91068f1645a474c6be7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08e537589e03d9c5afb1ef83eb7020c3f83459e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08e537589e03d9c5afb1ef83eb7020c3f83459e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
