)]}'
{"id":"openstack%2Fneutron~727274","triplet_id":"openstack%2Fneutron~stable%2Fussuri~I00a35884b14af3e2cf751c04312c847ecfe658c7","project":"openstack/neutron","branch":"stable/ussuri","topic":"cap-pycodestyle","hashtags":[],"change_id":"I00a35884b14af3e2cf751c04312c847ecfe658c7","subject":"Cap pycodestyle to be \u003c 2.6.0","status":"MERGED","created":"2020-05-12 16:57:19.000000000","updated":"2020-05-13 14:22:16.000000000","submitted":"2020-05-13 14:19:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727274-1589379560421-c4901b80","meta_rev_id":"967db821a362d0716150f86ad7395d665ac314f5","_number":727274,"virtual_id_number":727274,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-05-12 20:19:07.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-13 14:19:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-13 04:53:52.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-05-13 09:24:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-05-13 08:34:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-05-13 12:55:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2020-05-12 19:38:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2020-05-13 06:20:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-05-13 12:55:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 19:38:15.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-05-13 04:53:52.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-05-13 06:20:19.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-05-13 08:34:44.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-05-13 09:24:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-05-13 12:55:46.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-05-13 14:19:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb3e99f96d3c54b03b585c5f9245dec07a259e07","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-12 16:57:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b3412c721358c11bbbf81a0a3d817a7784c95f9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-12 17:23:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"471585dd43fa2c3683c1fa5ebb347746d8ae63ea","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-12 18:06:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"70b5e5629a7db140aaa2f900111a67908311eb39","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-12 18:07:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63d40789e1034daeaa0e801ec08b28f30ca4fb43","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-12 18:12:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3fad5ab5852d3477481ad83ceb43c1a189a32a73","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-12 18:28:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d58108acfda54b1a0ae865c8cecc1b1d7362bf80","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-12 18:40:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6cee02d85f6e5017d862472a70cdc3f88754e764","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-12 19:09:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d8c9d85f230f7689b6bb13caf5fe259791b79da7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-12 19:10:03.000000000","message":"Patch Set 3:\n\nRestored PS1","accounts_in_message":[],"_revision_number":3},{"id":"42e857a9fac838d93a2e19828e32c102fabc3f49","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-12 19:38:15.000000000","message":"Patch Set 3: Code-Review+1\n\nthis is best possible way forward. and in master we can remove pycodestyle as new flake8 handle the cap for it.","accounts_in_message":[],"_revision_number":3},{"id":"b4f511cc9137b404ae44a9ac40809cc7d327b4d2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-12 20:19:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"268291d553013a2b117a03f00d8d2f878b66d14f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-12 21:08:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/727274/3 : SUCCESS in 1h 58m 21s","accounts_in_message":[],"_revision_number":3},{"id":"a8fdc8001bea5f2b17d11190d2e0d0ff9a48acaa","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-12 23:35:32.000000000","message":"Patch Set 3:\n\nThanks for having this up, waiting for zuul","accounts_in_message":[],"_revision_number":3},{"id":"bb08d4628df612082bd54c4f105b9a8864d3f737","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 02:52:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7953d6f42d9542579f75877bca04e526 : SUCCESS in 33m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce8638b59a9a4ab29d5ea86dcecaeea7 : SUCCESS in 28m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcffe33f54864cd2b035a3158586ff40 : SUCCESS in 14m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7b0f5c3c5c245c38dcef9cb6a2f22a8 : SUCCESS in 30m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5510838d9c92417280093b8e1d4ca556 : SUCCESS in 29m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51089d7b587a43d8988fb9c2a215e4fc : SUCCESS in 25m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0f725ca2ac044a3b38db5e0e2e98986 : SUCCESS in 10m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c71bfba467d242629e65125a73534d3a : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95d67e7259ed4111a1501c74c4a55160 : SUCCESS in 21m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/da5c764b194e40c99fb68a02cbc1cf74 : SUCCESS in 47m 29s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6dea9fbfbdfe476dac6ec11f0f27a92d : SUCCESS in 1h 08m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e00e90936d5d46b79ba8e73a95c88a94 : FAILURE in 1h 08m 00s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/939e298d77ca40c4ad141214a3b79e44 : SUCCESS in 1h 19m 40s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d7e8b01b6d6d4059a6b59b477a7de887 : SUCCESS in 52m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"af5d269787d13411ea8243df1a31efa0282efe42","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-13 03:10:16.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"01fa5a5ca208c5d6b03e80ebb7742d6b7e1d7ba0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-13 04:53:52.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/727274/3 : SUCCESS in 1h 43m 00s","accounts_in_message":[],"_revision_number":3},{"id":"a321620d3115a5bf873e6e84060b224e95f47f54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 05:11:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a3304f84c434b0b9831984b9478f007 : SUCCESS in 38m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c87575328d54b5d8072206276a2278d : SUCCESS in 24m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0d605f98588422186c38c6416fd3a1a : SUCCESS in 13m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/819def158aaa4e5babd4971912ebf15a : SUCCESS in 26m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b62498a67a844ef872cd04dbf505ce3 : SUCCESS in 34m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89fe318f28db4ddc80ca0e0adc357431 : SUCCESS in 27m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b7d06c655904530a66ca1fda79776d9 : SUCCESS in 9m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/545113fb3a1e4187934cfc892f3bd9b0 : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a058279d59a4bc7bafafebabff119ef : SUCCESS in 24m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/604c8c981e2e4119b2c9d1619f2fa475 : SUCCESS in 56m 53s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/be92fb9f4054460e988a571770f68fa6 : SUCCESS in 1h 11m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6819b96da40f46058d1f8684ed18cb9d : SUCCESS in 1h 20m 33s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/92bcdf51bfaf445c8ca12172b690394d : SUCCESS in 1h 16m 36s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8c56d8c147674d1c98d005e4135ab6eb : SUCCESS in 56m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"551745a57054a542a177eb0a0a090b56a2b685fd","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-05-13 06:20:19.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"27b70110030db9bc0a58da93393b084e75681547","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-13 08:34:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6493921e4b118f30ccfc7557e6d672b83a37523c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-13 09:24:44.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8aa1d786a5d6ac49ab5e23e4f77968706e5073b5","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-13 10:23:14.000000000","message":"Patch Set 3:\n\nJust rechecked in 726245 a variant will be needed for rocky too: it fails even though it does not list pycodestyle in test-requirements.txt\n\n(and of course needed in train/stein)","accounts_in_message":[],"_revision_number":3},{"id":"31ed0ed0e2ffe633a415ac97ae4caf4f827d420e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-13 12:46:06.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 8ac909434cc98ecd2271c564b99cd300db681fdb","accounts_in_message":[],"_revision_number":3},{"id":"6001e37f9b483ef42ad89b0446c5bf230f97d9e6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-13 12:55:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6e29db9c1a24b68b83d2bdc13a48ff9711a371a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 12:56:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"926161d5f399e1ae6a58d06124b29ba84a165529","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 14:19:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"dd976dcacdc8dc4e8609c290ff155a3e3f211174","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 14:19:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/427ef10fdf544b5aaded461fe86b0122 : SUCCESS in 27m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5971cad4d4de441dad9fe8a7353b0fd8 : SUCCESS in 13m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbba9ad430b944959a6f6afbaef6946f : SUCCESS in 26m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4541c37ee6f54a70a8f460bf0aed4c5b : SUCCESS in 35m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f5011e5a002437b90fe071ba656fa01 : SUCCESS in 9m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/78ed6ca8af8b415ab08d1c07722983de : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9290e4849727438c8f353a53d450fa84 : SUCCESS in 25m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/636441cca9344b0eac0233cebe55e74b : SUCCESS in 46m 28s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/cfd530e69b37436794e857a8ea9a68f6 : SUCCESS in 1h 12m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8e4df1ffbcdc42738ad7aa2d46466470 : SUCCESS in 1h 17m 01s","accounts_in_message":[],"_revision_number":3},{"id":"0570d6fc488322b231383d931b167542d4117664","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 14:22:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec1a2ae68fb74abba828da09b545dd8c : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a3717ddb5404b68b8d3c21d1a26a451 : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"719cae183ae3f67da3ac834140af3f85f2113c10","revisions":{"d9dba9cf50542a7deebe7af3a33b3aee4488550a":{"kind":"REWORK","_number":1,"created":"2020-05-12 16:57:19.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/74/727274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/727274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/727274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/727274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/727274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/727274/1"}}},"commit":{"parents":[{"commit":"86b57718966dc2165b4cfb54bcae21b515ffe68f","subject":"[ovn]: Fix l3_plugin.add_router_interface to comply with RouterPluginBase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86b57718966dc2165b4cfb54bcae21b515ffe68f"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-12 16:56:41.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-12 16:56:41.000000000","tz":-240},"subject":"Cap pycodestyle to be \u003c 2.6.0","message":"Cap pycodestyle to be \u003c 2.6.0\n\nThe 2.6.0 version introduces some checks that cause failures\nwith the current code. To avoid that, cap pycodestyle to a\nversion that had been tested without errors.\n\nChange-Id: I00a35884b14af3e2cf751c04312c847ecfe658c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9dba9cf50542a7deebe7af3a33b3aee4488550a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9dba9cf50542a7deebe7af3a33b3aee4488550a"}]},"branch":"refs/heads/stable/ussuri"},"0dc8c4eec206141a14f1603351147f5303d4d9a6":{"kind":"REWORK","_number":2,"created":"2020-05-12 18:06:48.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/74/727274/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/727274/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/727274/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/727274/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/727274/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/727274/2"}}},"commit":{"parents":[{"commit":"86b57718966dc2165b4cfb54bcae21b515ffe68f","subject":"[ovn]: Fix l3_plugin.add_router_interface to comply with RouterPluginBase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86b57718966dc2165b4cfb54bcae21b515ffe68f"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-12 16:56:41.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-12 17:55:08.000000000","tz":-240},"subject":"Cap flake8 to be \u003c 3.8.0","message":"Cap flake8 to be \u003c 3.8.0\n\nThe 3.8.0 version introduces some checks that cause failures\nwith the current code. To avoid that, cap flake8 to a\nversion that had been tested without errors.\n\nChange-Id: I00a35884b14af3e2cf751c04312c847ecfe658c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0dc8c4eec206141a14f1603351147f5303d4d9a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0dc8c4eec206141a14f1603351147f5303d4d9a6"}]},"branch":"refs/heads/stable/ussuri"},"719cae183ae3f67da3ac834140af3f85f2113c10":{"kind":"REWORK","_number":3,"created":"2020-05-12 19:09:12.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/74/727274/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/727274/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/727274/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/727274/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/727274/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/727274/3"}}},"commit":{"parents":[{"commit":"86b57718966dc2165b4cfb54bcae21b515ffe68f","subject":"[ovn]: Fix l3_plugin.add_router_interface to comply with RouterPluginBase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86b57718966dc2165b4cfb54bcae21b515ffe68f"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-12 16:56:41.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-12 19:08:40.000000000","tz":-240},"subject":"Cap pycodestyle to be \u003c 2.6.0","message":"Cap pycodestyle to be \u003c 2.6.0\n\nThe 2.6.0 version introduces some checks that cause failures\nwith the current code. To avoid that, cap pycodestyle to a\nversion that had been tested without errors.\n\nChange-Id: I00a35884b14af3e2cf751c04312c847ecfe658c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/719cae183ae3f67da3ac834140af3f85f2113c10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/719cae183ae3f67da3ac834140af3f85f2113c10"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
