)]}'
{"id":"openstack%2Fneutron~730116","triplet_id":"openstack%2Fneutron~master~I2a4279c2de793a8ab8e42b3b218507c55bebe954","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"I2a4279c2de793a8ab8e42b3b218507c55bebe954","subject":"DNM, running periodic job","status":"ABANDONED","created":"2020-05-21 19:55:24.000000000","updated":"2020-05-22 15:53:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"782c62e420a0c7ac7d3b2b4225f97e8703918820","_number":730116,"virtual_id_number":730116,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-21 22:24:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"date":"2020-05-21 20:32:14.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-21 20:32:14.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-05-21 22:24:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"394202195f4aae70dc1521ddafdedaec89cfc2e1","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-21 19:55:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"509e352369a35b3c362cd7da4ddd913d81ee5fb2","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-21 19:55:35.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"6a1ca73614f825802496664d6d56e166cc9b997b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 19:55:42.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/neutron/zuul.d/project.yaml@master\", line 29, column 31","accounts_in_message":[],"_revision_number":1},{"id":"f26f0031c17a3071dfc39f29293b703970f6f9c3","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-21 20:00:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"da3d69875abaf4f051c4407f29b979f827448b7b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-21 20:32:14.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/730116/2 : FAILURE in 31m 08s","accounts_in_message":[],"_revision_number":2},{"id":"96e2e9cf322ba9b00e0aa189b6a0404a29b6cd77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 22:24:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/61b1f8e6095c488ebe907895e6b80968 : SUCCESS in 56m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8e82a05bad5c4bd584815400b37afae3 : SUCCESS in 39m 03s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/e64fe855621942c3bb6d8bc1a8caafa3 : FAILURE in 2h 17m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1ad1969171974451a66f6492eb6f65c6 : SUCCESS in 1h 18m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/261480b0bb6f45ee895710f9690df60e : SUCCESS in 1h 04m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7dc94d4d5f104ba4b0b020dc3372d7b5 : SUCCESS in 1h 32m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afb52742bb2b4c26b2a3905bbdd5c624 : SUCCESS in 36m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86cbeee60f9a4c53a70fd02947ea1dab : SUCCESS in 28m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69bdb25f3f634985acfa4e33c558e08d : SUCCESS in 11m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89f853b03e774aa9b151ba15bbf6150e : SUCCESS in 28m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a668f2f9fef44cb8397ea2a711aaaa7 : SUCCESS in 25m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bacdbf22ff874bb496f8e3d57a8d204f : SUCCESS in 9m 21s\n- openstack-tox-py36-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/99ef66f7c78b406c8c15729846c66ec9 : SUCCESS in 34m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6f8d756c4f9e4c0ab477c25ddf38082f : SUCCESS in 50m 22s","accounts_in_message":[],"_revision_number":2},{"id":"782c62e420a0c7ac7d3b2b4225f97e8703918820","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-22 15:53:15.000000000","message":"Abandoned\n\ntest passed. yay.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fa852c0995ae04bb35127ba059dced896bc577fd","revisions":{"f5dc59ffd16174e96b0b3cf04f23542603bff567":{"kind":"REWORK","_number":1,"created":"2020-05-21 19:55:24.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/16/730116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/730116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/730116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/730116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/730116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/730116/1"}}},"commit":{"parents":[{"commit":"8b2550d37636c2632f54d69d7dbd61b52ccfd10e","subject":"Merge \"Fix broken unit tests related to ovsdbapp indexing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b2550d37636c2632f54d69d7dbd61b52ccfd10e"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-21 19:55:17.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-21 19:55:17.000000000","tz":-300},"subject":"DNM, running periodic job","message":"DNM, running periodic job\n\nChange-Id: I2a4279c2de793a8ab8e42b3b218507c55bebe954\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f5dc59ffd16174e96b0b3cf04f23542603bff567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f5dc59ffd16174e96b0b3cf04f23542603bff567"}]},"branch":"refs/heads/master"},"fa852c0995ae04bb35127ba059dced896bc577fd":{"kind":"REWORK","_number":2,"created":"2020-05-21 20:00:33.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/16/730116/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/730116/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/730116/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/730116/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/730116/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/730116/2"}}},"commit":{"parents":[{"commit":"8b2550d37636c2632f54d69d7dbd61b52ccfd10e","subject":"Merge \"Fix broken unit tests related to ovsdbapp indexing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b2550d37636c2632f54d69d7dbd61b52ccfd10e"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-21 19:55:17.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-21 20:00:27.000000000","tz":-300},"subject":"DNM, running periodic job","message":"DNM, running periodic job\n\nChange-Id: I2a4279c2de793a8ab8e42b3b218507c55bebe954\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa852c0995ae04bb35127ba059dced896bc577fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa852c0995ae04bb35127ba059dced896bc577fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
