)]}'
{"id":"openstack%2Fneutron~730387","triplet_id":"openstack%2Fneutron~master~I6baedf05e5ddaf9413a5c5abc6e8f34bb1e874fb","project":"openstack/neutron","branch":"master","topic":"remove-six","hashtags":[],"change_id":"I6baedf05e5ddaf9413a5c5abc6e8f34bb1e874fb","subject":"Remove usage of six.next and iterators","status":"ABANDONED","created":"2020-05-22 18:08:46.000000000","updated":"2020-08-19 21:31:40.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"433a62ffc88e1a70acdb23086a894a49ca7349ff","_number":730387,"virtual_id_number":730387,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2020-05-27 04:58:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"date":"2020-06-06 11:03:11.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-27 00:08:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":-1,"date":"2020-05-26 19:41:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-22 20:11:30.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-05-26 15:34:52.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-05-26 18:59:24.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-05-27 00:08:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-27 04:58:42.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-06 11:03:11.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c3770b005e0fef22c1796966da052956805325a9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-22 18:08:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fa46ab3506f01f95436f9ca23da2d144f5d8ce8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-22 18:46:52.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/730387/1 : FAILURE in 35m 43s","accounts_in_message":[],"_revision_number":1},{"id":"fc7aff310ff0694020348170dc4ef1edb18ec0ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 20:52:53.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/01a29c92201a40e0a29f9a9b4b2f7071 : SUCCESS in 46m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0e9c1cc4f1cd4a51b248d8eb4a92ccf9 : SUCCESS in 39m 38s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/03d8e57eaaf643d6abc50ea2e23af8b0 : FAILURE in 2h 34m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/00cbe13b26a34b87ba730bbdbc476098 : SUCCESS in 1h 17m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/878103abc79b4d19aa886f01367fdd15 : SUCCESS in 1h 21m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/49eaea77084f47e4a472cf663e3a0d06 : SUCCESS in 1h 05m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fec299c4fa04b07a83fc1dbd51f7f4c : FAILURE in 30m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ff8cedd5e0a46e49dcb976cfa3a0c42 : FAILURE in 28m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d953cac33f44fb6a01583438072cb42 : FAILURE in 10m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85355ba5b9ff4e62bfce717defa05b42 : FAILURE in 29m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8373e870ddb84cc2b5189bb3d2a7d9fe : FAILURE in 27m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0632977387cd4760b93d92b4a6bdcc00 : SUCCESS in 12m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f99d645e455143ee9004526f851f77ee : SUCCESS in 47m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3060c74af59c444dbb6372bcaa764a4e : SUCCESS in 1h 14m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0e925ed7e4f8424c9a03b1965590f704 : SUCCESS in 1h 40m 03s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7778182314234d2b879db5bc0970740c : SUCCESS in 1h 50m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3313bc03dcb43d6bd87483bbfbb0ffc : SUCCESS in 1h 12m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/466edb96c6f6427e8c9b1dae48a0d87f : SUCCESS in 1h 14m 21s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2fde5abc52e74ad384518d11da7c3f3a : SUCCESS in 1h 15m 00s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7af56ee01be542bab6d3affb62ec2a31 : SUCCESS in 1h 10m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/364a1e273f354980828de2dfbaef92ee : SUCCESS in 1h 20m 09s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/703da9e59cb742a2bfc8173a82b1af3a : SUCCESS in 1h 39m 39s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/96e4636438c747beaaca101c575f8a9e : FAILURE in 1h 18m 25s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a70c020aba0c4baea713996a0f6292a5 : SUCCESS in 1h 18m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e5e0d6f7b4ca400eaee13a8c8fd2a61a : SUCCESS in 1h 05m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/89fbba8e5f844ef39b3b074373eea184 : FAILURE in 2h 29m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0537b589d364692b8c8a354cafb274d : SUCCESS in 1h 10m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b8c8934b27414de5927e6641dfde3cab : SUCCESS in 2h 22m 01s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/58e222b81ca0444f8686233c6571e3df : SUCCESS in 2h 19m 12s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/61339fab077940958078ba693462b028 : SUCCESS in 1h 23m 01s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0d98d810acb54607a2e8c66083b97c14 : SUCCESS in 45m 16s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/36e0695b5d96411a9467d5b79223bc57 : SUCCESS in 1h 42m 08s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fd4aaafa7aca4c84929923752bc39062 : SUCCESS in 2h 08m 43s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b8e23f946a0b4624bdbc5a367992ad9b : SUCCESS in 2h 29m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/92be63523bec4bf28464bfc57326e157 : FAILURE in 2h 05m 48s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/71fd7557cdff47bc920537637b9a6ec3 : SUCCESS in 2h 01m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"536354e26573c948690d6e5679a94c040439eca4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-05-22 20:53:46.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/May-2020/Arista_Tempest_Testrefs/changes/87/730387/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"e91dd3f4efc280d3a1b49c1367cb123453ff0bdd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-26 15:34:52.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nI think the unit tests are failing due to next(resource.keys()) and similar conversions","accounts_in_message":[],"_revision_number":1},{"id":"a85c5b758c67d076e23981493b8896c7f17bc79b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-26 16:48:19.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"29b56b0d7d5d47f10308589482472b073b758e1e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-26 16:48:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b84fb72bd30b434d5aceed1a24b0a2310526201b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-26 16:49:14.000000000","message":"Patch Set 2: Workflow-1\n\n-W until I get more eyes on these next() changes","accounts_in_message":[],"_revision_number":2},{"id":"dc65253c894466781e45a4a18c2085949f147d8b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-26 18:59:24.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/730387/2 : FAILURE in 2h 10m 01s","accounts_in_message":[],"_revision_number":2},{"id":"b11090805e72e009a2d3373a2690a16dcc581125","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-26 19:41:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b6d16db6c163febd33932280244439f44911f929","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-26 19:41:20.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"103acd22f4d47a6be8a8c34354e27f4415061676","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 00:08:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7e2ca864f0554b9f8861302ee40035a1 : SUCCESS in 55m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a62c58af423e44a687018c778ef311fc : SUCCESS in 48m 11s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/882528ac80ca4cacbe5fc7a98a1cabab : FAILURE in 2h 12m 40s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/58e8e7ba6edd4cc4bbc27ce1ff08e9a2 : SUCCESS in 1h 16m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e5ed41f8214c40668ed5cd1e3e9f6b19 : SUCCESS in 1h 09m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7d17a820730a4d7da910b23c4a732852 : SUCCESS in 1h 19m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ace989a11c48403094b832f9d580fd51 : SUCCESS in 42m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4eccdc18c214f84a329e9de7b062f6c : SUCCESS in 29m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21b9b86c7ab640db80c34286b1baecc3 : SUCCESS in 11m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6ea784cf5eb4870baa06a5f8dbaa820 : SUCCESS in 32m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c45c098874cc404ca6f81a08a0e797b2 : SUCCESS in 33m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e4b9ff5976144619d7978ff49031979 : SUCCESS in 10m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/234db4ab910f4a94bf80f05ca3fd06bc : SUCCESS in 54m 19s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/60c41dd80f42495da52bc5392a5638fc : SUCCESS in 1h 20m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d680e92e22f8469781ecb4dfee44ddae : SUCCESS in 1h 20m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/87e1719a9c204cb4ae5d4fb1916d699c : SUCCESS in 2h 10m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c62db45533aa4bc79cac2235a3dad344 : SUCCESS in 1h 04m 27s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c601c3427eda4858bd7c232e198e531b : SUCCESS in 1h 26m 20s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/cccf537b64ff4895a39d2c4ffb46011f : SUCCESS in 1h 19m 34s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/31190ceb3ce54b8da8d4898a968432d0 : SUCCESS in 1h 15m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b990786b6c40444a86b24c8deee3ff6e : SUCCESS in 1h 08m 24s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/117cc8d0aa5c4697a50d7e3afbe62d4b : SUCCESS in 1h 50m 35s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/41e99e3aa41b4e4080e2e889707a729d : SUCCESS in 1h 46m 29s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/02f485a0483a4b7094a2e4546ae60af1 : SUCCESS in 1h 12m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5943e71859df4e0391ccde1e115aa253 : SUCCESS in 1h 11m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6c5ffbf13f8c4abd8eb2151fa2c7e4f8 : SUCCESS in 2h 06m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df899e57d6d54fa684be66986c6bca7a : SUCCESS in 1h 00m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5358b152de4742d29314258e49470f59 : SUCCESS in 2h 07m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3ada42d8761e445a8a572022a0967840 : SUCCESS in 1h 35m 12s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fec3835dc06f43fe9944f627bf618633 : SUCCESS in 1h 21m 33s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/79de5b6594ed4540ac6305fe827ef5f8 : SUCCESS in 1h 01m 41s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ec3521282ff24177b3e6f414468cf951 : SUCCESS in 1h 44m 13s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d6058a4293ac4f9280e56a4d4780053e : SUCCESS in 1h 51m 19s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/81f1a0e70575467a871ad1dc11b4b20e : SUCCESS in 2h 41m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/aecc49ce201e4db0934ec74d5e000b90 : FAILURE in 1h 36m 27s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c90a307207294992bcc915f802062df1 : SUCCESS in 2h 35m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"48b495a01ec520aae4b4266818ad01c4a39541fb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-05-27 03:08:16.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/May-2020/Arista_Tempest_Testrefs/changes/87/730387/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"465d2f3d3a5401e469caf4c7be981dd5211eb403","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-05-27 04:58:42.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/May-2020/Arista_Tempest_Testrefs/changes/87/730387/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"cac9d39933371349738080cff122f4b98bc3e007","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-27 18:59:58.000000000","message":"Patch Set 3:\n\nHi Lajos - can you take a look at my changes since your last review?  They look Ok to me but you did correctly notice the difference with next() in py3.  At worst I can just remove those changes and do it in a separate patch.  Thanks.","accounts_in_message":[],"_revision_number":3},{"id":"54a42a92fb460428fa24765031c8ab173a6744d2","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-06-06 11:03:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-06-06/730387/3/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/5144be7 : EXCEPTION (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c0db749ab0131c1126e76bb5d930be6bde910762","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-19 21:31:40.000000000","message":"Abandoned\n\nRodolfo took care of these in https://review.opendev.org/#/c/740510/","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f3d583189d34c8c7c30e0d5e7ea4dd7b8606665a","revisions":{"d0c6ad81bc7602d9129d3df65b3bb187671ceade":{"kind":"REWORK","_number":1,"created":"2020-05-22 18:08:46.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/87/730387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/730387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/730387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/730387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/730387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/730387/1"}}},"commit":{"parents":[{"commit":"4f10c3bd3f57836aeb1b6ba4305301d63c046bd0","subject":"Remove usage of six.text_type and six.string_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f10c3bd3f57836aeb1b6ba4305301d63c046bd0"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-22 18:03:14.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-22 18:03:14.000000000","tz":-240},"subject":"Remove usage of six.next and iterators","message":"Remove usage of six.next and iterators\n\nWith python 3.x, six.next, six.itervalues, six.iteritems\nand six.iterkeys can be replaced by their python 3\nequivalents.\n\nAnother step in removing all of six usage from neutron.\n\nChange-Id: I6baedf05e5ddaf9413a5c5abc6e8f34bb1e874fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d0c6ad81bc7602d9129d3df65b3bb187671ceade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d0c6ad81bc7602d9129d3df65b3bb187671ceade"}]},"branch":"refs/heads/master"},"f187a975faa25184607fc380432b538570fac5b9":{"kind":"REWORK","_number":2,"created":"2020-05-26 16:48:34.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/87/730387/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/730387/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/730387/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/730387/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/730387/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/730387/2"}}},"commit":{"parents":[{"commit":"4f10c3bd3f57836aeb1b6ba4305301d63c046bd0","subject":"Remove usage of six.text_type and six.string_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f10c3bd3f57836aeb1b6ba4305301d63c046bd0"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-22 18:03:14.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-26 16:48:07.000000000","tz":-240},"subject":"Remove usage of six.next and iterators","message":"Remove usage of six.next and iterators\n\nWith python 3.x, six.next, six.itervalues, six.iteritems\nand six.iterkeys can be replaced by their python 3\nequivalents.\n\nAnother step in removing all of six usage from neutron.\n\nChange-Id: I6baedf05e5ddaf9413a5c5abc6e8f34bb1e874fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f187a975faa25184607fc380432b538570fac5b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f187a975faa25184607fc380432b538570fac5b9"}]},"branch":"refs/heads/master"},"f3d583189d34c8c7c30e0d5e7ea4dd7b8606665a":{"kind":"REWORK","_number":3,"created":"2020-05-26 19:41:10.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/87/730387/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/730387/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/730387/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/730387/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/730387/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/730387/3"}}},"commit":{"parents":[{"commit":"4f10c3bd3f57836aeb1b6ba4305301d63c046bd0","subject":"Remove usage of six.text_type and six.string_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f10c3bd3f57836aeb1b6ba4305301d63c046bd0"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-22 18:03:14.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-26 19:40:57.000000000","tz":-240},"subject":"Remove usage of six.next and iterators","message":"Remove usage of six.next and iterators\n\nWith python 3.x, six.next, six.itervalues, six.iteritems\nand six.iterkeys can be replaced by their python 3\nequivalents.\n\nAnother step in removing all of six usage from neutron.\n\nChange-Id: I6baedf05e5ddaf9413a5c5abc6e8f34bb1e874fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3d583189d34c8c7c30e0d5e7ea4dd7b8606665a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3d583189d34c8c7c30e0d5e7ea4dd7b8606665a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
