)]}'
{"id":"openstack%2Fneutron~731446","triplet_id":"openstack%2Fneutron~master~I7d937e39bf67ffb31f8a82f993323e9e70d6c344","project":"openstack/neutron","branch":"master","topic":"bug/1881070","hashtags":[],"change_id":"I7d937e39bf67ffb31f8a82f993323e9e70d6c344","subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","status":"ABANDONED","created":"2020-05-28 07:08:19.000000000","updated":"2020-11-26 07:24:45.000000000","total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"08cb47b58a92dc710bfe1821c7158d7b3ff5ad1b","_number":731446,"virtual_id_number":731446,"owner":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"date":"2020-07-28 01:52:27.000000000","_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-28 03:48:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-07-28 02:41:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":30744,"name":"uchenily","email":"uchenily@gmail.com","username":"uchenily"},{"date":"2020-07-28 09:33:44.000000000","_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"date":"2020-07-28 03:14:07.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"disliked":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-28 07:48:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},{"value":1,"date":"2020-09-21 08:07:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30744,"name":"uchenily","email":"uchenily@gmail.com","username":"uchenily"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":-1,"date":"2020-07-28 08:04:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":1,"date":"2020-07-28 05:50:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":30744,"name":"uchenily","email":"uchenily@gmail.com","username":"uchenily"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":30744,"name":"uchenily","email":"uchenily@gmail.com","username":"uchenily"},{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},{"_account_id":30744,"name":"uchenily","email":"uchenily@gmail.com","username":"uchenily"},{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 05:43:00.000000000","updated_by":{"_account_id":30744,"name":"uchenily","email":"uchenily@gmail.com","username":"uchenily"},"reviewer":{"_account_id":30744,"name":"uchenily","email":"uchenily@gmail.com","username":"uchenily"},"state":"REVIEWER"},{"updated":"2020-06-04 12:29:50.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-06 11:01:46.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-22 13:11:10.000000000","updated_by":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"reviewer":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"state":"REVIEWER"},{"updated":"2020-07-27 13:11:58.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-07-28 01:52:27.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-28 02:41:45.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-28 03:14:07.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-07-28 03:48:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-28 05:50:46.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-07-28 07:48:52.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-07-28 08:04:29.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2020-09-21 08:07:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"dbefcc1436bc38cde33506c06bf8404c22ee0126","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-05-28 07:08:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36800a527aabc87a43bbe66ed4daf27a4b8cfac2","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-05-28 08:09:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nalso need a UT please","accounts_in_message":[],"_revision_number":1},{"id":"241fe67dc415f821fd49b144fccea69ee6a217c7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-05-28 08:22:09.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/May-2020/Arista_Tempest_Testrefs/changes/46/731446/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"0199b8b38fde2e470048268bf88f9fa03212e666","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-28 08:56:25.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/731446/1 : FAILURE in 1h 47m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8847eda584b689b020c1cc7382e5b2d83377c288","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-28 08:57:18.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/46/731446/1/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"24f4a7a17035ae4c7fbf06016945feaaa762553e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 10:06:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d8793c1f1ed747a7ba5928e9a31667ff : SUCCESS in 59m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f9f5a2ac18dd402da11c3c2172d81d26 : SUCCESS in 24m 18s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/cdf2fa6e58274ce48d08ebfd3e6db462 : FAILURE in 1h 45m 22s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/27d29cf5b8c642ee95bf475e6230cd17 : SUCCESS in 1h 13m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0e7bbafd33b747efa8f30716f1da2159 : SUCCESS in 1h 26m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a72ed4e7746643768ca593c20de035c4 : SUCCESS in 1h 03m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9794881a7f0489aa89fb464b8d20b46 : FAILURE in 37m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9030778bf7a4bd490174ff347c9ea28 : FAILURE in 32m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb73654ea652425ba6b7e57d327d815b : SUCCESS in 13m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dc7ace3add7428facadc33156f3dbf6 : FAILURE in 28m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/521ce30d89f7496a98fd05da166e9333 : FAILURE in 25m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef76b1cb9d9c4f08a023c765fc30df47 : SUCCESS in 9m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3c5a2baa59a5478dace0a82d020168ef : SUCCESS in 49m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/27f859475b7643cbb4968f624c2b4fca : SUCCESS in 1h 12m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6f6f497969d14d39a8df98289e7124b8 : SUCCESS in 1h 11m 04s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/af4971dd86cc472dbb178f5da21c9606 : SUCCESS in 1h 52m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/869c8f1e467d4802a2a9768194433148 : SUCCESS in 1h 02m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9b75ddec9e594c24866a288c4424163a : SUCCESS in 1h 04m 41s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7fdc3db335e54c878a1a74b98b97e06d : SUCCESS in 1h 07m 36s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a26b3f485c9420ca318bff74ebbf554 : SUCCESS in 1h 09m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8de358b4ae97485694c172dbcdceece4 : SUCCESS in 1h 01m 30s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b26f8be8cb38406e811a12c55bab64d4 : SUCCESS in 1h 27m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5f93ce4788a547d4b8aecfdc01ddb0ea : FAILURE in 1h 43m 34s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0686becefc094eb4b787787e19b31a2a : SUCCESS in 1h 10m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/50b3a161658e422797dd805d87fb448e : SUCCESS in 56m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f24f5c59de524309b633580044ff402d : SUCCESS in 1h 55m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b720618f1454ea09ed0610e0dc4df5f : SUCCESS in 50m 40s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/90f330c1739f475f9408b4a7223befbd : FAILURE in 1h 51m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f53c9d7bd191425681f08e37efed5b2f : SUCCESS in 1h 53m 12s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1db7656015014a2390465751f096fe9e : SUCCESS in 1h 14m 49s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3519ad0c45e142aa9753b7ae32e988dd : SUCCESS in 46m 25s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6598ba14a87b4a4eb1271c690088b2f7 : SUCCESS in 1h 36m 36s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/23100057c0f840eab8cf93151ed874ac : TIMED_OUT in 2h 15m 57s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/98659b9bbf854c04bfa4ecc93078cb3e : SUCCESS in 2h 51m 55s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3074a019e57f42e1916bf72231be237f : FAILURE in 1h 51m 31s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f97bbff098f74cb0a6f3ffc60e647938 : SUCCESS in 2h 05m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c215ba5dc0ee1d824482bbf2f969c9e26041541","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-05-28 10:43:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3cefd58d65235f4a37a550fa2c71256e74e1a69","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-05-28 10:49:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b1c1ca7e1683a7d43637aff70d1cc5444d5e0d9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-28 13:10:51.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/46/731446/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/46/731446/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"91fe9ba8c6b5fae6569f8c8a2459d6a688c9b85e","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-05-29 14:14:30.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0e095b0729f1e148f4faf4f3118458ef6d939506","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-05-29 14:15:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df3ef0915b8345d8dc69ff0048a294bb89068c71","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-05-30 03:23:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"614ed49460407ad0156913e968ea61386c6fe02d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-05-30 04:41:36.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/May-2020/Arista_Tempest_Testrefs/changes/46/731446/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"7cdbf9806f1dfb240357d88cc1a20780027699b5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-05-30 05:05:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/2 : SUCCESS in 1h 40m 41s","accounts_in_message":[],"_revision_number":2},{"id":"d20e9b16e2458d031772199842ab61b53d1c5675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 05:48:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0be2ac3d1b134c379ccedaf7d6ae2114 : SUCCESS in 46m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/615e4e4396974b4db81cdba07585796d : SUCCESS in 31m 03s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/3eb4d8ee66ef4b95af9fee3bf06f79f9 : FAILURE in 2h 05m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2d48ced3ba824aec9a2f25cb6f19ffe0 : SUCCESS in 1h 01m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7e092daead2f465288a962b0c6f3127a : SUCCESS in 1h 10m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ab65678e22654bd8927fc7a80567fd32 : SUCCESS in 55m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/216c1194bda041fbbd24f83ddf1787f2 : FAILURE in 28m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e338c43897c4400be629de696fa6877 : FAILURE in 24m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/846e95f9da9047bcb093d5f69a90daf1 : SUCCESS in 9m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fab9be0c4874d2da1475e81458618ba : FAILURE in 23m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91310be5cddc41458a9f3f1e5428f8bd : FAILURE in 23m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17f609b4a2304cdf8b5303de9bf28fd2 : SUCCESS in 10m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/965a7ac73a374d32b3c02f13c39d3c1c : SUCCESS in 47m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a7fa33ff3811446faef60e9bba1f171e : SUCCESS in 44m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b8c18af14e2a4cd4a7608bd360f02104 : SUCCESS in 1h 06m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a017922242f04ba7bb85d8d834068072 : SUCCESS in 1h 09m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7c7bf3deca7f4ea685ecb83880b4ea0d : SUCCESS in 1h 37m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/941d22a4e78d4a828ea2254683597dc6 : SUCCESS in 1h 04m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5cdcc52c1f764b9f966f7585e8fd462e : SUCCESS in 1h 03m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0f536a5fbf5a4a7b8a113fc87a86b907 : SUCCESS in 1h 02m 05s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/25e011a2c5cb4eda8b09a7f8b918ae33 : SUCCESS in 1h 04m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/80ea2f1f79554f809c9a0ab25c27d521 : SUCCESS in 1h 04m 49s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2de8567ab5cb47568e77e72127d70c72 : SUCCESS in 1h 30m 51s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a418622c9ffb4da68d3cac6819b1a7fe : FAILURE in 1h 18m 04s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/418cbb7d909949658ab1031bc3101631 : SUCCESS in 1h 08m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/97c75dc788b345139cd15a63ed0ebd68 : SUCCESS in 50m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f2c68db54102441b8f56e0bca70321b5 : SUCCESS in 1h 57m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5954f62533034de8b0505cc8a634c9f1 : SUCCESS in 49m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6596dd744ded4df9943f52aba462031b : SUCCESS in 1h 45m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b246309a3dda4a43b22e11396e54c694 : SUCCESS in 1h 06m 59s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fc35e239356445e38b4f292298df9d71 : SUCCESS in 1h 13m 37s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/699ccb831f4c48b09a4ea3a0c7263ef4 : SUCCESS in 1h 30m 18s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/57ce61f4778d4bf38f89ed785cde79a9 : SUCCESS in 1h 47m 10s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d31f9d09627d43f897b8e7f3b71b9e1f : SUCCESS in 2h 22m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d682c44cd97c40aeb4c91ac50399a23d : SUCCESS in 2h 09m 09s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/db7b7cd8485b4755924693301ee88dd1 : SUCCESS in 2h 03m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aee21d90f68f607b53334fce95bcc140251c199d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-30 11:47:39.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/46/731446/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/46/731446/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ed01247d8ddfaf245c889505b9e481562080df1c","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-05-31 09:13:28.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c307e6420f60511d0e84dc621b78e8e89eadf810","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-01 08:54:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ca6e5bb5f51852ba8b45369498b9110a0969fdec","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-01 08:58:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ed6562f0c0ba72beed7292d36c36aa989b3d272f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-01 10:09:14.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/June-2020/Arista_Tempest_Testrefs/changes/46/731446/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"ac1ad0b4add4ca7458d318c617a316f6cf3b8110","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-01 10:53:49.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/731446/3 : FAILURE in 1h 58m 25s","accounts_in_message":[],"_revision_number":3},{"id":"d9223fc41c1685dd77c594e2133e924a685cc163","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-01 10:54:49.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/46/731446/3/SRIOV-neutron-macvtap : ABORTED\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/46/731446/3/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"aee1f8c4faa289fe56539a731eb53b98b190c42d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 11:20:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a911d98d84714917b2429bab508e0f51 : SUCCESS in 1h 01m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f602cfc370d44d49b1c32b1d5f1f3c00 : SUCCESS in 36m 38s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/d7634ef0f7fa488591671d05cb63bf7a : FAILURE in 2h 12m 04s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b5a83a55ca9042e29c4651f2570f9279 : SUCCESS in 1h 20m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5bdb3afd7ef94b929543acd17dc65625 : SUCCESS in 1h 10m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4e2292df101e47b0a467d6a5b57031e0 : SUCCESS in 39m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3192209482f443ca73f80ac842d8fbb : FAILURE in 39m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdee4b25ee624b14ae43d3ad2b4f3607 : FAILURE in 27m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ada61196b7d4f6da3158b8004797411 : SUCCESS in 11m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/267869403593411d86f1840afaa1b2fc : FAILURE in 25m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5f7732b299042da9c4998cd762dc63b : FAILURE in 33m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52f8863bd6dd4509a91612ead4ad5abb : SUCCESS in 11m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ce9ba0d758c9443fb5469ded2cddf14b : FAILURE in 1h 23m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ac6b3ffb16154528a2340272a157e5bb : FAILURE in 1h 26m 10s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a4fe4ceabd0f4ac89eeb1bcf30adf35d : TIMED_OUT in 2h 15m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/84e73d0a47e7402f91903a77bc76f71b : SUCCESS in 1h 10m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7d08f7cd82e746c4b8bcdf3ac98b4cab : SUCCESS in 1h 51m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/11cee39334344512b7cf3ec5bdec49d8 : SUCCESS in 1h 12m 40s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4852bd88a6cd41078629654e8108591b : SUCCESS in 1h 44m 43s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3163690f874241cc9f1331481c248348 : SUCCESS in 1h 14m 07s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f8be534d80654be397ff7bc46c174641 : SUCCESS in 1h 32m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/511f65c9806940389aa4a56880f3a490 : SUCCESS in 1h 04m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7ed031ff4bcb4bfda78d17f76ff24960 : SUCCESS in 1h 33m 16s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a52f364b6abb41b1b8c9f018d9fcc33c : FAILURE in 1h 33m 10s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5095efb29b8a42bcb0b5d302bdf92cd4 : SUCCESS in 1h 20m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f89bae05c6a64c6189bdab7cae2461f4 : SUCCESS in 1h 04m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3c86c0eb39ac43139ed059b1ccdddc1e : SUCCESS in 2h 23m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/170191df715c4d2abc10e4e631ff9789 : SUCCESS in 39m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1b537ef50fca4d29909b7bf8fdb7644c : FAILURE in 2h 02m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5b0dc0d0decf455088146171e83893dd : FAILURE in 1h 16m 02s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/20dd1e452ee14965b7f03c8b4c44c9c7 : SUCCESS in 58m 42s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8acf6b7fcafb47638ed9beffd4b1a565 : SUCCESS in 2h 05m 11s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c29b79a81a3c4e7d8510250d1423afb1 : SUCCESS in 2h 02m 59s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0e4762db72f0484c8cd09d25ccaad67f : SUCCESS in 1h 50m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ea4096dc720c43769501ec406e31c413 : SUCCESS in 2h 03m 23s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/bfef7d765be34ea58fc76e70dc35c1f3 : FAILURE in 2h 04m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"15784ad49dba42c483c397e3d2f571a40956f8ed","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-01 12:28:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e8f8c96221457584c15548e8ca6e95dfe0147228","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-01 13:42:24.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/June-2020/Arista_Tempest_Testrefs/changes/46/731446/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"34ed320256895bc1153734598608a7be7314bb55","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-01 14:25:31.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/4 : SUCCESS in 1h 56m 16s","accounts_in_message":[],"_revision_number":4},{"id":"26cdfa60cce46363b32a88951568de4c8477d3a2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-01 14:31:27.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/46/731446/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/46/731446/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"198df431226f185c25329ba4f4dfaf1853801e25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 15:12:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3bf260fcf3dd461cacdde8e3c62f344c : SUCCESS in 1h 00m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5b4d46ed32824e7eb69921b55d4979a4 : SUCCESS in 29m 14s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/5030061320f74757a653a56f04c45f58 : FAILURE in 1h 50m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c91652ac0d224ab8b707dde60ba4d452 : SUCCESS in 1h 02m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8aa79aab74cb49ea8fce34dcadfec412 : SUCCESS in 1h 02m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7317ac56c947482bad1ab6dc2fe72c8f : SUCCESS in 1h 00m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/607b5220bb1e44d8a0c2f7d6a9f09c92 : SUCCESS in 31m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e866c2796b0b4531b311bbdad76904df : SUCCESS in 31m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79fcf40ed8c24e27b5b1a9010bec2fef : SUCCESS in 11m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a085e8b0bf5e4b6790050fad64857e2a : FAILURE in 27m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e65f2c9eaa5444b6ae11596a5862d009 : FAILURE in 25m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56ab9e77e342452ca91b46039c334f50 : SUCCESS in 9m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/de50d3253c45405c861f46d93dbebc5a : SUCCESS in 56m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/535a8daaffd74f79a8ea087f6d8ec7df : SUCCESS in 52m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/72e7915a4bd24b258bec652dcd7e552e : SUCCESS in 1h 11m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6ba752b29dab4d439f024c60b16099ed : SUCCESS in 1h 14m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f5b77f3030714ff2b4bddd6e411b9d07 : SUCCESS in 1h 47m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1bd7aa7a1bc04454a3d38a7d0f545875 : SUCCESS in 1h 12m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/05df57cc438d416d821b2177d7269fe4 : SUCCESS in 1h 15m 41s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/56591b9259eb483396d05214f3bd32d2 : SUCCESS in 1h 40m 40s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4169ee0be2084f72aeccf24049a1bd12 : SUCCESS in 1h 21m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4604101c597741278a042ea6aa36ea5d : SUCCESS in 1h 33m 08s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/06bea6126cd74b5782be69e4ce1cd2f5 : SUCCESS in 1h 42m 13s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b8581dea7e3a40f1b621dc8f087d9404 : FAILURE in 1h 51m 10s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d6d551beb3c04c08a9ea9335dfb3afdf : SUCCESS in 1h 08m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8f6db453af414dc1b836076460fe9fb2 : SUCCESS in 59m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b0d81356153c404cb086bcee42f2b949 : SUCCESS in 2h 07m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2acc0fd905f2425b88bd8b4c9ad6ccd8 : SUCCESS in 57m 40s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/98df43f8d25a405ca03edef565a310ef : FAILURE in 1h 53m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1191a25c7b2d4d6a8c4903bd92bbf641 : FAILURE in 2h 07m 00s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4046d1ea576f437d97162e0bf2285260 : SUCCESS in 1h 25m 27s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5ee123bc5e0b4c8ea7ca0ddad5a9a4d3 : FAILURE in 2h 01m 59s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/67635d2761074026afafbd23b6000e05 : SUCCESS in 1h 56m 20s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/43302390b34e430bb56481ed8579a52d : SUCCESS in 2h 38m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9b4c227a8b304f85a191f2964243d034 : SUCCESS in 1h 42m 45s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/5f049d396d864fe4a1631d33128e01da : FAILURE in 2h 15m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3c587c79cee60b421fefd1eeeb75d7b03f563e86","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-06-01 15:21:14.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0a5515cf268c78ef3ff2ffa22a948cc34a1ab9bb","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-02 01:47:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ff60353451c6c4f4528311eacbb4d4a05e3f86a8","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-02 02:25:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8c2f0876d9731296ceb381dc21d298ee321a07fc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-02 04:06:53.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/3 : SUCCESS in 1h 40m 50s","accounts_in_message":[],"_revision_number":3},{"id":"46320f386de412015bcdadba8793452def34f77a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-06-02 06:14:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"da62364c6b6c77c4985ce9c84b57fd42ce29e845","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-02 13:44:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6bd55daca04ba1c03097e0e795ac42f99760b22f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-02 15:43:29.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/5 : SUCCESS in 1h 58m 32s","accounts_in_message":[],"_revision_number":5},{"id":"e04ac03fbbb22d40942879591b4de5b08b982dc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 16:32:26.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e637815e2d754dd787400b63f3c62050 : SUCCESS in 46m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b76faab66bd34de19e27ebf6a1e55f43 : SUCCESS in 35m 35s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/09dbce75252f4e2db89a91efd5909c18 : FAILURE in 2h 17m 12s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f20283ea911840febe25f9f25459c3d3 : SUCCESS in 1h 55m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cabee21e707b4e24bb10c40185c21699 : SUCCESS in 1h 02m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/91d3901c6cb54f06b839392ab2b6ee34 : SUCCESS in 1h 14m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59555089c29c45a58f827e8ed38ecdb1 : FAILURE in 32m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ab224d7f92e4bd69e08614412244224 : FAILURE in 28m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9c42a5aa01b4321985b476c041955bd : FAILURE in 11m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd8d16a109134005a3bb62cdb046513d : FAILURE in 25m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5104c8291034f2f92d05f40286c5992 : FAILURE in 24m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e128b8351e5c4f8f8217f8f04da742f4 : SUCCESS in 9m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/916534cbcff6485496ba1d305d2e005e : FAILURE in 57m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0a0d6b4c171742f0af36c4c4ae2c1e7e : FAILURE in 51m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5495ceb9c8004608a44243fcd86f9c33 : SUCCESS in 1h 00m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d23631d423a4bc980120b1dc5d8523f : SUCCESS in 1h 19m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/917d86d46b854d63add9ba376af2ad19 : SUCCESS in 1h 33m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/79a3a493faf94e8e8e4530fa47f7d24d : SUCCESS in 1h 10m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/909e38e3ab944dc7aeeeafa915d78583 : SUCCESS in 46m 01s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a7eaad4ff1f746f9930a3816d5a2bbef : SUCCESS in 53m 21s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e9f005c554864b7ab5088cd5e4db85ef : SUCCESS in 51m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/61cda183318d4d6a8ed3c02bc039802f : SUCCESS in 46m 23s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a19d8c2c4daa4a659c06c173bcb77dd0 : SUCCESS in 1h 07m 40s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e89ae766a8a54542808e5ad26096848c : FAILURE in 1h 00m 28s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/45ae37ff7d3d44fcbb9165b0e86fd6fd : SUCCESS in 52m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4f8ac096b9af48249adaabf9a2901378 : SUCCESS in 37m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5ae9762c032141cf95a66a9324e232c0 : SUCCESS in 1h 22m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf3c39b56b5c483c8355e19e5b77b048 : SUCCESS in 39m 40s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bd7addb724104e74b88515e98be38d00 : FAILURE in 1h 13m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/94bebeb2bcff4c899c1b2d5a60039ffa : FAILURE in 2h 04m 10s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/705526e16ec1463c834d3ac8f794e519 : SUCCESS in 1h 31m 45s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c3a9992e0ad640e2af280bb0a8c407bf : SUCCESS in 1h 46m 01s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cce309fe2472499189c2ad29af621bae : SUCCESS in 2h 07m 47s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/52d16e848ae34cf5a968122c3196e462 : SUCCESS in 2h 46m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/70c1e11a9743444b89b8cf158371fa45 : SUCCESS in 1h 16m 37s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/0a976411bbe94e178e74c090b26027e8 : FAILURE in 2h 20m 54s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/logapi/drivers/openvswitch/ovs_firewall_log.py","accounts_in_message":[],"_revision_number":5},{"id":"f4ce2e10a1e0e73155ecdc8f959892946eba44cf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-02 18:05:45.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/June-2020/Arista_Tempest_Testrefs/changes/46/731446/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"9d1c9419467ce0a08cacb441c8f08d323e1d6bde","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-03 04:21:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6de32e4ff199070f6e7179b82c615dd1373629a6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-03 05:37:48.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/June-2020/Arista_Tempest_Testrefs/changes/46/731446/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"e3b6e52a58ad4a49fcde345e5520f9710b56c63f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-03 06:12:58.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/6 : SUCCESS in 1h 50m 38s","accounts_in_message":[],"_revision_number":6},{"id":"92624fd3a03a207f2db258f2220890fcc02182c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 06:54:45.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c90e6464ca66467eb0d9d3113a0114d4 : SUCCESS in 37m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0920de44627d4c638c43c0d0c46621f8 : SUCCESS in 26m 34s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/403753adb47b435292df18f29d9e2097 : FAILURE in 1h 31m 17s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1ff5f045ff4c4f3e859a83f4c120e309 : SUCCESS in 49m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1431416a8b344d9f98e4de3b2fd6ffc8 : SUCCESS in 46m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1cacd33d9ca340ceb10c4225b901c5f2 : SUCCESS in 43m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1a0aa05c4044a4b89df387b5f5cdf7e : SUCCESS in 21m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22c1f267bd5d456d8eb250894c836558 : SUCCESS in 16m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46bc99a246784652a7401dcec7a0b46f : FAILURE in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6d2312fa80a435980bbf53d3dfdf87b : SUCCESS in 26m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06fac6fa19d747c7b3fd597770c2bb01 : SUCCESS in 25m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82561fa88fd64ef3b03109f4867678bd : SUCCESS in 7m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/732683f6c4a840338717be42eded8dbf : SUCCESS in 51m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/09181a7a09094ee5af728431ef91f491 : SUCCESS in 51m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/bea296df0c2643be9f8c3409460cfe95 : SUCCESS in 59m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/09736c218bae4845a3265675694f4e25 : SUCCESS in 1h 08m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5fe5026b448643b1bab6f53f12ba0804 : SUCCESS in 1h 49m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1428c9af0ece4ff5bf4eb0efec3e96e1 : SUCCESS in 45m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/098ed5a87bea4a12b12ca97bc44dc09e : SUCCESS in 1h 05m 35s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/fd14150f22764bb295239735c25e3b22 : SUCCESS in 1h 09m 42s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/dea9b6bc267d4ef396e211f4c0beaac8 : SUCCESS in 47m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0c722a12c14c4fa4be6b7d47f5b8be66 : SUCCESS in 1h 07m 09s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b3e2ca64ce6e4f89b1f0319bf0fbec8b : SUCCESS in 1h 04m 14s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3e39100166e946fabf2669609e420bb8 : SUCCESS in 1h 37m 31s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/04101e9ff1d6484bb414e12a8d599420 : SUCCESS in 1h 08m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1a6da9dbf09b4b7c8e45744a8da26265 : SUCCESS in 54m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ddec958589ca48ee9d24513bb6b7d0dc : SUCCESS in 2h 01m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93ed662c7b76495ab9b280ebbf633f41 : SUCCESS in 52m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ce4ce4fe286941b4aeb0b90329a5971c : FAILURE in 1h 20m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b64309760a6c4a5f81e23d349a04513d : FAILURE in 1h 44m 18s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/255210cc31d041c59fd3bc2aa4757ee4 : SUCCESS in 1h 14m 24s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5a46ac63db834395b282c6f3ebddc5e4 : SUCCESS in 1h 50m 51s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/56748bc33a424172904abaaf54ec2d05 : SUCCESS in 1h 14m 53s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a07aad8290084fbbadf3811410aa25f8 : SUCCESS in 2h 28m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/48fb666bdd944605bc23cc6de5c48c4b : SUCCESS in 1h 39m 26s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3ebae54000cc466bb304d77c7633cd4d : FAILURE in 2h 23m 28s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/logapi/drivers/openvswitch/ovs_firewall_log.py","accounts_in_message":[],"_revision_number":6},{"id":"2d74f60eab5ac750afa025516ff369fb3acde083","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-03 08:36:42.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"36c50bb7b6f3c40d70633e1fba875ed785ec6eeb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-03 10:18:50.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/6 : SUCCESS in 1h 41m 39s","accounts_in_message":[],"_revision_number":6},{"id":"c2b20832ebde488adda1ad75a112d12386da9eba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 11:43:35.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0d55d0cfd01b4481a412e31950848ab8 : SUCCESS in 54m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9c509e02a2be4a028fac64b9cbbf767d : SUCCESS in 40m 04s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/c0ee65a2041b4efc8d051dcda9b13391 : FAILURE in 1h 33m 15s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4cabf1ac6efb4683a39efd91d353417d : SUCCESS in 45m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b6436ed358b9421295be1c13fb1ef5e5 : SUCCESS in 39m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ed4325ef2d4147218cc452b371d9c05d : SUCCESS in 43m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a291dfa6526841babfd70063533adff1 : FAILURE in 32m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebfdbc3a04e747e893c973c83899cb80 : SUCCESS in 28m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e10f7da91a24e56a51dc9d928de9c88 : FAILURE in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d1112adcb3246c6975164338c5604a8 : SUCCESS in 25m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b055e38b950480a986d60653122c2d4 : SUCCESS in 29m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/424faafb56e94ec5b229ce17e9e0d2f6 : SUCCESS in 8m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8064826efaac4a14953ec0b9c51ba75d : FAILURE in 58m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/773a6d3843134166a819374ecbe63933 : SUCCESS in 1h 00m 34s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e5ce156c56c04570a1ea1abc96c97311 : SUCCESS in 1h 17m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8892f1bf3c7b42f5a6e838bfc98d5ff5 : SUCCESS in 1h 09m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/07871f75922e45c88d2a32968662aa51 : SUCCESS in 1h 57m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e7c9707da6f4a809fbb0596e8457cc0 : SUCCESS in 1h 15m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/571fe814fb8f49d984ea0092ebd6adfc : SUCCESS in 1h 15m 48s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/14c093953f054bdb9371792b0200affe : SUCCESS in 1h 26m 30s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e265da91528040c38f4a03afe45e4c0c : SUCCESS in 1h 22m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0774f5a37d3f46cd81cf6d4138ffad71 : SUCCESS in 1h 16m 58s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8133303dc7c54f71ae7c1b05076de7be : SUCCESS in 1h 31m 24s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e1d1aeec549140d888ec083b424966ea : FAILURE in 1h 19m 32s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d5371008f37c407b92bd659fd09c8198 : SUCCESS in 1h 14m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c2890e891f494d269966eb82d39eeea7 : SUCCESS in 1h 14m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1ef80a4cc4d04d7e8e28984fb9e99153 : SUCCESS in 2h 12m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fc08c208ca34a2fa2aa0a913bc6f5f9 : SUCCESS in 57m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/50c594ce788f4e48af75ffa100f1af8e : FAILURE in 1h 47m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1920b12913c04455a07ce1cd169c8621 : FAILURE in 1h 56m 06s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/355a5c4ee8b44e9681107f76554028e1 : SUCCESS in 1h 24m 01s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6d9fbc6dd281434aab0edd659bebfc8c : FAILURE in 2h 16m 56s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c9be7bd298764891a5e11041ff73aedd : SUCCESS in 2h 02m 33s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f2bb9661d7a04a5199221bd778eecf53 : SUCCESS in 3h 03m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/465c5c788f9f40c1a45bf2832bd73031 : SUCCESS in 2h 21m 47s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6a831a3d953e4b21a04fce82c08fbf89 : FAILURE in 2h 40m 22s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/logapi/drivers/openvswitch/ovs_firewall_log.py","accounts_in_message":[],"_revision_number":6},{"id":"be22caaf1c6538e52bd8eb5eeff175509ca78f1b","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-04 02:21:08.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"5e13e4f327454b98b5e2ee53e281f4fcfc711bbc","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-04 02:33:24.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ec3d917c3650b7d976caca1ec0c93167931f2e1b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-04 03:47:23.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/June-2020/Arista_Tempest_Testrefs/changes/46/731446/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"65a84789334e4b0e68f1e03b4de13c5f23b1d551","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-04 04:15:21.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/8 : SUCCESS in 1h 41m 07s","accounts_in_message":[],"_revision_number":8},{"id":"32ece3ae7b66c6d9ffc38d397e0bb948f934359e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 05:04:51.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/31aefa71946f4b16a4b888e99e67866b : SUCCESS in 53m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/beb2681a889d4656b5aa868795a5606e : SUCCESS in 37m 54s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/d8e42242730642ffb22a5148827b545f : FAILURE in 1h 52m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a1c2e1bdc3894982bff1cc20c046d6d4 : SUCCESS in 1h 09m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dbbccc6fa2fd4ade9257d8b8e8a59cdf : SUCCESS in 1h 12m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ddd96dd2b84a401cbf4cbb22ef1d6e9b : SUCCESS in 1h 13m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1bd1bc9fb1d43bea6ee8e5f903d8cd1 : SUCCESS in 33m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d76b775add5747f9a7e059c58cf60a96 : SUCCESS in 26m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f67d7627c1b44c6860e5c529fe37628 : SUCCESS in 11m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3eb6869c7edf44a0bb1d17bb10d58940 : FAILURE in 25m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f316f97aa5ce4b539f875cdc23064411 : SUCCESS in 27m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a28554eef574b8db05bb741ad403c25 : SUCCESS in 8m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/10cc443761564b40902c23ce46098335 : SUCCESS in 50m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8b2cfc4f45f5437cbb39f9d26c822295 : SUCCESS in 54m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1e7d182ac4e744b3a10a911ce8c535e0 : SUCCESS in 1h 16m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5cdc9b5f85ce4da2b8d2d12c5f7a9b5c : SUCCESS in 1h 19m 04s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/88dc700766514acaaeb0d9ec1315bf7d : SUCCESS in 2h 08m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c14c4dbb6efd41c2ab410a94e9198bbc : SUCCESS in 1h 10m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c3e127c53ad34989835bd69f2e11bbec : SUCCESS in 1h 12m 37s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/03ba6082004c4e168888d29ac5072b5d : SUCCESS in 1h 05m 31s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f791a24e8718477a8b42520e4aff1780 : SUCCESS in 1h 22m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a86d220c9a7c4467898b9b30443038c2 : SUCCESS in 59m 35s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d687022cb29e41c7a2de46a99582d3be : SUCCESS in 1h 41m 13s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d7497b96e1c74321865957c2a0158612 : SUCCESS in 1h 46m 30s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9008dc9ed8714e55ae1f4fffe257b3f8 : SUCCESS in 1h 01m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a24884a035b0435885ae2d8fddc77022 : SUCCESS in 56m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e4d07dd13dbc4f1297413894271a5295 : SUCCESS in 1h 51m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5e44a3636ad450d96d9d080f37908db : SUCCESS in 54m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2e9c66feefd24c0ba2c611a4d395214d : SUCCESS in 2h 14m 06s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/657994304b1d4eb9a64d25a6816c4c25 : SUCCESS in 1h 51m 47s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b3669011aee7471d9ba6b4a578811c48 : SUCCESS in 1h 24m 41s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/32aa90215c734b0e853b1e73cd0ed11a : SUCCESS in 1h 42m 54s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f77cc80446bc4b33b56c61e1bbb5e0f9 : TIMED_OUT in 2h 13m 26s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0b82e05c4e6f432393022fed4625bd74 : SUCCESS in 2h 26m 38s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6e72577dd62a42f89fb0871d3aa32b00 : SUCCESS in 1h 43m 22s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3ceef3253d544bfda8dfd17478b8ca34 : FAILURE in 2h 19m 52s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fbf91b141badbc1f0051d466702f706096361f5b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-04 05:13:34.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/June-2020/Arista_Tempest_Testrefs/changes/46/731446/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"3969fde73cad6345b3d37be936308ba80a41d816","author":{"_account_id":30744,"name":"uchenily","email":"uchenily@gmail.com","username":"uchenily"},"date":"2020-06-04 05:43:00.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"6ad8b535788843d528f580dff10c6815e2549d9c","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-04 06:37:33.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"3806fbfc504043fb6f65a3a926b70a995aa6e2d5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-04 08:18:26.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/8 : SUCCESS in 1h 40m 20s","accounts_in_message":[],"_revision_number":8},{"id":"9d6d8c95e8042c79c2e8be217d0e5b8e3a46fe26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 09:29:10.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/952f78ea7e1149f596ae5e79ddd1221a : SUCCESS in 50m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/796d18a29f1349988633fb7d690f283c : SUCCESS in 33m 01s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/8327ce84da2748b4b94c65edeb109611 : FAILURE in 2h 05m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/69b79e6aecd145deb1543191c5493daf : SUCCESS in 1h 25m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/97400416a25546a49c9626e52fce7155 : SUCCESS in 1h 29m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/85d6349073094907814a1fc96ec9de74 : SUCCESS in 1h 17m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c053dcf2a4944960bdba3f94d673d7c9 : SUCCESS in 44m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2a723b813884ad0baaad7747d08fc93 : SUCCESS in 34m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68ba777840cc48cab1ab1959fb3aa540 : SUCCESS in 13m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c6f97019f854fb3b3cf680a9b5cbd4d : FAILURE in 26m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25edda92fa31447a8f72ea81eb960a9c : SUCCESS in 26m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25e8973960a942379224c1d8c5aa6606 : SUCCESS in 9m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cdd94e665cdd4074af32d85a926189a8 : SUCCESS in 57m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/485c54fa86314d6dafb7b518f420f80d : FAILURE in 1h 34m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a6a89a8743b448f7b6e4852a4a226bb9 : SUCCESS in 1h 17m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/06979ecef0f0498597036691b9f10d6b : SUCCESS in 1h 15m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/81a2edb64e194768817c518e40d5d094 : SUCCESS in 1h 41m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/291a6303fc524da6ab0e1aca6ad15b0a : SUCCESS in 1h 19m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7d3e211c2ed64a0b90d67e847ed4d0c2 : SUCCESS in 1h 13m 52s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/959326e374fe472daf2b7402d676aa48 : SUCCESS in 1h 21m 57s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c2dea49ebab54f3b9437d4f23200f8e8 : SUCCESS in 1h 30m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e7b16355290d4e0293ad56fc1ebc77f2 : SUCCESS in 1h 03m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dc9c6279bc354f74b42a11aec66e16d0 : SUCCESS in 1h 49m 25s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ee5a59a265e5410bb38107e96f0ea1ec : FAILURE in 1h 48m 41s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/01b04a12c6c54f0ab5bb2812a0e560b3 : SUCCESS in 1h 04m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c414abd57ed74c22b37d71c93030a392 : TIMED_OUT in 1h 34m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/634a9e6837fa4b0097fc0b7f29e037a0 : SUCCESS in 2h 04m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0059c3e09e3d4764b32da73971bbfc1d : SUCCESS in 1h 00m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3f91c503542d48f1a603fb8074503776 : SUCCESS in 2h 05m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5c59e9d2d257460eb49d2440f460c244 : SUCCESS in 1h 50m 01s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3af036d5a9524a16a8e4573cc83890a4 : SUCCESS in 1h 14m 28s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/503a72ec668f4f19b54e8ef4fed0c10b : SUCCESS in 1h 47m 01s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f66473fa02f1475fa2265026df2e6c28 : SUCCESS in 1h 50m 32s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/be81f2eb09f44cb79e15dfaad54060d5 : SUCCESS in 2h 27m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bb5711a2ce2346c3afccff24751201d3 : SUCCESS in 1h 47m 51s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/97fce345252e49129e3770745dd665f5 : FAILURE in 1h 59m 37s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fa044c11e08b599fd58986ff3622267fc24d8df0","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-04 10:17:10.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"86796694450fa5b1556fef4bf2fb47fccb77566e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-04 10:25:42.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ http://swswe-opsci001p.aristanetworks.com:8000/June-2020/Arista_Tempest_Testrefs/changes/46/731446/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"2f84c5e862bf1160d32b3cfe47b3f227889508ee","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-04 12:04:36.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/9 : SUCCESS in 1h 46m 33s","accounts_in_message":[],"_revision_number":9},{"id":"2f8612fc9653cb80a3de67ba816ef0c98620801e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-04 12:29:50.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/46/731446/9/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/46/731446/9/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"c54961a02547e1731dc5894d12ff407d78451540","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:03:55.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ae934c7263574864a44c79317f6fe185 : SUCCESS in 45m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5d9b7792b67447b4b29db91fbd7bd9dc : SUCCESS in 36m 25s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/f6d96ccdadc04f5db9d8afab45c80160 : SUCCESS in 1h 58m 17s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/dc4b3e05698840edbdd762c4ef922e98 : SUCCESS in 1h 33m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c2975dccc2764260957221ed3d75e8c4 : FAILURE in 1h 45m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f064428ddeff451eb892ed54573c9b84 : FAILURE in 50m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bafffc77be1743388601e9f1cb9c54fb : FAILURE in 44m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/507c241c80df40ab88a0bf82e62cfd5f : FAILURE in 38m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ced9bfe7fb540a2b686207110fc0f86 : SUCCESS in 14m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8be2e59c6fdb4233b0dc4615ee99ca40 : FAILURE in 29m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3624fb86900148778a341d011ca8f7b1 : FAILURE in 30m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d58e349e12a4adab54502cba7a45a01 : SUCCESS in 11m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/47ddc56f499b45db9d44f02748902c54 : SUCCESS in 1h 01m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/711dc38007cf4cf6ba7e369e3f7b0623 : FAILURE in 53m 55s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/59dc0425f0ac4b48a7b5b536ac4399bd : SUCCESS in 1h 21m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4096274bdac34d65a3d836ad30c0c9f5 : SUCCESS in 1h 18m 14s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4c67be6ca9dd40ca9b63786994d9b0b6 : SUCCESS in 1h 49m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c08c19e04054a2482b168857ea7ec92 : SUCCESS in 1h 05m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c66181c522294a8bbbca2d10676d7326 : SUCCESS in 1h 17m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/760e448230f64108a51e851bbefb869e : SUCCESS in 1h 08m 55s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ac85dc20e8f84ef781cf55d85d822827 : FAILURE in 1h 07m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3bcd1f9fc8db46dbb4cb2de85dfe0ee8 : SUCCESS in 1h 06m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/83ea865499584a47b60ec9283d512895 : SUCCESS in 1h 35m 04s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/54cd06de4b1f44a7af898852c9b159a4 : SUCCESS in 1h 58m 48s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6b9dd95c46d94dae9f543bb3872f6991 : SUCCESS in 1h 09m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7c86384935384fcd96a4f4c18455240e : SUCCESS in 1h 09m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a56bc80a8cfa41978519e31a9b7f503b : SUCCESS in 2h 10m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bccae79d33d429fab1862425177b253 : SUCCESS in 41m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2069ce448dc24e658edf06fe454d58ba : SUCCESS in 1h 17m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c6951963131b41289f44a7096ab7bb16 : SUCCESS in 1h 53m 56s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c65bb50f2c494cb9b3689ba62abf6dff : SUCCESS in 1h 25m 20s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/24d91202d81e4e07acf0c8764cb3a881 : SUCCESS in 1h 44m 41s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/923fec97833b4f6cbc94f4a148f14f34 : SUCCESS in 1h 51m 35s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6b52922ed07f4792bbc5a7bd319b126b : SUCCESS in 2h 39m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6e80571ce54e434bae95f960851c6d98 : SUCCESS in 1h 49m 30s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/56a5709e41694f359c550b61217cca99 : FAILURE in 2h 28m 12s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e0949763f8d0030f098386fe559f6d208a7f6c64","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-05 01:15:28.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"deae7418d7f2d9783104d5345f2db1a10c6c4801","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-05 02:29:58.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/June-2020/Arista_Tempest_Testrefs/changes/46/731446/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"2f414885ee2778a7b478de5dc98d902e4fed3332","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-05 03:05:45.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/10 : SUCCESS in 1h 49m 32s","accounts_in_message":[],"_revision_number":10},{"id":"4eb1145f8955741cd3682e5990eeb56a0fe564bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 04:32:22.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/095c07b5acdc434a98e19b94ee650c5d : SUCCESS in 1h 12m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d39947990fac4c5ea061d7ec2e1aa7b6 : SUCCESS in 31m 32s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/24f55f9c91774035b03a693a277f278d : TIMED_OUT in 3h 06m 31s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/357626848532491bbc7d85679edc5e66 : SUCCESS in 1h 28m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1f5df8cc80004c60bc3cc642d28f13bc : SUCCESS in 1h 26m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4503678922b74ff881c5778812193b05 : SUCCESS in 1h 27m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b6de4c0a9da453e95eb3e880c371914 : FAILURE in 32m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44635a69716a4d2aafec071c534d3b6b : SUCCESS in 33m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68d284c12c9b4a90aaf9c563f3057c64 : SUCCESS in 22m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65672780453e45f59bef3d042e9a6f09 : SUCCESS in 32m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee1b47a88c5c4102bf33afc60791efaf : SUCCESS in 29m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4085dd73e9e54430b5f9e286febab2c2 : SUCCESS in 10m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/475378f0433548e39e56c821b71e40c9 : TIMED_OUT in 2h 20m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b85cc85e20854dd8aeaecf62cbb301ed : SUCCESS in 55m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9fd4896362ea400baac3b837240518bf : TIMED_OUT in 2h 19m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0ad4f40b07964f81a2aee3dc2ba25955 : SUCCESS in 1h 15m 24s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0e8b894ee8164342b7fa819f2bb0298a : TIMED_OUT in 3h 06m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/16d8eb3f8d2c484e9cea7881509b1272 : FAILURE in 1h 06m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0fccd6893f734c0e862e5a7934326281 : SUCCESS in 2h 56m 44s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ac8ccedff52642bda2350913a9bc3ebb : SUCCESS in 1h 47m 07s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/591eeb5c96e0453e840d101defadf08d : FAILURE in 1h 37m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f11c9399599f4920b99e32689af33eb7 : SUCCESS in 1h 44m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a4e7cf0580c4418e8ddcead9ef9362a2 : TIMED_OUT in 3h 07m 23s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/855a0b6375c44808abacc7a6f2cbfd0b : FAILURE in 2h 07m 48s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/38a5d159faea44febfdfc210d3c81592 : FAILURE in 1h 40m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/de51d30fe27541b2ac00afdc08a27409 : TIMED_OUT in 1h 36m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0877d47efc484a34b00897d1b5d09d73 : FAILURE in 2h 46m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce4753bf734947c299d39bf0a075d793 : SUCCESS in 2h 33m 05s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/22a321ca76b54d3daba3f8eb33369cc5 : SUCCESS in 1h 52m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4882c4d8de504a54bcdd2853aa0183af : SUCCESS in 1h 53m 08s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ac39dc16e87944399d4676de89e01f2c : SUCCESS in 1h 50m 07s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/02add6370ea849938e540d84a45b22ee : TIMED_OUT in 3h 15m 27s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/408a04cf99cb4651ad6dc6d65352be93 : TIMED_OUT in 2h 14m 34s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/30f31aa533ff451dbc3a3054ae960c8a : TIMED_OUT in 3h 12m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ce2a45fc23de4597a22e4a50ba20874a : SUCCESS in 1h 48m 36s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b0baccef9df0452b84a78df2c93d0f2c : FAILURE in 2h 33m 48s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"37be6f0201e9ce22b7eefd0f6c534aa77a5bf60f","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-05 04:35:36.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"56489ef6864376ccf539f9b37e974323eeb67759","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-05 06:16:58.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/10 : SUCCESS in 1h 40m 51s","accounts_in_message":[],"_revision_number":10},{"id":"096be0b780fa62674130ad495208f42b7d7bc0a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 07:34:13.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d38d01102c3b4ac1b7a9de61fa6b09f2 : SUCCESS in 1h 04m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cf234b19fea440f299359594b313ce1f : SUCCESS in 35m 53s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/77b82b2a998c4db9b8b1eaac9f319bf1 : FAILURE in 2h 30m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d1830d560bde45e6b9867583de1a5b86 : SUCCESS in 1h 36m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5501b370c56f4e70b550ae1a591e19b4 : SUCCESS in 1h 44m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/15324c4572fe482189ffad2e2206d37c : SUCCESS in 1h 14m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5224eaa8eef4730b61fc44d29752ac5 : FAILURE in 34m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1c25ab0d39346a0a358085cff30902d : SUCCESS in 28m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48b210819c4c4a719e9ca4d272e60973 : SUCCESS in 11m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1d9499d93674bf7a43e5973ced4be86 : SUCCESS in 25m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36750368cb554e27be8ad2bbfa1ba787 : SUCCESS in 28m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4610931514b44effad454b6cc443a8d2 : SUCCESS in 10m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2e563f26f74d4d029d03488154ca694f : SUCCESS in 50m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/40a4a9d8e25b47878be1daf717a74fff : SUCCESS in 57m 18s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f3b6e89b61c9490faec84b82eac0e8a3 : SUCCESS in 1h 11m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ffa328c0ce35491faf61f49f9a6127ae : SUCCESS in 1h 13m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f6e9760394864c46b546f5d98ed079ac : SUCCESS in 1h 46m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf4421b93baf44e1ac311f57b95538f3 : SUCCESS in 1h 07m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d4854f28d95f4a9093eea710179422f2 : SUCCESS in 1h 11m 10s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/dd63ec43d1ff406db71fa3392df621c7 : SUCCESS in 1h 09m 20s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/eaf84f163c2c43068729cbd93e592a63 : SUCCESS in 1h 34m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cc988507616445c3b181b0dafbc9cf68 : SUCCESS in 1h 11m 11s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/baa7b3c43f71456488b09cf13a1a5efc : SUCCESS in 1h 40m 49s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7ccbd81a98e04f469395f3c7277641ac : FAILURE in 1h 42m 30s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2943c2776f3349a080e7e5823d8251a1 : SUCCESS in 1h 12m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/295cf99df5344c3eaa63ef43bea6769d : SUCCESS in 55m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c3e5d0e679644675b21030be745b3217 : SUCCESS in 2h 06m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/685ddfa996e542d2aad1321a4fc8bae2 : SUCCESS in 57m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d000ba80e4484c86a46bb517c1752ee6 : SUCCESS in 1h 47m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/18da64a14e3f4e96822babb9d8085d69 : SUCCESS in 1h 46m 28s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/91199a83d1ca4d8284ff7a72f9b34364 : SUCCESS in 1h 19m 40s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/35cf2aba002b41c1b7f6d0ed6021981e : SUCCESS in 1h 52m 38s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/21825d71ef5a4ae4a518116c403d6733 : SUCCESS in 1h 53m 02s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/18dcb9a3642b4d8780cd5ed105331c9c : SUCCESS in 2h 52m 18s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c07ca96aa03a4b318ef80f1e594680cf : SUCCESS in 1h 54m 23s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/aea0ddaefef2439d8e508f0088a6b87a : FAILURE in 2h 12m 49s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"297a2561a94107f42cedc6facb164264056729fb","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-06 04:45:39.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"1e3bcff87f91961c6cbc37359663d0f9309e45a8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-06 05:54:18.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ http://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/46/731446/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"13bf9feffd4a6a189f57a056ee8cc82bba23c5ce","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-06 06:25:00.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/11 : SUCCESS in 1h 38m 36s","accounts_in_message":[],"_revision_number":11},{"id":"12e961d87e40296a173ea0f00e23396e967498fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-06 07:25:00.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a4a7ac4d17e54cde9af327ab6dae1473 : SUCCESS in 45m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f303ed9401fa4e108c635bfaf153dd30 : SUCCESS in 34m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/af0bb34b66c64b45a5c44357e8cdce2b : FAILURE in 1h 53m 12s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/38729b0424ee4244912b560053e805bd : SUCCESS in 1h 00m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/06b484adbfae449dacc60a87fe1d2669 : SUCCESS in 1h 03m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/38086bee53024450b97448778d3c85dd : SUCCESS in 58m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71d5d7448a264ab7a65744cec54c502b : FAILURE in 32m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b980fdb5586428897687fb5a998e40f : FAILURE in 26m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2be9f193c2145b48adee7378103c6b6 : SUCCESS in 10m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbca41ba04574af58af3b95d34016b3b : FAILURE in 25m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c82952c30e5e474186bce3fb30a6d42f : FAILURE in 24m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f21d41f32e944ae889cfad079d86e47 : SUCCESS in 8m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5e74ad0032634fe6a481733de4aad2b7 : SUCCESS in 49m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8cf0dccdd8284c99bad3b3a169cfdd00 : SUCCESS in 54m 39s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/49ef8e5591074263b01bce78a2014d72 : SUCCESS in 1h 13m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0161a985c13b40cdbe31fba8ec9d1aa7 : SUCCESS in 1h 27m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/836801a688ac4634a83888317469034c : SUCCESS in 1h 47m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/deb28d126747421c90282943be427960 : SUCCESS in 1h 06m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2966e984eb7646d1a56e826e96f57373 : SUCCESS in 1h 25m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ffe5cdb31798489bb1234b6f96c7b981 : SUCCESS in 58m 17s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/56233a65d1b14f8486af43038d128335 : SUCCESS in 1h 13m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/de028722ce604b18acb192f98f7aa676 : SUCCESS in 1h 18m 50s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d37c80cb045548899ab6510178ed43b4 : SUCCESS in 1h 43m 14s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/be3f7e84c3024a0f9eb58664161ebd37 : SUCCESS in 1h 54m 56s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/59fb26f09c8045e0b78d8a12104e3458 : SUCCESS in 45m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0e2dbe9a68d74b568fba9395f7013f3e : SUCCESS in 1h 15m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/549aaf8b9d894dd28cb40a6bd4fe420c : SUCCESS in 2h 15m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14722246449a46a4a84ceddf09b44b31 : SUCCESS in 1h 04m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a798609fc3c34a1eb43736c4d3c95401 : SUCCESS in 2h 13m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dfa3e0fd078c41ad89c51e3bb5505f0a : SUCCESS in 2h 10m 02s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c25229a0f866410d88e1014fa92a78fe : SUCCESS in 1h 29m 47s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6fe3697c8342422db82863486b9718df : SUCCESS in 2h 01m 18s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9863d7f261114f9f90b19664a59b13af : SUCCESS in 2h 00m 58s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f4366b10bebf4ce5b9b154c9b5822434 : SUCCESS in 2h 38m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e55901d70757460cb09096c2eec485ba : SUCCESS in 2h 02m 39s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/98f79233243749c197c60d1d5c91c501 : FAILURE in 1h 59m 37s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6e716adcf188a18af520cf35df0de24cbdd82b32","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-06-06 11:01:46.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-06-06/731446/10/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/7ebb526 : EXCEPTION (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c453ef74b56d1a332d8ef0dcb0973e83875a8571","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-06-06 11:03:23.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-06-06/731446/9/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/46f5b90 : EXCEPTION (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"07bf145367a392ec8388ac74d35e2d4e5136ba9b","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-08 01:19:21.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"99f9b59627cfe8c221aff077d5f326d35af2a54a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-08 02:32:49.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ http://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/46/731446/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"1af7f366e75ed3762a1f65ed806d3f5544fc7f0f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-08 03:01:06.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/12 : SUCCESS in 1h 41m 00s","accounts_in_message":[],"_revision_number":12},{"id":"de0f67491bbcb0c9942d1175e57a060a8e14ca12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 04:07:35.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c48ea99881544c12bcc76418b2918051 : SUCCESS in 57m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b4f9c9f0ce9f437693b2ed6dd77a036b : SUCCESS in 31m 37s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/bcd822095d5a4025bfc6f2a9221b62e4 : FAILURE in 2h 09m 31s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7f7e3796bce84ba0992ba71f4352cd85 : SUCCESS in 1h 10m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/15ae1e2cd1734b69b251e36748dbc6a1 : SUCCESS in 1h 29m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3bd4a3ec67ed4c19acdd12df7b990abd : SUCCESS in 1h 25m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae3244d798fc4873b88d343e0e6adf08 : FAILURE in 30m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff872470478a45459686344ded1f02bc : FAILURE in 26m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c3a77098a9e4968b4f791141a013c38 : SUCCESS in 11m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2246f5b5d5c740c7af39e0c6dfbe3e73 : FAILURE in 28m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa8db14614554fb4918164922b612f79 : SUCCESS in 23m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82b8d320298647ea816d6e86a0d0ca77 : SUCCESS in 7m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3710b5da8a674b4abdc89fe18e12f69e : SUCCESS in 50m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ba6ac70b6e904a5e82e393b6b8a781aa : SUCCESS in 56m 49s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/42f1647a2193476f8b174b8b89fc3033 : SUCCESS in 1h 08m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4e76a7a555864a45be3e61926d3b7171 : SUCCESS in 1h 29m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f8558d49193d49f3bd70e4b1b80d31d2 : SUCCESS in 2h 10m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebc424f040c5431eb0eaea3a09549072 : SUCCESS in 1h 33m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b6884a159bdd4e0283e024ebe6f1d938 : SUCCESS in 1h 16m 15s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1a5aea13357e459fbc330ec190a5f4d5 : SUCCESS in 1h 08m 14s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ea64aeb7b7643c18af7eead4fb71b53 : SUCCESS in 1h 35m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f859fb50b4f74b21bb4fcb4ab076ca54 : SUCCESS in 1h 11m 39s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/aa3375e22aa748c3900f16712e6b7e86 : SUCCESS in 1h 32m 37s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d255a96c16cd4bca9e2c59ec9725db2e : FAILURE in 1h 30m 08s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/68e3dbdd2b8f4e02ac493b1109b19162 : SUCCESS in 1h 04m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/65c6d27fb97d4985b3f70e113daa5433 : SUCCESS in 1h 26m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d700862fbbfe4aaebd83571216f676d9 : SUCCESS in 1h 54m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b5918c7247543be9b220885b8205b04 : SUCCESS in 1h 00m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c0c2db0af1424581a2cdc7357902d26b : SUCCESS in 2h 24m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8e5cee5b4ba44833b80cb2c32d51fc65 : SUCCESS in 1h 42m 42s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a26c0b1e9d824437837a734a065b7485 : SUCCESS in 1h 11m 29s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6bfa352a4d5d46a3aaf828d3d4ed8f67 : SUCCESS in 1h 53m 36s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cf71c2da758f479ebae81f37f1b3561d : SUCCESS in 2h 01m 42s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/903a04fd539748aea3177b7cc40b58bd : SUCCESS in 2h 45m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/79a1865217b3411b9f84061f222d3931 : SUCCESS in 1h 54m 39s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c2fa09a7e2884098b031779b904c5851 : FAILURE in 2h 02m 43s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d98ad75bb756f4c3088164c81a400fcc38a3e4c4","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-18 07:39:35.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"be87fc4b1acfba0598242e64f03594c1ed2e8c8e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-18 09:01:48.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/46/731446/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"ef61f5bd3f4ed021e2a033e954ae8c0d83300152","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-18 09:38:46.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/731446/13 : FAILURE in 1h 58m 19s","accounts_in_message":[],"_revision_number":13},{"id":"5e61b1edc86fa700ad3aa8e8faa8d8cdb0a0972d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 11:40:31.000000000","message":"Patch Set 13: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d61fea15ce5d476ca929d8f5e0289728 : SUCCESS in 50m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d435759ec71d4f8ab3eb96b606f300ef : SUCCESS in 34m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/3fba8d44d3e84571924452b5bda674da : FAILURE in 1h 24m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7ec38f6d63374cacab95aa14c0bc3219 : FAILURE in 1h 16m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/eeb521fca252421aadf7f3f6166b19af : SUCCESS in 1h 29m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3fb93d3622ec41fbb5a111876ec98082 : SUCCESS in 1h 20m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7439b2cd516b487aba391b92370030a2 : SUCCESS in 38m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edcaa5155b5f4161b3131966aebb1ecf : SUCCESS in 31m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fda03030d1c9411bb561b22fa10b1c57 : FAILURE in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49a6b30294474ee7ad477ed33f747100 : SUCCESS in 24m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1f3e940022c47b694f4c665b971644d : SUCCESS in 31m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b72161531ce497982accef5a667db19 : SUCCESS in 11m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e354a639f83a4eef975d0d3e0e1deffc : SUCCESS in 51m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a6aca988b2c74b6eba2c0bd289ae1797 : SUCCESS in 49m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6aba87bcb4514346964fbc492fa8dd7d : SUCCESS in 1h 01m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/80a11c342aae413bb09638c51bf1a02c : SUCCESS in 1h 01m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b611d54f6fb3442287dadbfd96cbf2e8 : SUCCESS in 1h 40m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b0cd8a76408424d80ecb41aa993c591 : SUCCESS in 1h 10m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/47eeac3d31d247d4ab566fe078b8a6c4 : SUCCESS in 1h 30m 54s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a206e4ae23044d33a19f118bb0764dba : SUCCESS in 47m 34s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e8120b51f4d46fc802ecd31a7df9867 : SUCCESS in 50m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f10259f05e904641826000078c9ec37d : SUCCESS in 1h 28m 30s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0fdb69c813ca404592309369647aec38 : SUCCESS in 1h 48m 26s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5eca46dbf9df43f78a854cccd52a9bfd : SUCCESS in 1h 42m 41s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/17eb999155bb4999bab99d699cb68713 : SUCCESS in 1h 02m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/51cb0dbebae049a7a67ca623bd7f53a3 : SUCCESS in 56m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/270b1d9cad3347259e7669370c5abc33 : SUCCESS in 2h 04m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f809b5410f44584bba2446429e22952 : SUCCESS in 51m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/63f394ffae8a4fc3923968cc1f940633 : FAILURE in 19m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1ab3943b27ac4417bfc5a342e37ad4e2 : FAILURE in 19m 49s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f6d082151baa4ecea4050077a2d10b66 : SUCCESS in 1h 14m 43s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/58bb20dfaa794292978530ecc42c2636 : SUCCESS in 1h 36m 17s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4649c9828c284ba49ccc2817936a50c1 : SUCCESS in 1h 42m 09s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a6704dc054ac419985fba70d21d9473b : SUCCESS in 2h 48m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cdfa61e93b3e4a6bb6e4f2f9a9ced984 : SUCCESS in 1h 45m 35s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b7aaab3bb76d46a9a74002c77bc516ae : SUCCESS in 2h 15m 39s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"59e8dd3ce19e2d804aed690150fff332316947b3","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-18 12:11:46.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"ecaca034c3bafd58538847693699575f8080e1a9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-18 13:31:51.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/46/731446/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"fd42dd4f6b7f8c637d319a7e1f3f8a825d324696","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-18 13:53:26.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/731446/14 : FAILURE in 1h 40m 50s","accounts_in_message":[],"_revision_number":14},{"id":"0acdc80bf863ea45d655677fa90128f556e92aa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 15:55:39.000000000","message":"Patch Set 14: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b30ebcfcb8754728a4d207f89e80e99f : SUCCESS in 54m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7ae2002f9f274604a37272fd80e2dc3d : SUCCESS in 38m 30s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/c8374fd270374663ac3886131b7f4e98 : FAILURE in 1h 45m 05s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ce3ada99bd8a40d5ba4d5a9a54b9cebe : SUCCESS in 1h 13m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9008adb089a94ee29aafc8f7843d0790 : SUCCESS in 1h 19m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a8ca45b7ba8f4e4cb8158b0d9939e785 : SUCCESS in 1h 04m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db6b8134dc2b40c693d24fab9d4bdec0 : SUCCESS in 32m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dca69fea95514355bb07b0408bdad908 : SUCCESS in 30m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a8526bd97c2477bac850c92f4be4c7b : FAILURE in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd9d60418d434a60ac601bc77ec918e9 : SUCCESS in 26m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe081455b8154d08bc4363f46bb9ab9f : SUCCESS in 33m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a458aed39c34cada79d9aafbdcd673b : SUCCESS in 10m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e7760d74454b46b28eb940925a63182d : SUCCESS in 48m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2cc701123d654a92a6e1d17c95df3531 : SUCCESS in 44m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b8904a748a73420686210c9ddc9c16b7 : SUCCESS in 1h 45m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0ba5c759f57d4b86b131e5fc142e6eee : SUCCESS in 1h 08m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5cfd5618b53d40108544b0dec36ea47e : SUCCESS in 1h 30m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4017023a719a43a2931120965295f82a : FAILURE in 1h 39m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e6913327401f4b53858f5e3a7f68ee05 : SUCCESS in 1h 31m 35s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6f7fd7339323431dbf70dc666a08e4a9 : SUCCESS in 1h 02m 43s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8dc5b1d0d3904f62a00fa39507292962 : SUCCESS in 1h 10m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/184d72bfdb824b5dafb50c5082b5c168 : SUCCESS in 51m 52s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a0d6be26c5d044e69f34b02879456a9d : SUCCESS in 1h 46m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/46188e7ca5c74cb4af25230f1cc8e13d : FAILURE in 1h 34m 30s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d537fb03291c475c8f7f0581188939a3 : SUCCESS in 1h 17m 01s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9650a80b890b4990a8249f76f05840da : SUCCESS in 54m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2a598a8677df46b686af66b179ffe3f8 : FAILURE in 2h 33m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e130c9516012408b97345f0cb947b6a5 : SUCCESS in 1h 09m 47s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/34219d3cd2354905b0326ad3d0f3a061 : FAILURE in 24m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/16b65021b7374d5b96171fbe662914bc : FAILURE in 18m 49s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e00840f42e41402f9091359632332674 : SUCCESS in 1h 22m 32s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c7b77241116d4b56a4aa2f26aea3b5a7 : SUCCESS in 1h 11m 38s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4fc59bc919d6410880c978d59bdfbcaa : SUCCESS in 1h 52m 37s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/97b78ef30fcb4ba8b82bc94e4b669924 : SUCCESS in 2h 31m 36s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c29d00c19e84476aba5bce40b374e88b : SUCCESS in 1h 48m 17s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/572a49ef254f4446972e2491cdd37150 : SUCCESS in 2h 09m 04s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"9bf96f4dcea8144cccc3676d568c26eaec88902f","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-19 01:11:50.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"5e8c35adb5cc99066cab369ea7ee59ad14dd0bc6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-19 02:30:33.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/46/731446/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"0fe0c67cb47ad5d245451e24a4de3e2342492017","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-19 02:54:02.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/15 : SUCCESS in 1h 41m 03s","accounts_in_message":[],"_revision_number":15},{"id":"ff676f851771d67b6ac7da69365a76c4028e61e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 04:33:47.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6db4418d623d4e518c200b78d075cca3 : SUCCESS in 57m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/107fd15186c14cc0a2c17cfab4c91e92 : TIMED_OUT in 1h 04m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1a89e05bba714e9b86ca58a530faa8ab : SUCCESS in 1h 12m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/999609497e574b6b9ec0fba9d5725f9d : SUCCESS in 2h 37m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9f5a192daf784f46b87db0477e759d3b : FAILURE in 1h 12m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de9f9537e4fa4747814e0f4ddd60b6c4 : SUCCESS in 36m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c33a3d4facbb4afa8ffcf2f7b78e8856 : SUCCESS in 50m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb4e3283e307416082ed89501f5de081 : SUCCESS in 10m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6c239e63737412f8f2dbfdc6efce096 : SUCCESS in 29m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba3aa842e5f3474e9b6951fa1a7072fa : SUCCESS in 45m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3df500b8645421685386e6dcaf2efb9 : SUCCESS in 9m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/837977e012b54a96be2ea47c8effca45 : RETRY_LIMIT in 9m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3a52b23692f9422bbc5146a1535ee8c0 : SUCCESS in 44m 03s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6cb5fe473fb34e6c96f3eee293be2bf6 : SUCCESS in 1h 13m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/afb6f6f32ed54b60bd638748f4fdbdfc : SUCCESS in 1h 10m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c3539edc15c84f6799bd6ca6811ebe1e : SUCCESS in 1h 50m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2f9cbbe4bad40929f6136ee5fcebf9c : SUCCESS in 1h 12m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/70848d1f22a84367a9da9f8c848fea31 : SUCCESS in 2h 47m 00s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ea46d113aff64c39a4f23b013115883d : SUCCESS in 1h 14m 06s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d9d558e91af34de8ae570bb06d55173c : SUCCESS in 1h 12m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/34f9ebbcdebd41e9b3ee0507dcd30773 : SUCCESS in 59m 11s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/82b63d2c6ba44400a0fc94e4cf8ec86b : SUCCESS in 1h 38m 12s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/62a709f0f2294728bb434fb7fa751080 : FAILURE in 1h 23m 36s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/481705465f56499ab46d2dfdd4b0635a : SUCCESS in 1h 08m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/baeed93a2508443682afc06677887d99 : SUCCESS in 1h 03m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ed32359c21124f1c906f1a5068591ac7 : SUCCESS in 2h 00m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2574d79d07f4d56b008eac1f14a8d46 : SUCCESS in 59m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/71b6bfb89a8a492d95cd001a3b88ec9b : FAILURE in 18m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/62501b53170745be8992347b7072ebfd : FAILURE in 19m 15s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7d7b498a55c04721830cdf95acacddaf : FAILURE in 31m 42s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6f4828fa4848474da1d2340a802bdb23 : SUCCESS in 3h 00m 09s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/abe41b18b4da469fa497f995eeb73897 : SUCCESS in 1h 57m 05s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a53d09556ee143c39ae52494aa4f4393 : TIMED_OUT in 3h 16m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a5fa4601fff54406b20ac2d15cf16c3d : SUCCESS in 1h 50m 24s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c97867ab16dd4cb7aa6f4ed1ea659fe4 : SUCCESS in 2h 10m 39s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"78c560d6bf663f7b3e48212fdf50451185e73360","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-19 04:39:55.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"1974c524038aa8bd8313847118202dde11152e46","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-19 06:20:53.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/15 : SUCCESS in 1h 40m 29s","accounts_in_message":[],"_revision_number":15},{"id":"739f13d7f9d2f7c6a0dbf3dc772f7001e373638a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 07:36:41.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a99dd5cec8104db99a765b9484f83ddf : SUCCESS in 44m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5d8a233504d64d5da4b385b6c34afbc3 : SUCCESS in 22m 33s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f34d34dce59645148f9e9198308239ae : FAILURE in 1h 18m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/68c7a9e38cc64c2198236fedef83aa8b : FAILURE in 1h 05m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f18d8a16a53342f486a4ebc33601fc5a : SUCCESS in 52m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bbcf017279b405493a8b61187ac70c3 : SUCCESS in 22m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f164f6483eb14c12b3eff9ea98bfef4b : SUCCESS in 32m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/530b52f146234a2c96148323f2a08710 : SUCCESS in 11m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c95bfe7be7604b8499070fa3c29eb94b : SUCCESS in 17m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e404feef74934e90b7f8b2352ab078bc : SUCCESS in 24m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3def7c41faa54cc391848fe24e21fe2f : SUCCESS in 10m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/eac5efe634544207b541bb482cdba166 : SUCCESS in 50m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bbf8bda9378141179d12c23330922bc4 : SUCCESS in 51m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0c1e81f452ea47dc8e549127317e4eef : SUCCESS in 1h 09m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9fa3693398984a158e72bebe0f7f92be : SUCCESS in 1h 07m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b1591fd65938426483261e8c2df75669 : SUCCESS in 1h 11m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e7718de2bd7427d87bc9c61426a423c : SUCCESS in 1h 04m 24s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ac36be25b95c440c8f528c2e70f6fc31 : SUCCESS in 46m 20s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/73311b86b63d495093d3ded9097d0b32 : SUCCESS in 1h 01m 41s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/09ec160d688d434ea2ffa1b728bd33da : SUCCESS in 1h 19m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/06173d66229c41aca49745fa27ab3d1d : SUCCESS in 1h 19m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c9db4f88f4e74a35831a9aabc3dfc86e : SUCCESS in 1h 41m 43s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8efa641f4a594bd5926dc181f205799a : SUCCESS in 1h 52m 19s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c936009d0abc44d5b822bdce0ba36c35 : SUCCESS in 1h 24m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f47eb1086a514faf8a6884b131577303 : SUCCESS in 1h 12m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/17859ac3e95842118b441ef0fbea3c2d : SUCCESS in 2h 08m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/330d66f7df9145d091c697096a00cdaf : SUCCESS in 37m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e4e7a124667440949cbea297cbd68e03 : FAILURE in 15m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/809e89be9da94809925a30bcb54bcf3e : FAILURE in 20m 59s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/56bae579149b4935b547d197b758e364 : FAILURE in 30m 51s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e7d0db1a44b74a3f94e0b73005eed6e7 : SUCCESS in 1h 52m 17s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1194de97b90a4d8fb144311d587f1eb0 : SUCCESS in 1h 56m 04s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7c9c20505a3749f8896ed9fe797f2e00 : SUCCESS in 2h 50m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/aec4e3dd71d34cf7b0b6510808abf438 : SUCCESS in 1h 10m 52s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/63546902bfc449989e293be03799ff85 : FAILURE in 2h 03m 18s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"55a83c6def72b4e8a33e85941cfba667c3541377","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-19 09:26:16.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"7f9ffb9124c2e29bf9dbc3abe73bb94839bf2ac1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-19 09:43:41.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/731446/15 : FAILURE in 16m 54s","accounts_in_message":[],"_revision_number":15},{"id":"7c98dbf830167eb7dd5b1a879effdae59c4efc44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 12:07:51.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6f31a8e8997e4487b99407f2d6034f0d : SUCCESS in 52m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0de989e7b36346aab022c39dd87eff1e : SUCCESS in 36m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/991d2836988345e0b958f63eca604a4e : SUCCESS in 1h 22m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dc1b50c71c104e6b89f55fff4f602235 : SUCCESS in 1h 01m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/692044d540c6481c93112b74089f8b97 : SUCCESS in 1h 01m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c6c6dd706964083949891eec39c46de : SUCCESS in 42m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a99e2afec5849749bde8b1d6d286dae : SUCCESS in 30m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a46dbc212f645cdbfabebad80c331cd : SUCCESS in 12m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/637465af544546c79abb26ffa907f7ff : SUCCESS in 34m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f66b0bde12d04a92871a7112dd7dc3f5 : SUCCESS in 26m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46b79d93f6b44dcdb13ff347ddbaffc8 : SUCCESS in 10m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2c7ba14679cd4a1e8db4724374c4c689 : RETRY_LIMIT in 14m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/80e35e5aa63547c5b2dff4dacb4c2a50 : RETRY_LIMIT in 11m 18s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fbc341be0d044afda725ceb260f7d1a0 : RETRY_LIMIT in 11m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6029e384df74420688f8eebd1d080211 : RETRY_LIMIT in 13m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d5887322df974892b1d0ae52f8def795 : FAILURE in 1h 43m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f5911fe3646948128fb095628723f61a : SUCCESS in 1h 06m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/943130ff559a465d9a33b7a6f03dceb3 : SUCCESS in 1h 03m 27s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/fd718cbec95946a98a3ea093677ae1c2 : SUCCESS in 1h 13m 58s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/612f6554306043a0ba618fca7a254a49 : SUCCESS in 1h 13m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5da8bea076014c14ba65d45da2fdfd40 : SUCCESS in 1h 21m 00s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8ebc7148c0f64717b8cfd095a930f597 : SUCCESS in 1h 34m 26s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/005cc7da28a5461f9c84cf17ef6ba0b7 : SUCCESS in 1h 56m 41s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ecfa7e7b8979482dae1177bb09f6d821 : SUCCESS in 1h 11m 35s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bcf0b7392c38406fae28190d52ec5127 : SUCCESS in 1h 02m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/657e53a047444a28aa6d4db9cfeef489 : SUCCESS in 2h 09m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b63b67f1fd7644d2b9b510f2605f7b73 : SUCCESS in 1h 04m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c29b9ab72f5949f3a84f9ffa6f95ff47 : FAILURE in 22m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bd192e2aa7eb43a5946bce67c96b8453 : FAILURE in 18m 43s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6505797fba584c6090aa8627be990bbf : POST_FAILURE in 23m 08s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3082772a232e4528b8703c18513afe0e : SUCCESS in 2h 04m 34s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3629d1892c384decb81c84c6609a150d : SUCCESS in 1h 52m 23s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/dcf863caf3be4c9cb5baf563c11038c1 : SUCCESS in 2h 27m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e77878f0613c42d1bcca32cc01cc0a90 : SUCCESS in 1h 53m 13s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/09bf497892fc494da830dceb5750145b : SUCCESS in 2h 12m 22s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"51d2c774e2687d7fe5d4bdbaed25ec5b55330eb7","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-22 02:15:05.000000000","message":"Uploaded patch set 16: Commit message was updated.","accounts_in_message":[],"_revision_number":16},{"id":"09e60b7e0656b10cd84c5cada41ed6eb3421a6f8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-22 03:30:41.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/46/731446/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"bf82117117b6a3c0c2d6de13858561be33a1c45e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-22 03:57:45.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/16 : SUCCESS in 1h 41m 51s","accounts_in_message":[],"_revision_number":16},{"id":"c01ccc3ce74019e427b2a17f21bc5466fc5bdee2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 05:21:04.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ab3a6a5cd501462bb8f01b2ef4c8d5ac : SUCCESS in 56m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/874da438c9084a268e6609ea21b5f751 : SUCCESS in 29m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e8b9da1099464449873542c967f84632 : SUCCESS in 1h 03m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f0d53141c31e4437b191bbf37b3031c5 : SUCCESS in 1h 03m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/20575cf616874a0689a44f294dd5b315 : FAILURE in 2h 24m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6d5cda97b6a4b519aa310562af26602 : SUCCESS in 41m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc2b7e994d9b406b8e7acce0fe455b32 : SUCCESS in 28m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f17ccdcceeca42cb8a97614ac8f8076b : SUCCESS in 10m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b15dbe26658f4351979f3a698239ed02 : SUCCESS in 29m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3246d6266e9647289b1dc9ea03b5becb : SUCCESS in 26m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88ed5bb35dea4995bd47731cb1ae2966 : SUCCESS in 11m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/18b09518016e46baaf44bfc997d73566 : SUCCESS in 1h 07m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c4fd2b4ab1e48ceb6dfd52045173f53 : SUCCESS in 1h 06m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5a9f4117c167406f94cfc6c2074d8ff6 : SUCCESS in 1h 17m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/04925d60472b4f1799e03dc502cc2995 : FAILURE in 1h 48m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8aec9b7ccd414fe78781009019fa61a7 : FAILURE in 1h 55m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/54a6972c54ea4467850b5ff6f204076c : SUCCESS in 1h 08m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b74c06d80c964a09834a713de07a4430 : SUCCESS in 1h 11m 33s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7b28c718516640fa8aab8e5e0245d9a3 : SUCCESS in 1h 14m 15s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/28738b12096f4e53a7520776ca6837ca : SUCCESS in 1h 30m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/32a137ed9df144b892bd2b2cad429305 : SUCCESS in 1h 22m 50s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e51d36f7ae684c36aef2a2ae5382d5a5 : SUCCESS in 1h 40m 13s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e9ba4e6e29544b919b00d5d7955c224c : FAILURE in 1h 16m 13s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e3c4898377c445ef89ca6083ee2f2bc6 : SUCCESS in 2h 34m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cbccaec98e814e82bf75418f7bff155a : SUCCESS in 46m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6c5c75d8bc244f5d9f6424d6b90641be : SUCCESS in 1h 23m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cf83e3590054c3ab327ac102fc10ec8 : SUCCESS in 57m 03s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a1771bf94bc749c3b3298344804e7b0d : SUCCESS in 1h 16m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/da3d7db7b574488bb9e069c384886ca2 : SUCCESS in 1h 50m 25s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3da031c286e947bfb7565f187f1258ed : POST_FAILURE in 35m 28s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bb6e84e5a385462f9eb405efdbc4caa9 : SUCCESS in 1h 46m 18s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b9a250161ff94f128feeff56c842282a : SUCCESS in 1h 45m 02s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f74b6adb9da74e24a5e9779f3a45270d : SUCCESS in 2h 33m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/39e38e1a7af5498baf6c795694339ae2 : SUCCESS in 1h 44m 17s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/df9d773771d146e7b6e5dc2abfab5204 : TIMED_OUT in 3h 04m 26s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"27c26a91154272677e96272e843528fe4d7d3f55","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-06-22 07:43:03.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"ab93968d21276c500411e6a7dab50ea254a5099b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-22 08:58:26.000000000","message":"Patch Set 17: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/46/731446/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"8b093a86469b691eb5b093544f3546809a4038bb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-22 09:24:00.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/17 : SUCCESS in 1h 40m 22s","accounts_in_message":[],"_revision_number":17},{"id":"8a624ae21518ff7e4a213a6598ef8d487b0a5893","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 10:32:00.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e6bbfc5bddce4c468f09bdf959f2c032 : SUCCESS in 56m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d16c061b9a9c4990b01ddce4367f1b94 : SUCCESS in 38m 02s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/702ee7a56b8c474f99e48071b5e7e7c0 : SUCCESS in 1h 36m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ace8b57ab4534db182176f38169b07b0 : SUCCESS in 1h 01m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/76832a0078d24c2ea3274f4589f17556 : SUCCESS in 1h 16m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abb761a7a45e427a923ee2886b3e66c8 : SUCCESS in 35m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f664e52cd3524b20b358caf6f9766239 : SUCCESS in 31m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37d96f019819423f9042a0bc21528f2e : SUCCESS in 12m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ca26c2ae59648a28fac4fb0a8d58be1 : SUCCESS in 25m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77e4247489c142cca6ab4c2cea88bc41 : SUCCESS in 33m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a5988d418d14981bada5e8bc6f4fffd : SUCCESS in 9m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4427c86ef39d40eea0c5ae05b4f057fd : SUCCESS in 52m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f7fb1e2b2ab243be90df3ba709964989 : SUCCESS in 51m 55s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fce96c4b515c4d109f23ea1a3913d399 : SUCCESS in 1h 13m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ac74439d10e44b028d090ff83b8b56c2 : SUCCESS in 1h 16m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/dc4a9b8b30e04242a8f0321c93fe9b85 : SUCCESS in 1h 52m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/50c7c00fae724b209eb08f4feb77242a : SUCCESS in 1h 10m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8bdbd86570fa4725acd6a6b62258f694 : SUCCESS in 1h 10m 59s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0d7fe423afe54d9f85a2286ce0c85d5d : SUCCESS in 1h 10m 06s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/fcf095c1da3245c4b31dbeaece7e7705 : SUCCESS in 1h 22m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a881151d356b4f47a603fe30867c3e97 : SUCCESS in 1h 15m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f5fb1c63327c402b89a029dfeda30987 : SUCCESS in 1h 32m 32s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/391d4f4dc1ac4455b27e24bed777b7b1 : FAILURE in 1h 22m 49s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/20438234c2cc4e9aacaf602b184d6fe0 : SUCCESS in 1h 17m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7b9ea807abf947d599e27fe0151cab98 : SUCCESS in 1h 10m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a5187e9feeb6477fb7955910bb510a11 : SUCCESS in 1h 59m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38d522d9058141a18cea3d29f0afcca3 : SUCCESS in 1h 01m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6cd0984e1eb145f2a690601360411de0 : SUCCESS in 1h 57m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/804eda392ded4e278763b08cdedab571 : SUCCESS in 1h 37m 18s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/03543bf86001439982938e4c7e8d3136 : POST_FAILURE in 28m 30s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2ad83fb18bed4bf4bd402778fe25b8d7 : SUCCESS in 1h 51m 05s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7daf3b2a0a7745ae85e638b9b98186ec : SUCCESS in 1h 47m 16s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/465da9c20aee42dc9663ff86f7880608 : SUCCESS in 2h 41m 50s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/daed3470202240ebbc1004a090489274 : SUCCESS in 1h 55m 29s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f48a12af4ebc4006b5a82f9166829789 : SUCCESS in 2h 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"8720f78eaa21b245cfa2cd176e66a71bacc15b70","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-06-22 13:11:10.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"1e4a9d9dd816bea3c3a777fc4282379e865cb9b9","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-07-02 05:37:16.000000000","message":"Patch Set 17: Code-Review-1\n\nSorry, but the test doesn\u0027t check anything","accounts_in_message":[],"_revision_number":17},{"id":"a02f4dde061183ebfeb2164a3b784fcd2c9a4b3b","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-07-08 02:36:43.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"4c72bf7803c4c2ed9e803bb681aae9bc0e393542","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-08 03:52:10.000000000","message":"Patch Set 18: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/46/731446/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"41ef30e5c4ec7e2572f8453f00abe7d904d66ba9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-07-08 04:18:19.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/731446/18 : SUCCESS in 1h 40m 56s","accounts_in_message":[],"_revision_number":18},{"id":"a3bd25774202103d3eb395f50919f0ec7f7cdb88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 05:33:01.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5c741a31ea5f4394b700d4cd40bc6b1e : SUCCESS in 45m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f936ca594e0a4ae3bb60e7fecd246922 : SUCCESS in 31m 09s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fef62750a5524045821d289b9c0ca5a8 : SUCCESS in 1h 02m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a2f4e9276db04bd6a94cd6ede70d9ee0 : SUCCESS in 57m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/20298ba1071449abb1b351ae8a1b36e5 : SUCCESS in 59m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/68ef5792049347179ca9bc23585b2248 : SUCCESS in 53m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d2b66d93c9b4c8cba5d0659b8b8d4d1 : SUCCESS in 31m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c89f5e1b6d346fb86836298c3b28f3d : SUCCESS in 26m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/821e3cfd85904ccdb53cbdb038f2e9dc : SUCCESS in 11m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0520f2f68de445891674a08201cba52 : SUCCESS in 24m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cccc3893034c4e74b3afbd4a1a97d7d0 : SUCCESS in 24m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cc5d595fb314c67b6d9740d1a8c7ccd : SUCCESS in 8m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/896326bd480943ad86a8befa9c6c48ef : SUCCESS in 45m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d1d476eed8894b6ca05c31b8ab8fd738 : SUCCESS in 55m 18s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/33e16c06be954087924a1a88c7451416 : SUCCESS in 1h 11m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/317ee98e3362429486d043300300e394 : SUCCESS in 1h 17m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2fcbc9689dca4831848171a2e756d7da : SUCCESS in 1h 55m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d98b963c2dd04523a67fed08efed7cf4 : SUCCESS in 1h 01m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3cc4b54949454e0e910d7a8b481b624f : SUCCESS in 1h 09m 05s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1b3826fd128c4427996a916d7ebfbc3e : SUCCESS in 1h 09m 14s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/9959e82982044f1487f40c6807aec7e4 : SUCCESS in 1h 08m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d55d89c894524d7e861326a50eb65fc6 : SUCCESS in 1h 05m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7f22bb92cdd74e6491fb572c987927ca : SUCCESS in 1h 44m 52s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b13f6b32af7a4b4d9eb38465fe85b79b : SUCCESS in 1h 27m 33s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/62c3026190184d538e3d476c6bf65c63 : SUCCESS in 1h 16m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/640c090a62ab48e8ab61d75df3948833 : SUCCESS in 51m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9117a86ca6a44696b20aaa2896f996c8 : SUCCESS in 1h 55m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4cd0aff9e16464183bd3808ca7219d0 : SUCCESS in 1h 01m 31s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/de46e60ede7649caa0c4505e60883a11 : SUCCESS in 1h 09m 49s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/95d03109e47d45d0957148a771f602f7 : SUCCESS in 2h 06m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3ac3a6aeb7fa40bcac4ac6473c1dd9bb : SUCCESS in 1h 47m 50s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/97b328cae892408d8035f4339ef710ce : SUCCESS in 1h 16m 32s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/df958e5b0b0f404a91e2d82a1e00f3fb : SUCCESS in 1h 42m 58s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e652c69cb9ff473a95c282713ef00ff9 : SUCCESS in 1h 54m 02s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f90efafc3ae44a6a8fc40c79290c5009 : SUCCESS in 2h 21m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0fc2b4113f6c4dbfa9e8a9dd14ed5b76 : SUCCESS in 1h 48m 55s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3fc7fe8942924e668ccfe685c2b1548d : FAILURE in 2h 50m 28s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"44381f3c7c1719f7031fe0803dae0b3263ec36fe","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-07-08 07:58:29.000000000","message":"Patch Set 18:\n\nUnit testing has been perfected.","accounts_in_message":[],"_revision_number":18},{"id":"74e5d4624cbbd0cc23bfa186489f06d1de12da4e","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-07-08 08:36:01.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"6bcf9bb16fdbbbf2dc57a3b2598ce3acaf13dd8d","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-07-09 02:02:49.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"984a2234f3f6b7e33d4f662056a243aa405a5ae1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-09 03:20:13.000000000","message":"Patch Set 19: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/46/731446/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"111e83d70bcf82e8f95e2adbe0269bffb54615fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 04:10:26.000000000","message":"Patch Set 19: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7cdcdc9037d54d9985409d6761176e5f : SUCCESS in 45m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f9afeff20a6b495e8da113c60874fb95 : SUCCESS in 31m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/82427bcbefcf47bd818d1fad0e7b910a : SUCCESS in 1h 01m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/02b271aa1a3142ca9c12fc80d2663c29 : SUCCESS in 1h 00m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a2c93e11767c43d48b6add90049e9b3b : SUCCESS in 59m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7493b2741a5b4bc59d717e40c59947ba : SUCCESS in 56m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4926929667ce43f0b0f8efc8d66a9eaf : SUCCESS in 30m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bef9f02d6d284ce896a131bea438b61d : SUCCESS in 23m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd8a1712bc6e4df38c69e08e79057f15 : FAILURE in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66942170b5cb40d5adaa06b47aaa716a : SUCCESS in 22m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12fafed22f9449de9f8546ec19b83660 : SUCCESS in 23m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3fc4f6079b442d48a1ab79c12699b4b : SUCCESS in 10m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/aa9e76dea277471c9276d68c3da5330b : SUCCESS in 47m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0fc0938e1d074452aca93f2a3d2e764b : SUCCESS in 46m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e2734019a93440c99f1866094d5a8b0b : SUCCESS in 1h 08m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cbd1ed314e16481d83e6deb7f338d85a : SUCCESS in 1h 06m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0f52d6097fc3400685bf05e35bc151d4 : SUCCESS in 1h 36m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9bd8043421124b0e8cfdfdb1ba116f97 : SUCCESS in 1h 00m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/61aed66f6af44ef489c2869622a6d478 : SUCCESS in 1h 04m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3370decda5b944ef8ecbaff397975103 : SUCCESS in 44m 33s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b0e37db7f6ef460a94bd3cfe0b3cedce : SUCCESS in 1h 26m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e5dff640589c4528a6e8ea2112eb0bc8 : SUCCESS in 1h 26m 36s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2191fb0f7a0c4e22b7966392350b5808 : SUCCESS in 1h 23m 41s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6eece6fde98144c6b84dcf3bf00d06fc : FAILURE in 1h 23m 48s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5a3e585b595b49d2b51fe8685849d49f : SUCCESS in 1h 04m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e61229e7cc1e43ddac97e1f64c107ed9 : SUCCESS in 1h 33m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ddc4596bce3f4032b239f3a1dcaf288c : SUCCESS in 2h 03m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0efe63478534a5c95b9c310e64a46f2 : SUCCESS in 49m 13s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/795f4e4d897a4c4d8e84a59bbcf6d7d1 : SUCCESS in 1h 03m 49s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f2a79457528c427c90bbf197a83a2fa4 : SUCCESS in 1h 35m 07s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b1a1b2489eba4c7f8421d873d0125d0f : SUCCESS in 36m 06s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2e2426640dec43b2aac1e343acf28a71 : SUCCESS in 1h 33m 43s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/03618e755cb4465aa596741afa123559 : SUCCESS in 1h 43m 06s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3be049da79294012b5117b43bf7d18bf : SUCCESS in 2h 05m 57s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d0c6d675d5d04991af39ff2d9662a1ed : SUCCESS in 1h 40m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f5f47fa9885940aea93314a46ec0e48a : SUCCESS in 1h 50m 09s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/0648b0673de6429585574d7678ff7872 : SUCCESS in 1h 58m 56s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"0b473275951315c8c5d1c4301e5f86947d8d8a4b","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-07-09 04:50:47.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"180d27de74ea771a6f5ff0584f0923f6ba40d5f0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-09 06:19:50.000000000","message":"Patch Set 20: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/46/731446/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"67316d52c083674cf0a9a7359b5034cc34f6ce5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 07:57:15.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d2235d7fbd104366bfa5451cb6a45a24 : SUCCESS in 58m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e40473f811144093ab1c6541409079f0 : SUCCESS in 43m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ffc9ec9f82ba4d4182d2ad86c8f25973 : SUCCESS in 1h 12m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dd1c9dbe9f584ec293bed5ff9a97251b : SUCCESS in 1h 47m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f0d89559767743c7bc47488d452db58a : SUCCESS in 1h 31m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1892e341f1c2447e9467bc98aa78fb40 : SUCCESS in 1h 13m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a99e8a385694fe39fba90829f5853b1 : SUCCESS in 37m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eaa746db031e414c9f5ef6df77b96f45 : SUCCESS in 33m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b98ef07e6c04b148389e40b30a611d3 : SUCCESS in 12m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a00070592bbe4a9ca5caa85e81e8ab1c : SUCCESS in 25m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b73503479916457abc5d27752602910a : SUCCESS in 25m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e666f9dbfe7d47ea834530167a432e5a : SUCCESS in 10m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a1a923cb1e7c475e890d4acdffc3c55d : SUCCESS in 53m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b8b6ab5276ff4b56bdc50329986e6ba7 : SUCCESS in 49m 54s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a7cc674ea69f4d60ae7106bc7b4aa69b : SUCCESS in 1h 12m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/87fe503b59eb4ff8b48da810d529a030 : SUCCESS in 1h 10m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bf704d4b6aea437883af4352a97930ab : SUCCESS in 2h 38m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce8415d6ae934db99db396f55e3a9fd2 : SUCCESS in 2h 16m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/01742151f83b4574bc62053333a0204d : SUCCESS in 1h 13m 51s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e37f0ec7d86e43b8b8baf401014f1f23 : SUCCESS in 1h 08m 51s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/128e10e419004e158de4d4dcb7896c04 : TIMED_OUT in 2h 27m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b8a34465325d4f999be25d0b1fef630b : SUCCESS in 1h 13m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/22f3d5b770a44136a27ebbb641a6092b : SUCCESS in 2h 41m 37s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/418c2b5a98344f889e14ad8888817031 : SUCCESS in 1h 23m 52s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/408f7cf88af94a8fb8f95b5af04656a6 : SUCCESS in 1h 15m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7693c08b58db4887bf070e7a04c205ed : SUCCESS in 1h 15m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/db9f20104c944c91b6cd077b3be96043 : TIMED_OUT in 3h 04m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d78c661438b42fb8ad9131ef171d14e : SUCCESS in 1h 15m 58s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/ff5379e6dc0b48fd9c4678647e787c20 : TIMED_OUT in 2h 17m 43s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9273bc9f8f844c8690c2e0e678d50ede : SUCCESS in 1h 34m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/264e89de3b27464f880102e5929da21f : SUCCESS in 1h 46m 49s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/95318631d9074f8899cf87c7a0361477 : SUCCESS in 1h 32m 15s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fdc4cb7f7be04d5ba89ea6507ac1944c : SUCCESS in 2h 45m 08s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ec5567efcb61408b9871f2c27e47c534 : SUCCESS in 2h 03m 40s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b2c60d47ca5d4dad84f3a403a10b5346 : SUCCESS in 2h 25m 23s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4432d6243f9a4d06a94d5e77a76406ba : SUCCESS in 2h 00m 30s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f9a7cf366fe1488298973c83930b0a2a : FAILURE in 2h 13m 39s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"0b3838d6921a450d1a29d3a3537501699bf811c0","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-07-09 08:03:44.000000000","message":"Patch Set 20: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"f5aecff58e9a796ea04512b7660b7fac425ca47a","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-07-09 10:13:37.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"83a575d6c1c8a925bec833149b8b472424a79f01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 10:59:33.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/743d1b2b8be44db8bf9cc59e5b50f597 : SUCCESS in 56m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2ce7a1864ab647bdabec67549ff71eab : SUCCESS in 52m 08s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2853918c1d4c4db6b7ff5c5a8ac95d3d : SUCCESS in 1h 16m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e78240fbf2d04b18baf089d6a64251b7 : SUCCESS in 1h 09m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d28a80ea6e8442e2bbb8eef9502d9e18 : SUCCESS in 1h 08m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ee56f1b6be49492db05256ddf396f7ac : SUCCESS in 1h 11m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe19b515cab54fcaab2e3c2ae31254d9 : SUCCESS in 34m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e2f76fae3f14b8797e4624730052eca : SUCCESS in 32m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/476df6eb06e94240b0f4b5e6f1cf0e60 : SUCCESS in 13m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/292dfbe8be4848d39afa64d98934f0e6 : SUCCESS in 26m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8190b7c9936a4214b1a33c1056823393 : SUCCESS in 27m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04d3ed797744455bb6f08557e7e7b5bd : SUCCESS in 11m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e658427eb60347d5994ba092dc893994 : SUCCESS in 1h 11m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0194bc098d3645c28d6f0513c25a7613 : SUCCESS in 58m 13s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/bc9138c95918450d9739307f8587c233 : SUCCESS in 1h 32m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7646ae43efc4414da8b626865819839f : SUCCESS in 1h 17m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/461906ff8509439fafc0f7562c25b251 : SUCCESS in 1h 44m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b812cf12566452da85a3c56c6eabdb8 : SUCCESS in 1h 15m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ec95e14ecda24d5b8be33a538b816735 : SUCCESS in 53m 38s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9b7b7a5ed4ee4437a2a04bdc7d0e7893 : SUCCESS in 1h 09m 45s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5cdd7b9fad54492aa7805ff234f477b : SUCCESS in 1h 14m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e047cdff33674b559456a0af8106aac2 : SUCCESS in 1h 14m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/524eb2a4052143cca18ba4c604ee0600 : SUCCESS in 1h 42m 48s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/882661e42038440b928d4a78575b7bb8 : FAILURE in 1h 26m 27s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ddc69671299d4215b85519074d2e3f4b : SUCCESS in 1h 02m 35s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0118a102c3e44afbaee84290f1fdee74 : FAILURE in 1h 21m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/89aba2aa58374573866ab8f8b9653a5d : SUCCESS in 2h 04m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/848b87e69d72434a983290afe6a220f7 : SUCCESS in 57m 21s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/3e6bd1be74d9402cb74e9d88d714d8a0 : SUCCESS in 1h 28m 47s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/62c284a786744082850f3809580898e2 : SUCCESS in 1h 15m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0d918332c6fd47ebbaea81a334089b5a : SUCCESS in 39m 04s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8fb4ba3b7d154c59a91776821c4547ca : SUCCESS in 1h 15m 17s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/43c86a1986254b5ea0e0c0db9956349b : SUCCESS in 1h 38m 16s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/25f3a9fdd6a547c5951963d07efb2a33 : SUCCESS in 2h 09m 38s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ee02903bbd254e7ca8bb6d3dc84a2677 : SUCCESS in 2h 25m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5caa2f88046a4abf9a9c2934e6a11197 : SUCCESS in 1h 44m 51s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/96609865f67f46419895847d21fb3fbc : FAILURE in 2h 18m 52s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"75550664be1a7fb9628ccd946abcb1d14c56d98a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-13 11:25:53.000000000","message":"Patch Set 20: Code-Review-1\n\n(3 comments)\n\nlooks good, just a few nits","accounts_in_message":[],"_revision_number":20},{"id":"3d21b6d485743f610a5ba3e94ed871bc43a14a8d","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-07-14 08:01:06.000000000","message":"Patch Set 20:\n\n(1 comment)\n\nThanks for your correction.","accounts_in_message":[],"_revision_number":20},{"id":"f39b195ae0dd6ecb0da79484eda09ae75b0ac2fa","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-07-14 08:14:44.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"81da2710ee515cf88875a9a240c519eefcffa888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 10:41:33.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d36120aa212b47888022226bad30011a : SUCCESS in 47m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/976a08d729cb4d6caf419fd8e1b6d2d3 : SUCCESS in 32m 36s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e3af507863f44ab88d57bafdb3f54169 : SUCCESS in 45m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d15d21bce1d342a7a46eccde6134edcd : SUCCESS in 1h 15m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8fba6d9625614780a315e61c7408d82e : SUCCESS in 53m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9a197914ba8547728e915e4f2823d5a3 : SUCCESS in 1h 05m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a85ae0a6f13e4d76964dde9065100888 : SUCCESS in 40m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1867ff94fa784cd2b5ab4cb252309ede : SUCCESS in 28m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c27a12224c1c48f4a728626203e7f1de : SUCCESS in 12m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dc05ccde7764e07862e55dd3dafbb09 : SUCCESS in 31m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/350ea4cde4af40dba20d3d538665bfc8 : SUCCESS in 27m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59bff1b495354965a21933432f067922 : SUCCESS in 10m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/57df5cf625fe4c7abe204ec66df4d19c : SUCCESS in 49m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3db05fcc6f0941cd80b1ac82e2a2f079 : SUCCESS in 52m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/714467a8cf194ddfa76921dc14f4ed04 : SUCCESS in 1h 16m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/20438ccf61414924a7f044719efc3a33 : SUCCESS in 1h 17m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/547ac3a5543245f393f931902c799145 : SUCCESS in 1h 34m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6353593ac8ee4d1b8fe0141c0965cd11 : SUCCESS in 1h 13m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bf98536922494c3a94d8ed855cfe29e9 : SUCCESS in 1h 19m 06s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/fb2ae151de7a4ded9717612c1c7dd1ae : SUCCESS in 1h 03m 13s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/470aa8b694074859b67075a0c64ac883 : SUCCESS in 1h 13m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/434728acad044d908925dced9c243fc2 : SUCCESS in 1h 17m 22s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9738fa41959c4675916cf88ad17eaf70 : SUCCESS in 1h 49m 36s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9cc05de3f7f945f491724986f4fc2af6 : SUCCESS in 1h 36m 49s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4645449ca28b4bed9d54dec3b25b6473 : SUCCESS in 1h 27m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3745adaa11b0468b89bfc6cd92639115 : SUCCESS in 1h 14m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3f8c3d0b23c64aacb79b58479b4d571b : SUCCESS in 2h 05m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3c4cad75905411398b702bdee9ec651 : SUCCESS in 1h 01m 08s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/8e02d28cbf954fe69ac395d3020f0575 : SUCCESS in 1h 15m 21s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/150409c580664eaf8068c20c4ed575b4 : SUCCESS in 1h 04m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b4dc5c7fd09a43d4a579a4c5e28ff689 : SUCCESS in 28m 14s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4b46a9967e3449ff8ebfee366f3eafaa : SUCCESS in 1h 31m 53s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/48091f05edfa4ebbb557d1044487cb52 : SUCCESS in 1h 18m 41s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/df8e4e27b1d046e58f828103f61183b0 : SUCCESS in 1h 17m 10s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ab0de1b3269343dc9e958f33b0f635f2 : SUCCESS in 2h 14m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f3b80795c8af431abde81271d16764a7 : SUCCESS in 1h 49m 57s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/72ca47cf715442139aa6dc7062f3d774 : SUCCESS in 1h 54m 35s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"8ca293802894dd58c79995a5ea7f4aa7984f2245","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-27 13:11:58.000000000","message":"Patch Set 21: Code-Review-1\n\n(1 comment)\n\nI have comment only for the commit msg now","accounts_in_message":[],"_revision_number":21},{"id":"5480fdf9a211f64ac58726b71e1cb03af4773cde","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-07-28 01:21:35.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"f0283a6f2d11b714836d7ad6e5cb645d4deaf61e","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-07-28 01:52:27.000000000","message":"Patch Set 22:\n\n* sriov-test http://52.27.155.124/sriov/731446/22 : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"e9d9149d36abe61d77f008844629d75a353ae3d1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-28 02:41:45.000000000","message":"Patch Set 22: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/46/731446/22 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":22},{"id":"513cc00f3e9db2b1715ba97150f8a1a9f4da94c4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-07-28 03:14:07.000000000","message":"Patch Set 22:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/731446/22 : FAILURE in 1h 51m 45s","accounts_in_message":[],"_revision_number":22},{"id":"ce428e1c44c0584714871857432b003aa3433331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 03:48:26.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a0bfa2e620a34eccb93f9bf14f0e1811 : SUCCESS in 1h 01m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/952c84e7e4964d2997a076ac904f5d61 : SUCCESS in 29m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2237c375db484c97b2dc2e421c049ee4 : SUCCESS in 1h 25m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/896105f92f8e458b84f6293f98a4c3b5 : SUCCESS in 1h 10m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/69ae3fb708f94ac4af4b94ccb1e4cf7c : SUCCESS in 38m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ef2e89f59db9423cbd3f49d7e3a97b57 : SUCCESS in 1h 09m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abbce22b0cdf4ef8ae214471e0ccf313 : SUCCESS in 35m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/369fef21a30041f59f3c21a9dce84fd1 : SUCCESS in 30m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dcab9651c864f74b1c4336830e50ac1 : SUCCESS in 10m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dbebea0653e4149bbc02e53a0c9be97 : SUCCESS in 28m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c7aa638281d405ca61a0f200e5d4960 : SUCCESS in 26m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3b1f6bb1234cc5a0621c265f90c049 : SUCCESS in 8m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4bb398322261482d8f73b949c517fcc9 : SUCCESS in 48m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2974bc36df5426cb673c4915906e7b5 : SUCCESS in 50m 48s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/71a50127f5f74d27a9f787dbce45b899 : SUCCESS in 1h 19m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/32afb62be77e45e1bf898ca3ea373a78 : SUCCESS in 1h 04m 21s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ee2bd10fcb54487b8ccd3e7b873e36d8 : SUCCESS in 1h 14m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6633099ad74c426caa4a672d954c5a6c : SUCCESS in 45m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5428785544cd47888d3069b2f0e506cf : SUCCESS in 1h 17m 37s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5a667f5603304de0a8f703662a6f61da : SUCCESS in 1h 23m 54s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c8af5c0e2fb4e60a453b5ffa9b04b0a : SUCCESS in 1h 19m 10s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/41962f9f46b9487c9c7d1f58a516f668 : SUCCESS in 1h 42m 22s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/96c2e854866c4a4495c4dbc03c217d07 : SUCCESS in 1h 21m 19s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7f2a1b7cdae14f778ed3cc1561d5f9ae : SUCCESS in 1h 07m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e219bf533bcf4015987542959952d452 : SUCCESS in 1h 12m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f6d0e8f9278f4bafb61ab2c0048e3c6c : SUCCESS in 2h 01m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6a513ef07e54d3d8fd7e37f19de619f : SUCCESS in 59m 12s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/b9c7fc6916724aa5aca33b184f01dea5 : SUCCESS in 1h 06m 37s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/eef778ba98d5401ab541f447751a41e4 : SUCCESS in 1h 14m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8ce45c19f48944c78307f3cfa4301e00 : SUCCESS in 46m 19s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cda27b0f0d4b46f28544063b1a37e0db : SUCCESS in 1h 15m 31s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/41a62ccca5724cde949bf743113d32bc : SUCCESS in 1h 45m 30s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/31c1c5796fa14f55a2f272d77d820cee : SUCCESS in 1h 11m 35s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2c0094014557455e8a86398247ef8f6b : SUCCESS in 2h 24m 23s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/be5a2c47248e4603bcc49b12d6f5f48c : FAILURE in 1h 48m 04s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/dbf4f59034054159905353f2a08e2bfe : SUCCESS in 1h 49m 22s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"df0595912c40fc2b3c8fb912f51188d673d0fe8c","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-07-28 05:50:46.000000000","message":"Patch Set 22: Code-Review+1","accounts_in_message":[],"_revision_number":22},{"id":"7fb5e07517c892b8d1ef5440b22c59f713e56583","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-07-28 07:48:52.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"5f9fc694296f5a5923fa71dfbb880c342c79189f","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-07-28 08:04:29.000000000","message":"Patch Set 22: Code-Review-1\n\n(2 comments)\n\nFYI: the issue is also covered by this fix\nhttps://review.opendev.org/#/c/738551/","accounts_in_message":[],"_revision_number":22},{"id":"053673bf0002a83af6edb4268c24c993acf64194","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-07-28 09:33:44.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"ab71652fb9579feb11a80f538c663cf2bd05d37b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-09-21 08:07:12.000000000","message":"Patch Set 22: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"f13e9287536fd8ff10b0fa36b26929c1b9741d56","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-23 12:46:51.000000000","message":"Removed Code-Review+2 by Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":22},{"id":"08cb47b58a92dc710bfe1821c7158d7b3ff5ad1b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2020-11-26 07:24:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"3060feafea9d02bf719962e43af0b255348d1eec","revisions":{"be0727a7ff65a1f22569bc9c04846af0ac1ef633":{"kind":"REWORK","_number":1,"created":"2020-05-28 07:08:19.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/1"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrate to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 was not executed. This will\nresule in stale flows in table 61. Add delete_accepted_egress_direct_flow()\nin port_unbound() to ensure delete these flows.\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070X\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be0727a7ff65a1f22569bc9c04846af0ac1ef633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be0727a7ff65a1f22569bc9c04846af0ac1ef633"}]},"branch":"refs/heads/master"},"ace0ef155e6b700ce1bd63c8e7a16e0e2186f67e":{"kind":"REWORK","_number":2,"created":"2020-05-30 03:23:48.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/2"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-30 03:08:41.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrate to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nMove the delete_accepted_egress_direct_flow() into port_unbound()\nto ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070X\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ace0ef155e6b700ce1bd63c8e7a16e0e2186f67e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ace0ef155e6b700ce1bd63c8e7a16e0e2186f67e"}]},"branch":"refs/heads/master"},"9562e330ad22170a972b0a579c0787051d5464ca":{"kind":"REWORK","_number":3,"created":"2020-06-01 08:54:41.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/3"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-06-01 08:52:31.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrate to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nMove the delete_accepted_egress_direct_flow() into port_unbound()\nto ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070X\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9562e330ad22170a972b0a579c0787051d5464ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9562e330ad22170a972b0a579c0787051d5464ca"}]},"branch":"refs/heads/master"},"5bb7d82f7befaeaaf9d4364ff3cc73301ccfdcbe":{"kind":"REWORK","_number":4,"created":"2020-06-01 12:28:30.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/4"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-06-01 12:25:39.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrate to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nMove the delete_accepted_egress_direct_flow() into port_unbound()\nto ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070X\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5bb7d82f7befaeaaf9d4364ff3cc73301ccfdcbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5bb7d82f7befaeaaf9d4364ff3cc73301ccfdcbe"}]},"branch":"refs/heads/master"},"beeac3ca066f06435347c0ab646f7d4bf8db0144":{"kind":"REWORK","_number":5,"created":"2020-06-02 13:44:09.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/5"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-06-02 13:41:10.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrate to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nSo create a seperate method for this and also call it inside port_unbound()\nby default to ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070X\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/beeac3ca066f06435347c0ab646f7d4bf8db0144"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/beeac3ca066f06435347c0ab646f7d4bf8db0144"}]},"branch":"refs/heads/master"},"322df97bb81cabe9a79d3cbe0ef3905fcee6b34f":{"kind":"REWORK","_number":6,"created":"2020-06-03 04:21:35.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/6"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-06-03 04:18:04.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrate to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nSo create a seperate method for this and also call it inside port_unbound()\nby default to ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070X\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/322df97bb81cabe9a79d3cbe0ef3905fcee6b34f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/322df97bb81cabe9a79d3cbe0ef3905fcee6b34f"}]},"branch":"refs/heads/master"},"2bac73418d16f0a08f83c43bbe43b2814061282f":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-06-04 02:21:08.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/7"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-06-04 02:19:18.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrate to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nSo create a seperate method for this and also call it inside the\nport_unbound() by default to ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070X\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bac73418d16f0a08f83c43bbe43b2814061282f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bac73418d16f0a08f83c43bbe43b2814061282f"}]},"branch":"refs/heads/master"},"609afbd74b6e6c9e4f0d2a7cac893a8fe819ff94":{"kind":"REWORK","_number":8,"created":"2020-06-04 02:33:24.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/8"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-06-04 02:31:09.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrate to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nSo create a seperate method for this and also call it inside the\nport_unbound() by default to ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070X\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/609afbd74b6e6c9e4f0d2a7cac893a8fe819ff94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/609afbd74b6e6c9e4f0d2a7cac893a8fe819ff94"}]},"branch":"refs/heads/master"},"2b027b19c36f03e809a26d4205003c238f6662aa":{"kind":"REWORK","_number":9,"created":"2020-06-04 10:17:10.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/9"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-06-04 10:16:43.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrate to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nSo create a seperate method for this and also call it inside the\nport_unbound() by default to ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070X\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b027b19c36f03e809a26d4205003c238f6662aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b027b19c36f03e809a26d4205003c238f6662aa"}]},"branch":"refs/heads/master"},"aa14aa229e24ead885c3ef47618771f5d4ac68fb":{"kind":"REWORK","_number":10,"created":"2020-06-05 01:15:28.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/10"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-06-05 01:15:11.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrate to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nSo create a seperate method for this and also call it inside the\nport_unbound() by default to ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070X\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aa14aa229e24ead885c3ef47618771f5d4ac68fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aa14aa229e24ead885c3ef47618771f5d4ac68fb"}]},"branch":"refs/heads/master"},"123a2ff0d0f710141004f6fee076d7a4fdd6df31":{"kind":"REWORK","_number":11,"created":"2020-06-06 04:45:39.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/11"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-06-06 04:44:58.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrate to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nSo create a seperate method for this and also call it inside the\nport_unbound() by default to ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070X\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/123a2ff0d0f710141004f6fee076d7a4fdd6df31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/123a2ff0d0f710141004f6fee076d7a4fdd6df31"}]},"branch":"refs/heads/master"},"b4ab91a052dfc19bb4fce29a9450567348d5eead":{"kind":"REWORK","_number":12,"created":"2020-06-08 01:19:21.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/12"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-06-08 01:18:36.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrate to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nSo create a seperate method for this and also call it inside the\nport_unbound() by default to ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070X\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b4ab91a052dfc19bb4fce29a9450567348d5eead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b4ab91a052dfc19bb4fce29a9450567348d5eead"}]},"branch":"refs/heads/master"},"cafe8ccee4f891f51c04dccbf945c7e0b40104f7":{"kind":"REWORK","_number":13,"created":"2020-06-18 07:39:35.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/13"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-06-18 07:38:55.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrate to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nSo create a seperate method for this and also call it inside the\nport_unbound() by default to ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070X\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cafe8ccee4f891f51c04dccbf945c7e0b40104f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cafe8ccee4f891f51c04dccbf945c7e0b40104f7"}]},"branch":"refs/heads/master"},"6b231aa32576602e1f6a560a343d1344a33e2bf8":{"kind":"REWORK","_number":14,"created":"2020-06-18 12:11:46.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/14"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-06-18 12:09:52.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrate to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nSo create a seperate method for this and also call it inside the\nport_unbound() by default to ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070X\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b231aa32576602e1f6a560a343d1344a33e2bf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b231aa32576602e1f6a560a343d1344a33e2bf8"}]},"branch":"refs/heads/master"},"c7647cc0785bc78e6b0009bc22c99ffd57cb634d":{"kind":"REWORK","_number":15,"created":"2020-06-19 01:11:50.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/15"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-06-19 01:10:53.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrate to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nSo create a seperate method for this and also call it inside the\nport_unbound() by default to ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070X\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c7647cc0785bc78e6b0009bc22c99ffd57cb634d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c7647cc0785bc78e6b0009bc22c99ffd57cb634d"}]},"branch":"refs/heads/master"},"66f34fed514885ae1894114242a8819c888de084":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2020-06-22 02:15:05.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/16"}}},"commit":{"parents":[{"commit":"805fb5c970c8b761ce7f4877052ffef74b524e41","subject":"Merge \"Add functional test for metadata agent monitoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805fb5c970c8b761ce7f4877052ffef74b524e41"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-06-22 02:10:40.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrated to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nSo create a seperate method for this and also call it inside the\nport_unbound() by default to ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070X\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66f34fed514885ae1894114242a8819c888de084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66f34fed514885ae1894114242a8819c888de084"}]},"branch":"refs/heads/master"},"66591240e40d15fecb0d90de284b34abfc25d510":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":17,"created":"2020-06-22 07:43:03.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/17"}}},"commit":{"parents":[{"commit":"b1dba996b52f14a1405800cf2e90485beaeb50d3","subject":"Merge \"Remove \"find\" rootwrap filter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1dba996b52f14a1405800cf2e90485beaeb50d3"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-06-22 07:41:22.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrated to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nSo create a seperate method for this and also call it inside the\nport_unbound() by default to ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66591240e40d15fecb0d90de284b34abfc25d510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66591240e40d15fecb0d90de284b34abfc25d510"}]},"branch":"refs/heads/master"},"831cda60d8fe7e0012beb1b97280d32dd97cd94b":{"kind":"REWORK","_number":18,"created":"2020-07-08 02:36:43.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/18"}}},"commit":{"parents":[{"commit":"a3c7aad790bd50a0c167ab5224eede7949737693","subject":"Merge \"ovn: Add TODO to remove Port Groups migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3c7aad790bd50a0c167ab5224eede7949737693"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-07-08 02:35:33.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrated to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nSo create a seperate method for this and also call it inside the\nport_unbound() by default to ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/831cda60d8fe7e0012beb1b97280d32dd97cd94b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/831cda60d8fe7e0012beb1b97280d32dd97cd94b"}]},"branch":"refs/heads/master"},"6d62fa30f4c45a88a72ef6fd4ba9fdcc8f7a4fb9":{"kind":"REWORK","_number":19,"created":"2020-07-09 02:02:49.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/19"}}},"commit":{"parents":[{"commit":"ad899146945d12cb6ab446a417d1415c35bb0603","subject":"Merge \"Sync the dns-assignment with the actual designate dns-domain\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ad899146945d12cb6ab446a417d1415c35bb0603"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-07-09 02:00:55.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrated to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nSo create a seperate method for this and also call it inside the\nport_unbound() by default to ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d62fa30f4c45a88a72ef6fd4ba9fdcc8f7a4fb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d62fa30f4c45a88a72ef6fd4ba9fdcc8f7a4fb9"}]},"branch":"refs/heads/master"},"d5a6828aaf988b0ac9bc6c117a2f6762a6a33c91":{"kind":"REWORK","_number":20,"created":"2020-07-09 04:50:47.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/20"}}},"commit":{"parents":[{"commit":"ad899146945d12cb6ab446a417d1415c35bb0603","subject":"Merge \"Sync the dns-assignment with the actual designate dns-domain\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ad899146945d12cb6ab446a417d1415c35bb0603"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-07-09 04:49:07.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrated to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will resule in stale flows in table 61.\nSo create a seperate method for this and also call it inside the\nport_unbound() by default to ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d5a6828aaf988b0ac9bc6c117a2f6762a6a33c91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d5a6828aaf988b0ac9bc6c117a2f6762a6a33c91"}]},"branch":"refs/heads/master"},"817029f4ab9faab365f5f94b73e00e90296e7c50":{"kind":"REWORK","_number":21,"created":"2020-07-14 08:14:44.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/21"}}},"commit":{"parents":[{"commit":"73557abefcba1c6ce0cef709d1082674c0217485","subject":"Merge \"Optionally use admin powers when deleting DNS records\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73557abefcba1c6ce0cef709d1082674c0217485"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-07-14 08:11:25.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrated to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will result in stale flows in table 61.\nSo create a seperate method for this and also call it inside the\nport_unbound() by default to ensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/817029f4ab9faab365f5f94b73e00e90296e7c50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/817029f4ab9faab365f5f94b73e00e90296e7c50"}]},"branch":"refs/heads/master"},"3060feafea9d02bf719962e43af0b255348d1eec":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":22,"created":"2020-07-28 01:21:35.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/46/731446/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/731446/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/731446/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/731446/22"}}},"commit":{"parents":[{"commit":"4e5d392a8f353f7d14c231f741340ca48b2b8def","subject":"Merge \"[OVN] Add segments to ML2_SUPPORTED_API_EXTENSIONS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e5d392a8f353f7d14c231f741340ca48b2b8def"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-05-28 06:51:57.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2020-07-28 01:19:08.000000000","tz":480},"subject":"Fix delete-accepted-egress-direct-flow issue on deleting VM","message":"Fix delete-accepted-egress-direct-flow issue on deleting VM\n\nWhen VM is deleted or migrated to other compute node, the function\n\u0027delete_accepted_egress_direct_flow()\u0027 in process_deleted_ports()\nwas not executed. This will result in stale flows in table 61.\nThe method should be called in the port_unbound() by default to\nensure delete these flows.\n\nIn addition, use \u0027br_int\u0027 instead of \u0027self\u0027 to call delete_flows()\nin delete_accepted_egress_direct_flow().\n\nChange-Id: I7d937e39bf67ffb31f8a82f993323e9e70d6c344\nCloses-Bug: #1881070\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3060feafea9d02bf719962e43af0b255348d1eec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3060feafea9d02bf719962e43af0b255348d1eec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
