)]}'
{"id":"openstack%2Fneutron~733033","triplet_id":"openstack%2Fneutron~master~Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f","project":"openstack/neutron","branch":"master","topic":"bug/1881316","hashtags":[],"change_id":"Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f","subject":"[OVN] Add logical port to its corresponding subnet port group","status":"ABANDONED","created":"2020-06-03 01:30:11.000000000","updated":"2020-06-25 15:23:16.000000000","total_comment_count":59,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cbdb546fab28a3d545cdc28d111ff9b1664f9988","_number":733033,"virtual_id_number":733033,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-06-20 22:15:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-20 22:12:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"date":"2020-06-20 21:23:54.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":-1,"date":"2020-06-20 19:43:03.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 01:39:51.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-06-04 07:51:02.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-06-05 11:55:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-06-05 20:26:32.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-06 11:02:27.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-09 21:35:07.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-06-11 12:46:05.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-06-17 15:11:36.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-06-20 21:23:54.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-06-20 22:12:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-20 22:15:57.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1b1bb838d14f5a265d92c08b8287af7a1fb54a6","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-03 01:30:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"227faa9cae69c8320198097eb44a4fef306cb7db","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-03 01:39:05.000000000","message":"Topic set to bug/1881316","accounts_in_message":[],"_revision_number":1},{"id":"2bc8f81848684ebeb6e9845adaa2816fc446a99c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-03 03:12:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/733033/1 : SUCCESS in 1h 41m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b7e14e4b9c5a6558e39ad5d888a2f09c3cec0793","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 04:37:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5106692a91a54bbdae65796011cb65c6 : SUCCESS in 50m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a2a5845efb2548809e9c25c72e0eb489 : SUCCESS in 33m 40s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/8be609d321c947c699bec7f1b464eaed : FAILURE in 1h 49m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4c1a2681cf1947479637f210c18a472e : SUCCESS in 45m 46s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/59db898132f24b08b5e4c0db3e8bd9c5 : SUCCESS in 1h 04m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c1059eb78eb24cfa9a291c7e34328ec1 : SUCCESS in 1h 06m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a2aee3a0a934077ac8a7e060f61d32f : FAILURE in 35m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7837ad362c9d407da14c1a7004498816 : FAILURE in 33m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7cb28ad31aa4905a4a80f17ca7c43c3 : FAILURE in 9m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6046df29e91a48498446de93c9bb3d90 : FAILURE in 25m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b999b5dcb3144021b17553c2202813e8 : FAILURE in 25m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99d2e44e11744eef872f6b5943605047 : SUCCESS in 9m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fc05b3770b224f34b4e68ce8ad764995 : FAILURE in 46m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d589f4245af64984814d603939010d21 : FAILURE in 49m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3808cdab04844b15ace3a7909785444e : SUCCESS in 1h 14m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/50eea46b85474a7fb738ee8ef1648000 : SUCCESS in 1h 08m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/77bd204e407549babdd2caf1d0d26ce7 : SUCCESS in 2h 00m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a32b5098ffe348d98aa9c4e33b0f3eac : TIMED_OUT in 3h 03m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/82aaa87aa21546f3bd9448d278eb9163 : SUCCESS in 1h 14m 02s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d7ed4834db2441c2ad40596ba6a55526 : SUCCESS in 1h 09m 53s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/790131bea8b545a5881cb2a7c11afd98 : SUCCESS in 1h 01m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/469ea21eac614667aa64dcc30a5319ff : SUCCESS in 1h 14m 18s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d92ddcb3d5e74e1491df757592ff7d05 : SUCCESS in 1h 40m 51s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9e084cd33fb9416aa063488405a0e4db : FAILURE in 1h 19m 02s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6cd6e841396b448390764598421cbe02 : SUCCESS in 1h 11m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/571674485d1d4ab3ade73a28c9be6112 : SUCCESS in 53m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/37b99799889f482ea1a9469152276740 : SUCCESS in 2h 05m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e19e61c91d904ad7bc4df6edf3d35f12 : SUCCESS in 55m 11s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6041839500694a0b8d9169bca7b41ef0 : FAILURE in 1h 41m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7ae3daecd0674c50ac32705b65ca7537 : FAILURE in 2h 01m 08s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/153e602f299b4fb4bff0ca9a6f281c80 : SUCCESS in 1h 14m 06s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/58cc41741e664b06b6b80dc1c02c5cf5 : SUCCESS in 1h 43m 40s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b8b95fbf1b724f34bc9c14e43f069548 : SUCCESS in 1h 45m 22s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6f5e2f373a4e4bb9ab018633a5f59764 : SUCCESS in 2h 20m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/343ad33f49124dbaa48923f9e036af77 : FAILURE in 1h 35m 50s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/efae950be9024521a81c15caad019aec : FAILURE in 2h 09m 39s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/logapi/drivers/openvswitch/ovs_firewall_log.py","accounts_in_message":[],"_revision_number":1},{"id":"a8b869a779d35ff0abfa1b01ebf92c61b6c431be","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-03 04:52:42.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"16dd7cdd7b88735c5a4101344d2f566a39ad5fee","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-03 10:10:37.000000000","message":"Patch Set 1: Workflow-1\n\nWhile all that is here is proper, I am missing code in maintenance task for check_subnet_port_groups. Will be adding that here next.","accounts_in_message":[],"_revision_number":1},{"id":"5577170ed469384c251b8113011d841e4b418839","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-04 02:17:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c243a6084f51b37515b16dcce6bb0425f0918db2","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-04 02:23:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3d8d1850dfe4354be99d34e53932889aea12b290","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-04 03:20:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"398d1e6483e5b806c2673acf31fbf34fb55573cd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-04 03:39:31.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"28daba4db10324908d8ef0c547c7bcc3e86f8418","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-04 05:01:00.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"8af1da4a75c914ba5530d127174db18f6018a2df","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-04 05:15:08.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/733033/4 : SUCCESS in 1h 53m 50s","accounts_in_message":[],"_revision_number":4},{"id":"d0d7111c1a2aae2f993db9e3df4338b30bacec32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 05:44:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a4d04b6f8941420d8e411fbdff480a7a : SUCCESS in 55m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/45f290bde6924ac6853260586fb51f96 : SUCCESS in 37m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/a1e2f4ba50d84f6689c42ceb31cfad13 : FAILURE in 1h 33m 10s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b425531469394c968c32a7ad472cce9b : SUCCESS in 1h 03m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/436a35866dff42529cc0b3c95aa2ee29 : SUCCESS in 1h 24m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6fbdee858511457d865fb982f4400fe3 : SUCCESS in 1h 00m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/384f0960b2024c41a43ef667bc261942 : SUCCESS in 38m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1002b5afe5454a86acd37f57e65d2ad0 : SUCCESS in 28m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a22b5560b4db4177a96f16888f9faae6 : SUCCESS in 10m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb5d73f8cdee472283abbd392210b7a8 : SUCCESS in 27m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b6d5cb0a15e4f49b367d1bb3624ed91 : SUCCESS in 28m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27ecdf0f08804af39f7decf2f43c41dc : SUCCESS in 9m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/346c137396224386978a7b711ad19404 : FAILURE in 52m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b2f578dd483491cac7b522c4dad142f : FAILURE in 56m 56s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/05cdb0272158443ebadba21ab09770f6 : SUCCESS in 1h 10m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2fa6393611bb4538b1dcbaf94e882696 : SUCCESS in 1h 09m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/14e4989e1b9e49e5a607d33f65e7bb82 : SUCCESS in 1h 55m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ceafda519b3b4043bdef9b5e3c2ce018 : SUCCESS in 1h 00m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4defd53930d64d1aa113c6c12058624c : SUCCESS in 1h 15m 12s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2805796ae6ad418d97b9f6fae584c514 : SUCCESS in 1h 03m 11s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/fbe962d023e346f080c33c63cbdf4e7b : SUCCESS in 1h 06m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b54aa924643546b78c863ff3b592629b : SUCCESS in 1h 14m 22s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cfbde838a50748dbb5425577b66d00a8 : SUCCESS in 1h 29m 26s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8ca9389fdf834d69bc4afaf9d55e4e5f : SUCCESS in 1h 28m 26s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/38885c8b8fcf47e7b53e0bfb82055eb3 : SUCCESS in 1h 04m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/13e712756439495eab92140d8cfd67f9 : SUCCESS in 51m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0d0690c45cbe4e5da4571d9e511e8c8a : SUCCESS in 2h 05m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d127f36ce2fd4e7e837d13f90b8d5920 : SUCCESS in 36m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c38b9c6ef112479a8352dc1f52f9c71e : FAILURE in 26m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/06f0dd5c07f84a438f5cf86db8fd6e19 : FAILURE in 19m 30s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/23eaf13895ef4d5bbdd382c97b2dbf6e : SUCCESS in 57m 23s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6a9a7fd6078f4d3d9f5db6c4783b75f7 : SUCCESS in 1h 34m 38s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c4bd432b6f04468e8bf4b009105914a6 : FAILURE in 35m 53s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6f6db692f40c46249ece93c9929b2995 : FAILURE in 2h 22m 56s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ed4fc7f125834037bf9f28f40b7d1a50 : FAILURE in 24m 40s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/4f4f99648ee449cfbd0e1cd63338b349 : FAILURE in 38m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"805ac43701ad25e8dcff40315fc89f2025415cc6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-04 06:18:00.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"185f23b42dfc623df27043e708af802db632d619","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-04 07:51:02.000000000","message":"Patch Set 4:\n\n(3 comments)\n\nHey! Thanks for the patch!\n\nI have a question related to default behavior while we don\u0027t have DHCP enabled on subnet.","accounts_in_message":[],"_revision_number":4},{"id":"efd0bb88d97388b466815c5ac089625dcc8657c4","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-04 09:44:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ff8418c89554a29e9232d82aaa15d043e78a55e8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-04 10:02:41.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/733033/5 : FAILURE in 17m 18s","accounts_in_message":[],"_revision_number":5},{"id":"2bc106ead99bb37441e3941a7a0ac29ebda6791b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-04 10:13:45.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ http://swswe-opsci001p.aristanetworks.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"30d2aa0ee72deb9991cc72e40e5fdb72cca6710f","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-04 10:32:42.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"7ae64701f164a2c7673022643161b617d67685f7","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-04 10:43:51.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f1d838bd059a2d7df5477104947fd9f3ab46a0a0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-04 11:51:43.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"7ce5d2738b780a389e2fc683f09b97a1f285d934","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-04 12:24:09.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/733033/6 : FAILURE in 1h 39m 32s","accounts_in_message":[],"_revision_number":6},{"id":"ebe7ebe01160443321bb1a9d0fd1d86f537539ac","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-04 13:19:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1e0ab4bba9a8de07aa0cdc43b205110c34683066","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-04 15:19:30.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/733033/7 : SUCCESS in 1h 59m 31s","accounts_in_message":[],"_revision_number":7},{"id":"2665a33d8ed7fc6e4d60b7f50e8804f4c338bd84","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-06-04 15:44:17.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"fc97db1da0d58664e7d4006fec92afbfcd0b82f6","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-06-04 15:47:42.000000000","message":"Patch Set 7: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"94d9af5b80cbc81ff5355e18fb76db647dc93903","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-04 16:17:57.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"9ffd95e21b6a73f6dbd3e689e6d68f6ca489dcd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 16:40:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/560e14fd162f4d9e8f7e173a3aaa7adb : SUCCESS in 46m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/927fe79ca399474299e57e4516ee1090 : SUCCESS in 45m 17s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/f7fb25fd7c8e4fd2ae772af9e53f711d : FAILURE in 2h 15m 23s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3374140eef454f4980c58cb66a5a95ca : SUCCESS in 49m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a678dcf628974539aa92e418abcd4d5a : SUCCESS in 1h 41m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c8e69cfd7b43447f86439a2dee4c8b52 : SUCCESS in 1h 24m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/499a8ba3f40a4063bb15a707165db2a6 : SUCCESS in 34m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/022686870dea4f9ea133b08e0f81b085 : SUCCESS in 31m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc480a9db72644d093c8ea82ffbe40f2 : SUCCESS in 12m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/905a02edca0a4a93b61976ab71c35539 : SUCCESS in 36m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80afd284778d44b1976013caefed1c4a : SUCCESS in 36m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38efdac36f89439eaaee35bcde96f77c : SUCCESS in 11m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7b31eec911524c8b954a97bda4c13fc8 : FAILURE in 59m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/687f267de518417e8d353a36bef3d033 : FAILURE in 1h 28m 04s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/bf395a6726d341fc9095dfabcdbde1fd : SUCCESS in 1h 46m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8b8b83a585064948ad48f73a669a136e : SUCCESS in 1h 15m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/82b405421e344d81a90669721da95e28 : SUCCESS in 1h 55m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f94a421b98d04b7d97f10f6f26740502 : SUCCESS in 1h 13m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/54bda0d8b84b4043b48b453af86cf96a : SUCCESS in 1h 18m 27s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c922abd6dc7b4f3ca45c88e18087b8de : SUCCESS in 1h 02m 40s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/31dea8e2e78e4e0ab64114978bebcd9e : SUCCESS in 53m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6517c7d7b358443aa03bddb847ed6f9c : SUCCESS in 1h 25m 39s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c659592c327141bb8f46e74f3074a71b : SUCCESS in 1h 54m 05s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4c45b08bc59449b3a522dfa60fbb3c4c : FAILURE in 1h 34m 09s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ca0745ea260349ed8d2b17e3152de128 : SUCCESS in 1h 12m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/15726a8378204012989199e188201b53 : SUCCESS in 54m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5f44691abafb42b3b33a00fa3029d094 : SUCCESS in 2h 15m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9281a1a0b5d142cda301d9357b3c9120 : SUCCESS in 51m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/76f235680dae46f18e79aa932280d1bf : FAILURE in 1h 46m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/85c4d9946c2b4b559bcd009eb2b86dbe : SUCCESS in 1h 45m 52s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ecb78d2f40594564a3071d1c327205e6 : SUCCESS in 1h 16m 50s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d20c9c00fe3a4bee9f7c0ab24e52db00 : SUCCESS in 1h 57m 50s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b37e5a6df73340349e4794bf71dbf235 : SUCCESS in 1h 47m 05s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/eb22e07345a14a83b7e36ee354370e05 : SUCCESS in 2h 34m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cfec4b427fbd44038576f18fb718e778 : SUCCESS in 1h 45m 29s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/5e1a6a753a9643f1b7df7eb52dced535 : FAILURE in 2h 15m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2af5afe580bc903f5e4393fd997879d074f9c152","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-04 21:29:35.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f3f921110104291649c80eefc543cc791161591a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-05 03:16:12.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0894a41fdbbd20cb4542a91bef889188fa008d93","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-05 04:30:35.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"98e9e0919c358efad1f8e418da2d5f9204e3c4aa","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-05 04:57:31.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/733033/8 : SUCCESS in 1h 40m 43s","accounts_in_message":[],"_revision_number":8},{"id":"1074b4654fa17e6fddf2e9719989f61dfa63d000","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 05:31:14.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fb9a950aa3ed4721b9f043ba1cdfb1f8 : SUCCESS in 43m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b7ed62ec5e264a28b6ffbe3a783b1311 : SUCCESS in 34m 32s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/bda838770818419890dc46c304ab9dd2 : FAILURE in 2h 09m 17s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/da0bbff0ea094889a49a30c390868578 : SUCCESS in 1h 11m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e54d4c38d39d4904b70b5da8f098bc2f : SUCCESS in 1h 12m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0769fd36fdc24af3975c904b1ed55569 : SUCCESS in 1h 12m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1959f500a9a1461e92d46cfa3aa89ec0 : FAILURE in 35m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76ee349110e2480cb6a74f5146b8b6bc : FAILURE in 26m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32c304e263e8471fb36849e485f00f6d : SUCCESS in 12m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/631d73a0d3254212a463d106f93f2850 : FAILURE in 26m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e863b8f3d694ef58d9dbc1a0f24c91c : FAILURE in 24m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fcf33b105314c24a406a76815ef110f : SUCCESS in 8m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/29a66795c5f349f69be8f24d0a1a87fb : FAILURE in 53m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/192621fa88e84b5bb7a463aa759075c0 : FAILURE in 45m 41s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ae4f7adb28944c038ba076abacde9817 : SUCCESS in 1h 10m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c4e73e129b3e485facf66c6021fd7597 : SUCCESS in 1h 05m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1a55c21d8f314afcb7dbdff6cac8d1f3 : SUCCESS in 1h 59m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6cbadb906bf54f079651a73e3281c3cc : SUCCESS in 1h 05m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/05f174c1a35c470a87f9df990ed287fc : SUCCESS in 1h 06m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6e43e090a44c4078b11d9020e074b5df : SUCCESS in 47m 33s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c80d982c3b1414d9e4be25bf7724b42 : SUCCESS in 1h 11m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c94c04662dc8456881e3ae5f4e1eff2f : SUCCESS in 1h 09m 35s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/41a6655559f2463ea2a745c02c956a1e : SUCCESS in 1h 08m 58s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9f39c78af6034bd59af2d73934ca263c : SUCCESS in 1h 35m 40s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7ae144c9ef4f4494b411e2e98a2bd889 : SUCCESS in 1h 06m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5168dea0c4394db48c824e65398d48ad : SUCCESS in 52m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a7c4a30c5d7d4770804aa25d3554dfee : SUCCESS in 1h 58m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67a317ccd4b44ff1ba13ff9a27ca90a1 : SUCCESS in 37m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/94cfda18594942f3a81e4b858cde7a1b : SUCCESS in 1h 34m 41s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/69902cc76bf7422285ead597087c1114 : SUCCESS in 1h 11m 21s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d5948c64e8d04bb085e7f377974a282f : SUCCESS in 1h 13m 36s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ac7651a3411e40a5a399fd0286e6a06c : SUCCESS in 1h 49m 53s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/609a151df3744e8b902bd762a7ef5ef8 : SUCCESS in 1h 45m 44s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/61d4b32668684cfdb3f5923d5cb7380c : SUCCESS in 1h 55m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/09faf3c334bd4a38977ceabdc8d7aca5 : SUCCESS in 1h 49m 54s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/589edb854df14eedb25a4770bb1ec8aa : FAILURE in 2h 01m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9d0fb8cea7406e3390c241b623a2385f688124bb","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-05 10:10:12.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2132c42b6e39dd1c19f9dec859837eac28d152cd","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-06-05 10:31:59.000000000","message":"Patch Set 9:\n\n(1 comment)\n\nI haven\u0027t reviewed the patch yet, just left a comment on the maint module.","accounts_in_message":[],"_revision_number":9},{"id":"3ef16df0c2a538f9f8b5d42a41192f6185f7326c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-05 11:52:54.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/733033/9 : SUCCESS in 1h 42m 07s","accounts_in_message":[],"_revision_number":9},{"id":"8b8e72f0adcb71c34239b65ed70f8c05fb0b2e32","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-05 12:34:57.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ http://swswe-opsci001p.aristanetworks.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"d7b3474f420a8cdcf1dbf53ead148243e5ca5d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 15:12:49.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/45a5363bde21463193741e8002c30f3c : SUCCESS in 57m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0e30df055dc14864bffd98e9591dcd8b : SUCCESS in 34m 16s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/ef6778ed2ccb40ae8ce370750d1e3272 : FAILURE in 1h 53m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2ecb365ba29743d5b0071347ee3df782 : SUCCESS in 1h 30m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a0d4ba2a587f427aa902455000ab54f4 : SUCCESS in 1h 05m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/adb49787077a4b7390cb62dbafe4db55 : SUCCESS in 1h 01m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82e00d4895c1486ab17eda72a68faadd : SUCCESS in 32m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c1ae759bb38439f8579ae1437448366 : SUCCESS in 29m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84dd76f4137f401a96f51521d7ddcc94 : SUCCESS in 11m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc449357dfbe422f8fc9af0eca51b168 : SUCCESS in 30m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b60d662b3233417b8a6315a614e9a5bd : SUCCESS in 26m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2ce8ef4907844aa8fcc49d79ba138b1 : SUCCESS in 10m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9fdf7309a1894832a9370aa7ef228a30 : FAILURE in 45m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/75b4d36769ef411786db4a3e5c4462f3 : FAILURE in 52m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5b23a15e1d084d48949e348e5861ab61 : SUCCESS in 1h 10m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6f31baf4963c49919b6ce4be5dc9447b : FAILURE in 51m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/04b1e99d048a4ff593177b606c90f2cc : SUCCESS in 1h 22m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed0bb6bfeaa14535abe3f2eb5eedea06 : SUCCESS in 53m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ac131f75650a45a7b97ae5d036474f73 : SUCCESS in 1h 23m 31s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/df3992e2eb35422b956d5a7f0758637f : SUCCESS in 1h 13m 26s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c82787f0883e41c7a760139fea8fb8fc : SUCCESS in 1h 25m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/568f876ab01e4a519c3f4b1271312c69 : SUCCESS in 1h 09m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/853d4f0308664aa09b151859778a8205 : SUCCESS in 1h 41m 12s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0d6c708c627044f8b44cc495a9c36ce2 : SUCCESS in 1h 43m 41s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/60adaea03f8a4723989cd5f5e9e4ad67 : SUCCESS in 1h 09m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/30bcec3b1f0b41c998b74ac5e951a8fd : SUCCESS in 53m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a82de267b9914b2d8b0c2282bfe03bcf : SUCCESS in 2h 02m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9750893645a34f599db8f2307f7ea082 : SUCCESS in 1h 03m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4604ec97d14a4fddae93e23e6cc332d3 : SUCCESS in 1h 14m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b7116b4db8e64212a03f1ffc0623bca4 : SUCCESS in 2h 15m 25s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/633a9dde5dfa4275939d1abe9f2f1134 : SUCCESS in 1h 18m 58s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3f345b1827dc4131a564e759e4282cce : SUCCESS in 1h 55m 35s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eb36bda31e7d4c0ba2bcd3affa936872 : SUCCESS in 1h 54m 28s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/92d8da5f257f4466b6f56db2265429a1 : SUCCESS in 2h 50m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cf1c2da4330443f3881b637b9e8d2ec3 : SUCCESS in 1h 18m 43s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/ea1afa31b7664450a0110741893b07d4 : FAILURE in 1h 56m 59s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c957904032e2393c7e459ca002f9205f2dbd2248","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-05 16:59:14.000000000","message":"Patch Set 9: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b4e943bdb5cbabf0edb265a291878c710b007aa4","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-05 19:21:29.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"091e7306e3a8712eb4595238fe349fd807c6db2c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-05 20:26:32.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"70feb9ba4cf55beb623b34ca4b2a09a79b92c728","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-05 21:03:15.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/733033/10 : SUCCESS in 1h 41m 11s","accounts_in_message":[],"_revision_number":10},{"id":"6f42dec498da834292c91cb9763b2d93edd84768","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-05 21:39:21.000000000","message":"Patch Set 10:\n\nArista third party testing FAILED [ http://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"74f28f8d9f86f28422efb498a66309bb30a8f6d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 22:05:40.000000000","message":"Patch Set 10: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6b655ee849a7457eba22759b7c2584a7 : SUCCESS in 53m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d0544775998646398dfad3d7a36e113b : SUCCESS in 40m 16s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/380d57faf00143ad9b554eda0426b4f1 : FAILURE in 2h 02m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5f67c9574f4c4d7aa46b3959904ad230 : SUCCESS in 1h 10m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8ece69924cc548e3a61a188215c012e4 : SUCCESS in 1h 24m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fa5ed205eea346018a1a6072a7775be8 : SUCCESS in 1h 21m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d88705d54514b7aa31d8a1ed7b585da : SUCCESS in 38m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5184f6b50bc4cc6834567a6ff92f60f : SUCCESS in 32m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95ec167f475e493cb1459fd0613fb975 : FAILURE in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1b09797c212465e91a27898e2ed93f2 : SUCCESS in 26m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3cb60cdde474fdd978f859733ea9db9 : SUCCESS in 24m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6f05f82b9f24e2793b9b19850181b7f : SUCCESS in 8m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e444b123866a475e9272e76d18d96b6e : FAILURE in 54m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb06e0618a7c4febaf57fe0501d417a0 : FAILURE in 40m 20s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b291ddf3bef048d0bc50de5e32ed2956 : SUCCESS in 1h 02m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8bbc813f6c684b0c8a163609a58b39ba : SUCCESS in 1h 10m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/306f777f187a4e29ad6c52fd4a7ceb6c : SUCCESS in 1h 22m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a9d19642b974524950d41ae028f2615 : SUCCESS in 1h 13m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0812714051d04785879b3f88697e766a : SUCCESS in 1h 11m 38s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8aa26a3ec70c4b92a116d79dc45e960e : SUCCESS in 43m 57s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/42d070fbe1df484095ca75b1d345f132 : SUCCESS in 59m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/26fac07cb555468dbb28cbecc5b8f591 : SUCCESS in 59m 41s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e9e63734605e49fdb243e5dc6d6a6e72 : FAILURE in 2h 08m 43s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/72684c66001645f5bb8742b29252cc9e : SUCCESS in 1h 38m 04s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5224744344044e4b935344684172c9fb : SUCCESS in 1h 07m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/63cc75a8758d40bcab32545c4bbd03fe : SUCCESS in 37m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0aaf09b107264cb5abcc4d1c72831338 : SUCCESS in 2h 07m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5264bbc0465b4803b921c273b0691795 : SUCCESS in 52m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/297270809a4d41e588e4396aee6ceb6f : SUCCESS in 1h 46m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7c359836ca604c0fa6978805be9f712b : SUCCESS in 1h 58m 37s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b3ee6c4e1c944b8eb08415332015e250 : SUCCESS in 1h 05m 20s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3c5d7b32089f48789203506c3d67ea35 : SUCCESS in 1h 44m 59s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6e825c1d90604bbc81dd6c05f1ad4831 : SUCCESS in 1h 45m 21s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/de72ee2f10eb47ee9cdbb255b5fec51e : SUCCESS in 2h 36m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/91b31e4df7b148ecbf8d1dbc6d6a93e8 : SUCCESS in 1h 48m 45s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/dd56b88bef864527a53a5287113116ae : FAILURE in 1h 41m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6962d96193ee40492212ec4ef7d27508a94b5ba5","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-06-06 11:02:27.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-06-06/733033/9/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/52f7113 : EXCEPTION (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"963f1747db06327c8f23dec6127b7540343bed2a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-06 11:12:46.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"938252ca8c35ccdc1c8ea5f04e2bff3d391747bf","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-06 11:45:57.000000000","message":"Patch Set 7:\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"7c5c520bf95a970b202c8567d27fc9b16952105a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-06 11:52:13.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"5ff6ff141c02ca6ce28c12490f5d4da02dff127d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-06 12:29:21.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ http://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"c6e2a2c1b131edb91290855f7dfc55dabb938579","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-06 13:33:08.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/733033/12 : FAILURE in 1h 40m 19s","accounts_in_message":[],"_revision_number":12},{"id":"f8e994ac19c4ff24a757e692feec05a993da7158","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-06 14:06:35.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ http://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"c4c37fef82021d7a9f3c76ac6a4c8b2e8048ac4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-06 14:48:59.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/943e97cdc8534becb326e8e43384bc8f : SUCCESS in 49m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/04b4c005f71547ab8318807a8c3550a8 : SUCCESS in 31m 06s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/fc9dadbc6939492eab6d876245e18be6 : FAILURE in 1h 37m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/26d0eb0ed4464dd18d03a9ef49524a43 : SUCCESS in 1h 12m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/733465c50ce74cc2b4fe9e6879b81689 : SUCCESS in 58m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4f9c0fcd855643109858d682a48442c6 : SUCCESS in 1h 07m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc33cdec609c4743b5d4aba99cb7f04b : SUCCESS in 31m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b2a5893eb57422c85e7fabecec2e820 : SUCCESS in 25m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f1201e943004cd8bf1f7b6f40402ed4 : SUCCESS in 10m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32ecb6793f2c475c9e647f7f60cb7c15 : SUCCESS in 24m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6608e462663f4a799fea8e42fd81dc32 : SUCCESS in 22m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25f7348ffb1d4dfdade91c3487988c4e : SUCCESS in 9m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5f8bbe04e6b949fe8aa1034dcd814f98 : SUCCESS in 45m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6133f51ef12d4614b276b463b0d80cd7 : SUCCESS in 48m 37s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/333dec83ecfe4d7eb34cc91cc5a45d50 : SUCCESS in 1h 06m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e4f64fb8d80c431b8eee2291ec3b8a7c : SUCCESS in 1h 04m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d07b4483c49440e49705f60bab91adc8 : SUCCESS in 1h 39m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/084b43d2f1f649809a8a075700b264da : SUCCESS in 1h 01m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5462cafead864236aca895fece4c0ad5 : SUCCESS in 1h 05m 44s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7fdb85a796a44bacbbb0aedd5d7170b0 : SUCCESS in 1h 35m 41s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ae1c8c69074f4ba3943d71b152ff3fa3 : SUCCESS in 1h 07m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/364594cef2ec42aabe6ff473b0f7d226 : SUCCESS in 1h 03m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ad24440bce41415385ae7fec12b6632b : SUCCESS in 1h 25m 58s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d72c5c714cea4a40b66fd0761c71446e : FAILURE in 1h 17m 32s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0642b46e980a493abe76583d6706d74b : SUCCESS in 1h 36m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9bb65324641044e5b5a677965681c431 : SUCCESS in 1h 08m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/231aa1612b29413d8503f10cc7006a47 : SUCCESS in 2h 25m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/701982a5eb874b0abed72297d51a8d7e : SUCCESS in 50m 55s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/883c8e7d35f74cfa83757c4e781d9d73 : SUCCESS in 2h 18m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/85f6d28f34814cce931b6ad913c4799d : SUCCESS in 2h 40m 36s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ac34dbcdb7e2448984dc0a6109224000 : SUCCESS in 1h 24m 17s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/851aadd96cd64da39e78ac605dcd7378 : SUCCESS in 1h 32m 29s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8325e44c8f2840389885fbb5d46d37da : SUCCESS in 2h 05m 24s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1ac5aad54c6f428298a541a406a2cd19 : SUCCESS in 2h 41m 05s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0da8c67f35c04f848f638eb4d4e2f4b2 : SUCCESS in 2h 06m 57s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/9775e116ea6f4420a6f8d7fa055429ae : FAILURE in 2h 55m 06s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"30f85265896d0b98a8edf32cbadb2ac936ccd0c2","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-06-09 21:35:07.000000000","message":"Patch Set 12: Code-Review+1\n\n(4 comments)\n\nLooks good to me. Just a couple of nit comments that I wouldn\u0027t hold anything up over. I\u0027m not super familiar with all of the PG code, so making sure other people get a stab at reviewing as well with the +1. ;)","accounts_in_message":[],"_revision_number":12},{"id":"55650b0353bdeff4cc08045294176e0bffedf5af","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-10 10:33:28.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"f327502a7297bb641aff749ac9ee240197471cb6","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-10 10:33:43.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"94dabad957f7cec0329a11fa490a4b64ead11a92","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-10 10:37:01.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"0218d9d977cb6806810c0d47e64430bf1aed9291","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-10 11:51:30.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ http://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"0da13c0787d4d2598b551b09d4f2831c25df1ff7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-10 12:32:15.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/733033/13 : SUCCESS in 1h 57m 57s","accounts_in_message":[],"_revision_number":13},{"id":"62884f8c2defb84d4c4e6d2f2894802e17c0af5f","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-06-10 12:49:48.000000000","message":"Patch Set 13: Code-Review-1\n\n(4 comments)\n\nThanks Flavio, I still don\u0027t grasp the external uuid part so left some comments and questions inline.","accounts_in_message":[],"_revision_number":13},{"id":"c532c066c72b09233f55180a6a626e310793c71a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 13:14:47.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/64eb376fa3e1469dbcf2d45295ea8158 : SUCCESS in 53m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b3741cf5c1c84d99a84b4d8d7608480d : SUCCESS in 40m 03s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/9affad6a773642cd9480e171471a14aa : FAILURE in 1h 37m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/aa5122a22b8f411e9187040a5d5e7550 : SUCCESS in 1h 20m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/67605b2c65654a5f909747b6d4710cc1 : SUCCESS in 1h 17m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1e6a98b5a176481f83a0aa24af30c822 : SUCCESS in 1h 10m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d86ec96ce5e948728482fa0a34939ab7 : SUCCESS in 33m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/586212d1bda74ee0b59ef9975960f4f9 : SUCCESS in 32m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f3491177e404b8fac0c33c529776d0e : SUCCESS in 12m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d19e3b097b34ebd88c0641358556979 : SUCCESS in 31m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fecd2cd8573e4f6097e8072f11f16d59 : SUCCESS in 26m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc9c89f76cb0466a87e425abad409c52 : SUCCESS in 10m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7dc714fad1c54f3ea84b4edb3582bd11 : FAILURE in 51m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b35ff8785dd4442bb67fd85b16fbb84 : FAILURE in 58m 00s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fd5a4c88add846fdae67cd0c0b491053 : SUCCESS in 1h 11m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0d0bea2c765e453fb983991d27db4748 : SUCCESS in 1h 21m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d87510824d15441e8b999dcc62394327 : SUCCESS in 1h 54m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8fa93c40ba5d462283edfa7826907194 : SUCCESS in 49m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0d0e8014b7ba4118abde0ab01c89bb6c : SUCCESS in 1h 12m 17s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/70f9e40194d14e29a4637166067ad58e : SUCCESS in 1h 13m 48s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ce14786ad77b44f0bd783afbd8426e88 : SUCCESS in 47m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d9c7e10c98294deab126b4bb33937f2a : SUCCESS in 1h 15m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/146a0ad74548480aa1253fbf86d4689c : SUCCESS in 1h 35m 24s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d2440982971a4a4f977314c57bf4ee58 : FAILURE in 1h 28m 14s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6c5a7d9f143f46fd85eac6a83aa8fbf7 : SUCCESS in 1h 15m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9d7e232a4c124a00b5746a6513b61dec : SUCCESS in 55m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a7f3d8a5718647b79d3e28ea4880efb1 : SUCCESS in 1h 59m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37952d8cb8f7460688337f25772efda4 : SUCCESS in 50m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ec288b6121b04d15afe4c7d13e90ccb8 : SUCCESS in 1h 11m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/eb889d4c3f304d0b8370d2fff57fb52b : SUCCESS in 1h 16m 40s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/18f2ad4fb60c4dcebeb4d84ef3869645 : SUCCESS in 59m 54s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d7a472b12c00421aacd65830a2950cc0 : FAILURE in 15m 18s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/103e157cda0b475caaba54070d70feb8 : SUCCESS in 1h 12m 41s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0e37158a333147c590fdfb53f6a2a495 : SUCCESS in 2h 33m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/57e466e485aa48369d3be52b9dd18676 : SUCCESS in 1h 10m 31s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/8b8bbc7f108a439c93b77a2026f86d5f : FAILURE in 2h 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"e06e30c0e80c74380c5ad3bdf80af052a09d371d","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-06-10 13:31:34.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"649ce1f6e9a99bbd16d27d623d8c23ae742a2506","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-06-10 14:19:27.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"94ee1e5f183d3ba0a09c66333442f234d20e01c1","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-10 14:51:11.000000000","message":"Patch Set 13: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":13},{"id":"eea1ad0d9178f451b4679d3f966e0954efd9de1f","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-10 16:17:37.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"174c60311076f9c8e8c12684bf2dbd877294a5f5","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-10 16:17:48.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"8e231c317d6735ad9578fce6ecaae52b380e7485","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-10 16:17:56.000000000","message":"Patch Set 15: Workflow-1","accounts_in_message":[],"_revision_number":15},{"id":"72fbf02d0a49608820d876cf84aa7e9012caaf4a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-10 18:02:42.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/733033/15 : SUCCESS in 1h 43m 48s","accounts_in_message":[],"_revision_number":15},{"id":"300f191f208d67db9e05d1dc431e5a46942b077c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-10 18:46:11.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ http://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"972c041bbb58324863b2bd8c7a8da392087e51b2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-10 19:20:01.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ http://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"801ed68146dffb976f3e295ea8081860a236fb6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 20:11:17.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/af6a501cc0a74817989d947ed8b9ffca : SUCCESS in 55m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ce1020a6bb044c519aa9649154d491c7 : SUCCESS in 40m 35s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/5729e8003aeb402599caf4db888c0052 : FAILURE in 2h 04m 31s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4f504e6e56d54d16bd4b337e1d94806c : SUCCESS in 1h 16m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c6225f491c034940972f890cdf0438fb : SUCCESS in 1h 03m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d5bab735ce9b4cecae02b32142cb131a : SUCCESS in 45m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/557364da7d1f4e1cb9a3a9d058792009 : SUCCESS in 23m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1914f13e8203436b980e0aeff1621c47 : SUCCESS in 18m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec6d399de5f345269ff74057c9fd61e6 : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80d9c6ddee1f45d89697f61347081e56 : SUCCESS in 20m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c46ed3d4c494f6cb739e7b149862fc9 : SUCCESS in 17m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e38dbddb9b1478d9511561878cf335e : SUCCESS in 11m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2aedc7bf29b242f2b9433bad0d641ae9 : SUCCESS in 47m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fb369884c3264185bce4beaae2749378 : SUCCESS in 51m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9b22c8d577444ecd84399b997d040f92 : SUCCESS in 1h 37m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/028f05f6e64d4e44ac763f6a09da9925 : SUCCESS in 1h 12m 05s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4d5242a14aac454381ca18ae70794fc6 : SUCCESS in 1h 46m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0fedeeb23c21439c80a39b3cd71fc2a0 : SUCCESS in 1h 10m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6c4bfd9b50264c519e9ce7f83e5a51c3 : SUCCESS in 1h 09m 26s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/59d30dc8f5244568add6b87bf0decb58 : SUCCESS in 1h 20m 39s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad01bc8e19114e8fb2e8ec729943e0e7 : SUCCESS in 1h 17m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/653769580eea41379f8f4770f17d2cfa : SUCCESS in 1h 12m 13s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/96f2910c2fcf438796b03cf6f992d8d7 : SUCCESS in 2h 04m 39s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1b89852aa6b945a8a99119119ae1188c : FAILURE in 1h 01m 28s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/95c2d3cf600d474089130066cbf9c7b2 : SUCCESS in 1h 14m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f0629d6d94744b7791cdd7021de2e138 : SUCCESS in 1h 13m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/126bbec6a366427ab18bb2bfe58404f1 : SUCCESS in 1h 59m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b50d041345d2499cb96f92f7836b2d7d : SUCCESS in 55m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fba0e56a632f44d5923e4f4d1cdf24b7 : SUCCESS in 1h 51m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1c82bbc22dc2402192c7f2678892a755 : SUCCESS in 1h 49m 18s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3e4fbcfbb2e64491a0d4fc28d8a1575f : SUCCESS in 1h 19m 40s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/79f628d14c474359bedbbce83c0ec7d7 : SUCCESS in 1h 54m 10s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0133be3acad2471db34b038013b1b63b : SUCCESS in 1h 59m 48s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e8bbd90e3c5a43938a6699a7e293ce49 : SUCCESS in 2h 50m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6afebd6edf2e4018a99b207c6e353e43 : SUCCESS in 2h 05m 38s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/5358bc5f955242cea7092452f82d4580 : FAILURE in 2h 06m 46s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"1134f4598c3c2aed20d2034111b669aa1c76574a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-10 22:14:42.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"777a622a5778938f92396b779afbd94626aadcb2","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-11 02:34:39.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"9dafd59ea704eb6d1a923755f2c923a18ae023a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 02:34:54.000000000","message":"Patch Set 16: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":16},{"id":"7f8f04bf38743f2ee6fdf9c5babbaab09f6edc0b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-11 04:06:36.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ http://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"279df9104df7ade41fc5a92b85ed4e0dd0052c6e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-06-11 09:18:25.000000000","message":"Patch Set 16:\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"6c23236a088940950107ec8f11856256940667c7","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-11 12:18:52.000000000","message":"Patch Set 16: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"3aaeb99282f2d8a4ffbcc473da37cfc1fa93a396","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-06-11 12:21:16.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"89041cb1eec2e47ce8ad854c2a7fde8765724ac2","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-06-11 12:24:36.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"83e9786138c29db89eb66c184be5a29561cabf7a","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-06-11 12:46:05.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"ebd416bc517a9b8be27fbb209823f4bb3f5a22d9","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-11 19:23:46.000000000","message":"Patch Set 16:\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"7875e8990a8a90d2e8669ef375f7a0b8f60f5922","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-11 21:53:44.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"8bd4e1e994be5c5c1cc7a008483b4f90fddff768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 21:54:00.000000000","message":"Patch Set 17: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":17},{"id":"ef96441d5bb3e222a39565ab6033e66e94b867ba","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-11 23:10:48.000000000","message":"Patch Set 17:\n\nArista third party testing FAILED [ http://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"44ecda52810fbad90a2f3c14ffa1dcd9eca1d5c0","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-12 02:18:39.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"cfea255b7cf4eb0cc29e299e4e4b15d950b74206","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-12 02:45:28.000000000","message":"Patch Set 18:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/733033/18 : FAILURE in 26m 15s","accounts_in_message":[],"_revision_number":18},{"id":"ca2dbdea17708e82dadcdddc765c722af0fb0eec","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-12 03:14:13.000000000","message":"Patch Set 18:\n\nArista third party testing FAILED [ http://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"07eaf673ee714df84c92f3a5e5f7e8cdb86bb85f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 04:54:48.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/341e85ed299f41b7990dd8bf6393d6f3 : SUCCESS in 1h 15m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2399a1a9dd8c41c78c84f728d22e6424 : SUCCESS in 36m 29s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/b70b820890e24df1ade0aa15d960893e : FAILURE in 1h 38m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a4b82945eb8640539ff46bd5ed0e3568 : SUCCESS in 1h 03m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/69c7a19e91a646f1b0f12bd7281e422c : SUCCESS in 1h 07m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4eaf7533197d47ba8ddff6775ee85dd1 : SUCCESS in 1h 00m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff24ea487ef642ddb66c7ab9a9e7a5d0 : FAILURE in 30m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e7e9942d8b44dd497997471bff798c3 : FAILURE in 27m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41c5518f9a864010bd166ddf5ca0aef7 : SUCCESS in 11m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cd53fafc072422abfbc6b24f6318c03 : FAILURE in 28m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da1bb49afd214ff7872d1f25f0f6c649 : FAILURE in 28m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa08249024e24a319ed7cf9fe8b0e480 : SUCCESS in 7m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/296b838fc84041caa10e00ab547106ef : SUCCESS in 1h 18m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9b9fe844c7c243109f293cf98febe9cb : SUCCESS in 1h 22m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9d431a6f0052436a8ba137870bf4ce6c : SUCCESS in 1h 41m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/63120f8f40514db08ed94d425dd8fcf9 : SUCCESS in 1h 42m 34s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/11bdf74571474b59b567c110c240d08b : FAILURE in 32m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa60ef4585b44be2bbb6007ed50e5a35 : SUCCESS in 1h 45m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c1261baca46a4c269cc7384c1f464ff0 : SUCCESS in 1h 34m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3d81731075d242f38c0c972f7c4c4f96 : SUCCESS in 1h 43m 04s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/9cb24c190082471ea2d7eefd323c73f1 : SUCCESS in 1h 04m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/70554815f0b54aa4af64b4d1ffe312a1 : SUCCESS in 1h 44m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/eeeb54129ad64ab38133840d6fe7a052 : FAILURE in 1h 20m 30s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/13184b38daef4e3d98dae57f269ec1ed : FAILURE in 1h 55m 13s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a7950b16e97240f3bf87702cca9123d9 : SUCCESS in 1h 15m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f20ba50a88c545f4b2ef7125ff3db7f2 : SUCCESS in 55m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9111f08b2cbb448caada719265450610 : FAILURE in 2h 15m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a271dfac24644841b88e764e2ae7f5c6 : SUCCESS in 1h 02m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2d7327ef3a9e4be49dda35429f52ca35 : SUCCESS in 2h 31m 16s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6089e872ab76444182694e95f4a6aaf5 : SUCCESS in 1h 14m 27s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ceaa12c1cfcc40fcb928964dfd78c116 : SUCCESS in 1h 17m 01s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4b07f04035df4487ab48697c8444ce4a : SUCCESS in 1h 44m 28s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/066660ee87b542e6826593200ef5932b : FAILURE in 30m 23s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/60c8d7f4dc9b461fb44f124e93f3c255 : SUCCESS in 2h 27m 58s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/48c9dee2ebb848f69121634376304f85 : SUCCESS in 1h 38m 49s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d4cf9197be274605b24a5a0be373b9df : FAILURE in 2h 25m 29s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"cfcd33eec7e5add0ea3e9f21034b2b0d504932f7","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-15 12:40:26.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"0eb69ec219478cff5d30e1586356dc53b9369c29","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-15 14:16:18.000000000","message":"Patch Set 19: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"c04727b3cdc1a7d4aaaf6aaf4f2a3f77eca9e1a2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-15 15:11:17.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/733033/19 : SUCCESS in 2h 30m 08s","accounts_in_message":[],"_revision_number":19},{"id":"9b645576a8cf84fcd383efd95a9983525052fa68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 15:19:06.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5a1f825e956c4eab9665e2160055502a : SUCCESS in 1h 03m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/52b60faed46a470a9ef7163ac0f90d5b : FAILURE in 33m 49s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/84d067374f2d4caf80ddeeaf232e7828 : FAILURE in 2h 00m 42s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c75146fdb44944fea8c4a53b83804f00 : SUCCESS in 1h 18m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5fe3ae150d8e41a389fdae527322a4fc : SUCCESS in 1h 31m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e423062c3a244c0fb3588ac092325d9b : SUCCESS in 1h 08m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fd09219be114cfb90e346e10b36bc78 : FAILURE in 34m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1ee7271d76e4174a16c4783caedb318 : FAILURE in 30m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae314a75bd9d4f75889e791ce5aaf6fe : SUCCESS in 13m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49c26e857b5449cdbc55c543743abd63 : FAILURE in 29m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e2e481c77e145f28d8a67fe24b65440 : FAILURE in 27m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d318b4b8da34610b433ca727b5003ec : SUCCESS in 11m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a7891bbb0d6b4c55bc149222c9697381 : SUCCESS in 1h 01m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f16c16fac032442ab633e5b10214337c : FAILURE in 1h 00m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/002d5ac753c64e58bf54bd58ec71e106 : SUCCESS in 1h 21m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7646005a059342e6a9ef09963f253fe1 : SUCCESS in 1h 14m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c118c264a2944cad818a7889aed32e40 : SUCCESS in 1h 57m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/924cf7b9401148689956eebd86a3cbf1 : FAILURE in 28m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/aecb5922690d4afd8a46079822968701 : FAILURE in 24m 44s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3711f835a70c47a583017647d2dd3b84 : SUCCESS in 1h 12m 28s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d1002d9d74a445e5b33e11227ccf2649 : SUCCESS in 1h 15m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3f94e7121c4c469e8fb216365b6255ba : SUCCESS in 1h 21m 25s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/40b1bc9b87fb4f5f86da370391a4b5e7 : FAILURE in 1h 37m 45s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/127ae85e693f449688c91f2799a2f28b : SUCCESS in 1h 08m 51s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/be365c60fdd8449591acf7b721aec90f : SUCCESS in 1h 28m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dd02cd35778541bfa89b9c7f81a56d54 : FAILURE in 34m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b1572d9c9b9743a98025908d0fca8412 : SUCCESS in 2h 20m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbfab53deb9945bbbbaedc25846f4417 : SUCCESS in 1h 27m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b00bc6fd3cbe4ce49e5656aee74ebe27 : FAILURE in 22m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/96b7bd2f7d214a69b787a1fb2e2b846d : FAILURE in 23m 37s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/be9bf8c2f7654eeb87b168a17de34f36 : FAILURE in 52m 58s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d4babcf9248d40c0859408467ff4aad4 : SUCCESS in 1h 52m 37s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5fd78b5ff1734d9884281ad624fa842a : SUCCESS in 2h 09m 02s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/18e7586e4ffb48a0b5a2d9da9ea8050b : SUCCESS in 2h 33m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6510ccc8794043db813a6f930d6286cb : SUCCESS in 1h 59m 24s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/26be32e2709242c3aed8e9637782b987 : SUCCESS in 2h 00m 03s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"0888f70b4e082073a19b34fd094237a7dab487b7","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-15 19:52:43.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"c9ccc945b23595b43c4d8ceb6d72e389733e7ac3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-15 19:54:55.000000000","message":"Patch Set 20:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/733033/20 : FAILURE in 1m 29s","accounts_in_message":[],"_revision_number":20},{"id":"ce83f5988df81ef692b50cca2b394da5691550dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 00:11:50.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/963f54a98fb94aa18d14b9ff07af9214 : SUCCESS in 1h 00m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/091f327012b24c898ce212560af82032 : FAILURE in 37m 13s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/985645b02ddc4223be4ef0faff1ccddd : FAILURE in 1h 46m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/686a5c8487624878b06fb6493ea7b628 : SUCCESS in 1h 15m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2daf3d005acc42f29058b001bf761e0e : SUCCESS in 57m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7adc0e6bc00d4cc08ef40fe3b2ce789d : SUCCESS in 1h 06m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb06cddf89ba4ae4be5a37a4fd778778 : SUCCESS in 40m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c24fdb9409c48e6a6a6d09570ffbda5 : SUCCESS in 29m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6231286f382401587d01747be2b12af : SUCCESS in 16m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f27c2cc4e3c5426c87865165b62cb024 : SUCCESS in 26m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b451199535c48b09f0a6d12ea3890de : SUCCESS in 27m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc4c261a60d42bc9dddbe672da08480 : SUCCESS in 11m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ee636030c53b4a9f8cf9bc21eada5126 : SUCCESS in 54m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/09ec356afd7d4d0d921d53c1152f1495 : SUCCESS in 58m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0f45991fac3140b28c9118bcd0e9f9d4 : SUCCESS in 1h 05m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d717dba269d45b79c3243ca633e4843 : SUCCESS in 1h 15m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/775250ee0bee464caec7fbcef4b483d7 : SUCCESS in 2h 28m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/424bdb4caf6740a493aa9a1aa767fb0d : FAILURE in 21m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5480f340c95f40198d01c4fbd28a55e0 : FAILURE in 24m 43s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a36f63ad55e4497ca62afcc06a3a6424 : SUCCESS in 1h 18m 29s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/115e033061fa44fe978e584f034512f6 : SUCCESS in 1h 14m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/86cae98f57b1496a84b376192da115bb : SUCCESS in 1h 16m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9be6a5a441d64bc68bfea53faabcae54 : SUCCESS in 1h 35m 36s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d4b8f3d4bcc54b46b2d0ca71e9df2ed2 : SUCCESS in 1h 31m 24s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/28fc26489dd04bb19685584106d63e7d : SUCCESS in 53m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bb9c504076224d4fbf956c1d938dcfdf : SUCCESS in 1h 02m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4fd7950fc81a42dfaa8cc2fa9a286685 : SUCCESS in 1h 57m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ec45a81a96945a3ae8d476b5d6b97a5 : SUCCESS in 55m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2400d9f412fe4ea8b3fd64ec511a98ae : FAILURE in 22m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/68f6daa997704f859ff8f08cc81af27a : FAILURE in 19m 54s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f55fe2e5f65641cfa67cbb3363c8e801 : FAILURE in 34m 46s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/17f59d0a830447b3be744708b1553549 : SUCCESS in 1h 17m 42s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/20b8fccdc7844ae5901e25cb0e3328c5 : SUCCESS in 1h 48m 22s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/03b745166cea4d278bd9c2c5ff82a15a : SUCCESS in 2h 30m 13s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/161055f171144bf799db1653a6478a8b : SUCCESS in 1h 49m 28s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/36a7111453e143aeb1d980e519538e89 : SUCCESS in 2h 22m 51s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"59f00c195d1013b52e24603abb1d24011a762dee","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-16 00:44:10.000000000","message":"Patch Set 20: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"6a1b5efbdc92a3f34b97fa0afd848d934467366b","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-16 01:48:06.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"2a5bf67f4c1693898966019fdbf10bf391fe4872","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-16 01:49:23.000000000","message":"Patch Set 20:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/733033/20 : FAILURE in 37s","accounts_in_message":[],"_revision_number":20},{"id":"b33cf5082911205e34af0a5968f10f4d5ba67fb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 04:07:13.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/34c8ae188d6f4fe989d67f6f4e0cc26e : SUCCESS in 1h 19m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d620711205e144bb9f80465cab782c19 : FAILURE in 34m 48s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/db0abc9001c741eca27aedc308ad0b03 : FAILURE in 2h 01m 28s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a627fd089b8748a9b3b92225ce10710e : SUCCESS in 1h 15m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2d090e50716443808c0a8ef936508b4f : SUCCESS in 1h 14m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/14bf9a0fe1444b888c85dab97562b37a : SUCCESS in 1h 23m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/132e17ed3f7f4db8a53e9f151eca0e10 : SUCCESS in 33m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b903c37ab4f40ceaad2ba0a232a24cc : POST_FAILURE in 32m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/272e70d437354521a01956520512144c : SUCCESS in 11m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a37045829bdb40e394faaeb42d5aa5fd : SUCCESS in 25m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/782d561b03ef42049da27e0c0cc0bd3b : SUCCESS in 24m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4add4223c408499794fb905161fcc3a4 : SUCCESS in 9m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cf40336edc6d4bcaa5f98404798e4f6b : SUCCESS in 1h 00m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/65661063540e42daa5b2aa6aa7f47214 : SUCCESS in 1h 02m 56s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/593841464a494eab9b2fbb5435e35d5e : SUCCESS in 1h 31m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fdc2dd36e66d4d29861e5fd7deef8d97 : SUCCESS in 1h 15m 00s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2978780c1b334c69837b46910ea6045c : SUCCESS in 1h 42m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e5af196a7b2f4acba5b4694f3974780e : FAILURE in 15m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e6b13a0273844789957247196d9eb56d : FAILURE in 21m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b610243e967442118f9c47656757210d : SUCCESS in 1h 27m 40s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/70c880b455b74f0ea3b87d8c7d340891 : SUCCESS in 1h 07m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d814054d67674a44bab88ff2c0e0c23a : SUCCESS in 1h 20m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b0e2d9a59a4149999be71a34147116fe : SUCCESS in 1h 36m 34s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8fff533ea478480198b8a6f37b953deb : FAILURE in 1h 39m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/982751e4fa9d43118c814cb363e04608 : SUCCESS in 1h 26m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0f32e5d4f8884b139d6effae62c21f08 : SUCCESS in 1h 14m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7ed90d385e4b443b9310514422daf7a3 : SUCCESS in 2h 01m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fae9b70cfc3940caaa0533951e81a3b0 : SUCCESS in 1h 16m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7ff40b1e3c5144e6a9834dd517fced7b : FAILURE in 22m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5c3814edc11945c8997d3257878352f0 : FAILURE in 20m 23s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c515225f6bfd4512a4a3e9dc13070ba2 : FAILURE in 32m 20s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6e88b3ae35404682b1776cb597f8d577 : FAILURE in 18m 23s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/92ca2c5f06ed463c92bd4c0f5fafeb01 : TIMED_OUT in 2h 13m 10s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/abddc77e4dc548d78b6fc1e5684ba4ef : FAILURE in 20m 13s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a4c7a525b5c847f7944d8129f6c5bc2e : SUCCESS in 1h 45m 44s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/432f09ce2bd14dd9895471325a9ad157 : SUCCESS in 2h 04m 58s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"c51932f45240cac9624bd9a84a9134821e0d37d0","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-06-16 09:56:12.000000000","message":"Patch Set 20: Code-Review+1\n\n(3 comments)\n\nThanks Flavio!\n\nI think that we need a test in tempest that doesn\u0027t make use of the Neutron default security group to make sure that we cover this scenario properly. Do you agree?","accounts_in_message":[],"_revision_number":20},{"id":"ef018f9d1a260f7434912ab5855cea2c84bf36e8","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-16 10:05:33.000000000","message":"Patch Set 20: Workflow-1\n\n(2 comments)\n\n\u003e I think that we need a test in tempest that doesn\u0027t make use of the Neutron default security group to make sure that we cover this scenario properly. Do you agree?\n\nYes, good point!","accounts_in_message":[],"_revision_number":20},{"id":"a34a11b80fb3724bbdde5ddcfc0947844e44e2c8","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-16 19:31:49.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"8eff061e3f08961eb87af80c20370b96e16b336b","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-16 19:32:05.000000000","message":"Patch Set 21:\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"38aa13a5e9bc8e152581804e34289c52933d1eaa","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-16 20:46:48.000000000","message":"Patch Set 21: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/21 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":21},{"id":"9380c5b70a595ca3f5e9c25a18a15570914a756f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-16 21:11:09.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/733033/21 : SUCCESS in 1h 38m 38s","accounts_in_message":[],"_revision_number":21},{"id":"b9a1e3d60b97bd9d553cf7b0dfa6fe038a7bcfa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 23:04:37.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a260eb47a5384ce6a1f459d006fdfbb0 : SUCCESS in 52m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/572c0bd54719480892290198334ecaf5 : FAILURE in 33m 45s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/44d0c6a61f314b2f9b4bc89f89ff2765 : FAILURE in 1h 38m 58s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/52cb31125fad43af81fdc19590afc43f : SUCCESS in 1h 14m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f7eb0857f4854f5683a9a4c8be7c6c4d : SUCCESS in 1h 11m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0d824e576a874ff7ab52bc6f8a7c2730 : SUCCESS in 1h 16m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92a5c5fd51a0488783b466e98782f6d3 : SUCCESS in 38m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea127f3a37f640ae89b2e62a64af6865 : SUCCESS in 35m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82763f24030d46ee99648e4d81f2a234 : SUCCESS in 16m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce340ac4aded48dfb0a71370fc551c5f : SUCCESS in 25m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cd5893a88b640768c54a6b338d26619 : SUCCESS in 31m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/466edca940044480b406276964d149fb : SUCCESS in 15m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/124e464983bb4c6ab71cf85936661100 : SUCCESS in 55m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5179bb62324f43ea81c68a3955f40a4c : SUCCESS in 50m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c462093d95a948c99318bbb82c5d4610 : SUCCESS in 58m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/41e536870504443d889ab8704f6cc0e4 : SUCCESS in 1h 08m 10s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e2ed670fcd87440d835bf4e197f90567 : SUCCESS in 1h 21m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0eb6aefc2d16400985e28ce13ff5ba04 : FAILURE in 20m 24s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2759306d8cbb4c87ba96b7d67d2c22ee : FAILURE in 30m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/01831aa588ad4c89bd5d810d61f45f46 : SUCCESS in 1h 05m 49s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e96befa700864ecb97d7f26d0ded7e53 : SUCCESS in 1h 24m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9db95d0877c44f32875f8eab20b28e0c : SUCCESS in 1h 11m 42s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d33d61db039b4698b5c7b8e702b80456 : SUCCESS in 1h 40m 20s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f27e2180c1f5410ebbf7cab815f2c14a : FAILURE in 1h 27m 44s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/24ee163b2b4f47a0ac233c9322ff9f63 : SUCCESS in 1h 08m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/835a5967d61b4ad4aa1b29688a27192a : SUCCESS in 55m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2349d82a89ad41348c1bc0266fc76025 : SUCCESS in 2h 00m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39d946f4b6fd485aa7e704d4fef49e67 : SUCCESS in 59m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cd2374cd27c84d2e9e76153a722e47ff : FAILURE in 24m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6cc9592f72774dedbc0b6c19a16e14e1 : FAILURE in 20m 27s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3a138cef74574a279aee7c5f427d8790 : POST_FAILURE in 27m 05s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/11a0e4dae7dd4e46b8cb91664a92e785 : SUCCESS in 2h 08m 47s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c4c5cbe0cb3849d09eaae8df19969ff8 : SUCCESS in 1h 52m 30s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b24ea10bd70f451ca5262f2365a7c2c8 : SUCCESS in 2h 41m 41s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a333a292e70c4c91bf7337fb705014b5 : SUCCESS in 1h 52m 55s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/a7a9cbf033f9494ea53d1fd350252b01 : SUCCESS in 2h 01m 15s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"426e6abb559e42357760e585c095ff4ea588d33f","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-06-17 15:11:36.000000000","message":"Patch Set 21: Code-Review+1\n\nLGTM! Haven\u0027t tested though... thanks for sending the patch also to tempest. Would it be possible to run it with a depends-on maybe?","accounts_in_message":[],"_revision_number":21},{"id":"6dd9770fb735a7a44a9dbea18c7d42ca955b35bf","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-17 18:28:25.000000000","message":"Patch Set 21:\n\n\u003e LGTM! Haven\u0027t tested though... thanks for sending the patch also to\n \u003e tempest. Would it be possible to run it with a depends-on maybe?\n\nYes, I added a depends-on clause on https://review.opendev.org/#/c/736076/ pointing at this Gerrit. We actually do not want that tempest test merged before this change.","accounts_in_message":[],"_revision_number":21},{"id":"4d2e81d4f525b54a965865dd23f757c1758d9d1b","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-17 18:31:41.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"6761759d0fdcae31e472387edc51e21f21619fa1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-17 20:12:00.000000000","message":"Patch Set 22:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/733033/22 : FAILURE in 1h 39m 28s","accounts_in_message":[],"_revision_number":22},{"id":"75a57aae267d9309892063c7fc4d3b208d2a23be","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-17 21:47:04.000000000","message":"Patch Set 22: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/22 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":22},{"id":"5f0693988e52cc13226b08b1f6f0a9875300d972","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 02:53:26.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7d1b04967d46412185006458c5796f1f : SUCCESS in 58m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/608ace3349db4a758a958970d06b27e8 : SUCCESS in 35m 12s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/7a50109f8e584a91a00f1b0880177e3d : FAILURE in 1h 57m 37s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a1ea04d52e7c4e35a4a26ca02494b16b : SUCCESS in 1h 28m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/237d644ba39646a498707f76131d8ac7 : SUCCESS in 1h 17m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a77bf5e502554f01b830e002da055f95 : SUCCESS in 1h 04m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a04153df3cdc4285a46c9952305d5b47 : SUCCESS in 32m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da6364e655264931bb78c7381341493a : SUCCESS in 36m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0478b517fd914f1c9f128369e502e604 : SUCCESS in 17m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/024cab9167144e19acc38e8477985d44 : SUCCESS in 36m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0eb6fc0dddd04a19911a27cfdadeddd5 : SUCCESS in 35m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88194328faae4b53abc5e4fb1ba058ad : SUCCESS in 12m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cbfaa4081560441d9fc9f76791b132ea : FAILURE in 19m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9cd146fae57c4e0eb1ab3c2b01e9ded9 : FAILURE in 21m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/17283cd13b2c4ff08ce681cb5d961dbe : SUCCESS in 1h 17m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/23c91019331a4905b2b26a5efe79703d : SUCCESS in 1h 13m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a8e87f247eb4425a88fee58fd82881ac : SUCCESS in 2h 07m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/869d5c8ce97944b0a0795dc7bdcafc21 : SUCCESS in 1h 13m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5c7afb4fec4248cea10005128a012cac : SUCCESS in 1h 05m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1fa7c41888874c90b787c36176c3e961 : SUCCESS in 1h 09m 39s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/9613bee197b447a4891396665ea44313 : SUCCESS in 1h 10m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/422bb8bc35e24ade84142b9fa2445f8c : SUCCESS in 1h 19m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d2c0d33c23464dc5877e11cdff3a08b3 : SUCCESS in 1h 45m 36s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7359dec494684c18b292e660ea64a03c : FAILURE in 1h 20m 33s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3236b4be3d7247158e3a25e15361f6cc : SUCCESS in 1h 17m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3e712e804cce42c498efdbfc9f75231b : SUCCESS in 1h 01m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2a08741ae82d4b4193ba8dfbde2894a5 : SUCCESS in 2h 10m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed08d44ae34b4b928b8af9a09ddc413c : SUCCESS in 53m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9a6524c37b86412ea40f5268401c44a2 : SUCCESS in 2h 03m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5e4171f2dc1e42bda374e2ea39b2959e : SUCCESS in 1h 40m 26s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8b1be662e1f54d7b9f68ed2c1712cb1c : SUCCESS in 1h 34m 11s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/641d6e5ed5324b4cad2225e6946f92e5 : SUCCESS in 1h 40m 12s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/173758bd01a04b1284084e3aaa4082c6 : SUCCESS in 1h 49m 26s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/05d20c24683e44bdb0f8f8e53cfa5a92 : SUCCESS in 2h 23m 06s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f62df82a88ef448c98aa698f77fdd9bb : SUCCESS in 1h 16m 19s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/02d905fa338b47018b1b6cb378b8f48e : FAILURE in 3h 00m 23s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"ca0600ad6fb51d9ec0a8ee10ebeb5f2456393fbd","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-18 16:29:21.000000000","message":"Patch Set 23: Patch Set 22 was rebased","accounts_in_message":[],"_revision_number":23},{"id":"3a0da05806efcdf94bb742997fee268225eabf9e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-18 18:10:37.000000000","message":"Patch Set 23: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/23 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":23},{"id":"3efbc80723e96dd44d6a8c5712a5cc87d2c386c7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-18 18:25:34.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/733033/23 : SUCCESS in 1h 55m 29s","accounts_in_message":[],"_revision_number":23},{"id":"36a252716b4b1c095b5bec963f45e3bcb862d442","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 20:52:30.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/70b14fdb8af44a408d21ea447863e85a : SUCCESS in 56m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4c7fa50ac8384cc4b0a8d3a2d12dc242 : SUCCESS in 37m 10s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9294ce3de94c462bbe80abea9db4c0be : SUCCESS in 1h 08m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/065f5047b2ef4863a1e6b676e7bc00b0 : SUCCESS in 1h 06m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/07ead96a04b040bbb1f0090737810d74 : SUCCESS in 1h 18m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e33be5558c054b32ae499c2c701e93a2 : SUCCESS in 34m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97e37ddc76b34ad285b292e82a136ee5 : SUCCESS in 37m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a0a286f2a9442a9b0def5cb89e1c4f7 : SUCCESS in 11m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b624f1928b94e629f773899ab584100 : SUCCESS in 31m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24ea406bfa4747c5aa4e6b762b959d1e : SUCCESS in 28m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f19cd2794a784b41ad61beb2cb5de24b : SUCCESS in 9m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/35b5143821d148c4aa6f0e8613284850 : SUCCESS in 58m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a5ddca0fd0cb405dabe6dd78df038d26 : RETRY_LIMIT in 9m 56s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7b6902d71fff4a9cb1ef34db20c30015 : SUCCESS in 1h 20m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f9728f4aa3ee4b8db7a43774723364c0 : SUCCESS in 1h 19m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ced1f18716f1475ba1643356dd9eb6c0 : SUCCESS in 1h 48m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/41a05b09fa704806ba4473af137aea1e : SUCCESS in 1h 31m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/25ad995dce8f4a4db656135fb5b16dd2 : SUCCESS in 1h 31m 06s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0aa17da5c5c247478f76ab105ae88b6f : SUCCESS in 1h 08m 55s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a40d5d5caa104627916b81c4bf42e8ca : SUCCESS in 1h 07m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/113cf3b02c7a4bea9cbfa74dbf30635a : SUCCESS in 1h 07m 14s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8d65643a2aac452896fed11c84ee2efc : SUCCESS in 1h 43m 25s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9dffd8cb9bb44d5eadf2509e0fbe2de8 : FAILURE in 1h 32m 11s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/93fe0a668f854e20b967856507640150 : SUCCESS in 1h 03m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6fdc8785bb354f9c84a98d7343362874 : SUCCESS in 49m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a97a3eb386b64439b0530b8da501dc5a : SUCCESS in 2h 13m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b1f15f697af419cb054a3f4e3379ffc : SUCCESS in 59m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/320880199e5e48e4bb2b5bf98c8b02ed : SUCCESS in 2h 00m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a1424fe18bb041259372b00a92019682 : SUCCESS in 1h 17m 13s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/011d4a9ccfcf44dabcd5e29680f97195 : FAILURE in 31m 48s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/661c568d29744f1aa1c75744131381b7 : SUCCESS in 1h 53m 33s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b3261ee5b3814e5c902a2d74584d4384 : SUCCESS in 1h 42m 34s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e636b206479d4e119340dd39697d2d31 : TIMED_OUT in 3h 20m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9b7547eed1e2467797182d062662e30e : SUCCESS in 1h 50m 10s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/729c600dec83401e9bc7ea4dca4622dc : SUCCESS in 2h 13m 42s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"7451d2741be76d94d207a91c0183c9cce425949c","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-18 22:35:19.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"8773c3adb516c90c912cffeeb0b94967c97808a3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-19 00:00:18.000000000","message":"Patch Set 24: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/24 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":24},{"id":"3fd8fd83d414d2199aea82fe13bd44025a40deef","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-19 00:20:52.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/733033/24 : SUCCESS in 1h 44m 39s","accounts_in_message":[],"_revision_number":24},{"id":"53a1c5287576b0f3ef0cd09eedd100a705e5576c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 01:15:30.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0cf0a4a0bc5e4df8bf8e0a0e9ef09a27 : SUCCESS in 54m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/37563b63f6e34cf3be9bdab4f618c294 : SUCCESS in 35m 21s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9502dd5b32d545dc8afc56cfd36c7cf7 : SUCCESS in 1h 14m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2ee3d6a87f274c87824f4235104756e7 : SUCCESS in 1h 28m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e345c56afe374a7ea7dd4b20cac7d6b7 : SUCCESS in 1h 08m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e0846b120e947c894c5b135d1ceb648 : SUCCESS in 39m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d3971ebdf974ce1bc0d8d3b3bb38ffe : SUCCESS in 27m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33e7b6c625e949e988f1db7b2d37fec3 : SUCCESS in 12m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/968fd17897894a878bde97ec7136afa8 : SUCCESS in 25m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/876e8fb8cd6f4529a283c3f1b5309beb : SUCCESS in 27m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4fdf6f43a504f4d87aad25d79568fbf : SUCCESS in 10m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/21bc385d679c4519babf39c93fa76d14 : SUCCESS in 56m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f66111f06bde4c688117540848c2de9b : RETRY_LIMIT in 8m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/90d42ea313ce466a9bdf996ac64ff411 : RETRY_LIMIT in 11m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f01564b2930f481e95d648d5baa42429 : SUCCESS in 1h 16m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/229f959b782d4023a85050f6fb2493e6 : SUCCESS in 1h 45m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4bdc55d51b2a48cc9049568f6343a985 : SUCCESS in 1h 06m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6be38ed60aac4297bfb62e854f961bfc : SUCCESS in 1h 09m 43s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/992c816cf4564d53830362f3ece98151 : SUCCESS in 1h 12m 03s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/90080d60a15347048cf6b21cf2140000 : SUCCESS in 1h 17m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/24d079f3f0a3440ca0052b4fb3a24871 : SUCCESS in 1h 05m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e8916ba9220d44c28ea5bcaae9f7cd32 : SUCCESS in 1h 29m 09s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dc865b4961ea4902901e18821f6cf0ff : FAILURE in 1h 17m 56s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b36e826472b34b6c8e3c54f1d83c8a7a : SUCCESS in 1h 01m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d7c27ca7a6cd4c82a5b8f3cf7097ca6c : SUCCESS in 55m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b8700bc0f9844fbd82fc8b820f5faa5e : SUCCESS in 1h 50m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7565a775f58a46ba969a3444c2adb10c : SUCCESS in 55m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5a3d947ae1174287bf91e79cfe7f4368 : SUCCESS in 2h 00m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/452c0491ac514025bf1d2e08547e9e96 : SUCCESS in 2h 01m 22s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b0e099f5346642b1be3aa3fb99770843 : POST_FAILURE in 29m 57s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f7f4a755973b48099434649ae9b05b33 : SUCCESS in 1h 43m 56s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/44f50871ffcf441da0a5c6ec5a8ff2ba : SUCCESS in 1h 51m 28s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/19d2e281c06b412cb5cb3c7fc807c80a : SUCCESS in 2h 34m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/857cea6180b64bca80a69b5be30b50fa : SUCCESS in 2h 24m 08s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/123d4aa576fa4d149cfa5360b5a1e638 : SUCCESS in 2h 00m 15s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"ad2010d8fdc06e0589b3c48217887cf4cd598021","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 01:16:09.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0cf0a4a0bc5e4df8bf8e0a0e9ef09a27 : SUCCESS in 54m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/37563b63f6e34cf3be9bdab4f618c294 : SUCCESS in 35m 21s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9502dd5b32d545dc8afc56cfd36c7cf7 : SUCCESS in 1h 14m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2ee3d6a87f274c87824f4235104756e7 : SUCCESS in 1h 28m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e345c56afe374a7ea7dd4b20cac7d6b7 : SUCCESS in 1h 08m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e0846b120e947c894c5b135d1ceb648 : SUCCESS in 39m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d3971ebdf974ce1bc0d8d3b3bb38ffe : SUCCESS in 27m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33e7b6c625e949e988f1db7b2d37fec3 : SUCCESS in 12m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/968fd17897894a878bde97ec7136afa8 : SUCCESS in 25m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/876e8fb8cd6f4529a283c3f1b5309beb : SUCCESS in 27m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4fdf6f43a504f4d87aad25d79568fbf : SUCCESS in 10m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/21bc385d679c4519babf39c93fa76d14 : SUCCESS in 56m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f66111f06bde4c688117540848c2de9b : RETRY_LIMIT in 8m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/90d42ea313ce466a9bdf996ac64ff411 : RETRY_LIMIT in 11m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f01564b2930f481e95d648d5baa42429 : SUCCESS in 1h 16m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/229f959b782d4023a85050f6fb2493e6 : SUCCESS in 1h 45m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4bdc55d51b2a48cc9049568f6343a985 : SUCCESS in 1h 06m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6be38ed60aac4297bfb62e854f961bfc : SUCCESS in 1h 09m 43s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/992c816cf4564d53830362f3ece98151 : SUCCESS in 1h 12m 03s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/90080d60a15347048cf6b21cf2140000 : SUCCESS in 1h 17m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/24d079f3f0a3440ca0052b4fb3a24871 : SUCCESS in 1h 05m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e8916ba9220d44c28ea5bcaae9f7cd32 : SUCCESS in 1h 29m 09s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dc865b4961ea4902901e18821f6cf0ff : FAILURE in 1h 17m 56s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b36e826472b34b6c8e3c54f1d83c8a7a : SUCCESS in 1h 01m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d7c27ca7a6cd4c82a5b8f3cf7097ca6c : SUCCESS in 55m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b8700bc0f9844fbd82fc8b820f5faa5e : SUCCESS in 1h 50m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7565a775f58a46ba969a3444c2adb10c : SUCCESS in 55m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5a3d947ae1174287bf91e79cfe7f4368 : SUCCESS in 2h 00m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/452c0491ac514025bf1d2e08547e9e96 : SUCCESS in 2h 01m 22s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b0e099f5346642b1be3aa3fb99770843 : POST_FAILURE in 29m 57s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f7f4a755973b48099434649ae9b05b33 : SUCCESS in 1h 43m 56s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/44f50871ffcf441da0a5c6ec5a8ff2ba : SUCCESS in 1h 51m 28s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/19d2e281c06b412cb5cb3c7fc807c80a : SUCCESS in 2h 34m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/857cea6180b64bca80a69b5be30b50fa : SUCCESS in 2h 24m 08s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/123d4aa576fa4d149cfa5360b5a1e638 : SUCCESS in 2h 00m 15s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"098bee1d56b830e6c566e85f56b6499fce439c76","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-19 08:09:45.000000000","message":"Patch Set 25: Patch Set 24 was rebased","accounts_in_message":[],"_revision_number":25},{"id":"a21aecf2faf5873ff78e84fb228b7491246e5431","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-19 08:25:58.000000000","message":"Patch Set 25:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/733033/25 : FAILURE in 15m 21s","accounts_in_message":[],"_revision_number":25},{"id":"cbde0949c17c7382e8788e5f8cc4201106a19a31","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-19 09:31:18.000000000","message":"Patch Set 25: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/25 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"a01d790478c07ad37d1e3024a3d0335c2e1accf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 11:17:37.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/87fa8085c8b641f7971eaafc57eb02bd : SUCCESS in 53m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6c5acccdebb241febda920f2d72e7f9e : SUCCESS in 38m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2a482ca86ccd437798e1546e30d02716 : SUCCESS in 1h 08m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/946877ebf89243d9a2690b69f1dde4bc : SUCCESS in 1h 11m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/babb54ea9c3642908ac61d81a45508d6 : SUCCESS in 1h 07m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcb7a21b6c0a4729b71588015741ddfc : SUCCESS in 39m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccefbda4a6da41829be5fda1eaa9437d : SUCCESS in 25m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83a9e870d980492f913596c4a712173d : SUCCESS in 11m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc7dcda294004bdfae1c6abc9e59088f : SUCCESS in 20m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b64820eeb9e143b38e197c6502ebddbd : SUCCESS in 30m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3740b37b3a4415daaca5a345d943978 : SUCCESS in 13m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fe26be8701b9450ba0e5451e80af9ec0 : RETRY_LIMIT in 10m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5fb86dddef51459c88fdcc6c871cfd83 : RETRY_LIMIT in 11m 54s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/cc2aa850df904576acdf5bd50b6d7319 : RETRY_LIMIT in 10m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0a2af75e6454a80963c221b6a4be47a : RETRY_LIMIT in 13m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/481d1861269d4e709b6897afc840990f : SUCCESS in 1h 49m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3e506f518774af48a81affe8c94bc9b : SUCCESS in 1h 13m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/485954890671485aaf39037312d4cdc8 : SUCCESS in 1h 17m 22s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/85ae3b3cca8f4229b327017702093832 : SUCCESS in 1h 39m 51s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c71a86017ce1404697be0c76a5bdf735 : SUCCESS in 1h 20m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2d1e280175694beabacbf2989b9fcc4e : SUCCESS in 1h 02m 32s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0568012c58b8442daad424a38cf31893 : FAILURE in 1h 19m 29s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d90e2a3d6bac441290e6af1922fedaee : FAILURE in 1h 17m 36s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7ac90611904f44f6add41629afb7ac64 : SUCCESS in 1h 07m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/adecf4e915cd43f08e19031e646eb9c1 : SUCCESS in 41m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5fc497aefa764da9ad3ac6bbf53ed554 : SUCCESS in 2h 07m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cb5ec7b52a74624a49635bd52fa442b : SUCCESS in 1h 01m 35s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6f48c31e671249b1bb14df2645c86e9f : SUCCESS in 1h 36m 37s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/59830c556c6544629f3dbeedcad239d6 : SUCCESS in 1h 53m 00s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5719e03dfe1547d489b47a8d51504f84 : POST_FAILURE in 19m 23s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/98b8f932e51244ed81bbadc7a266182b : SUCCESS in 1h 34m 57s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/64547cd0928f4062b91d0b981ab528d2 : SUCCESS in 2h 03m 13s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/dbaf6a3a389f48cb87ea09b0a20b4274 : SUCCESS in 2h 29m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1091a75871dc43d9b100acdb6d316e3c : SUCCESS in 1h 42m 55s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/ab616c7e7ba442b29dcac813a80eb616 : SUCCESS in 2h 17m 51s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"758b74c502d0c687f92c5a7e89c4f26443478e9a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-19 15:39:20.000000000","message":"Patch Set 26: Patch Set 25 was rebased","accounts_in_message":[],"_revision_number":26},{"id":"f304f3bdc7afe07f3dbd9b91ea6003c2b8a4acaf","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-19 15:40:10.000000000","message":"Patch Set 26: Workflow-1","accounts_in_message":[],"_revision_number":26},{"id":"8b4f46891498ccb40e883eb632c4eb8eb79adc2d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-19 17:55:23.000000000","message":"Patch Set 26: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/26 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":26},{"id":"db77575386ad9a06e44366da15e0524a8ff171e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 21:15:29.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b558f2ff3be945e8b79521f1d8bf7407 : SUCCESS in 53m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f02add74ffa04f16a7e446b9e22ab90e : SUCCESS in 36m 17s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/253a688a96b34966ba347fc8b3e60ffe : SUCCESS in 50m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5db33c90a25c48f3bc06294ee47b3734 : SUCCESS in 50m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/007a6c83d982441592c641edab1864a0 : SUCCESS in 41m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c350d254dc0148beb786f5cf29401ff4 : SUCCESS in 36m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71a3f161b4e04fbb9df65bd9f9783cfc : SUCCESS in 27m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf699efc79f2459787d083ecfcbeb715 : SUCCESS in 10m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc22b22b1df44833b849645f566eb9fe : SUCCESS in 29m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd0a26403bee410aa2bb3f761dbb16bf : SUCCESS in 24m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2910728fb88341b29b68df3e67d50814 : SUCCESS in 12m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/aa4e77b3c8c84a35a580c6ee6f6fa1c0 : RETRY_LIMIT in 10m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1dcc4d351bd141a7b737eb796718965f : RETRY_LIMIT in 16m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fb418e030ffc42d2aca93f3fe3c8003c : RETRY_LIMIT in 8m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f1154919d3014257b893b2d89c5480c5 : RETRY_LIMIT in 9m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/232aa143c16b4befa70613bd6b18d971 : SUCCESS in 1h 52m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b38d020bd84547869a623379cb65932a : SUCCESS in 1h 11m 24s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/117b178aad6d4673b7a2040e46b46668 : SUCCESS in 52m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c43dc2d65a684589a6d5f5f7f83ab10c : SUCCESS in 1h 20m 47s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/0963859af901446494cc433f792a5487 : SUCCESS in 1h 26m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e87918907c654947bb67d467aa28123c : SUCCESS in 1h 20m 03s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/518ad85237aa4473a4bcb12a289d6403 : FAILURE in 1h 36m 29s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b68eb0a930d542aaab853f7388c8da80 : FAILURE in 1h 16m 20s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9d9eb22b7e894a86ba8d3038c823f1e2 : SUCCESS in 1h 10m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/29fba77dd36e4f4d9774033114527bab : SUCCESS in 1h 02m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e4b90b343aed4ed9921435bd7967f10b : SUCCESS in 2h 12m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d31262ee3d347a39446c44ce74f6b0e : SUCCESS in 52m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/525921327dcd4b80add3ebec769d38bc : FAILURE in 17m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3ee7e7b2823e4712aeda97486638560d : FAILURE in 18m 59s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/65f7801101d44ec387531fc0f6ea229e : POST_FAILURE in 34m 21s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8ae38d51ff074809aa39850967185447 : SUCCESS in 2h 07m 21s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ff15f22e4fe14083b72fe6d10d1750d7 : SUCCESS in 2h 09m 09s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d7a596c93ea049f9809d5fd1cdcb79a9 : SUCCESS in 2h 45m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/884905beb7014f9eacc05e3c73ff4575 : SUCCESS in 1h 39m 04s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/fc4ab0ee3dee4ecdb2dbc3cd8ca6ae5b : SUCCESS in 2h 04m 37s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"7ced47de5930f3f897fb5db5dbfae41879eebe7e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-20 06:09:41.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/733033/26 : SUCCESS in 1h 50m 58s","accounts_in_message":[],"_revision_number":26},{"id":"72602b2ec3f29fcc4df8c0067da3867d0107a5a5","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-20 19:42:55.000000000","message":"Patch Set 27: Patch Set 26 was rebased","accounts_in_message":[],"_revision_number":27},{"id":"0c93c0980127431a738f85f7c167a36d5c8b5efc","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-20 19:43:03.000000000","message":"Patch Set 27: Workflow-1","accounts_in_message":[],"_revision_number":27},{"id":"b4ed1454732a0fe43f7f6c59fff8b0eb69b0f87f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-20 21:23:54.000000000","message":"Patch Set 27:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/733033/27 : FAILURE in 1h 40m 24s","accounts_in_message":[],"_revision_number":27},{"id":"be0c69c926970c65dc2db33900c89cecb2df3ef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 22:12:57.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7c70afa9555a48f9b83f4d5b9f993d92 : SUCCESS in 1h 05m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9f401d058fda4680bb89caff81b2dfc9 : SUCCESS in 41m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8baa581b659a49489809bee252a3e0ac : SUCCESS in 1h 13m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5087d85eec914669ab4b768121939a40 : SUCCESS in 1h 19m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c60c97c194f144c2b492d2f2898794f9 : SUCCESS in 1h 15m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4caa2746eea0411e870858421ecd2c18 : SUCCESS in 34m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/100420da9b6147398f4fcb9ea8b7a4bb : SUCCESS in 30m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c830917e06eb4428af073d5f96148a71 : SUCCESS in 11m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/284fc771c2c84509b78c42f78821e963 : SUCCESS in 25m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76071ef54094419fac26a8ebacfa72a9 : SUCCESS in 26m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e29c66a23c844a0085af59cddd289582 : SUCCESS in 9m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e6c0dc1e370a40bf9cc75e009d12a1d4 : SUCCESS in 56m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/43968e08ba734ffe99e14034ab6c9fbe : SUCCESS in 49m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2d16be738e5242d08f4400aaf5b6e95f : SUCCESS in 1h 09m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f8a01146fbe94bfd9e5a4b1069b9e4a7 : SUCCESS in 1h 19m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/487545698f04436e97c10b8088dd61bf : SUCCESS in 1h 56m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/57c749d7be6d4013a7fc4e85c6769d8e : SUCCESS in 1h 13m 40s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bb05b68223614bc28022fe99f3cfa3d5 : SUCCESS in 1h 20m 19s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b7a5d8dfa0e54520b68f0c9d6d86a3f9 : SUCCESS in 1h 21m 47s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa51e1c7828e498683bc7a3cbe7fc748 : SUCCESS in 1h 23m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0eaf300d6a7a4fa69388b0be510274b8 : SUCCESS in 1h 20m 11s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2c666887ceba4ba99702224b1ff2dc89 : SUCCESS in 1h 39m 50s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a11e5aca1291453a98c9faefc0b3b97a : FAILURE in 1h 17m 36s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/48ae1476bf1442eda9335320c040796e : SUCCESS in 1h 06m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/eb9024fdc5ef437e802f44339b14e8dc : SUCCESS in 58m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b793060ac9f9419faf4e0a02bc982a12 : SUCCESS in 2h 01m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de73fa76f63d4867b9004a9c91d9779e : SUCCESS in 1h 03m 51s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fe8351e8a46748be969d182763f64a46 : SUCCESS in 1h 46m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1b75cc0e98104b358753c0230853ef6e : SUCCESS in 2h 02m 40s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b1215382f50f44a7ad9cfb1fe454cb6b : POST_FAILURE in 29m 00s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b3685a6b9eb4407088ef11de57d49fd6 : SUCCESS in 1h 45m 21s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4e5a301711eb4fff893ccf79ece15568 : SUCCESS in 1h 48m 25s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5ad5746acbf2450cb83848242fef5289 : SUCCESS in 2h 24m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7444254f0fbd455cb58958cee6146239 : SUCCESS in 1h 42m 55s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/9b9cf1f89df7468b9d1e69fbd5fceb9b : SUCCESS in 1h 59m 06s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"6904661faf36d3948efa0ffd3c6583fa907677e1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-20 22:15:57.000000000","message":"Patch Set 27: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/33/733033/27 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":27},{"id":"cbdb546fab28a3d545cdc28d111ff9b1664f9988","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-25 15:23:16.000000000","message":"Abandoned\n\nWe are no longer convinced that\nthe code complexity introduced by the change justifies its benefits. Reasons for that are:\n\n1) More places for a bug to live;\n2) Default egress rules work just fine to give VMs access to dhcp and metadata; so this is\n   an unnecessary burden in memory and configuration for most use cases;\n3) If the operator wants total control over egress rules, implicitly adding SGR behind the\n   scenes is exactly what we should _not_ do.\n\nLastly, just a note that this is a slight departure from the default behavior that one gets\nwhen using ml2/ovs. Even though mls/ovs adds no \u0027implicit\u0027 rules for allowing vms to reach\nmetadata, it does add the dhcpv4 rules. When using ml2/ovn, operators will need to do add the\nrules on his/her own. It would look along these lines:\n\n   # Allow VM to contact dhcp server (ipv4)\n   openstack security group rule create --egress --ethertype IPv4 --protocol udp --dst-port 67 ${SEC_GROUP_ID}\n   # Allow VM to contact metadata server (ipv4)\n   openstack security group rule create --egress --ethertype IPv4 --protocol tcp --remote-ip 169.254.169.254 ${SEC_GROUP_ID}\n\n   # Allow VM to contact dhcp server (ipv6, non-slaac). Be aware that the remote-ip may vary depending on your use case!\n   openstack security group rule create --egress --ethertype IPv6 --protocol udp --dst-port 547 --remote-ip ff02::1:2 ${SEC_GROUP_ID}\n   # Allow VM to contact metadata server (ipv6)\n   openstack security group rule create --egress --ethertype IPv6 --protocol tcp --remote-ip fe80::a9fe:a9fe ${SEC_GROUP_ID}","accounts_in_message":[],"_revision_number":27}],"current_revision_number":27,"current_revision":"f266da26d23cd3f0b8b7b43367640c215e4f3297","revisions":{"a9f8b18543247268de149e1a2b6efec7ebe7bcb1":{"kind":"REWORK","_number":1,"created":"2020-06-03 01:30:11.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/1"}}},"commit":{"parents":[{"commit":"31280695a26cdcf211cb964ac5f401296398a19f","subject":"Merge \"[OVN] Blacklist test_port_security_macspoofing_port test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/31280695a26cdcf211cb964ac5f401296398a19f"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:55:16.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes a rule that allows packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n  to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow\n  to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow\n\nA caveat here is that Openstack operators may chosse remove these rules. As VMs\nmay use IPv4 DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder (aka ovn-controller) needs to be ensured via an\nexplicit rule that is not removable by Openstack operators. That is the purpose\nof the DHCP rules of the OVN\u0027s subnet port group. This particular port group is\nautomatically created/removed as subnets are added and they contain the following\nACL as part of creation:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\nThis change re-introduces (and fixes) the subnet port group, which was removed in\nbug #1790900.\n\nAll in all, this change reverts https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . I also addresses issues that were missing\nto ensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9f8b18543247268de149e1a2b6efec7ebe7bcb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9f8b18543247268de149e1a2b6efec7ebe7bcb1"}]},"branch":"refs/heads/master"},"e1bc933b4a72fa847704927794ee2966ee8f78b0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-04 02:17:03.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/2"}}},"commit":{"parents":[{"commit":"1a3811448d0a108e8e16bb32810e4f59db226a4b","subject":"Bump oslo.log to version 4.1.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a3811448d0a108e8e16bb32810e4f59db226a4b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-06-04 02:17:03.000000000","tz":0},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes a rule that allows packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n  to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow\n  to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow\n\nA caveat here is that Openstack operators may chosse remove these rules. As VMs\nmay use IPv4 DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder (aka ovn-controller) needs to be ensured via an\nexplicit rule that is not removable by Openstack operators. That is the purpose\nof the DHCP rules of the OVN\u0027s subnet port group. This particular port group is\nautomatically created/removed as subnets are added and they contain the following\nACL as part of creation:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\nThis change re-introduces (and fixes) the subnet port group, which was removed in\nbug #1790900.\n\nAll in all, this change reverts https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . I also addresses issues that were missing\nto ensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1bc933b4a72fa847704927794ee2966ee8f78b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1bc933b4a72fa847704927794ee2966ee8f78b0"}]},"branch":"refs/heads/master"},"90080694c9b0137e6ec86bcff3a1c3503234d51d":{"kind":"REWORK","_number":3,"created":"2020-06-04 02:23:40.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/3"}}},"commit":{"parents":[{"commit":"1a3811448d0a108e8e16bb32810e4f59db226a4b","subject":"Bump oslo.log to version 4.1.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a3811448d0a108e8e16bb32810e4f59db226a4b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-04 02:23:36.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes a rule that allows packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n  to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow\n  to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow\n\nA caveat here is that Openstack operators may chosse remove these rules. As VMs\nmay use IPv4 DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder (aka ovn-controller) needs to be ensured via an\nexplicit rule that is not removable by Openstack operators. That is the purpose\nof the DHCP rules of the OVN\u0027s subnet port group. This particular port group is\nautomatically created/removed as subnets are added and they contain the following\nACL as part of creation:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\nThis change re-introduces (and fixes) the subnet port group, which was removed in\nbug #1790900.\n\nAll in all, this change reverts https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . I also addresses issues that were missing\nto ensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90080694c9b0137e6ec86bcff3a1c3503234d51d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90080694c9b0137e6ec86bcff3a1c3503234d51d"}]},"branch":"refs/heads/master"},"af8e81f44c9210f230e509fc8297fb238717987a":{"kind":"REWORK","_number":4,"created":"2020-06-04 03:20:38.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/4"}}},"commit":{"parents":[{"commit":"1a3811448d0a108e8e16bb32810e4f59db226a4b","subject":"Bump oslo.log to version 4.1.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a3811448d0a108e8e16bb32810e4f59db226a4b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-04 03:20:32.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes a rule that allows packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n  to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow\n  to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow\n\nA caveat here is that Openstack operators may chosse remove these rules. As VMs\nmay use IPv4 DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder (aka ovn-controller) needs to be ensured via an\nexplicit rule that is not removable by Openstack operators. That is the purpose\nof the DHCP rules of the OVN\u0027s subnet port group. This particular port group is\nautomatically created/removed as subnets are added and they contain the following\nACL as part of creation:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\nThis change re-introduces (and fixes) the subnet port group, which was removed in\nbug #1790900.\n\nAll in all, this change reverts https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . I also addresses issues that were missing\nto ensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af8e81f44c9210f230e509fc8297fb238717987a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af8e81f44c9210f230e509fc8297fb238717987a"}]},"branch":"refs/heads/master"},"b4f59b08a182e1f47e56c527126d7593353c9d78":{"kind":"REWORK","_number":5,"created":"2020-06-04 09:44:43.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/5"}}},"commit":{"parents":[{"commit":"1a3811448d0a108e8e16bb32810e4f59db226a4b","subject":"Bump oslo.log to version 4.1.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a3811448d0a108e8e16bb32810e4f59db226a4b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-04 09:40:45.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes a rule that allows packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n  to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow\n  to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow\n\nA caveat here is that Openstack operators may chosse remove these rules. As VMs\nmay use IPv4 DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder (aka ovn-controller) needs to be ensured via an\nexplicit rule that is not removable by Openstack operators. That is the purpose\nof the DHCP rules of the OVN\u0027s subnet port group. This particular port group is\nautomatically created/removed as subnets are added and they contain the following\nACL as part of creation:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\nThis change re-introduces (and fixes) the subnet port group, which was removed in\nbug #1790900.\n\nAll in all, this change reverts https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . I also addresses issues that were missing\nto ensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b4f59b08a182e1f47e56c527126d7593353c9d78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b4f59b08a182e1f47e56c527126d7593353c9d78"}]},"branch":"refs/heads/master"},"9c62b49da5330b96665635ab6f0b23e45300d4e5":{"kind":"REWORK","_number":6,"created":"2020-06-04 10:43:51.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/6"}}},"commit":{"parents":[{"commit":"1a3811448d0a108e8e16bb32810e4f59db226a4b","subject":"Bump oslo.log to version 4.1.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a3811448d0a108e8e16bb32810e4f59db226a4b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-04 10:43:37.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes a rule that allows packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n  to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow\n  to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow\n\nA caveat here is that Openstack operators may chosse remove these rules. As VMs\nmay use IPv4 DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder (aka ovn-controller) needs to be ensured via an\nexplicit rule that is not removable by Openstack operators. That is the purpose\nof the DHCP rules of the OVN\u0027s subnet port group. This particular port group is\nautomatically created/removed as subnets are added and they contain the following\nACL as part of creation:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\nThis change re-introduces (and fixes) the subnet port group, which was removed in\nbug #1790900.\n\nAll in all, this change reverts https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . I also addresses issues that were missing\nto ensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c62b49da5330b96665635ab6f0b23e45300d4e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c62b49da5330b96665635ab6f0b23e45300d4e5"}]},"branch":"refs/heads/master"},"42ca5dd6d00bf8b34eb63ac760bf68002e58f238":{"kind":"REWORK","_number":7,"created":"2020-06-04 13:19:24.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/7"}}},"commit":{"parents":[{"commit":"1a3811448d0a108e8e16bb32810e4f59db226a4b","subject":"Bump oslo.log to version 4.1.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a3811448d0a108e8e16bb32810e4f59db226a4b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-04 13:15:03.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes a rule that allows packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n  to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow\n  to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow\n\nA caveat here is that Openstack operators may chosse remove these rules. As VMs\nmay use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder (aka ovn-controller) needs to be ensured via an\nexplicit rule that is not removable by Openstack operators. That is the purpose\nof the DHCP rules of the OVN\u0027s subnet port group. This particular port group is\nautomatically created/removed as subnets are added and they contain the following\nACL as part of creation:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow\n\nThis change re-introduces (and fixes) the subnet port group, which was removed in\nbug #1790900.\n\nAll in all, this change reverts https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . I also addresses issues that were missing\nto ensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/42ca5dd6d00bf8b34eb63ac760bf68002e58f238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/42ca5dd6d00bf8b34eb63ac760bf68002e58f238"}]},"branch":"refs/heads/master"},"20ba23713c1fed8b8509ad69db1ad6a010ffd520":{"kind":"REWORK","_number":8,"created":"2020-06-05 03:16:12.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/8"}}},"commit":{"parents":[{"commit":"1a3811448d0a108e8e16bb32810e4f59db226a4b","subject":"Bump oslo.log to version 4.1.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a3811448d0a108e8e16bb32810e4f59db226a4b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-05 03:13:01.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes a rule that allows packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-r\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-r\n\nA caveat here is that Openstack operators may want to remove these rules. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not removable by Openstack operators. That is the purpose of the DHCP rules\nof the OVN\u0027s subnet port group. This particular port group is automatically\ncreated/removed as subnets are added and they contain the following ACL as\npart of creation:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow\n\nThis change re-introduces (and fixes) the subnet port group, which was removed in\nbug #1790900.\n\nAll in all, this change reverts https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . I also addresses issues that were missing\nto ensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20ba23713c1fed8b8509ad69db1ad6a010ffd520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20ba23713c1fed8b8509ad69db1ad6a010ffd520"}]},"branch":"refs/heads/master"},"3bf5d3672b50448d9731f9dc5c59d58aec60e66e":{"kind":"REWORK","_number":9,"created":"2020-06-05 10:10:12.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/9"}}},"commit":{"parents":[{"commit":"1a3811448d0a108e8e16bb32810e4f59db226a4b","subject":"Bump oslo.log to version 4.1.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a3811448d0a108e8e16bb32810e4f59db226a4b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-05 10:07:25.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes a rule that allows packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-r\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-r\n\nA caveat here is that Openstack operators may want to remove these rules. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not removable by Openstack operators. That is the purpose of the DHCP rules\nof the OVN\u0027s subnet port group. This particular port group is automatically\ncreated/removed as subnets are added and they contain the following ACL as\npart of creation:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow\n\nThis change re-introduces (and fixes) the subnet port group, which was removed in\nbug #1790900.\n\nAll in all, this change reverts https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . I also addresses issues that were missing\nto ensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3bf5d3672b50448d9731f9dc5c59d58aec60e66e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3bf5d3672b50448d9731f9dc5c59d58aec60e66e"}]},"branch":"refs/heads/master"},"19cd35947c3f01ba6c32e2318951c0383cd25993":{"kind":"REWORK","_number":10,"created":"2020-06-05 19:21:29.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/10"}}},"commit":{"parents":[{"commit":"1a3811448d0a108e8e16bb32810e4f59db226a4b","subject":"Bump oslo.log to version 4.1.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a3811448d0a108e8e16bb32810e4f59db226a4b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-05 16:58:12.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes a rule that allows packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-r\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-r\n\nA caveat here is that Openstack operators may want to remove these rules. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not removable by Openstack operators. That is the purpose of the DHCP rules\nof the OVN\u0027s subnet port group. This particular port group is automatically\ncreated/removed as subnets are added and they contain the following ACL as\npart of creation:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow\n\nThis change re-introduces (and fixes) the subnet port group, which was removed in\nbug #1790900.\n\nAll in all, this change reverts https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . I also addresses issues that were missing\nto ensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19cd35947c3f01ba6c32e2318951c0383cd25993"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19cd35947c3f01ba6c32e2318951c0383cd25993"}]},"branch":"refs/heads/master"},"da7fbba967142b9cacb61385cec2481cf8de2448":{"kind":"REWORK","_number":11,"created":"2020-06-06 11:12:46.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/11"}}},"commit":{"parents":[{"commit":"1a3811448d0a108e8e16bb32810e4f59db226a4b","subject":"Bump oslo.log to version 4.1.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a3811448d0a108e8e16bb32810e4f59db226a4b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-06 11:12:38.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes rules that allow packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-related\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-related\n\nOpenstack operators are free to remove these rules, which has a side effect. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not visible to operators. That is the purpose of the DHCP rules of the OVN\u0027s\nsubnet port group. This particular port group is automatically managed as\nsubnets that have DHCP enabled are added. They contain these OVN ACLs, depending\non the IP version:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow\n\nThis change re-introduces (and fixes) subnet port group, which was removed in\nbug #1790900 by reverting https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . It also fixes issues needed to\nensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da7fbba967142b9cacb61385cec2481cf8de2448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da7fbba967142b9cacb61385cec2481cf8de2448"}]},"branch":"refs/heads/master"},"4a017675b1645509cf5c437ef3fc6f2914b6030d":{"kind":"REWORK","_number":12,"created":"2020-06-06 11:52:13.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/12"}}},"commit":{"parents":[{"commit":"1a3811448d0a108e8e16bb32810e4f59db226a4b","subject":"Bump oslo.log to version 4.1.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a3811448d0a108e8e16bb32810e4f59db226a4b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-06 11:49:44.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes rules that allow packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-related\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-related\n\nOpenstack operators are free to remove these rules, which has a side effect. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not visible to operators. That is the purpose of the DHCP rules of the OVN\u0027s\nsubnet port group. This particular port group is automatically managed as\nsubnets that have DHCP enabled are added. They contain these OVN ACLs, depending\non the IP version:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow\n\nThis change re-introduces (and fixes) subnet port group, which was removed in\nbug #1790900 by reverting https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . It also fixes issues needed to\nensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4a017675b1645509cf5c437ef3fc6f2914b6030d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4a017675b1645509cf5c437ef3fc6f2914b6030d"}]},"branch":"refs/heads/master"},"c5e3ed6164aa4f57d67e26ac501af2f73a045f90":{"kind":"REWORK","_number":13,"created":"2020-06-10 10:33:28.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/13"}}},"commit":{"parents":[{"commit":"1a3811448d0a108e8e16bb32810e4f59db226a4b","subject":"Bump oslo.log to version 4.1.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a3811448d0a108e8e16bb32810e4f59db226a4b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-10 10:33:01.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes rules that allow packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-related\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-related\n\nOpenstack operators are free to remove these rules, which has a side effect. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not visible to operators. That is the purpose of the DHCP rules of the OVN\u0027s\nsubnet port group. This particular port group is automatically managed as\nsubnets that have DHCP enabled are added. They contain these OVN ACLs, depending\non the IP version:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow\n\nThis change re-introduces (and fixes) subnet port group, which was removed in\nbug #1790900 by reverting https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . It also fixes issues needed to\nensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5e3ed6164aa4f57d67e26ac501af2f73a045f90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5e3ed6164aa4f57d67e26ac501af2f73a045f90"}]},"branch":"refs/heads/master"},"2adc6df91f1fc8711679c0ec5e31d1890d090560":{"kind":"REWORK","_number":14,"created":"2020-06-10 16:17:37.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/14"}}},"commit":{"parents":[{"commit":"1a3811448d0a108e8e16bb32810e4f59db226a4b","subject":"Bump oslo.log to version 4.1.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a3811448d0a108e8e16bb32810e4f59db226a4b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-10 15:24:42.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes rules that allow packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-related\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-related\n\nOpenstack operators are free to remove these rules, which has a side effect. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not visible to operators. That is the purpose of the DHCP rules of the OVN\u0027s\nsubnet port group. This particular port group is automatically managed as\nsubnets that have DHCP enabled are added. They contain these OVN ACLs, depending\non the IP version:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow\n\nThis change re-introduces (and fixes) subnet port group, which was removed in\nbug #1790900 by reverting https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . It also fixes issues needed to\nensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2adc6df91f1fc8711679c0ec5e31d1890d090560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2adc6df91f1fc8711679c0ec5e31d1890d090560"}]},"branch":"refs/heads/master"},"844a7672d76af33f0c970d74aacbce9b1660c513":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2020-06-10 16:17:48.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/15"}}},"commit":{"parents":[{"commit":"78511d958cfdb46d6f70abd80ca3e8d3fadd84a1","subject":"Merge \"[OVN] List a few missing feature parity gaps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78511d958cfdb46d6f70abd80ca3e8d3fadd84a1"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-06-10 16:17:48.000000000","tz":0},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes rules that allow packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-related\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-related\n\nOpenstack operators are free to remove these rules, which has a side effect. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not visible to operators. That is the purpose of the DHCP rules of the OVN\u0027s\nsubnet port group. This particular port group is automatically managed as\nsubnets that have DHCP enabled are added. They contain these OVN ACLs, depending\non the IP version:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow\n\nThis change re-introduces (and fixes) subnet port group, which was removed in\nbug #1790900 by reverting https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . It also fixes issues needed to\nensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/844a7672d76af33f0c970d74aacbce9b1660c513"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/844a7672d76af33f0c970d74aacbce9b1660c513"}]},"branch":"refs/heads/master"},"340019235e8c773d20e3bcf24dce929775241ba3":{"kind":"REWORK","_number":16,"created":"2020-06-11 02:34:39.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/16"}}},"commit":{"parents":[{"commit":"78511d958cfdb46d6f70abd80ca3e8d3fadd84a1","subject":"Merge \"[OVN] List a few missing feature parity gaps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78511d958cfdb46d6f70abd80ca3e8d3fadd84a1"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-11 02:09:59.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes rules that allow packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-related\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-related\n\nOpenstack operators are free to remove these rules, which has a side effect. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not visible to operators. That is the purpose of the DHCP rules of the OVN\u0027s\nsubnet port group. This particular port group is automatically managed as\nsubnets that have DHCP enabled are added. They contain these OVN ACLs, depending\non the IP version:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow\n\nThis change re-introduces (and fixes) subnet port group, which was removed in\nbug #1790900 by reverting https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . It also fixes issues needed to\nensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/340019235e8c773d20e3bcf24dce929775241ba3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/340019235e8c773d20e3bcf24dce929775241ba3"}]},"branch":"refs/heads/master"},"8f336de53fd786e46e4fcd1fc94ca97f727d59ba":{"kind":"REWORK","_number":17,"created":"2020-06-11 21:53:44.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/17"}}},"commit":{"parents":[{"commit":"78511d958cfdb46d6f70abd80ca3e8d3fadd84a1","subject":"Merge \"[OVN] List a few missing feature parity gaps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78511d958cfdb46d6f70abd80ca3e8d3fadd84a1"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-11 21:25:37.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes rules that allow packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-related\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-related\n\nOpenstack operators are free to remove these rules, which has a side effect. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not visible to operators. That is the purpose of the DHCP rules of the OVN\u0027s\nsubnet port group. This particular port group is automatically managed as\nsubnets that have DHCP enabled are added. They contain these OVN ACLs, depending\non the IP version:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow\n\nThis change re-introduces (and fixes) subnet port group, which was removed in\nbug #1790900 by reverting https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . It also fixes issues needed to\nensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f336de53fd786e46e4fcd1fc94ca97f727d59ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f336de53fd786e46e4fcd1fc94ca97f727d59ba"}]},"branch":"refs/heads/master"},"a22cc465c796d6b191fbd59b5b27d895fd0d8a08":{"kind":"REWORK","_number":18,"created":"2020-06-12 02:18:39.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/18"}}},"commit":{"parents":[{"commit":"4de18104ae88a835544cefbf30c878aa49efc31f","subject":"Merge \"[OVN] Create localnet port for each created segment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4de18104ae88a835544cefbf30c878aa49efc31f"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-12 00:53:33.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes rules that allow packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-related\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-related\n\nOpenstack operators are free to remove these rules, which has a side effect. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not visible to operators. That is the purpose of the DHCP rules of the OVN\u0027s\nsubnet port group. This particular port group is automatically managed as\nsubnets that have DHCP enabled are added. They contain these OVN ACLs, depending\non the IP version:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow\n\nThis change re-introduces (and fixes) subnet port group, which was removed in\nbug #1790900 by reverting https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . It also fixes issues needed to\nensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a22cc465c796d6b191fbd59b5b27d895fd0d8a08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a22cc465c796d6b191fbd59b5b27d895fd0d8a08"}]},"branch":"refs/heads/master"},"48a9d5596ceab68810478df0a6e422dc1fc44776":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2020-06-15 12:40:26.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/19"}}},"commit":{"parents":[{"commit":"05e783f039aab5a188654880e65f3f0b3ee627ab","subject":"[OVN] Load segments plugin in case not loaded in maintanance task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05e783f039aab5a188654880e65f3f0b3ee627ab"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-06-15 12:40:26.000000000","tz":0},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes rules that allow packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-related\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-related\n\nOpenstack operators are free to remove these rules, which has a side effect. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not visible to operators. That is the purpose of the DHCP rules of the OVN\u0027s\nsubnet port group. This particular port group is automatically managed as\nsubnets that have DHCP enabled are added. They contain these OVN ACLs, depending\non the IP version:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow\n\nThis change re-introduces (and fixes) subnet port group, which was removed in\nbug #1790900 by reverting https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . It also fixes issues needed to\nensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/48a9d5596ceab68810478df0a6e422dc1fc44776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/48a9d5596ceab68810478df0a6e422dc1fc44776"}]},"branch":"refs/heads/master"},"888bbae055a48844a0badaf361605fe127a81b22":{"kind":"REWORK","_number":20,"created":"2020-06-15 19:52:43.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/20"}}},"commit":{"parents":[{"commit":"05e783f039aab5a188654880e65f3f0b3ee627ab","subject":"[OVN] Load segments plugin in case not loaded in maintanance task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05e783f039aab5a188654880e65f3f0b3ee627ab"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-15 19:36:33.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes rules that allow packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-related\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-related\n\nOpenstack operators are free to remove these rules, which has a side effect. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not visible to operators. That is the purpose of the DHCP rules of the OVN\u0027s\nsubnet port group. This particular port group is automatically managed as\nsubnets that have DHCP enabled are added. They contain these OVN ACLs, depending\non the IP version:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow\n\nThis change re-introduces (and fixes) subnet port group, which was removed in\nbug #1790900 by reverting https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . It also fixes issues needed to\nensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/888bbae055a48844a0badaf361605fe127a81b22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/888bbae055a48844a0badaf361605fe127a81b22"}]},"branch":"refs/heads/master"},"b74f51a9a14bca93a574d55048962563f5fc606f":{"kind":"REWORK","_number":21,"created":"2020-06-16 19:31:49.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/21"}}},"commit":{"parents":[{"commit":"05e783f039aab5a188654880e65f3f0b3ee627ab","subject":"[OVN] Load segments plugin in case not loaded in maintanance task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05e783f039aab5a188654880e65f3f0b3ee627ab"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-16 18:51:23.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes rules that allow packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-related\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-related\n\nOpenstack operators are free to remove these rules, which has a side effect. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not visible to operators. That is the purpose of the DHCP rules of the OVN\u0027s\nsubnet port group. This particular port group is automatically managed as\nsubnets that have DHCP enabled are added. They contain these OVN ACLs, depending\non the IP version:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow\n\nThis change re-introduces (and fixes) subnet port group, which was removed in\nbug #1790900 by reverting https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . It also fixes issues needed to\nensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b74f51a9a14bca93a574d55048962563f5fc606f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b74f51a9a14bca93a574d55048962563f5fc606f"}]},"branch":"refs/heads/master"},"7eb06596b9e619ce97b69481e5073e7f76bcdbc4":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2020-06-17 18:31:41.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/22"}}},"commit":{"parents":[{"commit":"4ac4873fde688b3e4274c224ce6a8104072256d1","subject":"[OVN] Load segments plugin in case not loaded in maintanance task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ac4873fde688b3e4274c224ce6a8104072256d1"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-06-17 18:31:41.000000000","tz":0},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes rules that allow packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-related\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-related\n\nOpenstack operators are free to remove these rules, which has a side effect. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not visible to operators. That is the purpose of the DHCP rules of the OVN\u0027s\nsubnet port group. This particular port group is automatically managed as\nsubnets that have DHCP enabled are added. They contain these OVN ACLs, depending\non the IP version:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow\n\nThis change re-introduces (and fixes) subnet port group, which was removed in\nbug #1790900 by reverting https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . It also fixes issues needed to\nensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7eb06596b9e619ce97b69481e5073e7f76bcdbc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7eb06596b9e619ce97b69481e5073e7f76bcdbc4"}]},"branch":"refs/heads/master"},"8cdb009a405c3f49de4f5e5b03a5339a0231a837":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2020-06-18 16:29:21.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/23"}}},"commit":{"parents":[{"commit":"f5b6bd8f55a66a856ec00b2333e1968d0c47d5f8","subject":"ovn: Use newer OVS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f5b6bd8f55a66a856ec00b2333e1968d0c47d5f8"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-06-18 16:29:21.000000000","tz":0},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes rules that allow packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-related\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-related\n\nOpenstack operators are free to remove these rules, which has a side effect. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not visible to operators. That is the purpose of the DHCP rules of the OVN\u0027s\nsubnet port group. This particular port group is automatically managed as\nsubnets that have DHCP enabled are added. They contain these OVN ACLs, depending\non the IP version:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow\n\nThis change re-introduces (and fixes) subnet port group, which was removed in\nbug #1790900 by reverting https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . It also fixes issues needed to\nensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cdb009a405c3f49de4f5e5b03a5339a0231a837"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cdb009a405c3f49de4f5e5b03a5339a0231a837"}]},"branch":"refs/heads/master"},"2429c0971e59a470536c3516e5648914f815df29":{"kind":"REWORK","_number":24,"created":"2020-06-18 22:35:19.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/24"}}},"commit":{"parents":[{"commit":"f5b6bd8f55a66a856ec00b2333e1968d0c47d5f8","subject":"ovn: Use newer OVS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f5b6bd8f55a66a856ec00b2333e1968d0c47d5f8"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-18 22:27:47.000000000","tz":-240},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes rules that allow packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-related\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-related\n\nOpenstack operators are free to remove these rules, which has a side effect. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not visible to operators. That is the purpose of the DHCP rules of the OVN\u0027s\nsubnet port group. This particular port group is automatically managed as\nsubnets that have DHCP enabled are added. They contain these OVN ACLs, depending\non the IP version:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow-related\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow-related\n\nThis change re-introduces (and fixes) subnet port group, which was removed in\nbug #1790900 by reverting https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . It also fixes issues needed to\nensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2429c0971e59a470536c3516e5648914f815df29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2429c0971e59a470536c3516e5648914f815df29"}]},"branch":"refs/heads/master"},"558525a710850f1a43047fe10ed5bf53f377aae9":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2020-06-19 08:09:45.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/25"}}},"commit":{"parents":[{"commit":"d381a22b3f6910c371d827a93d63bcdef5e26eef","subject":"[OVN] Load segments plugin in case not loaded in maintanance task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d381a22b3f6910c371d827a93d63bcdef5e26eef"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-06-19 08:09:45.000000000","tz":0},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes rules that allow packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-related\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-related\n\nOpenstack operators are free to remove these rules, which has a side effect. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not visible to operators. That is the purpose of the DHCP rules of the OVN\u0027s\nsubnet port group. This particular port group is automatically managed as\nsubnets that have DHCP enabled are added. They contain these OVN ACLs, depending\non the IP version:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow-related\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow-related\n\nThis change re-introduces (and fixes) subnet port group, which was removed in\nbug #1790900 by reverting https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . It also fixes issues needed to\nensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/558525a710850f1a43047fe10ed5bf53f377aae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/558525a710850f1a43047fe10ed5bf53f377aae9"}]},"branch":"refs/heads/master"},"f171c06b03a7f8a36627667fa7e1aa4877b623e9":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2020-06-19 15:39:20.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/26"}}},"commit":{"parents":[{"commit":"78e16c77428e071ac14fbd3727840cd6348d1e7a","subject":"Merge \"Refactor assertEqual statement that includes \u0027True\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78e16c77428e071ac14fbd3727840cd6348d1e7a"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-06-19 15:39:20.000000000","tz":0},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes rules that allow packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-related\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-related\n\nOpenstack operators are free to remove these rules, which has a side effect. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not visible to operators. That is the purpose of the DHCP rules of the OVN\u0027s\nsubnet port group. This particular port group is automatically managed as\nsubnets that have DHCP enabled are added. They contain these OVN ACLs, depending\non the IP version:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow-related\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow-related\n\nThis change re-introduces (and fixes) subnet port group, which was removed in\nbug #1790900 by reverting https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . It also fixes issues needed to\nensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f171c06b03a7f8a36627667fa7e1aa4877b623e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f171c06b03a7f8a36627667fa7e1aa4877b623e9"}]},"branch":"refs/heads/master"},"f266da26d23cd3f0b8b7b43367640c215e4f3297":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2020-06-20 19:42:55.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/733033/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/733033/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/733033/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/733033/27"}}},"commit":{"parents":[{"commit":"0580d03a2b9f133024deceff4fc5c136a86d01b9","subject":"Merge \"Workaround for TCP checksum issue with ovs-dpdk and  veth pair\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0580d03a2b9f133024deceff4fc5c136a86d01b9"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-02 22:47:19.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-06-20 19:42:55.000000000","tz":0},"subject":"[OVN] Add logical port to its corresponding subnet port group","message":"[OVN] Add logical port to its corresponding subnet port group\n\nThe default security group normally includes rules that allow packets that\noriginate from the same subnet where the VM is connected:\n\n  direction: ingress, ether_type: IPv4, remote_ip_prefix: \"0.0.0.0/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\n  direction: ingress, ether_type: IPv6, remote_ip_prefix: \"::/0\"\n  remote_group_id:   \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as security_group_id\n  security_group_id: \u003cUUID_OF_SEC_GROUP\u003e  \u003c-- same as remote_group_id\n\nOVN translates them into ACLS that look like:\n\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip4 \u0026\u0026 ip4.src \u003d\u003d $\u003cPG_ID\u003e_ip4) allow-related\n to-lport 1002 (outport \u003d\u003d @\u003cPG_ID\u003e \u0026\u0026 ip6 \u0026\u0026 ip6.src \u003d\u003d $\u003cPG_ID\u003e_ip6) allow-related\n\nOpenstack operators are free to remove these rules, which has a side effect. As\nVMs use DHCP to boot, the removal of these security group rules means that\nconnectivity to DHCP responder needs to be ensured via an explicit rule that\nis not visible to operators. That is the purpose of the DHCP rules of the OVN\u0027s\nsubnet port group. This particular port group is automatically managed as\nsubnets that have DHCP enabled are added. They contain these OVN ACLs, depending\non the IP version:\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv4_SUBNET\u003e \u0026\u0026 ip4 \u0026\u0026\n  ip4.dst \u003d\u003d {255.255.255.255, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 68 \u0026\u0026\n  udp.dst \u003d\u003d 67) allow-related\n\n  from-lport  1002 (inport \u003d\u003d @pg_\u003cOVN_PG_ID_FOR_IPv6_SUBNET\u003e \u0026\u0026 ip6 \u0026\u0026\n  ip6.dst \u003d\u003d {ff02::1:2, \u003cSUBNET_PREFIX\u003e} \u0026\u0026 udp \u0026\u0026 udp.src \u003d\u003d 546 \u0026\u0026\n  udp.dst \u003d\u003d 547) allow-related\n\nThis change re-introduces (and fixes) subnet port group, which was removed in\nbug #1790900 by reverting https://review.opendev.org/#/c/681567/ and\nhttps://review.opendev.org/#/c/600365/ . It also fixes issues needed to\nensure that subnet port groups work as expected.\n\nChange-Id: Iaf31a1827a6d1eb8b143207ffc63d5ab933e4c0f\nRelated-Bug: #1790900\nCloses-Bug: #1881316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f266da26d23cd3f0b8b7b43367640c215e4f3297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f266da26d23cd3f0b8b7b43367640c215e4f3297"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
