)]}'
{"id":"openstack%2Fneutron~733082","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I7c7c8a4c371d6f4afdaab51ed50950e2b20db30f","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1864822-stable/queens","hashtags":[],"change_id":"I7c7c8a4c371d6f4afdaab51ed50950e2b20db30f","subject":"Ensure that stale flows are cleaned from phys_bridges","status":"MERGED","created":"2020-06-03 09:47:32.000000000","updated":"2020-07-07 21:33:17.000000000","submitted":"2020-07-07 21:33:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733082-1594157597434-86bf99d6","meta_rev_id":"d8cb28f8e2e8c506e3a4eeecc610926ed520afb1","_number":733082,"virtual_id_number":733082,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-07-07 15:12:50.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-07 21:33:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-07 14:54:41.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-07 14:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2020-07-07 14:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2020-07-07 14:55:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 20:22:26.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-07-07 14:54:41.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-07-07 14:55:42.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-07-07 15:12:50.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-07 21:33:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b7bbc9b2bcabffd597e58be2cf09bf3cc1f8f62","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-03 09:47:32.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"3b86348ef0b4facb40b311b7032eadffe2e983f6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-03 09:55:49.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/733082/1 : FAILURE in 7m 38s","accounts_in_message":[],"_revision_number":1},{"id":"58b8e5b1ed1d8fe7b103b8b08f069d7b8a7293d6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-03 10:58:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5af2795960ad4f1ca73a6c59c3d87c14de03401a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-03 11:09:04.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/733082/2 : FAILURE in 9m 27s","accounts_in_message":[],"_revision_number":2},{"id":"0f624945a979ad0c7b7fb968f0e6e15b051d98bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 12:52:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/13d0381b6eba4819b9c175cac111a53c : SUCCESS in 57m 17s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/f035ce8aa51b4a388474e0f7e93cc6e7 : SUCCESS in 48m 15s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/614b4ace89d74ffbb65433621c08e36d : FAILURE in 1h 50m 17s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/b20907e243354ee89360fc96de0325a9 : SUCCESS in 54m 45s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/1df8095dd38b4ebe9782249ca7217134 : SUCCESS in 53m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56d370c5f5e549b287e3668f5e5fba9d : SUCCESS in 10m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef305fd6bf144e1bbd98a522aaa357b1 : SUCCESS in 20m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/05d8c28c413a4f51aa75ce9d228d2ad2 : SUCCESS in 24m 52s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9c3ad4c23c3749718a21446854843d5f : SUCCESS in 5m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1ac3b68dfb4743238a1c0906eb5f87ae : SUCCESS in 1h 11m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/911d950e70ad46c5b3334d0ddcbb2f31 : SUCCESS in 1h 00m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fc921bca88c4ba6bb079c6557f31bbe : SUCCESS in 1h 16m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f16b19a4171e4e53a7bf3c3449143af0 : FAILURE in 43m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5c6c614dcf4842ad857ede5aa5c0f3d8 : FAILURE in 38m 21s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/2f211dc31687418dbd82269aacbd3e83 : SUCCESS in 1h 33m 19s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b5a386c84f53474b8f25fbedb598938c : SUCCESS in 1h 39m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/cbaf3f21daa54bcd9b44416a039a7c59 : SUCCESS in 53m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/469b5db5fab54840bcf9d23cab96fe6e : SUCCESS in 1h 00m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7a595af546fb47e59c0db02cbb665ac4 : SUCCESS in 1h 02m 27s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9280dedc704542b987f92a2363acc462 : SUCCESS in 56m 49s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9e1284140c9e40c39e4d40976822f95c : SUCCESS in 59m 49s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/0f19ccc298b24ebcb3331301755d1cfb : SUCCESS in 1h 14m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2c5f077dbc34eb1a39ae1f0a91ed9d0 : SUCCESS in 27m 01s","accounts_in_message":[],"_revision_number":2},{"id":"0a24fcdb775d3fb90d9b9a5fb24d2f2dc0c81db4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-03 16:28:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"86b63d5f588b6d8e6d15bd5cfc0cb56b42facbad","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-03 16:36:14.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/733082/2 : FAILURE in 7m 37s","accounts_in_message":[],"_revision_number":2},{"id":"cbf8b5d0084efcd73e608d85b27fe457a943c287","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 18:30:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/d8bc1e6486ee478eaa3b48b4e293d674 : SUCCESS in 50m 48s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/917bb6186fbd4ae9a88cd0b90e54675f : SUCCESS in 36m 19s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/282980d2368b48c58211b3466d3d4509 : FAILURE in 1h 29m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/d4e05a2ba7dd470d8b34f83bd134b4e3 : SUCCESS in 40m 40s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/42dd4fdf2528487da3c6eabe94a5e8ba : SUCCESS in 48m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cddec83b48a4eedacb85831c0fef0ec : SUCCESS in 9m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49f2de886d1e4deb8e2384f09c3795b0 : SUCCESS in 30m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b0ab4b84ae0646558cc205d2a5108a65 : SUCCESS in 24m 56s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3f785420777a45fa96c10c6a725aa94d : SUCCESS in 5m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5b4fd7a7e06840fea04e217cbb2f943d : SUCCESS in 1h 19m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e97d42bfa98d41efae2ffbf57644e338 : SUCCESS in 1h 00m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d01569e1d65243dd9cacfd7fc890434d : SUCCESS in 1h 13m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3fb41b35355c4e0a86664ce467ff9436 : SUCCESS in 44m 18s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2cbeac0e778643a7b1eebea283f8f44c : FAILURE in 1h 30m 05s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/130ce84e20d140c69f98528aec34312e : SUCCESS in 1h 54m 07s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8d447d260e444c35b42cd3799f1eee26 : SUCCESS in 1h 33m 15s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/45f2b6c3299648fc94f464c1b10da3b3 : SUCCESS in 1h 39m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b847e7e2118f421ca8b59cf809df6335 : SUCCESS in 1h 29m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f71f75d5c618486699320955e9ddbcc0 : SUCCESS in 1h 06m 25s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b5e0fdc9517a45b9b8d09837e47f31a4 : SUCCESS in 2h 00m 19s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e732739e78ca4093bb3de2bef3afa58d : SUCCESS in 1h 36m 21s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/7c4c0c2653484736a689fcaa946a1888 : SUCCESS in 1h 04m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16f98d5577ba46b6940526ad77524500 : SUCCESS in 26m 07s","accounts_in_message":[],"_revision_number":2},{"id":"64ac5af39090aba2b9954f3f74ceea23850750f2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-04 01:51:56.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/82/733082/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/82/733082/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"79058d1b15894844a52c53885b0695e31cff8cc0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-06-05 12:34:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0a4b2d2bc9fa0e6c7c1ae50e57832edaa3bc4d6f","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-06-16 20:22:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"95231fb384fb81fd135ab89f5bbe748a791b8125","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 20:22:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d5046c33b2ca5604a0e5f0f2809a9613bebd0cc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 22:00:45.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/62dc107dc6d74a7989a192f22af9e79d : FAILURE in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a0ce633c0364fc895cae5131f6c0be7 : SUCCESS in 13m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a10f4ec465fd467b8fa5bba4bf179141 : SUCCESS in 26m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b569aaca0a83400fbeafc64058626db5 : SUCCESS in 16m 17s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/114ea28099f8463d8139d90defa11079 : SUCCESS in 8m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cc35665ec73d42af86d7a6c6897742e1 : FAILURE in 14m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9f8270e5693f42dfb7b5b6ae08dbabd5 : FAILURE in 20m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c2c10e8b7e44c929e579bae232248c5 : FAILURE in 17m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7a9a092162124419a9358a4b2ce720b0 : FAILURE in 11m 34s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ed10d11a6362470bb8e07a54e4b13e7b : FAILURE in 14m 19s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2566cc0dc23d4b11a8bf68449f14f224 : FAILURE in 13m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7925b0586ce47db9d68408d034c6348 : FAILURE in 18m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d174a57d558f4c37805cc58c82a51523 : FAILURE in 19m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/227154c8da5c4898b06e188a240247f3 : SUCCESS in 29m 26s","accounts_in_message":[],"_revision_number":2},{"id":"3984d2d0785d7eab1a875bc7cd3d0475efbc89f7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-26 07:01:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"82add6f4b3459f07da74dd6a8288af253dc8691e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-26 07:11:24.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/733082/2 : FAILURE in 9m 03s","accounts_in_message":[],"_revision_number":2},{"id":"6c01ca9182ee0b6a11ed898a7624d0f045b10dc1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-26 07:15:43.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/82/733082/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/82/733082/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"1342677b73ef29e828a2d700da0dfb098beaa655","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 08:17:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/f32ab067b6ca49baaf08c9de917fcb07 : FAILURE in 9m 35s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/51ebf0d7cac54de4aca0e28e5d6b7fce : FAILURE in 18m 45s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/21166544537f462a87d862f76e249e3b : FAILURE in 15m 14s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/158cc96db7274aa3a1754b009b29cf34 : FAILURE in 14m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11b66d4ed23b456a8cabeb19c807ad2e : SUCCESS in 11m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4ecc0553ccff4fc9836f3eaf68f95044 : SUCCESS in 24m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/60d5532b5b6446da9f48fd46a4a5b01e : SUCCESS in 24m 34s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f3c47c3fa2b74e6093c2bfbfdc5ec670 : SUCCESS in 8m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7ea49eca467340c6ba3bc03269342c2e : FAILURE in 13m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ef8bff5e12584f02b5bbf8cc299656f0 : FAILURE in 17m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/891ad938ff404069baa3538a7deff361 : FAILURE in 11m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bbcbaf5ef9d5425694ca9b65d5c2e575 : FAILURE in 16m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ec98f9d6241b457680850f88cca48b46 : FAILURE in 9m 42s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/85a3d38cf45e41f2bebcdae9a3fe0353 : FAILURE in 14m 51s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f36c98f3d109446e88c18b59a6d0748d : FAILURE in 16m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4fb8c6954f3e4fe79babcb2df1a9841e : FAILURE in 15m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c0f8769e09a4f6ca09d7db5c6119749 : FAILURE in 17m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/14cccd5e1d6c484da9f1fe8e677a571a : FAILURE in 22m 05s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f94d149d334e424ab9a0efd4587cfb90 : FAILURE in 19m 54s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/140391cdf3c140f3a89d7d879c7c1264 : FAILURE in 23m 27s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/859089f91f39422eb74b3bf5f8715ff9 : FAILURE in 14m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef4d5641b63d4cf89805e593b6c35a6b : SUCCESS in 30m 15s","accounts_in_message":[],"_revision_number":2},{"id":"64c019722c99b962275664d44853a7fdf5fad0e8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-07-07 14:44:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"fa97d03ef787ad5e37d50d48934da46b60c40473","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-07-07 14:54:41.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/733082/3 : FAILURE in 8m 49s","accounts_in_message":[],"_revision_number":3},{"id":"24b4561e2afcb94c46a2b12cc38355bd06d62490","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-07-07 14:55:42.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d7e3ff95cf46293e7e127ddfc0eb8fbac46d0cba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-07 15:12:50.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/82/733082/3/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/82/733082/3/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"c3e00fb840e95e7c95ba3b1c7a3df59ba36e8a07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 20:02:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/0f91ac9f1f634f2e8ac108d8678c6d49 : SUCCESS in 47m 08s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/e79fdd5a88e44eb281e1aeb705f9a306 : FAILURE in 1h 34m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/15b7ddfbcf7c434dbae604d7291fc735 : SUCCESS in 56m 54s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/155adc602a3c4a5d8cef0f3214ef5927 : SUCCESS in 56m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ad8ec8618d541e3b5eb6ac4244cf3f5 : SUCCESS in 12m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/682a8320258b49398667c671f5f5df37 : SUCCESS in 23m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f284c98f4e5d46cca2f9a17b0e16dac9 : SUCCESS in 24m 01s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/fa15fa6e623a4d9488453d85d670da8c : SUCCESS in 7m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/241fb6aa3ae34e3f8a5e0cbff314979a : SUCCESS in 1h 18m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75a919736d3f4d3e8f9c013d88febd9a : SUCCESS in 1h 16m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c2d2dd103e7a424e8e36a0f90daf080a : SUCCESS in 37m 11s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/c1efbb02bf804777a454255d95fb0f8d : SUCCESS in 1h 45m 34s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b5d969b58e94475c980cf08b434bcfbb : SUCCESS in 1h 29m 03s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a91350912b334d0aaa59db6f7a648f5c : SUCCESS in 1h 18m 39s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/29227d33996b4830ad9b4091c0b3d232 : SUCCESS in 1h 27m 16s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a619dd0d90e2472d9f390125ddc88f1c : SUCCESS in 1h 40m 04s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/bf98b279ea4b4d3aa982365cb8852540 : SUCCESS in 56m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2e123c8ef3e4215a55e5da77b321aff : SUCCESS in 34m 51s","accounts_in_message":[],"_revision_number":3},{"id":"a8287a1d5f78720cf538dfb5fdb1e19eb47de3a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 20:02:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"375f44518def834ab0416d59d426c0b88a4589b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 21:33:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4a54a7d46e1116dead2f96c8ac924796d0e0f4f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 21:33:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/896bb56d4ac5455d898b433d0f0819f4 : SUCCESS in 1h 04m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba44be9a2f8d4b14b6b20c870e208d1f : SUCCESS in 14m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/20568138598e410594ab45bab5233dc1 : SUCCESS in 28m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/527b74b6c0524f1f9834919cb053f8ff : SUCCESS in 25m 44s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/35a9c7ef23bd4249bdfe735dc7ebffc8 : SUCCESS in 7m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9dc8aee427fd41d8af6a08415d8eb9ef : SUCCESS in 1h 18m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e22aef62094e42a485044e4612df7c17 : SUCCESS in 1h 25m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/574888e5eddd4d5fa9a04142b795ccd3 : SUCCESS in 41m 17s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/81579e334ca049d59a3a94cf9cd12ec2 : SUCCESS in 1h 24m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/15973c23a9ca4a8d9bdf64638ea628f9 : SUCCESS in 1h 19m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f566910ede694a9ab339b0073505c141 : SUCCESS in 29m 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2baf0aa83f551b5c88caad64da837630b9daea46","revisions":{"05894eaeee9dc6be83bb4d1eacaacd8ce4337b2c":{"kind":"REWORK","_number":1,"created":"2020-06-03 09:47:32.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/82/733082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/733082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/733082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/733082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/733082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/733082/1"}}},"commit":{"parents":[{"commit":"625505ecc77c36b6269af4ae94ac349e1677a8af","subject":"Merge \"Allow usage of legacy 3rd-party interface drivers\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/625505ecc77c36b6269af4ae94ac349e1677a8af"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-04-09 12:37:38.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-03 09:47:32.000000000","tz":0},"subject":"Ensure that stale flows are cleaned from phys_bridges","message":"Ensure that stale flows are cleaned from phys_bridges\n\nIn case when neutron-ovs-agent will notice that any of physical\nbridges was \"re-created\", we should also ensure that stale Open\nFlow rules (with old cookie id) are cleaned.\nThis patch is doing exactly that.\n\nConflicts:\n    neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py\n\nChange-Id: I7c7c8a4c371d6f4afdaab51ed50950e2b20db30f\nRelated-Bug: #1864822\n(cherry picked from commit 63c45b37666efc7fb33fc154fa828050b3dc7945)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05894eaeee9dc6be83bb4d1eacaacd8ce4337b2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05894eaeee9dc6be83bb4d1eacaacd8ce4337b2c"}]},"branch":"refs/heads/stable/queens"},"c1e39eab985af754f271845c374553fd993ab0da":{"kind":"REWORK","_number":2,"created":"2020-06-03 10:58:43.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/82/733082/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/733082/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/733082/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/733082/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/733082/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/733082/2"}}},"commit":{"parents":[{"commit":"625505ecc77c36b6269af4ae94ac349e1677a8af","subject":"Merge \"Allow usage of legacy 3rd-party interface drivers\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/625505ecc77c36b6269af4ae94ac349e1677a8af"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-04-09 12:37:38.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-03 10:58:43.000000000","tz":0},"subject":"Ensure that stale flows are cleaned from phys_bridges","message":"Ensure that stale flows are cleaned from phys_bridges\n\nIn case when neutron-ovs-agent will notice that any of physical\nbridges was \"re-created\", we should also ensure that stale Open\nFlow rules (with old cookie id) are cleaned.\nThis patch is doing exactly that.\n\nConflicts:\n    neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py\n\nChange-Id: I7c7c8a4c371d6f4afdaab51ed50950e2b20db30f\nRelated-Bug: #1864822\n(cherry picked from commit 63c45b37666efc7fb33fc154fa828050b3dc7945)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1e39eab985af754f271845c374553fd993ab0da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1e39eab985af754f271845c374553fd993ab0da"}]},"branch":"refs/heads/stable/queens"},"2baf0aa83f551b5c88caad64da837630b9daea46":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-07 14:44:41.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/82/733082/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/733082/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/733082/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/733082/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/733082/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/733082/3"}}},"commit":{"parents":[{"commit":"774b3ff89e5ec01494d7ca66773aba8984f0cd80","subject":"Fix queens gates, multiple fixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/774b3ff89e5ec01494d7ca66773aba8984f0cd80"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-04-09 12:37:38.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-07-07 14:44:41.000000000","tz":0},"subject":"Ensure that stale flows are cleaned from phys_bridges","message":"Ensure that stale flows are cleaned from phys_bridges\n\nIn case when neutron-ovs-agent will notice that any of physical\nbridges was \"re-created\", we should also ensure that stale Open\nFlow rules (with old cookie id) are cleaned.\nThis patch is doing exactly that.\n\nConflicts:\n    neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py\n\nChange-Id: I7c7c8a4c371d6f4afdaab51ed50950e2b20db30f\nRelated-Bug: #1864822\n(cherry picked from commit 63c45b37666efc7fb33fc154fa828050b3dc7945)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2baf0aa83f551b5c88caad64da837630b9daea46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2baf0aa83f551b5c88caad64da837630b9daea46"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
