)]}'
{"id":"openstack%2Fneutron~735099","triplet_id":"openstack%2Fneutron~stable%2Fussuri~I1588b88279ba1e25706cc74f4a888901e5aadcb6","project":"openstack/neutron","branch":"stable/ussuri","topic":"fix_ovn_migration_ssh-stable/ussuri","hashtags":[],"change_id":"I1588b88279ba1e25706cc74f4a888901e5aadcb6","subject":"Fix ssh to nodes during ML2/OVS to ML2/OVN migration","status":"MERGED","created":"2020-06-11 10:06:37.000000000","updated":"2020-06-11 15:51:23.000000000","submitted":"2020-06-11 15:44:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735099-1591890280347-00e32a87","meta_rev_id":"ee0bcae255aa351fc6e5639f2af7a9e3b3adddba","_number":735099,"virtual_id_number":735099,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 15:44:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-11 11:53:30.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2020-06-11 13:57:56.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},{"value":2,"date":"2020-06-11 13:44:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":1,"date":"2020-06-11 13:57:56.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"date":"2020-06-11 10:06:37.000000000","_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-11 10:06:37.000000000","updated_by":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"reviewer":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"state":"REVIEWER"},{"updated":"2020-06-11 11:53:30.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-06-11 13:44:50.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-06-11 13:57:56.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-06-11 15:44:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"48b5e877f8db71c9119b50aece55fa4f9bc0eb75","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-06-11 10:06:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0ad50108871611aa1bd06d81e0afd8eef3402fde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 11:32:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b64e97d858e7427594744987d678cd34 : SUCCESS in 31m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fafc1444f70e44aa8d2165e0b63ea639 : SUCCESS in 30m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12b36dad5e5341b29674b2ca3faf95d3 : SUCCESS in 14m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90cfc5f9cb184042adafb9bfd8dce4a7 : SUCCESS in 24m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ddb5a737c75c4c63ae4e0c895f710928 : SUCCESS in 26m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e86583274ca049e9ba19f10720f44a7b : SUCCESS in 28m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4abd84f93304e78b609791a94b847f6 : SUCCESS in 9m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ed179b8926fc4e30a4e2f54e11788ae0 : SUCCESS in 53m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/85790b36df024a4a9c3d999c28ba3402 : SUCCESS in 1h 25m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7090c107444440b8a696e151c70b3f59 : SUCCESS in 1h 12m 23s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/34a58e30b18046bebbb15ce4b2e213b1 : SUCCESS in 1h 19m 30s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/931be77cae2746dfa0007dfa8a273b46 : SUCCESS in 48m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d624599b09bdba284583559b823828f2a35391b8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-11 11:53:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/735099/1 : SUCCESS in 1h 46m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e76b11a9fb86c8f23bfd66c856989d8cde1a9f18","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-06-11 13:44:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c1afc8901b4322fc5cdeaee2bdc4b192e440a2e7","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-06-11 13:57:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a8730b9176d0e83b9095420eb5a507042597d1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 13:58:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5453fbc13493fe029612842b9c547c34b642a73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 15:44:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b659a738d07a4efe8e3d06bcb4d051a2 : SUCCESS in 25m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8b2abc792564d2a86c1c2e0da52581f : SUCCESS in 11m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e66467644cf54c9d843489a81396d530 : SUCCESS in 26m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92b76b8c8bf643659e50d49e461415eb : SUCCESS in 17m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f08280b90bb4f7298d76c2313e937e6 : SUCCESS in 7m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4396749f801e49109d3227c66a6d4ef0 : SUCCESS in 39m 41s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d95bce2cec444682ab00c64183681e9f : SUCCESS in 1h 22m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/52592b4d94aa423dbed77c8b8a55c7fd : SUCCESS in 1h 21m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0d42997736e6534b6834491509b2a9365c5b80ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 15:44:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"17a5b750c400aae033356f2f22441dfac827e59c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 15:51:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56968ab86241491894a6b1de7549f46c : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd49fae3f4b6462cbc17a49bd47f19554dd834ef","revisions":{"dd49fae3f4b6462cbc17a49bd47f19554dd834ef":{"kind":"REWORK","_number":1,"created":"2020-06-11 10:06:37.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/99/735099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/735099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/735099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/735099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/735099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/735099/1"}}},"commit":{"parents":[{"commit":"14652b3ea2d508a0a7f44e5c607268404a9a2ebd","subject":"Merge \"Don\u0027t check if any bridges were recrected when OVS was restarted\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/14652b3ea2d508a0a7f44e5c607268404a9a2ebd"}]}],"author":{"name":"Roman Safronov","email":"rsafrono@redhat.com","date":"2020-05-27 22:19:41.000000000","tz":180},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2020-06-11 10:06:37.000000000","tz":0},"subject":"Fix ssh to nodes during ML2/OVS to ML2/OVN migration","message":"Fix ssh to nodes during ML2/OVS to ML2/OVN migration\n\nMigration ansible playbooks are not able to connect to overcloud\nnodes using ssh due to bad settings in the generated ansible\ninventory file. Ctlplane subnet ip addresses should be used by\nansible in order to be able to establish connection successfully.\n\nCloses-bug: #1881029\nChange-Id: I1588b88279ba1e25706cc74f4a888901e5aadcb6\n(cherry picked from commit 0402337c151d6db3b6e04569df3735e0dc44125f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd49fae3f4b6462cbc17a49bd47f19554dd834ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd49fae3f4b6462cbc17a49bd47f19554dd834ef"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
