)]}'
{"id":"openstack%2Fneutron~736091","triplet_id":"openstack%2Fneutron~stable%2Fpike~If034770fda98a0f74c0a14f954a16b5c4375bd9b","project":"openstack/neutron","branch":"stable/pike","topic":"bug/1883636","hashtags":[],"change_id":"If034770fda98a0f74c0a14f954a16b5c4375bd9b","subject":"fix a bug that when create a subnet the gateway of it out range of subnet","status":"ABANDONED","created":"2020-06-17 04:40:05.000000000","updated":"2020-09-25 15:41:02.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0f14dcc8408e74e5541994f177ab7d43de5cae59","_number":736091,"virtual_id_number":736091,"owner":{"_account_id":30129,"name":"yongjin wu","email":"626448787@qq.com","username":"wuyongjin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-06-17 07:20:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"date":"2020-06-17 04:40:17.000000000","_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"date":"2020-09-25 15:41:02.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-17 05:22:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-06-17 04:58:38.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":-1,"date":"2020-06-17 16:17:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-17 04:40:17.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2020-06-17 04:58:38.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-06-17 05:22:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-17 07:20:05.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-17 16:17:13.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-25 15:41:02.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"5fa74e5d8c9ca03d998e81c613e91af784c0a3a9","author":{"_account_id":30129,"name":"yongjin wu","email":"626448787@qq.com","username":"wuyongjin"},"date":"2020-06-17 04:40:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5e7efa839d9367aad375f5c0222b377d985a129","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2020-06-17 04:40:17.000000000","message":"Patch Set 1:\n\nCongratulations, you\u0027ve proposed your first change in OpenDev.\n\nYour submission will now be tested automatically by Zuul, our gatekeeper,\nand reviewed by other friendly developers. They will give you feedback and\nmay require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches often take days (and sometimes weeks) to get reviewed, so be\npatient. Don\u0027t hesitate to ask for help, and answer questions about your\nwork promptly if you can. The more you get to know reviewers and get to be\nknown by them, the smoother the review and approval process will become. The\nfastest way to accomplish this is by reviewing other proposed changes\nyourself: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for participating in OpenDev, we look forward to seeing you\naround.\n\nWorkflow Guide: https://docs.openstack.org/infra/manual/developers.html","accounts_in_message":[],"_revision_number":1},{"id":"058aed14194fd10e2bcbf971a73d4b4fb79b2a0b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-17 04:58:38.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/736091/1 : FAILURE in 17m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e42b48f5784fd582006a9716135752bf33f10ca8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-17 04:58:58.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/91/736091/1/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"ab715d35e04660f3458a550d393688ad17c04bf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 05:22:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4012e8e46a8749cdaca9763eef6fc021 : FAILURE in 7m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2a222919a3e24a1f88670284622a284c : FAILURE in 13m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/298e13959a204215b89695b10fcd61bb : TIMED_OUT in 41m 02s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/89b38eb5fb174b5abf93e430eb8704de : SUCCESS in 5m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3b5313b7331741cf9641f5ab5954dfcd : FAILURE in 7m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/00c519d76110491e87fe0109a829629b : FAILURE in 18m 18s (non-voting)\n- neutron-functional https://zuul.opendev.org/t/openstack/build/950be0191d364ec0b1f1d98b9359c999 : FAILURE in 6m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d11d9079714f4701b4f9ca939314b694 : FAILURE in 6m 42s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/970551d3233c4141a3d3a87a476d8236 : FAILURE in 20m 06s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e1f62b1fc26d4d56a641f2c10aa31d95 : FAILURE in 8m 37s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/73eca5bd880c49a8a20f0a4f5f93ced2 : FAILURE in 14m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f780ad220d594d9e90d4881e1612d7ac : FAILURE in 23m 32s (non-voting)\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dda00dffdeb349948c6c74b0d89fb624 : FAILURE in 23m 27s (non-voting)\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4e9f8290a4c34379b9f8a77cd053f153 : FAILURE in 10m 53s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4d392aa949a241c28d364f66cc6fd3b5 : FAILURE in 25m 56s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/d3993756c7e5468cb7111106490da653 : FAILURE in 18m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/874c4da4866b4254be3a02d1bd432a8f : FAILURE in 32m 01s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/c9f02d93be0649b7ac38a14edb820ff6 : FAILURE in 18m 25s\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/a52d78f35f5842f2aebc0ec637c27ec3 : FAILURE in 14m 47s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bf4899ddef614c6283f7207764af1063 : FAILURE in 6m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6910e5ac08d12279b24ce3fc94952528407aa1b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-17 07:20:05.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/91/736091/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/91/736091/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"4fc9dc2e9b35c96556030e08a50c86b835887fa7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-17 16:17:13.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nIs this a cherry-pick from another upstream branch?","accounts_in_message":[],"_revision_number":1},{"id":"9364160044b1a2987d075d3954752dff427607aa","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-25 15:41:02.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e3966922dc6d0b9bab87bfeafcdf3fcfff8fe3e","revisions":{"4e3966922dc6d0b9bab87bfeafcdf3fcfff8fe3e":{"kind":"REWORK","_number":1,"created":"2020-06-17 04:40:05.000000000","uploader":{"_account_id":30129,"name":"yongjin wu","email":"626448787@qq.com","username":"wuyongjin"},"ref":"refs/changes/91/736091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/736091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/736091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/736091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/736091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/736091/1"}}},"commit":{"parents":[{"commit":"8f280339096ed702928f704fd20543ce4370c0d4","subject":"Revert iptables TCP checksum-fill code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f280339096ed702928f704fd20543ce4370c0d4"}]}],"author":{"name":"wuyongjin","email":"626448787@qq.com","date":"2020-06-17 04:21:39.000000000","tz":-240},"committer":{"name":"wuyongjin","email":"626448787@qq.com","date":"2020-06-17 04:21:39.000000000","tz":-240},"subject":"fix a bug that when create a subnet the gateway of it out range of subnet","message":"fix a bug that when create a subnet the gateway of it out range of subnet\n\nFixes: bug #1883636\nChange-Id: If034770fda98a0f74c0a14f954a16b5c4375bd9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e3966922dc6d0b9bab87bfeafcdf3fcfff8fe3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e3966922dc6d0b9bab87bfeafcdf3fcfff8fe3e"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
