)]}'
{"id":"openstack%2Fneutron~736865","triplet_id":"openstack%2Fneutron~stable%2Frocky~I10f5de4e5f7deb1c2fefdab28faf9aef2c79caa5","project":"openstack/neutron","branch":"stable/rocky","topic":"test-sanity","hashtags":[],"change_id":"I10f5de4e5f7deb1c2fefdab28faf9aef2c79caa5","subject":"[DO NOT MERGE]: Test branch sanity","status":"ABANDONED","created":"2020-06-18 23:08:34.000000000","updated":"2020-06-19 01:48:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5e1ca42ac9ded66b17338d817a968b74046ae788","_number":736865,"virtual_id_number":736865,"owner":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-18 23:41:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-06-18 23:17:24.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":-1,"date":"2020-06-18 23:08:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 23:17:24.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-06-18 23:41:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f9ee8b7d1e3cfa40b842c2147a6d9f06154c2ae8","author":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"date":"2020-06-18 23:08:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"597af6fb0b087943aab39d3f9d282d95b9fd3136","author":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"date":"2020-06-18 23:08:58.000000000","message":"Patch Set 1: Code-Review-1\n\nTesting upstream gate for stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"84cc635b11bc05c37a4d34abcf605781ad851e4d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-18 23:17:24.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/736865/1 : FAILURE in 8m 10s","accounts_in_message":[],"_revision_number":1},{"id":"99dc0b3fb268f1e2d62c8ab22f4ab7867a631579","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 23:41:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ce399cd226142dcb31947836a2843b8 : SUCCESS in 11m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a5f906f003849eb8a1d4fbae90c2c2f : SUCCESS in 23m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ad97934cc4f6447b8bb4b378c80ad755 : SUCCESS in 22m 13s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ea1340fb9ff34a7c8a1526988b52e0fc : SUCCESS in 8m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e880deab09a4e73831b185bcacb0900 : FAILURE in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f10173b5a3074ac7bdfd12633bf59c65 : SUCCESS in 8m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e37fbd851b8a42719e01b7c14a665053 : FAILURE in 10m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/cd8154877c5a464d91fa9d4ff0cef60b : FAILURE in 10m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a953a3cebe2a4fc797cf8e87aa493778 : SUCCESS in 27m 02s\n- neutron-fullstack-python35 https://zuul.opendev.org/t/openstack/build/b96fccd7596d4ccd8f5600ed0fcebd42 : FAILURE in 10m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63034774dc9747b4bb7d432a54338d49 : SUCCESS in 31m 21s","accounts_in_message":[],"_revision_number":1},{"id":"811ed91fd12f74cd5af2173ba40574c50bb49bdf","author":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"date":"2020-06-19 01:48:23.000000000","message":"Abandoned\n\nnot needed.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1220a73dd2990960b07c9e1e53852008f6f25cb","revisions":{"c1220a73dd2990960b07c9e1e53852008f6f25cb":{"kind":"REWORK","_number":1,"created":"2020-06-18 23:08:34.000000000","uploader":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"ref":"refs/changes/65/736865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/736865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/736865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/736865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/736865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/736865/1"}}},"commit":{"parents":[{"commit":"daa4737cf3e10650800d8364e524175e40932d7b","subject":"Merge \"Dynamically increase l3 router process queue green pool size\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/daa4737cf3e10650800d8364e524175e40932d7b"}]}],"author":{"name":"Thomas Bachman","email":"tbachman@yahoo.com","date":"2020-06-18 23:07:23.000000000","tz":0},"committer":{"name":"Thomas Bachman","email":"tbachman@yahoo.com","date":"2020-06-18 23:08:30.000000000","tz":0},"subject":"[DO NOT MERGE]: Test branch sanity","message":"[DO NOT MERGE]: Test branch sanity\n\nChcking upstream gate for stable/rocky branch.\n\nChange-Id: I10f5de4e5f7deb1c2fefdab28faf9aef2c79caa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1220a73dd2990960b07c9e1e53852008f6f25cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1220a73dd2990960b07c9e1e53852008f6f25cb"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
