)]}'
{"id":"openstack%2Fneutron~736951","triplet_id":"openstack%2Fneutron~stable%2Fussuri~I83688031951b97bfe64f3aaa761ad7afc1d5ea55","project":"openstack/neutron","branch":"stable/ussuri","hashtags":[],"change_id":"I83688031951b97bfe64f3aaa761ad7afc1d5ea55","subject":"ovn: Use newer OVS","status":"ABANDONED","created":"2020-06-19 08:30:17.000000000","updated":"2020-06-22 09:41:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"34d29e854112734087afe493bbe534f9d6315972","_number":736951,"virtual_id_number":736951,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-06-22 09:17:42.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 08:30:17.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-06-19 09:17:03.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-06-19 10:38:43.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-06-19 13:36:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-22 09:17:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"cb0239bbd5e828964f0adac53e663644e33bc3f5","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-06-19 08:30:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1449f79b72247a8f72ac0007c74603abbfee7971","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-19 08:46:27.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/736951/1 : FAILURE in 15m 31s","accounts_in_message":[],"_revision_number":1},{"id":"42903728c44f366a465f7a6e99a3cb4ed54a7495","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-06-19 09:17:03.000000000","message":"Patch Set 1:\n\nTo pass gates you will need to have this one based on https://review.opendev.org/#/c/736286 (and re-enable relevant jobs here)\nOr wait for it to be merged, but it has that six issue we have in master","accounts_in_message":[],"_revision_number":1},{"id":"f869a7b70021df61b866df4974290d90bb06609f","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-06-19 10:08:59.000000000","message":"Patch Set 1:\n\n\u003e To pass gates you will need to have this one based on\n \u003e https://review.opendev.org/#/c/736286 (and re-enable relevant jobs\n \u003e here)\n \u003e Or wait for it to be merged, but it has that six issue we have in\n \u003e master\n\nThat\u0027s a good point! Let me rebase on it and re-enable the ovn jobs","accounts_in_message":[],"_revision_number":1},{"id":"3dd45123db29a6241c41795acfabe2fc261654a0","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-06-19 10:19:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a1c536d31737f960b0e50c9ed0d873418d660706","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-19 10:38:43.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/736951/2 : FAILURE in 18m 09s","accounts_in_message":[],"_revision_number":2},{"id":"f04d364dbdf92848d3c4fff0f345b6c4d54c2609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 13:36:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/03e39c767dee4aab8cf7c91ed31f1dae : SUCCESS in 50m 28s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/c952f66a7dc44f51b5c37dcb4d41f9a2 : SUCCESS in 25m 08s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/454c5147e35e4c07907c1118e3e512b1 : SUCCESS in 1h 08m 51s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/341edf6b04af44cfa2c60a1af4f51a4d : SUCCESS in 45m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/59a8c8f63437448a9bbf3bc55ad83480 : SUCCESS in 53m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77a029f3ba324add8bb4990d08519b1a : SUCCESS in 23m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4f6954c1b13403f958dd595842f4077 : SUCCESS in 18m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d17aff3c7354aa8af11821821ab6ea8 : SUCCESS in 11m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83c51338a87b4351a8d973bbef39d450 : SUCCESS in 29m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15270a9ee1674af097a1bb9fe2078458 : SUCCESS in 18m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45239e29b3604cc587691e877eecf1c7 : SUCCESS in 18m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e8b0f02628b456d91c1d92100d62ba2 : SUCCESS in 9m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6f2f7ecb8c5e45b2bcd2f9e386a4bd41 : RETRY_LIMIT in 7m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/261a303b5542455380aed4b3a8eb2092 : RETRY_LIMIT in 12m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca648830ac2c47e89cdbef75ed3c9be2 : RETRY_LIMIT in 8m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/668f8a4d8c964c9cb2f72d072d54d68e : SUCCESS in 1h 07m 57s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5d43f90d382a46d3be068d39839cb61c : SUCCESS in 48m 53s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8895dd55e96a43ea852c2d652820e44a : SUCCESS in 1h 21m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c72ad59b52a94b9d84eb157e341041eb : SUCCESS in 52m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bb87bcc35cb8414ea5d2032b54ebe9e6 : SUCCESS in 1h 09m 08s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5df1f16464f74f2fb25d16d3d4b5d22c : FAILURE in 1h 27m 32s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2f41bf6a505f4f1285d16c4a14c1a0ce : SUCCESS in 42m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a0b8906ff63e464697173f49275f0e31 : SUCCESS in 41m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e9e401aca00a4a3d8181c489ee91a024 : SUCCESS in 1h 32m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b94d9fbbffde4d46a51b4febc3391ad1 : SUCCESS in 40m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c7e109fbfd694c0f898cbe5a8fdab54a : SUCCESS in 1h 10m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1a3362b8ad58420fadfb933aa13bbd31 : SUCCESS in 1h 14m 51s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f0d877a04b894f87919021936279f0da : SUCCESS in 1h 03m 28s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a65779c7447d40ef92ed84db81857fae : RETRY_LIMIT in 8m 16s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/940a35fc58db424cb4bf4362c76446d5 : SUCCESS in 2h 25m 49s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3af57e5ad691472aadbbefbb1c1a4a1d : SUCCESS in 2h 14m 21s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/741ee25487844010b36404d4b39828a4 : SUCCESS in 3h 13m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4f30888f9a034f4f8b5ebf7fa7c6e10b : FAILURE in 2h 29m 02s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/ae20672ba4e34ac788d3a64cae78b5da : SUCCESS in 2h 43m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dbe70db368097926b84e55b0e9279f947f14620e","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-06-22 09:02:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"34d29e854112734087afe493bbe534f9d6315972","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-06-22 09:41:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"49870573d1784f26e83f15eaf208c004bec2f4c1","revisions":{"1369b6a5b0124436283e9f369bc4db7a204bfffc":{"kind":"REWORK","_number":1,"created":"2020-06-19 08:30:17.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/51/736951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/736951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/736951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/736951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/736951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/736951/1"}}},"commit":{"parents":[{"commit":"e975365ffbc1c5a94621cadd4a87f86003f88011","subject":"Merge \"[OVN] Create localnet port for each created segment\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e975365ffbc1c5a94621cadd4a87f86003f88011"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2020-06-16 08:07:49.000000000","tz":120},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-06-19 08:30:15.000000000","tz":60},"subject":"ovn: Use newer OVS","message":"ovn: Use newer OVS\n\nOVS tag v2.13.0 is not compatible with current used kernel on upstream\nBionic Ubuntu kernel 4.15.0. This patch sticks to commit hash to unblock\nthe gate. We can either change to newer 2.13 tag once released or better\nstop compiling OVS and use one packaged by Ubuntu.\n\nChange-Id: I83688031951b97bfe64f3aaa761ad7afc1d5ea55\nCloses-bug: #1883601\n(cherry picked from commit f5b6bd8f55a66a856ec00b2333e1968d0c47d5f8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1369b6a5b0124436283e9f369bc4db7a204bfffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1369b6a5b0124436283e9f369bc4db7a204bfffc"}]},"branch":"refs/heads/stable/ussuri"},"76b188211ac1b1198bff3c76f84b3b7cdfdda68e":{"kind":"REWORK","_number":2,"created":"2020-06-19 10:19:33.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/51/736951/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/736951/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/736951/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/736951/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/736951/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/736951/2"}}},"commit":{"parents":[{"commit":"d4bdad10af7503c825445e7592315f2721c771be","subject":"Temporarily disable grenade-multinode and ovn jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4bdad10af7503c825445e7592315f2721c771be"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2020-06-16 08:07:49.000000000","tz":120},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-06-19 10:17:51.000000000","tz":60},"subject":"ovn: Use newer OVS","message":"ovn: Use newer OVS\n\nOVS tag v2.13.0 is not compatible with current used kernel on upstream\nBionic Ubuntu kernel 4.15.0. This patch sticks to commit hash to unblock\nthe gate. We can either change to newer 2.13 tag once released or better\nstop compiling OVS and use one packaged by Ubuntu.\n\nThis patch also re-enables the OVN jobs in the check and gate pipeline\nthat were disabled temporarily due to this versioning problem.\n\nChange-Id: I83688031951b97bfe64f3aaa761ad7afc1d5ea55\nCloses-bug: #1883601\n(cherry picked from commit f5b6bd8f55a66a856ec00b2333e1968d0c47d5f8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/76b188211ac1b1198bff3c76f84b3b7cdfdda68e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/76b188211ac1b1198bff3c76f84b3b7cdfdda68e"}]},"branch":"refs/heads/stable/ussuri"},"49870573d1784f26e83f15eaf208c004bec2f4c1":{"kind":"REWORK","_number":3,"created":"2020-06-22 09:02:42.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/51/736951/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/736951/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/736951/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/736951/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/736951/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/736951/3"}}},"commit":{"parents":[{"commit":"e975365ffbc1c5a94621cadd4a87f86003f88011","subject":"Merge \"[OVN] Create localnet port for each created segment\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e975365ffbc1c5a94621cadd4a87f86003f88011"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2020-06-16 08:07:49.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2020-06-22 09:02:42.000000000","tz":0},"subject":"ovn: Use newer OVS","message":"ovn: Use newer OVS\n\nOVS tag v2.13.0 is not compatible with current used kernel on upstream\nBionic Ubuntu kernel 4.15.0. This patch sticks to commit hash to unblock\nthe gate. We can either change to newer 2.13 tag once released or better\nstop compiling OVS and use one packaged by Ubuntu.\n\nRun functional/fullstack tests with selected OVN and OVS versions\n\nPreviously for functional tests OVS version was hardcoded - v2.12.0,\nand OVN was installed from OVS repository.\n\nAfter we merged OVN driver to Neutron tree we run both\nNeutron/OVS and Neutron/OVN functional tests in one job.\n\nThis patch adds possibility to specify from which tag/branch OVS and OVN\nshould be checkout.\n\nChange-Id: I83688031951b97bfe64f3aaa761ad7afc1d5ea55\nCloses-Bug: #1883601\nCloses-Bug: #1878160\n(cherry picked from commit d0860588957730b97fe3ebbefa77c6421da45adc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49870573d1784f26e83f15eaf208c004bec2f4c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49870573d1784f26e83f15eaf208c004bec2f4c1"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
