)]}'
{"id":"openstack%2Fneutron~737044","triplet_id":"openstack%2Fneutron~master~I70cb1fe651a50b2f6495d7a365a6beb2ba111c6d","project":"openstack/neutron","branch":"master","topic":"bug/1884273","hashtags":[],"change_id":"I70cb1fe651a50b2f6495d7a365a6beb2ba111c6d","subject":"Fix the wrong value for QoS rate conversion to bytes/s","status":"MERGED","created":"2020-06-19 16:03:43.000000000","updated":"2020-06-24 03:18:38.000000000","submitted":"2020-06-24 03:16:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737044-1592968575737-5e2ae216","meta_rev_id":"725694aa60f835f2502d23385a9207ab98c9edce","_number":737044,"virtual_id_number":737044,"owner":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},{"value":1,"date":"2020-06-20 21:03:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-24 03:16:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-20 21:24:31.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-23 22:56:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2020-06-22 20:30:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-23 22:56:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 17:19:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-06-20 21:03:16.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-20 21:24:31.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-06-22 20:30:30.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-06-23 22:56:00.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-06-24 03:16:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1565ece5a3c4486bbab2eee935b1179f2d476ff","author":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"date":"2020-06-19 16:03:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0cf0ea30f3efbea95911d592b2e94658e867505","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-19 17:19:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(9 comments)\n\nGood catch!","accounts_in_message":[],"_revision_number":1},{"id":"b2464bdba78f9463bc8d4f5fc0dfbc321861a8bf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-19 19:20:24.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/44/737044/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"88e4c20fa6b88f07a8a6382dace68ad3af674667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 20:37:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f3df116cb9bf40dead851c50636bce15 : SUCCESS in 58m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ad5b041979a4ad9919817be4056e109 : SUCCESS in 32m 02s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5ed40d6e6b5d42098b70e1651e1cf79f : SUCCESS in 1h 13m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8d053774c91c45aeaf8f8dda9d8fa301 : SUCCESS in 1h 11m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5498e6787f284707aa22d6e964750256 : SUCCESS in 1h 11m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75f1417e4e5b4f76be239726344766cf : SUCCESS in 34m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ace1d52b2b0e48e4bf8246b702393b9a : SUCCESS in 27m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88e352c15b0849fcb9ad08bfe2035e29 : SUCCESS in 11m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c218e88470a4425b53d75339751e3e1 : SUCCESS in 29m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/093f71f3499441bfa02a6f66f14d1e52 : SUCCESS in 28m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/522695a8cf234aa69bba7465007fb48c : SUCCESS in 11m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ef85303851394ecdab436c8724dd3b88 : RETRY_LIMIT in 10m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/274d5a6a4939457ebd6b69e6e888700a : RETRY_LIMIT in 9m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f9159699a51344488a3b0a6082c3d5e6 : RETRY_LIMIT in 15m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/83b9bd1614e84a64b6d1250cea247167 : RETRY_LIMIT in 13m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6bd58d87ce7c4d67af8d5089e3ff6c11 : SUCCESS in 1h 42m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/88d5965c45484a21965c048c8837bcba : SUCCESS in 1h 03m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/edaee819e369436d997c0c631b41b8da : SUCCESS in 53m 37s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d0084b4d99524c43a6ee96b2d78e0204 : SUCCESS in 1h 11m 20s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/0d573dddb7c943389be54b518bdce39b : SUCCESS in 1h 14m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/76bd6a9003464aaca1eb587879dcd5a4 : SUCCESS in 1h 10m 58s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6a87168ce742493aae589212915fcd23 : SUCCESS in 1h 29m 33s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b21c436cf19a4f34966a189c0eb6db38 : SUCCESS in 1h 39m 32s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2c3d79a2b0e94a5087cbb9e7e7f7f346 : SUCCESS in 46m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/66f2271b57984f07954501016fff86dd : SUCCESS in 54m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/04b15a4222f443c38191fd2094d0f666 : SUCCESS in 2h 02m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c90b0c4b7dbb46bdaa3d638a39e9782f : SUCCESS in 52m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9df4b6342ac0401c8946378665eff87d : FAILURE in 17m 35s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0c3b0ad2a0f947ad8cd4fed56acb0bfc : FAILURE in 16m 12s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e59753b687cb4b95b42d486d0a862e6a : POST_FAILURE in 25m 25s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/125f739d20254bc3ab81651cedf99e43 : SUCCESS in 1h 48m 22s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1d0d265fbb40496a9c48b0b0206318a3 : SUCCESS in 2h 00m 43s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b576c376c1c54a4484066c8393dfc757 : FAILURE in 1h 00m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ac927867d349416cb133406a6ffb2113 : SUCCESS in 1h 12m 56s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/eb88f61c15194217bb5d04d55ddc716b : FAILURE in 2h 24m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f262a9e53b355efe8da7a0f80addd1e37c1777d","author":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"date":"2020-06-19 22:38:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"858a5b9f71fe2091afa3b9c63eba2e130da1c32f","author":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"date":"2020-06-19 22:41:41.000000000","message":"Patch Set 1:\n\n(8 comments)\n\n\u003e (9 comments)\n \u003e \n \u003e Good catch!\n\nThanks. Sorry for missing those other values, I actually completely forgot to check other methods, hahah!","accounts_in_message":[],"_revision_number":1},{"id":"20b581562cb9a07f0e9d8f0c9869478069586965","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-20 00:17:19.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/44/737044/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"7a5dd21e5a147d0d9286440d680db7dc175d3182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 01:28:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8d850427a69e46d2b9b4968689d834c5 : SUCCESS in 1h 02m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ac2022bbebff480299c20b2fcfc0efd8 : SUCCESS in 38m 28s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/29dccbc4837143fb8de02b2ecd3df7c6 : SUCCESS in 1h 17m 46s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b712478e5cf940de95846ac453f4339d : SUCCESS in 1h 15m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7abb7af999684ae39b3f73f84886bb37 : SUCCESS in 59m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1ed2a1c17fb4e2d8b3768bef2aea82b : SUCCESS in 37m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fdc58a60c31345449e8d444b91d6e7f6 : SUCCESS in 35m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fc356b038bc4465ad516b7cc122dffe : SUCCESS in 14m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a425870e23e742bfacd850292fcccc7f : SUCCESS in 24m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7df7356c160f417ebc506350014df295 : SUCCESS in 27m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6b25e4dbf164150b82243888608a4a0 : SUCCESS in 13m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d2eb49f22650489b9a4645d2617c4337 : FAILURE in 58m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5b8c4793d07c4d498bc8134ae6e0bb68 : FAILURE in 59m 20s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f3bc1ff2652b46e384b036a95f221fe3 : SUCCESS in 1h 16m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2da922ef3faf4ae29ca9bf9f0c4ce63a : SUCCESS in 1h 35m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/05539697507848a4a95889e488a936c3 : SUCCESS in 1h 47m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f7261e8011514dffa6ec782d5266a745 : FAILURE in 1h 15m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8b81f2fb7bf5417eb8fd3d306c8db38b : SUCCESS in 1h 12m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4b080be66b8045ca90709896172fe7c2 : SUCCESS in 1h 38m 52s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c725e996794409b9fdd8590f884823a : SUCCESS in 1h 08m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/43c1e4afe40e49f5914283d083e44d4d : SUCCESS in 1h 19m 06s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2b161beef936496ba076323f4a6c9170 : SUCCESS in 1h 37m 59s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7844f8cfe91e43b6a9b69415ca56cbc4 : SUCCESS in 2h 03m 59s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/915fbe405bef4031a043bd0934397aa0 : SUCCESS in 1h 29m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/44219f8cc62949e0bd6c7f5c4e5ca742 : SUCCESS in 56m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fb044955c06b4652824ae6e33fea6524 : SUCCESS in 2h 14m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b77e0f82d4af4a4a928fda8282d74fa3 : SUCCESS in 59m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fce3bd2300ce45718ed605b74a10bd50 : SUCCESS in 1h 53m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bd1bcbd02cbf439a86b8ad800ac7ff9a : SUCCESS in 1h 59m 19s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6b74739fc777421da115e6195820f0a6 : POST_FAILURE in 36m 27s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5bb050448311457b9a2195d03aea5a42 : SUCCESS in 1h 47m 37s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0d2213579dc4457dba0a917b40fd6a13 : SUCCESS in 2h 10m 53s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/bb21b88344b54bb7b7753bb57e13d6ff : SUCCESS in 2h 34m 56s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/196a45d1340341ee95f42e641e2699a4 : SUCCESS in 1h 43m 50s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c35044cbe9e44449af724db1d853f1c5 : SUCCESS in 2h 20m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fa2b124961a5346d4a5a3bc27f1c1019b7919227","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-20 08:31:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/737044/2 : SUCCESS in 1h 50m 40s","accounts_in_message":[],"_revision_number":2},{"id":"e6e483da772a3922eb4562173737ace07189ddec","author":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"date":"2020-06-20 13:38:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0f2ad50ff0bc91124b272783f0b96e5b12d1d9a0","author":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"date":"2020-06-20 13:41:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"607c49f023d43ddd749de046170a2f01c90224a3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-20 15:24:30.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/44/737044/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"11645208827e67df43b2ba747830d1e6796198ba","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-20 15:34:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/737044/4 : SUCCESS in 1h 52m 53s","accounts_in_message":[],"_revision_number":4},{"id":"51510abf8346032cfbcb30df00baa5df9011e443","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-20 15:38:40.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/44/737044/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"9709ecf1495078364a58fe9c2a43fd79fc1f4590","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 16:58:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4a5c704ae99c426c8c642e19388f6314 : SUCCESS in 50m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b935b8cb0f6b45288b7aa4129f96f698 : SUCCESS in 34m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/571f3c83aa50498d89f9938d665cd3b3 : SUCCESS in 1h 08m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1c63ce97c4f24c67a913560d342b3129 : SUCCESS in 1h 04m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/121ec88a3e8a4663adc79be15527502f : SUCCESS in 1h 00m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0b0032b438c484295e9035bffafef57 : SUCCESS in 33m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed1b3e8089fe4123af69699b25c2be54 : SUCCESS in 27m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86a6eb6c86a04455ab191cb33eff311f : SUCCESS in 10m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88f18a1df7c84461a42e4348c49d083f : SUCCESS in 25m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1582b101ffa443d0856caffdc90b68df : SUCCESS in 28m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af1b5fedfb264149b92b75bbd23d6bc9 : SUCCESS in 8m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ffecccd71ba04ac8ab796f7aa3fa3c36 : SUCCESS in 52m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/030abe7ebe2649ec96f8af7693690845 : SUCCESS in 51m 16s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ed7c9e5df02d45fc83d3711f35843b82 : SUCCESS in 1h 12m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/875deb37d04343bda28fb1129dea6701 : SUCCESS in 1h 10m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b8dc9b17acc74e9b8d4957dff9f07108 : SUCCESS in 1h 54m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb96c4b430734eebb443e1330f98f393 : SUCCESS in 1h 08m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/46a23e305d63476e9e9c9917f0fbb993 : SUCCESS in 1h 09m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1ab2547d974f4a1eb1483c949369ab43 : SUCCESS in 1h 10m 37s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6d9877b0d9b4a7e990754f4b8c934d7 : SUCCESS in 58m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/69f8270cb6e942e085185af7ba894035 : SUCCESS in 45m 31s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4f03a41ac95c455691d6a9c0db753a37 : SUCCESS in 1h 39m 39s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bc49f802c14240ac82e6fe2e47aa49d1 : FAILURE in 1h 32m 02s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e42e51ff422542b091584446cf02faa5 : TIMED_OUT in 3h 03m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/41f31a53d88c40c080ff240b786c6321 : SUCCESS in 35m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f1828297b359482ca67161adcf1d07cc : SUCCESS in 1h 57m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7768cb99383641fa8903aa174b7eabab : SUCCESS in 1h 04m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/29e1026e41e945528456144262b59845 : SUCCESS in 1h 41m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8b212e00ca7d4fa7a69033d548541953 : SUCCESS in 2h 07m 48s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4dfc2783a7894bb3b86fea711d86d1a0 : POST_FAILURE in 30m 24s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fb89e60bf97145d492b8113d4f510f5d : SUCCESS in 1h 14m 14s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f7c1b726e05344919f17e955d5b4452d : SUCCESS in 1h 46m 18s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fd91dfa7c75d4e4497dfc787927232fa : SUCCESS in 2h 25m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/dab57f18688a4e68a811e3c7f1c8da8e : SUCCESS in 2h 00m 56s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/15a1dcf20d52428c869343068f2fa4bc : SUCCESS in 1h 54m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"50f04e2f132f4405181e068995dc15f74748a09a","author":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"date":"2020-06-20 19:38:30.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"f88ec9a767ea237a5542a1d1bb3259c0f724fb7e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-20 21:03:16.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/44/737044/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"f33c5c69e9823aefb56582490cc3e34e4ddd3aea","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-20 21:24:31.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/737044/5 : SUCCESS in 1h 45m 17s","accounts_in_message":[],"_revision_number":5},{"id":"b0d257437967971d3c3daeabf5c782536d18ba41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 22:18:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a8cb2b0292b64db085cc4b06182ca5e6 : SUCCESS in 55m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c7a75dc616254db0b8beaef5f54ad66b : SUCCESS in 35m 25s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3aaa25e7f24e42b1a8a60c9c65db19d4 : SUCCESS in 1h 11m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f034458735d24e3a96ce134046ea1035 : SUCCESS in 1h 07m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5b15a3ffaa714000ac312aa15bc67042 : SUCCESS in 37m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79ba111cf400475a90b0066712f084d6 : SUCCESS in 33m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e2d64c97a9c48f1ad29ca8f48a71a9f : SUCCESS in 27m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/521851980145403784aec2843a77a1d2 : SUCCESS in 10m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/064cbeaa2c8342f99dc0bd5b5311be35 : SUCCESS in 25m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5166dac628734561a9b3c6d1acecce09 : SUCCESS in 26m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7919172d68bf4f8cb39fb1f4e48f5ea6 : SUCCESS in 7m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0180c0da9a844150a84cb29d658155c8 : SUCCESS in 50m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/28f997f3f48a4c58a081f99ee8836e14 : SUCCESS in 40m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a4167b61ab12454eace71910ea8e093b : SUCCESS in 1h 10m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cf69f59da4db492d82ad5e0640e62483 : SUCCESS in 1h 09m 21s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/25889da2600d488ea42705c3f33cd39c : SUCCESS in 1h 30m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1333145fe6a74a66b2096572bcba95d3 : SUCCESS in 1h 08m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/be9183d1cd8947b683aee7bed31715e1 : SUCCESS in 1h 06m 27s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d46564ed53864584a2d7335d9b5dcef8 : SUCCESS in 1h 10m 44s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/80ed2562ebb94644a043055152968482 : SUCCESS in 1h 10m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/93c83c8dbc844567b379193235d2bff3 : SUCCESS in 1h 05m 01s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2744c48e6e6b47a0ab0f1c46e95c5b63 : SUCCESS in 1h 43m 23s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/67e3f4a4abd0474abbd1b8b297eee9cd : SUCCESS in 1h 34m 35s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/876b6971c84f4f5eb39b438a5a2e04c2 : SUCCESS in 1h 11m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/357e2cabdbcf4286a557333b586bd86b : SUCCESS in 56m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a682a5fcdd454c02804bfc3fd0dd69c3 : SUCCESS in 2h 02m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0af2dc5cddd24242a419d369e8064c31 : SUCCESS in 59m 51s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/006b3760c5eb4dd18c5b1c49ee7c7e31 : SUCCESS in 1h 08m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/22ab2824f6b3453eb0fe45190221a3db : SUCCESS in 1h 42m 16s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/45d8dff29d584472b146d2df7c1944d3 : POST_FAILURE in 17m 47s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3daf108a9d964bcaa901a8ff40bc0387 : SUCCESS in 1h 45m 16s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dacd45802b414a3e95d05c3c651ed12b : SUCCESS in 1h 44m 27s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a5ecb5e642dc417a8a56feabe259d04d : SUCCESS in 2h 34m 57s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0aa068cdd9fa45d993cb7a03d2938f0c : SUCCESS in 1h 13m 11s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/10e4eae3102d4d87a7f4d7988cf195e6 : SUCCESS in 1h 25m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f140d208f055cd2ff70f0e8c172de5afc49d7d26","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-22 20:30:30.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f69369eb9faab7e45efcd92df249d175009b0a48","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-06-23 22:56:00.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"51e517bc1923e9e06d4f9d171a45390698fc9cec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 22:56:43.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d3b2e45ae23a8207e74b809479f2c2cb5ad5f1f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 03:16:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"a66e87d55d62b9898892fc703442577c214d5dcb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 03:16:15.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fa23a758ac89492098e6e1125e5c9c85 : SUCCESS in 50m 26s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3bee4692ea7943c3bd30c39b94c7de5b : SUCCESS in 1h 36m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f7312e20b0e24f6aa39c00cea6d79b03 : SUCCESS in 1h 20m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/61f1f88b79734a95b55ce06d6ecab609 : SUCCESS in 1h 05m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/375a82f03e3941629f326675511573e6 : SUCCESS in 34m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42b3656eefde4fcab2016b802cc437a5 : SUCCESS in 11m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22f13d0de4f14253815102e6e9f494a4 : SUCCESS in 29m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/386604893fda4d16ac7e8ef1ccda21b3 : SUCCESS in 33m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff13d343e6e94fb38450e3b8906a57bf : SUCCESS in 9m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c78d7464555a47acb7ab9afa48d86c6f : SUCCESS in 58m 47s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6e3ab118779f43f19cbe2d16b55ed57d : SUCCESS in 1h 18m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6dd4f034b35644228c75e630ed047add : SUCCESS in 1h 14m 26s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/18cdeb2f5a104daca06f474c36c7b8ed : SUCCESS in 1h 22m 32s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/22870cf3cca847ea8098698ef2bc8049 : SUCCESS in 1h 10m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/126a272c507f471b8e289775f51ac920 : SUCCESS in 1h 18m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f7b08b95281c4b21a3d9c10f1cd42add : SUCCESS in 1h 13m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/27a5e6b207ef402e9627d3c220ad9a79 : SUCCESS in 2h 14m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d4db072c4b04d8b9382b112aeb19339 : SUCCESS in 58m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/56954c192aec44a9b433acf25b3254d9 : SUCCESS in 2h 14m 58s","accounts_in_message":[],"_revision_number":5},{"id":"14853512138b5d7d68ea06b4a4eab22757c735f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 03:18:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09aac44f91de4568872114d3f9acec81 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"94d6e38fa08f8a52b6f18077c2a05a05b89b42b5","revisions":{"ecc819deff1f3cb1956e02b11e541dd9e028b0a2":{"kind":"REWORK","_number":1,"created":"2020-06-19 16:03:43.000000000","uploader":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"ref":"refs/changes/44/737044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/737044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/737044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/737044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/737044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/737044/1"}}},"commit":{"parents":[{"commit":"78e16c77428e071ac14fbd3727840cd6348d1e7a","subject":"Merge \"Refactor assertEqual statement that includes \u0027True\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78e16c77428e071ac14fbd3727840cd6348d1e7a"}]}],"author":{"name":"Weronika Sikora","email":"weronika.sikora@corp.ovh.com","date":"2020-06-19 15:16:29.000000000","tz":120},"committer":{"name":"Weronika Sikora","email":"weronika.sikora@corp.ovh.com","date":"2020-06-19 16:03:33.000000000","tz":120},"subject":"Fix the wrong value for QoS rate conversion to bytes/s","message":"Fix the wrong value for QoS rate conversion to bytes/s\n\nThe rate value is converted to bytes per second before being\nsent to Pyroute2, but it used the wrong value for the calculations.\nThis resulted in incorrect rates.\nIt should be multiplied by 1000 (kbit), not 1024 (Kibit).\nThe same applies to the burst value.\n\nChange-Id: I70cb1fe651a50b2f6495d7a365a6beb2ba111c6d\nCloses-Bug: #1884273\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ecc819deff1f3cb1956e02b11e541dd9e028b0a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ecc819deff1f3cb1956e02b11e541dd9e028b0a2"}]},"branch":"refs/heads/master"},"e2625d389d026915a4576a84d1da2a2b669ec938":{"kind":"REWORK","_number":2,"created":"2020-06-19 22:38:54.000000000","uploader":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"ref":"refs/changes/44/737044/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/737044/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/737044/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/737044/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/737044/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/737044/2"}}},"commit":{"parents":[{"commit":"78e16c77428e071ac14fbd3727840cd6348d1e7a","subject":"Merge \"Refactor assertEqual statement that includes \u0027True\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78e16c77428e071ac14fbd3727840cd6348d1e7a"}]}],"author":{"name":"Weronika Sikora","email":"weronika.sikora@corp.ovh.com","date":"2020-06-19 15:16:29.000000000","tz":120},"committer":{"name":"Weronika Sikora","email":"weronika.sikora@corp.ovh.com","date":"2020-06-19 22:36:52.000000000","tz":120},"subject":"Fix the wrong value for QoS rate conversion to bytes/s","message":"Fix the wrong value for QoS rate conversion to bytes/s\n\nThe rate value is converted to bytes per second before being\nsent to Pyroute2, but it used the wrong value for the calculations.\nThis resulted in incorrect rates.\nIt should be multiplied by 1000 (kbit), not 1024 (Kibit).\nThe same applies to the burst value (kb).\n\nChange-Id: I70cb1fe651a50b2f6495d7a365a6beb2ba111c6d\nCloses-Bug: #1884273\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2625d389d026915a4576a84d1da2a2b669ec938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2625d389d026915a4576a84d1da2a2b669ec938"}]},"branch":"refs/heads/master"},"a38f340e50d39b21d41dc982cc7821e10c60f80c":{"kind":"REWORK","_number":3,"created":"2020-06-20 13:38:21.000000000","uploader":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"ref":"refs/changes/44/737044/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/737044/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/737044/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/737044/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/737044/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/737044/3"}}},"commit":{"parents":[{"commit":"78e16c77428e071ac14fbd3727840cd6348d1e7a","subject":"Merge \"Refactor assertEqual statement that includes \u0027True\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78e16c77428e071ac14fbd3727840cd6348d1e7a"}]}],"author":{"name":"Weronika Sikora","email":"weronika.sikora@corp.ovh.com","date":"2020-06-19 15:16:29.000000000","tz":120},"committer":{"name":"Weronika Sikora","email":"weronika.sikora@corp.ovh.com","date":"2020-06-20 13:37:30.000000000","tz":120},"subject":"Fix the wrong value for QoS rate conversion to bytes/s","message":"Fix the wrong value for QoS rate conversion to bytes/s\n\nThe rate value is converted to bytes per second before being\nsent to Pyroute2, but it used the wrong value for the calculations.\nThis resulted in incorrect rates.\nIt should be multiplied by 1000 (kbit), not 1024 (Kibit).\nThe same applies to the burst value (kb).\n\nChange-Id: I70cb1fe651a50b2f6495d7a365a6beb2ba111c6d\nCloses-Bug: #1884273\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a38f340e50d39b21d41dc982cc7821e10c60f80c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a38f340e50d39b21d41dc982cc7821e10c60f80c"}]},"branch":"refs/heads/master"},"ffa80555e5876519cb47c1c5a56ef897162fe177":{"kind":"REWORK","_number":4,"created":"2020-06-20 13:41:01.000000000","uploader":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"ref":"refs/changes/44/737044/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/737044/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/737044/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/737044/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/737044/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/737044/4"}}},"commit":{"parents":[{"commit":"78e16c77428e071ac14fbd3727840cd6348d1e7a","subject":"Merge \"Refactor assertEqual statement that includes \u0027True\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78e16c77428e071ac14fbd3727840cd6348d1e7a"}]}],"author":{"name":"Weronika Sikora","email":"weronika.sikora@corp.ovh.com","date":"2020-06-19 15:16:29.000000000","tz":120},"committer":{"name":"Weronika Sikora","email":"weronika.sikora@corp.ovh.com","date":"2020-06-20 13:40:40.000000000","tz":120},"subject":"Fix the wrong value for QoS rate conversion to bytes/s","message":"Fix the wrong value for QoS rate conversion to bytes/s\n\nThe rate value is converted to bytes per second before being\nsent to Pyroute2, but it used the wrong value for the calculations.\nThis resulted in incorrect rates.\nIt should be multiplied by 1000 (kbit), not 1024 (Kibit).\nThe same applies to the burst value (kb).\n\nChange-Id: I70cb1fe651a50b2f6495d7a365a6beb2ba111c6d\nCloses-Bug: #1884273\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ffa80555e5876519cb47c1c5a56ef897162fe177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ffa80555e5876519cb47c1c5a56ef897162fe177"}]},"branch":"refs/heads/master"},"94d6e38fa08f8a52b6f18077c2a05a05b89b42b5":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-06-20 19:38:30.000000000","uploader":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"ref":"refs/changes/44/737044/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/737044/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/737044/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/737044/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/737044/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/737044/5"}}},"commit":{"parents":[{"commit":"0580d03a2b9f133024deceff4fc5c136a86d01b9","subject":"Merge \"Workaround for TCP checksum issue with ovs-dpdk and  veth pair\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0580d03a2b9f133024deceff4fc5c136a86d01b9"}]}],"author":{"name":"Weronika Sikora","email":"weronika.sikora@corp.ovh.com","date":"2020-06-19 15:16:29.000000000","tz":120},"committer":{"name":"Weronika Sikora","email":"weronika.sikora@corp.ovh.com","date":"2020-06-20 19:38:30.000000000","tz":0},"subject":"Fix the wrong value for QoS rate conversion to bytes/s","message":"Fix the wrong value for QoS rate conversion to bytes/s\n\nThe rate value is converted to bytes per second before being\nsent to Pyroute2, but it used the wrong value for the calculations.\nThis resulted in incorrect rates.\nIt should be multiplied by 1000 (kbit), not 1024 (Kibit).\nThe same applies to the burst value (kb).\n\nChange-Id: I70cb1fe651a50b2f6495d7a365a6beb2ba111c6d\nCloses-Bug: #1884273\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94d6e38fa08f8a52b6f18077c2a05a05b89b42b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94d6e38fa08f8a52b6f18077c2a05a05b89b42b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
