)]}'
{"id":"openstack%2Fneutron~737097","triplet_id":"openstack%2Fneutron~master~I76562302119c6727a78bc72c4cf5346b3b8befe6","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"I76562302119c6727a78bc72c4cf5346b3b8befe6","subject":"[OVN] Updates to tools/migrate_names.txt","status":"MERGED","created":"2020-06-19 22:22:32.000000000","updated":"2020-06-24 02:48:09.000000000","submitted":"2020-06-24 02:46:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737097-1592966762939-cf4bca08","meta_rev_id":"afac18d596f596760e5ebc830109f623e135e926","_number":737097,"virtual_id_number":737097,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-23 17:09:11.000000000","_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-24 02:46:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-23 19:01:10.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2020-06-23 08:33:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-06-22 16:40:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":1,"date":"2020-06-23 08:33:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-22 16:40:45.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-23 08:33:46.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-06-23 19:01:10.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-06-24 02:46:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6170eaa5156cf2dd8702b61ef640afe2c1b0c04b","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-19 22:22:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b0a8d31f1b85f1495ef098338be3cf93b994806","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 00:22:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b500d4e64cf43a2bf09c3baa78660a1 : SUCCESS in 34m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8522a3d3ed2640779cf7a306d25e5b04 : SUCCESS in 25m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f60650ff5cc343e99d7c3337ec4c6616 : SUCCESS in 11m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/498a1373185d493ab52cd37491713a2f : SUCCESS in 20m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f3c5c97a41c4be681eb265bc5cdbfc3 : SUCCESS in 30m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f42d228956f4491998d3ab63af3cffc4 : SUCCESS in 10m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f6a1ed243d31417b9eb037dc091748d6 : SUCCESS in 52m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/06537411ded6456e8ad7b12e889d2574 : SUCCESS in 1h 04m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/239a6727a009416e83a0efd65091a72c : FAILURE in 1h 18m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a7587c7b95da47dc914a26e378891628 : SUCCESS in 1h 58m 31s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/19a649529c2d45bc99831ac3a82d2e87 : POST_FAILURE in 25m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"46cde2cffbcff1c3e97ffa9d79291159945fa12e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-20 08:14:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/737097/1 : SUCCESS in 1h 47m 12s","accounts_in_message":[],"_revision_number":1},{"id":"74621eaeb4e07459f910b7ce2ab1732c90733847","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-20 19:41:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2fb751eb23c44b49462bef33d5aae2561e11de5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 20:56:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b67abe00babd48e787f23b5f90e6cea6 : SUCCESS in 34m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3a46de936e742ff8d52f158d9b979b2 : SUCCESS in 27m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d624e01d51c74c65930b76c1a2f40d6f : SUCCESS in 10m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/914c7bf26f794a9ea54195d6c540a533 : SUCCESS in 27m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e7e56ec49b34543a07c3e0f805e868d : SUCCESS in 24m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/418339923a094c4599017f8caf593670 : SUCCESS in 9m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fa0f7c62b61849a2aa887daee1e45fb2 : SUCCESS in 53m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/53492ea5f22c4724b2fd32213dca6d64 : SUCCESS in 53m 13s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/32c09a4417d84948abdea61e2d9b6e75 : SUCCESS in 1h 06m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d637ba6c75c143399ee3da5fbb98ce29 : SUCCESS in 1h 09m 51s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/88c2bb3415c04efeb094e7a0e8318759 : POST_FAILURE in 25m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e7421fa4a682246da6a99919a4cba06442abf2ac","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-20 21:24:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/737097/2 : SUCCESS in 1h 41m 59s","accounts_in_message":[],"_revision_number":2},{"id":"394de5d4305e9aad5d64ed98b73cb070cbb20eac","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-22 16:40:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ad88ed44170684f72fb8e870e453929c6ba5e719","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-23 08:33:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"34655660fa964d380f773422cc9cb52214637399","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 08:34:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"acd48ec0f37ce2105cf61d92d47b7701a610aeae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 13:57:32.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/642cdc07697947be876207ff76ce9782 : POST_FAILURE in 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c4f8d31c8504cd1bd0d27c8ef8eed68 : SUCCESS in 12m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a0a77543fa4402b8d4c3d45f2e8899e : POST_FAILURE in 31m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ddd6726bbb8431eabf2228615f87134 : SUCCESS in 25m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb08a5d57e1747ae9008e471d39f0332 : SUCCESS in 12m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/19e6bbef2ed3463c89c1e91179467e20 : SUCCESS in 50m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a049f8d9759d44cf82a982ea92710d02 : SUCCESS in 1h 32m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b55cf8bea20740ab8abf384dc1c45a95 : SUCCESS in 1h 24m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d788263d60927cf230429caba0dfe81d1d1f01f2","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-23 17:09:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"02f775b8711e71f4cc58cc2391a6675ae8aacfd5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-23 19:01:10.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/737097/2 : FAILURE in 1h 51m 31s","accounts_in_message":[],"_revision_number":2},{"id":"70352bfa7ac48f8c0d251375fc251b0714f2658b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 22:26:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/613f02edb47c410eb44e7487a18ff47c : SUCCESS in 34m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/982f4a168f9643ddad566266c90c37ab : SUCCESS in 25m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7562353c240e4305877b1c38b503f910 : SUCCESS in 11m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb1b0072ac5e4683a83f11a7c4261517 : SUCCESS in 25m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0735975dfd2149819e5d15c6a7c26737 : SUCCESS in 27m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a5f85a1c6844508b552865a87dbbbb9 : SUCCESS in 7m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0b8c3e6ee7534a88a49bda05ddc16bdd : SUCCESS in 58m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/664387bf4e1c4f368aed9969bfbbc81a : SUCCESS in 1h 26m 41s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/dd83e3281095447c98bb5dffbb1d518c : SUCCESS in 1h 16m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c13141b9daa474fb482073bba5a939d : SUCCESS in 1h 09m 34s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/108c345c8c4c407bb445da9c3643c209 : POST_FAILURE in 27m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b97d0b92a586e90f246ff58db44b2a99ba13e568","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 22:26:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a186b3a5b1afc336e049fb4aeaf35265b654424b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 02:46:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1e07234d49145a2aeacb13e643c04d5 : SUCCESS in 27m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01bf212f985649b1826ba2d96ee3e566 : SUCCESS in 11m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9331582d8f60431e8f744bb040d6ea31 : SUCCESS in 34m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/450ea92363e04203a1db0aaea0c5fb9e : SUCCESS in 20m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b712ca3332de4e8da8e85ed03653a034 : SUCCESS in 10m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0e4d81dac5d4447f96efdc30fcc06971 : SUCCESS in 57m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/490d5831b00d4017a80dd87eae2cd9ce : SUCCESS in 1h 40m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fbdc9d0878fb40749f0be367e92fa8fd : SUCCESS in 1h 15m 48s","accounts_in_message":[],"_revision_number":2},{"id":"f6884a304e011457ce3cfac065da7f43e77f8387","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 02:46:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"afac18d596f596760e5ebc830109f623e135e926","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 02:48:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/965b4185f8e34d028117ff0c59c0e3f0 : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e3e1d994d8301192eb07980aec125501a1d76b62","revisions":{"b0ea0dddd1e18431f6c7d74bdb45d240ebf64eaa":{"kind":"REWORK","_number":1,"created":"2020-06-19 22:22:32.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/97/737097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/737097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/737097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/737097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/737097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/737097/1"}}},"commit":{"parents":[{"commit":"78e16c77428e071ac14fbd3727840cd6348d1e7a","subject":"Merge \"Refactor assertEqual statement that includes \u0027True\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78e16c77428e071ac14fbd3727840cd6348d1e7a"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-19 22:20:28.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-19 22:20:46.000000000","tz":-240},"subject":"[OVN] Updates to tools/migrate_names.txt","message":"[OVN] Updates to tools/migrate_names.txt\n\nUpdates to tools/migrate_names.txt to correctly represent\nmapping between neutron and networking-ovn.\n\nChange-Id: I76562302119c6727a78bc72c4cf5346b3b8befe6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0ea0dddd1e18431f6c7d74bdb45d240ebf64eaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0ea0dddd1e18431f6c7d74bdb45d240ebf64eaa"}]},"branch":"refs/heads/master"},"e3e1d994d8301192eb07980aec125501a1d76b62":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-20 19:41:54.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/97/737097/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/737097/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/737097/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/737097/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/737097/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/737097/2"}}},"commit":{"parents":[{"commit":"0580d03a2b9f133024deceff4fc5c136a86d01b9","subject":"Merge \"Workaround for TCP checksum issue with ovs-dpdk and  veth pair\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0580d03a2b9f133024deceff4fc5c136a86d01b9"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-19 22:20:28.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-06-20 19:41:54.000000000","tz":0},"subject":"[OVN] Updates to tools/migrate_names.txt","message":"[OVN] Updates to tools/migrate_names.txt\n\nUpdates to tools/migrate_names.txt to correctly represent\nmapping between neutron and networking-ovn.\n\nChange-Id: I76562302119c6727a78bc72c4cf5346b3b8befe6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e3e1d994d8301192eb07980aec125501a1d76b62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e3e1d994d8301192eb07980aec125501a1d76b62"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
