)]}'
{"id":"openstack%2Fneutron~737313","triplet_id":"openstack%2Fneutron~stable%2Ftrain~I9b4161946daa5863ddab94a57ad282e82bcf6e5c","project":"openstack/neutron","branch":"stable/train","topic":"bug/1884256-stable/train","hashtags":[],"change_id":"I9b4161946daa5863ddab94a57ad282e82bcf6e5c","subject":"Add ensure-tox in functional job","status":"MERGED","created":"2020-06-22 16:32:21.000000000","updated":"2020-06-23 01:57:03.000000000","submitted":"2020-06-23 01:55:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737313-1592877310916-3f73efbb","meta_rev_id":"7310492c2b8908dccadb7be5bb3307e23acca831","_number":737313,"virtual_id_number":737313,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-23 01:55:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-22 19:10:46.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-06-22 21:01:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-06-22 21:07:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-06-22 21:07:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-22 19:10:46.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-06-22 21:01:01.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-06-22 21:07:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-23 01:55:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07f8be9d636ce5e854220a35ca498ac700a5e8ce","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-22 16:32:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0916b2bbddb11fa7c4c04676348b4d7669f5e4d0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-22 19:10:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/737313/1 : SUCCESS in 1h 48m 57s","accounts_in_message":[],"_revision_number":1},{"id":"54c42caf35f2fc556904f5db0433fe5b6b6dad69","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-06-22 21:01:01.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good (and ussuri was done in squashed gate fix)","accounts_in_message":[],"_revision_number":1},{"id":"b762e8f8a70a808a88f98c6039ae399aa6cc882a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 21:04:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/4c6f5765722c44a598b3aeaba4dafe6c : SUCCESS in 48m 47s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/d738cd518f864fc4b3ea83c4edb582c2 : SUCCESS in 46m 51s\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/660fe37173d743f09169093e44d98562 : SUCCESS in 48m 00s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/40ecd6271e374de7a2943c494b40a40c : SUCCESS in 48m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/31c027a081ee416e8a0d98eadffb270f : SUCCESS in 1h 15m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d674c87e302f42c099ba65b95a4bd9db : SUCCESS in 29m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d20a49b07c204563b6ea9a50999cc554 : SUCCESS in 22m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc1519b7ce1d47358ec849bdd25c2fb8 : SUCCESS in 13m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d04a29f46bc1415b9a934f62c46b027c : SUCCESS in 25m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14682f0450a44383a8eeb3f256f2b932 : SUCCESS in 22m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30a89bc76a12403dbc60293e46bbb8ee : SUCCESS in 25m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42cd6564fddf44229fcddfe6659270d9 : SUCCESS in 10m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/8e206096e756438b97cd4fa10fd3ae15 : SUCCESS in 59m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4ad43984541c40c8a84017a6f5d253f0 : SUCCESS in 1h 14m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0f3e740cd81b498e847e0bcfce2000e2 : SUCCESS in 46m 22s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/e79057077d5c46ca87845cf80ed673cc : SUCCESS in 48m 19s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/983cc8985fa04161aa1bed8b9cf1df6e : SUCCESS in 1h 05m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/aba8ab7d56ce49888357bc396cf443ba : SUCCESS in 1h 51m 31s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d79a529fca6446fe880f9aeab35896d1 : SUCCESS in 1h 10m 12s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/081504c466824a25b3c20c021e6ed235 : SUCCESS in 1h 11m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/abedef668943400c8374319204dada5b : SUCCESS in 1h 17m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4bd6e73ba72f4f1792e56a59c2eecc25 : SUCCESS in 1h 11m 03s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5950219ee907466d848297d9f699f0b8 : SUCCESS in 1h 20m 34s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/06165a7163b6423591975330d543158d : FAILURE in 1h 30m 53s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/186d964e46ff42459dc132bcb0c80d18 : SUCCESS in 50m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/924948f5eb764f13b0bce99a6fe4b08f : SUCCESS in 59m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/eee3a81883ba4c1fba58b1b2cf722924 : SUCCESS in 56m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/70d6d0a6c0d94bd7a1a45298cc300632 : SUCCESS in 2h 06m 53s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a0991a334a5246febf270aba2b9ea09d : SUCCESS in 1h 15m 58s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/2e5eebc5d662403fa3380b3bdcd391e1 : FAILURE in 17m 17s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f6a05f83bd654047aa950483a48c69c5 : SUCCESS in 1h 19m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bda124ca20d453b9d5acb4b258b66a0 : SUCCESS in 1h 02m 51s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/068029af0d46484fadeb0df05d2263e6 : POST_FAILURE in 1h 11m 59s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/849e6dbdeb1d4e59b75587c0f4eded53 : SUCCESS in 37m 01s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f63b7e5921c14ca0b3a6baa74ae9de0e : SUCCESS in 56m 03s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/22513120055d415cb71b704577996202 : SUCCESS in 1h 11m 49s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f11c22de0e764bb299d6282a43e26483 : FAILURE in 1h 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba42b681a5dafab963d8ad0e4c283e7e939e12db","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-22 21:07:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"78c56afeaf7d44bfbfe17c03d6adf81ecd2d614c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 21:07:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5b0f66248615907f3a2b04590386ad8cde3583d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 01:55:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/66dcc0db8b9d4774bc1a5e65c1bb2931 : SUCCESS in 52m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e7e9e7ae41948a3908c2538a138f690 : SUCCESS in 24m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67bd30eb4ffd402184314641760cb355 : SUCCESS in 11m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4016fc6b2cfe4a2a8c1b36856133441a : SUCCESS in 24m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8476f63df534b3d955567f2f0e13724 : SUCCESS in 26m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/922be1233a114da989dff16509183bc9 : SUCCESS in 25m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f64e2d3e8345d0b8c468a24fe816ac : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/ac514386086e4fe8bc33e5201537074e : SUCCESS in 1h 24m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c9b4c136f61b49e3af354f69b22b3c8c : SUCCESS in 1h 08m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2d31d308d9674e649c63264cb20a939c : SUCCESS in 53m 14s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/f7bc0888d7fb4c59a665ce6379e2445d : SUCCESS in 44m 38s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/92c99e4e817e46c6bebacd0ad3564ed1 : SUCCESS in 1h 15m 04s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7df06d1f001d4e86a95d6f0e844ceef4 : SUCCESS in 1h 12m 42s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/36a0f303338b4bf7a60362577ee7fba2 : SUCCESS in 1h 10m 46s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1a24da31686b479f9d0d0eabe4d777c4 : SUCCESS in 1h 16m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e9bd12960754b2382b26c865a58dba7 : SUCCESS in 1h 16m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/17ce20791270455a9614e7fe25d56fff : SUCCESS in 1h 14m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cd9dd1aedfaf462e820d6f8d94d3b2cb : SUCCESS in 58m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/da113913f2f04f8395377be5af794b9f : SUCCESS in 2h 02m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4d3530162e1d4324985387c18efeade3 : SUCCESS in 1h 01m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0487f43814424d8c9ed4b3cba25ce824 : SUCCESS in 54m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2a2cc1de25a210f72316b819ac0d7e927516a413","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 01:55:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9dc366ce0b0c655dd290c6e3717470fe028010db","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 01:57:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/486d186d3a9b4e908541dc35eb74188e : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e2cf239c8c1409a3d9a25452930f766894e3657","revisions":{"8e2cf239c8c1409a3d9a25452930f766894e3657":{"kind":"REWORK","_number":1,"created":"2020-06-22 16:32:21.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/13/737313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/737313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/737313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/737313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/737313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/737313/1"}}},"commit":{"parents":[{"commit":"c5e00e551c74490f073aaca14b18d82a17a95d64","subject":"Migrate neutron grenade multinode jobs to be native Zuul v3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5e00e551c74490f073aaca14b18d82a17a95d64"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-19 14:19:23.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-22 16:32:21.000000000","tz":0},"subject":"Add ensure-tox in functional job","message":"Add ensure-tox in functional job\n\nneutron functional jobs run tests using the tox env.\nEven tempest jobs need tox to be present to run the tempest\ntests in venv. But functional job derived from devstack-minimal\njobs does not make sure tox is present or not. devstack jobs does not\nmak sure that as those are not only used for tox run but also for\nother integration testing.\n\nCurrent neutorn functional job failure-\n- https://zuul.opendev.org/t/openstack/build/59865004855c404ab18f06fc0ec1d005\n\nlet\u0027s call ensure-tox role in job to make sure tox is present.\n\nChange-Id: I9b4161946daa5863ddab94a57ad282e82bcf6e5c\nCloses-Bug: 1884256\n(cherry picked from commit 86e90bfce2169194524cabf4fe6eff5e232c7b91)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e2cf239c8c1409a3d9a25452930f766894e3657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e2cf239c8c1409a3d9a25452930f766894e3657"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
