)]}'
{"id":"openstack%2Fneutron~737736","triplet_id":"openstack%2Fneutron~master~I0a89bca0734d3e735fb357e488f85589e81d709f","project":"openstack/neutron","branch":"master","topic":"bug/1884906","hashtags":[],"change_id":"I0a89bca0734d3e735fb357e488f85589e81d709f","subject":"L3 agent scheduler should return a valid index if manual scheduling","status":"MERGED","created":"2020-06-24 10:05:54.000000000","updated":"2020-06-27 12:28:36.000000000","submitted":"2020-06-27 12:25:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737736-1593260723619-5872017d","meta_rev_id":"f0a543c2c311e6e29d141e1ec43e67de4a9c707c","_number":737736,"virtual_id_number":737736,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-26 09:55:27.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-06-24 20:15:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-27 12:25:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-26 11:45:38.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-27 08:27:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-06-26 13:43:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2020-06-25 05:38:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-27 08:27:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 20:15:55.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-25 05:38:24.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-06-26 11:45:38.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-06-26 13:43:49.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-27 08:27:08.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-06-27 12:25:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c1e6f0e0cbcaeb02f95d46d762d71330630fe6e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-24 10:05:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f73a0d578fd0a42b734431ff200d8999dd42723","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-24 12:01:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/737736/1 : SUCCESS in 1h 54m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c13ef56a6f35cdeeb0648569e652867cba260173","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-24 12:10:17.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/36/737736/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a44e10f5c08140d4e67de7eaa05882575bcd6532","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-24 15:37:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8c274a1f5c2a784c7b07e0f991ef3b123acc1372","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-24 17:29:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/737736/2 : SUCCESS in 1h 50m 33s","accounts_in_message":[],"_revision_number":2},{"id":"bf1c0699180d1a7df346c055b175ba655a0aa28f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-24 20:15:55.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/36/737736/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"b176ecfc4c3066a4a3ba4fe25044927aa504e6ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 22:58:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ea67576ff7654f669c8b24790689a4cd : SUCCESS in 57m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/124e9a3282f1433fb31478bd7e8d24b4 : SUCCESS in 37m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5323e92d56bc4943af6dcb20f3bca5dd : FAILURE in 1h 27m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1c2ff0ffd7cb43d98147132ba33ca0a6 : SUCCESS in 1h 17m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/47ba757c20f9468ab95276c305206fa9 : SUCCESS in 58m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9e56899fe3da4349b32b2c3e719647ca : SUCCESS in 1h 35m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e38a3550dfd4f9abcb1956580aa5f49 : SUCCESS in 31m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c94c2e0d33634c8d90d09aa7e570cdd5 : SUCCESS in 30m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ea2f66273db43b8940f147471f71a68 : SUCCESS in 17m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eade39d3bfe34e7a83401749052570e6 : SUCCESS in 27m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f85fe07e0c94c5c9abd5dc9e8b774c5 : SUCCESS in 28m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b83c5eb57264769b6a54c695f080bd0 : SUCCESS in 16m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5aeec4c278704b48a0ff3729c2460ee6 : SUCCESS in 57m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7a75e2f7cbd14d9ebd719ac4e57e6fc3 : SUCCESS in 1h 02m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e0ffeb847beb457f8388c9bcc1e56e93 : SUCCESS in 1h 11m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fbbaab6e443a4126bde2b37013230d9b : SUCCESS in 1h 23m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ef659c47cfe8406089504bcd6c67bb46 : SUCCESS in 1h 50m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e5264ad315f1447da5027bd3a958df03 : SUCCESS in 1h 09m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fedbc08172f84a9e94acd46ad30679c3 : SUCCESS in 1h 17m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/37baa8c3a5ea4d45a6425256d056db35 : SUCCESS in 1h 19m 21s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b1a55cb715b41f5b4cd35772556ecd4 : SUCCESS in 1h 05m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9679c3e86040442eaa0392c6722caf1a : SUCCESS in 52m 41s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5d1f8beb564446f9ac67b0cd9aa797e8 : SUCCESS in 1h 59m 06s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/50f242116bf441438862fdcc3e23fea4 : FAILURE in 1h 32m 07s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c302d6f80dbc40bcafc0755b56024632 : SUCCESS in 1h 08m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/58a06538bedc47648b001a6cf174f571 : SUCCESS in 54m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5b98bf19e0fc40c5968a0821e62bb57f : SUCCESS in 2h 04m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24eab1b72dba4ef89b3fcfe96452e0aa : SUCCESS in 1h 11m 09s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/ebc49e3a94a84c0d9a8d1e73643f72b3 : SUCCESS in 1h 16m 29s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c1c1d1da999e46fea6131e4996bad00b : SUCCESS in 1h 42m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3b6878ca01134e87a5d83b6eed359209 : SUCCESS in 2h 05m 16s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9713c6b196ac46e985607163c8f32d14 : POST_FAILURE in 25m 39s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/12e549af538842a18986c6cea28e0ae8 : SUCCESS in 1h 47m 50s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/caec144810ef49949d6573d5bc6c2bca : SUCCESS in 1h 46m 13s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e43cf3a0189c4c4886348cd4922555b0 : SUCCESS in 2h 42m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2b87240e9e38499596ec83ebd31eff09 : SUCCESS in 1h 49m 28s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c7ec717f28014102a5ade1f78304e411 : SUCCESS in 2h 54m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4ddae9f4e9b9b53b7997489c3896096a51f91357","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-06-25 05:38:24.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"db08579c805ac0a0f773e8a2c5754bc934d56d2e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-25 07:21:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/737736/2 : SUCCESS in 1h 42m 44s","accounts_in_message":[],"_revision_number":2},{"id":"bf07bb44db5be473ee1557bb340a4350b56bc609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 09:01:53.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4abcdf0840954583a05c576bafc13a9c : SUCCESS in 55m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c7774fb074294e5bb1f239a9d1825d4b : SUCCESS in 30m 40s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/857e72b876d54107b2ea161bebbd4e7f : SUCCESS in 1h 14m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a22e76f882c84ea6a2c3758ddb364e68 : SUCCESS in 1h 07m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/555987f185774c258c375739929b4417 : SUCCESS in 1h 04m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3a53057cdea34cfd83bf3631e354a193 : SUCCESS in 57m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c95a23bfa364e5abf0b40da644c53fe : SUCCESS in 34m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b617a2ab2b64a798c1a99657ad209e1 : SUCCESS in 31m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79b16bceacfc458690e5ccd5f1c0d5ba : SUCCESS in 12m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13bebc7f882642ba95586d5f538c438d : SUCCESS in 31m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e19bc6c2a5c42c49616ebe9cee83948 : SUCCESS in 22m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a69b431b62c46fabe91ec46934464a6 : SUCCESS in 12m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cd16ae40be4a43b190a63d48b8f906f2 : SUCCESS in 55m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5a1b361dffa14d0d8393bf6df8fb6ed6 : SUCCESS in 59m 47s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/44c8b08972c443699f2fcf669ad5798f : SUCCESS in 1h 25m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff88560675ee4ca692db892b05fefaac : SUCCESS in 1h 23m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/679319e4eb364748817e00749bbf30be : SUCCESS in 2h 16m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e433c7e7145d4cb399dc0342a50527af : SUCCESS in 1h 08m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a1feda4eef3e42298dc54d337fe4ed4d : SUCCESS in 1h 09m 15s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6c695d30af034113bb38a4dd6c61317e : SUCCESS in 1h 16m 32s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/36033f3314814d5e936d78e882dcea75 : SUCCESS in 1h 32m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b3f6789efd264e0694d24d0ec2a2c030 : SUCCESS in 1h 11m 36s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/859b9ec1e5494de69216d6892f3e2213 : SUCCESS in 1h 38m 15s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c8c01714425f4e449e341c7a0859c9bb : FAILURE in 1h 27m 11s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/29c8a59881b346fcbafe2e024e8b8e27 : SUCCESS in 1h 06m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0690f3382f3e4c5e8c4a21320c2df024 : SUCCESS in 1h 05m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c3a65a8b97ca4adba1cb51efaa951371 : SUCCESS in 2h 20m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a4722269c134925b99dc0ebf5a2ce11 : SUCCESS in 59m 22s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/6be4d6b1a42145b4af69f9fe719c1bac : SUCCESS in 1h 23m 45s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8f337110dca54b2dbe5ce3eeb54f1eda : FAILURE in 1h 54m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/497c1ab7860f4c1887637c9f3c5b69c1 : FAILURE in 1h 40m 33s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/769aa9781bf64e478efa31c72b194c28 : POST_FAILURE in 30m 44s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/749a0c90272f4d2c87a93d8c6d9f842b : FAILURE in 1h 48m 26s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f61728201c1347279a3de6428364bb66 : SUCCESS in 1h 57m 48s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d0b2ed7424144ca78dd598bffd51ac77 : SUCCESS in 3h 01m 57s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/467cf3e97a2e45189b02bcb1e7bb94a8 : SUCCESS in 1h 44m 57s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/fa851e5f601846f482622cd9b45083da : SUCCESS in 2h 06m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1a6d045bd3f36e52128c8b1432069dfb5dfc6c73","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-26 09:55:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"333d1c6ee11bcf76f91585d6e4628b146cbc8be6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-26 11:45:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/737736/2 : SUCCESS in 1h 49m 14s","accounts_in_message":[],"_revision_number":2},{"id":"7ee583b6b90c860a122a743609164919db1a864c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-26 13:43:49.000000000","message":"Patch Set 2: Code-Review+2\n\nI guess you could almost squash this with the parent?","accounts_in_message":[],"_revision_number":2},{"id":"ecd29efedb200ce9281772c3b81a38ef40dbe4ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 14:54:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/73c08f176e9c47cfa8d54e6274758002 : SUCCESS in 56m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3f2749acf6e341ad84463750408b34bd : SUCCESS in 36m 58s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e0480083831d47f2aba2bce0ee9687f5 : SUCCESS in 1h 24m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d999452cb5b44a849d9e43527e7e75c3 : SUCCESS in 1h 10m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d9edd4f68fa94ac9a58262dac2ba4325 : SUCCESS in 1h 09m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b14104b71b574b4593bde1e4edeedf0b : SUCCESS in 1h 08m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb40ec5fdaaf4902b8264a67cb5f91c9 : SUCCESS in 36m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41ea5e239b2d4143a0c8faa640b7c621 : SUCCESS in 26m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/239a16c8ab3e4ae286b10e16fd470b9e : SUCCESS in 13m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/685bc31906e747a88e26681ea7e445b8 : SUCCESS in 24m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22b68b087240422a971e1e7d6ae0aae5 : SUCCESS in 23m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4caddf33ea04fcb878146ce5f40bd11 : SUCCESS in 9m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e6949d5758634db5b569d93eae1c3598 : SUCCESS in 40m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ba78786369864c2188a5bf97835d5224 : SUCCESS in 1h 00m 55s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/189f1aeee61a4d7ca7aeedc0533f8b36 : SUCCESS in 1h 15m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/355e99a7e13a42758113d54108644a8a : SUCCESS in 1h 13m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/946d74c294a241cf8c0337ebd950f9d3 : SUCCESS in 1h 55m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8849f22640f4b7d8c9a77be7d49c664 : SUCCESS in 1h 17m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f5e8083401424649ba61fc0fd86a1926 : SUCCESS in 1h 11m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/edbb36da981e40ffa4dd5b0f86de2f84 : SUCCESS in 48m 28s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/22f4c32aa281424cb29f0ea5974fcc1a : SUCCESS in 48m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7164fe082f2d4fb49e45cf94f99942a2 : SUCCESS in 46m 06s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f6eeff6579e04f50863098065b8ca0cf : SUCCESS in 1h 10m 00s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8ffe0e0e3e4c4f368ed73f7037a96982 : SUCCESS in 1h 42m 08s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ea1131179dac4acd980a7397cadf2cba : SUCCESS in 1h 05m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cdf5ce53d9ab4c3fa64748a08f34f496 : SUCCESS in 53m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5b7d92411c4042b090162a0780bfaefc : SUCCESS in 1h 58m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/856108a414e049cab5aea4ba6169ceea : SUCCESS in 54m 46s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a73a1e8f8a364ac6b216a581cdd93703 : SUCCESS in 1h 02m 09s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6398dcf0bed5495994035c7134eecd09 : SUCCESS in 1h 50m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4f74cf93d07649a9a45ac31193589409 : SUCCESS in 1h 47m 25s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8c76941fb4ab4d1b968fc05b71b0220e : SUCCESS in 1h 16m 43s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0aa9910fec0e48a6b5b2b80a816689d1 : FAILURE in 1h 56m 49s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ec8c939f7792480794cb323fa8d3a7fc : SUCCESS in 1h 43m 08s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0b0457eb446d4c43b4669c106bcf8c33 : SUCCESS in 1h 58m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/233a3466a5164f0e81fee75246ad4e42 : SUCCESS in 1h 15m 04s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/11db79267b654c699d8e4ac4f423e980 : SUCCESS in 1h 56m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"68321b159d5ae5c10ee0583e69829c790756173c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-27 08:27:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5ada28c614e529970b6dd6eb49711fb10cdecfe5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 08:35:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9e0f0ae5f8ee15d36eeef22cb67d0243b8c38a0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 12:25:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f3b9ce881458dc50cf8651216c276017cd7df508","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 12:25:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/31de9b82d86d49679a434c4eb91e6e53 : SUCCESS in 50m 46s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/eb1b1f56cb5b40d1ace484c8ec10719a : SUCCESS in 1h 01m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/40bc20f97b064aa58d4ae239e5cf9649 : SUCCESS in 57m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9ba59a5afd3149a2a75b210014d6b394 : SUCCESS in 1h 03m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a871cdfe3a034e059e92c8bc63158206 : SUCCESS in 1h 10m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/630b4170c4fd430a84ee15690c0d6a58 : SUCCESS in 25m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e3cb715c6354237bbd999c99ec14c59 : SUCCESS in 11m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d2bec7dffca4e76ab6854634aea0eb9 : SUCCESS in 18m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12d4c86493b2428387d3d2c00e1f1292 : SUCCESS in 24m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3234d230ffa48fa98443288a927ebce : SUCCESS in 7m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c4bf2465daef4a60852e5c64407d458f : SUCCESS in 50m 56s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c48e88b834ab4e458894d0ef3eb72b0d : SUCCESS in 1h 12m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9b1c94f589274676b25c23974bc334f9 : SUCCESS in 56m 50s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/de348668a5d04313a5c508900ffc2382 : SUCCESS in 1h 12m 02s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a7440b4df1b144b9851776abe5afe784 : SUCCESS in 1h 09m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/26f0724c17914159b65a93354c00e91a : SUCCESS in 1h 07m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/403cbb3562a04104bfe65bc8bbabc90a : SUCCESS in 1h 07m 27s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/77b382763ae649fa9685da796545cb0d : SUCCESS in 1h 57m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdfb61a5ede34e95a136c13bcfe9065c : SUCCESS in 48m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/144c848b30b24bb79e0fb5f64d6aa83a : SUCCESS in 1h 40m 56s","accounts_in_message":[],"_revision_number":2},{"id":"4b55f726ae89c6e4ebfd204241cd48285cc414cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 12:28:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fcec80b93df4233b08ec23586222b54 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2bb514f2d73a64b6f954990a2ff12f5baa3a97cd","revisions":{"f7ecaffb966d03a1fae53ad53864ce19fe2c932a":{"kind":"REWORK","_number":1,"created":"2020-06-24 10:05:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/36/737736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/737736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/737736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/737736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/737736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/737736/1"}}},"commit":{"parents":[{"commit":"ed56429548dd8d08de10a7d451680a43e34c865c","subject":"DhcpFilter should always return a valid index if \"force_scheduling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed56429548dd8d08de10a7d451680a43e34c865c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-06-24 09:58:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-06-24 10:05:37.000000000","tz":0},"subject":"L3 agent scheduler should return a valid index if manual scheduling","message":"L3 agent scheduler should return a valid index if manual scheduling\n\nWhen retrieving a vacant L3 agent binding index, if\n\"is_manual_scheduling\" is set, the method \"get_vacant_binding_index\"\nshould always return a valid binding index. If the existing binding\nindexes are sequentially aligned, the method will return a new one\non top; if there is a gap in the binding indexes list, the first\nfree index will be returned.\n\nCloses-Bug: #1884906\n\nChange-Id: I0a89bca0734d3e735fb357e488f85589e81d709f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7ecaffb966d03a1fae53ad53864ce19fe2c932a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7ecaffb966d03a1fae53ad53864ce19fe2c932a"}]},"branch":"refs/heads/master"},"2bb514f2d73a64b6f954990a2ff12f5baa3a97cd":{"kind":"REWORK","_number":2,"created":"2020-06-24 15:37:44.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/36/737736/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/737736/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/737736/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/737736/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/737736/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/737736/2"}}},"commit":{"parents":[{"commit":"ed56429548dd8d08de10a7d451680a43e34c865c","subject":"DhcpFilter should always return a valid index if \"force_scheduling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed56429548dd8d08de10a7d451680a43e34c865c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-06-24 09:58:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-06-24 15:37:23.000000000","tz":0},"subject":"L3 agent scheduler should return a valid index if manual scheduling","message":"L3 agent scheduler should return a valid index if manual scheduling\n\nWhen retrieving a vacant L3 agent binding index, if\n\"is_manual_scheduling\" is set, the method \"get_vacant_binding_index\"\nshould always return a valid binding index. If the existing binding\nindexes are sequentially aligned, the method will return a new one\non top; if there is a gap in the binding indexes list, the first\nfree index will be returned.\n\nCloses-Bug: #1884906\n\nChange-Id: I0a89bca0734d3e735fb357e488f85589e81d709f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bb514f2d73a64b6f954990a2ff12f5baa3a97cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bb514f2d73a64b6f954990a2ff12f5baa3a97cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
