)]}'
{"id":"openstack%2Fneutron~738145","triplet_id":"openstack%2Fneutron~master~I8284b22c5d691bfd9eadeb8590c3d4b27d261b04","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"I8284b22c5d691bfd9eadeb8590c3d4b27d261b04","subject":"port_forwarding: validate args before invoking db update","status":"MERGED","created":"2020-06-26 03:37:32.000000000","updated":"2020-08-05 21:26:09.000000000","submitted":"2020-07-27 23:11:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738145-1595891481599-e63e72be","meta_rev_id":"6eab11e0db0eec71bec5c51c944976bec529d979","_number":738145,"virtual_id_number":738145,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-21 09:08:41.000000000","_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-19 16:10:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-27 23:11:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-07-27 20:08:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-07-27 15:04:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":2,"date":"2020-07-20 14:50:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-07-27 20:08:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-26 18:33:20.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-07 18:08:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-07-07 23:53:23.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-07-08 11:57:41.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-07-19 16:10:00.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-20 14:50:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-07-27 15:04:33.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2020-07-27 20:08:37.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-07-27 23:11:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"473eac3a330a73007e797b5d1e394fb4a586e775","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-26 03:37:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0bb73c6d157b7be881d66fae2fd052e7cba5324b","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-26 03:39:38.000000000","message":"Patch Set 1:\n\nPlease -1 if you think we need UT for this. I\u0027m on the fence whether this change warrants one or not.","accounts_in_message":[],"_revision_number":1},{"id":"f72d4f9ef3bf7f630102722f11d7311b2db2798f","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-26 03:46:04.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"f58b99f6794dbe31ab3e9cb02848c4afd2bd5459","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-26 05:20:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/738145/1 : SUCCESS in 1h 42m 38s","accounts_in_message":[],"_revision_number":1},{"id":"cbe5ffff5809206651b751a49a50accf1790b8fd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-26 05:30:03.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/45/738145/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/45/738145/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"4a80eb763d1113c3730d7ec74d08c91abc58cc66","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-26 06:36:11.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/45/738145/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a22caf55f93826a1cfc2e8a178d56cda134047c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 06:48:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ac30f03e2249485cb885eed5da53a116 : SUCCESS in 56m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0be84ab58a3548a9963bed62e4b4f432 : SUCCESS in 39m 50s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/605f30c2353e4b979b119b3b4e2038c8 : SUCCESS in 1h 02m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b7c1601791cb4a0bb46b9201585f8c3c : SUCCESS in 1h 17m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6d364da5c73b4de7a8d3e6de06b4edf4 : SUCCESS in 1h 09m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cb20e57ee4c24b5980ff932879931a6c : SUCCESS in 1h 03m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d93ab27cc5034ec2ad133db2c200ca68 : FAILURE in 34m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc3dce3385ba4aaba9c657ec6e068102 : FAILURE in 27m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1885e34e98d04b719fd61a7fa18ec12c : SUCCESS in 13m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83f5d8a7f2c842609e955905bafdd495 : FAILURE in 28m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9df7a88519cc4114b2897b220885ab0e : FAILURE in 26m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e77d906cb6b84d1fb9517a461264c72f : SUCCESS in 9m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d5b4a4efd57344a5a517fe0da7f7bdfb : SUCCESS in 50m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/024ce983340f4e9cafe29852e33fb9b0 : SUCCESS in 37m 38s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8501c04ea6c44dc88f276ee5b4ad9f93 : SUCCESS in 1h 12m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/89c85384cd3b42be9eaf69b9c4881ae4 : SUCCESS in 1h 17m 14s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/683aeb7354104649ad90455d6e7fdf0f : SUCCESS in 1h 43m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c0883698f734c97bf82fee4d060e018 : SUCCESS in 1h 06m 31s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/00b36e3d4a7743458efaba68c130b72c : SUCCESS in 1h 11m 10s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ad74646949de487c9923e6cabe7eade6 : SUCCESS in 1h 06m 17s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/74920f3e9f524c669745477b77df95e9 : SUCCESS in 49m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/11feb492546b4227b801e0fdd3a6ddbc : SUCCESS in 1h 11m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fa13c6b919ff4eb2bb8aba5de9bdb170 : SUCCESS in 1h 28m 45s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/300f3a0925304b54a3b9d619b01ded4c : FAILURE in 1h 32m 27s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8cd905ad88bf44b58b096ac2410614f6 : SUCCESS in 1h 11m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3856576abe9b423c9fb2ef241f4874af : SUCCESS in 56m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e557ea737dcb4629b46b5a9592e31648 : SUCCESS in 2h 11m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f49c47ea92c04bed986179f2f4a373c8 : SUCCESS in 52m 29s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/0d6dbc5125894f6ab5035d85e5fd13be : SUCCESS in 44m 43s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bd688f825a4342e3b2acc1cd7695b70f : SUCCESS in 1h 48m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/66dd412aa16a408aa1b0997820734710 : SUCCESS in 1h 48m 45s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4ab18c4793504cae9590cbb7da6c3476 : SUCCESS in 1h 34m 03s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/10aa876c5e1c42199d37daf24bddc187 : FAILURE in 1h 45m 01s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7a532e19b71b47248470ef6dc24ac902 : SUCCESS in 1h 46m 10s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2319255fe1b34f03930098067a3c172b : SUCCESS in 3h 03m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/67ef6e95743b4f0ea45871ca81d0b369 : SUCCESS in 1h 42m 28s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/16907c8a892040c4a07ae593a54512bc : SUCCESS in 1h 54m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"19aed32a9032be6eb59598b886f596159dc374ed","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-26 11:50:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7cd72bb4d7fa83edd0673904b9187c6cead5cfb5","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-26 11:51:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"723edbf047071aa1d1e21aeb0cfc4feecb82807f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-26 11:59:33.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/45/738145/3/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/45/738145/3/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"9d666cd2109cf682df717644a693014f38a6eb19","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-26 13:53:58.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/738145/3 : SUCCESS in 2h 02m 01s","accounts_in_message":[],"_revision_number":3},{"id":"ca7db16418c9922c114776b009cb762d0defd549","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-26 16:08:24.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/45/738145/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"6a94ecf8453ece77d2842bb8c59399b1477ba4b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 16:35:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f630fe6eeb6f475aae08a35a7c54d42b : SUCCESS in 1h 01m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e593e325bed84855807759965f0d95cf : SUCCESS in 31m 26s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/81785ac2226645bb92df96db5a8281b0 : FAILURE in 1h 12m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1e6b332295e14f92b1d209159f216a2f : SUCCESS in 1h 23m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e23bc61f911749b3bb2e820118a160b2 : SUCCESS in 1h 10m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8dff44040a2841e1b37b960fa27cfbd8 : SUCCESS in 1h 00m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ad32c30b5904b54b315b74b198656ce : FAILURE in 31m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff2f626bd92e479e96bc586f433e7c32 : FAILURE in 32m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4875b8cc6ce64b2b8090e54b57fe3e39 : SUCCESS in 11m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afc734f3e4a24c32b98a3c319537583d : FAILURE in 27m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11bc97491efa4b399b22a705507500ed : FAILURE in 31m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99a95c0e93814993a52d4666ad4eb0c2 : SUCCESS in 11m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/019d2352bf8948fdbda42a6e9d1c58c9 : SUCCESS in 1h 00m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf381d9f0d0942f1befb429916536a2b : SUCCESS in 55m 55s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6a6c626047614d92a4396d21c572ab47 : SUCCESS in 1h 18m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3cae69858034cdc9dcc172e6fe5c98b : SUCCESS in 1h 20m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/857587a72d7846e9a3d4695e66c8e5f8 : SUCCESS in 1h 53m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/62569b50fd7f485c9be77b13d610cab0 : SUCCESS in 1h 13m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/16c25da30fc441bb9a8c795ff1ec36a4 : SUCCESS in 1h 14m 52s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e89c8a2a1efe409e8964c62800ebdf2e : SUCCESS in 1h 05m 10s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/831d8efe2ba343989b6ac3ccc492ff7b : SUCCESS in 1h 07m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ed2fd637c67840c38e0228c04cfce09c : SUCCESS in 1h 14m 48s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7c3ae618d88948c2a20a59e2ad0b3f54 : SUCCESS in 1h 51m 10s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e2933c8b747f4f638655961598630d3b : FAILURE in 1h 22m 52s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f47d66855dd04471a1c7b1f6b35d3f1d : SUCCESS in 1h 24m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bfbd523742794f85a02a0b2daf2d6133 : SUCCESS in 59m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/419a95bc7871452fb798a22261eebfdd : SUCCESS in 2h 01m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4b5960d7bfd403abe5fde727fb968d4 : SUCCESS in 36m 57s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/3fb5497f347f4d25b6d5462f6cd8ff89 : SUCCESS in 1h 16m 13s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f9aef7552b7c47819ac054cafc4b724e : SUCCESS in 1h 51m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d2fc4448754e46eda82252bc0729d965 : SUCCESS in 1h 53m 08s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/64445d15071e4e56810f89e809b79b88 : SUCCESS in 1h 11m 24s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cf63a3757992494bb13ee44f3f50a81d : FAILURE in 1h 31m 10s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ec6c2d90bd8744bda677543e7e01ae97 : SUCCESS in 1h 58m 14s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/eb1de6f297a043748836f0cae5728362 : SUCCESS in 2h 25m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7feb1b55112b47c48e45c506c490630c : SUCCESS in 1h 43m 21s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/4b4c36c9d7424b99b201feec4ac478d6 : SUCCESS in 2h 07m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"68caf57b4c074445146ba62c7f034e5254f2e0af","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-26 17:45:50.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/45/738145/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"9d77ef0629f1844bec41990bc30df79f0fe7764a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-26 17:50:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0e40462c29d83f500b2e6084a3eb9aafc643b592","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-26 18:33:20.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/45/738145/3/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/45/738145/3/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"a83520c06d6a011cdaab372d121ca2e60f8b9e24","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-26 19:39:58.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/738145/3 : SUCCESS in 1h 48m 47s","accounts_in_message":[],"_revision_number":3},{"id":"32a437f21b1a20a83e836536ca61e77db7e0d0f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 23:06:36.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/34c0dfbd7db24080a4b29dac884a5ae8 : SUCCESS in 49m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/61e1b39c1a7c406bbdaff0903e643dc5 : SUCCESS in 37m 35s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a45161cd21f143aebfe6df08e493c27c : SUCCESS in 1h 13m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/964373b276ab4796bcb91008619abb82 : SUCCESS in 1h 16m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/311d4f2f53b749aabb5f76576ffd3460 : SUCCESS in 1h 23m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/68431a124c1148348874d8f0a6c9b9ff : SUCCESS in 1h 06m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a148d193915946eaa0329bc2fa9013cd : FAILURE in 32m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fb6b1c6f9a44288828e619cd87603b0 : FAILURE in 26m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4305cfed95be410ba788e875b5b74da4 : SUCCESS in 16m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86f5495159414348889a841a84a40a23 : FAILURE in 28m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ccc5bda45674dee986a1122b45eb5fa : FAILURE in 34m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/987d424beeb446a0bc5126c50a08e1da : SUCCESS in 10m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7f82209161a740ffb9b8c9ebbb9d76d3 : FAILURE in 1h 11m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2bff9fabb98d441c9a10fe884dd19e7b : SUCCESS in 58m 03s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/51ec980e6c704c2e920a14df15bf0dff : SUCCESS in 1h 22m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a673e731f9854f67a506a47f64b94a82 : SUCCESS in 1h 12m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/14629e40cfd64f22a046b4dff42125be : SUCCESS in 2h 06m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/97df371fb4c243009c2f3c544256aefd : SUCCESS in 1h 13m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cf465c864883411bab8b86f6321eb6c0 : SUCCESS in 1h 12m 01s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/07231a6acd53436ea0dfafdd00e340fe : SUCCESS in 1h 19m 09s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a698d175e46c4282869687daf43d63c6 : SUCCESS in 1h 04m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/459a448aea9148a4ba88e08cb21fb55b : SUCCESS in 1h 06m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8a56ba49b7e74f7dae5261a91b3daa81 : SUCCESS in 1h 33m 58s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/24186d09dfb04e629f35a6d80cab7020 : SUCCESS in 1h 46m 04s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e25fcbfcad87465da4511887ed244293 : SUCCESS in 1h 13m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/61726629e2ce4739b71bc56fd4a86778 : SUCCESS in 37m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6a6266ddbd2648e688460adc48418538 : SUCCESS in 1h 57m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d60b67965e3f4ba790f2cffbb4d5ce13 : SUCCESS in 58m 00s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/4330b2748a2a4997953ac00fef4f638a : SUCCESS in 1h 14m 18s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2b0cfb283e03436ca26e9c907cfdcafc : SUCCESS in 1h 54m 51s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/22743761bb65421d8991c7f76b97dd75 : SUCCESS in 2h 04m 51s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4d26d08212dd4a4a9b5ac9794c339c20 : SUCCESS in 1h 17m 23s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/46a6ccdeb88f4054bcc5229bab90d0b8 : FAILURE in 1h 36m 59s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b63d76dc4e374b7f935c1571865bd5c2 : SUCCESS in 1h 49m 26s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/feb4e79e4c1341daba4863a68ef569de : SUCCESS in 2h 58m 05s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/08e82f2702ec4a9bad0f259f7a27288c : SUCCESS in 1h 40m 12s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/4ca36ae13daa4adc94aa7aa8b85f2ef3 : SUCCESS in 1h 59m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"41d4404040cdd8d07f8a566aaafeea2075689a08","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-27 14:44:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6f795f238bf379f5bbe8ef516073a398d5542228","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-27 14:47:01.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"0cd49a71246a6f9783d92236f3787a0f4f35f6f7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-27 16:16:57.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/45/738145/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"48f195539c710202b81efed3880c28da0f9de13f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-27 16:22:15.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/45/738145/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"6016fd559445c5f643ed84a9192f7344a84c4180","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-27 16:27:38.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/738145/5 : SUCCESS in 1h 40m 03s","accounts_in_message":[],"_revision_number":5},{"id":"1e7d19a4224e49590c4314a0ccbab8a55e23b00b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 17:44:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d8c9b14e2db4450b84fd42647e388ddf : SUCCESS in 43m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7fc124ad4e604858b2da86450e4730f6 : SUCCESS in 33m 07s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1e7528228da4431689f495eb1b7bec1f : SUCCESS in 1h 06m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/12731112f931429a9266a365306abd91 : SUCCESS in 1h 04m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/70bb1d6f0a52489789cc8a2fdd1da44b : SUCCESS in 59m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0b6255e214df4c70b8de21186f224ec9 : SUCCESS in 1h 09m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f960bd04c89c42c59c9279073e8f5e03 : SUCCESS in 33m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae5d3861f4e3430b9067a72eb3eef205 : SUCCESS in 26m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7447dc88327549608980138ce9a77f40 : SUCCESS in 11m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb2b1f8a2b73461d865f91ba17f62fe4 : SUCCESS in 25m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1d404824648493fbf2abf57104bbc82 : SUCCESS in 28m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e126be84336c4bc3bb440f3fd904ecec : SUCCESS in 8m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5985c4d9123346cb82fe9290cba08bf3 : SUCCESS in 49m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c47dc386a79a4616963c921037f18ce3 : SUCCESS in 1h 01m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9ee8b70982d64c9cb1cc9d6ec0ca872a : SUCCESS in 1h 11m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/325790d1cae44414baf103839e943aea : SUCCESS in 1h 25m 17s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4f86973fda424801a359bb5ab5754821 : SUCCESS in 1h 46m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9062181e509042a59868f0194010dfbe : SUCCESS in 1h 06m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/28f8863077bc4bab87bbf15852c1f16d : FAILURE in 1h 06m 11s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ae6033753740478191d7340ad3b4d28e : SUCCESS in 1h 32m 28s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/cc651d1d58f742c4ae46b95ec7b86bdf : SUCCESS in 1h 11m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/258db6cdd9214826b7feeb83e9b12394 : SUCCESS in 1h 11m 01s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2331ed85adc94db0b29396be8a55bc9f : SUCCESS in 1h 55m 53s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fae1e5045504406b8607c3eb14f3c4a0 : FAILURE in 1h 25m 47s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/830db8fe707c40eca00d9a66e5b5a6c5 : SUCCESS in 1h 10m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/97431b373bb14261bc6a66f9ea109cfb : SUCCESS in 55m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2220ff096e10488296f62312f25a21b9 : SUCCESS in 1h 28m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed4cb0cbf38f47f7ba4a66af69c470e1 : SUCCESS in 56m 47s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/9d74682a6e50451bbbf0dc2e947e77b3 : SUCCESS in 1h 13m 09s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e34635f4bd284ac6b03a482dd296f897 : SUCCESS in 1h 52m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bc0fe13bd31c4535b957fd0ca0b2f84d : SUCCESS in 1h 46m 26s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/51eb8f7ec97c491fa7db75969a3f28e5 : SUCCESS in 1h 17m 07s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2636a7b56ab94909b18087d10b04f69b : FAILURE in 1h 40m 44s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6cde688352814bf5bca96b588b1f7df2 : SUCCESS in 1h 48m 52s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/30b05e78c778458ab1fc1af0850d24cc : SUCCESS in 2h 56m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1585c64c34134474809862f3b8cac3a2 : SUCCESS in 1h 48m 50s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/40084d99e8134d4eb595bd90b1dda2db : SUCCESS in 2h 26m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"572f53db82b23d63b9e22de00416782d0fb52895","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-29 09:58:22.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"14e25c9c62c10902ab93552bf4c44e7cc67aaa53","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-29 10:46:40.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/738145/5 : FAILURE in 26m 08s","accounts_in_message":[],"_revision_number":5},{"id":"1c5a65a76eb2ff47b1148b4f126ef7e1dd86fab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 12:57:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/90b16c4354e84d84bb5d7f834a99a413 : SUCCESS in 1h 09m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b7dc18d1a754404b9ee69da9d3bef99b : SUCCESS in 47m 07s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f4814ce8c55e4bcca7fd8ab07c114ec2 : SUCCESS in 1h 05m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c174109ef00b469b82f38f8b0280a3f7 : SUCCESS in 1h 11m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a845142eeb8a412790f231571b129aad : SUCCESS in 1h 14m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/376ca201124949699972daf9c1839325 : SUCCESS in 1h 20m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ee0f108d3054e479065432397ffe006 : SUCCESS in 32m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0deb09d9d0640a588ff460441b7cec0 : SUCCESS in 25m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79cecb2a348e40ce9bdfc6d3380fd541 : SUCCESS in 12m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cef574f441444a6692cce3c3543d5944 : SUCCESS in 26m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1749fa4a50964f30ad0f1f9489bd186a : SUCCESS in 25m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/543224d610ae4d3ca15ecdbf1b26b5fc : SUCCESS in 11m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/35d53c867598424399acd159ea2c3129 : SUCCESS in 56m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f95d55516ca47ada9af1ef8e4dd3d30 : SUCCESS in 50m 47s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6cb2d58cb6d34ddf93d1482736e6dfdb : SUCCESS in 1h 00m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1caf3ab51c9a4c8f8366896f541495a1 : SUCCESS in 1h 01m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d5f14b52c2324205bd8b84237810030d : SUCCESS in 1h 23m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/715dbb1af5ce429b9c9d23365914e8b3 : SUCCESS in 1h 12m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ff461469efa04e9da3d816f9dd6da861 : SUCCESS in 1h 25m 14s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/973527b02cbe4149a57d01f82241251b : SUCCESS in 1h 28m 27s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/298bce7007da4eccb6a253634684fd4d : SUCCESS in 1h 16m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/48ad6bbe494d4a5c8ea68fda29c14c57 : SUCCESS in 1h 10m 25s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cc85e74fa925448e8a851a289b738d95 : SUCCESS in 1h 37m 47s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f93eda3f55704e978b2218aa0c0f12bd : FAILURE in 1h 38m 13s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d2ac61f233104201a60b945cfb35a686 : SUCCESS in 1h 21m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d20f4b5c834847c799c3d75fb10cb312 : SUCCESS in 52m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/41cbe71896584e959862175af5f80bfa : SUCCESS in 1h 51m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cea60c6c0d6f4f52bcb95cf3021be1e5 : SUCCESS in 56m 05s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/0b11c9d5d3504d29beaf81239481ee9b : SUCCESS in 1h 15m 25s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9754b5ede36b4eb4ab2712ceacb2bcc4 : SUCCESS in 2h 02m 01s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/24bc1143a19a474bb63319e4241c87d2 : SUCCESS in 1h 27m 49s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/05a644afa63f4908b7f0b26a28d5f253 : SUCCESS in 1h 12m 30s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bb8846ec790c4eba9a407c74aca9c0fb : SUCCESS in 1h 41m 08s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/31b3dbeab1c64510885859f942bc1a0b : SUCCESS in 2h 06m 19s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/40194e956ae94b6a9f49c4de95de3da8 : SUCCESS in 2h 48m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/eeee4918953a4255aed206d60764abcf : SUCCESS in 1h 38m 28s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/172f91ad0eae4fbea2bc4fe1683ea1e2 : SUCCESS in 1h 56m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"35a9b90edb1114499c5274aefc7b52a45fcf0a0c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-29 14:10:01.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d34fc8deb060129cdd9025e5b75386ee13b1bbe0","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-29 16:36:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d2d1f8011d288d5055eda9df1583b6c59630caf6","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-29 16:37:54.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"1b64e8a7cb75edbcef132fc1f360336774ad800a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-29 18:23:52.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/738145/6 : SUCCESS in 1h 46m 19s","accounts_in_message":[],"_revision_number":6},{"id":"32f3a85a820e25672a458e1868634acc543175a2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-29 18:44:22.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/45/738145/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"ebf5990239697b19206791ad7101ff6d0301e122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 21:32:20.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b95c643e7166478bb3c23a82eeb9bedf : SUCCESS in 54m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7b9f785ba54b4b1d81b450f2aa8fe6a7 : SUCCESS in 33m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5779bb07eb1546989a99856e191545b9 : SUCCESS in 1h 19m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/377c1880b86346efa7569e5727825600 : SUCCESS in 1h 14m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/775586ddede2444c86ba4ad20e62d362 : SUCCESS in 1h 06m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fa13786b8d0744dbaf5dfce2c6c8883d : FAILURE in 1h 13m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6389f4d49b74582966d72eff7e59dec : SUCCESS in 39m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9baa6b0b87e14fefb344a382874da06d : SUCCESS in 32m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ce9e0c4c16b463087f6280a2e1f137e : SUCCESS in 12m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7238b71dc4b4dea9e702adbc9de883d : SUCCESS in 28m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d229dba81a043ec82d8c5619ed1dc44 : SUCCESS in 23m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a3cb468be342febfe736873c0005f6 : SUCCESS in 9m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/26d45564a46f410fba21043f69e7dcc5 : SUCCESS in 57m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/da0b67a63f4145c9a92a3a1399981da4 : SUCCESS in 52m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/cd1007695aa4423ca00803f188f575b6 : SUCCESS in 1h 13m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3421f4661a54539acfe3a3513576391 : SUCCESS in 1h 31m 01s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d0ee1053d9c944fc9232e9777e2c0ff8 : SUCCESS in 1h 34m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/53bc3f4bc0a647479edc2bc48be77bc3 : SUCCESS in 1h 01m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/913fb95c26974ab8a2b4db99e8d2bce9 : SUCCESS in 1h 08m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1e2636614bd64f99b2c00a963c0d1174 : SUCCESS in 1h 09m 47s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8bb18229910242e186c146dd254248ed : SUCCESS in 1h 04m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f8769df48ba14ee8b032d3ca029b61de : SUCCESS in 47m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6d4f27ea77fb4b6a808575c8c81836fc : SUCCESS in 1h 28m 44s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/914fa685d96d4b26856aede4a387992e : SUCCESS in 1h 33m 13s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bbe545de783c4112bc40c38bbe2b4ba7 : SUCCESS in 1h 06m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/35d1ee209ed347ae91cc20945db002a4 : SUCCESS in 39m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8a8cc5dff8734155bff8bc2e0d6e9131 : SUCCESS in 2h 01m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b3c40b85c4743f1a5820d19cc5d8606 : SUCCESS in 53m 39s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/1c7b7c8ec2ce45f5b7879cd3262824e6 : SUCCESS in 1h 08m 57s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/53cb1a09dfc54328b4de8b0179e60689 : SUCCESS in 1h 13m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9ccbeb261bbc40589758e6ef163831d7 : SUCCESS in 1h 35m 47s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/12a5375c2c5243eb9a4fd52202eac21b : SUCCESS in 1h 16m 29s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0a0d620196f04995b243c47b7ddaed37 : SUCCESS in 1h 41m 42s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0a2b795f2c6247d092a78a6903d05c6e : SUCCESS in 1h 14m 51s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0930540eeafe47afba970aaf242ecb3b : FAILURE in 30m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ecbef1877ddb4dc4b94694e12a6776c3 : SUCCESS in 1h 40m 27s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/df0f80a458774e519fcc8360d41c7f09 : SUCCESS in 2h 03m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"32809dcad6f99130c4181e6598f30ed2ea25a3de","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-30 10:33:05.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"3096d7fe48c5f4ca10934b807d7027e994bc4aa2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-30 11:04:04.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/738145/7 : FAILURE in 30m 24s","accounts_in_message":[],"_revision_number":7},{"id":"7895e13c5b17d9c745bf7ee9750b203bbb1e3f33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 13:19:24.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ef7fae490cde4411a7bbe63ce9b57d31 : SUCCESS in 53m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ea0cc17ab5dd4b95a524b8f4f04c6cfa : SUCCESS in 35m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f14d9deae03c4af3996ee0dea6628315 : SUCCESS in 1h 12m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2d06bb4521ce4aab9bf7832ecfc849d7 : SUCCESS in 1h 21m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b1538f96f0ce43fc85830386e74f5c9b : SUCCESS in 1h 04m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2a756f37c1dc43d0b5eb370b4940a786 : SUCCESS in 1h 08m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6f2ffdf333f41a99ed6cc1685638565 : SUCCESS in 32m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4b2004c3e31459baf631a62d7229c77 : SUCCESS in 28m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/327e4faaf33e4eb0a38a18b0c980bc25 : SUCCESS in 12m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dec638045ad24657ab47477b47601da5 : SUCCESS in 30m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acde503ea7ef42ae84132dbcbe9859d8 : SUCCESS in 26m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3bad12da28d43c0b068815a236b997d : SUCCESS in 9m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e8b2398da71743bd9228a15624bf21c3 : FAILURE in 55m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0a00a5255d0e480a802f3bdd24a939cc : SUCCESS in 55m 03s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/660ad5109eb4407487b7dad1ea5e7603 : SUCCESS in 1h 15m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bc9a5c2dd4f4419bb39af52542ef504d : SUCCESS in 1h 15m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c906c72d583448eeb4fe1a97fef86ab8 : SUCCESS in 1h 48m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b210351b4b504734a393c025ee0ed9b1 : SUCCESS in 1h 08m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3d652c632740421ebf5fc8e262ba1fe7 : SUCCESS in 1h 19m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a448ffcd5b5a40e4b8a9a6eb392405ae : SUCCESS in 1h 08m 01s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/02d59ec9a08c409284376e5f16389b79 : SUCCESS in 1h 07m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/67ceb19d0744407189cf336ca872dbbf : SUCCESS in 1h 11m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/292642e99594497e9b55d77622938ca3 : SUCCESS in 1h 58m 26s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/091953095af34879966edfa5d876e7df : FAILURE in 1h 38m 28s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d589979f4eb447b994a38c7e36254066 : SUCCESS in 54m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/09b75018eb454c6cbdf5364eaa48ee67 : SUCCESS in 1h 29m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ac59656678b84dfd9a2202ef41493aca : FAILURE in 2h 42m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe071e389edb4660b0bcf465bde2829b : SUCCESS in 1h 00m 33s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/1d9c7d635e3a4911b3a4ab7982a56d3a : SUCCESS in 1h 17m 01s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5fce5eabcf7b4ee1931c3f33b42bb072 : SUCCESS in 1h 51m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e18c004684644ca2af812181a62427af : SUCCESS in 1h 47m 45s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8d1899b85cb14565908eaccba819bbcb : SUCCESS in 1h 20m 00s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/047febb30d3440eaa5145a4df5687e29 : SUCCESS in 1h 43m 39s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bbc5be308ee8495f9544f42af4853910 : SUCCESS in 1h 51m 48s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/cc5770366725407e95c443cfc3ca5e30 : SUCCESS in 2h 13m 58s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/19b7f64829b640ffa65f74054c161946 : SUCCESS in 1h 43m 02s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/2c4c5479b8724242b9031a986118a351 : SUCCESS in 2h 14m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"15e682e48fbebcc1a53c7ec0dd4c16c46a59b232","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-30 13:29:11.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"348b8565ff95679ff629e2e3dcd6e50894e4821e","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-30 13:30:23.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"921a6da8c97bd7a15d1568e05ef8b3099a1787a8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-06-30 13:50:58.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/738145/8 : FAILURE in 19m 53s","accounts_in_message":[],"_revision_number":8},{"id":"11fdffc46810ecd9df38de8477c610012d7b1c3c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-30 13:55:52.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/45/738145/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"a3b83c0e49b5757ab29d1e9d6897cc7ba3da417f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-06-30 16:23:48.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2020/Arista_Tempest_Testrefs/changes/45/738145/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"a127914a2414fd865595352747177308f080f314","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 17:46:42.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0a927839b8964e0da33e27429fa1405b : SUCCESS in 37m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a170dc9b7a4a45e8832f3e5c197cac9b : SUCCESS in 23m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fd7eed3c3faf40958aa192011fab2560 : SUCCESS in 1h 37m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ffc0978735454145bc44247b9182284e : SUCCESS in 1h 30m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2ae067cda13b42fe9b4df8aa892de7f3 : SUCCESS in 57m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a3a9d61534c74d14a20fa3b3ccfb7917 : SUCCESS in 1h 09m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56da4691c27542ffb7ba59ee460c63b0 : SUCCESS in 34m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b00ce6dd646548cb8765054a0a22dcd3 : SUCCESS in 25m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f73ff3ed52d54ac98ea4aa4b9626c507 : SUCCESS in 14m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6491b4ffc6c48b18de6040b9b78b895 : SUCCESS in 33m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5ea59ff292f4ebfb3b981b588903011 : SUCCESS in 34m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fc86c332f054e9aaf910461b602bfaa : SUCCESS in 11m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fe7ba01dc8be4cb08ff9b1a2d7112403 : SUCCESS in 1h 20m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9e811f924e648d08471537193ff5527 : SUCCESS in 59m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4d11edaa29d5479b99905329018ed950 : SUCCESS in 1h 26m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/43d48b439daa4fd385b77f8da737f609 : SUCCESS in 1h 47m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/be84fcc062ce476b99af9797fc2b2647 : SUCCESS in 1h 41m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/08bce75c99f745cea843a36d1441be50 : SUCCESS in 1h 50m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e96cf926782643a88d08b67ea33e4af0 : SUCCESS in 1h 24m 12s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/74796f56c6784bb48a900c6c1a7361ee : SUCCESS in 51m 05s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e8e2303468fc4a108869a75102d60ce9 : SUCCESS in 56m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4bc8f59391784653b89538a1cc3d570c : SUCCESS in 1h 10m 06s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a821124cb53d4646b4fceec92411ea6f : SUCCESS in 1h 35m 56s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8a74f685896f4adc9310273778a87e05 : FAILURE in 1h 15m 30s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/55d8f3c6f8534ac7bdfa1c0c03659117 : SUCCESS in 54m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e9cd4a2cba9e450face435ab349a8d1b : SUCCESS in 57m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/81ff71cc4bb049dc9ec2d956fd2c85f6 : SUCCESS in 2h 02m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2152e4c8e0845838e8700268fd632c7 : SUCCESS in 1h 07m 09s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5553675eafb0491fbf68afbfb8e7dd38 : SUCCESS in 1h 24m 01s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1c8c898b24774d4c91694f7c9684147c : SUCCESS in 2h 04m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e2f63635cd8b4e2a865a288764fccdf6 : SUCCESS in 1h 59m 13s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2e959ddd6a7a4e2d8cd45044a32c17b5 : SUCCESS in 1h 23m 52s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4c58bf7b682e49a08eb30147438c2111 : SUCCESS in 1h 48m 05s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/681bb38db2644c27b07f3b75b67777ad : SUCCESS in 2h 15m 56s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/76298e7aeae442368055ddac623d2299 : FAILURE in 2h 37m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ea54325c1c1649b69b5609891c60d7ba : SUCCESS in 1h 46m 41s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b1ad8a2d947343668cce5233b2cda676 : SUCCESS in 2h 18m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bc9593b1eeadeca5e9c35e41df5fd05fa0247ff0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-01 13:51:35.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"65c00d60b3f2845d01b628a7748ef95314ba3bd7","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-07-06 10:10:29.000000000","message":"Patch Set 8: Code-Review-1\n\nneed to make some changes to address bhaley\u0027s comments","accounts_in_message":[],"_revision_number":8},{"id":"45e9220b896243d8328fc0fd46ec269d93c52e9f","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-07-06 20:40:54.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"57cfee2c18f281ea2fa034facbdbc497fc7544c8","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-07-06 20:44:26.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"5ba4859c6b4e5c5cfae5e01e702c86d1d0910e40","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-07-06 22:22:57.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/738145/9 : SUCCESS in 1h 41m 28s","accounts_in_message":[],"_revision_number":9},{"id":"23bcf68b597a943600f22091194de5fefc4fe311","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 23:34:11.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7abd0252693a4fdeb3b34d16ba09b08c : SUCCESS in 1h 00m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/147233a6778944eca98ee89c0a950415 : SUCCESS in 36m 19s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5f34001eb492464a816df319dc613e9f : SUCCESS in 59m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e7d23931236740bfa0af6e31fe180f37 : SUCCESS in 1h 15m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/83118be3f41f4c1290b7115d0512d94b : SUCCESS in 1h 30m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/32894b23d9cb485d921a808cdeeee025 : SUCCESS in 1h 02m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59a6623a1f9a4361b7b2e93404f6b790 : SUCCESS in 37m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a591e1c59a5a4a0283c9850c89c343e5 : SUCCESS in 28m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab3b1a2760e04ea3986f3da6c869811d : FAILURE in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fcad17e03564fd1992677e28020edca : SUCCESS in 24m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d3a8fb74cef468a94b271207c1ac60c : SUCCESS in 23m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f69d15fb68eb4ca9873b2d827787ce55 : SUCCESS in 9m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0493b0d3e9804407a156e2d2658fac2c : SUCCESS in 54m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8fe0d143c06b4d6fabea70e94dc0d920 : SUCCESS in 41m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/610c4373684d408188763521c3441bf8 : SUCCESS in 1h 11m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/98f560fc299b400481e1d0f68843c754 : SUCCESS in 1h 18m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/19dce4b22e4e4377ab6296409b187e4f : SUCCESS in 1h 50m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/95ebd65893d642adb7332d92373bc54b : SUCCESS in 1h 03m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7d1c1c51a3624a07ae110607f4b95ae8 : SUCCESS in 1h 12m 01s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ff8c5e8424eb421b9fa5b2a6cf129690 : SUCCESS in 1h 26m 00s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d1ef2f5699dd467ebb8f6b3851b0166d : SUCCESS in 1h 06m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/407f4e733e034674ac4b9712853572e2 : SUCCESS in 1h 04m 20s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d7a1f0d1a09d4ffaa25b7779b579a9e1 : SUCCESS in 1h 13m 47s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/678aa77060264bbd9c166ae0d05568c9 : FAILURE in 1h 51m 26s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6cc53fd2640e430c94f98ee518bc2370 : SUCCESS in 1h 25m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1e104e5c24ff4da380dcfa65aa1033d9 : SUCCESS in 1h 02m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a2a1752dd5bf4fd7986c6e21511a013b : SUCCESS in 1h 58m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/733520763901410898c7cc8383408551 : SUCCESS in 53m 43s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/d23130e9d5324247a943b2b33a701c0e : SUCCESS in 1h 13m 10s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5b070e4ef66d4089ac18fcc9a9b65b6f : SUCCESS in 1h 56m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2a05bbb628194379be1bce9177fe3278 : SUCCESS in 2h 05m 52s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6f583383a1ad4d4fa86f26aeb49c3f92 : SUCCESS in 1h 25m 43s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/56433005d11c48328b90f051ac5980fe : FAILURE in 1h 39m 21s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/696d63a8728d496b91dc4c1c91c50fe4 : SUCCESS in 2h 03m 23s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6d14e3998cbd454698db4adda641142a : SUCCESS in 2h 40m 16s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/649bb6c20ebb47da8cebd5b52f10c1c3 : SUCCESS in 1h 46m 50s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/81e4a74196cc4ee2879266e65a6bc2e0 : FAILURE in 2h 06m 11s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b8bf4937f262b73b7fe4ea7366abdbe61baf3cfd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-07 06:31:25.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/45/738145/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"3ef0f1c4b1284c5668cd988cccded05a5171f27a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-07-07 14:56:17.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"0e05d71414d3c8ce701fb3b21541ed49ae9648df","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-07-07 16:51:55.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/738145/10 : SUCCESS in 1h 51m 06s","accounts_in_message":[],"_revision_number":10},{"id":"e4227d01655b4dd8bafe521aaf6152352a234803","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-07 19:29:15.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/45/738145/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"f4e5e24f8614b855a3164d92614461073a7af172","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-07-07 22:06:38.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"6dfe96b18994a2e4e13ce11ca37ff9fd417a3f52","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-07-07 23:53:23.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/738145/11 : SUCCESS in 1h 46m 07s","accounts_in_message":[],"_revision_number":11},{"id":"7bcc3f30d54f1651c5e3dacf253cd169b4260d08","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-08 02:19:49.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/45/738145/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"4214a028a52ff69ec4fb48c3d6294e922159eec7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 02:59:59.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/52cc04ad2d72475392acc3cecaea90f8 : SUCCESS in 1h 14m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1b7066d4814f472383ffb586e826ec7b : SUCCESS in 31m 28s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/032e16b56e4d45b7923af452ba737cc8 : SUCCESS in 1h 19m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dee4f88f920e42d9b50b410f39e6ec7e : SUCCESS in 1h 17m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/429f67ac7ee746b38925373b8513056e : SUCCESS in 1h 22m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c40afbebbfa14a62b17f27cd249b4672 : SUCCESS in 1h 17m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b4e7cec7ba54162b91eef1a5deaed46 : SUCCESS in 33m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d612439ed9c43c29c54788b98d174b5 : SUCCESS in 25m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc86e9d0270b426490e0099381acd724 : SUCCESS in 15m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fe4eeaffadf4b9484e34d3cba4975b3 : SUCCESS in 26m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bdd646d12384a71b012954a36954a30 : SUCCESS in 28m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54a8c8eb64d04f27b0508bfd769252b5 : SUCCESS in 11m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4e203d3304274737a54041b976038cdc : SUCCESS in 57m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee7d2abc57bb4e33a038064392699a9d : SUCCESS in 56m 51s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4a3296805cd24e09af1f9f32391ca60d : SUCCESS in 1h 18m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff1a77dc072c4f8f86d8789666d46668 : SUCCESS in 1h 08m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d5e4b757700c4846b9d5450bb5bf4764 : SUCCESS in 1h 42m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4e6cecf25b24acfae7bebc835194282 : SUCCESS in 1h 09m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7ce0f74286b046d0b463e0babd36c98d : SUCCESS in 1h 35m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/eccf2e0e0df247ffab25debe231e00c2 : SUCCESS in 1h 08m 57s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/32ddb29ef18d41cfb0e2c96ce54ecc09 : SUCCESS in 1h 14m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cd0258a416cf48688c8e5f9e2efd4e68 : SUCCESS in 1h 23m 41s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/66abe284d764403d958aee2f97655f2f : SUCCESS in 1h 46m 23s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6ed9a35b342443069549a30f79126a55 : FAILURE in 1h 25m 39s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6d1ab1e9f83f45b2a39e15cda19a4d02 : SUCCESS in 49m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/724d80573fd24809ad265038cbdb14da : SUCCESS in 41m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dba6afbfdb6c4af3a09287a3e0a786b5 : SUCCESS in 1h 28m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e23c17991da46d8a435e94d9ab37ca7 : SUCCESS in 49m 39s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/ab8ef9f5426c494daa2bc6cb6c797210 : SUCCESS in 1h 01m 31s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1068d16f9629456691ac2c2408d38264 : SUCCESS in 1h 40m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e75041c6643d41c19c078808231073ed : SUCCESS in 1h 56m 50s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c7c7efc9e615448196f22f6abb6c1293 : SUCCESS in 1h 23m 34s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/08157500ae0e400fa4c99ee0ff2c3641 : SUCCESS in 1h 48m 16s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b0213d4e3a874121bfd33a37b53be08b : SUCCESS in 1h 50m 02s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a171533e5bbf4cc9930ce5b19809bd6b : SUCCESS in 2h 19m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c6354983918144ddab02355bb0b1be5d : SUCCESS in 1h 45m 17s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/9de47643bb7f48439215f0154f7096f8 : FAILURE in 2h 11m 29s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7308c421acceaceaf574968913a447a2c9a06b1d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-07-08 11:57:41.000000000","message":"Patch Set 11: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":11},{"id":"1806ca35403c6340ec1de740d8294b104fac6715","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-08 19:30:34.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"8c30842493bf4287fe29fc44343257b4f0a17d40","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-07-09 08:46:31.000000000","message":"Patch Set 11: Code-Review-1 Workflow-1\n\n(2 comments)\n\nNeeds a few more tweaks.","accounts_in_message":[],"_revision_number":11},{"id":"7f48cb382df0aaaf77127b773f069321b448a35f","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-07-09 10:06:15.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"0bf189538173f2990f6fd5a6c22eb866b952ac96","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-07-09 10:11:57.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"68f5524e8888217149cecd566bad54637a390a0c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-09 12:09:47.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/45/738145/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"bee326c468dee0f9791a144f9333da50ad578076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 15:10:50.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/80662a379b9f4af2abc7b6760eb2c795 : SUCCESS in 51m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/03273006de614a03bee15bb3599a6e8e : SUCCESS in 37m 02s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1e378c720859493aa2f9b12538d7e94c : SUCCESS in 1h 12m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/26cd23d0f5f04b469c513c30b6eda62e : SUCCESS in 1h 24m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/627a81ecc112494fbf032373027e12d2 : SUCCESS in 1h 15m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fb86bbac9c17448b8ce816c84e9d8922 : SUCCESS in 1h 17m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/120ab281134b405d9df12be0a26d6da0 : SUCCESS in 33m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a96699c0a26846ce8a9dbb5576aa5c56 : SUCCESS in 29m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b221f2ca0fbc48d5b6a4f187ad0a2eb2 : SUCCESS in 11m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e2905db273346c2874e9f07e2331f7a : SUCCESS in 28m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21d921127b6b4689ad309000d396e639 : SUCCESS in 31m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90ccec21971a459bba4a0c993e891ad2 : SUCCESS in 12m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ae0c22705e514de397bc28f45287de28 : SUCCESS in 53m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c27ca078cf6541efa2b14e7282f0625e : FAILURE in 55m 16s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/090035e8d57048b3a980f2afc45ce16e : SUCCESS in 1h 16m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4f38a52e0fef48f8a3b392193d47c441 : SUCCESS in 1h 13m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f50371651bc5442bbe71ea8d1d2f6b0a : FAILURE in 1h 48m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f739f8259bd4e45a0a3ae5596f027d0 : SUCCESS in 1h 10m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/24baefa09c82438183bee313428cfcb0 : SUCCESS in 1h 06m 15s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b8ccce60d4d4405c8df89dad8d9cc2f3 : SUCCESS in 1h 13m 27s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/df29560e3d8047faa8a302c61aebbb48 : SUCCESS in 1h 28m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/173ca908cd1c4050b8c484aba8b809ec : SUCCESS in 1h 11m 31s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/148c7c36057f4a0f929e87231259477f : SUCCESS in 1h 51m 41s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a4c49886bc2541ceaa81b686b207096b : SUCCESS in 1h 45m 13s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9a0be9ba65f24268a08b0be3a1c84386 : SUCCESS in 1h 24m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9567f943702e4301901bdfabb5e72680 : SUCCESS in 1h 00m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0a7baab918bc4e42a2aadfcb61b2a193 : SUCCESS in 2h 00m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4cdecb4b77641ef892efa43684ac0ae : SUCCESS in 1h 00m 18s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/d2c451d02b85481e94aaf0721b542889 : SUCCESS in 1h 18m 24s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8aff903586684d9a9c210860365375f1 : SUCCESS in 1h 27m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/65ec66776653463f9496792857bc03de : SUCCESS in 47m 24s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/36267684caeb402dbeb75a6cc66dc943 : SUCCESS in 1h 31m 35s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/021428ad992d493587a055f042f01dd2 : SUCCESS in 1h 38m 33s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/37bfcd3dbd08470f8d60b9d31422c0bd : SUCCESS in 1h 37m 15s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ece3677ec90144bcbbce170c502481a1 : SUCCESS in 2h 16m 10s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2177af38203d4f85a9a00eb2b96f9afd : SUCCESS in 1h 52m 36s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/cac4177cc16c45f9bbc992e32e643222 : SUCCESS in 2h 24m 46s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"2dcb0fc0982719f230c7d5790ef5383d98a1425d","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-07-09 18:34:11.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"1e8a6b33f87fd3f1747e404594d96b2c167798df","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-09 21:31:42.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7bf2e869b21ed139788312d1414fe8ed5fd7a7d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 23:03:24.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d2af74229ea54a68b9be45e5db231381 : SUCCESS in 51m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2f18f2d99c1a4528b1c6969593720aa7 : SUCCESS in 33m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/468fbe5f516b47d99911cbf69da4bf8e : SUCCESS in 1h 10m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1c11182c0da64b7eada45c03cca55c89 : SUCCESS in 1h 08m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3fa1f4d8b646483a83c8456c3bdd6163 : SUCCESS in 1h 16m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/60953d7ce8964561a6ef6e5066f98f7e : SUCCESS in 1h 05m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb87c578badd413498840d3cf960537c : SUCCESS in 43m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d986b77ddee64247953b4f63c86483ea : SUCCESS in 29m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0324d70b3294557b6973f0cdfdc425c : SUCCESS in 13m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbc34cc723754b1586aa2af1af7cdc33 : SUCCESS in 17m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/587346c309e34a75bb456661ff251739 : SUCCESS in 17m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bd3f5cfda26428092dc5c3f5d622aed : SUCCESS in 15m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/18129f3bdfa14b8294f564da89cca0c6 : SUCCESS in 1h 05m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f6137d634eb24aa0bf2eadb45661e802 : FAILURE in 1h 28m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/72aa90597ac740eba8dc1a2a76f4eea4 : SUCCESS in 1h 13m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6440e0606bd94e8a9cfefb0ac225af6b : SUCCESS in 1h 26m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5daecde3997242768e370af14ba4b07c : SUCCESS in 1h 57m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a3af2085c7454c34a1f3aa12c6d808c3 : SUCCESS in 1h 06m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9ddf67291ed94a32a6c9912c1bb34784 : SUCCESS in 1h 11m 32s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9b51702b76f84adcad60f5f1b755372e : SUCCESS in 1h 06m 58s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a58b41dbcf314028843cd5aa018f6184 : SUCCESS in 1h 12m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c32bb21ff37f4e6cb80025296604e436 : SUCCESS in 1h 04m 31s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/898b0c3b55dc448c95dc0fd31cc2bd0a : SUCCESS in 1h 35m 39s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3bd0b3a89d2c4800a3161afef5a8c0b9 : SUCCESS in 1h 33m 17s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2035e01295ab446c883a847aa6bbf9fa : SUCCESS in 48m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/24339d0a85044e1cab5b502899f0ee65 : SUCCESS in 1h 12m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cb30ba409fa14f208a2099acf53c527b : SUCCESS in 1h 48m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3be04cbbcec745c8ab2f3f624dfa444b : SUCCESS in 57m 40s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/1a05753b5ab74fd9a51d20ea16f957e5 : SUCCESS in 1h 03m 43s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8a2fe39c0afd4c8a99d00467d710129d : SUCCESS in 1h 12m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3da41eac046747a889b0a76bb72f7594 : SUCCESS in 38m 46s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cd48d72cdfe8448db07528291a0dc2ef : SUCCESS in 1h 40m 02s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1a59e32bf4044f6a951f918aea0bd461 : SUCCESS in 2h 02m 43s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fe36e13754b447d286b0bcb7dfae3812 : SUCCESS in 1h 45m 50s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/711b2dce3f114d80bf3f7dc83c994c67 : SUCCESS in 2h 39m 33s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5137dd175ed34bf588f7177382d9fe30 : SUCCESS in 1h 12m 17s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6de3aa85547940438444045571a1d844 : FAILURE in 1h 49m 42s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"aae0e2a652aeb893e9ea08102ac40b8082060a8d","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-07-10 08:55:48.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"70933977473e7fed753b1425ce276b6b5a146296","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-10 10:54:35.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/45/738145/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"495a30fd8e2a791592d2e64dee3290bd52dc8a55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 11:52:03.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/37116622313949f1995048f836896903 : SUCCESS in 54m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/55e9152e5d424067a5dc189813e46a04 : SUCCESS in 39m 33s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/15e98f0aace74705828fb1209e495fc4 : SUCCESS in 1h 19m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0592261e87614bc894b50facd5b08601 : SUCCESS in 53m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1f90892770a64febbee9f8cd0310c12e : SUCCESS in 45m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d8e90be69d0a4926b89fa4f477e6e162 : SUCCESS in 53m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5025fb51446c4f289afa541bc5b9f8f2 : SUCCESS in 41m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc3e7b2f278e4a60ad779cd82cde82ad : SUCCESS in 27m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e494794965042228d22ff35ab3c0090 : SUCCESS in 12m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f55cfa380a53403cbdd19b45b562a149 : SUCCESS in 27m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51ba337283d3476c86ad329493088a9f : SUCCESS in 25m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0751046e61748cdb01518348ea714dd : SUCCESS in 11m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/072485ac06c04ec4905cae485c7fd96d : SUCCESS in 57m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c33682e229f45c99a633efb19125f50 : SUCCESS in 51m 00s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/065863e9cc0f40fb90b7a0df3fc67b35 : SUCCESS in 1h 12m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a28f553c76dc4169bd390038cda635de : SUCCESS in 1h 13m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/334a2e0cc49c47fc83ea361ecfa084b6 : SUCCESS in 1h 43m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f490c50305b548569cbf0914d061c193 : SUCCESS in 1h 12m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/be44724638844a4aaeaec825a8971dd8 : SUCCESS in 1h 15m 55s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ac724183e7e24255a923d681be6f51cf : SUCCESS in 1h 28m 22s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/66313e3e50f64365acdd8fce38a2ea18 : SUCCESS in 1h 30m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fb92231dfe3b4f10a8ce97e254304726 : SUCCESS in 1h 13m 32s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/31d328dbb3224c14b9a8e1ca44891804 : SUCCESS in 1h 56m 43s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/58aa378289e241128a87ac61b18459d4 : FAILURE in 1h 30m 21s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/47f90f8fe5f94ce5a4bd1e3d52cbecaa : SUCCESS in 1h 05m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3513102bdbf144059133431d4b83440d : SUCCESS in 1h 13m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6683d777af704562b9e5f3b5d985cfb6 : SUCCESS in 2h 05m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e9ff2abb7254ac6b12f0d55e36b30df : SUCCESS in 1h 00m 16s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/010cc7fa882b4be19ef64b5497448350 : SUCCESS in 1h 36m 34s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dffc55de421742c5814b0960d1e1d858 : SUCCESS in 1h 14m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8aa0b2b00d6a466b8a3f7ac0b9f249d0 : SUCCESS in 45m 18s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3eea1059ebbe421ca39af4705f25da57 : SUCCESS in 1h 23m 48s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3c0d7ce7b5264c098c0161c51d5a513a : SUCCESS in 1h 43m 13s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/795bc41443474d07ae3ebaaca0d4038c : SUCCESS in 2h 10m 40s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/71f02734d5294773a6eb5264c40567b3 : SUCCESS in 2h 20m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3d77d72554cc47c2971ee5d14aca1abc : SUCCESS in 1h 49m 30s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/5ea25e46aae9434d83eb30d4d399b58d : FAILURE in 2h 39m 20s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"80dc9bb182f83d46b3d637548b3c1bcea355aa80","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-16 15:17:58.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d58bb04d4df0f807fc48a807d80bc2eddee33c1a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-07-19 13:19:40.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"e5bbbe8fd79287833422049822e1dd37b5bb2e3b","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-07-19 13:32:23.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"5ead5e1d5384256613d2cfae7981f881057f8923","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-07-19 13:33:12.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d2502b000561136e051c58bace623537841324c0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-19 14:53:24.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/45/738145/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"f1eed5ea843654a64e1100f0d372c9dfdb685667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 16:00:13.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/83f7a5fe98d84284a52524e491bfc2ca : SUCCESS in 46m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8269a86f78c847de8bbaf9588f6f0a19 : FAILURE in 30m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/16634db668b945609867ab129ad4eff6 : SUCCESS in 2h 11m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2f4a24203ab2486ba19198fc40738e8f : SUCCESS in 1h 14m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/edc42f253aa94aedbf9cdc7976a0cfcf : SUCCESS in 1h 07m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/282cd589cb7e48279826ebf8fa08167e : SUCCESS in 1h 15m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ead952169094854b5fda03ae0696c58 : SUCCESS in 41m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d64e60e3e89745b4a400d635ba08caf1 : SUCCESS in 24m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c8cf09edce4824b20bf019315b3919 : SUCCESS in 11m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc146a69ae0f448d8b1f404207d38dd8 : SUCCESS in 16m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7df0e680f564d198856071bc64b032b : SUCCESS in 50m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/798ae1b1122d4561a3ee6db547fe91e6 : SUCCESS in 9m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f76dae2f2d0d48b4a900325512fe4c70 : FAILURE in 1h 32m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a954183a0164463886f05d857d736ec2 : FAILURE in 59m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/db48c428a55f4294b36ad582581f768a : FAILURE in 1h 03m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cfcf198222684ed5be38eb881e861623 : FAILURE in 1h 07m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/66bcc23b275c4d3e8bb182e1855e32f4 : SUCCESS in 1h 43m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d602d2eb12de44ceba09141e3e26171e : SUCCESS in 1h 06m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e48e9a3b9a7449d8bcfb5946a4d5182c : SUCCESS in 1h 16m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1676b92e860745549170f8ca8e7fd08e : SUCCESS in 1h 05m 04s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/155822d31c0547bcb49b2de5a0428bda : SUCCESS in 1h 15m 36s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/229a11c2622344c488cd2d03e9fdde01 : SUCCESS in 1h 25m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/101cb1d85f2641cca26ba828002fe71a : SUCCESS in 1h 54m 37s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/73ff7ad483c742079ba0d21d69e86e93 : SUCCESS in 1h 13m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/28d0934f0ff940f593d3f6247032949e : SUCCESS in 1h 07m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/091430b121764802b68b6f75b06ca201 : SUCCESS in 2h 10m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7a8f985760743f39b8a8f2319e4c927 : SUCCESS in 1h 06m 54s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/be4b0c2461764ac79a745713657a8494 : SUCCESS in 1h 07m 20s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dc7a0c5e2b7c4d1da170126f9abeb01c : SUCCESS in 1h 12m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b976dc8551634810a6345b0d443ed483 : SUCCESS in 1h 23m 23s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3338e293a2f7419ea511a1c24f73ec68 : SUCCESS in 1h 12m 59s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6ff384684918428ab38e23d7e9a872b9 : SUCCESS in 1h 06m 13s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6407493000604d318ca6033ec72db95d : SUCCESS in 1h 56m 56s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2c3e723534294f569c3764823c1116a6 : SUCCESS in 2h 22m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ad7875ff3f3c4e78962ea637759bb57b : SUCCESS in 1h 45m 54s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b5a56fa2461b4b729952f39f1b0e5aa6 : SUCCESS in 1h 23m 11s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"19783d874629917e2fa2fca1549367ca2addc382","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-19 16:10:00.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/45/738145/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"20512d51efb1876850028aad401f9291972fc870","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-07-20 09:43:05.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"1fd95f89c8852ed8fb61249a7ed331dc192223de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 12:14:28.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f53e9b1f092e46789c71d45b13e6b8d5 : SUCCESS in 46m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b3d55c249c76473d94d1e2c6359bc4c5 : FAILURE in 40m 25s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/51d28f5c2b154945a2839ecbb8fb02f2 : SUCCESS in 1h 27m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e1bd80731084472492897ee3d01c84ce : SUCCESS in 1h 12m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/18d61ecbe1a34571b6cd1da2ffdd79d7 : SUCCESS in 1h 10m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cb5e125647e845aa9ee72e38b19ad4e0 : SUCCESS in 1h 14m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b9d82caf252405a8aa3e299f8cfa9e0 : SUCCESS in 32m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1453e9648384466991956abf876eba46 : SUCCESS in 27m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a749bf879e4347b19b1d314658117374 : SUCCESS in 14m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c984d2e9a74457da6f089339b024a3b : SUCCESS in 25m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59a11d97cb834f50818c6859281ec8fa : SUCCESS in 25m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4c378c79bdb4a678bac0ab72bbc3264 : SUCCESS in 10m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/74d3cb05604f471d8d633bb65a882bb1 : FAILURE in 52m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/125bc9d2a2004a24a54cffa951559fa9 : FAILURE in 47m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fd189aed437f4999b2ca13ac115ac3bd : FAILURE in 1h 07m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6879590492bf4269b51752a234b285b5 : FAILURE in 1h 09m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1a0d6d399c9c4a41b56d089ccc752cf9 : SUCCESS in 1h 51m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ed02fe61d1f4502ad0be695860c6e9a : SUCCESS in 48m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ca89e7655b014550898f2c3ac920eb63 : SUCCESS in 1h 06m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e19f8f36eada488daf54b8a6f11756fb : SUCCESS in 1h 19m 06s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c21075d611f9420a854ca4a71e567828 : SUCCESS in 1h 14m 42s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/94aaae8fac5345e8835ec0f86b26b527 : SUCCESS in 1h 27m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fa8c7acdd132461b9ac260c18ce68c8a : FAILURE in 1h 12m 33s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b93755e0538a4485b0f95fbcdbe82adb : SUCCESS in 49m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4791d93c431848caaed3d8134f24ea58 : SUCCESS in 50m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9dc9fd2184e04000881ba44d6ebfd7dc : SUCCESS in 1h 55m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/367dda79f59f4e40bc223570bfed2034 : SUCCESS in 53m 59s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/f46ce932c2b14473b5112fe87cb74e94 : SUCCESS in 1h 09m 37s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e8b1788f292244c481c36cad31fb62d7 : SUCCESS in 1h 16m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/73f4a90ad2454607913e7c0008abba99 : SUCCESS in 39m 44s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6aa33674b6234c0aacb7e2055482e884 : SUCCESS in 1h 19m 18s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e826fefe61d148c9a0ef585dd9cfed17 : SUCCESS in 1h 45m 49s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e422cbe4014b44c3a1ed4733613a5154 : SUCCESS in 1h 45m 00s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d228bcf8cdd94d36a44dcfd35c4619c3 : SUCCESS in 2h 28m 46s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7eda35f8c5604c6ca013e9edbfc93ac4 : SUCCESS in 1h 44m 21s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e5e96c3703174058a45d62060841ed99 : SUCCESS in 1h 57m 17s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"9c11021d856300f28a78b008be3f53d96e2d1175","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-20 14:50:03.000000000","message":"Patch Set 15: Code-Review+2\n\nThanks Flavio, just have to wait for the functional test fix to merge now.","accounts_in_message":[],"_revision_number":15},{"id":"fe38764b67e3f80bf4b0d7e87a844493123eb490","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-07-21 09:08:41.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"fee468937631da04b08b2b9e0dd41cdc566b9e06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 12:27:26.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8235a75429d945059cf2fce9465edb2b : SUCCESS in 59m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/42a36ebd9c5341c4b5aa8b74e57485b9 : SUCCESS in 39m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/920596cae7d1457dac503fd016846825 : SUCCESS in 1h 25m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3d2b1a9979914382a0ba95afe283c557 : SUCCESS in 1h 12m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/96fdebbb67494b6b84e2f0dd7682b03a : SUCCESS in 1h 04m 06s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2dc88c57af1f4aa5a59bd5249b2a9beb : SUCCESS in 1h 19m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71f8ae4a43d44378a8ae7a37ff7a6296 : SUCCESS in 42m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d3f0817dc5d4db4b71659031732acad : SUCCESS in 31m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7dc4fd1564146f79e5acb5218b7aa04 : SUCCESS in 12m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13ca23c8a7424a7db871bade3fc6d884 : SUCCESS in 26m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a2dab7bb95241e992754f115df7fdb2 : SUCCESS in 25m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cae7a57a867e4a7e943654e48f3ba4b3 : SUCCESS in 9m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e5d0e84a9b1a4f2e941e5fa04c7b25ea : SUCCESS in 47m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9aae9d23fff4405e842c39219a4761d5 : SUCCESS in 51m 41s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1426271aff24433888bf1a17101dd8f4 : SUCCESS in 1h 21m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5687d6e82ff047fe938d0b91e56ac549 : SUCCESS in 1h 15m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/70f09615bfd94ebea77afc996000ec15 : SUCCESS in 1h 43m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b3f594c18564404854b07500e478196 : SUCCESS in 1h 03m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7e62bd4c2bfd4fc6b43cc5630a733598 : SUCCESS in 1h 10m 13s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/623ac1ad43154467b2bc4c2a4c245deb : SUCCESS in 1h 11m 24s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/393763efa9fa4654bbaa1b012b6bd623 : SUCCESS in 1h 08m 40s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6c358e4b4afb4ca2aa4d9c156a1e37fd : SUCCESS in 1h 26m 49s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8159f866182c4a9e904d3ee080103e4c : SUCCESS in 1h 46m 36s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fa041ca4b2e94ce6ac99a63b7f79c9cc : SUCCESS in 1h 31m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b2116dcff5334be584dd030a465ff8af : SUCCESS in 1h 21m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0f4110599ea542c6a29392f93b543197 : SUCCESS in 2h 29m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c316ac48f80244a8a2543d81cda12617 : SUCCESS in 1h 07m 48s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/01da0ac9db5f4765a8e76944ef90647a : SUCCESS in 1h 20m 13s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7a7b02471b66424c9342f902b20927b6 : SUCCESS in 1h 35m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5613ea0d10d942f5a46e896f70612bc0 : SUCCESS in 41m 50s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3af7cef0c5274cd0a0d411af59e24f4e : SUCCESS in 1h 14m 27s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b213b646ec144092811102095448beb7 : SUCCESS in 1h 34m 41s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9aaa70b4e69d42c3b8a4fe99cb84eb05 : SUCCESS in 1h 44m 44s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/85d4c8c36eeb4efe9b3ed56711adc430 : SUCCESS in 2h 18m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7469b2559aa44c888a4daa10f40328f8 : SUCCESS in 1h 57m 20s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/0404c58aba404d0f9cd3115d2b8cae48 : FAILURE in 1h 48m 50s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"70a5374066e71e2c6f3d79162a5f783ef1841f39","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-07-27 15:04:33.000000000","message":"Patch Set 15: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"5ffe8746e29b09288d39148e0c5c7dc954ba1188","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-07-27 20:08:37.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"505a584bc94e21ed7685b87c3e10ddd72b9a3640","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 20:09:00.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"ba40f661fdffc819fd22b61804dfcbc2eaf697a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 23:11:21.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/09198fce8dc144f9974eed98b55f4bf9 : SUCCESS in 53m 19s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/521912a1b2d044aca8723bdb74ba13f7 : SUCCESS in 1h 07m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3c964c966cb64100a903718d9454c5e3 : SUCCESS in 1h 22m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cf74302cd76a4de2877e9bdf9447c301 : SUCCESS in 1h 17m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d85c02b4a04e43eea4d46b490f07c6a8 : SUCCESS in 1h 05m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f13a6efe33c4d098ac1dbc172ffd440 : SUCCESS in 29m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f62478d223824003b40d14910530dc4e : SUCCESS in 12m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f9220b46bdd4b0489865497d2d9da69 : SUCCESS in 26m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65310773a2f34a19be1a77e089b4d544 : SUCCESS in 23m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db4c910e4c54346ab447731bdb1bbbe : SUCCESS in 11m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/193a9fc3f06040598c656bcc85730f3b : SUCCESS in 53m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ae7d97f5b0d44421b0236e003476b027 : SUCCESS in 1h 08m 20s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5476d1b6e3e0445f959d5581bfdf20ba : SUCCESS in 1h 26m 56s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/15a27cb1b2fc4dcfbe084712c964efb9 : SUCCESS in 1h 10m 54s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/051fb5082ede447995f1076f143c9596 : SUCCESS in 59m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d318e6d01a1248e1911866adaf223172 : SUCCESS in 59m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/27742782ff1a49628aaefbd4d361f717 : SUCCESS in 1h 12m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/11d17bb9607e4c0baabb90ee912ce6d9 : SUCCESS in 2h 07m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b630f8c799764157a81541f5c1265c25 : SUCCESS in 1h 12m 11s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/de50c6b9b2874d9a8a33fbe2cc4e8178 : SUCCESS in 1h 17m 29s","accounts_in_message":[],"_revision_number":15},{"id":"9c144c672e54b306953f0109ccaea0231e2842b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 23:11:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"f9a1216d9c7a2e852a33e052eb99302e319fc311","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 23:14:58.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aecca1fa30fd4711934ca9ee841f7972 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":15},{"id":"6eab11e0db0eec71bec5c51c944976bec529d979","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-08-05 21:26:09.000000000","message":"Patch Set 15:\n\n(2 comments)\n\nAddressed review feedback in: https://review.opendev.org/744993","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"f3797403486ba43100ff9ec8917d38c8723e12b2","revisions":{"4369d61e664971d0debd0877cc846e69afe4fb1b":{"kind":"REWORK","_number":1,"created":"2020-06-26 03:37:32.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/45/738145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/738145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/738145/1"}}},"commit":{"parents":[{"commit":"1f9268aafc575d562d1244fd43692d637e807358","subject":"Merge \"[OVN] rename get_port_groups to get_sg_port_groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f9268aafc575d562d1244fd43692d637e807358"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-26 03:34:32.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-26 03:35:28.000000000","tz":-240},"subject":"port_forwarding: validate args to avoid getting stuck in dp update","message":"port_forwarding: validate args to avoid getting stuck in dp update\n\nAdd validator to update_floatingip_port_forwarding so codepath does not\nattempt performing invalid database operation. With that, operation fails\nright away, with a hint on the offending argument.\n\nChange-Id: I8284b22c5d691bfd9eadeb8590c3d4b27d261b04\nCloses-Bug: #1878299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4369d61e664971d0debd0877cc846e69afe4fb1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4369d61e664971d0debd0877cc846e69afe4fb1b"}]},"branch":"refs/heads/master"},"85d259d9ac4a4bdb4c21c712ab8d3b60e29b922f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-26 11:50:13.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/45/738145/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/738145/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/738145/2"}}},"commit":{"parents":[{"commit":"d9d273e53864acaa1266cda8996dfb6594e804ab","subject":"Merge \"[OVN] Add router availability zones documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9d273e53864acaa1266cda8996dfb6594e804ab"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-26 03:34:32.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-06-26 11:50:13.000000000","tz":0},"subject":"port_forwarding: validate args to avoid getting stuck in dp update","message":"port_forwarding: validate args to avoid getting stuck in dp update\n\nAdd validator to update_floatingip_port_forwarding so codepath does not\nattempt performing invalid database operation. With that, operation fails\nright away, with a hint on the offending argument.\n\nChange-Id: I8284b22c5d691bfd9eadeb8590c3d4b27d261b04\nCloses-Bug: #1878299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85d259d9ac4a4bdb4c21c712ab8d3b60e29b922f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85d259d9ac4a4bdb4c21c712ab8d3b60e29b922f"}]},"branch":"refs/heads/master"},"a15610a22c18b181d0973e2ea0e6724075637f24":{"kind":"REWORK","_number":3,"created":"2020-06-26 11:51:21.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/45/738145/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/738145/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/738145/3"}}},"commit":{"parents":[{"commit":"d9d273e53864acaa1266cda8996dfb6594e804ab","subject":"Merge \"[OVN] Add router availability zones documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9d273e53864acaa1266cda8996dfb6594e804ab"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-26 03:34:32.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-26 11:51:18.000000000","tz":-240},"subject":"port_forwarding: validate args to avoid getting stuck in dp update","message":"port_forwarding: validate args to avoid getting stuck in dp update\n\nAdd validator to update_floatingip_port_forwarding so codepath does not\nattempt performing invalid database operation. With that, operation fails\nright away, with a hint on the offending argument.\n\nChange-Id: I8284b22c5d691bfd9eadeb8590c3d4b27d261b04\nCloses-Bug: #1878299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a15610a22c18b181d0973e2ea0e6724075637f24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a15610a22c18b181d0973e2ea0e6724075637f24"}]},"branch":"refs/heads/master"},"a6cddf8e900adf8475af6e8cc0f1b7d9f64e0e0c":{"kind":"REWORK","_number":4,"created":"2020-06-27 14:44:38.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/45/738145/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/738145/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/738145/4"}}},"commit":{"parents":[{"commit":"d9d273e53864acaa1266cda8996dfb6594e804ab","subject":"Merge \"[OVN] Add router availability zones documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9d273e53864acaa1266cda8996dfb6594e804ab"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-26 03:34:32.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-27 14:44:01.000000000","tz":-240},"subject":"port_forwarding: validate args to avoid getting stuck in db update","message":"port_forwarding: validate args to avoid getting stuck in db update\n\nAdd validator to update_floatingip_port_forwarding so codepath does not\nattempt performing invalid database operation. With that, operation fails\nright away, with a hint on the offending argument(s).\n\nChange-Id: I8284b22c5d691bfd9eadeb8590c3d4b27d261b04\nCloses-Bug: #1878299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a6cddf8e900adf8475af6e8cc0f1b7d9f64e0e0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a6cddf8e900adf8475af6e8cc0f1b7d9f64e0e0c"}]},"branch":"refs/heads/master"},"07a9cfff83d7bea93df67faadbc74c3d8fd0b657":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-06-27 14:47:01.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/45/738145/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/738145/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/738145/5"}}},"commit":{"parents":[{"commit":"d9d273e53864acaa1266cda8996dfb6594e804ab","subject":"Merge \"[OVN] Add router availability zones documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9d273e53864acaa1266cda8996dfb6594e804ab"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-26 03:34:32.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-27 14:44:55.000000000","tz":-240},"subject":"port_forwarding: validate args before invoking db update","message":"port_forwarding: validate args before invoking db update\n\nAdd validator to update_floatingip_port_forwarding so codepath does not\nattempt performing invalid database operation. With that, operation fails\nright away, with a hint on the offending argument(s).\n\nChange-Id: I8284b22c5d691bfd9eadeb8590c3d4b27d261b04\nCloses-Bug: #1878299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07a9cfff83d7bea93df67faadbc74c3d8fd0b657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07a9cfff83d7bea93df67faadbc74c3d8fd0b657"}]},"branch":"refs/heads/master"},"e8c8e3d88cb25d4876a5e0d9aa053a9c6bffa253":{"kind":"REWORK","_number":6,"created":"2020-06-29 16:36:58.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/45/738145/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/738145/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/738145/6"}}},"commit":{"parents":[{"commit":"d9d273e53864acaa1266cda8996dfb6594e804ab","subject":"Merge \"[OVN] Add router availability zones documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9d273e53864acaa1266cda8996dfb6594e804ab"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-26 03:34:32.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-29 16:35:10.000000000","tz":-240},"subject":"port_forwarding: validate args before invoking db update","message":"port_forwarding: validate args before invoking db update\n\nAdd validator to update_floatingip_port_forwarding so codepath does not\nattempt performing invalid database operation. With that, operation fails\nright away, with a hint on the offending argument(s).\n\nChange-Id: I8284b22c5d691bfd9eadeb8590c3d4b27d261b04\nCloses-Bug: #1878299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e8c8e3d88cb25d4876a5e0d9aa053a9c6bffa253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e8c8e3d88cb25d4876a5e0d9aa053a9c6bffa253"}]},"branch":"refs/heads/master"},"908be9d588b5d70414d1317d6da3fe8940952054":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-06-30 10:33:05.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/45/738145/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/738145/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/738145/7"}}},"commit":{"parents":[{"commit":"885cce39b7ad45d229a10c0378336cbbe125a94b","subject":"Merge \"Do not block connection between br-int and br-phys on startup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/885cce39b7ad45d229a10c0378336cbbe125a94b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-26 03:34:32.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-06-30 10:33:05.000000000","tz":0},"subject":"port_forwarding: validate args before invoking db update","message":"port_forwarding: validate args before invoking db update\n\nAdd validator to update_floatingip_port_forwarding so codepath does not\nattempt performing invalid database operation. With that, operation fails\nright away, with a hint on the offending argument(s).\n\nChange-Id: I8284b22c5d691bfd9eadeb8590c3d4b27d261b04\nCloses-Bug: #1878299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/908be9d588b5d70414d1317d6da3fe8940952054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/908be9d588b5d70414d1317d6da3fe8940952054"}]},"branch":"refs/heads/master"},"33e019ab03c2edb859b29e3bf7144bfcca49a630":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-06-30 13:30:23.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/45/738145/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/738145/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/738145/8"}}},"commit":{"parents":[{"commit":"598e0376c6497929a5abc1aa7c7ea1dada5de589","subject":"[OVN] Wait for WaitForDataPathBindingCreateEvent event in functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/598e0376c6497929a5abc1aa7c7ea1dada5de589"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-26 03:34:32.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-06-30 13:30:23.000000000","tz":0},"subject":"port_forwarding: validate args before invoking db update","message":"port_forwarding: validate args before invoking db update\n\nAdd validator to update_floatingip_port_forwarding so codepath does not\nattempt performing invalid database operation. With that, operation fails\nright away, with a hint on the offending argument(s).\n\nChange-Id: I8284b22c5d691bfd9eadeb8590c3d4b27d261b04\nCloses-Bug: #1878299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33e019ab03c2edb859b29e3bf7144bfcca49a630"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33e019ab03c2edb859b29e3bf7144bfcca49a630"}]},"branch":"refs/heads/master"},"af44a9f81a308d4c6ac576611c114014d997219b":{"kind":"REWORK","_number":9,"created":"2020-07-06 20:40:54.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/45/738145/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/738145/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/738145/9"}}},"commit":{"parents":[{"commit":"598e0376c6497929a5abc1aa7c7ea1dada5de589","subject":"[OVN] Wait for WaitForDataPathBindingCreateEvent event in functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/598e0376c6497929a5abc1aa7c7ea1dada5de589"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-26 03:34:32.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-07-06 19:45:12.000000000","tz":-240},"subject":"port_forwarding: validate args before invoking db update","message":"port_forwarding: validate args before invoking db update\n\nAdd validator to update_floatingip_port_forwarding so codepath does not\nattempt performing invalid database operation. With that, operation fails\nright away, with a hint on the offending argument(s).\n\nChange-Id: I8284b22c5d691bfd9eadeb8590c3d4b27d261b04\nCloses-Bug: #1878299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af44a9f81a308d4c6ac576611c114014d997219b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af44a9f81a308d4c6ac576611c114014d997219b"}]},"branch":"refs/heads/master"},"79313bf9ede1b0d3c1837f4277439e2bd8c845b1":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-07-07 14:56:17.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/45/738145/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/738145/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/738145/10"}}},"commit":{"parents":[{"commit":"75970c4cda175d246b3519171e8cd96da2c65776","subject":"Fix pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/75970c4cda175d246b3519171e8cd96da2c65776"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-26 03:34:32.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-07-07 14:56:17.000000000","tz":0},"subject":"port_forwarding: validate args before invoking db update","message":"port_forwarding: validate args before invoking db update\n\nAdd validator to update_floatingip_port_forwarding so codepath does not\nattempt performing invalid database operation. With that, operation fails\nright away, with a hint on the offending argument(s).\n\nChange-Id: I8284b22c5d691bfd9eadeb8590c3d4b27d261b04\nCloses-Bug: #1878299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79313bf9ede1b0d3c1837f4277439e2bd8c845b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79313bf9ede1b0d3c1837f4277439e2bd8c845b1"}]},"branch":"refs/heads/master"},"f649a9673d665d35e17a2f78dcacea90a9896890":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2020-07-07 22:06:38.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/45/738145/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/738145/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/738145/11"}}},"commit":{"parents":[{"commit":"a3c7aad790bd50a0c167ab5224eede7949737693","subject":"Merge \"ovn: Add TODO to remove Port Groups migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3c7aad790bd50a0c167ab5224eede7949737693"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-26 03:34:32.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-07-07 22:06:38.000000000","tz":0},"subject":"port_forwarding: validate args before invoking db update","message":"port_forwarding: validate args before invoking db update\n\nAdd validator to update_floatingip_port_forwarding so codepath does not\nattempt performing invalid database operation. With that, operation fails\nright away, with a hint on the offending argument(s).\n\nChange-Id: I8284b22c5d691bfd9eadeb8590c3d4b27d261b04\nCloses-Bug: #1878299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f649a9673d665d35e17a2f78dcacea90a9896890"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f649a9673d665d35e17a2f78dcacea90a9896890"}]},"branch":"refs/heads/master"},"8c867da72816aea853805194e70f4ebd23a46955":{"kind":"REWORK","_number":12,"created":"2020-07-09 10:06:15.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/45/738145/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/738145/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/738145/12"}}},"commit":{"parents":[{"commit":"a3c7aad790bd50a0c167ab5224eede7949737693","subject":"Merge \"ovn: Add TODO to remove Port Groups migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3c7aad790bd50a0c167ab5224eede7949737693"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-26 03:34:32.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-07-09 10:05:27.000000000","tz":-240},"subject":"port_forwarding: validate args before invoking db update","message":"port_forwarding: validate args before invoking db update\n\nAdd validator to update_floatingip_port_forwarding so codepath does not\nattempt performing invalid database operation. With that, operation fails\nright away, with a hint on the offending argument(s).\n\nChange-Id: I8284b22c5d691bfd9eadeb8590c3d4b27d261b04\nCloses-Bug: #1878299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c867da72816aea853805194e70f4ebd23a46955"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c867da72816aea853805194e70f4ebd23a46955"}]},"branch":"refs/heads/master"},"76c25df66582607408d67393869ecbc3ede97966":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2020-07-10 08:55:48.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/45/738145/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/738145/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/738145/13"}}},"commit":{"parents":[{"commit":"5999716cfc4a00ac426e016eabbb51247ba0b190","subject":"Merge \"Implement \"BridgeDevice\" with Pyroute2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5999716cfc4a00ac426e016eabbb51247ba0b190"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-26 03:34:32.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-07-10 08:55:48.000000000","tz":0},"subject":"port_forwarding: validate args before invoking db update","message":"port_forwarding: validate args before invoking db update\n\nAdd validator to update_floatingip_port_forwarding so codepath does not\nattempt performing invalid database operation. With that, operation fails\nright away, with a hint on the offending argument(s).\n\nChange-Id: I8284b22c5d691bfd9eadeb8590c3d4b27d261b04\nCloses-Bug: #1878299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/76c25df66582607408d67393869ecbc3ede97966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/76c25df66582607408d67393869ecbc3ede97966"}]},"branch":"refs/heads/master"},"f30368bfa1fc37fddf89e42a03b4636529aed00d":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2020-07-19 13:19:40.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/45/738145/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/738145/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/738145/14"}}},"commit":{"parents":[{"commit":"a7010e9c4e844940ef33aae3309ca4e323bef78e","subject":"Merge \"[OVN] DevStack: Make the transition of DevStack module easier\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a7010e9c4e844940ef33aae3309ca4e323bef78e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-26 03:34:32.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-07-19 13:19:40.000000000","tz":0},"subject":"port_forwarding: validate args before invoking db update","message":"port_forwarding: validate args before invoking db update\n\nAdd validator to update_floatingip_port_forwarding so codepath does not\nattempt performing invalid database operation. With that, operation fails\nright away, with a hint on the offending argument(s).\n\nChange-Id: I8284b22c5d691bfd9eadeb8590c3d4b27d261b04\nCloses-Bug: #1878299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f30368bfa1fc37fddf89e42a03b4636529aed00d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f30368bfa1fc37fddf89e42a03b4636529aed00d"}]},"branch":"refs/heads/master"},"f3797403486ba43100ff9ec8917d38c8723e12b2":{"kind":"REWORK","_number":15,"created":"2020-07-19 13:32:23.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/45/738145/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/738145/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/738145/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/738145/15"}}},"commit":{"parents":[{"commit":"a7010e9c4e844940ef33aae3309ca4e323bef78e","subject":"Merge \"[OVN] DevStack: Make the transition of DevStack module easier\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a7010e9c4e844940ef33aae3309ca4e323bef78e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-26 03:34:32.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-07-19 13:28:16.000000000","tz":-240},"subject":"port_forwarding: validate args before invoking db update","message":"port_forwarding: validate args before invoking db update\n\nAdd validator to update_floatingip_port_forwarding so codepath does not\nattempt performing invalid database operation. With that, operation fails\nright away, with a hint on the offending argument(s).\n\nChange-Id: I8284b22c5d691bfd9eadeb8590c3d4b27d261b04\nCloses-Bug: #1878299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3797403486ba43100ff9ec8917d38c8723e12b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3797403486ba43100ff9ec8917d38c8723e12b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
