)]}'
{"id":"openstack%2Fneutron~738744","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I231e624f460510decc6d2237040c8bef207e2e8e","project":"openstack/neutron","branch":"stable/queens","topic":"improve-ovs-agent-logging-stable/ussuri-stable/train-stable/stein-stable/rocky-stable/queens","hashtags":[],"change_id":"I231e624f460510decc6d2237040c8bef207e2e8e","subject":"Improve log message when port losts its vlan tag","status":"MERGED","created":"2020-07-01 07:15:13.000000000","updated":"2020-07-08 00:23:17.000000000","submitted":"2020-07-08 00:23:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738744-1594167797794-de16a12a","meta_rev_id":"c5b79d74af2d72809506a5b4fc6cae8f769e3ec2","_number":738744,"virtual_id_number":738744,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-07-07 15:02:34.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-08 00:23:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-07 14:52:42.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-07 14:55:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-07-07 22:44:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-07-07 22:44:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-07 14:52:42.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-07-07 14:55:00.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-07-07 15:02:34.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-07 22:44:59.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-07-08 00:23:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e70f4243e83f07e2552b2be8a063445385cd8de","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-07-01 07:15:13.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"f09efe086470b04e3e716f774b3facb939e0eb44","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-07-01 07:25:37.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/738744/1 : FAILURE in 8m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a9948b0c554b72987ba6f1f3d6fa0aaf883440a6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-01 07:35:36.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/44/738744/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/44/738744/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"836d0cf1db99781a59cd3433bf607ab93788cfeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 08:42:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/4eb683514fa14cb2a39b6210fae6ac9f : FAILURE in 16m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/61d4274961f84f0cac16f60c0e0e7dbb : FAILURE in 24m 10s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/cdda15362b464ce19d804df1ffcc076d : FAILURE in 14m 26s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/5b9bf958de8b445abbf96c43c8ceb4c3 : FAILURE in 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f06d16b633a54bf0b144d75335b91277 : SUCCESS in 14m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4ced4e047cdd46c4b5ff1cd5aa33bf5f : SUCCESS in 25m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0b9ecaf6b5a24863a4d56c55f6b696dd : SUCCESS in 28m 08s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/dc66f31ebe8a47eb9924d358421c65c7 : SUCCESS in 8m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/790ffc6279614457befe23fcd6149027 : FAILURE in 10m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/67a9fa5528a5446d96c75e6de9ac5e4f : FAILURE in 15m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4d8601b63da4754a612004f4f7041f9 : FAILURE in 12m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f4e4094dcbcb4afdb555d3fc85b585be : FAILURE in 9m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ecdeaa7c616d4a469432944e8e492ce8 : FAILURE in 13m 21s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/cbed9a27e57842b8b999ae679a908c03 : FAILURE in 13m 26s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e84c70e7add249b88a49fe181bf1187c : FAILURE in 17m 10s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/aa453efb2e6e4ebe8b95dbac9b5c808d : FAILURE in 14m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff3c15a1c6d6452292f1f8427d13d19e : FAILURE in 19m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bec31f2100de4656bb43da1d7196aa0b : FAILURE in 18m 06s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7d965313d50e488294178c938521aa95 : FAILURE in 23m 16s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/73603154c39e4793ba222629fa72fea9 : FAILURE in 18m 29s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/1413e5c8e041442290525c20aad903e1 : FAILURE in 14m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bf07f748d6e4a9bb9c85ccb5734627f : SUCCESS in 30m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a3a40f5f42618ecd0499d0da2a30f0745bcf1e40","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-07-07 14:44:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"710d7524c31b99c8d4df575cfc9016f9fef60b9f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-07-07 14:52:42.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/738744/2 : FAILURE in 7m 46s","accounts_in_message":[],"_revision_number":2},{"id":"bc601d1258f05f7673adee6bf18ac020e85780b7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-07 14:53:13.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/44/738744/2/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/44/738744/2/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"ff04a07cbb13320984e68612dd7193dfcb4d7c30","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-07-07 14:55:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"45f6fa56f152eec4f3b7ab766b08d1296330bf5f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-07 15:02:34.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/44/738744/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/44/738744/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"2be4fa928dd0505c722db9230199f41dc1c014b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 19:46:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/b5ee11741f1c4295861c521cfdb8952d : SUCCESS in 1h 22m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/0d3583f28d144895a51b546d0f5d1125 : FAILURE in 1h 53m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/1f4482af7d794cacb951fd753afcdd01 : SUCCESS in 1h 00m 10s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/b15035f27bfd44b09ca0a07aadd9b544 : SUCCESS in 56m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d63efe6286c04b99b8928e8aef752950 : SUCCESS in 13m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a0ce8a118fb4a2ea41d79b719e922bd : SUCCESS in 27m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/02a236f7d7634189977ee4e19e1edfb0 : SUCCESS in 29m 01s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/343d87f2e223437cb4c77e83f168aa21 : SUCCESS in 7m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a73767c2b3f2469f9aae212c3b47412d : SUCCESS in 1h 15m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26e592cfd0d349ba9f035a60381c3cff : SUCCESS in 1h 17m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/da583e272d9446dfa78628591143b5bd : SUCCESS in 47m 43s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/07b6a3bbbabf4ecfa80177f938a74df6 : SUCCESS in 1h 46m 32s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/17b8d498e63b4ba0829600f8d05417bf : SUCCESS in 1h 27m 35s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8f7414961bcc4825b65bf4f06d9b7bde : SUCCESS in 1h 24m 59s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/644564d181d84c1f9daca2e549500599 : FAILURE in 21m 41s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/699f633f875b4118835a09ce87ddcf54 : SUCCESS in 1h 47m 37s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/c647fa362c974f5699c91ffa41eccd50 : SUCCESS in 1h 22m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df492ac9b17c44558951f5c3674581df : SUCCESS in 30m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c369893595f0b92f96cffd4b1eb97393607fcacf","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-07-07 22:44:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bf44de071d5c49410678bef50340ad3346ff290d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 22:45:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"015a62d2150f50f5e91f8af889cf12665e197677","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 00:23:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/e17ffddecd9b4525897ca0d2b67de756 : SUCCESS in 44m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4514b152d5364ff2b0f31058dfbc02d4 : SUCCESS in 14m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18210ac1638b427192492512500a61fa : SUCCESS in 30m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3ff9d345b2bf4bf783fd675fbdfd1527 : SUCCESS in 26m 16s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4f58cff90769496c8ede10f9adaa9388 : SUCCESS in 7m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0a49c0885f0c4aaaa4631797c6213a69 : SUCCESS in 1h 23m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d61390b9947d44f9929de26fdb541f38 : SUCCESS in 1h 32m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8a804e7a58ca4cff8a817e0c641c3e4f : SUCCESS in 39m 26s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/79c5941988964af4bbd551195835cdfe : SUCCESS in 1h 19m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b241892c94024652bc83cd4d2ff6fb3d : SUCCESS in 1h 21m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1334df5dc6674f14a90db96b1644bfb4 : SUCCESS in 31m 52s","accounts_in_message":[],"_revision_number":2},{"id":"2632c07210c0668ef93ceca100ca87d09d524289","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 00:23:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0ee2fc0e15621461cd0f342e8eb43bf633c8ab42","revisions":{"30ff804b74781e3a6c44393de25bf2dc8122529e":{"kind":"REWORK","_number":1,"created":"2020-07-01 07:15:13.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/44/738744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/738744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/738744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/738744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/738744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/738744/1"}}},"commit":{"parents":[{"commit":"625505ecc77c36b6269af4ae94ac349e1677a8af","subject":"Merge \"Allow usage of legacy 3rd-party interface drivers\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/625505ecc77c36b6269af4ae94ac349e1677a8af"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-30 08:20:28.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-07-01 07:15:13.000000000","tz":0},"subject":"Improve log message when port losts its vlan tag","message":"Improve log message when port losts its vlan tag\n\nIn some cases it may be useful to log new vlan tag which is found\non the port when it losts old vlan tag which should is expected to\nbe there.\nSo this patch adds such value to the log message.\n\nTrivialFix\n\nChange-Id: I231e624f460510decc6d2237040c8bef207e2e8e\n(cherry picked from commit 3ac63422eaebf6edfd581571406266a07a85b883)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30ff804b74781e3a6c44393de25bf2dc8122529e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30ff804b74781e3a6c44393de25bf2dc8122529e"}]},"branch":"refs/heads/stable/queens"},"0ee2fc0e15621461cd0f342e8eb43bf633c8ab42":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-07 14:44:20.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/44/738744/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/738744/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/738744/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/738744/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/738744/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/738744/2"}}},"commit":{"parents":[{"commit":"774b3ff89e5ec01494d7ca66773aba8984f0cd80","subject":"Fix queens gates, multiple fixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/774b3ff89e5ec01494d7ca66773aba8984f0cd80"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-30 08:20:28.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-07-07 14:44:20.000000000","tz":0},"subject":"Improve log message when port losts its vlan tag","message":"Improve log message when port losts its vlan tag\n\nIn some cases it may be useful to log new vlan tag which is found\non the port when it losts old vlan tag which should is expected to\nbe there.\nSo this patch adds such value to the log message.\n\nTrivialFix\n\nChange-Id: I231e624f460510decc6d2237040c8bef207e2e8e\n(cherry picked from commit 3ac63422eaebf6edfd581571406266a07a85b883)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ee2fc0e15621461cd0f342e8eb43bf633c8ab42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ee2fc0e15621461cd0f342e8eb43bf633c8ab42"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
