)]}'
{"id":"openstack%2Fneutron~740977","triplet_id":"openstack%2Fneutron~master~Ibd3343fc74a9a78f3ec974c204a5bbf5310758b7","project":"openstack/neutron","branch":"master","topic":"bug/1887523","hashtags":[],"change_id":"Ibd3343fc74a9a78f3ec974c204a5bbf5310758b7","subject":"quota: use native oslo.db retry","status":"ABANDONED","created":"2020-07-14 14:32:49.000000000","updated":"2020-09-25 15:39:45.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5748e7212855591f755368fe3e15c02071fd0d4b","_number":740977,"virtual_id_number":740977,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-07-29 13:36:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"date":"2020-07-28 14:29:27.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-14 20:55:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":-1,"date":"2020-07-28 10:06:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":-1,"date":"2020-07-29 12:06:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},{"date":"2020-07-16 15:19:58.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"date":"2020-07-20 05:35:34.000000000","_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-14 14:42:42.000000000","updated_by":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"reviewer":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"state":"REVIEWER"},{"updated":"2020-07-14 17:44:06.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2020-07-14 17:44:28.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-07-14 20:55:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-16 15:19:58.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-07-20 05:35:34.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2020-07-28 10:06:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-07-29 12:06:22.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-07-29 13:36:54.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1280c9254827c1b4ee15b5636f370ff6272b2134","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-07-14 14:32:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55f2b86f473fd0b813bb4700d8ea294cad287cd7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-14 17:18:00.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/77/740977/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/77/740977/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a0331546cdf197bdbf134d372edc6f850bcc8457","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 17:20:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/73633d4f501740dab7cfcdbc8592543b : SUCCESS in 50m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6480ced0840b44d3956e4eda95fd604c : SUCCESS in 33m 15s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0616c2d3eb654e439cdfc3add0f8e95a : FAILURE in 48m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/18a1bafcb32e47a4b0700bb0699379d9 : FAILURE in 1h 34m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0758cbd561634f65a4f25ccc7fef036f : SUCCESS in 1h 28m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/197f55fdf3f04184bc0d2767b65e1d0a : SUCCESS in 1h 11m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cd1ea6c751e42669fd12c53d6399c72 : SUCCESS in 38m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/560f51207e1c42b4b3e89657d5a5ff67 : SUCCESS in 30m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8181341894749de9ea6f37010e06c53 : SUCCESS in 12m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb18a47b74e941198a62b4048619ecc5 : SUCCESS in 29m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/984cb99fd3074e43b534117113e88196 : SUCCESS in 29m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af8f7382db2043dc953ff7263c4f33af : SUCCESS in 11m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b491d87ff3e143b49148fd23c9ad96b2 : SUCCESS in 48m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dfd351f938b744aba81bec268da4c6cf : SUCCESS in 46m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/73977a0d1f584ee8930c5ce91b9408c8 : SUCCESS in 1h 14m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/07d599c479084734ad5f79512f447cd5 : FAILURE in 1h 16m 01s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a5b0e3ba2021486dbd732018e905c37a : SUCCESS in 2h 04m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a2fb7355a3064af5b303dd3fcebe7350 : SUCCESS in 1h 12m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7afdbf1817c1487ba0b98448d56d91a6 : SUCCESS in 1h 07m 40s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/dfc6dcbda4a945be9a2efabce7327ecc : SUCCESS in 1h 18m 16s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e54a3c880774c7c8558404073cdd86c : SUCCESS in 1h 25m 47s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4b5fd69ac46f49dbbf358ccb8839cff7 : SUCCESS in 1h 21m 12s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1c50f8942a9a420f8e4ada3a42e27acc : SUCCESS in 1h 39m 51s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e7fc0c0b09654d5ab772e3512b3ae445 : SUCCESS in 52m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fe7af8290eb84c1399dcd5305e08c8c4 : SUCCESS in 1h 14m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/170cb6337bdd428fae45fb1204d595e3 : SUCCESS in 2h 07m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87af9512637f4429b324b3e6c3cdd45f : SUCCESS in 51m 32s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/4f97c869a2064cc9847355cbe5562f0a : SUCCESS in 1h 03m 56s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/42b95fb21e6641ce8cf6b18152b0722e : SUCCESS in 56m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6035b024131345a7bd73f2b40efa63d7 : SUCCESS in 33m 31s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a9ef86166d644c92a250d7c435ed48db : SUCCESS in 1h 13m 44s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/09c05958b9c54ac8931a81d363c38b4e : SUCCESS in 1h 05m 14s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/41ac4422826d4869a63f8a956d0b2c91 : SUCCESS in 1h 34m 53s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/99fc04e69b434cc584f9e430c528b354 : SUCCESS in 2h 10m 15s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3ee8e7be69c94db9b0e3a2e7b79e505c : SUCCESS in 1h 55m 20s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b05e12999feb42ab8c14c549a14415b5 : FAILURE in 2h 03m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7aed8be25213d762d8c34245ba6282b46e10948c","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-07-14 17:31:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dc7e08bda4c124ee6349b9840572107d31f074f7","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-07-14 17:55:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e58ba708fd3dc12e9be001e37e91b4bfb36ecfcf","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-07-14 18:27:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4af1b74fbb72847bff89e8d1e1a175c5e696348f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 20:55:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e33bcee1ea014990b3eae3b659116feb : SUCCESS in 58m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/060c933dd7214743bd28f2f4af672d5f : SUCCESS in 38m 28s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/af49838af7f744dcb2c9462486d231a7 : FAILURE in 1h 04m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c018bda02ab64d268693b31bce8ddf3f : SUCCESS in 1h 07m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c8efca40539d441b88ee9798d233d5dd : SUCCESS in 57m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f69b0af7adca49ccbdd943c203f208a2 : FAILURE in 1h 01m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c149b3f768164b4191c4642a6050f72a : SUCCESS in 34m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e98bbb70bc884169b61a78ee58706b90 : SUCCESS in 31m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6203763d04f04119a1021bb121628f42 : SUCCESS in 13m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b856e6e90924d23bc2246b66ada305a : SUCCESS in 30m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c93c6cd1d474dae83f8bca4bd1aadd2 : SUCCESS in 28m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c36e4e6bf0e1495b8ae2580eef9ce2fe : SUCCESS in 8m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7e12fe4ef1e74284844b8a1cfad1796d : SUCCESS in 56m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/83fe302233fc407ebab57bb7219f30ca : SUCCESS in 54m 13s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/29c5681dcd82465e9fd711e9f1fcfc67 : SUCCESS in 1h 14m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b426fbdb9f23410ea3a5353eb7554fee : SUCCESS in 1h 21m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/39e8dd1d137a40ad8fce27c98fbc83c2 : SUCCESS in 1h 47m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb633797c45f452b8a36fc87d0a6e5df : SUCCESS in 1h 07m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c61cfcbe94ea40e0a09df274db95ce3f : SUCCESS in 1h 09m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f5c0cc9698444a28825dbcdf64c3367f : SUCCESS in 1h 12m 36s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/05fa1df68eaf4d87b90c3fb73ea237db : SUCCESS in 1h 10m 42s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/77a6c08b350f4052a17b79b51b038ad1 : SUCCESS in 1h 36m 11s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/27d9bd58af714508b19e01e4e5f167c9 : FAILURE in 1h 45m 13s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/35140ca32b7545a09e2ac921bf5db84f : SUCCESS in 1h 22m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b30d12af02d344d9b4a69a88c92e9fc1 : SUCCESS in 1h 07m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/af289cf26e54415a821d391dc61c5c10 : SUCCESS in 2h 09m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98147782667948e8ba89ee72781d2e1f : SUCCESS in 59m 26s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/c26c5b041b7a4dff866ce84b00481b7d : SUCCESS in 1h 14m 55s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b50ef8f93af5464e99da3756ccc230f4 : SUCCESS in 1h 18m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/afbbad3b0ab84ee6ae158a4c5309f751 : SUCCESS in 41m 34s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7dcbdc7c78754753ab6c3b79eeec731a : SUCCESS in 1h 20m 35s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b3e18bb091a04df1944c605c8421536e : SUCCESS in 1h 34m 09s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b8f1ab589870412fb203a9e4bca49c7b : FAILURE in 2h 06m 45s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e84de6685425411f87e2af5dca8c53d6 : SUCCESS in 2h 22m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/644e5fdc13d44bd68b272e5df2d79378 : SUCCESS in 1h 51m 57s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e4c8f2361b9d4a808eff90be7b31f974 : SUCCESS in 1h 59m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"416dc565956a67edc8a4f3daa8f22f438992ef8c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-14 23:07:19.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/77/740977/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/77/740977/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"09b82bbb7598c11970241bc51b11d50f36d67660","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-28 10:06:28.000000000","message":"Patch Set 3: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"acff96bd684cbab39049cec699e05dd7e4133cc2","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-07-28 14:29:27.000000000","message":"Patch Set 3:\n\nsorry -- unfortuantely i don\u0027t think i have time to actually pursue this and push this forward at he moment.  :(","accounts_in_message":[],"_revision_number":3},{"id":"c9c2db1f35652dacf5a2265dfab424fe751e4e2a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-07-29 12:06:22.000000000","message":"Patch Set 3: Code-Review-1\n\nIssues in scenario jobs, like e.g. https://b2793fc951b6cda052a8-c455553d95560f1580667e93cc59b7bd.ssl.cf2.rackcdn.com/740977/3/check/neutron-tempest-plugin-scenario-linuxbridge/af49838/controller/logs/screen-q-svc.txt seems like may be related to this change. Please check it before rechecking","accounts_in_message":[],"_revision_number":3},{"id":"9436638fed4dab82329a01281123c3c1dc494ea4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-29 13:36:54.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/77/740977/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/77/740977/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"86d42f05277e7845f0e4937ec4736d5ac3e58472","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-25 15:39:45.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"62ca8ce5d5fc1a6480c03fc8486699fd101068b2","revisions":{"9d9d399793f76e73d11d3759c38068bee46ad4aa":{"kind":"REWORK","_number":1,"created":"2020-07-14 14:32:49.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/77/740977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/77/740977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/740977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/740977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/740977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/77/740977/1"}}},"commit":{"parents":[{"commit":"73557abefcba1c6ce0cef709d1082674c0217485","subject":"Merge \"Optionally use admin powers when deleting DNS records\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73557abefcba1c6ce0cef709d1082674c0217485"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-07-14 14:24:39.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-07-14 14:32:47.000000000","tz":-240},"subject":"quota: use native oslo.db retry","message":"quota: use native oslo.db retry\n\nThere are some issues with setting quota usages to dirty which cause\ndeadlocks that have been worked around by adding the global retry hook,\nthis has actually caused more issues in our larger scale deployments and\nended up giving retries without proper stacktraces and missing\nsavepoints.\n\nAll OpenStack projects adopt native use of oslo_db retry function\nnatively, the reason for the use of the current Neutron model is to\navoid nested retries, which really is a workaround rather than solving\nthe fundamental issue of clearing out the queries that deadlock.\n\nThis patch removes any retries from SELECT operations, as those will\nlikely not really be affected much (and don\u0027t need to be retried).  We\nwill be retrying the UPDATE operation anyways (if it\u0027s deadlocked on\nthat).\n\nAll of the update operations have been updated to use the oslo_db retry\ndecorator, we can probably drop some of those eventually (even a project\nlike Nova has very little of those (in ~250-ish database API functions,\nonly 13 are actually tagged to retry on deadlock).\n\nIt also removes the API-level retry for all operations, as it\u0027s much\nbetter to hit and narrow down on the deadlock-able operations.  This\nshould eventually be moved onto all other components but for now, this\npatch covers quota as it\u0027s a hard hit (plugging 250 ports concurrently\nresults in 504s and deadlocks everywhere).\n\nChange-Id: Ibd3343fc74a9a78f3ec974c204a5bbf5310758b7\nRelated-Bug: #1777965\nRelated-Bug: #1771293\nRelated-Bug: #1887523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d9d399793f76e73d11d3759c38068bee46ad4aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d9d399793f76e73d11d3759c38068bee46ad4aa"}]},"branch":"refs/heads/master"},"ddec81e448d256b20e826f7bf2e273bd7c6a0384":{"kind":"REWORK","_number":2,"created":"2020-07-14 17:55:03.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/77/740977/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/77/740977/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/740977/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/740977/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/740977/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/77/740977/2"}}},"commit":{"parents":[{"commit":"73557abefcba1c6ce0cef709d1082674c0217485","subject":"Merge \"Optionally use admin powers when deleting DNS records\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73557abefcba1c6ce0cef709d1082674c0217485"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-07-14 14:24:39.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-07-14 17:54:15.000000000","tz":-240},"subject":"quota: use native oslo.db retry","message":"quota: use native oslo.db retry\n\nThere are some issues with setting quota usages to dirty which cause\ndeadlocks that have been worked around by adding the global retry hook,\nthis has actually caused more issues in our larger scale deployments and\nended up giving retries without proper stacktraces and missing\nsavepoints.\n\nAll OpenStack projects adopt native use of oslo_db retry function\nnatively, the reason for the use of the current Neutron model is to\navoid nested retries, which really is a workaround rather than solving\nthe fundamental issue of clearing out the queries that deadlock.\n\nThis patch removes any retries from SELECT operations, as those will\nlikely not really be affected much (and don\u0027t need to be retried).  We\nwill be retrying the UPDATE operation anyways (if it\u0027s deadlocked on\nthat).\n\nAll of the update operations have been updated to use the oslo_db retry\ndecorator, we can probably drop some of those eventually (even a project\nlike Nova has very little of those (in ~250-ish database API functions,\nonly 13 are actually tagged to retry on deadlock).\n\nIt also removes the API-level retry for all operations, as it\u0027s much\nbetter to hit and narrow down on the deadlock-able operations.  This\nshould eventually be moved onto all other components but for now, this\npatch covers quota as it\u0027s a hard hit (plugging 250 ports concurrently\nresults in 504s and deadlocks everywhere).\n\nThis patch also makes some changes to the quota driver to avoid having\ncalls for retries there as they are handled at a much lower level inside\nthe quota database API.\n\nChange-Id: Ibd3343fc74a9a78f3ec974c204a5bbf5310758b7\nRelated-Bug: #1777965\nRelated-Bug: #1771293\nRelated-Bug: #1887523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ddec81e448d256b20e826f7bf2e273bd7c6a0384"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ddec81e448d256b20e826f7bf2e273bd7c6a0384"}]},"branch":"refs/heads/master"},"62ca8ce5d5fc1a6480c03fc8486699fd101068b2":{"kind":"REWORK","_number":3,"created":"2020-07-14 18:27:45.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/77/740977/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/77/740977/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/740977/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/740977/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/740977/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/77/740977/3"}}},"commit":{"parents":[{"commit":"73557abefcba1c6ce0cef709d1082674c0217485","subject":"Merge \"Optionally use admin powers when deleting DNS records\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73557abefcba1c6ce0cef709d1082674c0217485"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-07-14 14:24:39.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-07-14 18:27:38.000000000","tz":-240},"subject":"quota: use native oslo.db retry","message":"quota: use native oslo.db retry\n\nThere are some issues with setting quota usages to dirty which cause\ndeadlocks that have been worked around by adding the global retry hook,\nthis has actually caused more issues in our larger scale deployments and\nended up giving retries without proper stacktraces and missing\nsavepoints.\n\nAll OpenStack projects adopt native use of oslo_db retry function\nnatively, the reason for the use of the current Neutron model is to\navoid nested retries, which really is a workaround rather than solving\nthe fundamental issue of clearing out the queries that deadlock.\n\nThis patch removes any retries from SELECT operations, as those will\nlikely not really be affected much (and don\u0027t need to be retried).  We\nwill be retrying the UPDATE operation anyways (if it\u0027s deadlocked on\nthat).\n\nAll of the update operations have been updated to use the oslo_db retry\ndecorator, we can probably drop some of those eventually (even a project\nlike Nova has very little of those (in ~250-ish database API functions,\nonly 13 are actually tagged to retry on deadlock).\n\nIt also removes the API-level retry for all operations, as it\u0027s much\nbetter to hit and narrow down on the deadlock-able operations.  This\nshould eventually be moved onto all other components but for now, this\npatch covers quota as it\u0027s a hard hit (plugging 250 ports concurrently\nresults in 504s and deadlocks everywhere).\n\nThis patch also makes some changes to the quota driver to avoid having\ncalls for retries there as they are handled at a much lower level inside\nthe quota database API.\n\nChange-Id: Ibd3343fc74a9a78f3ec974c204a5bbf5310758b7\nRelated-Bug: #1777965\nRelated-Bug: #1771293\nRelated-Bug: #1887523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62ca8ce5d5fc1a6480c03fc8486699fd101068b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62ca8ce5d5fc1a6480c03fc8486699fd101068b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
