)]}'
{"id":"openstack%2Fneutron~741017","triplet_id":"openstack%2Fneutron~master~I98d2cd98c06066def324f22f2cda15ea326d79c8","project":"openstack/neutron","branch":"master","topic":"test_osloprivsep","hashtags":[],"change_id":"I98d2cd98c06066def324f22f2cda15ea326d79c8","subject":"[DNM] Test oslo.privsep patch","status":"ABANDONED","created":"2020-07-14 17:54:56.000000000","updated":"2020-07-31 17:44:48.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a34c3fd8806710e48121d72209f210f09568e8c5","_number":741017,"virtual_id_number":741017,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-07-21 17:45:06.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-07-21 18:13:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-18 03:35:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-17 20:40:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-17 20:40:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-18 03:35:12.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-21 18:13:19.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5fb0da730e0b9188523db6a9c9d1e0fde90945be","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-14 17:54:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1da7721644b6ef6da61d1d2870e67598014ae79","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-14 20:25:12.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/741017/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/741017/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"3d951631c2d34db7ab077ab216ee3db1100717af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 20:28:54.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8594f221e429482aa8ee2e85adfa42cb : SUCCESS in 56m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2c8ef95c84c84f418ee76e032bc23e6f : SUCCESS in 43m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/df6f033364364afe8c329b9457160e70 : SUCCESS in 1h 15m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/625204eb0c5d44c8bd985c9b789d8eae : SUCCESS in 1h 12m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d5163a1964574df9bebab3cce10a16e6 : SUCCESS in 2h 19m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f6931caac4d04d90906dfb3790884011 : SUCCESS in 1h 08m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2592b00c9d37468f9693f830d12d1ec5 : SUCCESS in 39m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75b95571bdbf4e7e856de43b06646686 : SUCCESS in 31m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef8ab1a39a334e60818d22d635b6405b : FAILURE in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3de39c5bff954451b9ed57a42d2bad4f : SUCCESS in 28m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbe64804d6944d9a8e8bd400b6b023b4 : SUCCESS in 25m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12df8bae95144f3da8e7b27070ba4c50 : SUCCESS in 11m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4f3ed1c0e4e14a679bbbdf0f44adeeca : SUCCESS in 54m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9d3446e3f0504c50850256be8d198741 : SUCCESS in 1h 01m 34s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d8d062be493145bb8672311be6d16288 : SUCCESS in 1h 13m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f692def2a4b64ed4bd2bb91557ae264f : SUCCESS in 1h 12m 14s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0f640141dff4483a9b0d3181040c318e : SUCCESS in 1h 45m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/14ac788c5a004dbfbdb0b967935781a7 : SUCCESS in 1h 09m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0f139d1430a1416bb87a84d30b5360e8 : SUCCESS in 1h 16m 13s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9ca7354fa499465a8ba7eede5098971c : SUCCESS in 47m 16s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c1d7ad632b64e5c825840a0c7e0602d : SUCCESS in 1h 10m 17s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/982160aaa62644f8ae3f65c22b147f4d : SUCCESS in 1h 30m 22s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2a86e0b5c9b34c8cb0941b7e20d7315b : SUCCESS in 1h 39m 48s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5c9cc63f05884ecaa6c9a85d79bf27a5 : SUCCESS in 1h 09m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3a9ceeca19b246adad8791c69615e125 : SUCCESS in 1h 20m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/94075e7e6a214b3c986588a9e954403f : SUCCESS in 2h 05m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a659469b8184150a91f4add111605e7 : SUCCESS in 1h 05m 12s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/ea2127fcdd144373ba635e2b72d5151b : FAILURE in 1h 26m 02s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4991a8c9be514d6bb94dd9fc524859ef : SUCCESS in 1h 15m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ad04aa09405343c596a8bd7390cf96a5 : SUCCESS in 43m 02s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b53a113b2d114c099297cc3abfb0f22f : SUCCESS in 1h 24m 15s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a70176e346a04917a999e480f155b01d : SUCCESS in 1h 39m 43s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a0aa27daa00e43b59832b30a9e99009b : SUCCESS in 1h 53m 50s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2259cdbdac1c42a7b207b53d5391e66a : SUCCESS in 2h 27m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/35811d4a317141e9a3a67e93b9e0a388 : SUCCESS in 1h 51m 34s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b6996c8cda024b1a9e82d55cf9a0dab9 : SUCCESS in 2h 01m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"980a9a4f33ecd15d6525dc908ca2d5b697bccc68","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-15 11:26:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fb5438182e3747b2c1b495ef124035e974510bec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-15 12:33:59.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/741017/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/741017/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"eca133025163de3a2e2695c63ca8250af5e13729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 12:47:40.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"984687c937ef854bd272c588e8793cbeb6846a95","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-15 12:49:51.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/17/741017/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"f93eaf33ef0f139cd1cfd353482c58c15d4eef79","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-15 14:48:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fd66bed6a9ae5a21043409464d4b1b2a4e7705bb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-15 15:50:54.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/741017/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/741017/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"c6d0f5daf1087f048c2312f9106ef19468813879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 21:24:01.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c3e7e0a03d1846709b10e81625217fac : SUCCESS in 1h 01m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f94ef844ce3e4912be507fb29c58aafd : SUCCESS in 45m 44s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5d95131297ed4aec98c501b568826cf3 : SUCCESS in 1h 09m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/95d697be8ed2432bbc3f63ba0db1301a : SUCCESS in 1h 14m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/095aaf5901f54e038e2c70c8ac3b3793 : SUCCESS in 1h 14m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d7accf92812a4a09a8182fce06fe1116 : SUCCESS in 1h 05m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dde50787cfc144a1bcd0ee194629837b : SUCCESS in 38m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1625814be8404e789019dcb892568997 : SUCCESS in 26m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1399413ab20544c5bfc9fdfb924d2555 : FAILURE in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cb877e63d3d44b09f9655276617d75a : SUCCESS in 26m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2530fcf030bf490a8e640f412c9bec08 : SUCCESS in 24m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1158c1cf63b541f79c7b0d923b2570f7 : SUCCESS in 12m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b1047abab4a54364b11645daa8cd959d : SUCCESS in 4m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/69235bbcda7e4d78ad80d3fd48ee3ce6 : SUCCESS in 51m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9422673e592741c7b764c675f0880027 : SUCCESS in 51m 37s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ad5eb6b2b8d54d79a82f2f8514d34b5b : SUCCESS in 1h 09m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a4323fcb5cb4a429df21314c8fc6e72 : SUCCESS in 1h 12m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c5e20573a1e24a33958ee8e3b1cb7c11 : SUCCESS in 1h 59m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/38278c0bae3040d2993422fca72878fd : SUCCESS in 1h 09m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1b3cda0c26174464ba59b23e76f5b9f0 : SUCCESS in 1h 00m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4db7ca9f6d1d42d99144f6aec1c40580 : SUCCESS in 50m 28s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3700a1ea0f2348eb9431dccc89ec6e55 : SUCCESS in 50m 30s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/60bbea8bb5a44a2084cc83dffaa4c998 : SUCCESS in 1h 23m 51s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d0d4d0b18bb045eca252781e53921668 : FAILURE in 1h 05m 20s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/34023c9297da4fa0a7b76455532fd505 : SUCCESS in 1h 28m 44s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f56931f4e72548338fdfebca96aa50ca : SUCCESS in 1h 15m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f01cd3e7c00a47649c3d777f27881124 : SUCCESS in 2h 15m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/288d3ec44a65401da269f2c3286a4862 : SUCCESS in 1h 13m 42s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/8260e84ef0bc4faca2e22252a6f9ce41 : SUCCESS in 1h 10m 00s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ce1500b463a14febae7d3c1ebc4d107a : SUCCESS in 1h 45m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dfab4fcaee1d4c9abfb72bfa9a5953be : SUCCESS in 44m 08s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1c2389db6fe54beaae1a04dbd4900104 : SUCCESS in 1h 29m 01s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/48cdc4a816c14c83ad568512b7e7459e : SUCCESS in 1h 45m 20s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/022a4e0738294f42b0872a5d649b2a42 : SUCCESS in 2h 00m 59s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7b6b1643cd0f4db398a08e73412c5df4 : SUCCESS in 1h 43m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8ce752d0d84b4c0f807e78761ada91ed : SUCCESS in 1h 42m 56s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/9c24c8c014f74635b42b3c0cb322c109 : SUCCESS in 2h 00m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"61dc962ddc481e773e38180d957f6603c61156e9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-17 15:13:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3f9066a1ba6eb3b14958e5fe148c01ae9e77ba3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 15:14:17.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  Job neutron-functional-with-uwsgi-2 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - neutron-tempest-plugin-jobs\n        - openstack-cover-jobs\n        - openstack-lower-constraints-jobs\n        - openstack-python3-victoria-jobs\n        - publish-openstack-docs-pti\n        - periodic-stable-jobs\n        - check-requirements\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - neutron-functional\n          - neutron-functional-with-uwsgi\n          - neutron-functional-2\n          - neutron-functional-with-uwsgi-2\n          - neutron-functional-3\n          - neutron-functional-with-uwsgi-3\n          - neutron-functional-4\n          - neutron-functional-with-uwsgi-4\n          - neutron-functional-5\n          - neutron-functional-with-uwsgi-5\n          - neutron-functional-6\n          - neutron-functional-with-uwsgi-6\n          - neutron-functional-7\n          - neutron-functional-with-uwsgi-7\n          - neutron-functional-8\n          - neutron-functional-with-uwsgi-8\n          - neutron-functional-9\n          - neutron-functional-with-uwsgi-9\n          - neutron-functional-10\n          - neutron-functional-with-uwsgi-10\n  #        - neutron-fullstack\n  #        - neutron-fullstack-with-uwsgi\n  #        - neutron-rally-task\n  #        - neutron-grenade-multinode\n  #        - neutron-grenade-dvr-multinode\n  #        - neutron-tempest-linuxbridge\n  #        - neutron-tempest-with-uwsgi\n  #        - neutron-tempest-multinode-full-py3\n  #        - neutron-tempest-dvr-ha-multinode-full\n  #        - neutron-tempest-iptables_hybrid\n  #        - ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa:\n  #            voting: false\n  #            # We don\u0027t run the job on things like neutron docs-only changes\n  #            irrelevant-files: \u0026irrelevant-files\n  #              - ^(test-|)requirements.txt$\n  #              - ^.*\\.rst$\n  #              - ^doc/.*$\n  #              - ^neutron/locale/.*$\n  #              - ^neutron/tests/unit/.*$\n  #              - ^releasenotes/.*$\n  #              - ^setup.cfg$\n  #              - ^tools/.*$\n  #              - ^tox.ini$\n  #              - ^vagrant/.*$\n  #              - ^migration/.*$\n  #              - ^devstack/.*\\.sample$\n  #        - tempest-slow-py3:\n  #            irrelevant-files: *irrelevant-files\n  #        - tempest-ipv6-only:\n  #            irrelevant-files: *irrelevant-files\n  #        - neutron-tempest-with-neutron-lib-master:\n  #            voting: false\n  #            irrelevant-files: *irrelevant-files\n  #        - neutron-ovn-tempest-ovs-release\n  #        - neutron-ovn-tempest-ovs-release-ipv6-only:\n  #            voting: false\n  #        # TODO(slaweq): add this job again to the check queue when it will be\n  #        # working fine on python 3\n  #        #- networking-midonet-tempest-aio-ml2-centos-7:\n  #        #    voting: false\n  #        #    irrelevant-files: *irrelevant-files\n  #        - openstacksdk-functional-devstack-networking:\n  #            voting: false\n  #        - neutron-centos-8-tripleo-standalone\n  #        - neutron-ovn-rally-task:\n  #            voting: false\n  #        # TripleO jobs that deploy OVN.\n  #        # Note we don\u0027t use a project-template here, so it\u0027s easier\n  #        # to disable voting on one specific job if things go wrong.\n  #        # In Stein and beyond, fs010 will run using\n  #        # networking-ovn-tripleo-ci-centos-7-containers-multinode.\n  #        # If you need any support to debug these jobs in case of\n  #        # failures, please reach us on #tripleo IRC channel.\n  #        - neutron-ovn-tripleo-ci-centos-8-containers-multinode:\n  #            voting: false\n  #        - neutron-ovn-tempest-slow:\n  #            voting: false\n  #        - neutron-ovn-tempest-full-multinode-ovs-master:\n  #            voting: false\n  #        - openstack-tox-py36:  # from openstack-python3-victoria-jobs template\n  #            timeout: 3600\n  #        - openstack-tox-py38:  # from openstack-python3-victoria-jobs template\n  #            timeout: 3600\n  #        - openstack-tox-lower-constraints:  # from openstack-tox-lower-constraints template\n  #            timeout: 3600\n  #        - openstack-tox-cover:  # from openstack-cover-jobs template\n  #            timeout: 4800\n      gate:\n        jobs:\n          - neutron-functional\n          - neutron-fullstack\n          - neutron-tempest-linuxbridge\n          - neutron-tempest-iptables_hybrid\n          - neutron-tempest-multinode-full-py3\n          - neutron-grenade-multinode\n          - neutron-grenade-dvr-multinode\n  #        - tempest-slow-py3:\n  #            irrelevant-files: *irrelevant-files\n  #        - tempest-ipv6-only:\n  #            irrelevant-files: *irrelevant-files\n          - neutron-ovn-tempest-ovs-release\n          - openstack-tox-py36:  # from openstack-python3-victoria-jobs template\n              timeout: 3600\n          - openstack-tox-lower-constraints:  # from openstack-tox-lower-constraints template\n              timeout: 3600\n          #- neutron-ovn-rally-task\n          #- neutron-ovn-tripleo-ci-centos-8-containers-multinode\n  \n      experimental:\n        jobs:\n          - neutron-ovn-tempest-ovs-master\n          - neutron-ovn-grenade\n  \n      periodic:\n        jobs:\n          - neutron-functional\n          - neutron-tempest-postgres-full\n          - neutron-tempest-mariadb-full\n          - neutron-tempest-with-os-ken-master\n          - neutron-ovn-tempest-ovs-master-fedora\n\n  in \"openstack/neutron/zuul.d/project.yaml@master\", line 5, column 3","accounts_in_message":[],"_revision_number":3},{"id":"03b92a0162c29839dfa3a0b23a647e8f68bf7a36","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-17 15:16:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cf6941b0064e8d65ae2083d603688550269b3d9b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-17 16:48:04.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/741017/4/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/741017/4/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"58aaa82f42b4864729cd50c3dc37d7776a872f74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 20:40:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/15104407f94a4af0a001567d2cf09d2c : FAILURE in 33m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/666dd255a3c7453eae7d8c04116f77db : FAILURE in 30m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8f5d387b0475429f9375def50de346da : FAILURE in 18m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/89850cf0578d4f1ebbdcaa5b0704bd5f : FAILURE in 26m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/86c63cfc26cf4c0696f72bcda8081b04 : FAILURE in 29m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6c3375eb3fbf4cdc81520d89c845d9dd : SUCCESS in 46m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28ec5de8f2794483a090f200bf2f967d : SUCCESS in 21m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7820a3edc38146519ed822bd6b85514f : SUCCESS in 30m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91c4fa3708d1422bbbabb8bd54e3bf25 : SUCCESS in 12m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbd0ef9aafff4c28acd11374a70a3436 : SUCCESS in 28m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5753013e1c084197addbdf2c0e4e542e : SUCCESS in 27m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dd3c1a077ff447891a828fa6ce39c17 : SUCCESS in 9m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/516da689fdac4424bc7bccf3e5558f70 : SUCCESS in 3m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bb5998e2b2774616b8ad6090efc16130 : SUCCESS in 50m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3c3c87ec028b4b4889abc08231ff1752 : SUCCESS in 1h 01m 16s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/e44ecd729d6c45c69dab23b8fceb705b : FAILURE in 57m 55s\n- neutron-functional-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/480cd3aba6e8414f8c3b526f51d1c422 : SUCCESS in 54m 11s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/324d7e8b58464c4bb8f04c2313485f67 : SUCCESS in 56m 28s\n- neutron-functional-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/7f46a85ffa934daba4d02403c2060c73 : SUCCESS in 58m 16s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/9ce04e743d7f498f9c8f93f8db69796c : SUCCESS in 52m 33s\n- neutron-functional-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/90d76247596c46a2923f33f2fd81e83f : FAILURE in 1h 01m 13s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/08889bc20a3b410990a5c0690fd1be32 : FAILURE in 1h 14m 27s\n- neutron-functional-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/6caf368343634cf480a226e04a445e0e : SUCCESS in 1h 06m 55s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/98b381f208a44b2eadebcdcc96bca918 : SUCCESS in 57m 47s\n- neutron-functional-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/6e96a14f82844ca595e578fce2d6c282 : SUCCESS in 49m 18s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/5a9317bad40740cab027154970202637 : SUCCESS in 51m 37s\n- neutron-functional-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/2709c16eb1aa46139f645e232cc11423 : SUCCESS in 48m 06s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/6f1fc989a4614519ad6938494c4517f2 : SUCCESS in 59m 14s\n- neutron-functional-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/004927cd5cf44831bca5b7a24989bc49 : SUCCESS in 48m 33s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/3e7855d74a8248229a58f389e269c1e0 : FAILURE in 50m 43s\n- neutron-functional-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/e4a22249175645a6b5f0df03965fff09 : SUCCESS in 49m 41s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/b41cf631b99243979dfad91ae5f2b30b : SUCCESS in 49m 16s\n- neutron-functional-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/2c26dba047ea4639bda0f5e3282bac99 : SUCCESS in 1h 09m 11s","accounts_in_message":[],"_revision_number":4},{"id":"da536a3642243ff57e9f12a59991b06114b18c7f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-18 03:30:02.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/17/741017/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"244da9ea772c8778339bee8249230500f5ec612d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-18 03:35:12.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/17/741017/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"f1d2d57ac3e1412bc2f993bc2ce73d934942b8b6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-18 23:12:45.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ad63b42c4adae91d93a2a52aa631b866640a40fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 00:47:40.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5382ba0f74c64cf48089868827c9fad7 : SUCCESS in 1h 08m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4562f75e19b3415885d0e86d73af26ac : SUCCESS in 43m 19s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5400d50fff7246549f1d15f1ed05c888 : SUCCESS in 1h 31m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/840e3d5faaf3474a881dc6a39f2b6373 : SUCCESS in 1h 31m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6b8cb41835f34af081da367f49b74f9c : SUCCESS in 44m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9f7999f7ec1c4fe5ba01f19f73705e8a : SUCCESS in 1h 03m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6929608c01344018a658a9b14a070fb : SUCCESS in 31m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20c00dedfbc84f49adab4b65700818b7 : SUCCESS in 17m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8efcf7f3ad264b1887b44e5fcaa6d7c8 : SUCCESS in 13m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2078ddcb188a49099a0bf39366874aca : SUCCESS in 23m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a2c4e4896c44f65bf4fea5cee29428e : SUCCESS in 18m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5372180ab45f42cd812b9987f228b0f5 : SUCCESS in 9m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e3b2dea26a4e4e169716a9cfba55a523 : SUCCESS in 3m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/842f453e16454e789783c13a033aadd8 : SUCCESS in 54m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c0eb45af09b649aaa8823fb5577287ad : SUCCESS in 41m 42s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/2791c3606b6a4ea7b7b2e9fc3230f14c : SUCCESS in 44m 18s\n- neutron-functional-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/02b36ac2a115414cb5211c44a0cdded5 : FAILURE in 1h 02m 21s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/d5fd200700ae491f9868b72008ca65a2 : SUCCESS in 44m 22s\n- neutron-functional-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/5e464d0034764b75b5c0bf4f188555d8 : FAILURE in 50m 27s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/613f84b769834a129e3f77b394e21827 : SUCCESS in 57m 48s\n- neutron-functional-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/99c3857bdd3943bdb8c86486d6841e94 : SUCCESS in 1h 01m 26s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/5cfef5c6b52a43d6b0a4d55fec165c68 : SUCCESS in 57m 14s\n- neutron-functional-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/0ac81d5eb5f447b98c768318de870348 : SUCCESS in 50m 30s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/ca999f373e49472b958affc9c9fcf7ed : SUCCESS in 51m 31s\n- neutron-functional-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/7a204b1a274644ef8b0716db9e781d05 : SUCCESS in 55m 00s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/3f4d5d14684649dc997a99c2fda1132c : SUCCESS in 52m 14s\n- neutron-functional-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/8f40a6a4cb874c37ac713d3634456938 : SUCCESS in 44m 55s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/b66ac4bd52e1425ea128349a1a162517 : SUCCESS in 41m 44s\n- neutron-functional-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/262edc054c4c4ca3ba60e5c5408890ff : SUCCESS in 39m 18s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/f71e2ca93e6f4ea2bdd46b843666ef8c : SUCCESS in 53m 17s\n- neutron-functional-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/1b88d60b2d0649d0a0904b9a1a5e46b1 : SUCCESS in 47m 06s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/9246645f4cb6495485d55cf521dca395 : SUCCESS in 44m 54s\n- neutron-functional-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/e581f889de53498bb598a80fc83d70cb : SUCCESS in 45m 35s","accounts_in_message":[],"_revision_number":4},{"id":"62c412c9e28130e472dab32d99d8534f0600ab2b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-19 02:49:20.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/741017/4/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/741017/4/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"af420bf93b489902599c5a290a9a3494e425853d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-20 11:00:05.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"476ef0a31656f9dee0ee45104a68af2adb05505b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 12:28:19.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/37a86e68c1fe4fdba37acb06d3118163 : SUCCESS in 50m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e6787398f1bd49a98ee7437bac4341bf : FAILURE in 35m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8955b273cd9a4babab6464fd4827b459 : SUCCESS in 1h 21m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e112a237a312434fbacc1fa88a6b2d35 : SUCCESS in 1h 03m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f9703a9358ea409bb229fed45133f65a : SUCCESS in 1h 07m 12s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/54b437f1d7284d73bbf7b4c16c214969 : SUCCESS in 1h 05m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f78889c1a224295aa53fda4835f4a86 : SUCCESS in 23m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df0bf662132245eca61c0c44ef623bef : SUCCESS in 25m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25303faf3a11427eacbe399f679589ec : SUCCESS in 12m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b23d81f42cc49b2a7f5781490f2f92c : SUCCESS in 26m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e932a2d93c6e464b92ff4315bfcf71ad : SUCCESS in 31m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65a2ff7d527140aeb02a5ad26951158e : SUCCESS in 10m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/22e4fbfd2625458c8b4459dfd783ec68 : SUCCESS in 3m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/32e2ab01fc8046fca0d9219c5e3f6699 : FAILURE in 49m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5bc5828bc6224292bcdd40cb2af0585d : FAILURE in 49m 32s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/a00302dc3c05474cb90dc82b186e4823 : FAILURE in 54m 01s\n- neutron-functional-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/0971d1b12dc54ea9a0e690f6e09eff0c : FAILURE in 1h 03m 04s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/d339952f74944a3d9eebc8922529e565 : FAILURE in 52m 05s\n- neutron-functional-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/ee4eb0efac674a79a83efdfbcd4aede8 : FAILURE in 1h 01m 41s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/70bec254ef78468aa26b5095c1e82f79 : FAILURE in 54m 26s\n- neutron-functional-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/0a56f003ff7245a996ad74ad49e8ae7f : FAILURE in 52m 51s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/09ab80bb5256463daa3ddc95a3fddb04 : FAILURE in 50m 47s\n- neutron-functional-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/448a4e526bf849a998426ba904a271d6 : FAILURE in 53m 33s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/7620b9227ada4528bc8395c704fb3699 : FAILURE in 52m 58s\n- neutron-functional-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/d060244f64704249ad913354f699b27a : FAILURE in 54m 36s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/f5aa06bb6d51495f99ecc33249405ac8 : FAILURE in 51m 41s\n- neutron-functional-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/7695385fae3f416c86e244e24dd8cb45 : FAILURE in 40m 24s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/ce54977203e4408aa90d68864f67d4fb : FAILURE in 52m 35s\n- neutron-functional-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/a25162306f7945f5893ef56f4a18db8b : FAILURE in 48m 58s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/8d420e7c7f7f462dabd84c678eb00575 : FAILURE in 54m 25s\n- neutron-functional-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/fd2072045ada4b37a4331a12e869236f : FAILURE in 48m 04s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/f32a620ccfff4694ba60f9f2e7faff2f : FAILURE in 53m 17s\n- neutron-functional-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/afd642077f3a4885a3127607b478106c : FAILURE in 46m 53s","accounts_in_message":[],"_revision_number":4},{"id":"8d5ab473e30812d5df653547c5ab20639ebbf15d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-20 12:38:17.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/741017/4/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/741017/4/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"838e61372701fe0c950eb03d42bdbd44710f9ea6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-21 17:45:06.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"85fec059ed99bca9c4ec9028572f49c2692455e8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-21 18:13:19.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/741017/4/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/741017/4/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"1989c036223d27ea2e068df6bede5e020f068227","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 01:52:37.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/74221ab92c2748009ec6be96c723f0fe : SUCCESS in 1h 07m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f5c41fefea2e4cc9b696f85e6c13f1d3 : SUCCESS in 39m 09s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/316942d64283485da784a5cd486a4a00 : SUCCESS in 45m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/283b2728e60f4f128793e0da2806da8a : SUCCESS in 1h 35m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/db27854fce4d49b6b23cdd0543dd91a9 : SUCCESS in 1h 28m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b334fafd18c3493eb4bad5ce1f2bbf44 : SUCCESS in 41m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef19d13fffed46f4900866dca5387218 : SUCCESS in 38m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40f4699f373a4258bc3a70bbdbcf30b9 : SUCCESS in 26m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ad1683d4b5649b2999e27e5d2f065fe : SUCCESS in 11m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f26808d8007d467aa2451f95c546c478 : SUCCESS in 16m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56c043b904d145569b969e520aab8acd : SUCCESS in 25m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b93c62f60f74a54b518ebefd65e1e13 : SUCCESS in 8m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/91aa4111415b416fb5e6cab2bbdca271 : SUCCESS in 4m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7f4f232212294ed89c762610bec460a0 : SUCCESS in 50m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b0d4d293a2c842e09a226415f81e3867 : SUCCESS in 57m 18s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/b19225b36de34338a1ee14a753cc3299 : SUCCESS in 58m 24s\n- neutron-functional-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/c30b05ac13884b42ac51bf19be5b87a2 : SUCCESS in 47m 51s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/e5db1dd7ef6546d5b148825ecfbd69e3 : SUCCESS in 1h 06m 27s\n- neutron-functional-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/a3401c1ef7a34295959d8d24994353a5 : SUCCESS in 51m 31s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/23a9b1c0ed7547ec9e62fd1256da6d80 : SUCCESS in 51m 23s\n- neutron-functional-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/c4954c6c0f9b46f09f1e83242edb1205 : SUCCESS in 52m 30s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/5f6ee34b149e4feeb4ca73297ade9c61 : SUCCESS in 51m 09s\n- neutron-functional-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/69aa61de86a143ac9aaab7eb40b12281 : FAILURE in 43m 17s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/c994eb6b19e44caa939c7d2860f299f0 : SUCCESS in 42m 03s\n- neutron-functional-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/809407e9d7954a0fa4c60862c5b91b64 : SUCCESS in 46m 24s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/07927cd4953345809789edeed18ba93a : SUCCESS in 1h 03m 41s\n- neutron-functional-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/cea80b0852ef4aa6be0dcf4c58c71981 : FAILURE in 1h 07m 20s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/fd904c980e8e4483b2f57eab9e6b7592 : SUCCESS in 40m 41s\n- neutron-functional-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/705d0b3db9f54169ab650df708778b26 : SUCCESS in 38m 05s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/1c84ca15601746ecb14330761f2b590a : SUCCESS in 1h 00m 56s\n- neutron-functional-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/1bf3eb87e8bb4fe88d27028fe284b429 : SUCCESS in 53m 29s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/db3174da833645e39457a7d00e6c5255 : SUCCESS in 55m 34s\n- neutron-functional-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/87d36348376c4b4a89d15eddfc2f0535 : SUCCESS in 1h 16m 28s","accounts_in_message":[],"_revision_number":4},{"id":"e52141337905bd7cb18ea20537ab8fdf6376ec6b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-31 17:44:48.000000000","message":"Abandoned\n\nPatch https://review.opendev.org/#/c/740970/ was merged and it\u0027s being released right now (oslo.privsep 2.3.0)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5847c71ad9c6c8d536b4b4aa22c092c3e48baac3","revisions":{"4d7496559befe51d7a75d8ce56be937624a33198":{"kind":"REWORK","_number":1,"created":"2020-07-14 17:54:56.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/17/741017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/17/741017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/741017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/741017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/741017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/17/741017/1"}}},"commit":{"parents":[{"commit":"69cac6c64061dbc929654d1b224b66ca720077c2","subject":"Merge \"Stop disabling peakmem_tracker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69cac6c64061dbc929654d1b224b66ca720077c2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-14 17:48:26.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-14 17:49:17.000000000","tz":0},"subject":"[DNM] Test oslo.privsep patch","message":"[DNM] Test oslo.privsep patch\n\n\"Undo the eventlet monkey patch for the privileged daemon\"\n\nDepends-On: https://review.opendev.org/#/c/740970/\nChange-Id: I98d2cd98c06066def324f22f2cda15ea326d79c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d7496559befe51d7a75d8ce56be937624a33198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d7496559befe51d7a75d8ce56be937624a33198"}]},"branch":"refs/heads/master"},"8b3066db25850925284346ba5c96495b5f0a5189":{"kind":"REWORK","_number":2,"created":"2020-07-15 11:26:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/17/741017/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/17/741017/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/741017/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/741017/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/741017/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/17/741017/2"}}},"commit":{"parents":[{"commit":"69cac6c64061dbc929654d1b224b66ca720077c2","subject":"Merge \"Stop disabling peakmem_tracker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69cac6c64061dbc929654d1b224b66ca720077c2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-14 17:48:26.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-15 11:26:19.000000000","tz":0},"subject":"[DNM] Test oslo.privsep patch","message":"[DNM] Test oslo.privsep patch\n\n\"Undo the eventlet monkey patch for the privileged daemon\"\n\nDepends-On: https://review.opendev.org/#/c/740970/\nChange-Id: I98d2cd98c06066def324f22f2cda15ea326d79c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b3066db25850925284346ba5c96495b5f0a5189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b3066db25850925284346ba5c96495b5f0a5189"}]},"branch":"refs/heads/master"},"119e75120a6af141b359a5d6463be34b56640299":{"kind":"REWORK","_number":3,"created":"2020-07-17 15:13:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/17/741017/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/17/741017/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/741017/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/741017/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/741017/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/17/741017/3"}}},"commit":{"parents":[{"commit":"d28654a0c740844ab726d557480ebb3fd3ac66c7","subject":"Merge \"Remove tempest-integrated-networking job from neutron queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d28654a0c740844ab726d557480ebb3fd3ac66c7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-14 17:48:26.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-17 15:13:34.000000000","tz":0},"subject":"[DNM] Test oslo.privsep patch","message":"[DNM] Test oslo.privsep patch\n\n\"Undo the eventlet monkey patch for the privileged daemon\"\n\nDepends-On: https://review.opendev.org/#/c/740970/\nChange-Id: I98d2cd98c06066def324f22f2cda15ea326d79c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/119e75120a6af141b359a5d6463be34b56640299"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/119e75120a6af141b359a5d6463be34b56640299"}]},"branch":"refs/heads/master"},"5847c71ad9c6c8d536b4b4aa22c092c3e48baac3":{"kind":"REWORK","_number":4,"created":"2020-07-17 15:16:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/17/741017/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/17/741017/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/741017/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/741017/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/741017/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/17/741017/4"}}},"commit":{"parents":[{"commit":"d28654a0c740844ab726d557480ebb3fd3ac66c7","subject":"Merge \"Remove tempest-integrated-networking job from neutron queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d28654a0c740844ab726d557480ebb3fd3ac66c7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-14 17:48:26.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-17 15:16:10.000000000","tz":0},"subject":"[DNM] Test oslo.privsep patch","message":"[DNM] Test oslo.privsep patch\n\n\"Undo the eventlet monkey patch for the privileged daemon\"\n\nDepends-On: https://review.opendev.org/#/c/740970/\nChange-Id: I98d2cd98c06066def324f22f2cda15ea326d79c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5847c71ad9c6c8d536b4b4aa22c092c3e48baac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5847c71ad9c6c8d536b4b4aa22c092c3e48baac3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
