)]}'
{"id":"openstack%2Fneutron~741691","triplet_id":"openstack%2Fneutron~master~I664eeb87d019ba4898800ec8c90e9785164f5f26","project":"openstack/neutron","branch":"master","topic":"test_ntp_ci_jobs","hashtags":[],"change_id":"I664eeb87d019ba4898800ec8c90e9785164f5f26","subject":"[DNM] Test devstack patch to unblock NTP CI jobs","status":"ABANDONED","created":"2020-07-17 17:48:46.000000000","updated":"2020-07-18 23:08:20.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"db3804053547a4b836e185892b20eb0f91303368","_number":741691,"virtual_id_number":741691,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-07-18 10:03:53.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-07-18 11:57:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-18 07:49:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"date":"2020-07-18 11:17:03.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-18 13:41:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-18 07:49:24.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-18 11:17:03.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-07-18 11:57:24.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-18 13:41:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"19f734ca04d92192b9c10789afc87035fe33e7f7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-17 17:48:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16277c05881c0be11be20d793f99bce4a0919781","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-17 17:49:02.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9d8b8653d78d79676175fe9e0e3e44bdd45503b1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-17 17:51:00.000000000","message":"Patch Set 2:\n\nRelated bug: https://bugs.launchpad.net/neutron/+bug/1887992","accounts_in_message":[],"_revision_number":2},{"id":"c9131157affb37ac0762e2c4a03c59f0cdac2629","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-17 20:09:37.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/91/741691/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/91/741691/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"63832eb45638ddf65ec0b97e7d84a2d535c77835","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 02:52:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/da5b2ab892cc420aad916ee307600e71 : SUCCESS in 57m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b22ca469e3854efabe643fc57d9e1ea8 : SUCCESS in 38m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/60e805972eb6477194a86479bf24cb73 : SUCCESS in 1h 14m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c9486a203918430892901ddf190d03b2 : SUCCESS in 1h 21m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9bca70f185034aa89e92d11b8312a454 : SUCCESS in 1h 12m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1b7040f7344a4547bffe13f9604ebdb9 : SUCCESS in 1h 22m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e4a3c6deeec408a9a2ef342c0fe8a7f : SUCCESS in 37m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dcc2d8daec14852aefe70cdc6168378 : SUCCESS in 25m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9daea9dea3024f6bab47d5ff7edf6473 : SUCCESS in 13m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d933b8b49d645d5928d6458537a55c6 : SUCCESS in 28m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e66ec8fd8c2c4279817dfb7a75109a9e : SUCCESS in 29m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d3849285883411890139bd9dc408849 : SUCCESS in 10m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/295ea062bd1844cc80bbd67011c9045d : SUCCESS in 56m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/06c233ba70bb4ccc8059618aa7ee065c : SUCCESS in 51m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ee7d0b7ec9bb40d3ba03deed9c2a0f45 : SUCCESS in 1h 17m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/10ea08e5806947c0b49a286cf5a49a56 : SUCCESS in 1h 23m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d1c34b909daf4f02a7d8ea81c7594444 : SUCCESS in 1h 39m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6af3797180db44c58e34c24a5897e235 : SUCCESS in 1h 12m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8bc8d93425eb428aa1cff2b077c7a2e3 : SUCCESS in 1h 04m 50s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f3f58b4ca028404880e65b7cb85b9312 : SUCCESS in 1h 06m 36s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/240f5f275ce54b8eadea0619ecd9528e : SUCCESS in 1h 33m 57s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/35fccfc9faf84f35a2f76204b09de1f3 : SUCCESS in 1h 28m 56s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e4b971853d4b4723b15439e27efb2425 : FAILURE in 1h 24m 15s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/497af40518e5410292d4c79599688dd4 : SUCCESS in 1h 07m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a310a04afe5f475eaf1ba3007c20dad3 : SUCCESS in 1h 15m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ba3e793df0bb45a098fd5db9319ab805 : SUCCESS in 2h 13m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b856822bfb364bb4ae775a98e1f4546f : SUCCESS in 1h 11m 50s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/416bda60226f43a2bb9b37fbaca228fa : SUCCESS in 1h 23m 56s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/97f267b47a1c478eac52c2c8c2238e90 : SUCCESS in 1h 19m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/63052227fce04fb38096b9982c5d8452 : SUCCESS in 34m 28s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d7c805d5f9d74206ba5be57be1914c26 : SUCCESS in 1h 14m 37s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/35b38e5cc140477e96306c76a8d1e052 : SUCCESS in 1h 34m 46s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/79562c109f5a47949189ae947aa089d7 : SUCCESS in 1h 47m 27s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/685c3d29f3e84b7ca676b455a7d0c8b8 : SUCCESS in 2h 13m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/15d4385b329f4ecdbc5ee728395cb778 : SUCCESS in 1h 52m 59s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/da43161cc146412d9b44188a7583acf4 : SUCCESS in 1h 58m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4842a7319c21705f430843864e0d9bbc149fd056","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-18 07:22:24.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/91/741691/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"437b99491dd74350e8a9a3acdfe1570d9cde7b4d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-07-18 07:49:24.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2020/Arista_Tempest_Testrefs/changes/91/741691/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"a320b3dcb75954ad1864e1256891395aacf36be7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-18 09:33:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5e9e0e9ae0ccba25b3c61de7d89641bdd15ab777","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-18 10:03:47.000000000","message":"Patch Set 2:\n\nlooks like the devstack patch is working. I +1\u0027d it","accounts_in_message":[],"_revision_number":2},{"id":"cd3b6e497451e00c16a377df3adacbc65ee2376a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-18 10:03:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9dc67029fe650e27b08e2a1b1e22d3aa9d852035","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-18 10:21:08.000000000","message":"Patch Set 2:\n\n\u003e looks like the devstack patch is working. I +1\u0027d it\n\nthanks, Rodolfo; I\u0027m actually testing a cleaner approach but first going for minimum failure scenario in https://review.opendev.org/741807","accounts_in_message":[],"_revision_number":2},{"id":"8f374f2a303135064e1cbea38cb5d7fd45ab8738","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-18 10:40:20.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/91/741691/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/91/741691/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"4d936978886c616a5a2db5ce1a12a79f26b9193b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 11:16:29.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"b5581ad380a5710c421cd42aca0d8013bc777ffe","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-18 11:17:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"39ee7926938c18c642389e5de252c76b71105988","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-18 11:57:24.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/91/741691/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/91/741691/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"323f1955d4e852e5d5561755a19e8ca270e3223f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 13:41:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e4ab5cab257e4cc586a55173e8313cef : SUCCESS in 55m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/55f3e582df884219ac2c1b89055e38f3 : SUCCESS in 30m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7baf461441e04bef85513f5f1986a85c : SUCCESS in 1h 07m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ee819e7cb5894cc581de0cf58fe23500 : SUCCESS in 1h 18m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6d60df45ec22414588ab92977477a756 : SUCCESS in 37m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/dccc80c686024a3f867e5292caea7683 : SUCCESS in 1h 05m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f3f0237a5434dea8c2d8a5a4e85726d : SUCCESS in 34m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92fc51603ee9403d95e4f8a904d8a325 : SUCCESS in 24m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/777b0c2e7833401abf7fedcb4ea441ef : SUCCESS in 11m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/875dd54437364d67b33922d69196c1ed : SUCCESS in 16m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80f6517512e64dc0a3374c4caa406d56 : SUCCESS in 27m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29d1ac2ed96e486eb04ca8a24dafd80e : SUCCESS in 9m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9e308f73b4a94c79a6d2be4b9f9e36ad : SUCCESS in 55m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c8d9466da48a4f8b9075fedecb90fcdd : SUCCESS in 39m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/598b8e6f316d4fafb94e3d2bdb57ba2d : SUCCESS in 1h 11m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/97e5f2d195194eb7ba5c71f6f78e54a3 : SUCCESS in 1h 07m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2d3712390cb145d9993c92c9017858ba : SUCCESS in 1h 50m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/339fcde24f5448fa88978236d06258b9 : SUCCESS in 1h 12m 27s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/28d85bfa9ef44027a1e82ea494912f39 : SUCCESS in 1h 21m 02s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5558aa9054d846e38fdc0fc1908d4547 : SUCCESS in 1h 20m 32s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e456df75f6aa472f84a53049d7cf6327 : SUCCESS in 1h 10m 37s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3d7f45eb06314d32a79abb036e733eb3 : SUCCESS in 1h 30m 32s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4e9db8f946dc46e9a125344b2f93bcae : FAILURE in 1h 23m 55s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/faa431fe9f094be7be754e19613ce684 : SUCCESS in 1h 13m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/74c94c87c7cd49f18a0fc44e9a33bca7 : SUCCESS in 1h 08m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d581f755b34945af9815b0576c07d888 : SUCCESS in 2h 08m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/196db9f56b584bc3828da5c74bc2c65a : SUCCESS in 55m 33s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/e62076abe8264cbc87177f3a5b177bd8 : SUCCESS in 1h 04m 25s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a0c82f0291e44f309a2a2719d2bb7445 : SUCCESS in 48m 01s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dc8b6d87fe674e9f89450c72271f1ccd : SUCCESS in 44m 37s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9a8d6e3b3b2c4cd5ad4ba4cfc851ee54 : SUCCESS in 1h 13m 34s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1ab8eb991f934ddfb928faf1bdf13119 : SUCCESS in 1h 23m 47s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6c839acac1b045b4a3494e991815a18a : SUCCESS in 1h 47m 18s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/204b91de56b646a0a20429083b460387 : SUCCESS in 2h 19m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/43dec16bd99a499e99429adc4b350ae1 : SUCCESS in 1h 46m 44s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/1243d3efb8c54601b58a2b3f5b1901bd : SUCCESS in 1h 56m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7465abe8834f7fa4a8cbcfa3346556f5649bf63a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-18 23:08:20.000000000","message":"Abandoned\n\ndevstack patch is working and merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6d94b3c6e75a2538f05fab3efea5dfad5e6bf26b","revisions":{"10fcb04f7ac5e7959206213191b0399013b11785":{"kind":"REWORK","_number":1,"created":"2020-07-17 17:48:46.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/91/741691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/741691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/741691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/741691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/741691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/741691/1"}}},"commit":{"parents":[{"commit":"d28654a0c740844ab726d557480ebb3fd3ac66c7","subject":"Merge \"Remove tempest-integrated-networking job from neutron queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d28654a0c740844ab726d557480ebb3fd3ac66c7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-17 17:47:44.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-17 17:47:44.000000000","tz":0},"subject":"[DNM] Test devstack patch to unbloc NTP CI jobs","message":"[DNM] Test devstack patch to unbloc NTP CI jobs\n\nDepends-On: https://review.opendev.org/#/c/741687/\nChange-Id: I664eeb87d019ba4898800ec8c90e9785164f5f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10fcb04f7ac5e7959206213191b0399013b11785"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10fcb04f7ac5e7959206213191b0399013b11785"}]},"branch":"refs/heads/master"},"6d94b3c6e75a2538f05fab3efea5dfad5e6bf26b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-17 17:49:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/91/741691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/741691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/741691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/741691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/741691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/741691/2"}}},"commit":{"parents":[{"commit":"d28654a0c740844ab726d557480ebb3fd3ac66c7","subject":"Merge \"Remove tempest-integrated-networking job from neutron queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d28654a0c740844ab726d557480ebb3fd3ac66c7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-17 17:47:44.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-17 17:49:00.000000000","tz":0},"subject":"[DNM] Test devstack patch to unblock NTP CI jobs","message":"[DNM] Test devstack patch to unblock NTP CI jobs\n\nDepends-On: https://review.opendev.org/#/c/741687/\nChange-Id: I664eeb87d019ba4898800ec8c90e9785164f5f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d94b3c6e75a2538f05fab3efea5dfad5e6bf26b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d94b3c6e75a2538f05fab3efea5dfad5e6bf26b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
