)]}'
{"id":"openstack%2Fneutron~742364","triplet_id":"openstack%2Fneutron~stable%2Frocky~I42c33fefaae6a7bee134779c840f35632823472e","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1887148","hashtags":[],"change_id":"I42c33fefaae6a7bee134779c840f35632823472e","subject":"Ensure drop flows on br-int at agent startup for DVR too","status":"MERGED","created":"2020-07-22 07:58:30.000000000","updated":"2020-07-23 00:43:03.000000000","submitted":"2020-07-23 00:43:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742364-1595464983688-78954b85","meta_rev_id":"0cd9c9da811bb0c9c3317f592410d5abc7c76932","_number":742364,"virtual_id_number":742364,"owner":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-23 00:43:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-22 15:19:56.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-07-22 08:20:43.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-07-22 22:58:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2020-07-22 15:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-07-22 22:58:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2020-07-22 08:23:03.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 08:20:43.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-07-22 08:23:03.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-07-22 15:11:42.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-07-22 15:19:56.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-07-22 22:58:06.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-07-23 00:43:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9187f75675bdd32fc0530d2358238b324a2c882d","author":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"date":"2020-07-22 07:58:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0390a88e4b860ddf8446cae0f65bf361435e22ee","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-22 08:20:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e8745bd27efcd39ef1d518841b03dd98da51353f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 10:36:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/14561896754e44b6a77ef97cb1585f2e : SUCCESS in 54m 26s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/cab29cd13bda48d491a1341e42cb9603 : SUCCESS in 42m 39s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/47ed99712f754279ba742e44777fe045 : SUCCESS in 1h 30m 05s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/2e701ac358ce47b9a07bb771aa1324a6 : SUCCESS in 1h 03m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/c964f9f44f094595ac58b1b406163d10 : SUCCESS in 57m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4f5f7408c7b4216b13d933a23487b63 : SUCCESS in 12m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0789681a29394072bf5ceaab4ae8659c : SUCCESS in 25m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ab8684e1093744eaa3f421aacc8c22c4 : SUCCESS in 25m 18s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/764d78ba1c874701a333c5c3d0f7344c : SUCCESS in 7m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bb8125f08f79421f8577c352058495db : SUCCESS in 1h 35m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2352c3b17aa2463c87a3793dc78503d4 : SUCCESS in 1h 01m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb062fa6d6774d97b870d7b516ae42d8 : SUCCESS in 1h 18m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/616c1ea9db4747a7b19f6da1bd373a41 : SUCCESS in 46m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6cb683f33984454e8e2b513ad2c66c6d : FAILURE in 21m 58s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c3b07d08499a46819eea26c392c25626 : SUCCESS in 1h 25m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5ce09679119b4840b3ef9f87d42049b1 : SUCCESS in 1h 21m 32s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fb084b700c4948e48c84be6fd75ac41b : SUCCESS in 1h 29m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c314b2737bef4baaa50f45f9d97c9cd1 : SUCCESS in 23m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43a03693b14c4e2698838ad3d34b681e : SUCCESS in 30m 24s","accounts_in_message":[],"_revision_number":1},{"id":"ccb3a9a998fe61a9ed098306eceed72eb8e46fc1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-22 15:11:42.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1cf96df08f61eb2bc701e9f187cc97d82ff3cf88","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-07-22 15:19:56.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/742364/1 : FAILURE in 7m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8ffcf6b27913160c7c8dfb71daa23fd0a4ee3010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 19:46:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/96e2a85b3717481eab3b885e3725be11 : SUCCESS in 1h 06m 10s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/a964e278a012463699adfa2526f87046 : SUCCESS in 32m 30s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/ab1c4f046d984f18be2364e1917cd32d : SUCCESS in 1h 01m 53s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/57c111bb2b0d41ffa636355e0025a755 : SUCCESS in 58m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/ad6f4b19b93b4dd787f9eea419dae1c5 : SUCCESS in 58m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c70a778a9eb421286ecb0d6c5da572e : SUCCESS in 15m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf8830c42cbd4437acd2b3ab3024d9c4 : SUCCESS in 27m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4523ebaa17814512a398e4045f2add12 : SUCCESS in 27m 53s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/cd8ce186c5194c2e9f221377601ffb68 : SUCCESS in 7m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e9aa2ef16890479bbc8a0296090a0f5b : SUCCESS in 1h 25m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/61e94aef06184485bbe258f618781246 : SUCCESS in 1h 01m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d50343f434f4552a642bf273b420051 : SUCCESS in 1h 27m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bed84bab642c4b3e95906b00ffcbe7e9 : SUCCESS in 49m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9e4b13f3055c4cd98c5919e2964eacbc : SUCCESS in 1h 39m 53s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8ad1dfa1fc0c4beab32e92d93c7c03b7 : SUCCESS in 1h 23m 33s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8e088b62344148168da24a594180bb98 : SUCCESS in 1h 25m 54s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/20b5b9ef4d644a9685e4063bc85c374b : SUCCESS in 1h 20m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c63a2692ca64303a3ecffede5f7c59c : SUCCESS in 26m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f43450bb0ae64e2fa89fab0d5bf537c5 : SUCCESS in 38m 48s","accounts_in_message":[],"_revision_number":1},{"id":"0237a350a9c867e256843633bd93028d70384364","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-07-22 22:58:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e48f3ec138a67c1c962a40d732c568211af99e75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 22:58:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a62a475e8ae6a843affcb57ba074b72bf6fbc47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 00:43:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/894ef28810f441ef843331161d258fa5 : SUCCESS in 1h 03m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3527e89a3e412e9300e9217bbe0b3c : SUCCESS in 11m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/526a36728c764d10885f43401d124bf4 : SUCCESS in 22m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/70ba9dce059e443d82da3c9c38384fd9 : SUCCESS in 26m 18s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/36b852189e8e410387bbcf22f75d68a4 : SUCCESS in 7m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e7802944eafa4e69a7690c042b9832a6 : SUCCESS in 1h 42m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ba1e927a069246109bd64c9f630e3657 : SUCCESS in 1h 11m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5ed0002bbc34c159839553db2fa4138 : SUCCESS in 1h 34m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/508c551c3a504bfda3290ba6550e5f86 : SUCCESS in 50m 22s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/db560b53af084956a452b8a77583b966 : SUCCESS in 1h 36m 22s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ccf7adf465014190bf360756cc66afd6 : SUCCESS in 1h 24m 48s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ed9601f72b674bb6a3634fdf8403dcf4 : SUCCESS in 1h 40m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a9fa8dbb4d34ae78419e2208a47fb28 : SUCCESS in 25m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61e4ada865ee44a28c52341c663bc72a : SUCCESS in 39m 09s","accounts_in_message":[],"_revision_number":1},{"id":"944e796a4ca29bc6f95d54ea3df2805ca236dd23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 00:43:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"47ec363f5faefd85dfa33223c0087f4444afb5b9","revisions":{"47ec363f5faefd85dfa33223c0087f4444afb5b9":{"kind":"REWORK","_number":1,"created":"2020-07-22 07:58:30.000000000","uploader":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"ref":"refs/changes/64/742364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/742364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/742364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/742364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/742364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/742364/1"}}},"commit":{"parents":[{"commit":"d6afd6e63bb8fd35f2fb3557312af6015b89d623","subject":"Merge \"[stable only] Configure logging in keepalived_state_change\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6afd6e63bb8fd35f2fb3557312af6015b89d623"}]}],"author":{"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","date":"2020-07-13 14:48:10.000000000","tz":0},"committer":{"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","date":"2020-07-22 07:52:34.000000000","tz":0},"subject":"Ensure drop flows on br-int at agent startup for DVR too","message":"Ensure drop flows on br-int at agent startup for DVR too\n\nCommit 90212b12 changed the OVS agent so adding vital drop flows on\nbr-int (table 0 priority 2) for packets from physical bridges was\ndeferred until DVR initialization later on. But if br-int has no flows\nfrom a previous run (eg after host reboot), then these packets will hit\nthe NORMAL flow in table 60. And if there is more than one physical\nbridge, then the physical interfaces from the different bridges are now\nessentially connected at layer 2 and a network loop is possible in the\ntime before the flows are added by DVR. Also the DVR code won\u0027t add them\nuntil after RPC calls to the server, so a loop is more likely if the\nserver is not available.\n\nThis patch restores adding these flows to when the physical bridges are\nfirst configured. Also updated a comment that was no longer correct and\nupdated the unit test.\n\nChange-Id: I42c33fefaae6a7bee134779c840f35632823472e\nCloses-Bug: #1887148\nRelated-Bug: #1869808\n(cherry picked from commit c1a77ef8b74bb9b5abbc5cb03fb3201383122eb8)\n(cherry picked from commit 143fe8ff89ba776618ed6291af9d5e28e4662bdb)\n(cherry picked from commit 6a861b8c8c28e5675ec2208057298b811ba2b649)\n(cherry picked from commit 8181c5dbfe799ac6c832ab67b7eab3bcef4098b9)\nConflicts:\n\tneutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47ec363f5faefd85dfa33223c0087f4444afb5b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47ec363f5faefd85dfa33223c0087f4444afb5b9"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
