)]}'
{"id":"openstack%2Fneutron~742365","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I42c33fefaae6a7bee134779c840f35632823472e","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1887148","hashtags":[],"change_id":"I42c33fefaae6a7bee134779c840f35632823472e","subject":"Ensure drop flows on br-int at agent startup for DVR too","status":"MERGED","created":"2020-07-22 08:06:14.000000000","updated":"2020-07-23 00:56:49.000000000","submitted":"2020-07-23 00:56:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742365-1595465809202-56289871","meta_rev_id":"c82284ffc66253d19aa09fc10767fe9cd314c8ca","_number":742365,"virtual_id_number":742365,"owner":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-23 00:56:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-07-22 08:20:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-07-22 22:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2020-07-22 15:12:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-07-22 22:58:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 08:20:48.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-07-22 15:12:09.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-07-22 22:58:34.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-07-23 00:56:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb515adaffd9a32b8deaa0fbe87f77d053a3f565","author":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"date":"2020-07-22 08:06:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88087c4b1388d277acfcb2c5eb3b338d625f38c3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-22 08:20:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5d018c649fcf549fb460e67dcaea358bdb5c4c12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 10:42:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/82e80a7c7deb4ea3a05db46338c100bd : SUCCESS in 49m 33s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/ade9f4300c39422782afb24d1d78ec62 : FAILURE in 1h 35m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/052ed65261a545dc9b26141b29c9bd47 : SUCCESS in 54m 42s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/9df2bf1d12bc4d8baa653e0d4e86a2cb : SUCCESS in 49m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e59dd66b15a4c5e9e00f07f1ee28757 : SUCCESS in 11m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8e662989d6b4e04bd685d0cb1a511e7 : SUCCESS in 22m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3b4c7e1d4b2748d79a473746eeacd3de : SUCCESS in 21m 42s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7f9a530b4ae148ed9e1175acebf0049e : SUCCESS in 8m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/aca37ff1f5764087b24c130ddfc6208b : SUCCESS in 1h 12m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d2701a1c9cb045fbab7def7deb264e87 : SUCCESS in 59m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d81a216ca21402283e84a0cf2761ba7 : SUCCESS in 1h 18m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f5aada952e1147fabf062b46ab12d4a8 : SUCCESS in 35m 56s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/4621415a6c2a4935a288e679b7179e21 : SUCCESS in 1h 35m 15s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d27e3c87740c476d844690dbc0e77b06 : SUCCESS in 1h 19m 20s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/60ca0bd020cd4dd99896d6bef290e664 : SUCCESS in 51m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe31c0af946a425ba8a05bd9e1e608b5 : SUCCESS in 39m 04s","accounts_in_message":[],"_revision_number":1},{"id":"21e433f0ebecc586e79de9b8fef025f5b42875ca","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-22 15:12:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7d4803db6051965548fd4e0cfa1566e43e51c716","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-07-22 22:58:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4e49276481c879cfa1ae325d23dbc1419e7ddde1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 22:58:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b491f953b5fe7282015b30b663093900e73dc63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 00:56:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/63d9362e2c46431880bf6be298f271e1 : SUCCESS in 48m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e24689ee2472448d91b6d1d3d0e457cb : SUCCESS in 11m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1790b841eef041918fb89c0d3a6c9470 : SUCCESS in 23m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e2d90a9ce5c14181898c9f7218754f4f : SUCCESS in 26m 30s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d4eae4ed7c164e1080aa9d1506d5ad75 : SUCCESS in 6m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/131594bc910247af99ae9e31b663f099 : SUCCESS in 1h 27m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bdc382736041488ba36119565fa6c8fd : SUCCESS in 1h 48m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/260dbd681e8c48228d3c181ef99fb5cd : SUCCESS in 1h 31m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cc62a7d66a58446c8559a2d33a889399 : SUCCESS in 39m 09s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/fbadc609a8524b35931cacb15167191f : SUCCESS in 1h 30m 57s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0cd5229a3d9d4f3f8f026dfad50db5ff : SUCCESS in 1h 22m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3ed972172934903a5fd11cb536150a5 : SUCCESS in 32m 35s","accounts_in_message":[],"_revision_number":1},{"id":"f3458cf9f1f05b0a0fbed330f8e021783aba6108","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 00:56:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8a173ec29ac1819c3d28c191814cd1402d272bb9","revisions":{"8a173ec29ac1819c3d28c191814cd1402d272bb9":{"kind":"REWORK","_number":1,"created":"2020-07-22 08:06:14.000000000","uploader":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"ref":"refs/changes/65/742365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/742365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/742365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/742365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/742365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/742365/1"}}},"commit":{"parents":[{"commit":"a19b0cb09ee51d299ff0a6f11bff623059774678","subject":"Merge \"[stable only] Configure logging in keepalived_state_change\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a19b0cb09ee51d299ff0a6f11bff623059774678"}]}],"author":{"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","date":"2020-07-13 14:48:10.000000000","tz":0},"committer":{"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","date":"2020-07-22 08:03:21.000000000","tz":0},"subject":"Ensure drop flows on br-int at agent startup for DVR too","message":"Ensure drop flows on br-int at agent startup for DVR too\n\nCommit 90212b12 changed the OVS agent so adding vital drop flows on\nbr-int (table 0 priority 2) for packets from physical bridges was\ndeferred until DVR initialization later on. But if br-int has no flows\nfrom a previous run (eg after host reboot), then these packets will hit\nthe NORMAL flow in table 60. And if there is more than one physical\nbridge, then the physical interfaces from the different bridges are now\nessentially connected at layer 2 and a network loop is possible in the\ntime before the flows are added by DVR. Also the DVR code won\u0027t add them\nuntil after RPC calls to the server, so a loop is more likely if the\nserver is not available.\n\nThis patch restores adding these flows to when the physical bridges are\nfirst configured. Also updated a comment that was no longer correct and\nupdated the unit test.\n\nChange-Id: I42c33fefaae6a7bee134779c840f35632823472e\nCloses-Bug: #1887148\nRelated-Bug: #1869808\n(cherry picked from commit c1a77ef8b74bb9b5abbc5cb03fb3201383122eb8)\n(cherry picked from commit 143fe8ff89ba776618ed6291af9d5e28e4662bdb)\n(cherry picked from commit 6a861b8c8c28e5675ec2208057298b811ba2b649)\n(cherry picked from commit 8181c5dbfe799ac6c832ab67b7eab3bcef4098b9)\nConflicts:\n\tneutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py\n(cherry picked from commit 47ec363f5faefd85dfa33223c0087f4444afb5b9)\nConflicts:\n\tneutron/tests/unit/plugins/ml2/drivers/openvswitch/agent/test_ovs_neutron_agent.py\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a173ec29ac1819c3d28c191814cd1402d272bb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a173ec29ac1819c3d28c191814cd1402d272bb9"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
