)]}'
{"id":"openstack%2Fneutron~744137","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I2e3aa7c400d7bb17cc117b65faaa160b41013dde","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1867119","hashtags":[],"change_id":"I2e3aa7c400d7bb17cc117b65faaa160b41013dde","subject":"[Security] fix allowed-address-pair 0.0.0.0/0 issue","status":"MERGED","created":"2020-07-31 11:31:45.000000000","updated":"2020-11-17 14:34:08.000000000","submitted":"2020-08-07 11:06:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744137-1596798388352-453d58e8","meta_rev_id":"f0581ed5bc28aed72c6b7664c63c72ef3f41aeec","_number":744137,"virtual_id_number":744137,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-07 11:06:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-07 07:12:08.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-08-05 08:50:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-08-06 22:20:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-08-06 22:20:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"date":"2020-07-31 11:31:45.000000000","_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-31 11:31:45.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2020-08-05 08:50:50.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-08-06 22:20:53.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-08-07 07:12:08.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-08-07 11:06:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3930af3015a4b0ba80df1059bf0ae5d6494449ee","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-07-31 11:31:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47fd6879929e76ba5cf4901a43f07e18abd317ae","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-07-31 11:33:07.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/744137/1 : FAILURE in 46s","accounts_in_message":[],"_revision_number":1},{"id":"48691faec09f2cdd39971468506b623b1c79a789","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 17:35:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/7d2a9c3ff37546fbaf0e76488e71e186 : SUCCESS in 47m 37s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/e690111a3b9a4e0a8cd94390e7591f8d : FAILURE in 1h 29m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/c6f90f8d83ca406788becd342dadb0b2 : SUCCESS in 54m 02s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/ed0a2f57ddd3406ca4a364434b602bca : SUCCESS in 29m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64622ed4b04240d9953efbef19bdfa29 : SUCCESS in 11m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2d246484da64990bae5133de5262c51 : SUCCESS in 24m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/40d679a0ed114108a53fb50a17f36353 : SUCCESS in 28m 57s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5c46f6b0caa740e7afc40ac439d65963 : SUCCESS in 6m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/65f8034cc39d49dfb9c0e8425171cac2 : SUCCESS in 1h 10m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b4677931be5a46388a2b6d6cacf2ba72 : SUCCESS in 51m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3393944b53e4cd1b49894855c04635b : SUCCESS in 1h 21m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/953b13d95ba941729e4eb30f9d330c7f : SUCCESS in 41m 04s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/529473ee7b5b48e799977875de18a6c1 : SUCCESS in 1h 49m 19s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2d3ae9f1e10e4440ace026cf340950a8 : SUCCESS in 1h 30m 55s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a18c35ffd297454594418d5ab991045d : SUCCESS in 1h 32m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9330be92fca74202961c956687c6e14a : SUCCESS in 31m 24s","accounts_in_message":[],"_revision_number":1},{"id":"1b408899a96680241eb932ff2b9f4cc6b73cc51a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-05 08:50:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"04c544ebd6bd4b6e64e132b1e70a8c09f27edb0b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-08-06 22:20:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"94755b55e1dfa7969c39d8c2d566c2d3cc3d9450","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 22:21:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3381349836c5a91567cbe5cf5f4a6be723edfd11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 03:30:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/347ca4a7fd0c4bd084ae2b050207c91e : SUCCESS in 43m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f901dbeff8a46cf94f35c5903145e71 : SUCCESS in 10m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4137d4d061474580a034b2a255464228 : SUCCESS in 20m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/32a69ec7b4b44c8c969efbe2f7522728 : SUCCESS in 20m 53s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/632d354fdde24cd5881c223a010dc1a4 : SUCCESS in 7m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/672afa62d40e4bbd8f036b5685040998 : FAILURE in 1h 06m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/39a8dd81a2784c5fbac56301ca9379ae : SUCCESS in 54m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a927f960c454b1bba2ef5e6a7d654bc : SUCCESS in 1h 18m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f4e936e9e00340b2845a0005af90e0a0 : SUCCESS in 36m 39s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8eb98e65620c46a9bc2ad525965973ad : SUCCESS in 1h 15m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d88340dfe0494702b30c59d764517b59 : SUCCESS in 1h 12m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc7e65ca969441f5860ea5e84f3abd67 : SUCCESS in 28m 11s","accounts_in_message":[],"_revision_number":1},{"id":"fa0c8800421436bf5963f8b8931a8d9ec33df067","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-07 07:01:40.000000000","message":"Patch Set 1:\n\nrecheck tempest-full","accounts_in_message":[],"_revision_number":1},{"id":"701ba92f653c2c9caae6d9b2cb8b7b60396d020f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-07 07:12:08.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/744137/1 : FAILURE in 9m 25s","accounts_in_message":[],"_revision_number":1},{"id":"bba503e486fb114505b31ed8fab64b7fa46a4b9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 09:09:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/84372a8c18d048b395f98135039dfeaa : SUCCESS in 30m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/3ea8114dafb249a8b74d37c0e9ffd7e4 : FAILURE in 56m 37s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/aa69557d65854127acfd059ae4a0494c : SUCCESS in 55m 13s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/3c5d2a87e8584651953a0b58a4140408 : SUCCESS in 58m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53ea4382ea5b467ba326932aaabe5edc : SUCCESS in 12m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ac3b18929884f14a090b93c8dbc598f : SUCCESS in 23m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cff3ada378a646228da22e83128bc944 : SUCCESS in 22m 35s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7419309e55304c8891f59f39b2206681 : SUCCESS in 7m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b258d08eff4c4b03b01fccf295429b4c : SUCCESS in 1h 36m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/82518a5625734bba8db90aa936c49c30 : SUCCESS in 1h 09m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c047898be1043ba847c5736a55f24b6 : SUCCESS in 1h 24m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/12ae24a8e3a04d42ba1975e540005fd5 : SUCCESS in 42m 52s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/12efcd1ee8df4415b341b1d545f60174 : SUCCESS in 1h 28m 43s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3bbc7bdc9d344c8a8e3992d98cd67c13 : SUCCESS in 1h 39m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/af01735df980427ba6009acd09ba8e4b : SUCCESS in 1h 26m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f15266b9bb414735a95b50d13d9d7486 : SUCCESS in 31m 00s","accounts_in_message":[],"_revision_number":1},{"id":"6ed4f728611487aa4c51b243362157674ff96556","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 09:10:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c67eea0ff702740680eacd3631ea3bc4be681a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 11:06:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/f6266ceed4ae4bd1afc8608cc2d03d87 : SUCCESS in 49m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a506f05d3244233898b6c6763957c0b : SUCCESS in 12m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cd1c8fb5bfb84bbca3f3f1453c4cfdd2 : SUCCESS in 27m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/67b05a4080ed479f951cea9fba4033fb : SUCCESS in 27m 44s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/19617e5c226f47649d0c2a01244818cb : SUCCESS in 7m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4803d7f150f446519e2faca63750cb1c : SUCCESS in 1h 27m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8a985141816d47919d920ced601c51a9 : SUCCESS in 58m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b22350d4e44490f87dd746127679bcd : SUCCESS in 1h 16m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3adfaf2025fc4f289d178ca06cc2328c : SUCCESS in 41m 55s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/630314faa21f43c49ff85b76860c0d70 : SUCCESS in 1h 16m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9f697ef382e140ff9b85b1125ff7a577 : SUCCESS in 1h 30m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c73b432cc9014e8694f8bbff191ee011 : SUCCESS in 32m 40s","accounts_in_message":[],"_revision_number":1},{"id":"95ff9660e5d051338d6e082b4053bc7d72045fcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 11:06:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6d687c16bcf9970c7b223c1f9b2bc5ec8dc9f434","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-17 14:34:08.000000000","message":"Created a revert of this change as Ida8a8848813f06dff7a8d98bb0264534dde37573","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd6203b2c7e1e4af63813b307bc4ec1b49516bd5","revisions":{"bd6203b2c7e1e4af63813b307bc4ec1b49516bd5":{"kind":"REWORK","_number":1,"created":"2020-07-31 11:31:45.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/37/744137/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/744137/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/744137/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/744137/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/744137/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/744137/1"}}},"commit":{"parents":[{"commit":"8a173ec29ac1819c3d28c191814cd1402d272bb9","subject":"Ensure drop flows on br-int at agent startup for DVR too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a173ec29ac1819c3d28c191814cd1402d272bb9"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2020-03-13 10:18:04.000000000","tz":480},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-07-31 11:31:18.000000000","tz":120},"subject":"[Security] fix allowed-address-pair 0.0.0.0/0 issue","message":"[Security] fix allowed-address-pair 0.0.0.0/0 issue\n\nWhen add allowed-address-pair 0.0.0.0/0 to one port, it will\nunexpectedly open all others\u0027 protocol under same security\ngroup. IPv6 has the same problem.\n\nThe root cause is the openflow rules calculation of the\nsecurity group, it will unexpectedly allow all IP(4\u00266)\ntraffic to get through.\n\nFor openvswitch openflow firewall, this patch adds a source\nmac address match for the allowed-address-pair which has\nprefix lenght 0, that means all ethernet packets from this\nmac will be accepted. It exactly will meet the request of\naccepting any IP address from the configured VM.\n\nTest result shows that the remote security group and\nallowed address pair works:\n1. Port has 0.0.0.0/0 allowed-address-pair clould send any\n   IP (src) packet out.\n2. Port has x.x.x.x/y allowed-address-pair could be accepted\n   for those VMs under same security group.\n3. Ports under same network can reach each other (remote\n   security group).\n4. Protocol port number could be accessed only when there\n   has related rule.\n\nConflicts:\n    neutron/tests/unit/agent/linux/openvswitch_firewall/test_rules.py\n\nCloses-bug: #1867119\nChange-Id: I2e3aa7c400d7bb17cc117b65faaa160b41013dde\n(cherry picked from commit 00298fe6e84cd7610b39af50e9517885a182f47c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd6203b2c7e1e4af63813b307bc4ec1b49516bd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd6203b2c7e1e4af63813b307bc4ec1b49516bd5"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
