)]}'
{"id":"openstack%2Fneutron~744210","triplet_id":"openstack%2Fneutron~master~If1e533a61aeebbb4761d669c516fe86a4381765c","project":"openstack/neutron","branch":"master","topic":"bug/1888487","hashtags":[],"change_id":"If1e533a61aeebbb4761d669c516fe86a4381765c","subject":"Change \"propagate_uplink_status\" default value to True","status":"MERGED","created":"2020-07-31 16:03:23.000000000","updated":"2020-09-09 01:03:59.000000000","submitted":"2020-09-09 01:02:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744210-1599613332206-b51fe808","meta_rev_id":"2ae2076be0f1ee6c8f64769adc32e11676109dd8","_number":744210,"virtual_id_number":744210,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-09-07 17:09:13.000000000","_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"date":"2020-09-08 16:44:20.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-09 01:02:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-07 15:53:53.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-09-07 18:17:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"date":"2020-09-08 20:33:55.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-08 16:06:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-09-08 08:34:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-08 13:07:07.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-08 04:00:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":1,"date":"2020-09-08 06:22:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-08 13:07:07.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-04 07:38:56.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-08-06 00:36:24.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-08 20:58:56.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2020-09-07 17:09:13.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-07 18:17:32.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-08 04:00:42.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-09-08 06:22:26.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-09-08 08:34:17.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-09-08 13:07:07.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-09-08 16:06:35.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-09-08 16:44:20.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-09-08 20:33:55.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-09-09 01:02:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8759147294f294a533e998159a4c1d9f56dce8ad","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-31 16:03:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7def2e3fa3fe6dbb04576b8f7832b3e662eab717","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-07-31 16:58:24.000000000","message":"Patch Set 1:\n\n* sriov-test http://52.27.155.124/sriov/744210/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"ada6bc323fa1a3de7ae5553779396a47f8658f4c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-07-31 17:53:39.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/744210/1 : FAILURE in 1h 49m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7742576fcff5f0893831f1127ea46f81dbb9da69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 00:22:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4a01295bffee4108945c21501499f3fc : SUCCESS in 36m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b6dbc83498024bbea7b6ffa4f7420b13 : SUCCESS in 41m 35s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a6e528a4b0de4f9daff7a22fa39d7c1c : SUCCESS in 44m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4e1a21cf508340ad92f929ce1a012955 : SUCCESS in 41m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/300eb559256d4421a6d4d02623bf6fe7 : SUCCESS in 47m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4846b6ba63f54a7a8dd30f34d461901d : SUCCESS in 1h 08m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad88596388734c299112e060ed01d0ec : FAILURE in 34m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0058b17057d6486eba2b0cf964c79520 : FAILURE in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2326903260a74020bd37f1a876a039cd : FAILURE in 10m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5a0351e5f264218bf14a3276c167bb3 : FAILURE in 27m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f94e97c4af3b4c3ca74155810ce037ef : FAILURE in 27m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4c63acbd8f54a42b796870e942109cd : SUCCESS in 9m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b608c162e51f40c0b3dcda823469a38f : FAILURE in 54m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a1dffab99147476babc3e743903a671e : SUCCESS in 44m 22s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/83fb199d64ed4bbdaabd4d1d1f99e794 : SUCCESS in 1h 23m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cf505c4745fe46d8985ead364600a7e1 : SUCCESS in 1h 08m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0d56ac541ace4294ac4ebd55fdc2e3da : SUCCESS in 1h 54m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/af7b625607de4a49ba565f722fadef95 : SUCCESS in 1h 13m 32s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/aca88659a66e4a319f55b17c3b755efa : SUCCESS in 1h 05m 47s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/515d55648d434ae9af5fe495abe899d2 : SUCCESS in 1h 13m 38s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d993e2d886b4496785465019c288e221 : SUCCESS in 1h 22m 53s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ded27ac47ba340a991f01134b8cdd9cf : SUCCESS in 1h 35m 42s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0ef3bec6f5da4f2eb14876ffe225260b : SUCCESS in 1h 38m 46s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2fbf2e2c126d44b29c3041b186ebf643 : SUCCESS in 1h 20m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/301063d95a8845a9b14757f7000df57e : SUCCESS in 1h 29m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8e436483183e420bb274c98094e40742 : SUCCESS in 1h 57m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91a67179fa714183a1f9509329da23ff : SUCCESS in 1h 02m 22s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/e6d9e732bfd3482784f66d13a8e40eda : SUCCESS in 45m 55s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a67b5eb353e3417aaf8b8db13b1a5bfc : SUCCESS in 1h 29m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/91836eb81a604e10bd69d26ef3f6c691 : SUCCESS in 34m 13s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5bd48ddf7a844ac9bad5dea6892c816e : SUCCESS in 1h 13m 50s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/738b9c38eb2a47ae861745128f5a92b9 : SUCCESS in 1h 11m 36s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4fa499db24ac4c3a9a33d26112c789e9 : SUCCESS in 2h 02m 42s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ab17f0e66b5f475b881f041bcd5ad756 : SUCCESS in 2h 42m 40s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5006918cc7704455ae5e0ae43bd13a85 : FAILURE in 1h 53m 04s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/a059d8b0e37a4513acd1ecee4eedeb7f : TIMED_OUT in 3h 05m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ecec0b0a462dc8bb2a4745a679d61cf41cadef5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-01 12:35:51.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/10/744210/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/10/744210/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"04b953020db129eacf6b3275cfa651da3375885e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-04 07:38:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97224763ee00d3eaa271a004c53783816ca1f395","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-08-04 13:38:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0b303a4ad9a5bdf94be9203d127607002736502","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-08-04 14:00:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d96126e4f3977c7e5254ed3a6915ea869ed45a6a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-08-04 15:11:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0c35097c06f74249068e03ee970b7d98076b8c3e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-04 15:53:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/744210/2 : SUCCESS in 1h 52m 33s","accounts_in_message":[],"_revision_number":2},{"id":"6faa60c4af0d38ab985a9faaaf21aa51e771c7c7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-04 16:45:37.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/10/744210/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/10/744210/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"360ca8031f1ffec78e22c837c3348011e6cde13d","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-08-04 18:21:56.000000000","message":"Patch Set 2:\n\n* sriov-test http://52.27.155.124/sriov/744210/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"4ff1714e798ebc2608a27a9a8bb24e6287ebde94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 18:53:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/65336bd4f04c4dbfa02bf9b6d528156b : SUCCESS in 57m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a8d838551aaa4b1fac91fc967a6a42e2 : SUCCESS in 41m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d42628ab551447ddb8cc10bd54c5f669 : SUCCESS in 1h 18m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4e2930362f614a3bb5b592029aec12f8 : SUCCESS in 1h 31m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ed5657f6bedd4c08b9b5a7366328e4a7 : SUCCESS in 1h 30m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/331f9ece0be54099808664e053f592bd : SUCCESS in 1h 08m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4cf5a5ee0d14941a797e94911f2b7a9 : FAILURE in 34m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f558f9b6b61543e8b07312661b40ed85 : FAILURE in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e997db1474f24de8ae6e3e0e8aeddfbc : FAILURE in 14m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87a644f0119f46698298425f1b4af958 : FAILURE in 32m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/069333dba851415792dda6df1d156bcf : FAILURE in 31m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4787b510114a4ba5b251a100dc35d4c3 : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d1d1fe139fb4a9db48623f6db29915c : SUCCESS in 11m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/951a8fd61f0a4fdbbdb2c4da894b38ce : SUCCESS in 52m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e5805ae795242baaebc51b117a29bac : SUCCESS in 56m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6491132cd08d4d598db4461c3cababfb : SUCCESS in 1h 32m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7021ce8b9b0b4fbf99112c96ded0497d : SUCCESS in 1h 17m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/95a8e5dc6853403f89930d50e81debf9 : SUCCESS in 1h 58m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9f5ea27429424d509292204e32b39791 : FAILURE in 1h 04m 31s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/03ca52479c2b4623a2ea0174a19ef58c : FAILURE in 1h 12m 52s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/793b6e40d671446294fe2fdb19aabdbb : SUCCESS in 1h 16m 57s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c270d27de33f4615813eb1695b22c4c3 : SUCCESS in 1h 22m 55s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/14ce1b0948044437b14abadd7708ed84 : SUCCESS in 1h 19m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2053a392a40e42afb69f0d986bb3f3ca : FAILURE in 1h 31m 47s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e7faa0ef825249bcae85d8a41ec188c4 : SUCCESS in 1h 10m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/738d387e7b9e467dbae124008ffaed99 : SUCCESS in 1h 38m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a8c75a1c045a41c3aa92e2d9e6189d87 : SUCCESS in 2h 11m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5e32291de504123902f3203ec5deff0 : SUCCESS in 55m 24s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a2c9c990601245569ca63dcd563145cc : SUCCESS in 1h 05m 19s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e0aa23f9e497476087adc322bf6699f4 : SUCCESS in 1h 07m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6df1cacd39e44fe1b608a49a66630c57 : SUCCESS in 39m 09s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6015aff145b24b6e912d78be049fefcd : SUCCESS in 1h 22m 10s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ad37cd18b3ad44f486e39ed6150bdf16 : SUCCESS in 1h 35m 48s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/afca081b7ca84fae9a07bb4891dbe4e3 : TIMED_OUT in 2h 14m 38s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7536960f18fd4f3686bfc88a8957bfdb : SUCCESS in 2h 41m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cbb4608f2e24419ab8ab45f97c46b4d6 : FAILURE in 1h 46m 45s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d6d28be387dc4b149812e8b25c638259 : TIMED_OUT in 3h 06m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c1139b84750ab96cdee9ea20b1334f843be53efb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-08-04 20:22:44.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2020/Arista_Tempest_Testrefs/changes/10/744210/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"bdff88ee2374657b00aa573d0804ea17bbbdb810","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-08-04 23:08:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3133f7a209eae06ffffbeb35528429df3d0f5c08","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-05 01:56:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/744210/2 : SUCCESS in 1h 58m 42s","accounts_in_message":[],"_revision_number":2},{"id":"b4b4ab7d880dddcd44d673d71e38b2e676ca1ddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 02:23:35.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b3b985e557f94424b4533201a2ea93ef : SUCCESS in 57m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e35b72b7a02748ff9841ece9773d8536 : SUCCESS in 32m 50s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f29672bfddfe4e319581c5dfe0a12c35 : SUCCESS in 1h 34m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2aa435ebd9f04c59a7f44ea70bd5e321 : SUCCESS in 1h 33m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/de452a15cf074aa296c3b5b4568448b1 : SUCCESS in 1h 03m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b4a2aa9d3e474e708fced717ed7059f7 : SUCCESS in 55m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49f32b8bf3a24e2ea2242cb3133c8389 : FAILURE in 31m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c95653d20a84e44a937df0778e2f0be : FAILURE in 26m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e41b0d3ac5fe46ab8bcc73517133f75a : FAILURE in 11m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2941f1c7973b4ad5a526de2613e70ad3 : FAILURE in 39m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68c0e71bd6e84547860f87b82c6f8148 : FAILURE in 29m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3f3c546a71844349dd79b189cd2422a : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4fdfb3a855e41bbad26c1d39268cff2 : SUCCESS in 8m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/18b04999242745bda80cfc97ce2b6113 : SUCCESS in 44m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dfe82e14738540a9a2da825623097b8e : SUCCESS in 47m 51s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5cac9e4fdb484144a004b7e6d49481fd : SUCCESS in 1h 10m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f26f1ccffa8442c497a0950811e7eb16 : SUCCESS in 1h 08m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/af4c326ca33b4844bc9251639fa67b40 : SUCCESS in 1h 57m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8dd2cae514b44da59fa999c8edec423e : SUCCESS in 1h 16m 24s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/934da44958954eb0818868ad9a8b1d0c : SUCCESS in 1h 21m 57s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/43cf21e4ba3b4af0826d1d0c3ba21a32 : SUCCESS in 1h 26m 55s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5bd697c0eb674b3c9dc17810dce9c8c5 : SUCCESS in 1h 13m 14s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a118b4df666f4748a9671872e089e081 : SUCCESS in 1h 20m 41s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1a02d7863204448ab5a6f4afa3f3fc56 : SUCCESS in 1h 32m 22s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/610eed9896ec4b678c6ba02b3bd3c611 : SUCCESS in 1h 04m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b119adcc920740909fd46c11ecb91016 : SUCCESS in 1h 20m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/604455e6094b473d9e9d083db499a0f2 : SUCCESS in 2h 09m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/955dca05b5a44ed988ec998f4f621ded : SUCCESS in 57m 03s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/3d3564d254214385a7fc58ebb91b6d87 : SUCCESS in 1h 19m 16s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/07440f47159447318bd6ada124b27b5a : SUCCESS in 1h 18m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b6393d63be654c828c494b928a1257c3 : SUCCESS in 41m 10s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b39d948c275549dab59990559935dd60 : SUCCESS in 1h 23m 40s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/99c816381ea14b2c9a8f520d02047919 : SUCCESS in 1h 41m 34s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/518fca2b9bfb4bd690d48d1f730f4fc1 : SUCCESS in 2h 03m 31s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0a2bbfab273c46a7ae8c81e2e0ff0293 : SUCCESS in 2h 22m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3b4f38bc855348f58034ae5b4e309a7b : FAILURE in 1h 50m 27s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b129dc35fb014d83a0753b619a87e424 : TIMED_OUT in 3h 04m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4360a6615c15c1d891c469dd2f1402416f7a0636","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-05 05:26:04.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/10/744210/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/10/744210/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"1c3bece8cdafb74872a9d9361a74b3582cbd3ade","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-08-05 16:37:38.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"270bd8cdb93e391d60eee50b27678cf9a9f720d5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-05 16:39:37.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/744210/3 : FAILURE in 45s","accounts_in_message":[],"_revision_number":3},{"id":"295f2dc86dceacc4f4aabf6cd123ee5bf9a059cd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-08-05 16:41:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f9c8620018012d14f25fd378b6219a508ead7ade","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-08-05 16:41:52.000000000","message":"Patch Set 3:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":3},{"id":"13143dc2f4d4ea60e28b484cbf83f77bbc0df676","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-05 16:47:17.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/744210/4 : FAILURE in 46s","accounts_in_message":[],"_revision_number":4},{"id":"dc03eaaa9357100a7080d58c6201705c70070a28","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-08-05 17:38:34.000000000","message":"Patch Set 4:\n\n* sriov-test http://52.27.155.124/sriov/744210/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"3fcae060d073cc1ff30fd160d1a15c0360daeaf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 21:58:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/80109a04c7ce4383acac9e336afd35a5 : SUCCESS in 56m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/30632362db634746aa938e18c8527978 : SUCCESS in 39m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0a98ebde84a34ef1828cc1b66b884f8c : SUCCESS in 1h 11m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f9cea08a0c05444f89e76e67cf3c1032 : SUCCESS in 1h 09m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/13864e9919a94042876665f8b2a89fd4 : SUCCESS in 1h 04m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d2f94312d1db4c0fb2be5969cfe6e810 : SUCCESS in 1h 00m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/088ef51686744348b08ef64ac1f24ab8 : SUCCESS in 32m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/964c7d3e1ebd430ebf54ca6aec0b40ed : SUCCESS in 28m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50d9abb7d99b41e985de976d359bb33a : SUCCESS in 12m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7682d4302604678ac5575c9abc681dc : SUCCESS in 25m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91200da44b1149dab3301726593f1507 : SUCCESS in 24m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24dacbfea1ca4feb9582cea80fc4dcbd : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/394f877df43e49f892917606763c7fde : SUCCESS in 9m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/200bb8be9e254af5ac50b5015a7d9248 : FAILURE in 50m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/065dd69fc789443d95e1d0476f1dc6d4 : SUCCESS in 41m 18s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/db76e0cb56ae4f7faf6162015c5bebcd : SUCCESS in 1h 23m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3609b8afe26847ee8273be0a2efc0816 : SUCCESS in 1h 20m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f0fc1084f2834f2b805e3ed9ea8f1b53 : SUCCESS in 1h 57m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a57762fd7e8145338ec9565f5c9c9abd : SUCCESS in 1h 06m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/31c61cd5f48f4f5a93472bb652d45552 : SUCCESS in 1h 36m 01s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/62e79cc8f14e47fa9d1e4a4f12b0bc37 : SUCCESS in 1h 32m 55s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4f8775a36a0145dd8f5593295b5a76d7 : SUCCESS in 1h 12m 04s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/86890f16d60444da98ad26fcd5fc2a72 : SUCCESS in 1h 56m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/99a27a22e0674db7b4e951deaa30dc91 : SUCCESS in 2h 00m 05s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b4bf6461dac147ce883e6cdb5b0601ec : SUCCESS in 1h 26m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a3871934ef204ce9afe1f75cd057e1b7 : SUCCESS in 1h 21m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f9e11749182f4753a2a096eb8725cf77 : SUCCESS in 1h 58m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66971e9214aa461cb8b9abde3c9f8e15 : SUCCESS in 52m 45s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a1f80267a2af4ef09e0842f09767413c : SUCCESS in 1h 02m 19s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/222885022d2b42d2b29080fff2491564 : SUCCESS in 1h 30m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b118ede9774c4fb0bce1a574a2009785 : SUCCESS in 46m 52s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/998ef6c60f314797990718a5224f3904 : SUCCESS in 1h 20m 41s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/95b400944e47417a9fbf31176008b405 : SUCCESS in 1h 49m 37s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f603caa9ae98433190d2c84e86698c01 : SUCCESS in 2h 01m 17s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b257e12e4f4746d4bec402a651e1ac62 : SUCCESS in 2h 53m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2642b746a63c4cf29e82031bc075c157 : FAILURE in 2h 05m 28s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3daf88359ebf404faf721b5cfd00c9aa : TIMED_OUT in 3h 04m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9c64293a23f39f400ef55a78719152000bcca187","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-06 00:36:24.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/10/744210/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/10/744210/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"499449381098d33ba2ace163938d471a36bfa53a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-08-06 06:23:36.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b604509a0ed697713be051ea5b8919e7a991c15d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-08-06 07:09:18.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2020/Arista_Tempest_Testrefs/changes/10/744210/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"8c362827e961845f9147b8c66cace95a32787844","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-08-06 09:33:41.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5f86ce72f1502f28b376b5cbd03f57e536af54a6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-08-06 09:38:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7ef0b46159aee033b9c99d112f6570e5b8584c90","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-06 10:04:26.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/744210/5 : FAILURE in 25m 23s","accounts_in_message":[],"_revision_number":5},{"id":"ad7ce220cc80659acd5dc8cfdec095a6c0b9490c","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-08-06 10:38:56.000000000","message":"Patch Set 5:\n\n* sriov-test http://52.27.155.124/sriov/744210/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"e3945ce9b1852b1a53745d5af17c228abb1cd706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 13:47:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8a7e31ad4a324ae2b8e0fc092482c2b2 : SUCCESS in 54m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a017b29a635f4fc9a7dad7d775f068f1 : SUCCESS in 38m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/633e78a369414d469ebad97307d8ba6c : SUCCESS in 1h 25m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/63f2fd6ef97b418b9abf07b67985cf1d : SUCCESS in 1h 15m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b17cb96ce1f14d5f81689eb9996daa28 : SUCCESS in 1h 07m 03s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4dd0112559304ed2b87b1b9d10033b54 : SUCCESS in 1h 14m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b4532331f614e3f989ed5637c2d4bbb : SUCCESS in 37m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb4ef1fdfb6c420b85756cf5cf2ba7c6 : SUCCESS in 26m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eace493b3a154c0b86a3e011e509012d : SUCCESS in 12m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69949dddf4a945e1be3625d9eab6e8fd : SUCCESS in 17m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/add36f4ab79e4e19a7438ba6d54fb2cb : SUCCESS in 25m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dfcf29eeb344a7aba3345016af12c8a : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7422e7ac726e4534be805dc06d8a17f5 : SUCCESS in 8m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/84fed6e888df4a0e9f007f06da1dace6 : SUCCESS in 51m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cf928300606841bdaf863c29d2a8e15e : SUCCESS in 53m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/24329621ceeb42368c2c660c91bc2007 : SUCCESS in 1h 15m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b8d245015dc44d96a698588458c134c8 : SUCCESS in 1h 11m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a07552ad9f7d4bc5b0f4f5f54d0e7e68 : SUCCESS in 1h 13m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f61cc178fdb148358669fdc3cd09ab3a : SUCCESS in 48m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f51f965b45a840a08f0d14749d1a0328 : SUCCESS in 1h 13m 55s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/51ea434b4b3e4b708c59a00139db79eb : SUCCESS in 1h 18m 44s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e957883e5afe43e08f53d9f42bb54473 : SUCCESS in 1h 08m 14s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/65215201a9ad4f40a9787dc32e87437a : SUCCESS in 1h 13m 01s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b481054b230444d4824f9df87cef4621 : FAILURE in 1h 22m 34s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/31e8388dd57646ab8f6b512d708d3678 : SUCCESS in 1h 19m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e83fc7302992438b8747ed8599ea1bcd : SUCCESS in 1h 28m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a829157c2b0646388267cd8a1a55df18 : SUCCESS in 2h 12m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/440b33e3946a4d56a6471baa589cb1d3 : SUCCESS in 58m 31s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/af0fc7b5f59e489cbba8759abdde0d56 : SUCCESS in 1h 04m 05s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f991aab3fa814cb1b24943cc1bf8b723 : SUCCESS in 1h 09m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b3144faeb3414256b7005015227031a9 : SUCCESS in 37m 57s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d720f280a0aa4dd0994d07724e459210 : SUCCESS in 1h 31m 59s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d4a2d5b5ca7f440f93d83e47dfd502da : SUCCESS in 1h 30m 26s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cd2bd7b71f9f411c9f2ceecb3719e074 : SUCCESS in 2h 13m 15s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/dad7e8fdad1e4dac892b84fe93e81b8f : SUCCESS in 2h 38m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/abe2680588754ba1920e8e44889adca5 : FAILURE in 1h 52m 00s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/0bcb0d981daa44ae8186d38493a1eced : TIMED_OUT in 3h 05m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8801aab22af0dbbe9da3278467c461ce8e1d4927","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-08-06 14:15:35.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"6d86a733929201d1f686df7f26c4eb941d75346f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-08-07 09:43:42.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2020/Arista_Tempest_Testrefs/changes/10/744210/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"4d214fe8f7a804d421e0e5d01c0a379fb82fb4df","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-07 11:01:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"33adde19cab216e17f509ba38d8b3751d754d4b5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-08-07 11:19:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d11603fdfa3ccc9b75cb215481484823ddf0e6c6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-08-07 13:08:26.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2020/Arista_Tempest_Testrefs/changes/10/744210/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"492609688181ca2d16942ec1369f4ead36684d8b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-08-08 01:13:24.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2020/Arista_Tempest_Testrefs/changes/10/744210/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"438620f521eef8cbe52249015fe35ed72e6d7f44","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-08-08 20:58:56.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"ffeba6ce4bc68d3f85cafe37c9b9b08fca75f75d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-11 21:17:55.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"8f7892805857d2a947bb1dbc69276de6253ae0b8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-26 08:55:36.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"a7e632acdb5ccc846c60beb274e4030c8cc2c388","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-09-07 15:53:08.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4fb4d22d4feabd6d73892f403ff6b6ac18162a2c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-09-07 15:53:53.000000000","message":"Patch Set 6:\n\nPS6 just solves a merge problem with the alembic migrations EXPAND_HEAD","accounts_in_message":[],"_revision_number":6},{"id":"73f6a042cb0b88e00cc6ee2da5d80823962d6d30","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-07 16:48:11.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/744210/6 : FAILURE in 54m 30s","accounts_in_message":[],"_revision_number":6},{"id":"3e208126fa5cdac1bd414af59bccc98b551ece28","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-07 16:51:27.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/10/744210/6/check/devstack-networking-ovs-dpdk/06e9a85/ : RETRY_LIMIT in 1m 56s","accounts_in_message":[],"_revision_number":6},{"id":"0c53e742b27f87982587d047e8ae19465dc4b994","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-07 17:09:13.000000000","message":"Patch Set 6:\n\n* sriov-test http://52.27.155.124/sriov/744210/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"df3bbf49deafa8d67ebe3f25f5bbf366d5b66b69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 18:07:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9c13329f61a24748b8c4b49c077a753e : SUCCESS in 33m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3b6cf814d498403aa039a579e3a94a42 : SUCCESS in 23m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0f7e985128c54c5794a30d4c6c0e9ac7 : SUCCESS in 43m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a7095c3618f34e95b2ef2b9302b86962 : SUCCESS in 43m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/83f4c78c7a47476ba8b63d1fc3b75082 : SUCCESS in 55m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/132cde51688a440686522348040c9ebb : SUCCESS in 44m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50fd8f4e60d24b5b9b5b5de33ba34051 : SUCCESS in 21m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e76c0b9f14674a5aa2613475e0fa577c : SUCCESS in 25m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b2c3abb8fef446aa0cf1db6ea7404a4 : SUCCESS in 13m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e6fe0aff2d44cb381d7b8aea10290ce : SUCCESS in 27m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4386b7b187d4977a6e3ea09dfdc7704 : SUCCESS in 24m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6be0554bddfc4968815ac036cdcded38 : SUCCESS in 10m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c1489896fad4c78be88f7ac6e707987 : SUCCESS in 9m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd4ac3bc8c4c4461b1b06de4f73d3f48 : SUCCESS in 49m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fe7583d6a9a9417ca5a6cfe8cd6bc020 : SUCCESS in 1h 06m 46s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d75a602407944ec88aa3980b298c443f : SUCCESS in 1h 46m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb41ea9a8a134acc820a00f322aef02f : SUCCESS in 1h 16m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/962724761ae145be8b63fd570df02944 : SUCCESS in 1h 08m 26s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/34c058943a39444ea289554dd04c3a30 : SUCCESS in 1h 11m 54s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d25e0575bd8f476cb03feb1d9b20216c : SUCCESS in 1h 04m 54s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f63f953e458b45f4a5fe971a4fd88914 : SUCCESS in 1h 21m 32s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e51ddd59884e459c8ec8e13b6059909e : FAILURE in 1h 30m 05s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d06a1f578c044e4f91c485e92b88ddc3 : SUCCESS in 1h 10m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3d90ec42c33344b1b66714c9fe1c6efc : SUCCESS in 1h 15m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/088862e40f174cc3809fea10e1ef608a : SUCCESS in 2h 04m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f67fdc400ab421ebda6997bd68ddbb6 : SUCCESS in 52m 25s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/53309d9efae2400ba4312c575d154d61 : SUCCESS in 1h 09m 43s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3a8aba215da24d8081078d7aaf42a2e6 : SUCCESS in 1h 18m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3d7b8fe58e244eaf89e706484a22cf9a : FAILURE in 36m 15s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/83e9287b726c419a90162cf4e4546476 : SUCCESS in 1h 16m 51s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9dfbfccb04bc45dda7b423f9663e7b78 : SUCCESS in 1h 32m 32s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6fdf58916f6d412391798f78597ef3fe : FAILURE in 1h 46m 58s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/3ed274132e3e486287b44fdf8fad4f1b : SUCCESS in 2h 12m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/dffdcf937a3f4072bac6ac9271204ccd : SUCCESS in 1h 37m 50s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/434f04a8ca944afeaf63331fb329650d : FAILURE in 1h 47m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b2aeb5b7cbfd5c208a7753c550e311f090ba5eb0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-07 18:17:32.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/10/744210/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"263c61aadb26299cabc2d80f12a2d63bfb018a08","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-09-08 04:00:42.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"b9055bdb5a320778b23eac6a8d02670139268ebb","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-09-08 06:22:26.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"c529f0f5603e6804e423c176864ab68dc920859e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-08 08:34:17.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"7823cd5c999615fc50e4d6872bb8ef33729593a6","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-09-08 13:07:07.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"992b5f53f8d92f26ac3c9d204d215d574f4efa82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 13:07:32.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"11ff132e43c2bca8734c780b9722bad8b4d9ac30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 15:18:41.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e8dad344df674a9aa8d2673911affcc1 : SUCCESS in 1h 22m 55s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d9fde1c9ac0f46ccbfdcf5ad8bb1e368 : SUCCESS in 2h 06m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b0faf99460a24fc7a7a1e9d8cde72647 : SUCCESS in 1h 44m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fcccf25a9a8249ad8ecd7fdeb2d078a7 : SUCCESS in 1h 48m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1a3b6a8fb98f4879b3fee9b6cf080606 : SUCCESS in 1h 58m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21926f1b7d134b31bd8803752524f3f7 : SUCCESS in 26m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/915590b36c564cddb6a1ca39e9122a44 : SUCCESS in 13m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/addbdd4b925b490db55bd13d6fffa336 : SUCCESS in 16m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f172da0046f4be49303c5b9732b26fd : SUCCESS in 24m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a0c424229a147bb995c99c73ed94c08 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05ecd649a8b54ab5b86d668bd0984b20 : SUCCESS in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c85abe2cee374bca8516766c46006f95 : SUCCESS in 39m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f2294a9c94bb48f0ac10c674a67e6fee : SUCCESS in 1h 03m 04s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f345efe7939f40a0b72ec27526bccbe1 : SUCCESS in 1h 19m 06s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ce59520087b34c73ad6caa2fde33c9c9 : SUCCESS in 1h 04m 52s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/94285fc755dd407293b9a6ecf6dd60e4 : SUCCESS in 1h 18m 45s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/572a78691e8142449ce970f8923cf731 : SUCCESS in 1h 18m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/baf54e14ca1b4b76bd92801653e8bcab : SUCCESS in 48m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ffb243ee97e6495d83928eb9d8d7f28e : SUCCESS in 1h 09m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3d66f25826024587807450dba394aa24 : SUCCESS in 1h 53m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65d7c497d7ff4d2cb4963159d819e244 : SUCCESS in 40m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c3a3a10a02e04bc9b68ef7ad5cb8bd8c : FAILURE in 1h 11m 01s","accounts_in_message":[],"_revision_number":6},{"id":"0b8d28029dea9fe3a0f122f168b5181b65bab051","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-08 16:06:35.000000000","message":"Patch Set 6: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"5399c7bf124c579584220c2358016fb434be2090","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-08 16:44:20.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/10/744210/6/check/devstack-networking-ovs-dpdk/8915fdb/ : RETRY_LIMIT in 1m 58s","accounts_in_message":[],"_revision_number":6},{"id":"f9ee14fb5b4bda3414d8f3036b452cc38d77b7bc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-08 16:53:49.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/744210/6 : FAILURE in 46m 40s","accounts_in_message":[],"_revision_number":6},{"id":"d8f5c4db2a577068b144b0fc618ff0dcd4795185","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-08 19:48:12.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"ee1fd19f8cac9813abbdaca761f11ac4b654d3c8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-08 20:33:55.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/744210/6 : FAILURE in 45m 09s","accounts_in_message":[],"_revision_number":6},{"id":"07da7c36bf5965e30a07037543f2a256baed7e65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 22:09:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fbe721dd2ccf49bfaf09900109c772ba : SUCCESS in 47m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ef22e52e5cd4cd3bc7db63e112632ac : SUCCESS in 31m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4e91d72d9a634a3e9d622a9534a75213 : SUCCESS in 1h 30m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1239edec6df04eae975877af667cba9e : SUCCESS in 59m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b56f507856e9480fa7c176b5513e3673 : SUCCESS in 1h 27m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c2acb395767d48838647f5b03916347d : SUCCESS in 1h 06m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f2b6e0c64774062990d8d0c30c5add8 : SUCCESS in 33m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/482254a7a36a46e2aa4ff3be119a9945 : SUCCESS in 28m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7491266becb249929c6d7785b833b14f : SUCCESS in 13m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66f801b836c74d348112b938be0815a5 : SUCCESS in 24m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/353a1777fe844f68b29abbeb04fcff5e : SUCCESS in 24m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e924b3b21f9d4f6d98a467961b39ad4b : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a93040c6bea24c70b2f1072bc64ae77f : SUCCESS in 10m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5dfc8c2434cb4c2e8c43ece0e0f0e13a : SUCCESS in 46m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0f6559d537324e298894f8809b422120 : SUCCESS in 1h 17m 24s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/52f44c4965aa4db088c4fc57fe0aec8f : SUCCESS in 1h 45m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe78b5c2038e426497e48b3bdb6630ed : SUCCESS in 1h 05m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/59ff649a9869444799c729b594025a65 : SUCCESS in 1h 06m 14s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9bb1f2cefb72455d9d90fe70292b1d0e : SUCCESS in 1h 08m 09s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/524ba89a965046328b171df392c0bfcb : SUCCESS in 1h 10m 36s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/21c07217073c4f7781e7b84262b1d245 : SUCCESS in 1h 22m 06s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/84fb30528bc144829c228c1770db0721 : TIMED_OUT in 3h 08m 04s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9dcf017367324203b2522c9757e44bdb : SUCCESS in 1h 13m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ce76c7168fd04a91b6e4dae500f5f3ad : SUCCESS in 1h 16m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e87b25c7215c4924973094c4a4bea2b9 : SUCCESS in 2h 06m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ed2c37023e74b1ab42d5c02bb3e1695 : SUCCESS in 51m 33s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/88785538cd474ccabb32f43aa9d9c4d8 : SUCCESS in 1h 04m 18s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/580547b5a0cb428b93022fffd685d616 : SUCCESS in 1h 17m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/43789d842d2a4b3fb38d6dfd6f90f850 : FAILURE in 35m 54s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6596a443caa548518fb88df69030b0aa : SUCCESS in 1h 14m 12s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2960a6ec7d9447fcb8b26e01d50f2682 : SUCCESS in 1h 40m 41s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aad43b0433d34deba31444eb9c4fea67 : FAILURE in 1h 45m 06s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ec41b1dcaabd45f4982eb0047b44a39c : SUCCESS in 2h 20m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d1295875ca444c9396bc27bc84f9fff7 : FAILURE in 1h 48m 07s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/57ba0287989646e694f3942d8278f01a : FAILURE in 1h 59m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b894f4867d9ffc133327de6968ba9d98895a06e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 22:09:56.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e7d7cb0453bd7b23c4c39486741386a31994acea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 01:02:11.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e3d95c1fab744918a712a68f468af831 : SUCCESS in 1h 03m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/549829f445c34905a9ce797acb63485b : SUCCESS in 1h 37m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a8ff0871ccae440e83d5107db3e0b741 : SUCCESS in 1h 16m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/71ff1537ab324accbec7e8984f81b116 : SUCCESS in 1h 11m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e94ce3a892034132858ba29d7fa0dbc2 : SUCCESS in 1h 23m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e3ec366eb8a4954842a7108593c9758 : SUCCESS in 38m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bad598a4fd7e412b91c055893f640d32 : SUCCESS in 18m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80001506089b4f4c9ba1fb34664cd822 : SUCCESS in 37m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c77f8b5626ab47e3816e6958ea628665 : SUCCESS in 35m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12bc2b7384fd4b96a6b56d416fefcbc2 : SUCCESS in 11m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89ca19a206c54f14873724c7215e000b : SUCCESS in 10m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f39a8959ad54dd9919e1d5aecb8cf7f : SUCCESS in 46m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8cec0e36956447dca03ed974632fd0ab : SUCCESS in 1h 44m 04s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8038d29fb4ee4490a447a38ca0b695a6 : SUCCESS in 1h 02m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b6097d200fca4742aba9088e0e0a82d3 : SUCCESS in 53m 22s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/903eeaecf900443bbd103c4b695f5655 : SUCCESS in 52m 39s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/58eac56d9bf74b9e9911706956679d3a : SUCCESS in 1h 26m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/92e7b81188ad4af7ac06e7c70c563b29 : SUCCESS in 1h 07m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dd38ecb94bfe42dda0a1d4d725bff3cd : SUCCESS in 1h 11m 25s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c666bb15e09f492abb437c72ebaa4a1b : SUCCESS in 2h 03m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74754b36c1d946c690ab7828c6eef98f : SUCCESS in 56m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/55a3369abd98484ba9dbf0d840a1eede : SUCCESS in 1h 13m 39s","accounts_in_message":[],"_revision_number":6},{"id":"2c9f7d06650add1ae17de48e8feb5c05257e05dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 01:02:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"dd20b5051fc0189e76c50ba3a5e9a6526352794c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 01:03:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/093f2b81daf44ce59727796b52e783af : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e926457ce4304903b159c5930891700f : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d4ae3f75a542112c92130828f992eec17d5bf0ba","revisions":{"0e2844746f7a9a4be120b0879f55374b79faa752":{"kind":"REWORK","_number":1,"created":"2020-07-31 16:03:23.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/10/744210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/744210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/744210/1"}}},"commit":{"parents":[{"commit":"fe79ef22b8edadc02c54435537a6cc947960906f","subject":"Merge \"[OVS][FW] Remote SG IDs left behind when a SG is removed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe79ef22b8edadc02c54435537a6cc947960906f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-31 15:33:47.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-31 16:02:53.000000000","tz":0},"subject":"[WIP] Change \"propagate_uplink_status\" default value to True","message":"[WIP] Change \"propagate_uplink_status\" default value to True\n\nNOTE: I still need to add a release note\n\nCloses-Bug: #1888487\n\nChange-Id: If1e533a61aeebbb4761d669c516fe86a4381765c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e2844746f7a9a4be120b0879f55374b79faa752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e2844746f7a9a4be120b0879f55374b79faa752"}]},"branch":"refs/heads/master"},"3d726c92b0c57c9b48a1f273315228fdbf686c71":{"kind":"REWORK","_number":2,"created":"2020-08-04 14:00:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/10/744210/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/744210/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/744210/2"}}},"commit":{"parents":[{"commit":"fe79ef22b8edadc02c54435537a6cc947960906f","subject":"Merge \"[OVS][FW] Remote SG IDs left behind when a SG is removed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe79ef22b8edadc02c54435537a6cc947960906f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-31 15:33:47.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-08-04 13:59:45.000000000","tz":0},"subject":"Change \"propagate_uplink_status\" default value to True","message":"Change \"propagate_uplink_status\" default value to True\n\nWhen \"uplink-status-propagation\" extension is enabled, new ports\ncreated will default the value of \"propagate_uplink_status\" to True.\n\nCloses-Bug: #1888487\n\nChange-Id: If1e533a61aeebbb4761d669c516fe86a4381765c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d726c92b0c57c9b48a1f273315228fdbf686c71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d726c92b0c57c9b48a1f273315228fdbf686c71"}]},"branch":"refs/heads/master"},"1a5dbba03eef1bb4e64585051d324ba782ed7e72":{"kind":"REWORK","_number":3,"created":"2020-08-05 16:37:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/10/744210/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/744210/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/744210/3"}}},"commit":{"parents":[{"commit":"fe79ef22b8edadc02c54435537a6cc947960906f","subject":"Merge \"[OVS][FW] Remote SG IDs left behind when a SG is removed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe79ef22b8edadc02c54435537a6cc947960906f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-31 15:33:47.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-08-05 16:37:35.000000000","tz":0},"subject":"Change \"propagate_uplink_status\" default value to True","message":"Change \"propagate_uplink_status\" default value to True\n\nWhen \"uplink-status-propagation\" extension is enabled, new ports\ncreated will default the value of \"propagate_uplink_status\" to True.\n\nCloses-Bug: #1888487\n\nChange-Id: If1e533a61aeebbb4761d669c516fe86a4381765c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a5dbba03eef1bb4e64585051d324ba782ed7e72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a5dbba03eef1bb4e64585051d324ba782ed7e72"}]},"branch":"refs/heads/master"},"901369854651404397a379d4ed49b6c41331a762":{"kind":"REWORK","_number":4,"created":"2020-08-05 16:41:51.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/10/744210/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/744210/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/744210/4"}}},"commit":{"parents":[{"commit":"fe79ef22b8edadc02c54435537a6cc947960906f","subject":"Merge \"[OVS][FW] Remote SG IDs left behind when a SG is removed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe79ef22b8edadc02c54435537a6cc947960906f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-31 15:33:47.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-08-05 16:41:23.000000000","tz":0},"subject":"Change \"propagate_uplink_status\" default value to True","message":"Change \"propagate_uplink_status\" default value to True\n\nWhen \"uplink-status-propagation\" extension is enabled, new ports\ncreated will default the value of \"propagate_uplink_status\" to True.\n\nCloses-Bug: #1888487\n\nChange-Id: If1e533a61aeebbb4761d669c516fe86a4381765c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/901369854651404397a379d4ed49b6c41331a762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/901369854651404397a379d4ed49b6c41331a762"}]},"branch":"refs/heads/master"},"bc7d18826a1eb95b4f0e9ae10a58f4cac1eee510":{"kind":"REWORK","_number":5,"created":"2020-08-06 09:38:29.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/10/744210/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/744210/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/744210/5"}}},"commit":{"parents":[{"commit":"fe79ef22b8edadc02c54435537a6cc947960906f","subject":"Merge \"[OVS][FW] Remote SG IDs left behind when a SG is removed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe79ef22b8edadc02c54435537a6cc947960906f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-31 15:33:47.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-08-06 09:38:01.000000000","tz":0},"subject":"Change \"propagate_uplink_status\" default value to True","message":"Change \"propagate_uplink_status\" default value to True\n\nWhen \"uplink-status-propagation\" extension is enabled, new ports\ncreated will default the value of \"propagate_uplink_status\" to True.\n\nCloses-Bug: #1888487\n\nChange-Id: If1e533a61aeebbb4761d669c516fe86a4381765c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc7d18826a1eb95b4f0e9ae10a58f4cac1eee510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc7d18826a1eb95b4f0e9ae10a58f4cac1eee510"}]},"branch":"refs/heads/master"},"d4ae3f75a542112c92130828f992eec17d5bf0ba":{"kind":"REWORK","_number":6,"created":"2020-09-07 15:53:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/10/744210/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/744210/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/744210/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/744210/6"}}},"commit":{"parents":[{"commit":"b78a953d7f14a337662cdbae9328498ba0c9168c","subject":"Merge \"Force ncat install for functional/fullstack jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b78a953d7f14a337662cdbae9328498ba0c9168c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-31 15:33:47.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-09-07 15:52:52.000000000","tz":0},"subject":"Change \"propagate_uplink_status\" default value to True","message":"Change \"propagate_uplink_status\" default value to True\n\nWhen \"uplink-status-propagation\" extension is enabled, new ports\ncreated will default the value of \"propagate_uplink_status\" to True.\n\nCloses-Bug: #1888487\n\nChange-Id: If1e533a61aeebbb4761d669c516fe86a4381765c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4ae3f75a542112c92130828f992eec17d5bf0ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4ae3f75a542112c92130828f992eec17d5bf0ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
