)]}'
{"id":"openstack%2Fneutron~744439","triplet_id":"openstack%2Fneutron~stable%2Frocky~I3cc1f1b7dc480d54a7cee369e4638f9fd597c759","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1864822","hashtags":[],"change_id":"I3cc1f1b7dc480d54a7cee369e4638f9fd597c759","subject":"Don\u0027t check if any bridges were recrected when OVS was restarted","status":"MERGED","created":"2020-08-03 09:54:36.000000000","updated":"2020-08-07 13:26:26.000000000","submitted":"2020-08-07 13:26:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744439-1596806786901-de949bf5","meta_rev_id":"86f55988abd57fdb47f1b49fc5aae2e69453aeb8","_number":744439,"virtual_id_number":744439,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-08-07 10:22:44.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-07 13:26:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-07 10:05:16.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-05 13:57:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-08-06 20:28:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-08-06 20:28:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-05 13:57:53.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-08-06 20:28:17.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-08-07 10:05:16.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-08-07 10:22:44.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-07 13:26:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"899b94837a43eb04b397ec063f43d8c5b2322916","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-03 09:54:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26064384a8e52d0f6c915894b59718d1d0cc88df","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-03 09:55:55.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 2f4bb76338433f2d3afc10facadee95876b9a5b7","accounts_in_message":[],"_revision_number":1},{"id":"08d996124dfe77d36b913f56221aca2bf1e1b478","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-03 10:03:52.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/744439/1 : FAILURE in 8m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4e79205ae6d031e5d2dabf7a34ac5ef48e052751","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-03 10:13:33.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/39/744439/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/39/744439/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f07f2496cb85c6607cb0708089dc7ccf52336acc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 15:37:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/e4e71d8a1a0944899272c7bc00a2093a : SUCCESS in 48m 10s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/584c7e7f34be44cba3bd6c6abadb04af : SUCCESS in 37m 36s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/0d51b568370246c78062ec14577ac1d8 : SUCCESS in 1h 09m 38s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/f28471de159c4eeeb35cce90ad5e6af0 : SUCCESS in 35m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/12b78989fc734c0a8266e2ae5922fd3c : SUCCESS in 46m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f66feff622c1423da54917d4ad137b22 : SUCCESS in 13m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d20d4721a1a64b4e86c2c88d279588cb : SUCCESS in 28m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f21dfc45fbc64186a0a6de009bcc08a4 : SUCCESS in 22m 22s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/60108608ce754865b4a69e8ee43a3faa : SUCCESS in 7m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/83e380256b25482ea640916c5d8b154c : SUCCESS in 1h 18m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fe04fb6ff8674fa5a4f33dd3a6390a2e : SUCCESS in 33m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/46b19115dcce4b97ba480c7d1c1e6761 : SUCCESS in 1h 26m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/79e40eed0316414b91bb375d0b02b763 : FAILURE in 40m 25s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e9edfe603dad4be2905824d43c4e700f : SUCCESS in 1h 49m 35s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a4ac0db67a7145c68632554004a7228f : SUCCESS in 1h 21m 05s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8e4a5462728d401ba27d20d503d428db : SUCCESS in 1h 33m 23s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/21bb15391f4f41e3b0509a80f34569b5 : SUCCESS in 1h 25m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fc984ed141d4f6b8f24e846891b6e7d : FAILURE in 11m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ad4afc3115a42cd9d17a39e71885d05 : SUCCESS in 36m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8f64ce9b6cc527ca7f41c9fbc0bedcad5b47cbc5","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-05 13:57:53.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck LGTM and passing in queens","accounts_in_message":[],"_revision_number":1},{"id":"fa76d52f8e9a859e7b3150033e7af8221a133060","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-05 14:03:05.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/744439/1 : FAILURE in 2m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b4035062198a7b453142bceb00a1c83fe3ad50d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 18:46:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/461cc4acfaff46d0903a285cfcbddc3d : SUCCESS in 59m 38s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/f1a4aeb0c48f48dea208026f866b3ba9 : SUCCESS in 34m 19s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/3d9974d0111e4261a716744f5c9d2baa : SUCCESS in 1h 09m 26s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/d4ad061e55ca40a3adb32c35586a72b2 : SUCCESS in 1h 15m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/8c8176080457452390b45bb939744bd3 : SUCCESS in 57m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2f7f187b76d4fddb9f79c365d7b1c93 : SUCCESS in 13m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4902383e5edf4c5aaa1a608521672c9d : SUCCESS in 24m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e98b9a98526a40c9b2bba54062305dbc : SUCCESS in 25m 24s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c6ec8faf10a441c99323f5e27d744ba9 : SUCCESS in 8m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/eda64380595e4bef9dc0b91981574155 : SUCCESS in 1h 40m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/04f61196f6184f14b6da03bc9615f732 : SUCCESS in 1h 10m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4bdce416a364b32bcd1ca59cdc82a0c : SUCCESS in 1h 23m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f2c7006057584a3ea2b425cb799d7e05 : SUCCESS in 53m 00s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b22eb0fca0ff4836961dd148a05fb818 : SUCCESS in 1h 47m 36s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1c679d84969a43bdb1d1cd353c043edc : SUCCESS in 1h 26m 59s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0a5d907722eb4e42a8177abb0285a2c1 : SUCCESS in 1h 27m 01s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/74fb98ab06cb45a197fb522dc569cf7f : SUCCESS in 1h 34m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f7c332bc82c45bab5eefdf1bc511605 : SUCCESS in 37m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09520bb5e2a74672898c27a8dcaae097 : SUCCESS in 35m 16s","accounts_in_message":[],"_revision_number":1},{"id":"90ba272cadc27bb56ea37211dce26aa2ad6d5070","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-05 21:48:09.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/39/744439/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/39/744439/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"b3ddd70a15187558214b5fd30125d0a2402798fc","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-08-06 20:28:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e3d857a8f400517f6c243a786121d1bff14e75ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 20:28:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"98db854bfddc623480ba296afc87a57626b5de9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 22:37:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/cce2900da94943be8a8c6aa98852255a : SUCCESS in 1h 30m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f53193ad6c694687bbe6c53599d120f6 : SUCCESS in 13m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc3b4b26f53d4e51b235cbdc419da75e : SUCCESS in 27m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1a3329d66fa24f6c85202075c3064610 : SUCCESS in 27m 35s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d032a315fbda407792d6b4f1f46a9044 : SUCCESS in 7m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c6837fa1a2ed49fcb1e2cbf5da66b5f5 : SUCCESS in 1h 37m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a1edf22bb8844b94944fac149aaccd11 : SUCCESS in 1h 03m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b751027df2b8435c987f28a9b61de0be : SUCCESS in 1h 56m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2cf271b8ea574110a3d3e0fcda9a656e : SUCCESS in 45m 35s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/fc2d70d190b14bd7badcd95b95463e16 : SUCCESS in 1h 31m 40s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ea579bd0f7a84c18be9e9f9d06584c13 : SUCCESS in 1h 33m 12s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d6b08be60dfd4345aa2515fede119c7f : FAILURE in 1h 07m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef78628a39b2455699f6d7b236fc56fd : SUCCESS in 34m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f6f1a10ca4c4dcab719f3907f49ba61 : SUCCESS in 33m 40s","accounts_in_message":[],"_revision_number":1},{"id":"9a7a8dcdc0ed15435f2b51ac905129aebedb40b1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-07 09:56:47.000000000","message":"Patch Set 1:\n\nrecheck neutron-tempest-iptables_hybrid","accounts_in_message":[],"_revision_number":1},{"id":"96502e9f123a385c37892814080c3eacdb41b323","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-07 10:05:16.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/744439/1 : FAILURE in 7m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a415c4d6d50f78ab4ffa0edf149dc02dbc6291b9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-07 10:22:44.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/39/744439/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/39/744439/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"b9804ffc9eb4a053551e014495a99343db27b66a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 11:48:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/fd41fd3f4c22414dae168d1572335f87 : SUCCESS in 53m 29s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/2a0b801b344f4a6ea117b06774d28ed0 : SUCCESS in 38m 59s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/4471cf58353041ebaca42373d3316054 : SUCCESS in 1h 17m 20s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/56114ae567534098af31b363fc56a711 : SUCCESS in 1h 10m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/9976595b42f14825a3630ba9aa463c4d : SUCCESS in 1h 18m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85067899120547c7b5fb343c81a86718 : SUCCESS in 13m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47e5ccf4be5345a7a64e82d4559df292 : SUCCESS in 23m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4f2cccca432c467bac676747e619e541 : SUCCESS in 31m 47s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/01cbf0d4078b44c38833ab9005a55ca4 : SUCCESS in 7m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f8c9218edd654786833a2c1dde4653df : SUCCESS in 1h 22m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/90d393c613b24171a6248a61ba2ddc58 : SUCCESS in 54m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69b15a80af164ec1ac5b3537406199db : SUCCESS in 1h 33m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7fd9b3b2e2954e79b4a2f3cf3e868a14 : SUCCESS in 51m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c9e1767c0c0c4054affff9f66d17019b : SUCCESS in 1h 47m 43s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a418e3a23b66443395e7708d5e2079d4 : SUCCESS in 1h 21m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/bb9218dd7dcd43faa1fbe51677129d11 : SUCCESS in 1h 24m 17s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/76c0e2895e644ff79034927d1f6a4219 : SUCCESS in 1h 23m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b69e52b6fb904edda2a171a95ff8f908 : SUCCESS in 24m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f09299ac47d64b5c857ba640d7e481b3 : SUCCESS in 37m 55s","accounts_in_message":[],"_revision_number":1},{"id":"df374e881c4f31a5bfaed3ced3a170dbfa997fef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 11:48:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7450a784cf3547f8633bc87731b25d4dc74deb59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 13:26:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/2fe7e091fdb341e48cbc94977b7317bf : SUCCESS in 56m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1caa00be3584178afd438c468220feb : SUCCESS in 10m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e5c63fe9df94463bc5a530581c1ae9d : SUCCESS in 28m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fd27498246904472b5640627d4f8a3e8 : SUCCESS in 22m 10s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/64449620d2db4d0ca7579c613f97ea43 : SUCCESS in 7m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/523d8ca975ff478bbd6dfe12e2ecafed : SUCCESS in 1h 30m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c567dfdd7b0442a7aeaae4d096df6d8b : SUCCESS in 49m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae2b0707cbbd42c7bc65e2ff58e98c6d : SUCCESS in 1h 21m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/00e9dd75d59b4dd9be70228c7af3c005 : SUCCESS in 39m 05s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7864433f85ed405e9bf85b352699b15b : SUCCESS in 1h 28m 16s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3fd9426467044e6dae94e1494b81817e : SUCCESS in 1h 24m 13s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/32021297995a47df90b39a5960c8847d : SUCCESS in 1h 23m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ae4b656083442ebb103bbc8a92cf3bb : SUCCESS in 28m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4970bcdd2dd4795bbb7a3346d5198de : SUCCESS in 28m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7057a70bf8dd1de59701d69b628c0cc32fe0b514","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 13:26:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8bba2a8f6e6500c1b996ada7cc77e6d48379b995","revisions":{"8bba2a8f6e6500c1b996ada7cc77e6d48379b995":{"kind":"REWORK","_number":1,"created":"2020-08-03 09:54:36.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/39/744439/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/744439/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/744439/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/744439/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/744439/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/744439/1"}}},"commit":{"parents":[{"commit":"47ec363f5faefd85dfa33223c0087f4444afb5b9","subject":"Ensure drop flows on br-int at agent startup for DVR too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47ec363f5faefd85dfa33223c0087f4444afb5b9"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-05-11 10:09:46.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-03 09:53:51.000000000","tz":120},"subject":"Don\u0027t check if any bridges were recrected when OVS was restarted","message":"Don\u0027t check if any bridges were recrected when OVS was restarted\n\nIn case when openvswitch was restarted, full sync of all bridges will\nbe always triggered by neutron-ovs-agent so there is no need to check\nin same rpc_loop iteration if bridges were recreated.\n\nConflicts:\n    neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py\n\nChange-Id: I3cc1f1b7dc480d54a7cee369e4638f9fd597c759\nRelated-bug: #1864822\n(cherry picked from commit 45482e300aab781064281b3e06a17ae488831151)\n(cherry picked from commit b8e7886d8b52a144241c0cc2c30b670ce5ddf926)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bba2a8f6e6500c1b996ada7cc77e6d48379b995"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bba2a8f6e6500c1b996ada7cc77e6d48379b995"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
