)]}'
{"id":"openstack%2Fneutron~744440","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I3cc1f1b7dc480d54a7cee369e4638f9fd597c759","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1864822-stable/queens","hashtags":[],"change_id":"I3cc1f1b7dc480d54a7cee369e4638f9fd597c759","subject":"Don\u0027t check if any bridges were recrected when OVS was restarted","status":"MERGED","created":"2020-08-03 09:55:55.000000000","updated":"2020-08-06 22:20:04.000000000","submitted":"2020-08-06 22:20:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744440-1596752404224-1ab9881c","meta_rev_id":"44c34218264d26ee91a5ffc58cc1b5b871466dba","_number":744440,"virtual_id_number":744440,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-08-03 10:16:39.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 22:20:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-03 10:04:14.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-05 13:58:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-08-06 20:27:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-08-06 20:27:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 10:04:14.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-08-03 10:16:39.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-05 13:58:00.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-08-06 20:27:56.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-08-06 22:20:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"787116c9c6265a8222111368787872855561272d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-03 09:55:55.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"b8467985a06e523871a7d758cb8e071fbc7102f0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-03 10:04:14.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/744440/1 : FAILURE in 7m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c79740b1174514ad1bc9b4632215d63d6310e672","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-03 10:16:39.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/40/744440/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/40/744440/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"81cb4a3c0c1dbb2c6b92b59be5ce56e14d58bf0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 15:53:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/2310c82755654b8092500a3c25861b80 : SUCCESS in 58m 48s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/e6c4b009964c47bd9428754c9943e768 : FAILURE in 1h 18m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/e47dd085ec964845833a0576e45fe120 : SUCCESS in 59m 08s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/b687b868b1964d74b39de849c4d457fb : SUCCESS in 43m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f70ba4b7f72f4e10b143f0f9af1d8f50 : SUCCESS in 11m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c3b2efe37c743eabb262d4740f22bcc : SUCCESS in 13m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/382b3aa3c68942dfae0a20ce559dd886 : SUCCESS in 25m 36s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a5d09d65eb604a8f8d27143c0be47ae9 : SUCCESS in 6m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ead1c8a9ce9b40df80718b86d836dcde : SUCCESS in 1h 14m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/46a54f2eb0a94605a03d5308da2bd59a : SUCCESS in 55m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c76093194e98458b8f6821faa0ebed31 : SUCCESS in 1h 14m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/72a73c5f6d554e52b8834834b73b76a2 : SUCCESS in 44m 24s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/8fa1f2e3bb5f49f0b6137ac963da74c6 : SUCCESS in 1h 46m 29s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bfb93a3b579446e69bdaae20521ac466 : SUCCESS in 1h 04m 02s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/96b5ed23fc474c9c969a9ff1e35a03d4 : SUCCESS in 1h 15m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8758f0a471640a4a8fb0878698be3b6 : SUCCESS in 17m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c9123f14054cc2b60e9e2fba845a98581939167a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-05 13:58:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e587d8a4123cc96ea7177be839534021da7c256f","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-08-06 20:27:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"be75fc92c0d56ccc804fdcd7a3d6d7a4465f5e05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 20:28:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9c05698878614782334c2876d028318dceeda7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 22:20:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/e47bd85774e64643bf471fed923e78c3 : SUCCESS in 1h 00m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06ac51faffc340b4a17decb1fd071a8e : SUCCESS in 10m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6812668448f844c2b389b36bb74148fc : SUCCESS in 22m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3b764bd6529b48f8b28f079fdb73aba6 : SUCCESS in 25m 23s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/984c2089d02c4b7fafa73876b79c8853 : SUCCESS in 6m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/92ba9d36543b41adbac4f36a90c2d1f0 : SUCCESS in 1h 12m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/95ad9eaf974242bd9e445d97618cf2ef : SUCCESS in 1h 13m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1f97c5c414d4fbd87a21aad3d7335f2 : SUCCESS in 1h 28m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a2adf0b9acce4eb981b84db7ae417b01 : SUCCESS in 46m 37s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a3187696e39a4098b83f07f415045d2a : SUCCESS in 1h 49m 02s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/43f9f417ecb34ec4818bb6a35dc338ab : SUCCESS in 1h 16m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62493d8df7b5452099780ae86cebb48a : SUCCESS in 35m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b84f71c08218ddd7293b73846bcd7fac5d90b72d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 22:20:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f4bb76338433f2d3afc10facadee95876b9a5b7","revisions":{"2f4bb76338433f2d3afc10facadee95876b9a5b7":{"kind":"REWORK","_number":1,"created":"2020-08-03 09:55:55.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/40/744440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/744440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/744440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/744440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/744440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/744440/1"}}},"commit":{"parents":[{"commit":"8a173ec29ac1819c3d28c191814cd1402d272bb9","subject":"Ensure drop flows on br-int at agent startup for DVR too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a173ec29ac1819c3d28c191814cd1402d272bb9"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-05-11 10:09:46.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-03 09:55:55.000000000","tz":0},"subject":"Don\u0027t check if any bridges were recrected when OVS was restarted","message":"Don\u0027t check if any bridges were recrected when OVS was restarted\n\nIn case when openvswitch was restarted, full sync of all bridges will\nbe always triggered by neutron-ovs-agent so there is no need to check\nin same rpc_loop iteration if bridges were recreated.\n\nConflicts:\n    neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py\n\nChange-Id: I3cc1f1b7dc480d54a7cee369e4638f9fd597c759\nRelated-bug: #1864822\n(cherry picked from commit 45482e300aab781064281b3e06a17ae488831151)\n(cherry picked from commit b8e7886d8b52a144241c0cc2c30b670ce5ddf926)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f4bb76338433f2d3afc10facadee95876b9a5b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f4bb76338433f2d3afc10facadee95876b9a5b7"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
