)]}'
{"id":"openstack%2Fneutron~744465","triplet_id":"openstack%2Fneutron~master~I499693db900d88ad5e041cbb25baedf83817313e","project":"openstack/neutron","branch":"master","topic":"bug/1890064","hashtags":[],"change_id":"I499693db900d88ad5e041cbb25baedf83817313e","subject":"Fix issue with decorator in lower-constraints job","status":"ABANDONED","created":"2020-08-03 13:01:56.000000000","updated":"2020-08-06 06:22:40.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"74f35f9d2ce09ec609941d828c8ee474395141ea","_number":744465,"virtual_id_number":744465,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"date":"2020-08-04 12:57:18.000000000","_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"date":"2020-08-04 20:10:34.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-04 14:29:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"date":"2020-08-05 00:54:06.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2020-08-04 17:24:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":1,"date":"2020-08-04 12:52:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"value":2,"date":"2020-08-04 12:09:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-08-04 12:50:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-08-05 04:15:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"value":2,"date":"2020-08-04 12:09:18.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 14:08:55.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-08-04 12:09:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-08-04 12:50:55.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-08-04 12:52:38.000000000","updated_by":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"reviewer":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"state":"REVIEWER"},{"updated":"2020-08-04 12:57:18.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-04 14:29:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-04 17:24:44.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2020-08-04 20:10:34.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-08-05 00:54:06.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"}],"messages":[{"id":"93f3f14294cd949ffbca5dd2281445e59311fdad","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-03 13:01:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd10792430f13381fab8b16e8eb58e2da73cba41","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-03 13:29:12.000000000","message":"Patch Set 1:\n\nNice, I was planning to send similar fix but better when it is already done!\n\nMaybe add a Closes-Bug (I mentioned neutorn in https://bugs.launchpad.net/neutron/+bug/1890064 ) as we will need it in unssuri and train too","accounts_in_message":[],"_revision_number":1},{"id":"40d7c825f8a7f875c756429d0cc7e87c30452262","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-08-03 13:34:01.000000000","message":"Patch Set 1:\n\n* sriov-test http://52.27.155.124/sriov/744465/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d94a525c1c11511d0ce475e907c9e9cde8f03fa4","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-08-03 14:08:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae0a6218123a181189e9a33b3e4e26f6ba5f3702","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-03 14:55:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/744465/1 : SUCCESS in 1h 52m 31s","accounts_in_message":[],"_revision_number":1},{"id":"db559ceb86ea0604813007ef24becae09e96c03e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 19:30:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2196870b592747019356fb456be476ca : SUCCESS in 54m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7abfc741bec449ab8a4ce9f0dac80b59 : FAILURE in 33m 33s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7483f5fae2ba4bcfa48b2db6dc8e458d : SUCCESS in 1h 01m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0da86b2086ee49329c2c37f62dc1eae7 : SUCCESS in 1h 01m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2e077f92e2964a50b1aac89e96c986ea : SUCCESS in 1h 02m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4f5a0468c43045c7bb5b9839638666ee : SUCCESS in 56m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32c6b169453641c6abbc0f38e196c926 : SUCCESS in 33m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fb1ff33295d4bf18b178032b06c4b20 : SUCCESS in 27m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb3e4f1b3b8f424aae51005106ca0328 : SUCCESS in 12m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/def9968462b14154945a3d51f4517f11 : SUCCESS in 26m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8583fbe3ab54793a71b7dbc9fc719b4 : SUCCESS in 27m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74780da2385341098f080b2b84f23705 : SUCCESS in 9m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a7fc725a0891479791dc5138725f21ff : FAILURE in 3m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b96f34e7293746c1ba35848d64b7ead4 : SUCCESS in 47m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bde4dc12b075419a8782f68e9045bf01 : SUCCESS in 50m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a6b1281339df4327b08cd038293dc688 : SUCCESS in 1h 22m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/18d5d841eb694851b10ae9db09d87c11 : SUCCESS in 1h 08m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/20a71076a7b44b4d80ae00d271368075 : SUCCESS in 1h 11m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c089637ef42843ed9fd3026d7486f11a : SUCCESS in 1h 00m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/406b5c9576dc41a2b223002bebb417c3 : SUCCESS in 1h 22m 42s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/526feef370f549d9a86598dc9aaece29 : SUCCESS in 1h 12m 18s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/699550d04adc4544a6d699a49abf10b8 : SUCCESS in 1h 28m 39s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6c936ba4571143cf986cb874eb11119a : SUCCESS in 1h 30m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/37f63e3e9dcf43b0ad3afd7faf7c130e : FAILURE in 1h 18m 22s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1ce9d28a97c84f1ea06f2bc5141ce805 : SUCCESS in 1h 22m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/612ba31f40234429ad7db1d07d185815 : SUCCESS in 1h 12m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/16ca7b95c3ee4607bd996a7937712d5c : SUCCESS in 2h 07m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b29bf3a8db14fa9b7e94148511ad69d : SUCCESS in 1h 03m 06s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/10d78474c6324a9aa020f3d42bc20d1b : SUCCESS in 1h 29m 39s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/04ba09794ada4355adf3e2f9143180f2 : SUCCESS in 1h 09m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c4e28d512cbc4ded9f788fcd01bf6483 : SUCCESS in 33m 08s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/dea4440414554a78ae182d1a75e22c4d : SUCCESS in 1h 27m 16s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2acfd72915b74273a87cc6d708424faa : SUCCESS in 1h 47m 17s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f0bfa0dbdb684324afb95b1b779c2f13 : SUCCESS in 2h 00m 53s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9c5b459719fb4648a998f8d59146cb75 : FAILURE in 2h 40m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d88793618fc44a369d7d8b9920f2d155 : FAILURE in 2h 03m 13s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/932f44801a154f33a2081ccdd29e54da : TIMED_OUT in 3h 07m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6eb068e9aecaff79519a90e04af2419e2f5bd335","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-04 06:58:35.000000000","message":"Patch Set 1: Code-Review-1\n\nNeeds requirements.txt too:\nERROR: Package \u0027decorator\u0027 is constrained to 4.0.0 which does not match the minimum version specifier 3.4.0 in requirements.txt\nValidating lower constraints of test-requirements.txt\n\n(+ the bug link)","accounts_in_message":[],"_revision_number":1},{"id":"c09d0be50a3520bca552c96b7611017aea482aa2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-04 07:28:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c9c1b11cc9559b31eb1142e468e7636b92aceb9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-04 07:29:22.000000000","message":"Patch Set 1:\n\n\u003e Needs requirements.txt too:\n \u003e ERROR: Package \u0027decorator\u0027 is constrained to 4.0.0 which does not\n \u003e match the minimum version specifier 3.4.0 in requirements.txt\n \u003e Validating lower constraints of test-requirements.txt\n \u003e \n \u003e (+ the bug link)\n\nThanks Bernard, I changed my direction, and as I see it is better to wait some common OpenstackWide solution :-)","accounts_in_message":[],"_revision_number":1},{"id":"3dcc9d361e494462de80038f50b5a653a323854d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-04 07:29:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"314f5fce4b72cb04d6cc485643935ad73011b8c5","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-08-04 08:07:13.000000000","message":"Patch Set 1:\n\nReading the mailing list thread, my understanding is that the problem happens due to inconsistency of lower-constraints file while pip behavior changes across versions.\n\nLooking at our lower-constraints job, for example [3], lower-constraints-2.log [4] which installs neutron updates some requirements. Theoretically it should be done in lower-constraints-1.log [5], but some entries in our lower-constraints.txt is out-of-date, so the second step which installs neutron updates some requirements including dogpile.cache.\nI believe we should fix lower-constraints to make all entries consistent.\n\noslo.cache recently updated dogpile.cache requirement to \u003e\u003d0.9.2 as part of focal migration [1].\ndogpile.cache 0.9.2 or later requires decorator\u003e\u003d4.0.0 [2].\n\nI think this change makes sense and it is a good chance to clean up lower-constraints inconsistency.\n\n[1] https://review.opendev.org/#/c/734706/\n[2] https://github.com/sqlalchemy/dogpile.cache/blob/rel_0_9_2/setup.py\n[3] https://zuul.opendev.org/t/openstack/build/0058b17057d6486eba2b0cf964c79520/logs\n[4] https://496c145d7be383fffb9e-43001c5c460bd4e8e3c2be96e8c7a69c.ssl.cf1.rackcdn.com/744210/1/check/openstack-tox-lower-constraints/0058b17/tox/lower-constraints-2.log\n[5] https://496c145d7be383fffb9e-43001c5c460bd4e8e3c2be96e8c7a69c.ssl.cf1.rackcdn.com/744210/1/check/openstack-tox-lower-constraints/0058b17/tox/lower-constraints-1.log","accounts_in_message":[],"_revision_number":1},{"id":"67271d47dce9713df6c6cdfd647f636fc5e9913f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-04 10:56:59.000000000","message":"Patch Set 1:\n\nbased on last Akihiro\u0027s comment I think that we should restore this patch and move on with it.","accounts_in_message":[],"_revision_number":1},{"id":"a18ab54a8ed89234cf287f8d25e8b4afc8672e15","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-04 11:14:29.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"61101ce74824af9e4e0627023120364c7a01dae4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-04 11:18:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"53c5ecf64eb147453c0b16614fb3638849f0a267","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-08-04 12:09:18.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+2\n\nrequirements-check is passing now","accounts_in_message":[],"_revision_number":2},{"id":"b366c48ebf12139759964270c9a053be8308de83","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-04 12:50:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1367fbe081a47c31a0cb7c9ec6eae3e547fe9941","author":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"date":"2020-08-04 12:52:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"803b250c00b256c5c3d497ba0bf99894db668b4a","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-08-04 12:57:18.000000000","message":"Patch Set 2:\n\n* sriov-test http://52.27.155.124/sriov/744465/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ba4cef8142d1ec0acfbacf0d99e74ef2d9f25530","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-04 13:06:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/744465/2 : SUCCESS in 1h 47m 39s","accounts_in_message":[],"_revision_number":2},{"id":"7bed5ff804233f29a087b56a33ac4ed94d489bd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 14:29:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a46a12dfd11d46ffa1ae4ed56415ef7d : SUCCESS in 56m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/232e097799c64eaf8f7934c660b1f3ed : SUCCESS in 36m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/08070ad58129441abc815040d9c00685 : SUCCESS in 1h 04m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/df59bc7387134d8eace4113d43de75e0 : SUCCESS in 1h 42m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3d24b8e710aa49b78d90e76f97c8dad1 : SUCCESS in 1h 48m 40s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0f229f14e38d4b79a5d38fe40483fcfa : SUCCESS in 1h 40m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c64b144fa3140eba469ec4bd13a5622 : SUCCESS in 56m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2db54dff67b94747be14880ac4218d34 : SUCCESS in 50m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3cfa3b9a4c045e2b25c3c6825a9a348 : SUCCESS in 19m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d38a1f06240d40d6bc6f20ab8489d57d : SUCCESS in 48m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/482b86db07d84716a60c183602509f71 : SUCCESS in 48m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db9bf46c818a4c148bdeba252b99b005 : SUCCESS in 9m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/088ad1206ae947be80eab336422cbbed : SUCCESS in 4m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c95fc561b2bb4514a4795224f899f0d8 : SUCCESS in 49m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c15ebf5b338c43dfb58130b0aab0062b : SUCCESS in 48m 28s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6d25bfdeb6dd4d07a05639f90711fd33 : SUCCESS in 1h 00m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/449c26f9e2d946ccb5f0fffbe008e314 : SUCCESS in 1h 13m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5333687a16334c28bb0395b3d8fa8569 : SUCCESS in 1h 54m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cbf18ef4cb114e1e870207a7142df014 : FAILURE in 1h 00m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/06e7aef929d741438dbdf6795d747217 : FAILURE in 1h 06m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d8e7fab19ecb426fbd7db5ff71717084 : SUCCESS in 1h 17m 02s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f4102b627ba44cefac251a7d885ebce4 : SUCCESS in 1h 19m 51s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1cf479afbe8a480980addc4a98d1ec6c : SUCCESS in 1h 17m 27s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cfef4986136d4875b5998acf51661989 : SUCCESS in 1h 43m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3ff07d13efae400392614e2d86ae6a2a : SUCCESS in 1h 05m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9e42b254a7124652a45d067d2dece246 : SUCCESS in 1h 14m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/166c6276b126493daf9b45b005c679a2 : SUCCESS in 1h 53m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4aebeb69648241e6ac9b0d5c2fb1b1f6 : SUCCESS in 55m 52s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/8b2558dab3d141a786b9a28077dda71b : SUCCESS in 1h 10m 17s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8134b99f703e4faabcbb24b399f15964 : SUCCESS in 1h 10m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3a91accb37f44bc691a292fd1ccca7ba : SUCCESS in 38m 10s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/22efeab4a2544bbfad07789e526977b5 : SUCCESS in 1h 17m 48s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/747ace36b83146e7892379d97f0a09cd : SUCCESS in 2h 19m 44s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/82b2c2ea85024770b5fade33a235946e : SUCCESS in 1h 46m 33s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1a4488bfbe554d17aeb84a6780830883 : SUCCESS in 2h 20m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1601d2b5edc04eb18c074bfdf45cc6fa : FAILURE in 1h 45m 16s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/12d5d4dac5124e76807642c106827883 : TIMED_OUT in 3h 05m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"46f3630b3bb4fad8503026cb568b5a39cce37f73","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-08-04 17:24:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6bad3506d6488aeb969cdf184a46c20b44f47d9d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-04 20:10:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d3c6a20776b92bfaafbeec7ce89221425ec19b89","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-08-04 23:04:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ec06a26880bb7abbd8c4dc81a240783431b538d0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-05 00:54:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/744465/2 : SUCCESS in 1h 49m 05s","accounts_in_message":[],"_revision_number":2},{"id":"fa0d998adf83ae38610b3be9e93484d8ae15ad1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 02:10:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/edc05ae146924bd9bb591aa4ab766bc2 : SUCCESS in 35m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/df43f987a55547d5bac93acddcf58351 : SUCCESS in 22m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/920b149ac2da435facf6a847c49cb4cc : SUCCESS in 46m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a6df1fc319bd4fc2b0e66151bd79e1f5 : SUCCESS in 48m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ffb1357062234e738ba066083153bdf5 : SUCCESS in 42m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f97bab4ca45b4dc8ac19bd61130991a3 : SUCCESS in 39m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3b26e6d94eb4a0cbff3ab47d7ee782b : SUCCESS in 21m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1caeb6cdc6434fbda9c08003a2dceb53 : SUCCESS in 38m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e47a98cb7ab43fdb75d192e33be69c8 : SUCCESS in 15m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26a40885f3794789a41b2f0a61e9a733 : SUCCESS in 16m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02d96e855c2040c3b698e5c5ecacd010 : SUCCESS in 33m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40305804634b48c9a5d5ed0cab2d94f4 : SUCCESS in 8m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/88836e71e0e042708a51e40bc9aa883f : SUCCESS in 3m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/055a42c23d754c01a91aa294132ae113 : SUCCESS in 1h 05m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5751912264a940e1bbaf604f8348afca : FAILURE in 44m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2830ce8d3e5e43c38ad70783bde1270f : SUCCESS in 1h 22m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2618c289686242ffbd1593ac24a6e37f : SUCCESS in 1h 26m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ecb65086d735454ca93d619a8fadd917 : SUCCESS in 1h 14m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0f5cebf8e5564915ab81591bf029af6e : SUCCESS in 1h 06m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/830e56f561a9460e8ddee88c9e1ca5a3 : SUCCESS in 1h 27m 02s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b0319523757449c3b3f0961043eb6ff4 : SUCCESS in 1h 29m 32s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/40170732a0f649338905bc29da17c96b : SUCCESS in 1h 11m 47s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ab2507a3ea9b4a59a424f535775bb2e7 : SUCCESS in 1h 40m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0fb5bafcb3fe42e4afaf5b9ad266f7f8 : FAILURE in 1h 21m 35s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f9dadb7a7a664b7f961c2fc7f77741f2 : SUCCESS in 1h 31m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c84117f32adc4814b51331124923275d : SUCCESS in 1h 34m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1bea2834494246e7a5a2a321b1b2c132 : SUCCESS in 1h 57m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4015149d51c4f29a32962ecc42af640 : SUCCESS in 57m 54s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/71ab087ad6344db6b55693b073038d20 : SUCCESS in 46m 03s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0242b0fb648340729cc0cb0e85997f2e : SUCCESS in 1h 36m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c977f597c4e3452b8df08a1d3bbcfade : SUCCESS in 42m 17s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ce72e76f2a644ecf99139b29ad03ab27 : SUCCESS in 57m 59s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/13cdd04c67124a6ca1f1136276757863 : SUCCESS in 1h 46m 31s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4250bb1fbbe848879f6399d666ac51ae : SUCCESS in 1h 15m 29s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/3dceb1dfa84b4e05a4cfd7fd349668d0 : SUCCESS in 2h 37m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/79ee2d02afc44a43ac6d562a23821da3 : FAILURE in 1h 56m 57s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/842932f2ff144f47a4aab25e00f6a3ab : TIMED_OUT in 3h 03m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b14c94e698bbefe9fc6c966e02ce9a92e0659a2e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-05 04:15:11.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)\n\nIt seems that virtualenv bumb solved this issue","accounts_in_message":[],"_revision_number":2},{"id":"7fa74d83f95ed079ba001937d8e0f907e5baaea6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-06 06:22:40.000000000","message":"Abandoned\n\nit is not needed anymore","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2de14b9964a5609fb231d5d0ccc392d945ae13e8","revisions":{"ed470be31832ec32dca88adcd7b95e49d14df2b7":{"kind":"REWORK","_number":1,"created":"2020-08-03 13:01:56.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/65/744465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/744465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/744465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/744465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/744465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/744465/1"}}},"commit":{"parents":[{"commit":"fe79ef22b8edadc02c54435537a6cc947960906f","subject":"Merge \"[OVS][FW] Remote SG IDs left behind when a SG is removed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe79ef22b8edadc02c54435537a6cc947960906f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-03 12:57:15.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-03 12:57:15.000000000","tz":120},"subject":"Fix issue with decorator in lower-constraints job","message":"Fix issue with decorator in lower-constraints job\n\nChange-Id: I499693db900d88ad5e041cbb25baedf83817313e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed470be31832ec32dca88adcd7b95e49d14df2b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed470be31832ec32dca88adcd7b95e49d14df2b7"}]},"branch":"refs/heads/master"},"2de14b9964a5609fb231d5d0ccc392d945ae13e8":{"kind":"REWORK","_number":2,"created":"2020-08-04 11:18:06.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/65/744465/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/744465/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/744465/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/744465/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/744465/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/744465/2"}}},"commit":{"parents":[{"commit":"fe79ef22b8edadc02c54435537a6cc947960906f","subject":"Merge \"[OVS][FW] Remote SG IDs left behind when a SG is removed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe79ef22b8edadc02c54435537a6cc947960906f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-03 12:57:15.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-04 11:15:41.000000000","tz":120},"subject":"Fix issue with decorator in lower-constraints job","message":"Fix issue with decorator in lower-constraints job\n\nFor related mail thread, please see:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-August/016236.html\n\nCloses-Bug: #1890064\nChange-Id: I499693db900d88ad5e041cbb25baedf83817313e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2de14b9964a5609fb231d5d0ccc392d945ae13e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2de14b9964a5609fb231d5d0ccc392d945ae13e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
