)]}'
{"id":"openstack%2Fneutron~746154","triplet_id":"openstack%2Fneutron~master~Ia313b9c4028f0242cb0da8764b9921c5c30c465a","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"Ia313b9c4028f0242cb0da8764b9921c5c30c465a","subject":"\"round\" may bump 1 second to 2 if sleep takes more than 1.49 sec","status":"MERGED","created":"2020-08-13 15:08:00.000000000","updated":"2020-10-12 13:43:49.000000000","submitted":"2020-08-26 04:14:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746154-1598415295286-038e877c","meta_rev_id":"11ee7a1a05ed2c3bf1a26bc146f5b7df3a32f1c7","_number":746154,"virtual_id_number":746154,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-25 23:10:16.000000000","_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":1,"date":"2020-08-18 14:42:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-08-18 13:58:15.000000000","_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 04:14:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-26 00:06:41.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2020-08-18 13:43:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-25 09:45:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-08-18 15:44:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-25 09:45:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 15:09:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-08-13 15:09:42.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-08-13 15:09:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-13 15:31:13.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2020-08-18 13:43:44.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-08-18 13:58:15.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-18 14:42:14.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-18 15:44:15.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-08-25 09:45:34.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-08-26 00:06:41.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-08-26 04:14:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c561ea1551abeb06bfff1825b71f67c344556d1e","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-08-13 15:08:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3248ca32e9c61316e055b751e85ed76067800f8b","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-08-13 15:31:13.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"129b3dbb165ace81ae614ba8671bee4943dcff38","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-08-13 15:42:25.000000000","message":"Patch Set 1:\n\n* sriov-test http://52.27.155.124/sriov/746154/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"087e70f4417b466b4637c6b4bb3088af67605459","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-08-13 16:20:11.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2020/Arista_Tempest_Testrefs/changes/54/746154/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"5cab7f1dc1e614fe024d4ac5d1828d89f2fd4b3c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-13 17:00:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/746154/1 : SUCCESS in 1h 52m 14s","accounts_in_message":[],"_revision_number":1},{"id":"45f0377231c4fa7e1fa8336843a434d7aca8b141","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 18:10:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8044011a0d44feaa667c4b635265629 : SUCCESS in 34m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc0d770a7ffa4373b7dc328ed44d8a13 : SUCCESS in 26m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a42461578714b00adc69f25eb3e11e0 : SUCCESS in 13m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd395f02fb374e7b8add5ee43a5c517b : SUCCESS in 25m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5de111b84fba4a2e828893e7201480b5 : SUCCESS in 24m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f80f2c33d3c445f1bce9b3c47a971e14 : SUCCESS in 10m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/706da07a95fa496390a04c2c7fe9bc7f : SUCCESS in 59m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/64b210ba559b441ebce54e31749669aa : SUCCESS in 57m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7062a66b391b4e629bceb09ff6f1daad : SUCCESS in 1h 31m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0cf5f5221c8c443397bca1d94a224ace : SUCCESS in 1h 11m 21s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5483f2d49f764cafadd396091927a22f : SUCCESS in 1h 15m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aef739fd019c12bd98956ff0af05f8b03bc7ea14","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-18 09:42:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f47f6aad72c5318f5c8d026931acb49aa2d8280","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-08-18 09:44:58.000000000","message":"Patch Set 1: Workflow-1\n\nvalid points! will make some changes, so this is easy to read in the future. TY","accounts_in_message":[],"_revision_number":1},{"id":"347a4458a5c04acb0197df177ad92a9c9a0b7b36","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-08-18 13:26:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42ca44c1a3fe7eddef32375c0977a984cd346536","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-18 13:43:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3a5e9bb136287ed33412342146131ed56f59c05c","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-08-18 13:58:15.000000000","message":"Patch Set 2:\n\n* sriov-test http://52.27.155.124/sriov/746154/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"3b10bbec502f0f6afd53099fce210e33e5ac4eb4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-08-18 14:42:14.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2020/Arista_Tempest_Testrefs/changes/54/746154/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"0081f54bc67e89561d85f302e03e06008222499b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 14:50:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f7f552714b64418bdc2d90779745dfd : SUCCESS in 37m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/337ef60ce6e445da87eb653c47ab0fcf : SUCCESS in 28m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/badb2b1394274fa3ba6d4cacc6131ed2 : SUCCESS in 12m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c783ca2e0b7a4940a047e146fe8306cb : SUCCESS in 31m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26fa8fa31a3d4fc299e060f20931d6fd : SUCCESS in 25m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80574f5c97c64aab951ee4a8c261e022 : SUCCESS in 11m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/17adbb83002e48de9372fed8f55c70ca : SUCCESS in 58m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b8fd664413b402494eeb364aeb46ded : SUCCESS in 53m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9823de65d0a041ed8890ddc061e30da7 : SUCCESS in 1h 18m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d68a0b91002e4bfeb111e15185389d8b : SUCCESS in 1h 17m 51s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/dc79db80ae7549899402a948ff66bd33 : SUCCESS in 1h 17m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"55fd3d09a2e31847a5a5bdc03dfb9319b1eb8067","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-18 15:14:58.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/746154/2 : SUCCESS in 1h 47m 21s","accounts_in_message":[],"_revision_number":2},{"id":"35f5c472b5778a94fd01be305fc44af7e7ec0429","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-08-18 15:44:15.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3d6205625a35873fa30460d309070d1c7d3c9bfb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-25 09:45:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"beeb8b4aa8478c2d18afe672a1bf2ea7cafb66ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 09:45:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aac2ae2dfa009c2c9cb821f05dbfa666f5bff75a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 11:08:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5821f529a1224ea9a431aee96a253f35 : SUCCESS in 27m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4442355319014b30826e7104117119ad : FAILURE in 11m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c49560f5b8b48f1a28e6200d24f998f : SUCCESS in 25m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f95352677144412ad5099a2bd036cbe : SUCCESS in 28m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9d287e9499148f9bae4fdc1f3dc42c5 : SUCCESS in 8m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a92064724104430abf6c8c0b12a26a9e : SUCCESS in 58m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1107ceffa3b249c49fcb38bf3230a6c4 : SUCCESS in 1h 15m 50s\n\nWarning:\n  Comments left for invalid file neutron/agent/linux/dhcp.py","accounts_in_message":[],"_revision_number":2},{"id":"c7084379529c828fdd65b71170c5055069557a56","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-25 11:13:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c835aeff9d44932a8dfc2e4093a6febab3bc08f4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-25 12:09:15.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/746154/2 : FAILURE in 55m 44s","accounts_in_message":[],"_revision_number":2},{"id":"eb67b6863a21be39b640c702abd39e16e56cd1d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 15:53:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d54c6ef477114c6cadacf1e61a94d08f : SUCCESS in 39m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b42b793bb4bd4f6db73e12fd290be98e : SUCCESS in 27m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/247c77289bee415bb3ee8f6114510095 : FAILURE in 13m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79b5d4653f17407f89d51bdd05acdd48 : SUCCESS in 32m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec278f796da244aeb00aea7af47464e5 : SUCCESS in 29m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f73136f5e9104b2ca5a251a3383e47c1 : SUCCESS in 17m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/81afc519c63e41d4b6e2d402ae4d2200 : SUCCESS in 52m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/61ffbf101cd045b3aaed69b1f1d2e265 : SUCCESS in 51m 28s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5a27b2a7ae144dfc8c527fa4311ed965 : SUCCESS in 1h 11m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/801bfc036f9d4436a7175e715b08fa27 : SUCCESS in 1h 10m 34s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c266470a2a5f451c9b2bf88567f0f0db : SUCCESS in 1h 19m 16s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/linux/dhcp.py","accounts_in_message":[],"_revision_number":2},{"id":"4135e305f40deb9e331880d7689c7f7441ce0126","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-08-25 23:10:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6292bb1c5c5faf933a2847ee8830383640bb2bb9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-26 00:06:41.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/746154/2 : FAILURE in 52m 38s","accounts_in_message":[],"_revision_number":2},{"id":"8e1d8c5302ecbb845b2d05ce57e4bff10dedb006","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 00:36:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/945044fd082849e885c8b42c54ca9f51 : SUCCESS in 40m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/966de7324a9844d0add3e82744183598 : SUCCESS in 32m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55ee7a83f46341cc9b6e509757fa6047 : SUCCESS in 12m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa09de33737e4e95b6177d5d78616e99 : SUCCESS in 30m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6262e279d3b04df68c971ebd24cc216f : SUCCESS in 24m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6387d8758a43431e856120b037ce6987 : SUCCESS in 8m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e4421287ecb74296ac643e8f64f2efeb : SUCCESS in 58m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/80544f6efaf940a783777bbbebecb3e1 : SUCCESS in 54m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d6c967fdc3e24ea38c66611eef455c3e : SUCCESS in 1h 19m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c02311e3760b42bc8ed21397e5ae6e7f : SUCCESS in 1h 10m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5f2c022560a04585979899004dff8afd : SUCCESS in 1h 17m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0a2bfa0acc9018561fe6bf5946aa7f83051be883","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 00:36:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"32ccb154ad3503e1a57b2024cf00997af668a1bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 04:14:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72e96e6164c14793ab652b1e39267692 : SUCCESS in 17m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faa2b5ecd3f6429489ec18a1fbc2fcc6 : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7a953fe58a548b087c0bf224ec1b4e9 : SUCCESS in 15m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce1b5b55e1764fa280a02f2b38423936 : SUCCESS in 15m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bd24b1b5bd24f71a74006ed2e362297 : SUCCESS in 7m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f9ff2f861f2247238e9a486dfe6be5e4 : SUCCESS in 36m 54s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/95bf37b0d818406992bf7ebbe9774979 : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":2},{"id":"167359c49cd3bc3980ae05187b4d1e9a55bc1345","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 04:14:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"89294a500a5b9e2ca3a4a2e9a83b9a801a58f43b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 04:17:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a73cb93ce474035abdb5a8f09f7f6f5 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2},{"id":"11ee7a1a05ed2c3bf1a26bc146f5b7df3a32f1c7","author":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"date":"2020-10-12 13:43:49.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 010d8f74a92f471d9c45bc3d4451b4be0985c9bf","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6a9fce39d6cf0051c6ea0f3bf5d57b9a7a03c3b8","revisions":{"c3baba82924650e40e54e4d17fe58475b6347e65":{"kind":"REWORK","_number":1,"created":"2020-08-13 15:08:00.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/54/746154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/54/746154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/746154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/746154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/746154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/54/746154/1"}}},"commit":{"parents":[{"commit":"b4dad5bb3dbeb52c116b84a16f97ea6cb7d6abc2","subject":"Merge \"Clean up some of the OVN agent API methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b4dad5bb3dbeb52c116b84a16f97ea6cb7d6abc2"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-08-13 14:56:23.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-08-13 15:07:16.000000000","tz":-240},"subject":"\"round\" may bump 1 second to 2 if sleep takes more than 1.49 sec","message":"\"round\" may bump 1 second to 2 if sleep takes more than 1.49 sec\n\nThis is a follow up to: https://review.opendev.org/#/c/691883/\n\nTrivial-Fix\n\nChange-Id: Ia313b9c4028f0242cb0da8764b9921c5c30c465a\nFixes-Bug: #1891517\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3baba82924650e40e54e4d17fe58475b6347e65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3baba82924650e40e54e4d17fe58475b6347e65"}]},"branch":"refs/heads/master"},"6a9fce39d6cf0051c6ea0f3bf5d57b9a7a03c3b8":{"kind":"REWORK","_number":2,"created":"2020-08-18 13:26:57.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/54/746154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/54/746154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/746154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/746154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/746154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/54/746154/2"}}},"commit":{"parents":[{"commit":"54da471a005d9956d9dd980f766b31a28a024201","subject":"Merge \"[ovn]: port forwarding -- Functional test for IDL changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54da471a005d9956d9dd980f766b31a28a024201"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-08-13 14:56:23.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-08-18 13:25:39.000000000","tz":-240},"subject":"\"round\" may bump 1 second to 2 if sleep takes more than 1.49 sec","message":"\"round\" may bump 1 second to 2 if sleep takes more than 1.49 sec\n\nUse less than 2 as a way to account for test on slower systems.\nThis is a follow up to: https://review.opendev.org/#/c/691883/\n\nTrivial-Fix\n\nChange-Id: Ia313b9c4028f0242cb0da8764b9921c5c30c465a\nFixes-Bug: #1891517\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a9fce39d6cf0051c6ea0f3bf5d57b9a7a03c3b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a9fce39d6cf0051c6ea0f3bf5d57b9a7a03c3b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
